)]}'
{"id":"openstack%2Fkolla~661975","triplet_id":"openstack%2Fkolla~master~I13f64fcbfb3150efcb94af267f463eaddb11621d","project":"openstack/kolla","branch":"master","topic":"bug/1830900","hashtags":[],"change_id":"I13f64fcbfb3150efcb94af267f463eaddb11621d","subject":"Use the default policy.json for Qinling","status":"MERGED","created":"2019-05-29 13:03:57.000000000","updated":"2019-06-03 19:25:51.000000000","submitted":"2019-06-03 19:25:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"661975-1559589951226-20f53d74","meta_rev_id":"f67150d14645cd799cd2dadb79bbc9fc58f77600","_number":661975,"virtual_id_number":661975,"owner":{"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-06-03 16:44:50.000000000","_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-06-03 19:25:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-06-03 17:28:58.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},{"value":2,"date":"2019-06-03 15:38:22.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2019-05-29 14:28:47.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},{"value":1,"date":"2019-06-03 15:38:22.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-29 14:28:47.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-06-03 15:38:22.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-06-03 17:28:58.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-06-03 19:25:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"555dbf53fcfb62399a11430a55238d2484475f9d","author":{"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},"date":"2019-05-29 13:03:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"50295d7f63c7de97543aaec7f839c7da603f45b4","author":{"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},"date":"2019-05-29 13:07:51.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7153c58aad4251311c059c684fb4b6437ea25deb","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-05-29 13:38:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/75/661975/1/openstack-check/tripleo-build-containers-fedora-28-master/a593633/ : SUCCESS in 31m 42s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/75/661975/1/openstack-check/tripleo-build-containers-fedora-28-stein/8eeb930/ : SUCCESS in 33m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"713af7929fd0bf184edec9c2356223378cdb690f","author":{"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},"date":"2019-05-29 14:10:54.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"82b1df5a57fc2d035842dbfcb77406ee75f21c18","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-05-29 14:28:47.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"e72ba63d437c4e5ca37733cfbed389738c735725","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-05-29 14:47:52.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/75/661975/2/openstack-check/tripleo-build-containers-fedora-28-master/7039cdd/ : SUCCESS in 33m 00s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/75/661975/2/openstack-check/tripleo-build-containers-fedora-28-stein/7216d20/ : SUCCESS in 34m 28s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e97d8797b31fa86121245eeaed0c796d237ccdce","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-29 15:57:43.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/75/661975/2/check/kolla-build-centos-binary/cf443be/ : SUCCESS in 42m 41s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/75/661975/2/check/kolla-build-centos-source/b4f7ff6/ : SUCCESS in 58m 28s\n- kolla-ansible-centos-source http://logs.openstack.org/75/661975/2/check/kolla-ansible-centos-source/7f6ae1d/ : SUCCESS in 1h 11m 14s\n- kolla-ansible-centos-binary http://logs.openstack.org/75/661975/2/check/kolla-ansible-centos-binary/799f643/ : SUCCESS in 1h 02m 16s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/75/661975/2/check/kolla-ansible-centos-source-upgrade/29a26ec/ : SUCCESS in 1h 19m 28s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/75/661975/2/check/tripleo-build-containers-centos-7/d865b91/ : SUCCESS in 45m 38s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/75/661975/2/check/kolla-build-oraclelinux-binary/b67bc6d/ : SUCCESS in 1h 04m 01s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/75/661975/2/check/kolla-build-oraclelinux-source/44ebaca/ : SUCCESS in 50m 35s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/75/661975/2/check/kolla-ansible-oraclelinux-source/7a77b02/ : SUCCESS in 1h 19m 49s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/75/661975/2/check/kolla-ansible-oraclelinux-binary/0972f01/ : SUCCESS in 1h 03m 48s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/75/661975/2/check/openstack-tox-lower-constraints/566fb11/ : SUCCESS in 3m 52s\n- openstack-tox-pep8 http://logs.openstack.org/75/661975/2/check/openstack-tox-pep8/2e3a29a/ : SUCCESS in 3m 05s\n- openstack-tox-py27 http://logs.openstack.org/75/661975/2/check/openstack-tox-py27/03065e3/ : SUCCESS in 3m 04s\n- openstack-tox-py36 http://logs.openstack.org/75/661975/2/check/openstack-tox-py36/8db5fd7/ : SUCCESS in 3m 14s\n- openstack-tox-py37 http://logs.openstack.org/75/661975/2/check/openstack-tox-py37/aeda335/ : SUCCESS in 4m 05s\n- openstack-tox-docs http://logs.openstack.org/75/661975/2/check/openstack-tox-docs/2a72ecb/html/ : SUCCESS in 2m 50s\n- kolla-build-ubuntu-binary http://logs.openstack.org/75/661975/2/check/kolla-build-ubuntu-binary/f86d05a/ : SUCCESS in 35m 44s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/75/661975/2/check/kolla-build-ubuntu-source/6313cb7/ : SUCCESS in 31m 23s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/75/661975/2/check/kolla-ansible-ubuntu-source/d847ae8/ : SUCCESS in 51m 46s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/75/661975/2/check/kolla-ansible-ubuntu-binary/140c757/ : SUCCESS in 44m 33s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/75/661975/2/check/kolla-ansible-ubuntu-source-upgrade/7cf37b9/ : SUCCESS in 1h 21m 48s","accounts_in_message":[],"_revision_number":2},{"id":"844700a36b8ce9e4b705883d37314579a50f75d3","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-03 15:38:22.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"82987e999293af991054e39ccd8f0d19a91c7aae","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-03 15:38:39.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"7fa7bafc4140a3dd6ebafa4df8205dcf8a36cc69","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-03 16:39:17.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-source http://logs.openstack.org/75/661975/2/gate/kolla-build-centos-source/bf5a8d8/ : SUCCESS in 54m 29s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/75/661975/2/gate/tripleo-build-containers-centos-7/ce8f51a/ : SUCCESS in 49m 56s\n- kolla-build-oraclelinux-source http://logs.openstack.org/75/661975/2/gate/kolla-build-oraclelinux-source/e89bfb3/ : SUCCESS in 54m 44s\n- openstack-tox-lower-constraints http://logs.openstack.org/75/661975/2/gate/openstack-tox-lower-constraints/a5b75b5/ : SUCCESS in 5m 12s\n- openstack-tox-pep8 http://logs.openstack.org/75/661975/2/gate/openstack-tox-pep8/e9cc00f/ : SUCCESS in 3m 01s\n- openstack-tox-py27 http://logs.openstack.org/75/661975/2/gate/openstack-tox-py27/bc5cf4c/ : SUCCESS in 3m 57s\n- openstack-tox-py36 http://logs.openstack.org/75/661975/2/gate/openstack-tox-py36/05ff4a9/ : SUCCESS in 3m 01s\n- openstack-tox-py37 http://logs.openstack.org/75/661975/2/gate/openstack-tox-py37/b4f43f4/ : SUCCESS in 3m 39s\n- openstack-tox-docs http://logs.openstack.org/75/661975/2/gate/openstack-tox-docs/977dbed/html/ : SUCCESS in 2m 24s\n- kolla-build-ubuntu-source http://logs.openstack.org/75/661975/2/gate/kolla-build-ubuntu-source/c582039/ : RETRY_LIMIT in 1m 23s","accounts_in_message":[],"_revision_number":2},{"id":"e409cc6bd0e62f6fc9b741932ee8b042a8cebb57","author":{"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},"date":"2019-06-03 16:44:50.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"ff75057b611bc0d0d981adacacb6ebc0fd2f9e86","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-03 17:28:58.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/75/661975/2/openstack-check/tripleo-build-containers-fedora-28-master/93fcd76/ : SUCCESS in 40m 32s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/75/661975/2/openstack-check/tripleo-build-containers-fedora-28-stein/ff0ce02/ : SUCCESS in 37m 31s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c6c1619ec201282e619ea74e8ff45ece9298198b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-03 18:28:30.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/75/661975/2/check/kolla-build-centos-binary/38cb681/ : SUCCESS in 36m 49s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/75/661975/2/check/kolla-build-centos-source/c1ef117/ : SUCCESS in 47m 52s\n- kolla-ansible-centos-source http://logs.openstack.org/75/661975/2/check/kolla-ansible-centos-source/be7b5fb/ : SUCCESS in 1h 06m 10s\n- kolla-ansible-centos-binary http://logs.openstack.org/75/661975/2/check/kolla-ansible-centos-binary/f0448be/ : SUCCESS in 1h 08m 25s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/75/661975/2/check/kolla-ansible-centos-source-upgrade/e308543/ : SUCCESS in 1h 19m 56s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/75/661975/2/check/tripleo-build-containers-centos-7/391e977/ : SUCCESS in 43m 11s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/75/661975/2/check/kolla-build-oraclelinux-binary/a2ca107/ : SUCCESS in 1h 08m 38s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/75/661975/2/check/kolla-build-oraclelinux-source/83ad99c/ : SUCCESS in 42m 55s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/75/661975/2/check/kolla-ansible-oraclelinux-source/a7ef1a8/ : SUCCESS in 1h 05m 48s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/75/661975/2/check/kolla-ansible-oraclelinux-binary/7dcca4c/ : SUCCESS in 1h 10m 48s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/75/661975/2/check/openstack-tox-lower-constraints/dab2cba/ : SUCCESS in 3m 51s\n- openstack-tox-pep8 http://logs.openstack.org/75/661975/2/check/openstack-tox-pep8/9ca5ac4/ : SUCCESS in 3m 14s\n- openstack-tox-py27 http://logs.openstack.org/75/661975/2/check/openstack-tox-py27/56a08e7/ : SUCCESS in 3m 08s\n- openstack-tox-py36 http://logs.openstack.org/75/661975/2/check/openstack-tox-py36/4b703f8/ : SUCCESS in 3m 05s\n- openstack-tox-py37 http://logs.openstack.org/75/661975/2/check/openstack-tox-py37/7e09fb8/ : SUCCESS in 3m 15s\n- openstack-tox-docs http://logs.openstack.org/75/661975/2/check/openstack-tox-docs/8e49677/html/ : SUCCESS in 2m 53s\n- kolla-build-ubuntu-binary http://logs.openstack.org/75/661975/2/check/kolla-build-ubuntu-binary/c6c4d07/ : SUCCESS in 57m 35s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/75/661975/2/check/kolla-build-ubuntu-source/4a2dbda/ : SUCCESS in 56m 22s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/75/661975/2/check/kolla-ansible-ubuntu-source/c536cf2/ : SUCCESS in 57m 36s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/75/661975/2/check/kolla-ansible-ubuntu-binary/3bf30e8/ : SUCCESS in 46m 03s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/75/661975/2/check/kolla-ansible-ubuntu-source-upgrade/e965b80/ : SUCCESS in 1h 03m 25s","accounts_in_message":[],"_revision_number":2},{"id":"6b0de2a458b02330fc1667572a8c3039f8bae29c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-03 18:28:43.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6c64e6fe8c44b850ed2361c056f531127bd0a636","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-03 19:25:50.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source http://logs.openstack.org/75/661975/2/gate/kolla-build-centos-source/79c066d/ : SUCCESS in 51m 38s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/75/661975/2/gate/tripleo-build-containers-centos-7/b33a59c/ : SUCCESS in 43m 10s\n- kolla-build-oraclelinux-source http://logs.openstack.org/75/661975/2/gate/kolla-build-oraclelinux-source/861caab/ : SUCCESS in 46m 07s\n- openstack-tox-lower-constraints http://logs.openstack.org/75/661975/2/gate/openstack-tox-lower-constraints/5a3780b/ : SUCCESS in 3m 26s\n- openstack-tox-pep8 http://logs.openstack.org/75/661975/2/gate/openstack-tox-pep8/c24c7b0/ : SUCCESS in 2m 55s\n- openstack-tox-py27 http://logs.openstack.org/75/661975/2/gate/openstack-tox-py27/e9c37ed/ : SUCCESS in 2m 55s\n- openstack-tox-py36 http://logs.openstack.org/75/661975/2/gate/openstack-tox-py36/eb71d05/ : SUCCESS in 3m 11s\n- openstack-tox-py37 http://logs.openstack.org/75/661975/2/gate/openstack-tox-py37/b0fa483/ : SUCCESS in 4m 05s\n- openstack-tox-docs http://logs.openstack.org/75/661975/2/gate/openstack-tox-docs/e01ac42/html/ : SUCCESS in 2m 55s\n- kolla-build-ubuntu-source http://logs.openstack.org/75/661975/2/gate/kolla-build-ubuntu-source/7e7be25/ : SUCCESS in 38m 11s","accounts_in_message":[],"_revision_number":2},{"id":"a6d4b5f7af4eb58ed74a227673fea4364dd17865","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-03 19:25:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c0f9ced6f2788a86c46b3214a6f3827c6a0dd122","revisions":{"98f281ca0a8537cfb78de400f9047406a48f16ee":{"kind":"REWORK","_number":1,"created":"2019-05-29 13:03:57.000000000","uploader":{"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},"ref":"refs/changes/75/661975/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/75/661975/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/75/661975/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/75/661975/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/75/661975/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/75/661975/1"}}},"commit":{"parents":[{"commit":"3e9cf08ac59c6b38b1542ac12e19ff44af92eb12","subject":"Merge \"Install fluentd gem on debian/ubuntu for non x86_64\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3e9cf08ac59c6b38b1542ac12e19ff44af92eb12"}]}],"author":{"name":"Gaetan Trellu","email":"gaetan.trellu@incloudus.com","date":"2019-05-29 12:59:05.000000000","tz":-240},"committer":{"name":"Gaetan Trellu","email":"gaetan.trellu@incloudus.com","date":"2019-05-29 12:59:05.000000000","tz":-240},"subject":"Use the default policy.json for Qinling","message":"Use the default policy.json for Qinling\n\nTo avoid complexity and possible issue during the Qinling usage\nwe should build the Qinling Docker images with the policy.json file.\n\nThe reason why is because Qinling doesn\u0027t have default policies.\n\nChange-Id: I13f64fcbfb3150efcb94af267f463eaddb11621d\nCloses-Bug: #1830900\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/98f281ca0a8537cfb78de400f9047406a48f16ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/98f281ca0a8537cfb78de400f9047406a48f16ee"}]},"branch":"refs/heads/master"},"c0f9ced6f2788a86c46b3214a6f3827c6a0dd122":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-05-29 14:10:54.000000000","uploader":{"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},"ref":"refs/changes/75/661975/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/75/661975/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/75/661975/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/75/661975/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/75/661975/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/75/661975/2"}}},"commit":{"parents":[{"commit":"3e9cf08ac59c6b38b1542ac12e19ff44af92eb12","subject":"Merge \"Install fluentd gem on debian/ubuntu for non x86_64\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3e9cf08ac59c6b38b1542ac12e19ff44af92eb12"}]}],"author":{"name":"Gaetan Trellu","email":"gaetan.trellu@incloudus.com","date":"2019-05-29 12:59:05.000000000","tz":-240},"committer":{"name":"Gaetan Trellu","email":"gaetan.trellu@incloudus.com","date":"2019-05-29 14:10:32.000000000","tz":-240},"subject":"Use the default policy.json for Qinling","message":"Use the default policy.json for Qinling\n\nTo avoid complexity and possible issue during the Qinling usage\nwe should build the Qinling Docker images with the policy.json file.\n\nThe reason why is because Qinling doesn\u0027t have default policies.\n\nChange-Id: I13f64fcbfb3150efcb94af267f463eaddb11621d\nCloses-Bug: #1830900\nImplements: blueprint qinling-support\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c0f9ced6f2788a86c46b3214a6f3827c6a0dd122"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c0f9ced6f2788a86c46b3214a6f3827c6a0dd122"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
