)]}'
{"id":"openstack%2Fkolla~662429","triplet_id":"openstack%2Fkolla~master~If8434c6c05e415ac8d601260d02b25facbdef1f5","project":"openstack/kolla","branch":"master","topic":"662429","hashtags":[],"change_id":"If8434c6c05e415ac8d601260d02b25facbdef1f5","subject":"grafana/aarch64: use package from upstream repositories","status":"MERGED","created":"2019-05-31 10:31:35.000000000","updated":"2019-06-17 16:09:06.000000000","submitted":"2019-06-17 16:09:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"662429-1560787746138-b95397b7","meta_rev_id":"015cad92ac72cb488900c63ee1b55a4c752d1802","_number":662429,"virtual_id_number":662429,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},{"value":0,"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"date":"2019-06-07 21:55:36.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-06-17 16:09:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-06-07 22:32:19.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":2,"date":"2019-06-17 14:41:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},{"value":0,"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2019-06-11 14:26:13.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":1,"date":"2019-06-17 14:41:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},{"value":0,"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},{"value":0,"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},{"value":0,"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-03 13:02:45.000000000","updated_by":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"reviewer":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"state":"REVIEWER"},{"updated":"2019-06-04 10:26:43.000000000","updated_by":{"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},"reviewer":{"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},"state":"REVIEWER"},{"updated":"2019-06-05 05:49:23.000000000","updated_by":{"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},"reviewer":{"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},"state":"REVIEWER"},{"updated":"2019-06-06 11:14:16.000000000","updated_by":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"reviewer":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"state":"REVIEWER"},{"updated":"2019-06-07 04:19:14.000000000","updated_by":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"reviewer":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"state":"REVIEWER"},{"updated":"2019-06-07 22:32:19.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-06-11 14:26:13.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-06-17 14:41:24.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2019-06-17 16:09:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d4459b1bec452a893e9e702cd73b807a17cd70b4","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-05-31 10:31:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"54006570f4d7eaab475a090837cbbdf77dd8c8b4","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-05-31 11:21:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/29/662429/1/openstack-check/tripleo-build-containers-fedora-28-master/f203314/ : SUCCESS in 47m 49s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/29/662429/1/openstack-check/tripleo-build-containers-fedora-28-stein/adb8c7b/ : SUCCESS in 46m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9ba4495d011f83ef114fd39c3e3ed2860fe7f484","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-31 11:54:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/29/662429/1/check/kolla-build-centos-binary/aff195f/ : SUCCESS in 31m 28s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/29/662429/1/check/kolla-build-centos-source/66b97fc/ : SUCCESS in 51m 47s\n- kolla-ansible-centos-source http://logs.openstack.org/29/662429/1/check/kolla-ansible-centos-source/b2bad7e/ : SUCCESS in 1h 09m 20s\n- kolla-ansible-centos-binary http://logs.openstack.org/29/662429/1/check/kolla-ansible-centos-binary/110d405/ : FAILURE in 29m 12s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/29/662429/1/check/kolla-ansible-centos-source-upgrade/1fb4a96/ : SUCCESS in 1h 12m 30s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/29/662429/1/check/tripleo-build-containers-centos-7/ea28f95/ : SUCCESS in 40m 48s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/29/662429/1/check/kolla-build-oraclelinux-binary/d1bb1c3/ : SUCCESS in 38m 31s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/29/662429/1/check/kolla-build-oraclelinux-source/387aa42/ : SUCCESS in 44m 56s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/29/662429/1/check/kolla-ansible-oraclelinux-source/fff1929/ : FAILURE in 36m 46s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/29/662429/1/check/kolla-ansible-oraclelinux-binary/6db68b9/ : SUCCESS in 1h 12m 28s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/29/662429/1/check/openstack-tox-lower-constraints/2656eaf/ : SUCCESS in 3m 40s\n- openstack-tox-pep8 http://logs.openstack.org/29/662429/1/check/openstack-tox-pep8/07e3868/ : SUCCESS in 4m 02s\n- openstack-tox-py27 http://logs.openstack.org/29/662429/1/check/openstack-tox-py27/39b467a/ : SUCCESS in 3m 32s\n- openstack-tox-py36 http://logs.openstack.org/29/662429/1/check/openstack-tox-py36/e670ad6/ : SUCCESS in 4m 02s\n- openstack-tox-py37 http://logs.openstack.org/29/662429/1/check/openstack-tox-py37/bc63f13/ : SUCCESS in 3m 46s\n- openstack-tox-docs http://logs.openstack.org/29/662429/1/check/openstack-tox-docs/40f59d7/html/ : SUCCESS in 3m 44s\n- kolla-build-ubuntu-binary http://logs.openstack.org/29/662429/1/check/kolla-build-ubuntu-binary/82e04d9/ : SUCCESS in 34m 25s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/29/662429/1/check/kolla-build-ubuntu-source/ac62d33/ : SUCCESS in 35m 46s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/29/662429/1/check/kolla-ansible-ubuntu-source/d0dc850/ : SUCCESS in 56m 47s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/29/662429/1/check/kolla-ansible-ubuntu-binary/6831e84/ : SUCCESS in 51m 22s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/29/662429/1/check/kolla-ansible-ubuntu-source-upgrade/aa99ab4/ : SUCCESS in 1h 04m 47s","accounts_in_message":[],"_revision_number":1},{"id":"dd4a60fc31f6aa19dc8d1a57e599f8486a756f8e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-03 13:03:38.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"17c12dd6440573aa0d0657609b3368d3e3edf319","author":{"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},"date":"2019-06-04 10:26:43.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"75880246ab2fb24edb69ffde5d8775e5456991d0","author":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"date":"2019-06-06 11:14:16.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"620ddf6525f198391534b7e10b7b1049e430847a","author":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"date":"2019-06-07 04:19:14.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"520ace8edb1d18039365ef17918c5b6b7dc7b953","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-07 15:54:35.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0334deaf0b7411cb77b5ec6d06d57b76f628f7ab","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-07 16:42:34.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/29/662429/2/openstack-check/tripleo-build-containers-fedora-28-master/c484bec/ : SUCCESS in 41m 07s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/29/662429/2/openstack-check/tripleo-build-containers-fedora-28-stein/70c31f5/ : SUCCESS in 43m 37s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"45b940ca5254dd6e8e5c27060ba62b5da0e3a02a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 17:35:54.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/29/662429/2/check/kolla-build-centos-binary/e793695/ : SUCCESS in 42m 50s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/29/662429/2/check/kolla-build-centos-source/ed694ed/ : FAILURE in 10m 26s\n- kolla-ansible-centos-source http://logs.openstack.org/29/662429/2/check/kolla-ansible-centos-source/068a5f8/ : SUCCESS in 1h 18m 47s\n- kolla-ansible-centos-binary http://logs.openstack.org/29/662429/2/check/kolla-ansible-centos-binary/5524d8e/ : FAILURE in 12m 41s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/29/662429/2/check/kolla-ansible-centos-source-upgrade/d04c9c4/ : SUCCESS in 1h 21m 06s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/29/662429/2/check/tripleo-build-containers-centos-7/c108fc1/ : SUCCESS in 36m 44s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/29/662429/2/check/kolla-build-oraclelinux-binary/328b4a5/ : SUCCESS in 40m 17s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/29/662429/2/check/kolla-build-oraclelinux-source/6123f8c/ : SUCCESS in 49m 48s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/29/662429/2/check/kolla-ansible-oraclelinux-source/f25c712/ : SUCCESS in 1h 28m 02s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/29/662429/2/check/kolla-ansible-oraclelinux-binary/8428f6c/ : SUCCESS in 1h 18m 01s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/29/662429/2/check/openstack-tox-lower-constraints/9348be8/ : SUCCESS in 3m 51s\n- openstack-tox-pep8 http://logs.openstack.org/29/662429/2/check/openstack-tox-pep8/5af504c/ : SUCCESS in 3m 14s\n- openstack-tox-py27 http://logs.openstack.org/29/662429/2/check/openstack-tox-py27/2e8d74c/ : SUCCESS in 3m 12s\n- openstack-tox-py36 http://logs.openstack.org/29/662429/2/check/openstack-tox-py36/67d0522/ : SUCCESS in 4m 29s\n- openstack-tox-py37 http://logs.openstack.org/29/662429/2/check/openstack-tox-py37/ec4534e/ : SUCCESS in 5m 47s\n- openstack-tox-docs http://logs.openstack.org/29/662429/2/check/openstack-tox-docs/06ccbf7/html/ : SUCCESS in 4m 00s\n- kolla-build-ubuntu-binary http://logs.openstack.org/29/662429/2/check/kolla-build-ubuntu-binary/4b80c0d/ : SUCCESS in 28m 04s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/29/662429/2/check/kolla-build-ubuntu-source/8fefa6b/ : SUCCESS in 32m 19s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/29/662429/2/check/kolla-ansible-ubuntu-source/e41955a/ : SUCCESS in 44m 32s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/29/662429/2/check/kolla-ansible-ubuntu-binary/8f8f5e5/ : SUCCESS in 45m 03s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/29/662429/2/check/kolla-ansible-ubuntu-source-upgrade/86f8495/ : SUCCESS in 46m 45s","accounts_in_message":[],"_revision_number":2},{"id":"b0af5771b217a58152eef7cbe7b757dce8694328","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-07 21:55:36.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"c3b07b3023f2bf75b87cbb3441380523a455f08d","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-07 22:32:19.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/29/662429/2/openstack-check/tripleo-build-containers-fedora-28-master/94dfa75/ : SUCCESS in 35m 42s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/29/662429/2/openstack-check/tripleo-build-containers-fedora-28-stein/c2d166f/ : SUCCESS in 35m 08s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6033c78f554328ef9880cbf2e54819b6c0ab1ed8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 23:29:51.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/29/662429/2/check/kolla-build-centos-binary/55f93b7/ : SUCCESS in 39m 44s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/29/662429/2/check/kolla-build-centos-source/5b40e0b/ : SUCCESS in 47m 56s\n- kolla-ansible-centos-source http://logs.openstack.org/29/662429/2/check/kolla-ansible-centos-source/060fd85/ : SUCCESS in 1h 22m 00s\n- kolla-ansible-centos-binary http://logs.openstack.org/29/662429/2/check/kolla-ansible-centos-binary/8667a78/ : SUCCESS in 1h 09m 52s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/29/662429/2/check/kolla-ansible-centos-source-upgrade/ed2c3a6/ : SUCCESS in 1h 27m 33s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/29/662429/2/check/tripleo-build-containers-centos-7/b408841/ : SUCCESS in 42m 08s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/29/662429/2/check/kolla-build-oraclelinux-binary/5bba882/ : SUCCESS in 41m 18s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/29/662429/2/check/kolla-build-oraclelinux-source/5c8888a/ : SUCCESS in 53m 42s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/29/662429/2/check/kolla-ansible-oraclelinux-source/27d3f24/ : SUCCESS in 1h 23m 53s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/29/662429/2/check/kolla-ansible-oraclelinux-binary/656fc0f/ : SUCCESS in 1h 03m 55s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/29/662429/2/check/openstack-tox-lower-constraints/fd8cf8b/ : SUCCESS in 3m 29s\n- openstack-tox-pep8 http://logs.openstack.org/29/662429/2/check/openstack-tox-pep8/8c20f46/ : SUCCESS in 3m 15s\n- openstack-tox-py27 http://logs.openstack.org/29/662429/2/check/openstack-tox-py27/cb57b88/ : SUCCESS in 4m 22s\n- openstack-tox-py36 http://logs.openstack.org/29/662429/2/check/openstack-tox-py36/23c1c0d/ : SUCCESS in 3m 28s\n- openstack-tox-py37 http://logs.openstack.org/29/662429/2/check/openstack-tox-py37/5f20848/ : SUCCESS in 3m 16s\n- openstack-tox-docs http://logs.openstack.org/29/662429/2/check/openstack-tox-docs/64146cd/html/ : SUCCESS in 2m 40s\n- kolla-build-ubuntu-binary http://logs.openstack.org/29/662429/2/check/kolla-build-ubuntu-binary/73735d2/ : SUCCESS in 27m 01s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/29/662429/2/check/kolla-build-ubuntu-source/8ee2e90/ : SUCCESS in 29m 38s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/29/662429/2/check/kolla-ansible-ubuntu-source/8a3149d/ : SUCCESS in 43m 04s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/29/662429/2/check/kolla-ansible-ubuntu-binary/6307b83/ : SUCCESS in 42m 55s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/29/662429/2/check/kolla-ansible-ubuntu-source-upgrade/1243610/ : SUCCESS in 56m 35s","accounts_in_message":[],"_revision_number":2},{"id":"287d5613e8c992752b0ac986ab539de4c096c41b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-11 14:26:13.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"53fdb64a131c8a5a0d58f7f2bc28e239043a0caa","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-06-17 14:41:24.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"2b1b80ab559a6da6f23b3eac66482d79e4b58224","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-17 14:41:41.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"96aedefa4d4ba1561880c1c81b55e1cc99b4869a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-17 16:09:05.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source http://logs.openstack.org/29/662429/2/gate/kolla-build-centos-source/3c990ad/ : SUCCESS in 1h 09m 30s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/29/662429/2/gate/tripleo-build-containers-centos-7/14c9345/ : SUCCESS in 45m 26s\n- kolla-build-oraclelinux-source http://logs.openstack.org/29/662429/2/gate/kolla-build-oraclelinux-source/33549f0/ : SUCCESS in 41m 55s\n- openstack-tox-lower-constraints http://logs.openstack.org/29/662429/2/gate/openstack-tox-lower-constraints/30dd9b6/ : SUCCESS in 3m 19s\n- openstack-tox-pep8 http://logs.openstack.org/29/662429/2/gate/openstack-tox-pep8/bf916bb/ : SUCCESS in 2m 48s\n- openstack-tox-py27 http://logs.openstack.org/29/662429/2/gate/openstack-tox-py27/8a0a139/ : SUCCESS in 4m 25s\n- openstack-tox-py36 http://logs.openstack.org/29/662429/2/gate/openstack-tox-py36/1c662a7/ : SUCCESS in 2m 55s\n- openstack-tox-py37 http://logs.openstack.org/29/662429/2/gate/openstack-tox-py37/1fb2c0b/ : SUCCESS in 4m 06s\n- openstack-tox-docs http://logs.openstack.org/29/662429/2/gate/openstack-tox-docs/e9369b9/html/ : SUCCESS in 2m 27s\n- kolla-build-ubuntu-source http://logs.openstack.org/29/662429/2/gate/kolla-build-ubuntu-source/8afd654/ : SUCCESS in 28m 53s","accounts_in_message":[],"_revision_number":2},{"id":"3b20067e3300c99e534dd6a941ace1e845635a07","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-17 16:09:06.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1e7a78932dc0461c60d9fd6e613b252d8b22ff01","revisions":{"e9123ec25a67390770254c9f30d79c56a4b155fc":{"kind":"REWORK","_number":1,"created":"2019-05-31 10:31:35.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/29/662429/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/29/662429/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/662429/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/662429/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/662429/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/29/662429/1"}}},"commit":{"parents":[{"commit":"3e9cf08ac59c6b38b1542ac12e19ff44af92eb12","subject":"Merge \"Install fluentd gem on debian/ubuntu for non x86_64\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3e9cf08ac59c6b38b1542ac12e19ff44af92eb12"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-05-31 10:21:08.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-05-31 10:21:08.000000000","tz":120},"subject":"grafana/aarch64: use package from upstream repositories","message":"grafana/aarch64: use package from upstream repositories\n\nWe were installing 5.2.4 version instead of just latest one like it is\ndone on x86-64.\n\nChange-Id: If8434c6c05e415ac8d601260d02b25facbdef1f5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e9123ec25a67390770254c9f30d79c56a4b155fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e9123ec25a67390770254c9f30d79c56a4b155fc"}]},"branch":"refs/heads/master"},"1e7a78932dc0461c60d9fd6e613b252d8b22ff01":{"kind":"REWORK","_number":2,"created":"2019-06-07 15:54:35.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/29/662429/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/29/662429/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/662429/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/662429/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/662429/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/29/662429/2"}}},"commit":{"parents":[{"commit":"498c57fcac31da9268918ec6c188fe2aae0c4906","subject":"Merge \"Use packagecloud/ubuntu rabbitmq repo for all archs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/498c57fcac31da9268918ec6c188fe2aae0c4906"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-05-31 10:21:08.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-06-07 15:54:13.000000000","tz":120},"subject":"grafana/aarch64: use package from upstream repositories","message":"grafana/aarch64: use package from upstream repositories\n\nWe were installing 5.2.4 version instead of just latest one like it is\ndone on x86-64.\n\nChange-Id: If8434c6c05e415ac8d601260d02b25facbdef1f5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1e7a78932dc0461c60d9fd6e613b252d8b22ff01"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1e7a78932dc0461c60d9fd6e613b252d8b22ff01"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
