)]}'
{"id":"openstack%2Fkolla~662456","triplet_id":"openstack%2Fkolla~master~Ic32c19109d8521b93b7e124ac4ea9224767aad5c","project":"openstack/kolla","branch":"master","topic":"ci/add-centos-aarch64-job","hashtags":[],"change_id":"Ic32c19109d8521b93b7e124ac4ea9224767aad5c","subject":"Add CentOS/source aarch64 experimental job","status":"ABANDONED","created":"2019-05-31 13:00:07.000000000","updated":"2019-12-13 12:13:33.000000000","total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"bd26deceae5ba25aa709b0697b05d5a8495a1fd8","_number":662456,"virtual_id_number":662456,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"tag":"autogenerated:zuul:check","value":1,"date":"2019-12-11 11:21:35.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22076,"name":"Kevin Zhao","email":"kevin.zhao@linaro.org","username":"KevinZhao"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22076,"name":"Kevin Zhao","email":"kevin.zhao@linaro.org","username":"KevinZhao"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":-1,"date":"2019-12-11 19:47:22.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22076,"name":"Kevin Zhao","email":"kevin.zhao@linaro.org","username":"KevinZhao"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22076,"name":"Kevin Zhao","email":"kevin.zhao@linaro.org","username":"KevinZhao"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22076,"name":"Kevin Zhao","email":"kevin.zhao@linaro.org","username":"KevinZhao"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":22076,"name":"Kevin Zhao","email":"kevin.zhao@linaro.org","username":"KevinZhao"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-15 11:42:17.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-12-11 05:37:36.000000000","updated_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"reviewer":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"state":"REVIEWER"},{"updated":"2019-12-11 07:01:17.000000000","updated_by":{"_account_id":22076,"name":"Kevin Zhao","email":"kevin.zhao@linaro.org","username":"KevinZhao"},"reviewer":{"_account_id":22076,"name":"Kevin Zhao","email":"kevin.zhao@linaro.org","username":"KevinZhao"},"state":"REVIEWER"},{"updated":"2019-12-11 11:21:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c4e0a3d6c33b21c116fe6cebd31a04c501fdeab2","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-05-31 13:00:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"525cf8af3b162e7ab820795c22feeb1f823172e0","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-05-31 13:00:27.000000000","message":"Patch Set 1:\n\ncheck experimental","accounts_in_message":[],"_revision_number":1},{"id":"18abe2e2b8b69a5984dfe5ad9fd050a313d900ac","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-31 13:01:24.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/kolla on branch master.  The error was:\n\n  The nodeset \"kolla-debian-arm64\" was not found.\n\nThe error appears in the following job stanza:\n\n  job:\n      name: kolla-build-centos-source-arm64\n      parent: kolla-base\n      nodeset: kolla-debian-arm64\n      voting: false\n      vars:\n        action: build\n        base_distro: centos\n        install_type: source\n\n  in \"openstack/kolla/.zuul.d/centos.yaml@master\", line 49, column 3","accounts_in_message":[],"_revision_number":1},{"id":"3ee9bb7f5fa417e505a17ac54295984af2fea91a","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-05-31 13:04:04.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f306e00aecc8fc3827b74df59a966c7afd9212f3","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-05-31 13:04:27.000000000","message":"Patch Set 2:\n\ncheck experimental","accounts_in_message":[],"_revision_number":2},{"id":"a23f7948607e06edf06e896ee5428e91c9dfd680","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-31 13:09:47.000000000","message":"Patch Set 2:\n\nBuild succeeded (experimental pipeline).\n\n- kolla-build-centos-source-arm64 http://logs.openstack.org/56/662456/2/experimental/kolla-build-centos-source-arm64/720b207/ : RETRY_LIMIT in 4m 07s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a0fbca838d32cc182d3719ad2a0456cd4f6d7a03","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-05-31 13:18:51.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"2f224995f040116df9d39dcd0842f66f7fa84155","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-05-31 13:19:19.000000000","message":"Patch Set 3:\n\ncheck experimental","accounts_in_message":[],"_revision_number":3},{"id":"707ace68dabeaaa1d6269808e84dcf31cfaf80bc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-31 14:02:54.000000000","message":"Patch Set 3:\n\nBuild succeeded (experimental pipeline).\n\n- kolla-build-centos-source-arm64 kolla-build-centos-source-arm64 : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c330e9c5ff73760979be0f00cec08a70b4bc6be0","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-05-31 14:17:41.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"f0750c7915d620d17881d1ce16a7da8108864d5c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-31 14:44:43.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-source-arm64 kolla-build-centos-source-arm64 : NODE_FAILURE in 0s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/56/662456/4/check/openstack-tox-lower-constraints/e262404/ : SUCCESS in 3m 57s\n- openstack-tox-pep8 http://logs.openstack.org/56/662456/4/check/openstack-tox-pep8/2b528c6/ : SUCCESS in 3m 32s\n- openstack-tox-py27 http://logs.openstack.org/56/662456/4/check/openstack-tox-py27/5753f0f/ : SUCCESS in 3m 54s\n- openstack-tox-py36 http://logs.openstack.org/56/662456/4/check/openstack-tox-py36/9d42af2/ : SUCCESS in 3m 46s\n- openstack-tox-py37 http://logs.openstack.org/56/662456/4/check/openstack-tox-py37/8cee84c/ : SUCCESS in 3m 35s\n- openstack-tox-docs http://logs.openstack.org/56/662456/4/check/openstack-tox-docs/2820223/html/ : SUCCESS in 2m 39s","accounts_in_message":[],"_revision_number":4},{"id":"b8b06ca290a8fb8072d92b65c3e91c60e1adcbeb","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-05-31 18:12:23.000000000","message":"Patch Set 4: Workflow-1\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"5a6ea44b4004f7905c7306f7637ad87596158324","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-31 18:25:25.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-source-arm64 kolla-build-centos-source-arm64 : NODE_FAILURE in 0s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/56/662456/4/check/openstack-tox-lower-constraints/aee6772/ : SUCCESS in 3m 37s\n- openstack-tox-pep8 http://logs.openstack.org/56/662456/4/check/openstack-tox-pep8/8b5a940/ : SUCCESS in 3m 47s\n- openstack-tox-py27 http://logs.openstack.org/56/662456/4/check/openstack-tox-py27/263cafe/ : SUCCESS in 4m 22s\n- openstack-tox-py36 http://logs.openstack.org/56/662456/4/check/openstack-tox-py36/4a23cf0/ : SUCCESS in 3m 31s\n- openstack-tox-py37 http://logs.openstack.org/56/662456/4/check/openstack-tox-py37/2d9b182/ : SUCCESS in 4m 01s\n- openstack-tox-docs http://logs.openstack.org/56/662456/4/check/openstack-tox-docs/bd4cc33/html/ : SUCCESS in 2m 41s","accounts_in_message":[],"_revision_number":4},{"id":"1a275f949064ed02943fe656034f5c68d91c9e2e","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-05-31 18:26:55.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"9ad9e943d81d98d01a23fed44d583e8056cc542d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-31 19:24:04.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-source-arm64 kolla-build-centos-source-arm64 : NODE_FAILURE in 0s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/56/662456/5/check/openstack-tox-lower-constraints/5f5fc5d/ : SUCCESS in 4m 29s\n- openstack-tox-pep8 http://logs.openstack.org/56/662456/5/check/openstack-tox-pep8/4fe2b73/ : SUCCESS in 2m 37s\n- openstack-tox-py27 http://logs.openstack.org/56/662456/5/check/openstack-tox-py27/1e763e0/ : SUCCESS in 2m 59s\n- openstack-tox-py36 http://logs.openstack.org/56/662456/5/check/openstack-tox-py36/d27bb79/ : SUCCESS in 2m 40s\n- openstack-tox-py37 http://logs.openstack.org/56/662456/5/check/openstack-tox-py37/382787d/ : SUCCESS in 3m 06s\n- openstack-tox-docs http://logs.openstack.org/56/662456/5/check/openstack-tox-docs/0afb4b5/html/ : SUCCESS in 2m 26s","accounts_in_message":[],"_revision_number":5},{"id":"93993a3b1e511ee6106c126ecabe49d19ee3c26b","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-01 19:09:27.000000000","message":"Patch Set 5: Workflow-1","accounts_in_message":[],"_revision_number":5},{"id":"a35b553a3ed84b92a3741cb8ff865d2d4af2d53b","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-01 19:09:34.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"0d00fffab28a21c684087f9a2597fd19d5ac5d47","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-01 19:18:32.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-source-arm64 http://logs.openstack.org/56/662456/5/check/kolla-build-centos-source-arm64/802f864/ : RETRY_LIMIT in 4m 31s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/56/662456/5/check/openstack-tox-lower-constraints/a03ddc3/ : SUCCESS in 3m 42s\n- openstack-tox-pep8 http://logs.openstack.org/56/662456/5/check/openstack-tox-pep8/dee2506/ : SUCCESS in 2m 53s\n- openstack-tox-py27 http://logs.openstack.org/56/662456/5/check/openstack-tox-py27/1f9107c/ : SUCCESS in 3m 53s\n- openstack-tox-py36 http://logs.openstack.org/56/662456/5/check/openstack-tox-py36/d4b746e/ : SUCCESS in 3m 27s\n- openstack-tox-py37 http://logs.openstack.org/56/662456/5/check/openstack-tox-py37/249a458/ : SUCCESS in 3m 02s\n- openstack-tox-docs http://logs.openstack.org/56/662456/5/check/openstack-tox-docs/f3a594a/html/ : SUCCESS in 2m 15s","accounts_in_message":[],"_revision_number":5},{"id":"5b15b8ace634c3cc43164d766299e0ac63334b76","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-01 19:21:31.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"f77ac3b1ee27f473b8bdacd90c58215c06ec2f12","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-01 19:29:51.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-source-arm64 http://logs.openstack.org/56/662456/5/check/kolla-build-centos-source-arm64/0f84de9/ : RETRY_LIMIT in 4m 23s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/56/662456/5/check/openstack-tox-lower-constraints/fa47aaf/ : SUCCESS in 3m 18s\n- openstack-tox-pep8 http://logs.openstack.org/56/662456/5/check/openstack-tox-pep8/1bfb5ad/ : SUCCESS in 3m 09s\n- openstack-tox-py27 http://logs.openstack.org/56/662456/5/check/openstack-tox-py27/c9df502/ : SUCCESS in 3m 04s\n- openstack-tox-py36 http://logs.openstack.org/56/662456/5/check/openstack-tox-py36/753f3b3/ : SUCCESS in 2m 55s\n- openstack-tox-py37 http://logs.openstack.org/56/662456/5/check/openstack-tox-py37/8475666/ : SUCCESS in 3m 09s\n- openstack-tox-docs http://logs.openstack.org/56/662456/5/check/openstack-tox-docs/a967b56/html/ : SUCCESS in 2m 48s","accounts_in_message":[],"_revision_number":5},{"id":"66d1355df7afee96c6f62a479f76aef3ee84c94d","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-01 20:30:33.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"c05579d8732e84319cb9be59f1cfb7a4f28c5922","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-01 21:11:41.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-source-arm64 kolla-build-centos-source-arm64 : NODE_FAILURE in 0s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/56/662456/6/check/openstack-tox-lower-constraints/2157f39/ : SUCCESS in 3m 43s\n- openstack-tox-pep8 http://logs.openstack.org/56/662456/6/check/openstack-tox-pep8/00a4f62/ : SUCCESS in 2m 43s\n- openstack-tox-py27 http://logs.openstack.org/56/662456/6/check/openstack-tox-py27/6f9cccd/ : SUCCESS in 2m 50s\n- openstack-tox-py36 http://logs.openstack.org/56/662456/6/check/openstack-tox-py36/40325b4/ : SUCCESS in 2m 58s\n- openstack-tox-py37 http://logs.openstack.org/56/662456/6/check/openstack-tox-py37/cbe08f1/ : SUCCESS in 3m 08s\n- openstack-tox-docs http://logs.openstack.org/56/662456/6/check/openstack-tox-docs/29d1c90/html/ : SUCCESS in 2m 51s","accounts_in_message":[],"_revision_number":6},{"id":"e5bc2ef7092d2feadcd27410bf3d4b5b25ad2fa0","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-02 07:55:44.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"93d6ee263feff8658858306b1e252dd4040a34ac","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-02 08:07:24.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"fe317077a4590972ff5a4db5c0912ad5066e65d4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-02 08:45:04.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-source-arm64 kolla-build-centos-source-arm64 : NODE_FAILURE in 0s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/56/662456/7/check/openstack-tox-lower-constraints/1ab70dd/ : SUCCESS in 3m 38s\n- openstack-tox-pep8 http://logs.openstack.org/56/662456/7/check/openstack-tox-pep8/12df835/ : FAILURE in 2m 42s\n- openstack-tox-py27 http://logs.openstack.org/56/662456/7/check/openstack-tox-py27/421668f/ : SUCCESS in 2m 52s\n- openstack-tox-py36 http://logs.openstack.org/56/662456/7/check/openstack-tox-py36/c3e543c/ : SUCCESS in 3m 36s\n- openstack-tox-py37 http://logs.openstack.org/56/662456/7/check/openstack-tox-py37/a270f6f/ : SUCCESS in 3m 18s\n- openstack-tox-docs http://logs.openstack.org/56/662456/7/check/openstack-tox-docs/db3036a/html/ : SUCCESS in 2m 38s","accounts_in_message":[],"_revision_number":7},{"id":"16fa77f6f1ad7df5914ec30b022bd908303a8a08","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-02 13:32:49.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"3e38e77a4f4ccc1394c2d9990d309652317bb063","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-02 13:42:13.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-source-arm64 http://logs.openstack.org/56/662456/7/check/kolla-build-centos-source-arm64/66feada/ : FAILURE in 8m 57s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/56/662456/7/check/openstack-tox-lower-constraints/d9418ac/ : SUCCESS in 3m 24s\n- openstack-tox-pep8 http://logs.openstack.org/56/662456/7/check/openstack-tox-pep8/d6a14ea/ : FAILURE in 2m 54s\n- openstack-tox-py27 http://logs.openstack.org/56/662456/7/check/openstack-tox-py27/1139437/ : SUCCESS in 3m 02s\n- openstack-tox-py36 http://logs.openstack.org/56/662456/7/check/openstack-tox-py36/95a4d04/ : SUCCESS in 2m 52s\n- openstack-tox-py37 http://logs.openstack.org/56/662456/7/check/openstack-tox-py37/2d81fe9/ : SUCCESS in 3m 04s\n- openstack-tox-docs http://logs.openstack.org/56/662456/7/check/openstack-tox-docs/085da84/html/ : SUCCESS in 2m 39s","accounts_in_message":[],"_revision_number":7},{"id":"2ad73ae47d4f0a206aa7ecdb124a71e59a6b5b35","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-02 13:47:20.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"32188c172bd4f5baf9ad83e81abfb1514c2c6f43","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-02 14:24:53.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-source-arm64 kolla-build-centos-source-arm64 : NODE_FAILURE in 0s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/56/662456/8/check/openstack-tox-lower-constraints/b6fb1bd/ : SUCCESS in 3m 25s\n- openstack-tox-pep8 http://logs.openstack.org/56/662456/8/check/openstack-tox-pep8/a3b657f/ : SUCCESS in 2m 48s\n- openstack-tox-py27 http://logs.openstack.org/56/662456/8/check/openstack-tox-py27/98361f7/ : SUCCESS in 2m 37s\n- openstack-tox-py36 http://logs.openstack.org/56/662456/8/check/openstack-tox-py36/b0a1c50/ : SUCCESS in 2m 52s\n- openstack-tox-py37 http://logs.openstack.org/56/662456/8/check/openstack-tox-py37/50cab34/ : SUCCESS in 2m 58s\n- openstack-tox-docs http://logs.openstack.org/56/662456/8/check/openstack-tox-docs/e127ac5/html/ : SUCCESS in 2m 28s","accounts_in_message":[],"_revision_number":8},{"id":"8caa22c2319b070d9d15bdff4f1cd65ac7c02523","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-02 14:41:37.000000000","message":"Patch Set 8: Workflow-1\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"16bcfa5480b0adfbc0511d871686fab83e4f988e","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-02 14:55:30.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"e2b8806ec4458be02f82c8b8f809f650e73355c8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-02 15:16:29.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-source-arm64 http://logs.openstack.org/56/662456/9/check/kolla-build-centos-source-arm64/80109b1/ : FAILURE in 20m 23s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/56/662456/9/check/openstack-tox-lower-constraints/a52a417/ : SUCCESS in 3m 20s\n- openstack-tox-pep8 http://logs.openstack.org/56/662456/9/check/openstack-tox-pep8/562d1ff/ : SUCCESS in 2m 50s\n- openstack-tox-py27 http://logs.openstack.org/56/662456/9/check/openstack-tox-py27/a843ccd/ : SUCCESS in 3m 00s\n- openstack-tox-py36 http://logs.openstack.org/56/662456/9/check/openstack-tox-py36/afb20f0/ : SUCCESS in 2m 34s\n- openstack-tox-py37 http://logs.openstack.org/56/662456/9/check/openstack-tox-py37/864f76f/ : SUCCESS in 3m 08s\n- openstack-tox-docs http://logs.openstack.org/56/662456/9/check/openstack-tox-docs/0224f16/html/ : SUCCESS in 2m 08s","accounts_in_message":[],"_revision_number":9},{"id":"73ea18142cc43cda58a2441be1a4ba5085fd6a13","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-02 16:17:15.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"a6387318b0e5fe4b91a56cf054e0c434ce820713","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-02 16:33:31.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-source-arm64 http://logs.openstack.org/56/662456/10/check/kolla-build-centos-source-arm64/b3c1390/ : FAILURE in 9m 36s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/56/662456/10/check/openstack-tox-lower-constraints/c827d64/ : SUCCESS in 3m 48s\n- openstack-tox-pep8 http://logs.openstack.org/56/662456/10/check/openstack-tox-pep8/b224be4/ : SUCCESS in 2m 46s\n- openstack-tox-py27 http://logs.openstack.org/56/662456/10/check/openstack-tox-py27/7e469a6/ : SUCCESS in 3m 03s\n- openstack-tox-py36 http://logs.openstack.org/56/662456/10/check/openstack-tox-py36/d880bcd/ : SUCCESS in 3m 19s\n- openstack-tox-py37 http://logs.openstack.org/56/662456/10/check/openstack-tox-py37/7e17356/ : SUCCESS in 3m 22s\n- openstack-tox-docs http://logs.openstack.org/56/662456/10/check/openstack-tox-docs/c1dfc4d/html/ : SUCCESS in 2m 41s","accounts_in_message":[],"_revision_number":10},{"id":"cd9a2a2898efa551975464ac385219df985673b9","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-03 06:19:13.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"1d3d80fff65ceae46d55832a2dcfdf6d6911bdbd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-03 07:02:31.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-source-arm64 http://logs.openstack.org/56/662456/11/check/kolla-build-centos-source-arm64/03031a1/ : FAILURE in 42m 09s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/56/662456/11/check/openstack-tox-lower-constraints/8128814/ : SUCCESS in 4m 06s\n- openstack-tox-pep8 http://logs.openstack.org/56/662456/11/check/openstack-tox-pep8/9a6b80e/ : SUCCESS in 3m 05s\n- openstack-tox-py27 http://logs.openstack.org/56/662456/11/check/openstack-tox-py27/2864459/ : SUCCESS in 3m 28s\n- openstack-tox-py36 http://logs.openstack.org/56/662456/11/check/openstack-tox-py36/2d0aabc/ : SUCCESS in 3m 07s\n- openstack-tox-py37 http://logs.openstack.org/56/662456/11/check/openstack-tox-py37/2d44aaa/ : SUCCESS in 3m 35s\n- openstack-tox-docs http://logs.openstack.org/56/662456/11/check/openstack-tox-docs/e2f0139/html/ : SUCCESS in 2m 46s","accounts_in_message":[],"_revision_number":11},{"id":"bca4f63d2d11c54d4f4b276863047492c0d5d9be","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-07 14:07:53.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"619466a30dc3523d07c6223c7d9ac4775c038654","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 14:57:41.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-source-arm64 http://logs.openstack.org/56/662456/12/check/kolla-build-centos-source-arm64/8cc5bcb/ : FAILURE in 47m 54s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/56/662456/12/check/openstack-tox-lower-constraints/564d0d2/ : SUCCESS in 3m 57s\n- openstack-tox-pep8 http://logs.openstack.org/56/662456/12/check/openstack-tox-pep8/cda1f3e/ : SUCCESS in 3m 27s\n- openstack-tox-py27 http://logs.openstack.org/56/662456/12/check/openstack-tox-py27/88eb45a/ : SUCCESS in 3m 11s\n- openstack-tox-py36 http://logs.openstack.org/56/662456/12/check/openstack-tox-py36/7bd4a8d/ : SUCCESS in 3m 31s\n- openstack-tox-py37 http://logs.openstack.org/56/662456/12/check/openstack-tox-py37/15735be/ : SUCCESS in 3m 52s\n- openstack-tox-docs http://logs.openstack.org/56/662456/12/check/openstack-tox-docs/1a3c8eb/html/ : SUCCESS in 2m 50s","accounts_in_message":[],"_revision_number":12},{"id":"cebbff948cb693410a5473d1505bb691e8854198","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-07 15:01:54.000000000","message":"Patch Set 12:\n\nEPEL mirror is x86-64 only","accounts_in_message":[],"_revision_number":12},{"id":"3c2ac4cdaab92cf2a2bcacc1c1bbacd9abd96f56","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-07 15:05:10.000000000","message":"Patch Set 12:\n\nhttps://review.opendev.org/663973 sent to get EPEL mirror for aarch64","accounts_in_message":[],"_revision_number":12},{"id":"6b3b7a9371a21e67dc553626add7c49f449116f9","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-18 08:07:50.000000000","message":"Patch Set 12:\n\nrecheck","accounts_in_message":[],"_revision_number":12},{"id":"303454dd80bc1e8d1ff09b2a8c8cde501e7e95a2","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-18 08:56:00.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"2df61293a91f36d9805221f2f65a7527b3efb2c7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 10:10:54.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-source-arm64 kolla-build-centos-source-arm64 : NODE_FAILURE in 0s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/56/662456/13/check/openstack-tox-lower-constraints/f345203/ : SUCCESS in 3m 21s\n- openstack-tox-pep8 http://logs.openstack.org/56/662456/13/check/openstack-tox-pep8/58787b6/ : SUCCESS in 3m 38s\n- openstack-tox-py27 http://logs.openstack.org/56/662456/13/check/openstack-tox-py27/47efbee/ : SUCCESS in 3m 32s\n- openstack-tox-py36 http://logs.openstack.org/56/662456/13/check/openstack-tox-py36/b528cc3/ : SUCCESS in 3m 27s\n- openstack-tox-py37 http://logs.openstack.org/56/662456/13/check/openstack-tox-py37/4074d0f/ : SUCCESS in 3m 34s\n- openstack-tox-docs http://logs.openstack.org/56/662456/13/check/openstack-tox-docs/70d03ee/html/ : SUCCESS in 4m 02s","accounts_in_message":[],"_revision_number":13},{"id":"0fd3877ecddb8e9c0b0753816f79518ef596e0b9","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-18 11:35:14.000000000","message":"Patch Set 13:\n\nrecheck","accounts_in_message":[],"_revision_number":13},{"id":"31e84ce8ac0e0eb2429082ad604ae98df020b322","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-18 12:41:47.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"3437e7c13a1de959d57e64e439413ba87c9970a3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 15:38:17.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-source-arm64 http://logs.openstack.org/56/662456/14/check/kolla-build-centos-source-arm64/ddf4361/ : TIMED_OUT in 2h 55m 21s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/56/662456/14/check/openstack-tox-lower-constraints/69f2f8c/ : SUCCESS in 3m 47s\n- openstack-tox-pep8 http://logs.openstack.org/56/662456/14/check/openstack-tox-pep8/b443e9d/ : SUCCESS in 3m 02s\n- openstack-tox-py27 http://logs.openstack.org/56/662456/14/check/openstack-tox-py27/986acd2/ : SUCCESS in 3m 14s\n- openstack-tox-py36 http://logs.openstack.org/56/662456/14/check/openstack-tox-py36/c6bf0b1/ : SUCCESS in 3m 28s\n- openstack-tox-py37 http://logs.openstack.org/56/662456/14/check/openstack-tox-py37/96cca2b/ : SUCCESS in 3m 30s\n- openstack-tox-docs http://logs.openstack.org/56/662456/14/check/openstack-tox-docs/211d092/html/ : SUCCESS in 3m 06s","accounts_in_message":[],"_revision_number":14},{"id":"decf253f428d5fff6f755699c789443ff22b41c7","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-18 20:35:40.000000000","message":"Patch Set 14: Workflow-1\n\nneed to work on timeout ;(","accounts_in_message":[],"_revision_number":14},{"id":"f51a4a29a5a416d793f987689b01effaa6047020","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-28 06:08:47.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"97e566d73b71af3b779ec255f27b1d0772b3b80c","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-28 06:09:11.000000000","message":"Patch Set 15: Workflow-1","accounts_in_message":[],"_revision_number":15},{"id":"5b814e57415dfa067f47361959b3498aca67f4a2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-28 09:02:31.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-source-arm64 http://logs.openstack.org/56/662456/15/check/kolla-build-centos-source-arm64/39a1b02/ : TIMED_OUT in 2h 52m 12s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/56/662456/15/check/openstack-tox-lower-constraints/b190025/ : SUCCESS in 5m 10s\n- openstack-tox-pep8 http://logs.openstack.org/56/662456/15/check/openstack-tox-pep8/08028c4/ : SUCCESS in 3m 43s\n- openstack-tox-py27 http://logs.openstack.org/56/662456/15/check/openstack-tox-py27/f7c81c8/ : SUCCESS in 3m 03s\n- openstack-tox-py36 http://logs.openstack.org/56/662456/15/check/openstack-tox-py36/101d8c8/ : SUCCESS in 3m 34s\n- openstack-tox-py37 http://logs.openstack.org/56/662456/15/check/openstack-tox-py37/00cf62a/ : SUCCESS in 3m 35s\n- openstack-tox-docs http://logs.openstack.org/56/662456/15/check/openstack-tox-docs/09bf373/html/ : SUCCESS in 4m 50s\n- kolla-build-ubuntu-binary http://logs.openstack.org/56/662456/15/check/kolla-build-ubuntu-binary/59b131c/ : SUCCESS in 25m 00s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/56/662456/15/check/kolla-build-ubuntu-source/1d4aefa/ : SUCCESS in 27m 15s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/56/662456/15/check/kolla-ansible-ubuntu-source/048dcd5/ : SUCCESS in 46m 46s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/56/662456/15/check/kolla-ansible-ubuntu-binary/e87726e/ : SUCCESS in 44m 05s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/56/662456/15/check/kolla-ansible-ubuntu-source-upgrade/c4abb70/ : SUCCESS in 52m 56s","accounts_in_message":[],"_revision_number":15},{"id":"a829757b3dc43dad5c9e1a1cb588a9c2953fc407","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-28 10:07:04.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"e6aad33cb7b3ab28a03a68516b33fba7e6ca6289","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-28 10:40:59.000000000","message":"Patch Set 16:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/56/662456/16/openstack-check/tripleo-build-containers-fedora-28-master/2743cde/ : RETRY_LIMIT in 15m 59s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/56/662456/16/openstack-check/tripleo-build-containers-fedora-28-stein/28dfa5b/ : RETRY_LIMIT in 1m 02s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"662805feb8de41e20a3c53f7b554c02e8db6164b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-28 11:20:20.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-source-arm64 kolla-build-centos-source-arm64 : NODE_FAILURE in 0s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/56/662456/16/check/openstack-tox-lower-constraints/2ebda87/ : SUCCESS in 4m 02s\n- openstack-tox-pep8 http://logs.openstack.org/56/662456/16/check/openstack-tox-pep8/c8435d3/ : SUCCESS in 3m 29s\n- openstack-tox-py27 http://logs.openstack.org/56/662456/16/check/openstack-tox-py27/6bd4652/ : SUCCESS in 3m 22s\n- openstack-tox-py36 http://logs.openstack.org/56/662456/16/check/openstack-tox-py36/78cc995/ : SUCCESS in 3m 24s\n- openstack-tox-py37 http://logs.openstack.org/56/662456/16/check/openstack-tox-py37/081f84f/ : SUCCESS in 3m 40s\n- openstack-tox-docs http://logs.openstack.org/56/662456/16/check/openstack-tox-docs/379b47d/html/ : SUCCESS in 3m 10s\n- kolla-build-ubuntu-binary http://logs.openstack.org/56/662456/16/check/kolla-build-ubuntu-binary/11fc3ce/ : SUCCESS in 23m 32s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/56/662456/16/check/kolla-build-ubuntu-source/b507f63/ : SUCCESS in 26m 46s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/56/662456/16/check/kolla-ansible-ubuntu-source/2ab52da/ : SUCCESS in 43m 00s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/56/662456/16/check/kolla-ansible-ubuntu-binary/37c5a72/ : SUCCESS in 49m 11s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/56/662456/16/check/kolla-ansible-ubuntu-source-upgrade/581df77/ : SUCCESS in 53m 11s","accounts_in_message":[],"_revision_number":16},{"id":"aace5305e1184c039ea2f671fc6ff2267dde8a94","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-28 11:48:38.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"cedc6c7c55a936e5656bd5934b2ed17b49ee096d","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-28 12:27:54.000000000","message":"Patch Set 17:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/56/662456/17/openstack-check/tripleo-build-containers-fedora-28-master/0a20b83/ : FAILURE in 5m 15s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/56/662456/17/openstack-check/tripleo-build-containers-fedora-28-stein/c9c8e58/ : SUCCESS in 34m 06s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"04588241b3e01b60c0f8dcf1e9559fd3c78bd72c","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-28 12:39:37.000000000","message":"Patch Set 18: Published edit on patch set 17.","accounts_in_message":[],"_revision_number":18},{"id":"62bc1e0fcecafb9b924c420aa6bb2b818a5f5ca6","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-28 13:15:40.000000000","message":"Patch Set 18:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/56/662456/18/openstack-check/tripleo-build-containers-fedora-28-master/2f2d012/ : FAILURE in 5m 57s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/56/662456/18/openstack-check/tripleo-build-containers-fedora-28-stein/1da4983/ : SUCCESS in 35m 09s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"3dc56cf6c03f1239617c5e153235d387abf1dc57","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-28 14:53:09.000000000","message":"Patch Set 18: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":18},{"id":"8b025feae121fe7e1ccf028d0b16a8eaa3ca2db9","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-29 16:21:29.000000000","message":"Patch Set 18:\n\nrecheck","accounts_in_message":[],"_revision_number":18},{"id":"2fc7beb55c56617ce2d4a122c275f90674d82f2c","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-29 17:09:55.000000000","message":"Patch Set 18:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/56/662456/18/openstack-check/tripleo-build-containers-fedora-28-master/e8a4837/ : FAILURE in 5m 24s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/56/662456/18/openstack-check/tripleo-build-containers-fedora-28-stein/706a245/ : SUCCESS in 42m 51s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"28623e4558b7a060a6481b6788397de07f2115e9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-29 19:17:54.000000000","message":"Patch Set 18:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-source-arm64 http://logs.openstack.org/56/662456/18/check/kolla-build-centos-source-arm64/ad29ae5/ : TIMED_OUT in 2h 55m 46s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/56/662456/18/check/openstack-tox-lower-constraints/abb45cc/ : SUCCESS in 2m 58s\n- openstack-tox-pep8 http://logs.openstack.org/56/662456/18/check/openstack-tox-pep8/3f51f1d/ : SUCCESS in 3m 05s\n- openstack-tox-py27 http://logs.openstack.org/56/662456/18/check/openstack-tox-py27/8e4d65c/ : SUCCESS in 2m 42s\n- openstack-tox-py36 http://logs.openstack.org/56/662456/18/check/openstack-tox-py36/5574488/ : SUCCESS in 2m 44s\n- openstack-tox-py37 http://logs.openstack.org/56/662456/18/check/openstack-tox-py37/4fe55f3/ : SUCCESS in 5m 13s\n- openstack-tox-docs http://logs.openstack.org/56/662456/18/check/openstack-tox-docs/359ceb6/html/ : SUCCESS in 2m 23s\n- kolla-build-ubuntu-binary http://logs.openstack.org/56/662456/18/check/kolla-build-ubuntu-binary/0b88558/ : SUCCESS in 22m 02s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/56/662456/18/check/kolla-build-ubuntu-source/7b0b46b/ : SUCCESS in 24m 00s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/56/662456/18/check/kolla-ansible-ubuntu-source/a8b8e11/ : SUCCESS in 42m 57s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/56/662456/18/check/kolla-ansible-ubuntu-binary/971e3e9/ : SUCCESS in 43m 58s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/56/662456/18/check/kolla-ansible-ubuntu-source-upgrade/9287745/ : FAILURE in 39m 35s","accounts_in_message":[],"_revision_number":18},{"id":"8aa0f7138c3d6bf40dce0d749db621e1543a8c83","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-07-15 11:38:57.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"8ad5f07c74189cdb6abe9c03366250bbecfb47b7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-15 11:39:14.000000000","message":"Patch Set 19: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/kolla on branch master.  The error was:\n\n  Configuration file is not a list.  Each zuul.yaml configuration\n  file must be a list of items, for example:\n  \n  - job:\n      name: foo\n  \n  - project:\n      name: bar\n  \n  Ensure that every item starts with \"- \" so that it is parsed as a\n  YAML list.","accounts_in_message":[],"_revision_number":19},{"id":"66ae753090b43aef36cfa3d7175b803d71c026a5","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-15 11:39:15.000000000","message":"Patch Set 19:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/kolla on branch master.  The error was:\n\n  Configuration file is not a list.  Each zuul.yaml configuration\n  file must be a list of items, for example:\n  \n  - job:\n      name: foo\n  \n  - project:\n      name: bar\n  \n  Ensure that every item starts with \"- \" so that it is parsed as a\n  YAML list.","accounts_in_message":[],"_revision_number":19},{"id":"3d0457bec0af7bda01c5a6c813eaf5b3440dd3dc","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-07-15 11:41:58.000000000","message":"Uploaded patch set 20: Patch Set 19 was rebased.","accounts_in_message":[],"_revision_number":20},{"id":"4b3e8732eb7c9687460c73b708b0370ff605b770","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-15 11:42:14.000000000","message":"Patch Set 20:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/kolla on branch master.  The error was:\n\n  Configuration file is not a list.  Each zuul.yaml configuration\n  file must be a list of items, for example:\n  \n  - job:\n      name: foo\n  \n  - project:\n      name: bar\n  \n  Ensure that every item starts with \"- \" so that it is parsed as a\n  YAML list.","accounts_in_message":[],"_revision_number":20},{"id":"34f1dbe6efcaa1432a27636722d295552c7b4e6a","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-15 11:42:17.000000000","message":"Patch Set 20:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/kolla on branch master.  The error was:\n\n  Configuration file is not a list.  Each zuul.yaml configuration\n  file must be a list of items, for example:\n  \n  - job:\n      name: foo\n  \n  - project:\n      name: bar\n  \n  Ensure that every item starts with \"- \" so that it is parsed as a\n  YAML list.","accounts_in_message":[],"_revision_number":20},{"id":"8d81ede7e1f8771a9960914df83f54584a8820fb","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-07-15 11:44:38.000000000","message":"Uploaded patch set 21.","accounts_in_message":[],"_revision_number":21},{"id":"c2aae3cdaac105d0a34cd341f5fd6e1b00eb82bb","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-07-15 12:38:11.000000000","message":"Uploaded patch set 22.","accounts_in_message":[],"_revision_number":22},{"id":"e6b301374b38a065641d58b336ecd7c8e320bd37","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-15 14:05:03.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-source-arm64 kolla-build-centos-source-arm64 : NODE_FAILURE in 0s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/56/662456/22/check/openstack-tox-lower-constraints/9dcb877/ : SUCCESS in 5m 44s\n- openstack-tox-pep8 http://logs.openstack.org/56/662456/22/check/openstack-tox-pep8/e99078c/ : SUCCESS in 5m 38s\n- openstack-tox-py27 http://logs.openstack.org/56/662456/22/check/openstack-tox-py27/59e7daf/ : SUCCESS in 3m 52s\n- openstack-tox-py36 http://logs.openstack.org/56/662456/22/check/openstack-tox-py36/bffa9ec/ : SUCCESS in 3m 57s\n- openstack-tox-py37 http://logs.openstack.org/56/662456/22/check/openstack-tox-py37/5113be8/ : SUCCESS in 5m 27s\n- openstack-tox-docs http://logs.openstack.org/56/662456/22/check/openstack-tox-docs/bcac724/html/ : SUCCESS in 3m 23s\n- kolla-build-ubuntu-binary http://logs.openstack.org/56/662456/22/check/kolla-build-ubuntu-binary/8d88ab5/ : SUCCESS in 28m 19s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/56/662456/22/check/kolla-build-ubuntu-source/a567f6f/ : SUCCESS in 36m 04s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/56/662456/22/check/kolla-ansible-ubuntu-source/e69a5b8/ : SUCCESS in 52m 14s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/56/662456/22/check/kolla-ansible-ubuntu-binary/cbc536e/ : SUCCESS in 47m 03s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/56/662456/22/check/kolla-ansible-ubuntu-source-upgrade/e4d32f0/ : SUCCESS in 1h 04m 40s","accounts_in_message":[],"_revision_number":22},{"id":"27432c21127257f2b472e7c3f7488d461c52ee8c","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-07-15 14:29:10.000000000","message":"Patch Set 22:\n\nrecheck","accounts_in_message":[],"_revision_number":22},{"id":"8b0578758ce2f035aee7bb55fa0a564de1c3ab48","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-15 15:38:02.000000000","message":"Patch Set 22:\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-source-arm64 kolla-build-centos-source-arm64 : NODE_FAILURE in 0s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/56/662456/22/check/openstack-tox-lower-constraints/fa5dd35/ : SUCCESS in 3m 41s\n- openstack-tox-pep8 http://logs.openstack.org/56/662456/22/check/openstack-tox-pep8/bb6506f/ : SUCCESS in 4m 33s\n- openstack-tox-py27 http://logs.openstack.org/56/662456/22/check/openstack-tox-py27/b740df0/ : SUCCESS in 3m 13s\n- openstack-tox-py36 http://logs.openstack.org/56/662456/22/check/openstack-tox-py36/a3090e2/ : SUCCESS in 3m 13s\n- openstack-tox-py37 http://logs.openstack.org/56/662456/22/check/openstack-tox-py37/3202741/ : SUCCESS in 3m 33s\n- openstack-tox-docs http://logs.openstack.org/56/662456/22/check/openstack-tox-docs/0664b13/html/ : SUCCESS in 2m 55s\n- kolla-build-ubuntu-binary http://logs.openstack.org/56/662456/22/check/kolla-build-ubuntu-binary/3611c29/ : SUCCESS in 23m 19s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/56/662456/22/check/kolla-build-ubuntu-source/6801606/ : SUCCESS in 28m 09s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/56/662456/22/check/kolla-ansible-ubuntu-source/d7d7902/ : SUCCESS in 49m 09s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/56/662456/22/check/kolla-ansible-ubuntu-binary/d099bcb/ : SUCCESS in 50m 14s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/56/662456/22/check/kolla-ansible-ubuntu-source-upgrade/f2926b2/ : SUCCESS in 59m 18s","accounts_in_message":[],"_revision_number":22},{"id":"b9c96f985dd064138a543c007a2a0df353829973","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-07-15 16:09:05.000000000","message":"Patch Set 22:\n\nrecheck","accounts_in_message":[],"_revision_number":22},{"id":"5f148de6c61672ccddc5a4e346d1129763b242df","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-15 16:56:45.000000000","message":"Patch Set 22: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-source-arm64 kolla-build-centos-source-arm64 : NODE_FAILURE in 0s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/56/662456/22/check/openstack-tox-lower-constraints/50d6af0/ : SUCCESS in 4m 48s\n- openstack-tox-pep8 http://logs.openstack.org/56/662456/22/check/openstack-tox-pep8/9cab4b3/ : SUCCESS in 4m 57s\n- openstack-tox-py27 http://logs.openstack.org/56/662456/22/check/openstack-tox-py27/7286448/ : SUCCESS in 4m 40s\n- openstack-tox-py36 http://logs.openstack.org/56/662456/22/check/openstack-tox-py36/441b40e/ : SUCCESS in 3m 40s\n- openstack-tox-py37 http://logs.openstack.org/56/662456/22/check/openstack-tox-py37/9bb231e/ : SUCCESS in 5m 53s\n- openstack-tox-docs http://logs.openstack.org/56/662456/22/check/openstack-tox-docs/bb9c603/html/ : SUCCESS in 2m 55s\n- kolla-build-ubuntu-binary http://logs.openstack.org/56/662456/22/check/kolla-build-ubuntu-binary/481afa2/ : FAILURE in 24m 03s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/56/662456/22/check/kolla-build-ubuntu-source/23716f0/ : FAILURE in 18m 50s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/56/662456/22/check/kolla-ansible-ubuntu-source/7889263/ : FAILURE in 11m 31s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/56/662456/22/check/kolla-ansible-ubuntu-binary/f906481/ : FAILURE in 19m 24s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/56/662456/22/check/kolla-ansible-ubuntu-source-upgrade/30d0d41/ : FAILURE in 12m 16s","accounts_in_message":[],"_revision_number":22},{"id":"6d7186620044450055000c691e0482b9caf29c1a","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-07-16 07:53:57.000000000","message":"Patch Set 22:\n\nrecheck","accounts_in_message":[],"_revision_number":22},{"id":"7c7c4c43ea5a2cfb244a76b09f0c914432b1ea52","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-16 08:04:47.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-source-arm64 finger://ze09.openstack.org/cb08cc207c23436b90b7851705dceaf0 : RETRY_LIMIT in 1m 00s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/56/662456/22/check/openstack-tox-lower-constraints/3204db7/ : SUCCESS in 4m 22s\n- openstack-tox-pep8 http://logs.openstack.org/56/662456/22/check/openstack-tox-pep8/c927ad0/ : SUCCESS in 5m 30s\n- openstack-tox-py27 http://logs.openstack.org/56/662456/22/check/openstack-tox-py27/6da4517/ : SUCCESS in 4m 58s\n- openstack-tox-py36 http://logs.openstack.org/56/662456/22/check/openstack-tox-py36/9dc9ee7/ : SUCCESS in 3m 26s\n- openstack-tox-py37 http://logs.openstack.org/56/662456/22/check/openstack-tox-py37/48d99db/ : SUCCESS in 3m 46s\n- openstack-tox-docs http://logs.openstack.org/56/662456/22/check/openstack-tox-docs/ec6e9e9/html/ : SUCCESS in 3m 34s","accounts_in_message":[],"_revision_number":22},{"id":"afc2ae6d9da8ffac06122ff6daf45f1bc6d9c461","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-07-16 08:19:17.000000000","message":"Patch Set 22:\n\nrecheck","accounts_in_message":[],"_revision_number":22},{"id":"b93be546b5d44f79a0a77aeabd8c7dc90b11596a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-16 09:08:07.000000000","message":"Patch Set 22:\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-source-arm64 kolla-build-centos-source-arm64 : NODE_FAILURE in 0s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/56/662456/22/check/openstack-tox-lower-constraints/07bc0ac/ : SUCCESS in 3m 40s\n- openstack-tox-pep8 http://logs.openstack.org/56/662456/22/check/openstack-tox-pep8/5d20514/ : SUCCESS in 4m 32s\n- openstack-tox-py27 http://logs.openstack.org/56/662456/22/check/openstack-tox-py27/bddda1c/ : SUCCESS in 3m 15s\n- openstack-tox-py36 http://logs.openstack.org/56/662456/22/check/openstack-tox-py36/a95a87a/ : SUCCESS in 3m 31s\n- openstack-tox-py37 http://logs.openstack.org/56/662456/22/check/openstack-tox-py37/0d4fc81/ : SUCCESS in 3m 55s\n- openstack-tox-docs http://logs.openstack.org/56/662456/22/check/openstack-tox-docs/241fa11/html/ : SUCCESS in 3m 27s","accounts_in_message":[],"_revision_number":22},{"id":"61af865581a3cd8ca837641ab0ceef050736cbd9","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-07-17 12:00:07.000000000","message":"Patch Set 22:\n\nrecheck","accounts_in_message":[],"_revision_number":22},{"id":"ac70b92d3b0c0b019f61ee8858f1769256457c44","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-17 14:56:29.000000000","message":"Patch Set 22:\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-source-arm64 http://logs.openstack.org/56/662456/22/check/kolla-build-centos-source-arm64/c21a292/ : TIMED_OUT in 2h 55m 48s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/56/662456/22/check/openstack-tox-lower-constraints/c73c1fd/ : SUCCESS in 3m 58s\n- openstack-tox-pep8 http://logs.openstack.org/56/662456/22/check/openstack-tox-pep8/5fdcea6/ : SUCCESS in 4m 16s\n- openstack-tox-py27 http://logs.openstack.org/56/662456/22/check/openstack-tox-py27/877dd09/ : SUCCESS in 3m 09s\n- openstack-tox-py36 http://logs.openstack.org/56/662456/22/check/openstack-tox-py36/b1aae52/ : SUCCESS in 3m 48s\n- openstack-tox-py37 http://logs.openstack.org/56/662456/22/check/openstack-tox-py37/e38c36e/ : SUCCESS in 3m 21s\n- openstack-tox-docs http://logs.openstack.org/56/662456/22/check/openstack-tox-docs/034457b/html/ : SUCCESS in 2m 50s","accounts_in_message":[],"_revision_number":22},{"id":"253db836c4acb5577e09f49039275bd004da480c","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-07-23 08:43:14.000000000","message":"Patch Set 22:\n\nrecheck","accounts_in_message":[],"_revision_number":22},{"id":"6e78e5c8c17c8b29fef69dd8ae48721b6aefff2b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 11:40:06.000000000","message":"Patch Set 22:\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-source-arm64 http://logs.openstack.org/56/662456/22/check/kolla-build-centos-source-arm64/63dfbab/ : TIMED_OUT in 2h 56m 15s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/56/662456/22/check/openstack-tox-lower-constraints/eccc8e7/ : SUCCESS in 4m 58s\n- openstack-tox-pep8 http://logs.openstack.org/56/662456/22/check/openstack-tox-pep8/b621915/ : SUCCESS in 4m 41s\n- openstack-tox-py27 http://logs.openstack.org/56/662456/22/check/openstack-tox-py27/32699d5/ : SUCCESS in 4m 01s\n- openstack-tox-py36 http://logs.openstack.org/56/662456/22/check/openstack-tox-py36/f38bb42/ : SUCCESS in 4m 46s\n- openstack-tox-py37 http://logs.openstack.org/56/662456/22/check/openstack-tox-py37/54cded0/ : SUCCESS in 4m 20s\n- openstack-tox-docs http://logs.openstack.org/56/662456/22/check/openstack-tox-docs/6718a70/html/ : SUCCESS in 4m 31s","accounts_in_message":[],"_revision_number":22},{"id":"43ddb5eceed3b3acf156b9e27d6d30c8d404fad7","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-08-12 08:02:01.000000000","message":"Patch Set 22:\n\nrecheck","accounts_in_message":[],"_revision_number":22},{"id":"a69bdaf23cf74969960b8c22db7d819c1704d573","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-12 10:58:29.000000000","message":"Patch Set 22:\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-source-arm64 https://logs.opendev.org/56/662456/22/check/kolla-build-centos-source-arm64/8727f9d/ : TIMED_OUT in 2h 52m 39s (non-voting)\n- openstack-tox-lower-constraints https://logs.opendev.org/56/662456/22/check/openstack-tox-lower-constraints/ad0b1d4/ : SUCCESS in 3m 49s\n- openstack-tox-pep8 https://logs.opendev.org/56/662456/22/check/openstack-tox-pep8/6481da9/ : SUCCESS in 5m 16s\n- openstack-tox-py27 https://logs.opendev.org/56/662456/22/check/openstack-tox-py27/ab6af73/ : SUCCESS in 3m 23s\n- openstack-tox-py36 https://logs.opendev.org/56/662456/22/check/openstack-tox-py36/2231cdf/ : SUCCESS in 3m 40s\n- openstack-tox-py37 https://logs.opendev.org/56/662456/22/check/openstack-tox-py37/073a3da/ : SUCCESS in 3m 50s\n- openstack-tox-docs https://logs.opendev.org/56/662456/22/check/openstack-tox-docs/0acb537/html/ : SUCCESS in 3m 13s","accounts_in_message":[],"_revision_number":22},{"id":"d582de9cfa51370c251ab4505c9b20e1eaca4118","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-08-12 11:02:55.000000000","message":"Patch Set 22:\n\nrecheck\n\nlet\u0027s check will it get 16vcpu now","accounts_in_message":[],"_revision_number":22},{"id":"f6361bad788466e7cafd566c249e0a8aea05f1d5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-12 11:46:16.000000000","message":"Patch Set 22:\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-source-arm64 kolla-build-centos-source-arm64 : NODE_FAILURE in 0s (non-voting)\n- openstack-tox-lower-constraints https://logs.opendev.org/56/662456/22/check/openstack-tox-lower-constraints/7af23e4/ : SUCCESS in 3m 56s\n- openstack-tox-pep8 https://logs.opendev.org/56/662456/22/check/openstack-tox-pep8/ec68d8f/ : SUCCESS in 3m 58s\n- openstack-tox-py27 https://logs.opendev.org/56/662456/22/check/openstack-tox-py27/87885c5/ : SUCCESS in 3m 40s\n- openstack-tox-py36 https://logs.opendev.org/56/662456/22/check/openstack-tox-py36/d8516ff/ : SUCCESS in 3m 07s\n- openstack-tox-py37 https://logs.opendev.org/56/662456/22/check/openstack-tox-py37/4172ed3/ : SUCCESS in 4m 39s\n- openstack-tox-docs https://logs.opendev.org/56/662456/22/check/openstack-tox-docs/5cedf2a/html/ : SUCCESS in 3m 21s","accounts_in_message":[],"_revision_number":22},{"id":"def78fb9585bb35bb992d64d5bb87fe803204ff3","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-08-12 11:54:24.000000000","message":"Patch Set 22:\n\nrecheck","accounts_in_message":[],"_revision_number":22},{"id":"c156e0b13f6753837cc8d1648a99acb64b1cd2b2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-12 14:50:45.000000000","message":"Patch Set 22:\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-source-arm64 https://logs.opendev.org/56/662456/22/check/kolla-build-centos-source-arm64/28813d4/ : TIMED_OUT in 2h 51m 53s (non-voting)\n- openstack-tox-lower-constraints https://logs.opendev.org/56/662456/22/check/openstack-tox-lower-constraints/5accf5c/ : SUCCESS in 5m 11s\n- openstack-tox-pep8 https://logs.opendev.org/56/662456/22/check/openstack-tox-pep8/474e782/ : SUCCESS in 5m 18s\n- openstack-tox-py27 https://logs.opendev.org/56/662456/22/check/openstack-tox-py27/1b8cf9c/ : SUCCESS in 4m 40s\n- openstack-tox-py36 https://logs.opendev.org/56/662456/22/check/openstack-tox-py36/902a6f8/ : SUCCESS in 4m 29s\n- openstack-tox-py37 https://logs.opendev.org/56/662456/22/check/openstack-tox-py37/496eacc/ : SUCCESS in 5m 27s\n- openstack-tox-docs https://logs.opendev.org/56/662456/22/check/openstack-tox-docs/9622cb7/html/ : SUCCESS in 3m 39s","accounts_in_message":[],"_revision_number":22},{"id":"2fb89ce7e175a1db11b2c1f8b2366ddfc046944e","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-08-28 10:44:12.000000000","message":"Uploaded patch set 23.","accounts_in_message":[],"_revision_number":23},{"id":"bf89f40d6e429a2c65aa909545992e8696ecd7a6","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-08-28 10:49:49.000000000","message":"Uploaded patch set 24: Patch Set 23 was rebased.","accounts_in_message":[],"_revision_number":24},{"id":"ab7b2852c15d6760832acb7298be92775e3b046e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-28 16:06:33.000000000","message":"Patch Set 24: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-source-arm64 https://zuul.opendev.org/t/openstack/build/4884e091c5944da18ae0f7cda57c1537 : TIMED_OUT in 4h 51m 44s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/afcf2d3e40f54db8bd7cf1fe1ae56d30 : SUCCESS in 40m 23s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7a14c6a906a945dc8648653ea38ea39a : SUCCESS in 5m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/49ea3599969a4204bae6f7409e8986d3 : SUCCESS in 5m 02s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b6735a16d9964fee80ca5c90e2c76724 : SUCCESS in 3m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/92ed2af2012b45b8b469c65dbad9c4d6 : SUCCESS in 4m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e905360bdd5b47a2abd8185e57eddd64 : SUCCESS in 4m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d7357a154e147d69984cb72d923ffa7 : SUCCESS in 5m 10s","accounts_in_message":[],"_revision_number":24},{"id":"7b300b14d72c97746d539adaad6a73e05f80f577","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-09-20 15:54:19.000000000","message":"Uploaded patch set 25.","accounts_in_message":[],"_revision_number":25},{"id":"abe6084fc53eec449fc220bb7ed3ba7c41ead7e9","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-09-20 15:58:25.000000000","message":"Uploaded patch set 26.","accounts_in_message":[],"_revision_number":26},{"id":"0b039a001c9b88a4d989a3d9daf87c2870171e4c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 20:17:10.000000000","message":"Patch Set 26: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/e55c6eaa0eac4fd9b1430688b46f0990 : SUCCESS in 48m 12s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/31a6283ce4c7426882e238cdd09e28c8 : SUCCESS in 52m 25s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/4ecb428b73584becb6b989a410c93efe : SUCCESS in 1h 34m 02s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/0167228f38274ced9e1f2e69ee25b70d : SUCCESS in 1h 17m 00s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/7cce0a3500ac475eb1ba282e4d8edee9 : SUCCESS in 1h 39m 05s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/7f4aa13a280a4d4eb3c0c10a45e0ba38 : SUCCESS in 57m 40s\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/af1c783b29734fc1aafcdc9fc3139612 : SUCCESS in 57m 24s (non-voting)\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/31d24c1e6a044a73abb21d6f6af7725d : TIMED_OUT in 2h 35m 30s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/36a808bd1b2a434dac46a541edf817e5 : TIMED_OUT in 2h 34m 00s (non-voting)\n- kolla-ansible-centos-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/41a9958a5ba54d0bbfb6dd99b0df4b26 : SUCCESS in 2h 04m 22s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/0f7a283225de4cb0a90c64986272de6a : TIMED_OUT in 2h 03m 54s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/f3668eb4f28946408906b21dc522f378 : SUCCESS in 1h 44m 13s (non-voting)\n- kolla-build-centos-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/534a7d62f81f4f28b5881c57b901d2c9 : SUCCESS in 29m 56s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/964f62481ee643229ee8e09368044751 : SUCCESS in 3m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b3aed271bc1c42cca01aa8850fab8cac : FAILURE in 4m 18s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/84abe99e74a5429086f3ca910fbd0e67 : SUCCESS in 3m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/db549d4dde7e44a880ebf4cc0d273114 : SUCCESS in 3m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/96312bbb08484cd7beacd6fff5e8db70 : SUCCESS in 4m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d71e77d5a9ba486599662b31c6f24cba : SUCCESS in 6m 05s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/aa6a116a8d3f4b6ab566d056f04d08cb : SUCCESS in 23m 51s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/c257acd0929b42398c413a1d432152b4 : SUCCESS in 32m 24s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/da263a1a38204ba4a668e583fb63ee8b : SUCCESS in 55m 20s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/95adce51df0e4ca6bf3b277b21d2633b : SUCCESS in 1h 06m 16s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/993432def0384c41b909cf7ca2e99801 : SUCCESS in 1h 14m 11s\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/71a73e31a3d64e05b478a99cbe9f9efe : SUCCESS in 1h 22m 18s (non-voting)\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/5b2ee40ac943419b9fc09fec3c81e062 : TIMED_OUT in 2h 32m 52s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/cc7a816797b540c380bec1eb6d8d87df : SUCCESS in 2h 08m 12s (non-voting)\n- kolla-ansible-ubuntu-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/96c37d056f8a4d289ef197a3073c0753 : SUCCESS in 1h 25m 24s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/a8c68c7ea1e34181a6a9413b8c9d80c0 : SUCCESS in 1h 17m 48s (non-voting)","accounts_in_message":[],"_revision_number":26},{"id":"fd11d78017cd1b9a2d925ff1446b0c3e42e1ee0a","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-11-20 09:25:03.000000000","message":"Uploaded patch set 27.","accounts_in_message":[],"_revision_number":27},{"id":"d5e5c5fd3fceabebd7aecec5775434d5aa2fe1bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-20 11:27:23.000000000","message":"Patch Set 27: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/b14d4dbe19cc44bcaa4fa4b03622a1d5 : SUCCESS in 40m 39s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/aa5ab36de8854b8f910cb48b5c973dbc : SUCCESS in 57m 59s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/53ed4dd7fec6422894e13fc2526232c0 : SUCCESS in 1h 26m 03s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/59c9486e4e6f4d79be098245136bbd16 : SUCCESS in 1h 36m 55s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/0262ce8bf6164b9890af9dc0066b8861 : SUCCESS in 1h 40m 22s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/f2a340cd5b204d0fbd19681910568403 : SUCCESS in 47m 08s\n- kolla-build-centos-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/ce91c713100343e481fe8d07e7f562bb : SUCCESS in 34m 10s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/76850b35d523481e9f65e03ce2063a9a : SUCCESS in 18m 40s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/5bc2352e583b4873b68acdc197c1de42 : SUCCESS in 52m 59s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/585a6045ee81487f8fc3cd6211052d05 : SUCCESS in 3m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/059a636a715d4bcc957f74a5684eb382 : FAILURE in 3m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a162fca7f7df40a5826d49f9ed3b5a07 : SUCCESS in 2m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/628067c76eea4ddfbfe72100276f2222 : SUCCESS in 5m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02095f42496d45b1982d7ffefc0143a7 : SUCCESS in 4m 51s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/540ec6b07d3e4e06bb885e007b5248b1 : FAILURE in 24m 03s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/897b3977a427431b887921bdb250f734 : FAILURE in 21m 45s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/993e87aa711d4a26a4fdc6e81b7c8edf : FAILURE in 14m 02s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/93cc1ffc9ffd4ef2af7ab55256196d0e : FAILURE in 28m 38s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/740fa37fe75a499d9875cbcf3242a07f : POST_FAILURE in 9m 11s","accounts_in_message":[],"_revision_number":27},{"id":"eec400bdc02f361c126042e79da21fc75f9cc000","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-11-26 12:03:45.000000000","message":"Patch Set 27:\n\nrecheck","accounts_in_message":[],"_revision_number":27},{"id":"dafc9e2e1f190437c69b0ce6f357d0a6d53eb15c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-26 13:59:32.000000000","message":"Patch Set 27:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/02aa2851a92e406682f6f2e86d7c694c : SUCCESS in 45m 02s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/c5def1d397844e7e81da9e93ee6d6230 : SUCCESS in 1h 00m 23s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/c2e7046c940645e09bb05a5ee5a4fbeb : SUCCESS in 1h 49m 57s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/cfabeb5b32724a23a9dd7e4a29877ce6 : SUCCESS in 1h 26m 11s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/b5a1f3d008b248639786bcf145d45536 : SUCCESS in 1h 44m 48s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/09d9670082544aa5926de9c72917f046 : SUCCESS in 56m 23s\n- kolla-build-centos-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/2762998907404f1096f6a86fdebc237f : SUCCESS in 23m 19s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/974c510d0a49475b8de47ab71493be55 : SUCCESS in 24m 07s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/7febc5a6917e4c1c9a9126b24cdb108f : SUCCESS in 56m 31s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4c8f10eca6804d8196d75140d324f7d7 : SUCCESS in 3m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c136a2188754a5fb078396e3a9c8299 : FAILURE in 4m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6661873206284cae8e88377e859f7970 : SUCCESS in 3m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d1c154cc7866492c9dfca2d41f2b106c : SUCCESS in 3m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/acf4fbea0a3e4150b5e86907f553a64a : SUCCESS in 5m 01s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9332d3dbcce04222ab33fae4af265ade : SUCCESS in 23m 58s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/ad0b0c7f46314b32a81f02b0ddd74a8d : SUCCESS in 27m 13s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/230479c9cbab45bd860aa9e04513498d : SUCCESS in 51m 36s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9d1981d0296040dfbb4503aa86b728f5 : SUCCESS in 55m 26s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/8c6b7c2cb53d44c6bd6cd6a7cdbbd27c : SUCCESS in 1h 06m 59s","accounts_in_message":[],"_revision_number":27},{"id":"0b4c471dafd68926513dc36a42dfe0461667b289","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-12-06 08:56:53.000000000","message":"Patch Set 27:\n\nrecheck","accounts_in_message":[],"_revision_number":27},{"id":"c57034212ee85237f634298dfc0393bea49c072e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-06 10:53:14.000000000","message":"Patch Set 27:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/860edf95dc29425694253606c66ff02a : SUCCESS in 32m 02s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/78aa3b211e1e435b98179b43be5ea041 : SUCCESS in 32m 50s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/e8db5e0328e148508a5022c60550b961 : FAILURE in 1h 13m 32s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/443b419b58b24119a1acfade83708e0f : SUCCESS in 1h 34m 50s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/cf6bf28fe5e54361a875d389fde8de94 : FAILURE in 1h 53m 26s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/e9a8c85d65034c738a9a90ec041595e1 : SUCCESS in 50m 59s\n- kolla-build-centos-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/606bd13261a440568d4df50d11aaafdb : SUCCESS in 24m 25s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/0f9f2ecca5504232bfc86f8cc2379993 : FAILURE in 42m 42s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/d8075d3b645448c8a618b8606fa17416 : SUCCESS in 1h 13m 52s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/802c956d1efd4ad49334ad50adc109a6 : SUCCESS in 4m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37e766b9badf4090a641f960beaffa45 : FAILURE in 4m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/788c4324d59942d99ee70318936fadce : SUCCESS in 3m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/00cd75f65bc44e78a87fd088f44f83d5 : SUCCESS in 3m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d6bd89853c248c2819b1df2170fcdf3 : SUCCESS in 4m 28s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9d4765cc5b4b4c2a920ea52dd02f0aa2 : FAILURE in 48m 31s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/bac8d2f5ca7c45f5885f0184c9b45cc6 : SUCCESS in 31m 08s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/72ac8e57be2f412993e22902a0dddc73 : SUCCESS in 56m 15s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/03ba2c194ca3484082b3774aca4a0324 : SUCCESS in 1h 15m 56s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/b787dee422b2470384c3a9d20a9bb97d : SUCCESS in 1h 27m 37s","accounts_in_message":[],"_revision_number":27},{"id":"91d1688e6a57e7e57015231ca497da98e12daf89","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-12-10 08:09:48.000000000","message":"Patch Set 27:\n\nrecheck","accounts_in_message":[],"_revision_number":27},{"id":"a49dd99477b07284267528158d3cb7a41a12ba12","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-10 09:46:49.000000000","message":"Patch Set 27:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/40ab5ccaf3114df986d7a76bc888d41a : SUCCESS in 38m 02s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/f662866e7149441f81a657afd512ea03 : SUCCESS in 45m 40s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/50ededd4d2f043b98d97b04ba7d29c59 : SUCCESS in 1h 18m 19s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/ea250f9916c64298a94d6879b227db6b : SUCCESS in 1h 13m 35s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/8e2baa339ad140b3bf36431a6ad4ed52 : SUCCESS in 1h 35m 27s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/ba1f999e8ff949a2a973851904ef0d0a : SUCCESS in 53m 19s\n- kolla-build-centos-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/f442d784283047b3a459d55368beb963 : SUCCESS in 22m 29s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/65d42f876dc34f499d6f98e8d45c018f : SUCCESS in 20m 17s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/b21108f652364466853f33b0bdb56cbd : SUCCESS in 53m 32s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/467f13c5f4af4c5e9965543af447a1b4 : SUCCESS in 3m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e6f19efb8bb46e28ecf71881a8375da : FAILURE in 4m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5f0074e92b18414098c5cb563deec53e : SUCCESS in 3m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/77df9603e1cc4518bd4e85d25fab813c : SUCCESS in 3m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d97a7fad5134f359ea7aa53c43897d3 : SUCCESS in 4m 32s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/bd73ccc1eb8c40ddba6f586144741558 : SUCCESS in 23m 50s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/a4192668f8a34789941105e1cbed48a3 : SUCCESS in 26m 31s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/0b13eb362f3242d2b7950d9f8502733a : SUCCESS in 54m 54s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e7b609fdb736445383e4688d919eee81 : SUCCESS in 53m 28s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/0282ad3f931f4a0cb5806ca064ca3566 : SUCCESS in 1h 04m 56s","accounts_in_message":[],"_revision_number":27},{"id":"feb5e28cd252dc021eb7d21138bad194630c3dde","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2019-12-11 05:37:36.000000000","message":"Patch Set 27:\n\n(1 comment)","accounts_in_message":[],"_revision_number":27},{"id":"47c6a9be4ffa737e7a77ce2f494e08f555ae4409","author":{"_account_id":22076,"name":"Kevin Zhao","email":"kevin.zhao@linaro.org","username":"KevinZhao"},"date":"2019-12-11 06:34:21.000000000","message":"Uploaded patch set 28.","accounts_in_message":[],"_revision_number":28},{"id":"b8597a7fc9926aebdc9a5434874c1d2f198a6874","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-11 06:34:41.000000000","message":"Patch Set 28: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/kolla on branch master.  The error was:\n\n  The nodeset \"kolla-debian-arm64\" was not found.\n\nThe error appears in the following job stanza:\n\n  job:\n      name: kolla-build-centos-source-aarch64\n      parent: kolla-base\n      nodeset: kolla-debian-arm64\n      voting: false\n      vars:\n        action: build\n        base_distro: centos\n        install_type: source\n\n  in \"openstack/kolla/.zuul.d/centos.yaml@master\", line 94, column 3","accounts_in_message":[],"_revision_number":28},{"id":"cbba65b6e3c77d68b9aec511b3564e20fc3ca917","author":{"_account_id":22076,"name":"Kevin Zhao","email":"kevin.zhao@linaro.org","username":"KevinZhao"},"date":"2019-12-11 06:41:22.000000000","message":"Uploaded patch set 29.","accounts_in_message":[],"_revision_number":29},{"id":"b8bf62766fa432aed8b672c48c15cd3deee4a0b9","author":{"_account_id":22076,"name":"Kevin Zhao","email":"kevin.zhao@linaro.org","username":"KevinZhao"},"date":"2019-12-11 07:01:17.000000000","message":"Patch Set 29:\n\nrecheck","accounts_in_message":[],"_revision_number":29},{"id":"5b8d2b549179207b459e0eac5b31f1a865258e81","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-12-11 07:37:04.000000000","message":"Uploaded patch set 30.","accounts_in_message":[],"_revision_number":30},{"id":"eb4d367a0a41294b254f830744bc36dab8e13d11","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-11 09:07:16.000000000","message":"Patch Set 30: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/1b7b5cb64e8141e68ba3865a7575e9cc : SUCCESS in 37m 47s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/51c96c43bb7747b99cfd34c49e35fb3d : SUCCESS in 38m 57s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/309e0a6da4df4de1be9e7f22536f8a30 : SUCCESS in 1h 20m 41s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/5828388d5c5e4ca19b311de4a86d0830 : SUCCESS in 1h 02m 22s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/39a506eab8504c2580be9cee86ff423c : SUCCESS in 1h 23m 49s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/47a5950d08144dff8b88a2b038d4843d : SUCCESS in 47m 13s\n- kolla-build-centos-source-aarch64 https://zuul.opendev.org/t/openstack/build/10a24c9136e049a3b255390f376f9d14 : FAILURE in 1h 29m 39s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/f06bbea0c27a4430b655ed6afb5ef8e8 : SUCCESS in 26m 39s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/6ce8851b9d734687995f37f212372612 : SUCCESS in 18m 16s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/494384e662854ee0ad79d4e5b48e7bde : SUCCESS in 53m 09s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/daf108a34afa4d508b0c58426a1490f8 : SUCCESS in 3m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c8bb446f3af745bbad5e99163eb904ca : FAILURE in 4m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e4cae375769743c488163e5104df6317 : SUCCESS in 3m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a94b920999e04da984eb731efd2374f7 : SUCCESS in 3m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c2364cf438334e71b4ceca002b8cf78f : SUCCESS in 5m 01s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4f436e7a55484a80a9aedef657acb5ea : SUCCESS in 22m 50s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/ae07e1e7bc6646f1a1d4f5a1942a585f : SUCCESS in 31m 36s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/b82fa3aeb1454589b7427acb2cd7556c : SUCCESS in 1h 03m 55s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b7d10c11890649a8ab05b633ff7e3a9d : SUCCESS in 54m 56s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/177727dd7fa24c4c89ab71ab64ca2ed6 : SUCCESS in 58m 55s","accounts_in_message":[],"_revision_number":30},{"id":"aa5e1ca943f7d5226ef0e9b9ce2776c224feb7b3","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-12-11 09:35:09.000000000","message":"Uploaded patch set 31.","accounts_in_message":[],"_revision_number":31},{"id":"84446cec88302bc39c61dfb32098ae33f1de3ea0","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-12-11 09:36:51.000000000","message":"Uploaded patch set 32: Patch Set 31 was rebased.","accounts_in_message":[],"_revision_number":32},{"id":"760c11a366b59f79c540dcbf1b89447f15f03651","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-11 11:21:35.000000000","message":"Patch Set 32: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/64395eb717554de0b297c2ece9d4be5a : SUCCESS in 38m 22s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/64424815194b4f1cb4ab93526155b795 : SUCCESS in 46m 39s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/8e6a2b69a755411ca2965a4d6d31da3d : SUCCESS in 1h 39m 04s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/2832130d5e304548a693065d4bdbe0ee : SUCCESS in 1h 17m 32s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/0d7ebd784fca43b1bef34c24a3b962ef : SUCCESS in 1h 38m 48s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/0f2ec9c017cf4a1bb24fd9426136a5d8 : SUCCESS in 49m 21s\n- kolla-build-centos-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/55676242fdf94535a650b619243c31f5 : SUCCESS in 42m 15s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/49a32a194345471b80bfb8091b8c6283 : SUCCESS in 38m 59s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/dc3f74b3bed143ae8fa5c130877237fa : SUCCESS in 58m 21s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/31a44ff052ae48429c86c722032da8a2 : SUCCESS in 4m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f3bf9c9d6864fecb3913abe336ebe1b : SUCCESS in 6m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b44264962720452485d402f6513c7877 : SUCCESS in 5m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/445fd9031ef448f3bbf69650a19a3467 : SUCCESS in 4m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64004131aaca4a22944287ed8422f2b4 : SUCCESS in 6m 08s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1c97d3ea48214dff9c34765f0122b979 : SUCCESS in 31m 41s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/ce240abf6b064d349f72d08c3967c0a3 : SUCCESS in 40m 00s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/70c3b60acc264ed79e6ea710a8250fb9 : SUCCESS in 54m 03s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c7bc756a09fa41eaac7f8fb995fdd89a : SUCCESS in 1h 01m 32s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/1b5b6513c8cb41f895b01eebc8dce1e6 : SUCCESS in 1h 14m 32s","accounts_in_message":[],"_revision_number":32},{"id":"68b088590f343be95cc63e341abb577518c4e50c","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-12-11 19:47:22.000000000","message":"Patch Set 32: Workflow-1\n\nholding merging before we discuss how to not overflow 8 arm64 nodes with jobs","accounts_in_message":[],"_revision_number":32},{"id":"73c96ebff3c31a87e538fe870c8873165b7a5340","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-12-11 20:38:14.000000000","message":"Patch Set 32:\n\nDue to CentOS 7 Python issues I plan to abandon this patch. Debian one will be only one.","accounts_in_message":[],"_revision_number":32},{"id":"f05c73be92cf7c8028fa36f5cd14b4a055302271","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-12-13 12:13:33.000000000","message":"Abandoned\n\nCentOS 7 is py2 \u003d\u003d issues","accounts_in_message":[],"_revision_number":32}],"current_revision_number":32,"current_revision":"524622b4492c526b2da8c5be3ae558ee649de7a4","revisions":{"5c3f4a5816c855511f8ccd85d3f062b1ada7d8e8":{"kind":"REWORK","_number":1,"created":"2019-05-31 13:00:07.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/56/662456/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/56/662456/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/662456/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/662456/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/662456/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/56/662456/1"}}},"commit":{"parents":[{"commit":"3e9cf08ac59c6b38b1542ac12e19ff44af92eb12","subject":"Merge \"Install fluentd gem on debian/ubuntu for non x86_64\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3e9cf08ac59c6b38b1542ac12e19ff44af92eb12"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-05-31 12:28:55.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-05-31 12:59:54.000000000","tz":120},"subject":"Add CentOS/source aarch64 experimental job","message":"Add CentOS/source aarch64 experimental job\n\nAdd kolla CentOS 7 source aarch64 building job.\n\nImplements: blueprint aarch64-3rd-party-ci-kolla\n\nChange-Id: Ic32c19109d8521b93b7e124ac4ea9224767aad5c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5c3f4a5816c855511f8ccd85d3f062b1ada7d8e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5c3f4a5816c855511f8ccd85d3f062b1ada7d8e8"}]},"branch":"refs/heads/master"},"69ae00c95e1b150689cb84f351f7748fadf46f0b":{"kind":"REWORK","_number":2,"created":"2019-05-31 13:04:04.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/56/662456/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/56/662456/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/662456/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/662456/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/662456/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/56/662456/2"}}},"commit":{"parents":[{"commit":"3e9cf08ac59c6b38b1542ac12e19ff44af92eb12","subject":"Merge \"Install fluentd gem on debian/ubuntu for non x86_64\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3e9cf08ac59c6b38b1542ac12e19ff44af92eb12"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-05-31 12:28:55.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-05-31 13:03:48.000000000","tz":120},"subject":"Add CentOS/source aarch64 experimental job","message":"Add CentOS/source aarch64 experimental job\n\nAdd kolla CentOS 7 source aarch64 building job.\n\nImplements: blueprint aarch64-3rd-party-ci-kolla\n\nChange-Id: Ic32c19109d8521b93b7e124ac4ea9224767aad5c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/69ae00c95e1b150689cb84f351f7748fadf46f0b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/69ae00c95e1b150689cb84f351f7748fadf46f0b"}]},"branch":"refs/heads/master"},"2b2cf4b20100d9cba63b03b608193c5b5d40d1dd":{"kind":"REWORK","_number":3,"created":"2019-05-31 13:18:51.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/56/662456/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/56/662456/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/662456/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/662456/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/662456/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/56/662456/3"}}},"commit":{"parents":[{"commit":"3e9cf08ac59c6b38b1542ac12e19ff44af92eb12","subject":"Merge \"Install fluentd gem on debian/ubuntu for non x86_64\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3e9cf08ac59c6b38b1542ac12e19ff44af92eb12"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-05-31 12:28:55.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-05-31 13:18:41.000000000","tz":120},"subject":"Add CentOS/source aarch64 experimental job","message":"Add CentOS/source aarch64 experimental job\n\nAdd kolla CentOS 7 source aarch64 building job.\n\nImplements: blueprint aarch64-3rd-party-ci-kolla\n\nChange-Id: Ic32c19109d8521b93b7e124ac4ea9224767aad5c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2b2cf4b20100d9cba63b03b608193c5b5d40d1dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2b2cf4b20100d9cba63b03b608193c5b5d40d1dd"}]},"branch":"refs/heads/master"},"35b80520a86625171a5503333e23b7f26634a029":{"kind":"REWORK","_number":4,"created":"2019-05-31 14:17:41.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/56/662456/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/56/662456/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/662456/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/662456/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/662456/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/56/662456/4"}}},"commit":{"parents":[{"commit":"3e9cf08ac59c6b38b1542ac12e19ff44af92eb12","subject":"Merge \"Install fluentd gem on debian/ubuntu for non x86_64\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3e9cf08ac59c6b38b1542ac12e19ff44af92eb12"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-05-31 12:28:55.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-05-31 14:17:27.000000000","tz":120},"subject":"Add CentOS/source aarch64 experimental job","message":"Add CentOS/source aarch64 experimental job\n\nAdd kolla CentOS 7 source aarch64 building job.\n\nImplements: blueprint aarch64-3rd-party-ci-kolla\n\nNOTE: removed all normal jobs for now.\n\nChange-Id: Ic32c19109d8521b93b7e124ac4ea9224767aad5c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/35b80520a86625171a5503333e23b7f26634a029"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/35b80520a86625171a5503333e23b7f26634a029"}]},"branch":"refs/heads/master"},"285569c5b406da80d82e31fae7c7459444595c4d":{"kind":"REWORK","_number":5,"created":"2019-05-31 18:26:55.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/56/662456/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/56/662456/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/662456/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/662456/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/662456/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/56/662456/5"}}},"commit":{"parents":[{"commit":"3e9cf08ac59c6b38b1542ac12e19ff44af92eb12","subject":"Merge \"Install fluentd gem on debian/ubuntu for non x86_64\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3e9cf08ac59c6b38b1542ac12e19ff44af92eb12"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-05-31 12:28:55.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-05-31 18:26:42.000000000","tz":120},"subject":"Add CentOS/source aarch64 experimental job","message":"Add CentOS/source aarch64 experimental job\n\nAdd kolla CentOS 7 source aarch64 building job.\n\nImplements: blueprint aarch64-3rd-party-ci-kolla\n\nNOTE: removed all normal jobs for now.\n\nChange-Id: Ic32c19109d8521b93b7e124ac4ea9224767aad5c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/285569c5b406da80d82e31fae7c7459444595c4d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/285569c5b406da80d82e31fae7c7459444595c4d"}]},"branch":"refs/heads/master"},"8d7afdfbc36c856378635223c371b13b14d2c7cb":{"kind":"REWORK","_number":6,"created":"2019-06-01 20:30:33.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/56/662456/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/56/662456/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/662456/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/662456/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/662456/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/56/662456/6"}}},"commit":{"parents":[{"commit":"3e9cf08ac59c6b38b1542ac12e19ff44af92eb12","subject":"Merge \"Install fluentd gem on debian/ubuntu for non x86_64\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3e9cf08ac59c6b38b1542ac12e19ff44af92eb12"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-05-31 12:28:55.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-06-01 20:30:23.000000000","tz":120},"subject":"Add CentOS/source aarch64 experimental job","message":"Add CentOS/source aarch64 experimental job\n\nAdd kolla CentOS 7 source aarch64 building job.\n\nImplements: blueprint aarch64-3rd-party-ci-kolla\n\nNOTE: removed all normal jobs for now.\n\nChange-Id: Ic32c19109d8521b93b7e124ac4ea9224767aad5c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8d7afdfbc36c856378635223c371b13b14d2c7cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8d7afdfbc36c856378635223c371b13b14d2c7cb"}]},"branch":"refs/heads/master"},"f1a0d044c789f30f37e5867f906ae2fa12d9c720":{"kind":"REWORK","_number":7,"created":"2019-06-02 08:07:24.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/56/662456/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/56/662456/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/662456/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/662456/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/662456/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/56/662456/7"}}},"commit":{"parents":[{"commit":"c585ca5055c7a92fdbe5826030d3529e5c397540","subject":"Use td-agent-gem to install fluentd monasca plugin on ubuntu","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c585ca5055c7a92fdbe5826030d3529e5c397540"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-05-31 12:28:55.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-06-02 08:07:14.000000000","tz":120},"subject":"Add CentOS/source aarch64 experimental job","message":"Add CentOS/source aarch64 experimental job\n\nAdd kolla CentOS 7 source aarch64 building job.\n\nImplements: blueprint aarch64-3rd-party-ci-kolla\n\nNOTE: removed all normal jobs for now.\n\nChange-Id: Ic32c19109d8521b93b7e124ac4ea9224767aad5c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f1a0d044c789f30f37e5867f906ae2fa12d9c720"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f1a0d044c789f30f37e5867f906ae2fa12d9c720"}]},"branch":"refs/heads/master"},"6d3fc6c91df266e438e85bb051dadf0793e35a55":{"kind":"REWORK","_number":8,"created":"2019-06-02 13:47:20.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/56/662456/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/56/662456/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/662456/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/662456/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/662456/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/56/662456/8"}}},"commit":{"parents":[{"commit":"c585ca5055c7a92fdbe5826030d3529e5c397540","subject":"Use td-agent-gem to install fluentd monasca plugin on ubuntu","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c585ca5055c7a92fdbe5826030d3529e5c397540"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-05-31 12:28:55.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-06-02 13:46:58.000000000","tz":120},"subject":"Add CentOS/source aarch64 experimental job","message":"Add CentOS/source aarch64 experimental job\n\nAdd kolla CentOS 7 source aarch64 building job.\n\nImplements: blueprint aarch64-3rd-party-ci-kolla\n\nNOTE: removed all normal jobs for now.\n\nChange-Id: Ic32c19109d8521b93b7e124ac4ea9224767aad5c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6d3fc6c91df266e438e85bb051dadf0793e35a55"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6d3fc6c91df266e438e85bb051dadf0793e35a55"}]},"branch":"refs/heads/master"},"99b88ad0f9b14b6ed0f77c308f71000282b1b8b0":{"kind":"REWORK","_number":9,"created":"2019-06-02 14:55:30.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/56/662456/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/56/662456/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/662456/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/662456/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/662456/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/56/662456/9"}}},"commit":{"parents":[{"commit":"c585ca5055c7a92fdbe5826030d3529e5c397540","subject":"Use td-agent-gem to install fluentd monasca plugin on ubuntu","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c585ca5055c7a92fdbe5826030d3529e5c397540"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-05-31 12:28:55.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-06-02 14:55:21.000000000","tz":120},"subject":"Add CentOS/source aarch64 experimental job","message":"Add CentOS/source aarch64 experimental job\n\nAdd kolla CentOS 7 source aarch64 building job.\n\nImplements: blueprint aarch64-3rd-party-ci-kolla\n\nNOTE: removed all normal jobs for now.\n\nChange-Id: Ic32c19109d8521b93b7e124ac4ea9224767aad5c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/99b88ad0f9b14b6ed0f77c308f71000282b1b8b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/99b88ad0f9b14b6ed0f77c308f71000282b1b8b0"}]},"branch":"refs/heads/master"},"1aabbb6f2f62f871e5e6316cbbed7a874b0733b8":{"kind":"REWORK","_number":10,"created":"2019-06-02 16:17:15.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/56/662456/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/56/662456/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/662456/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/662456/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/662456/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/56/662456/10"}}},"commit":{"parents":[{"commit":"c585ca5055c7a92fdbe5826030d3529e5c397540","subject":"Use td-agent-gem to install fluentd monasca plugin on ubuntu","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c585ca5055c7a92fdbe5826030d3529e5c397540"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-05-31 12:28:55.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-06-02 16:16:48.000000000","tz":120},"subject":"Add CentOS/source aarch64 experimental job","message":"Add CentOS/source aarch64 experimental job\n\nAdd kolla CentOS 7 source aarch64 building job.\n\nImplements: blueprint aarch64-3rd-party-ci-kolla\n\nNOTE: removed all normal jobs for now.\nNOTE: disabled OpenStack infra mirrors for CentOS due to lack of altarch\n\nChange-Id: Ic32c19109d8521b93b7e124ac4ea9224767aad5c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1aabbb6f2f62f871e5e6316cbbed7a874b0733b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1aabbb6f2f62f871e5e6316cbbed7a874b0733b8"}]},"branch":"refs/heads/master"},"3ed310c6a4ae050d853844e55ebf5b6d1f8b34e8":{"kind":"REWORK","_number":11,"created":"2019-06-03 06:19:13.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/56/662456/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/56/662456/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/662456/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/662456/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/662456/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/56/662456/11"}}},"commit":{"parents":[{"commit":"c585ca5055c7a92fdbe5826030d3529e5c397540","subject":"Use td-agent-gem to install fluentd monasca plugin on ubuntu","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c585ca5055c7a92fdbe5826030d3529e5c397540"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-05-31 12:28:55.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-06-03 06:19:03.000000000","tz":120},"subject":"Add CentOS/source aarch64 experimental job","message":"Add CentOS/source aarch64 experimental job\n\nAdd kolla CentOS 7 source aarch64 building job.\n\nImplements: blueprint aarch64-3rd-party-ci-kolla\n\nNOTE: removed all normal jobs for now.\nNOTE: disabled OpenStack infra mirrors for CentOS due to lack of altarch\n\nChange-Id: Ic32c19109d8521b93b7e124ac4ea9224767aad5c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3ed310c6a4ae050d853844e55ebf5b6d1f8b34e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3ed310c6a4ae050d853844e55ebf5b6d1f8b34e8"}]},"branch":"refs/heads/master"},"dab7f5240cadcae6cdab1d9c804252f92d61536f":{"kind":"REWORK","_number":12,"created":"2019-06-07 14:07:53.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/56/662456/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/56/662456/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/662456/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/662456/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/662456/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/56/662456/12"}}},"commit":{"parents":[{"commit":"9cfd3a88aa1c07054b66495dbb069a0000a20738","subject":"Merge \"Ignore EM releases in version-check.py\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9cfd3a88aa1c07054b66495dbb069a0000a20738"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-05-31 12:28:55.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-06-07 14:07:33.000000000","tz":120},"subject":"Add CentOS/source aarch64 experimental job","message":"Add CentOS/source aarch64 experimental job\n\nAdd kolla CentOS 7 source aarch64 building job.\n\nImplements: blueprint aarch64-3rd-party-ci-kolla\n\nNOTE: removed all normal jobs for now.\n\nChange-Id: Ic32c19109d8521b93b7e124ac4ea9224767aad5c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dab7f5240cadcae6cdab1d9c804252f92d61536f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dab7f5240cadcae6cdab1d9c804252f92d61536f"}]},"branch":"refs/heads/master"},"efc30579d3251a43c9bd620e1c62beae0dbfaed8":{"kind":"REWORK","_number":13,"created":"2019-06-18 08:56:00.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/56/662456/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/56/662456/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/662456/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/662456/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/662456/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/56/662456/13"}}},"commit":{"parents":[{"commit":"b07241ff5537e6f3807943c1bb9bc26a9badc7aa","subject":"Merge \"grafana/aarch64: use package from upstream repositories\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b07241ff5537e6f3807943c1bb9bc26a9badc7aa"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-05-31 12:28:55.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-06-18 08:55:26.000000000","tz":120},"subject":"Add CentOS/source aarch64 experimental job","message":"Add CentOS/source aarch64 experimental job\n\nAdd kolla CentOS 7 source aarch64 building job.\n\nImplements: blueprint aarch64-3rd-party-ci-kolla\n\nNOTE: removed all normal jobs for now.\n\nChange-Id: Ic32c19109d8521b93b7e124ac4ea9224767aad5c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/efc30579d3251a43c9bd620e1c62beae0dbfaed8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/efc30579d3251a43c9bd620e1c62beae0dbfaed8"}]},"branch":"refs/heads/master"},"5de81fcdbd0896f9bb78145f0719c90bff53d9a2":{"kind":"REWORK","_number":14,"created":"2019-06-18 12:41:47.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/56/662456/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/56/662456/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/662456/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/662456/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/662456/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/56/662456/14"}}},"commit":{"parents":[{"commit":"b07241ff5537e6f3807943c1bb9bc26a9badc7aa","subject":"Merge \"grafana/aarch64: use package from upstream repositories\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b07241ff5537e6f3807943c1bb9bc26a9badc7aa"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-05-31 12:28:55.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-06-18 12:41:32.000000000","tz":120},"subject":"Add CentOS/source aarch64 experimental job","message":"Add CentOS/source aarch64 experimental job\n\nAdd kolla CentOS 7 source aarch64 building job.\n\nImplements: blueprint aarch64-3rd-party-ci-kolla\n\nNOTE: removed all normal jobs for now.\n\nChange-Id: Ic32c19109d8521b93b7e124ac4ea9224767aad5c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5de81fcdbd0896f9bb78145f0719c90bff53d9a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5de81fcdbd0896f9bb78145f0719c90bff53d9a2"}]},"branch":"refs/heads/master"},"efe893d376faebc5b7138ebe84a4deb1589f9428":{"kind":"REWORK","_number":15,"created":"2019-06-28 06:08:47.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/56/662456/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/56/662456/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/662456/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/662456/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/662456/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/56/662456/15"}}},"commit":{"parents":[{"commit":"b27e99c2b3f83b714759f023ffc3e80842ea930e","subject":"Merge \"Move back to CentOS StorageSIG repos for Ceph and Ganesha\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b27e99c2b3f83b714759f023ffc3e80842ea930e"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-05-31 12:28:55.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin@juszkiewicz.com.pl","date":"2019-06-28 06:07:13.000000000","tz":120},"subject":"Add CentOS/source aarch64 experimental job","message":"Add CentOS/source aarch64 experimental job\n\nAdd kolla CentOS 7 source aarch64 building job.\n\nImplements: blueprint aarch64-3rd-party-ci-kolla\n\nNOTE: removed all normal jobs for now.\n\nDepends-on: Ie259a38aad84bd8f5392f19791edfbd1f8ec4b1a\nDepends-on: Id13545224fa0fbb729d46386a10505e0989fc3c6\n\nChange-Id: Ic32c19109d8521b93b7e124ac4ea9224767aad5c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/efe893d376faebc5b7138ebe84a4deb1589f9428"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/efe893d376faebc5b7138ebe84a4deb1589f9428"}]},"branch":"refs/heads/master"},"1efc7b6de2f620d53c7f745276de52cc03d2a9c2":{"kind":"REWORK","_number":16,"created":"2019-06-28 10:07:04.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/56/662456/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/56/662456/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/662456/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/662456/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/662456/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/56/662456/16"}}},"commit":{"parents":[{"commit":"93ed9474b504fc16a7ce313c01cb30ec345e9fcb","subject":"Merge \"doc: Start using openstackdoctheme\u0027s extlink extension\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/93ed9474b504fc16a7ce313c01cb30ec345e9fcb"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-05-31 12:28:55.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin@juszkiewicz.com.pl","date":"2019-06-28 10:06:08.000000000","tz":120},"subject":"Add CentOS/source aarch64 experimental job","message":"Add CentOS/source aarch64 experimental job\n\nAdd kolla CentOS 7 source aarch64 building job.\n\nImplements: blueprint aarch64-3rd-party-ci-kolla\n\nNOTE: removed all normal jobs for now.\n\nNOTE: \u0027tooz\u0027 removed from openstack-base image. It fetches \u0027grpcio\u0027\nwhich takes a while to build. Let\u0027s see how it change build times.\n\nDepends-on: Ie259a38aad84bd8f5392f19791edfbd1f8ec4b1a\nDepends-on: Id13545224fa0fbb729d46386a10505e0989fc3c6\n\nChange-Id: Ic32c19109d8521b93b7e124ac4ea9224767aad5c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1efc7b6de2f620d53c7f745276de52cc03d2a9c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1efc7b6de2f620d53c7f745276de52cc03d2a9c2"}]},"branch":"refs/heads/master"},"b2454dc24367d26b5a45dca3479e23cbc3cc8c7d":{"kind":"REWORK","_number":17,"created":"2019-06-28 11:48:38.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/56/662456/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/56/662456/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/662456/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/662456/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/662456/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/56/662456/17"}}},"commit":{"parents":[{"commit":"93ed9474b504fc16a7ce313c01cb30ec345e9fcb","subject":"Merge \"doc: Start using openstackdoctheme\u0027s extlink extension\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/93ed9474b504fc16a7ce313c01cb30ec345e9fcb"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-05-31 12:28:55.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin@juszkiewicz.com.pl","date":"2019-06-28 11:48:09.000000000","tz":120},"subject":"Add CentOS/source aarch64 experimental job","message":"Add CentOS/source aarch64 experimental job\n\nAdd kolla CentOS 7 source aarch64 building job.\n\nImplements: blueprint aarch64-3rd-party-ci-kolla\n\nNOTE: removed all normal jobs for now.\n\nNOTE: \u0027tooz\u0027 removed from openstack-base image. It fetches \u0027grpcio\u0027\nwhich takes a while to build. Let\u0027s see how it change build times.\n\nNOTE: trying debian-buster-arm64 node\n\nDepends-on: Ie259a38aad84bd8f5392f19791edfbd1f8ec4b1a\nDepends-on: Id13545224fa0fbb729d46386a10505e0989fc3c6\n\nChange-Id: Ic32c19109d8521b93b7e124ac4ea9224767aad5c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b2454dc24367d26b5a45dca3479e23cbc3cc8c7d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b2454dc24367d26b5a45dca3479e23cbc3cc8c7d"}]},"branch":"refs/heads/master"},"d788ab210110a0ee4541cbcdb1c991b59721cd11":{"kind":"REWORK","_number":18,"created":"2019-06-28 12:39:37.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/56/662456/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/56/662456/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/662456/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/662456/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/662456/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/56/662456/18"}}},"commit":{"parents":[{"commit":"93ed9474b504fc16a7ce313c01cb30ec345e9fcb","subject":"Merge \"doc: Start using openstackdoctheme\u0027s extlink extension\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/93ed9474b504fc16a7ce313c01cb30ec345e9fcb"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-05-31 12:28:55.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-06-28 12:39:21.000000000","tz":0},"subject":"Add CentOS/source aarch64 experimental job","message":"Add CentOS/source aarch64 experimental job\n\nAdd kolla CentOS 7 source aarch64 building job.\n\nImplements: blueprint aarch64-3rd-party-ci-kolla\n\nNOTE: removed all normal jobs for now.\n\nNOTE: \u0027tooz\u0027 removed from openstack-base image. It fetches \u0027grpcio\u0027\nwhich takes a while to build. Let\u0027s see how it change build times.\n\nDepends-on: Ie259a38aad84bd8f5392f19791edfbd1f8ec4b1a\nDepends-on: Id13545224fa0fbb729d46386a10505e0989fc3c6\n\nChange-Id: Ic32c19109d8521b93b7e124ac4ea9224767aad5c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d788ab210110a0ee4541cbcdb1c991b59721cd11"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d788ab210110a0ee4541cbcdb1c991b59721cd11"}]},"branch":"refs/heads/master"},"b3792246d56607db08c80603d2fb387f0397f4d0":{"kind":"REWORK","_number":19,"created":"2019-07-15 11:38:57.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/56/662456/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/56/662456/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/662456/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/662456/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/662456/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/56/662456/19"}}},"commit":{"parents":[{"commit":"26ed9b8f4ff785cb5f90530de16d2c9dd7edd574","subject":"cleanup setup of Debian/Ubuntu CI nodes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/26ed9b8f4ff785cb5f90530de16d2c9dd7edd574"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-07-15 11:37:37.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-07-15 11:37:37.000000000","tz":120},"subject":"Add CentOS/source aarch64 experimental job","message":"Add CentOS/source aarch64 experimental job\n\nAdd kolla CentOS 7 source aarch64 building job.\n\nImplements: blueprint aarch64-3rd-party-ci-kolla\n\nNOTE: removed all normal jobs for now.\n\nChange-Id: Ic32c19109d8521b93b7e124ac4ea9224767aad5c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b3792246d56607db08c80603d2fb387f0397f4d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b3792246d56607db08c80603d2fb387f0397f4d0"}]},"branch":"refs/heads/master"},"97d2569749f93567b612606315f4f642d91aa4c9":{"kind":"TRIVIAL_REBASE","_number":20,"created":"2019-07-15 11:41:58.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/56/662456/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/56/662456/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/662456/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/662456/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/662456/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/56/662456/20"}}},"commit":{"parents":[{"commit":"0b796fbf9834a43d708096eff435cde2ce8aefaa","subject":"cleanup setup of Debian/Ubuntu CI nodes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0b796fbf9834a43d708096eff435cde2ce8aefaa"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-07-15 11:37:37.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-07-15 11:41:35.000000000","tz":120},"subject":"Add CentOS/source aarch64 experimental job","message":"Add CentOS/source aarch64 experimental job\n\nAdd kolla CentOS 7 source aarch64 building job.\n\nImplements: blueprint aarch64-3rd-party-ci-kolla\n\nNOTE: removed all normal jobs for now.\n\nChange-Id: Ic32c19109d8521b93b7e124ac4ea9224767aad5c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/97d2569749f93567b612606315f4f642d91aa4c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/97d2569749f93567b612606315f4f642d91aa4c9"}]},"branch":"refs/heads/master"},"8ff86cfb5f521bba7ffce0e7699d425d0f11732d":{"kind":"REWORK","_number":21,"created":"2019-07-15 11:44:38.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/56/662456/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/56/662456/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/662456/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/662456/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/662456/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/56/662456/21"}}},"commit":{"parents":[{"commit":"0b796fbf9834a43d708096eff435cde2ce8aefaa","subject":"cleanup setup of Debian/Ubuntu CI nodes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0b796fbf9834a43d708096eff435cde2ce8aefaa"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-07-15 11:37:37.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-07-15 11:44:11.000000000","tz":120},"subject":"Add CentOS/source aarch64 experimental job","message":"Add CentOS/source aarch64 experimental job\n\nAdd kolla CentOS 7 source aarch64 building job.\n\nImplements: blueprint aarch64-3rd-party-ci-kolla\n\nNOTE: removed all normal jobs for now.\n\nChange-Id: Ic32c19109d8521b93b7e124ac4ea9224767aad5c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8ff86cfb5f521bba7ffce0e7699d425d0f11732d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8ff86cfb5f521bba7ffce0e7699d425d0f11732d"}]},"branch":"refs/heads/master"},"50155fa933ba0a22b71814c52b2ee8b6e7603bee":{"kind":"REWORK","_number":22,"created":"2019-07-15 12:38:11.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/56/662456/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/56/662456/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/662456/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/662456/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/662456/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/56/662456/22"}}},"commit":{"parents":[{"commit":"0b796fbf9834a43d708096eff435cde2ce8aefaa","subject":"cleanup setup of Debian/Ubuntu CI nodes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0b796fbf9834a43d708096eff435cde2ce8aefaa"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-07-15 11:37:37.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-07-15 12:37:08.000000000","tz":120},"subject":"Add CentOS/source aarch64 experimental job","message":"Add CentOS/source aarch64 experimental job\n\nAdd kolla CentOS 7 source aarch64 building job.\n\nImplements: blueprint aarch64-3rd-party-ci-kolla\n\nNOTE: removed all normal jobs for now.\n\nChange-Id: Ic32c19109d8521b93b7e124ac4ea9224767aad5c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/50155fa933ba0a22b71814c52b2ee8b6e7603bee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/50155fa933ba0a22b71814c52b2ee8b6e7603bee"}]},"branch":"refs/heads/master"},"cf15894078094e51d7e5d4d4fcedc36d9ee74749":{"kind":"REWORK","_number":23,"created":"2019-08-28 10:44:12.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/56/662456/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/56/662456/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/662456/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/662456/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/662456/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/56/662456/23"}}},"commit":{"parents":[{"commit":"857cf3effcda2dcd178fd74c70d8384f0c22ca40","subject":"Merge \"Add fluentd_binary and fluentd_version labels\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/857cf3effcda2dcd178fd74c70d8384f0c22ca40"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-07-15 11:37:37.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-08-28 10:43:37.000000000","tz":120},"subject":"Add CentOS/source aarch64 experimental job","message":"Add CentOS/source aarch64 experimental job\n\nAdd kolla CentOS 7 source aarch64 building job.\n\nImplements: blueprint aarch64-3rd-party-ci-kolla\n\nNOTE: removed all normal jobs for now.\n\nChange-Id: Ic32c19109d8521b93b7e124ac4ea9224767aad5c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cf15894078094e51d7e5d4d4fcedc36d9ee74749"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cf15894078094e51d7e5d4d4fcedc36d9ee74749"}]},"branch":"refs/heads/master"},"6b47c82d1939dc76f5aabd094209639dcc573a5b":{"kind":"NO_CHANGE","_number":24,"created":"2019-08-28 10:49:49.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/56/662456/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/56/662456/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/662456/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/662456/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/662456/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/56/662456/24"}}},"commit":{"parents":[{"commit":"857cf3effcda2dcd178fd74c70d8384f0c22ca40","subject":"Merge \"Add fluentd_binary and fluentd_version labels\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/857cf3effcda2dcd178fd74c70d8384f0c22ca40"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-07-15 11:37:37.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-08-28 10:48:30.000000000","tz":120},"subject":"Add CentOS/source aarch64 experimental job","message":"Add CentOS/source aarch64 experimental job\n\nAdd kolla CentOS 7 source aarch64 building job.\n\nImplements: blueprint aarch64-3rd-party-ci-kolla\n\nNOTE: removed all normal jobs for now.\n\nChange-Id: Ic32c19109d8521b93b7e124ac4ea9224767aad5c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6b47c82d1939dc76f5aabd094209639dcc573a5b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6b47c82d1939dc76f5aabd094209639dcc573a5b"}]},"branch":"refs/heads/master"},"fcc77a172b3c5dacd4577f20025027c3be84481f":{"kind":"REWORK","_number":25,"created":"2019-09-20 15:54:19.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/56/662456/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/56/662456/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/662456/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/662456/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/662456/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/56/662456/25"}}},"commit":{"parents":[{"commit":"29e4cb118ae14f300717c0a80c82ddf37cffa385","subject":"Merge \"Fix retries when pushing images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/29e4cb118ae14f300717c0a80c82ddf37cffa385"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-09-20 15:44:06.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-09-20 15:51:30.000000000","tz":120},"subject":"Add CentOS/source aarch64 experimental job","message":"Add CentOS/source aarch64 experimental job\n\nAdd kolla CentOS 7 source aarch64 building job.\n\nImplements: blueprint aarch64-3rd-party-ci-kolla\n\nChange-Id: Ic32c19109d8521b93b7e124ac4ea9224767aad5c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fcc77a172b3c5dacd4577f20025027c3be84481f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fcc77a172b3c5dacd4577f20025027c3be84481f"}]},"branch":"refs/heads/master"},"40d448cb61df735948f72a81e723dcc71b7f8291":{"kind":"REWORK","_number":26,"created":"2019-09-20 15:58:25.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/56/662456/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/56/662456/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/662456/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/662456/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/662456/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/56/662456/26"}}},"commit":{"parents":[{"commit":"29e4cb118ae14f300717c0a80c82ddf37cffa385","subject":"Merge \"Fix retries when pushing images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/29e4cb118ae14f300717c0a80c82ddf37cffa385"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-09-20 15:44:06.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-09-20 15:58:01.000000000","tz":120},"subject":"Add CentOS/source aarch64 experimental job","message":"Add CentOS/source aarch64 experimental job\n\nAdd kolla CentOS 7 source aarch64 building job.\n\nImplements: blueprint aarch64-3rd-party-ci-kolla\n\nChange-Id: Ic32c19109d8521b93b7e124ac4ea9224767aad5c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/40d448cb61df735948f72a81e723dcc71b7f8291"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/40d448cb61df735948f72a81e723dcc71b7f8291"}]},"branch":"refs/heads/master"},"635f4f74c2005ef1b864be41a2d882799ce87719":{"kind":"REWORK","_number":27,"created":"2019-11-20 09:25:03.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/56/662456/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/56/662456/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/662456/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/662456/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/662456/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/56/662456/27"}}},"commit":{"parents":[{"commit":"25d336c105daacd37a2778444b1092311a1152ea","subject":"Merge \"Drop openstack-tox-py27 job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/25d336c105daacd37a2778444b1092311a1152ea"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-09-20 15:44:06.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-11-20 09:24:37.000000000","tz":60},"subject":"Add CentOS/source aarch64 experimental job","message":"Add CentOS/source aarch64 experimental job\n\nAdd kolla CentOS 7 source aarch64 building job.\n\nImplements: blueprint aarch64-3rd-party-ci-kolla\n\nChange-Id: Ic32c19109d8521b93b7e124ac4ea9224767aad5c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/635f4f74c2005ef1b864be41a2d882799ce87719"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/635f4f74c2005ef1b864be41a2d882799ce87719"}]},"branch":"refs/heads/master"},"2407215900cd307663ee904fe9d9224fdd46557f":{"kind":"REWORK","_number":28,"created":"2019-12-11 06:34:21.000000000","uploader":{"_account_id":22076,"name":"Kevin Zhao","email":"kevin.zhao@linaro.org","username":"KevinZhao"},"ref":"refs/changes/56/662456/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/56/662456/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/662456/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/662456/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/662456/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/56/662456/28"}}},"commit":{"parents":[{"commit":"25d336c105daacd37a2778444b1092311a1152ea","subject":"Merge \"Drop openstack-tox-py27 job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/25d336c105daacd37a2778444b1092311a1152ea"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-09-20 15:44:06.000000000","tz":120},"committer":{"name":"Kevin Zhao","email":"kevin.zhao@linaro.org","date":"2019-12-11 06:22:32.000000000","tz":480},"subject":"Add CentOS/source aarch64 experimental job","message":"Add CentOS/source aarch64 experimental job\n\nAdd kolla CentOS 7 source aarch64 building job.\n\nImplements: blueprint aarch64-3rd-party-ci-kolla\n\nChange-Id: Ic32c19109d8521b93b7e124ac4ea9224767aad5c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2407215900cd307663ee904fe9d9224fdd46557f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2407215900cd307663ee904fe9d9224fdd46557f"}]},"branch":"refs/heads/master"},"1f19a00fba2c74c3d0861544548ed3d1fcabb40d":{"kind":"REWORK","_number":29,"created":"2019-12-11 06:41:22.000000000","uploader":{"_account_id":22076,"name":"Kevin Zhao","email":"kevin.zhao@linaro.org","username":"KevinZhao"},"ref":"refs/changes/56/662456/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/56/662456/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/662456/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/662456/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/662456/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/56/662456/29"}}},"commit":{"parents":[{"commit":"25d336c105daacd37a2778444b1092311a1152ea","subject":"Merge \"Drop openstack-tox-py27 job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/25d336c105daacd37a2778444b1092311a1152ea"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-09-20 15:44:06.000000000","tz":120},"committer":{"name":"Kevin Zhao","email":"kevin.zhao@linaro.org","date":"2019-12-11 06:38:59.000000000","tz":480},"subject":"Add CentOS/source aarch64 experimental job","message":"Add CentOS/source aarch64 experimental job\n\nAdd kolla CentOS 7 source aarch64 building job.\n\nImplements: blueprint aarch64-3rd-party-ci-kolla\n\nChange-Id: Ic32c19109d8521b93b7e124ac4ea9224767aad5c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1f19a00fba2c74c3d0861544548ed3d1fcabb40d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1f19a00fba2c74c3d0861544548ed3d1fcabb40d"}]},"branch":"refs/heads/master"},"c8116f44e86894d791fa41fd58098300bd5bc9bb":{"kind":"REWORK","_number":30,"created":"2019-12-11 07:37:04.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/56/662456/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/56/662456/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/662456/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/662456/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/662456/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/56/662456/30"}}},"commit":{"parents":[{"commit":"888833a7f06fde36ad90d6222e1c732ad48761cb","subject":"Merge \"Revert \"Introduce macro to remove Py3 only plugins on CentOS 7\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/888833a7f06fde36ad90d6222e1c732ad48761cb"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-09-20 15:44:06.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-12-11 07:36:04.000000000","tz":60},"subject":"Add CentOS/source aarch64 experimental job","message":"Add CentOS/source aarch64 experimental job\n\nAdd kolla CentOS 7 source aarch64 building job.\n\nImplements: blueprint aarch64-3rd-party-ci-kolla\n\nChange-Id: Ic32c19109d8521b93b7e124ac4ea9224767aad5c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c8116f44e86894d791fa41fd58098300bd5bc9bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c8116f44e86894d791fa41fd58098300bd5bc9bb"}]},"branch":"refs/heads/master"},"cf0d0738aebd58abb306924e55f3b416ae02cc9f":{"kind":"REWORK","_number":31,"created":"2019-12-11 09:35:09.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/56/662456/31","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/56/662456/31","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/662456/31 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/662456/31 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/662456/31 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/56/662456/31"}}},"commit":{"parents":[{"commit":"888833a7f06fde36ad90d6222e1c732ad48761cb","subject":"Merge \"Revert \"Introduce macro to remove Py3 only plugins on CentOS 7\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/888833a7f06fde36ad90d6222e1c732ad48761cb"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-09-20 15:44:06.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-12-11 09:34:36.000000000","tz":60},"subject":"Add CentOS/source aarch64 experimental job","message":"Add CentOS/source aarch64 experimental job\n\nAdd kolla CentOS 7 source aarch64 building job.\n\nImplements: blueprint aarch64-3rd-party-ci-kolla\n\nDepends-On: https://review.opendev.org/698430\n\nChange-Id: Ic32c19109d8521b93b7e124ac4ea9224767aad5c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cf0d0738aebd58abb306924e55f3b416ae02cc9f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cf0d0738aebd58abb306924e55f3b416ae02cc9f"}]},"branch":"refs/heads/master"},"524622b4492c526b2da8c5be3ae558ee649de7a4":{"kind":"NO_CHANGE","_number":32,"created":"2019-12-11 09:36:51.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/56/662456/32","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/56/662456/32","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/662456/32 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/662456/32 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/662456/32 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/56/662456/32"}}},"commit":{"parents":[{"commit":"888833a7f06fde36ad90d6222e1c732ad48761cb","subject":"Merge \"Revert \"Introduce macro to remove Py3 only plugins on CentOS 7\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/888833a7f06fde36ad90d6222e1c732ad48761cb"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-09-20 15:44:06.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-12-11 09:36:21.000000000","tz":60},"subject":"Add CentOS/source aarch64 experimental job","message":"Add CentOS/source aarch64 experimental job\n\nAdd kolla CentOS 7 source aarch64 building job.\n\nImplements: blueprint aarch64-3rd-party-ci-kolla\n\nDepends-On: https://review.opendev.org/698430\n\nChange-Id: Ic32c19109d8521b93b7e124ac4ea9224767aad5c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/524622b4492c526b2da8c5be3ae558ee649de7a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/524622b4492c526b2da8c5be3ae558ee649de7a4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
