)]}'
{"id":"openstack%2Fkolla~662788","triplet_id":"openstack%2Fkolla~master~Ic1e5030c890c4473f792904556941bda97dafcd0","project":"openstack/kolla","branch":"master","topic":"bug/1831469","hashtags":[],"change_id":"Ic1e5030c890c4473f792904556941bda97dafcd0","subject":"Install pymysql for python3 in Ubuntu binary","status":"ABANDONED","created":"2019-06-03 15:27:53.000000000","updated":"2019-06-07 17:56:54.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d70f954ae8cbace3ccca77fe98673524697562d7","_number":662788,"virtual_id_number":662788,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"date":"2019-06-07 17:56:52.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":1,"date":"2019-06-05 14:45:15.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2019-06-05 13:29:52.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2019-06-07 07:22:00.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2019-06-05 13:42:26.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"date":"2019-06-07 16:03:52.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-05 13:29:52.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-06-05 13:42:26.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2019-06-05 14:45:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-06-07 07:22:00.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2019-06-07 16:03:52.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"}],"messages":[{"id":"c4f999243503146ea93e1799e60863782a621ed0","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-03 15:27:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cfed1e735beeb33961dd378c6f97e724d74eeded","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-03 16:15:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/88/662788/1/openstack-check/tripleo-build-containers-fedora-28-master/22904e6/ : SUCCESS in 39m 16s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/88/662788/1/openstack-check/tripleo-build-containers-fedora-28-stein/16c7d3d/ : SUCCESS in 42m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9f8215c77a6959853a4504181a6f254bbaad442a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-03 16:57:03.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/88/662788/1/check/kolla-build-centos-binary/b0651a7/ : FAILURE in 46m 06s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/88/662788/1/check/kolla-build-centos-source/8f33ddd/ : FAILURE in 58m 25s\n- kolla-ansible-centos-source http://logs.openstack.org/88/662788/1/check/kolla-ansible-centos-source/c992a56/ : SUCCESS in 1h 22m 43s\n- kolla-ansible-centos-binary http://logs.openstack.org/88/662788/1/check/kolla-ansible-centos-binary/cad7739/ : SUCCESS in 1h 02m 12s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/88/662788/1/check/kolla-ansible-centos-source-upgrade/a6fcb30/ : SUCCESS in 1h 15m 30s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/88/662788/1/check/tripleo-build-containers-centos-7/a81e585/ : SUCCESS in 40m 28s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/88/662788/1/check/kolla-build-oraclelinux-binary/2a87c7f/ : FAILURE in 53m 58s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/88/662788/1/check/kolla-build-oraclelinux-source/fdbea5e/ : FAILURE in 1h 08m 09s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/88/662788/1/check/kolla-ansible-oraclelinux-source/69d3ddd/ : SUCCESS in 1h 14m 55s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/88/662788/1/check/kolla-ansible-oraclelinux-binary/5cb48ec/ : SUCCESS in 1h 01m 36s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/88/662788/1/check/openstack-tox-lower-constraints/73b225e/ : SUCCESS in 4m 11s\n- openstack-tox-pep8 http://logs.openstack.org/88/662788/1/check/openstack-tox-pep8/2424d7a/ : SUCCESS in 3m 04s\n- openstack-tox-py27 http://logs.openstack.org/88/662788/1/check/openstack-tox-py27/ac7ff6e/ : SUCCESS in 3m 06s\n- openstack-tox-py36 http://logs.openstack.org/88/662788/1/check/openstack-tox-py36/92fade3/ : SUCCESS in 3m 15s\n- openstack-tox-py37 http://logs.openstack.org/88/662788/1/check/openstack-tox-py37/e1ef1f8/ : SUCCESS in 3m 34s\n- openstack-tox-docs http://logs.openstack.org/88/662788/1/check/openstack-tox-docs/63ea942/html/ : SUCCESS in 3m 11s\n- kolla-build-ubuntu-binary http://logs.openstack.org/88/662788/1/check/kolla-build-ubuntu-binary/8d81cff/ : RETRY_LIMIT in 1m 26s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/88/662788/1/check/kolla-build-ubuntu-source/d989987/ : RETRY_LIMIT in 1m 29s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/88/662788/1/check/kolla-ansible-ubuntu-source/53de560/ : RETRY_LIMIT in 1m 24s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/88/662788/1/check/kolla-ansible-ubuntu-binary/a31ff22/ : RETRY_LIMIT in 1m 17s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/88/662788/1/check/kolla-ansible-ubuntu-source-upgrade/720a96a/ : RETRY_LIMIT in 1m 31s","accounts_in_message":[],"_revision_number":1},{"id":"157e288d25985859f7d434143aa867e06b1ed10a","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-05 10:59:53.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"0e0521511d015c866655f85377f2ae94337101f0","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-05 11:39:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/88/662788/1/openstack-check/tripleo-build-containers-fedora-28-master/b5606ab/ : SUCCESS in 38m 40s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/88/662788/1/openstack-check/tripleo-build-containers-fedora-28-stein/f359194/ : SUCCESS in 36m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"19d37c14e49e97e137fd8503a8048073661c1286","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-05 12:38:50.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/88/662788/1/check/kolla-build-centos-binary/cdc5f97/ : SUCCESS in 48m 07s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/88/662788/1/check/kolla-build-centos-source/43035ad/ : SUCCESS in 53m 48s\n- kolla-ansible-centos-source http://logs.openstack.org/88/662788/1/check/kolla-ansible-centos-source/62eb9a5/ : SUCCESS in 1h 07m 17s\n- kolla-ansible-centos-binary http://logs.openstack.org/88/662788/1/check/kolla-ansible-centos-binary/12413ca/ : SUCCESS in 58m 26s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/88/662788/1/check/kolla-ansible-centos-source-upgrade/9179072/ : SUCCESS in 1h 13m 49s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/88/662788/1/check/tripleo-build-containers-centos-7/1ac3660/ : SUCCESS in 38m 48s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/88/662788/1/check/kolla-build-oraclelinux-binary/1265408/ : SUCCESS in 51m 26s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/88/662788/1/check/kolla-build-oraclelinux-source/63ae31d/ : FAILURE in 12m 27s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/88/662788/1/check/kolla-ansible-oraclelinux-source/2a2154c/ : SUCCESS in 1h 09m 43s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/88/662788/1/check/kolla-ansible-oraclelinux-binary/4ec2be0/ : SUCCESS in 1h 15m 07s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/88/662788/1/check/openstack-tox-lower-constraints/cec6459/ : SUCCESS in 3m 25s\n- openstack-tox-pep8 http://logs.openstack.org/88/662788/1/check/openstack-tox-pep8/c166a69/ : SUCCESS in 3m 42s\n- openstack-tox-py27 http://logs.openstack.org/88/662788/1/check/openstack-tox-py27/ca43b87/ : SUCCESS in 3m 22s\n- openstack-tox-py36 http://logs.openstack.org/88/662788/1/check/openstack-tox-py36/bb6bef0/ : SUCCESS in 3m 23s\n- openstack-tox-py37 http://logs.openstack.org/88/662788/1/check/openstack-tox-py37/6afe83a/ : SUCCESS in 2m 57s\n- openstack-tox-docs http://logs.openstack.org/88/662788/1/check/openstack-tox-docs/2ee6a92/html/ : SUCCESS in 2m 52s\n- kolla-build-ubuntu-binary http://logs.openstack.org/88/662788/1/check/kolla-build-ubuntu-binary/d37ce91/ : SUCCESS in 36m 04s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/88/662788/1/check/kolla-build-ubuntu-source/1cd169a/ : SUCCESS in 34m 25s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/88/662788/1/check/kolla-ansible-ubuntu-source/5a79dca/ : SUCCESS in 50m 04s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/88/662788/1/check/kolla-ansible-ubuntu-binary/e227338/ : SUCCESS in 46m 52s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/88/662788/1/check/kolla-ansible-ubuntu-source-upgrade/28413a8/ : SUCCESS in 1h 06m 06s","accounts_in_message":[],"_revision_number":1},{"id":"d7b87f9522ed070a5756d1bc7bf8c0bb3087021f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-05 12:48:26.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b6f3024e1d58dd21f533a6f43a18ccaebc0b3893","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-05 13:29:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/88/662788/1/openstack-check/tripleo-build-containers-fedora-28-master/9c29a9b/ : SUCCESS in 35m 02s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/88/662788/1/openstack-check/tripleo-build-containers-fedora-28-stein/72f5868/ : SUCCESS in 37m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d0c6896b32d59626b8b597293caaddfa38cb2aa9","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2019-06-05 13:42:26.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"9ba4cf7b382c9d3281bd95f37c8d00b4c6d3ff67","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-05 14:45:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/88/662788/1/check/kolla-build-centos-binary/aa07412/ : SUCCESS in 43m 52s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/88/662788/1/check/kolla-build-centos-source/4a90a54/ : SUCCESS in 54m 57s\n- kolla-ansible-centos-source http://logs.openstack.org/88/662788/1/check/kolla-ansible-centos-source/b1ad56f/ : SUCCESS in 1h 23m 42s\n- kolla-ansible-centos-binary http://logs.openstack.org/88/662788/1/check/kolla-ansible-centos-binary/11343da/ : SUCCESS in 1h 06m 49s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/88/662788/1/check/kolla-ansible-centos-source-upgrade/93f1047/ : SUCCESS in 1h 15m 50s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/88/662788/1/check/tripleo-build-containers-centos-7/3794fd0/ : SUCCESS in 47m 03s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/88/662788/1/check/kolla-build-oraclelinux-binary/8d7eca3/ : SUCCESS in 59m 26s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/88/662788/1/check/kolla-build-oraclelinux-source/36867bb/ : SUCCESS in 59m 09s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/88/662788/1/check/kolla-ansible-oraclelinux-source/224bb0d/ : SUCCESS in 1h 21m 21s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/88/662788/1/check/kolla-ansible-oraclelinux-binary/75c994d/ : SUCCESS in 1h 06m 07s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/88/662788/1/check/openstack-tox-lower-constraints/0478a6d/ : SUCCESS in 4m 07s\n- openstack-tox-pep8 http://logs.openstack.org/88/662788/1/check/openstack-tox-pep8/31742f9/ : SUCCESS in 3m 08s\n- openstack-tox-py27 http://logs.openstack.org/88/662788/1/check/openstack-tox-py27/d250f44/ : SUCCESS in 3m 02s\n- openstack-tox-py36 http://logs.openstack.org/88/662788/1/check/openstack-tox-py36/09e5c14/ : SUCCESS in 3m 15s\n- openstack-tox-py37 http://logs.openstack.org/88/662788/1/check/openstack-tox-py37/3b32967/ : SUCCESS in 4m 09s\n- openstack-tox-docs http://logs.openstack.org/88/662788/1/check/openstack-tox-docs/4ea639f/html/ : SUCCESS in 2m 56s\n- kolla-build-ubuntu-binary http://logs.openstack.org/88/662788/1/check/kolla-build-ubuntu-binary/66c8a8d/ : SUCCESS in 1h 05m 32s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/88/662788/1/check/kolla-build-ubuntu-source/44140b4/ : SUCCESS in 32m 48s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/88/662788/1/check/kolla-ansible-ubuntu-source/ec77f00/ : SUCCESS in 53m 10s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/88/662788/1/check/kolla-ansible-ubuntu-binary/f45509c/ : SUCCESS in 1h 21m 55s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/88/662788/1/check/kolla-ansible-ubuntu-source-upgrade/d170a59/ : SUCCESS in 59m 46s","accounts_in_message":[],"_revision_number":1},{"id":"4967b64429cd7c84ea9a5e3b930851153a4e237b","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-06-07 07:22:00.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"9ee4214bc7b61d67271d25664f1118b471bc4437","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-07 16:03:52.000000000","message":"Patch Set 1:\n\nif it is needed for ironic inspector then maybe install it there?","accounts_in_message":[],"_revision_number":1},{"id":"2c09d6e236721164ce74aec7a7c4cdeb3a32c62b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-07 17:00:49.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b67d825c7e8b95be22d67edb92908be809f76580","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-07 17:56:52.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"aaace52a6cc2e0d5bef4b4519aadb04ea4fe910b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-07 17:56:54.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"508d7da004ba5698321a14555707a18b796b34af","revisions":{"508d7da004ba5698321a14555707a18b796b34af":{"kind":"REWORK","_number":1,"created":"2019-06-03 15:27:53.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/88/662788/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/88/662788/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/662788/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/662788/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/662788/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/88/662788/1"}}},"commit":{"parents":[{"commit":"be775aa6cf725f6f2a3e44055ca9bc347a94b0db","subject":"Merge \"Update contributor guide\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/be775aa6cf725f6f2a3e44055ca9bc347a94b0db"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-06-03 15:27:00.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-06-03 15:27:00.000000000","tz":60},"subject":"Install pymysql for python3 in Ubuntu binary","message":"Install pymysql for python3 in Ubuntu binary\n\nThis is needed for ironic inspector.\n\nChange-Id: Ic1e5030c890c4473f792904556941bda97dafcd0\nCloses-Bug: #1831469\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/508d7da004ba5698321a14555707a18b796b34af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/508d7da004ba5698321a14555707a18b796b34af"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
