)]}'
{"id":"openstack%2Fkolla~662846","triplet_id":"openstack%2Fkolla~master~I353ecd3ed04c78fd21f0108c5e1355280d3f3bb9","project":"openstack/kolla","branch":"master","topic":"retire_tripleo_ui","hashtags":[],"change_id":"I353ecd3ed04c78fd21f0108c5e1355280d3f3bb9","subject":"Remove tripleo-ui","status":"MERGED","created":"2019-06-03 19:58:55.000000000","updated":"2019-06-06 23:53:01.000000000","submitted":"2019-06-06 23:53:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"662846-1559865181674-26b96e4b","meta_rev_id":"60d3abbf85eaec332ac2b4f3756f08d9d6b3fe87","_number":662846,"virtual_id_number":662846,"owner":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10112,"name":"Jason E. Rist","email":"jrist@redhat.com","username":"jrist"},{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"date":"2019-06-06 20:44:23.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-06-06 23:53:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-06-06 21:27:28.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":30413,"name":"chaiguanglong","email":"chaigl@awcloud.com","username":"chaiguanglong"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"recommended":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2019-06-06 12:59:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10112,"name":"Jason E. Rist","email":"jrist@redhat.com","username":"jrist"},{"value":1,"date":"2019-06-06 13:00:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},{"value":2,"date":"2019-06-06 13:03:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-06-05 16:25:42.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":30413,"name":"chaiguanglong","email":"chaigl@awcloud.com","username":"chaiguanglong"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":10112,"name":"Jason E. Rist","email":"jrist@redhat.com","username":"jrist"},{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},{"value":1,"date":"2019-06-06 13:03:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":30413,"name":"chaiguanglong","email":"chaigl@awcloud.com","username":"chaiguanglong"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10112,"name":"Jason E. Rist","email":"jrist@redhat.com","username":"jrist"},{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":30413,"name":"chaiguanglong","email":"chaigl@awcloud.com","username":"chaiguanglong"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":10112,"name":"Jason E. Rist","email":"jrist@redhat.com","username":"jrist"},{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":30413,"name":"chaiguanglong","email":"chaigl@awcloud.com","username":"chaiguanglong"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":10112,"name":"Jason E. Rist","email":"jrist@redhat.com","username":"jrist"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":30413,"name":"chaiguanglong","email":"chaigl@awcloud.com","username":"chaiguanglong"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-04 09:17:26.000000000","updated_by":{"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},"reviewer":{"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},"state":"REVIEWER"},{"updated":"2019-06-05 10:55:13.000000000","updated_by":{"_account_id":30413,"name":"chaiguanglong","email":"chaigl@awcloud.com","username":"chaiguanglong"},"reviewer":{"_account_id":30413,"name":"chaiguanglong","email":"chaigl@awcloud.com","username":"chaiguanglong"},"state":"REVIEWER"},{"updated":"2019-06-05 16:25:42.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-06-06 12:59:43.000000000","updated_by":{"_account_id":10112,"name":"Jason E. Rist","email":"jrist@redhat.com","username":"jrist"},"reviewer":{"_account_id":10112,"name":"Jason E. Rist","email":"jrist@redhat.com","username":"jrist"},"state":"REVIEWER"},{"updated":"2019-06-06 13:00:21.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-06-06 13:03:02.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2019-06-06 21:27:28.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-06-06 23:53:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f8678fae8d35547447c7618ebc1c08391ff4af9c","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-06-03 19:58:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6dc320e5d5d2ddd3afed27fb49d86e3305e1b4dc","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-06-03 19:59:22.000000000","message":"Topic changed from bug/1831478 to retire_tripleo_ui","accounts_in_message":[],"_revision_number":1},{"id":"e8fc4bc74f2316b25d3639dd9e2a042a786b10fa","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-03 20:38:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/46/662846/1/openstack-check/tripleo-build-containers-fedora-28-master/0b94850/ : SUCCESS in 36m 53s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/46/662846/1/openstack-check/tripleo-build-containers-fedora-28-stein/2579f9d/ : SUCCESS in 38m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c96349170f12f54d8e0bc29b195742b69a49f570","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-03 21:20:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/46/662846/1/check/kolla-build-centos-binary/d2e4d66/ : SUCCESS in 48m 56s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/46/662846/1/check/kolla-build-centos-source/04bb50a/ : SUCCESS in 48m 34s\n- kolla-ansible-centos-source http://logs.openstack.org/46/662846/1/check/kolla-ansible-centos-source/141dfa8/ : SUCCESS in 1h 11m 34s\n- kolla-ansible-centos-binary http://logs.openstack.org/46/662846/1/check/kolla-ansible-centos-binary/c6c3ec6/ : SUCCESS in 1h 15m 31s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/46/662846/1/check/kolla-ansible-centos-source-upgrade/bb86640/ : SUCCESS in 1h 08m 20s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/46/662846/1/check/tripleo-build-containers-centos-7/953cb43/ : SUCCESS in 46m 07s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/46/662846/1/check/kolla-build-oraclelinux-binary/006e74a/ : SUCCESS in 53m 26s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/46/662846/1/check/kolla-build-oraclelinux-source/b97c6e7/ : SUCCESS in 37m 37s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/46/662846/1/check/kolla-ansible-oraclelinux-source/22dc3bf/ : SUCCESS in 1h 09m 23s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/46/662846/1/check/kolla-ansible-oraclelinux-binary/7606d37/ : SUCCESS in 1h 05m 25s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/46/662846/1/check/openstack-tox-lower-constraints/0e3ab23/ : SUCCESS in 3m 35s\n- openstack-tox-pep8 http://logs.openstack.org/46/662846/1/check/openstack-tox-pep8/f43d8d7/ : SUCCESS in 3m 48s\n- openstack-tox-py27 http://logs.openstack.org/46/662846/1/check/openstack-tox-py27/f20b3df/ : SUCCESS in 3m 02s\n- openstack-tox-py36 http://logs.openstack.org/46/662846/1/check/openstack-tox-py36/c95675a/ : SUCCESS in 2m 48s\n- openstack-tox-py37 http://logs.openstack.org/46/662846/1/check/openstack-tox-py37/4aa13e7/ : SUCCESS in 3m 21s\n- openstack-tox-docs http://logs.openstack.org/46/662846/1/check/openstack-tox-docs/f36e2c3/html/ : SUCCESS in 2m 51s\n- build-openstack-releasenotes http://logs.openstack.org/46/662846/1/check/build-openstack-releasenotes/6e937b2/html/ : SUCCESS in 3m 34s\n- kolla-build-ubuntu-binary http://logs.openstack.org/46/662846/1/check/kolla-build-ubuntu-binary/f9d6e8f/ : SUCCESS in 32m 27s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/46/662846/1/check/kolla-build-ubuntu-source/1970858/ : SUCCESS in 35m 00s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/46/662846/1/check/kolla-ansible-ubuntu-source/e65bd45/ : SUCCESS in 1h 05m 39s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/46/662846/1/check/kolla-ansible-ubuntu-binary/fad5fb9/ : SUCCESS in 52m 29s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/46/662846/1/check/kolla-ansible-ubuntu-source-upgrade/d214aa9/ : SUCCESS in 1h 03m 24s","accounts_in_message":[],"_revision_number":1},{"id":"4b49938c3557f89cbc8a8f8f73da5ba4f7b8cf1a","author":{"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},"date":"2019-06-04 09:17:26.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"725b595cef9c3adf0119363caf49bf6267a93aa8","author":{"_account_id":30413,"name":"chaiguanglong","email":"chaigl@awcloud.com","username":"chaiguanglong"},"date":"2019-06-05 10:55:13.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"c048edf1f64b062983489724f23a7234999ddab3","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-05 11:25:29.000000000","message":"Patch Set 1: Code-Review+2\n\nThanks","accounts_in_message":[],"_revision_number":1},{"id":"3f8356b8dc99fb6015ad35da734cb0eb777f1b0e","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-06-05 15:21:20.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"075abefad4baf8833ccb54dbc316fa95919b7a89","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-05 16:01:30.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/46/662846/2/openstack-check/tripleo-build-containers-fedora-28-master/c905aec/ : SUCCESS in 33m 41s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/46/662846/2/openstack-check/tripleo-build-containers-fedora-28-stein/c566d2d/ : SUCCESS in 36m 57s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"519ec4e37bc84e797b01fd2c408ba43419b051fa","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-05 16:25:42.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"bdfe7c7cd8c66297cc2334670850386130d3c698","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-05 17:07:28.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/46/662846/2/check/kolla-build-centos-binary/a83f046/ : FAILURE in 3m 24s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/46/662846/2/check/kolla-build-centos-source/0ce836b/ : SUCCESS in 53m 01s\n- kolla-ansible-centos-source http://logs.openstack.org/46/662846/2/check/kolla-ansible-centos-source/99a84e7/ : SUCCESS in 1h 04m 02s\n- kolla-ansible-centos-binary http://logs.openstack.org/46/662846/2/check/kolla-ansible-centos-binary/c9407fe/ : SUCCESS in 1h 07m 27s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/46/662846/2/check/kolla-ansible-centos-source-upgrade/31e3734/ : SUCCESS in 1h 17m 44s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/46/662846/2/check/tripleo-build-containers-centos-7/cd66ce4/ : SUCCESS in 48m 48s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/46/662846/2/check/kolla-build-oraclelinux-binary/8d04aab/ : SUCCESS in 53m 37s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/46/662846/2/check/kolla-build-oraclelinux-source/255e79a/ : SUCCESS in 56m 00s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/46/662846/2/check/kolla-ansible-oraclelinux-source/ae07af2/ : SUCCESS in 1h 26m 18s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/46/662846/2/check/kolla-ansible-oraclelinux-binary/193bd93/ : SUCCESS in 1h 13m 28s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/46/662846/2/check/openstack-tox-lower-constraints/1133018/ : SUCCESS in 5m 28s\n- openstack-tox-pep8 http://logs.openstack.org/46/662846/2/check/openstack-tox-pep8/bd2b4ec/ : SUCCESS in 4m 07s\n- openstack-tox-py27 http://logs.openstack.org/46/662846/2/check/openstack-tox-py27/b5486a7/ : SUCCESS in 4m 02s\n- openstack-tox-py36 http://logs.openstack.org/46/662846/2/check/openstack-tox-py36/a62cb20/ : SUCCESS in 3m 51s\n- openstack-tox-py37 http://logs.openstack.org/46/662846/2/check/openstack-tox-py37/5c9b612/ : SUCCESS in 2m 44s\n- openstack-tox-docs http://logs.openstack.org/46/662846/2/check/openstack-tox-docs/26ab970/html/ : SUCCESS in 2m 46s\n- build-openstack-releasenotes http://logs.openstack.org/46/662846/2/check/build-openstack-releasenotes/cb84d1b/html/ : SUCCESS in 4m 20s\n- kolla-build-ubuntu-binary http://logs.openstack.org/46/662846/2/check/kolla-build-ubuntu-binary/e421162/ : SUCCESS in 35m 16s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/46/662846/2/check/kolla-build-ubuntu-source/2c4c215/ : SUCCESS in 33m 10s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/46/662846/2/check/kolla-ansible-ubuntu-source/7ac8702/ : SUCCESS in 53m 55s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/46/662846/2/check/kolla-ansible-ubuntu-binary/5a83c3e/ : SUCCESS in 1h 03m 19s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/46/662846/2/check/kolla-ansible-ubuntu-source-upgrade/a9e52f1/ : SUCCESS in 1h 07m 16s","accounts_in_message":[],"_revision_number":2},{"id":"0d11935729925c4abfbd19a0076c8dba594f35a8","author":{"_account_id":10112,"name":"Jason E. Rist","email":"jrist@redhat.com","username":"jrist"},"date":"2019-06-06 12:59:43.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"01bf503c9909496b65d7ad3e4b85970350d694f5","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-06-06 13:00:21.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"f5d925615cb0b6fffffa385aca87a643b0cd4616","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-06-06 13:03:02.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"bab559c63ad7e62ae5e0305d09d6edfb47cd5443","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 13:04:56.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e6ee2a602c9ffc949cd841edc4ce341596ab5599","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 14:17:32.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-source http://logs.openstack.org/46/662846/2/gate/kolla-build-centos-source/f4d2187/ : FAILURE in 15m 49s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/46/662846/2/gate/tripleo-build-containers-centos-7/1c31ce8/ : SUCCESS in 49m 01s\n- kolla-build-oraclelinux-source http://logs.openstack.org/46/662846/2/gate/kolla-build-oraclelinux-source/e540e0e/ : SUCCESS in 58m 54s\n- openstack-tox-lower-constraints http://logs.openstack.org/46/662846/2/gate/openstack-tox-lower-constraints/af2c9ef/ : SUCCESS in 3m 50s\n- openstack-tox-pep8 http://logs.openstack.org/46/662846/2/gate/openstack-tox-pep8/bec317b/ : SUCCESS in 3m 09s\n- openstack-tox-py27 http://logs.openstack.org/46/662846/2/gate/openstack-tox-py27/e6eb5dc/ : SUCCESS in 3m 19s\n- openstack-tox-py36 http://logs.openstack.org/46/662846/2/gate/openstack-tox-py36/a9fea32/ : SUCCESS in 3m 21s\n- openstack-tox-py37 http://logs.openstack.org/46/662846/2/gate/openstack-tox-py37/0bf2db2/ : SUCCESS in 3m 33s\n- openstack-tox-docs http://logs.openstack.org/46/662846/2/gate/openstack-tox-docs/5d5d10c/html/ : SUCCESS in 2m 40s\n- build-openstack-releasenotes http://logs.openstack.org/46/662846/2/gate/build-openstack-releasenotes/8eee136/html/ : SUCCESS in 4m 14s\n- kolla-build-ubuntu-source http://logs.openstack.org/46/662846/2/gate/kolla-build-ubuntu-source/fb88088/ : SUCCESS in 1h 07m 00s","accounts_in_message":[],"_revision_number":2},{"id":"c0159d7ffd54ffd7ece48fceb5642cca4aa02751","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-06 14:22:53.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"407ea8f096f9dd5bdd0bf17f963a1ae18ef95854","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-06-06 14:49:11.000000000","message":"Patch Set 2:\n\nrecheck curl: (7) Failed connect to www.percona.com:443; Connection timed out","accounts_in_message":[],"_revision_number":2},{"id":"52b975a4cc1add84633dcf7c313249e5a1238206","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-06 15:06:39.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/46/662846/2/openstack-check/tripleo-build-containers-fedora-28-master/06aa02f/ : SUCCESS in 36m 04s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/46/662846/2/openstack-check/tripleo-build-containers-fedora-28-stein/bfb0cd7/ : SUCCESS in 40m 30s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6e5278449cca289311ac9f6be665cc1c6bf47807","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 16:05:52.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/46/662846/2/check/kolla-build-centos-binary/f16320d/ : SUCCESS in 39m 10s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/46/662846/2/check/kolla-build-centos-source/9a0f6a6/ : SUCCESS in 48m 03s\n- kolla-ansible-centos-source http://logs.openstack.org/46/662846/2/check/kolla-ansible-centos-source/c672531/ : SUCCESS in 1h 23m 38s\n- kolla-ansible-centos-binary http://logs.openstack.org/46/662846/2/check/kolla-ansible-centos-binary/a4da3e8/ : SUCCESS in 1h 11m 08s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/46/662846/2/check/kolla-ansible-centos-source-upgrade/93b561b/ : SUCCESS in 1h 20m 51s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/46/662846/2/check/tripleo-build-containers-centos-7/ca20102/ : SUCCESS in 48m 49s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/46/662846/2/check/kolla-build-oraclelinux-binary/c9500ca/ : SUCCESS in 51m 59s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/46/662846/2/check/kolla-build-oraclelinux-source/805eaa3/ : SUCCESS in 58m 41s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/46/662846/2/check/kolla-ansible-oraclelinux-source/50e2eab/ : SUCCESS in 1h 14m 07s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/46/662846/2/check/kolla-ansible-oraclelinux-binary/d811333/ : SUCCESS in 1h 18m 01s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/46/662846/2/check/openstack-tox-lower-constraints/5ea250c/ : SUCCESS in 3m 46s\n- openstack-tox-pep8 http://logs.openstack.org/46/662846/2/check/openstack-tox-pep8/1f98650/ : SUCCESS in 3m 05s\n- openstack-tox-py27 http://logs.openstack.org/46/662846/2/check/openstack-tox-py27/d34f413/ : SUCCESS in 2m 43s\n- openstack-tox-py36 http://logs.openstack.org/46/662846/2/check/openstack-tox-py36/6fc9475/ : SUCCESS in 3m 05s\n- openstack-tox-py37 http://logs.openstack.org/46/662846/2/check/openstack-tox-py37/341919e/ : SUCCESS in 3m 09s\n- openstack-tox-docs http://logs.openstack.org/46/662846/2/check/openstack-tox-docs/9c28774/html/ : SUCCESS in 2m 59s\n- build-openstack-releasenotes http://logs.openstack.org/46/662846/2/check/build-openstack-releasenotes/47ac89b/html/ : SUCCESS in 4m 04s\n- kolla-build-ubuntu-binary http://logs.openstack.org/46/662846/2/check/kolla-build-ubuntu-binary/9956233/ : FAILURE in 7m 58s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/46/662846/2/check/kolla-build-ubuntu-source/7b62391/ : FAILURE in 5m 48s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/46/662846/2/check/kolla-ansible-ubuntu-source/2b8e437/ : SUCCESS in 1h 05m 35s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/46/662846/2/check/kolla-ansible-ubuntu-binary/8c7bdbd/ : SUCCESS in 56m 23s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/46/662846/2/check/kolla-ansible-ubuntu-source-upgrade/db935ff/ : SUCCESS in 1h 15m 44s","accounts_in_message":[],"_revision_number":2},{"id":"9e17361e592462f873db03e8b01aed5cf81aca73","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-06-06 16:14:48.000000000","message":"Patch Set 2:\n\nrecheck keyserver problem http://logs.openstack.org/46/662846/2/check/kolla-build-ubuntu-source/7b62391/job-output.txt.gz#_2019-06-06_14_40_21_889988","accounts_in_message":[],"_revision_number":2},{"id":"d7de35acf9a6f90da6e6fdc5630e669a122e4135","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-06 17:05:13.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/46/662846/2/openstack-check/tripleo-build-containers-fedora-28-master/ff79e9b/ : SUCCESS in 38m 29s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/46/662846/2/openstack-check/tripleo-build-containers-fedora-28-stein/fe1f1d6/ : SUCCESS in 40m 48s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a7ff3d3686421fd434218aea59a2f79f907893b7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 17:40:42.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/46/662846/2/check/kolla-build-centos-binary/0e78219/ : SUCCESS in 35m 20s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/46/662846/2/check/kolla-build-centos-source/46fe261/ : SUCCESS in 50m 28s\n- kolla-ansible-centos-source http://logs.openstack.org/46/662846/2/check/kolla-ansible-centos-source/1ddb259/ : SUCCESS in 1h 09m 17s\n- kolla-ansible-centos-binary http://logs.openstack.org/46/662846/2/check/kolla-ansible-centos-binary/3cde8a6/ : SUCCESS in 1h 02m 23s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/46/662846/2/check/kolla-ansible-centos-source-upgrade/5f01d88/ : FAILURE in 19m 33s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/46/662846/2/check/tripleo-build-containers-centos-7/7b2868b/ : SUCCESS in 41m 50s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/46/662846/2/check/kolla-build-oraclelinux-binary/0866791/ : SUCCESS in 48m 04s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/46/662846/2/check/kolla-build-oraclelinux-source/9e52a38/ : SUCCESS in 52m 41s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/46/662846/2/check/kolla-ansible-oraclelinux-source/e6ad5df/ : SUCCESS in 1h 19m 20s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/46/662846/2/check/kolla-ansible-oraclelinux-binary/2ab7126/ : SUCCESS in 1h 11m 20s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/46/662846/2/check/openstack-tox-lower-constraints/4b10bc9/ : SUCCESS in 4m 06s\n- openstack-tox-pep8 http://logs.openstack.org/46/662846/2/check/openstack-tox-pep8/10006f5/ : SUCCESS in 3m 19s\n- openstack-tox-py27 http://logs.openstack.org/46/662846/2/check/openstack-tox-py27/38c4243/ : SUCCESS in 3m 19s\n- openstack-tox-py36 http://logs.openstack.org/46/662846/2/check/openstack-tox-py36/0c5e626/ : SUCCESS in 3m 05s\n- openstack-tox-py37 http://logs.openstack.org/46/662846/2/check/openstack-tox-py37/eae2c51/ : SUCCESS in 4m 02s\n- openstack-tox-docs http://logs.openstack.org/46/662846/2/check/openstack-tox-docs/7d3e655/html/ : SUCCESS in 3m 02s\n- build-openstack-releasenotes http://logs.openstack.org/46/662846/2/check/build-openstack-releasenotes/d21ae3b/html/ : SUCCESS in 4m 22s\n- kolla-build-ubuntu-binary http://logs.openstack.org/46/662846/2/check/kolla-build-ubuntu-binary/2b2e27e/ : SUCCESS in 33m 15s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/46/662846/2/check/kolla-build-ubuntu-source/af8ab99/ : SUCCESS in 32m 35s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/46/662846/2/check/kolla-ansible-ubuntu-source/94111e5/ : SUCCESS in 52m 28s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/46/662846/2/check/kolla-ansible-ubuntu-binary/fce7b3d/ : SUCCESS in 48m 29s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/46/662846/2/check/kolla-ansible-ubuntu-source-upgrade/607cee6/ : SUCCESS in 1h 01m 47s","accounts_in_message":[],"_revision_number":2},{"id":"d5746a26f880517d23c228e5a783d05cc7787e09","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-06-06 18:47:53.000000000","message":"Patch Set 2:\n\nrecheck follow the bouncing failure, curl: (7) Failed connect to www.percona.com:443; Connection timed out","accounts_in_message":[],"_revision_number":2},{"id":"b71304ce9ed9b6806ccf5fad39b5f164ea3b3af2","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-06 19:38:55.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/46/662846/2/openstack-check/tripleo-build-containers-fedora-28-master/9895ae8/ : SUCCESS in 45m 45s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/46/662846/2/openstack-check/tripleo-build-containers-fedora-28-stein/1a9dabd/ : SUCCESS in 39m 19s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0ad86e073e0452a5c16987d55c28b9fac19f87ad","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 20:32:24.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/46/662846/2/check/kolla-build-centos-binary/02e6952/ : SUCCESS in 39m 31s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/46/662846/2/check/kolla-build-centos-source/245e7f6/ : SUCCESS in 59m 30s\n- kolla-ansible-centos-source finger://ze08.openstack.org/44e512c1a912444ca8da89ab9c3c40da : RETRY_LIMIT in 29m 35s\n- kolla-ansible-centos-binary http://logs.openstack.org/46/662846/2/check/kolla-ansible-centos-binary/31a531f/ : SUCCESS in 1h 14m 48s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/46/662846/2/check/kolla-ansible-centos-source-upgrade/63fb7b4/ : SUCCESS in 1h 32m 08s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/46/662846/2/check/tripleo-build-containers-centos-7/4339d44/ : SUCCESS in 46m 58s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/46/662846/2/check/kolla-build-oraclelinux-binary/91cc75c/ : SUCCESS in 57m 25s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/46/662846/2/check/kolla-build-oraclelinux-source/3480171/ : SUCCESS in 59m 56s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/46/662846/2/check/kolla-ansible-oraclelinux-source/1ce211e/ : SUCCESS in 1h 09m 27s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/46/662846/2/check/kolla-ansible-oraclelinux-binary/abb6cbd/ : SUCCESS in 1h 17m 02s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/46/662846/2/check/openstack-tox-lower-constraints/dc8b7b3/ : SUCCESS in 4m 14s\n- openstack-tox-pep8 http://logs.openstack.org/46/662846/2/check/openstack-tox-pep8/2c3cb75/ : SUCCESS in 4m 10s\n- openstack-tox-py27 http://logs.openstack.org/46/662846/2/check/openstack-tox-py27/43c4bf2/ : SUCCESS in 3m 23s\n- openstack-tox-py36 http://logs.openstack.org/46/662846/2/check/openstack-tox-py36/57a10c8/ : SUCCESS in 3m 07s\n- openstack-tox-py37 http://logs.openstack.org/46/662846/2/check/openstack-tox-py37/4c56b92/ : SUCCESS in 4m 00s\n- openstack-tox-docs http://logs.openstack.org/46/662846/2/check/openstack-tox-docs/e317d60/html/ : SUCCESS in 3m 52s\n- build-openstack-releasenotes http://logs.openstack.org/46/662846/2/check/build-openstack-releasenotes/9bb2cf0/html/ : SUCCESS in 4m 01s\n- kolla-build-ubuntu-binary http://logs.openstack.org/46/662846/2/check/kolla-build-ubuntu-binary/71b706a/ : SUCCESS in 31m 24s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/46/662846/2/check/kolla-build-ubuntu-source/db5b8ef/ : SUCCESS in 48m 31s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/46/662846/2/check/kolla-ansible-ubuntu-source/87a8eee/ : SUCCESS in 58m 03s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/46/662846/2/check/kolla-ansible-ubuntu-binary/d6700e7/ : SUCCESS in 46m 53s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/46/662846/2/check/kolla-ansible-ubuntu-source-upgrade/bbe5158/ : SUCCESS in 1h 05m 58s","accounts_in_message":[],"_revision_number":2},{"id":"0cd27a19a57abb5c2af71effdec56eb800d9d28b","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-06-06 20:44:23.000000000","message":"Patch Set 2:\n\nrecheck this change just doesn\u0027t want to merge it seems, kolla-ansible-centos-source finger://ze08.openstack.org/44e512c1a912444ca8da89ab9c3c40da : RETRY_LIMIT in 29m 35s","accounts_in_message":[],"_revision_number":2},{"id":"e0fd36df2f820da5616b5a99a4174a765aca1939","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-06 21:27:28.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/46/662846/2/openstack-check/tripleo-build-containers-fedora-28-master/fea16a7/ : SUCCESS in 38m 55s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/46/662846/2/openstack-check/tripleo-build-containers-fedora-28-stein/a87071d/ : SUCCESS in 41m 45s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"32d00438c57bdec71a961eba6b9093ed3d6cb784","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 22:39:21.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/46/662846/2/check/kolla-build-centos-binary/27b769e/ : SUCCESS in 43m 40s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/46/662846/2/check/kolla-build-centos-source/52eac9d/ : SUCCESS in 50m 46s\n- kolla-ansible-centos-source http://logs.openstack.org/46/662846/2/check/kolla-ansible-centos-source/017a210/ : SUCCESS in 1h 13m 20s\n- kolla-ansible-centos-binary http://logs.openstack.org/46/662846/2/check/kolla-ansible-centos-binary/6af98bb/ : FAILURE in 19m 18s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/46/662846/2/check/kolla-ansible-centos-source-upgrade/2072daa/ : SUCCESS in 1h 18m 31s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/46/662846/2/check/tripleo-build-containers-centos-7/ac71427/ : SUCCESS in 43m 51s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/46/662846/2/check/kolla-build-oraclelinux-binary/1ab5087/ : SUCCESS in 52m 58s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/46/662846/2/check/kolla-build-oraclelinux-source/63525af/ : SUCCESS in 50m 28s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/46/662846/2/check/kolla-ansible-oraclelinux-source/ba2dd0f/ : SUCCESS in 1h 14m 57s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/46/662846/2/check/kolla-ansible-oraclelinux-binary/153bcf7/ : SUCCESS in 1h 05m 02s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/46/662846/2/check/openstack-tox-lower-constraints/20bb0cc/ : SUCCESS in 5m 46s\n- openstack-tox-pep8 http://logs.openstack.org/46/662846/2/check/openstack-tox-pep8/8d7236a/ : SUCCESS in 4m 13s\n- openstack-tox-py27 http://logs.openstack.org/46/662846/2/check/openstack-tox-py27/164393a/ : SUCCESS in 4m 07s\n- openstack-tox-py36 http://logs.openstack.org/46/662846/2/check/openstack-tox-py36/9c83196/ : SUCCESS in 4m 55s\n- openstack-tox-py37 http://logs.openstack.org/46/662846/2/check/openstack-tox-py37/70a330c/ : SUCCESS in 3m 03s\n- openstack-tox-docs http://logs.openstack.org/46/662846/2/check/openstack-tox-docs/11765ff/html/ : SUCCESS in 3m 21s\n- build-openstack-releasenotes http://logs.openstack.org/46/662846/2/check/build-openstack-releasenotes/cbe76e5/html/ : SUCCESS in 5m 06s\n- kolla-build-ubuntu-binary http://logs.openstack.org/46/662846/2/check/kolla-build-ubuntu-binary/6f0aff3/ : SUCCESS in 34m 25s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/46/662846/2/check/kolla-build-ubuntu-source/70fd892/ : SUCCESS in 29m 24s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/46/662846/2/check/kolla-ansible-ubuntu-source/4307baa/ : SUCCESS in 53m 33s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/46/662846/2/check/kolla-ansible-ubuntu-binary/20cb48e/ : SUCCESS in 53m 47s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/46/662846/2/check/kolla-ansible-ubuntu-source-upgrade/46fd69a/ : SUCCESS in 1h 07m 50s","accounts_in_message":[],"_revision_number":2},{"id":"3925f054071969a42d1af1965dcf6a4efd9ddba5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 22:39:51.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e29440e2fabfd56d6e2cc0f82e46136791f2aba1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 23:53:01.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"60d3abbf85eaec332ac2b4f3756f08d9d6b3fe87","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 23:53:01.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source http://logs.openstack.org/46/662846/2/gate/kolla-build-centos-source/224831d/ : SUCCESS in 54m 01s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/46/662846/2/gate/tripleo-build-containers-centos-7/9207fc0/ : SUCCESS in 39m 07s\n- kolla-build-oraclelinux-source http://logs.openstack.org/46/662846/2/gate/kolla-build-oraclelinux-source/b4e54d4/ : SUCCESS in 1h 01m 34s\n- openstack-tox-lower-constraints http://logs.openstack.org/46/662846/2/gate/openstack-tox-lower-constraints/6c99fae/ : SUCCESS in 5m 14s\n- openstack-tox-pep8 http://logs.openstack.org/46/662846/2/gate/openstack-tox-pep8/3157998/ : SUCCESS in 3m 10s\n- openstack-tox-py27 http://logs.openstack.org/46/662846/2/gate/openstack-tox-py27/6f5af8f/ : SUCCESS in 3m 40s\n- openstack-tox-py36 http://logs.openstack.org/46/662846/2/gate/openstack-tox-py36/2cafbe9/ : SUCCESS in 4m 18s\n- openstack-tox-py37 http://logs.openstack.org/46/662846/2/gate/openstack-tox-py37/5923b18/ : SUCCESS in 4m 15s\n- openstack-tox-docs http://logs.openstack.org/46/662846/2/gate/openstack-tox-docs/9cc38ca/html/ : SUCCESS in 3m 24s\n- build-openstack-releasenotes http://logs.openstack.org/46/662846/2/gate/build-openstack-releasenotes/e39f27b/html/ : SUCCESS in 4m 09s\n- kolla-build-ubuntu-source http://logs.openstack.org/46/662846/2/gate/kolla-build-ubuntu-source/820cd28/ : SUCCESS in 36m 59s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e0eb7b79b22a7577eb3b6eeabc19e8b79b068d9c","revisions":{"538feabca48a4387648200297aa54b7afdcf305e":{"kind":"REWORK","_number":1,"created":"2019-06-03 19:58:55.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/46/662846/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/662846/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/662846/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/662846/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/662846/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/662846/1"}}},"commit":{"parents":[{"commit":"4e083196d14877de458ae16129331c37fe1f6f21","subject":"Merge \"Use the default policy.json for Qinling\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4e083196d14877de458ae16129331c37fe1f6f21"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-06-03 19:58:05.000000000","tz":-360},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-06-03 19:58:05.000000000","tz":-360},"subject":"Remove tripleo-ui","message":"Remove tripleo-ui\n\nThe tripleo-ui project has been removed from the upstream. Remove the\nreferences to the tripleo-ui containers.\n\nChange-Id: I353ecd3ed04c78fd21f0108c5e1355280d3f3bb9\nCloses-Bug: #1831478\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/538feabca48a4387648200297aa54b7afdcf305e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/538feabca48a4387648200297aa54b7afdcf305e"}]},"branch":"refs/heads/master"},"e0eb7b79b22a7577eb3b6eeabc19e8b79b068d9c":{"kind":"REWORK","_number":2,"created":"2019-06-05 15:21:20.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/46/662846/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/662846/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/662846/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/662846/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/662846/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/662846/2"}}},"commit":{"parents":[{"commit":"4926adb62465bf819fca2dacd42454d4e9eb74df","subject":"Merge \"create missing apache2 directory on Debian/Ubuntu\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4926adb62465bf819fca2dacd42454d4e9eb74df"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-06-03 19:58:05.000000000","tz":-360},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-06-05 15:21:14.000000000","tz":-360},"subject":"Remove tripleo-ui","message":"Remove tripleo-ui\n\nThe tripleo-ui project has been removed from the upstream. Remove the\nreferences to the tripleo-ui containers.\n\nChange-Id: I353ecd3ed04c78fd21f0108c5e1355280d3f3bb9\nCloses-Bug: #1831478\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e0eb7b79b22a7577eb3b6eeabc19e8b79b068d9c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e0eb7b79b22a7577eb3b6eeabc19e8b79b068d9c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
