)]}'
{"id":"openstack%2Fkolla~663300","triplet_id":"openstack%2Fkolla~master~I880a33b56a2fe9b86de57326c80310a6a83c5e50","project":"openstack/kolla","branch":"master","hashtags":[],"change_id":"I880a33b56a2fe9b86de57326c80310a6a83c5e50","subject":"Use packagecloud/ubuntu rabbitmq repo for all archs","status":"MERGED","created":"2019-06-05 11:06:47.000000000","updated":"2019-06-07 15:06:15.000000000","submitted":"2019-06-07 15:06:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"663300-1559919975272-f1351380","meta_rev_id":"0ef26a432a43fd32e6645c4463e36b57e1a6f775","_number":663300,"virtual_id_number":663300,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-06-07 15:06:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-06-06 09:00:12.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":2,"date":"2019-06-07 14:06:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2019-06-07 09:19:10.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":1,"date":"2019-06-07 14:06:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"date":"2019-06-07 12:27:37.000000000","_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2019-06-05 11:09:54.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-06 09:00:12.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-06-07 09:19:10.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-06-07 12:27:37.000000000","updated_by":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"reviewer":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"state":"REVIEWER"},{"updated":"2019-06-07 14:06:13.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2019-06-07 15:06:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5e8b1accc6f320a25552a30ec06f3c0a90b2705b","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-05 11:06:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"933443b3814e901d60e38421257e274d1d8b4383","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-05 11:09:54.000000000","message":"Patch Set 1: Review-Priority+1\n\nbumping review priority as it unblocks rabbitmq/ubuntu builds for !x86","accounts_in_message":[],"_revision_number":1},{"id":"471702c6b00930cb3ff7ee26ba763d023a397d33","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-05 11:33:58.000000000","message":"Patch Set 1:\n\nWhat was the issue with the distro rabbit on !x86?\n\nWhat version of Rabbit do we have before and after?\n\nAre there any concerns around deployment and upgrade?","accounts_in_message":[],"_revision_number":1},{"id":"bd87e47b626512718397240a74a5362f4bc6580e","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-05 11:44:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/00/663300/1/openstack-check/tripleo-build-containers-fedora-28-master/bda7619/ : SUCCESS in 34m 14s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/00/663300/1/openstack-check/tripleo-build-containers-fedora-28-stein/824d5f9/ : SUCCESS in 34m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"be49503b736eb58f95fffea0d04e6f4049841e4e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-05 12:32:52.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/00/663300/1/check/kolla-build-centos-binary/8aefdc6/ : SUCCESS in 45m 16s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/00/663300/1/check/kolla-build-centos-source/01a61e1/ : FAILURE in 10m 07s\n- kolla-ansible-centos-source http://logs.openstack.org/00/663300/1/check/kolla-ansible-centos-source/94c57fa/ : SUCCESS in 1h 13m 44s\n- kolla-ansible-centos-binary http://logs.openstack.org/00/663300/1/check/kolla-ansible-centos-binary/06ee71a/ : SUCCESS in 1h 08m 47s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/00/663300/1/check/kolla-ansible-centos-source-upgrade/f0fa162/ : SUCCESS in 1h 15m 36s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/00/663300/1/check/tripleo-build-containers-centos-7/0d3e80a/ : SUCCESS in 48m 11s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/00/663300/1/check/kolla-build-oraclelinux-binary/fb0e942/ : SUCCESS in 1h 02m 19s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/00/663300/1/check/kolla-build-oraclelinux-source/07e1d37/ : SUCCESS in 1h 02m 51s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/00/663300/1/check/kolla-ansible-oraclelinux-source/0047921/ : SUCCESS in 1h 14m 14s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/00/663300/1/check/kolla-ansible-oraclelinux-binary/94bd00c/ : SUCCESS in 1h 13m 32s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/00/663300/1/check/openstack-tox-lower-constraints/6a88cca/ : SUCCESS in 3m 37s\n- openstack-tox-pep8 http://logs.openstack.org/00/663300/1/check/openstack-tox-pep8/b64befa/ : SUCCESS in 3m 14s\n- openstack-tox-py27 http://logs.openstack.org/00/663300/1/check/openstack-tox-py27/96b00f0/ : SUCCESS in 4m 03s\n- openstack-tox-py36 http://logs.openstack.org/00/663300/1/check/openstack-tox-py36/6d92e1e/ : SUCCESS in 3m 12s\n- openstack-tox-py37 http://logs.openstack.org/00/663300/1/check/openstack-tox-py37/2ab86ae/ : SUCCESS in 3m 42s\n- openstack-tox-docs http://logs.openstack.org/00/663300/1/check/openstack-tox-docs/11177ea/html/ : SUCCESS in 2m 30s\n- kolla-build-ubuntu-binary http://logs.openstack.org/00/663300/1/check/kolla-build-ubuntu-binary/1115d3e/ : SUCCESS in 33m 41s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/00/663300/1/check/kolla-build-ubuntu-source/0cef487/ : SUCCESS in 32m 36s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/00/663300/1/check/kolla-ansible-ubuntu-source/3b8c58a/ : SUCCESS in 53m 49s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/00/663300/1/check/kolla-ansible-ubuntu-binary/d08f192/ : SUCCESS in 44m 17s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/00/663300/1/check/kolla-ansible-ubuntu-source-upgrade/8ea358d/ : SUCCESS in 1h 03m 40s","accounts_in_message":[],"_revision_number":1},{"id":"1d79b7290e55d9c9ca5925fdf158410bcb6669bb","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-06 08:17:51.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"8e86e963edf408bf86e2b4d8b4da3898b5727175","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-06 09:00:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/00/663300/1/openstack-check/tripleo-build-containers-fedora-28-master/fc3b7c7/ : SUCCESS in 38m 43s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/00/663300/1/openstack-check/tripleo-build-containers-fedora-28-stein/cdc8f71/ : SUCCESS in 36m 44s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"808c2c01edb40874598d5cf4502e585537b329d0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 09:51:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/00/663300/1/check/kolla-build-centos-binary/2b18dc3/ : SUCCESS in 25m 14s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/00/663300/1/check/kolla-build-centos-source/a29e587/ : SUCCESS in 1h 00m 42s\n- kolla-ansible-centos-source http://logs.openstack.org/00/663300/1/check/kolla-ansible-centos-source/a897f3e/ : SUCCESS in 1h 15m 37s\n- kolla-ansible-centos-binary http://logs.openstack.org/00/663300/1/check/kolla-ansible-centos-binary/c6e20d0/ : SUCCESS in 1h 05m 34s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/00/663300/1/check/kolla-ansible-centos-source-upgrade/2c47f7c/ : SUCCESS in 1h 22m 42s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/00/663300/1/check/tripleo-build-containers-centos-7/4d0e30b/ : SUCCESS in 38m 57s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/00/663300/1/check/kolla-build-oraclelinux-binary/4ec8de9/ : SUCCESS in 47m 23s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/00/663300/1/check/kolla-build-oraclelinux-source/5e6c672/ : SUCCESS in 51m 54s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/00/663300/1/check/kolla-ansible-oraclelinux-source/b8bd94e/ : SUCCESS in 1h 13m 59s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/00/663300/1/check/kolla-ansible-oraclelinux-binary/d2c1a88/ : SUCCESS in 1h 11m 45s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/00/663300/1/check/openstack-tox-lower-constraints/54a6164/ : SUCCESS in 3m 31s\n- openstack-tox-pep8 http://logs.openstack.org/00/663300/1/check/openstack-tox-pep8/1cb65a2/ : SUCCESS in 3m 00s\n- openstack-tox-py27 http://logs.openstack.org/00/663300/1/check/openstack-tox-py27/f7e2149/ : SUCCESS in 3m 38s\n- openstack-tox-py36 http://logs.openstack.org/00/663300/1/check/openstack-tox-py36/3bdcfe6/ : SUCCESS in 3m 19s\n- openstack-tox-py37 http://logs.openstack.org/00/663300/1/check/openstack-tox-py37/8b38336/ : SUCCESS in 4m 04s\n- openstack-tox-docs http://logs.openstack.org/00/663300/1/check/openstack-tox-docs/70c9509/html/ : SUCCESS in 2m 52s\n- kolla-build-ubuntu-binary http://logs.openstack.org/00/663300/1/check/kolla-build-ubuntu-binary/b634f40/ : SUCCESS in 1h 10m 05s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/00/663300/1/check/kolla-build-ubuntu-source/30b0e29/ : SUCCESS in 52m 49s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/00/663300/1/check/kolla-ansible-ubuntu-source/8d149a6/ : SUCCESS in 53m 59s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/00/663300/1/check/kolla-ansible-ubuntu-binary/dd1732d/ : SUCCESS in 45m 19s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/00/663300/1/check/kolla-ansible-ubuntu-source-upgrade/b5d5c24/ : SUCCESS in 1h 01m 40s","accounts_in_message":[],"_revision_number":1},{"id":"e256cfe30668fb2ac0830e29fcb168c5f1ee2939","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-07 09:11:45.000000000","message":"Patch Set 1:\n\n\u003e What was the issue with the distro rabbit on !x86?\n\nUbuntu forces 3.7.10-1 version so build was failing.\n\n\u003e What version of Rabbit do we have before and after?\n\nNone - build failed for Ubuntu unless either sources.list or rabbitmq/Dockerfile was edit.\n \n\u003e Are there any concerns around deployment and upgrade?\n\nAs it fixes build failure there is no concerns.","accounts_in_message":[],"_revision_number":1},{"id":"94381ad6d4bd9fcc5d63c6088d6955c30e77a3c1","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-07 09:19:10.000000000","message":"Patch Set 1: Code-Review+2\n\n\u003e \u003e What was the issue with the distro rabbit on !x86?\n \u003e \n \u003e Ubuntu forces 3.7.10-1 version so build was failing.\n \u003e \n \u003e \u003e What version of Rabbit do we have before and after?\n \u003e \n \u003e None - build failed for Ubuntu unless either sources.list or\n \u003e rabbitmq/Dockerfile was edit.\n \u003e \n \u003e \u003e Are there any concerns around deployment and upgrade?\n \u003e \n \u003e As it fixes build failure there is no concerns.\n\nI missed the aarch64 and ppc64ie suffix on those files, thought this would affect x86 but looks like it will not. Thanks.","accounts_in_message":[],"_revision_number":1},{"id":"7500dcd3f9943bf4c99c11008a14ebd3e3ed72f4","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-06-07 14:06:13.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"839b7305cc187dbc38f39fa5eaf439be16d86929","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 14:06:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"692845feacdb3708d77feed6ca29b525ca41c295","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 15:06:15.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"0ef26a432a43fd32e6645c4463e36b57e1a6f775","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 15:06:15.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source http://logs.openstack.org/00/663300/1/gate/kolla-build-centos-source/d993657/ : SUCCESS in 37m 17s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/00/663300/1/gate/tripleo-build-containers-centos-7/cbeac89/ : SUCCESS in 45m 34s\n- kolla-build-oraclelinux-source http://logs.openstack.org/00/663300/1/gate/kolla-build-oraclelinux-source/2135af8/ : SUCCESS in 53m 49s\n- openstack-tox-lower-constraints http://logs.openstack.org/00/663300/1/gate/openstack-tox-lower-constraints/39ff490/ : SUCCESS in 5m 19s\n- openstack-tox-pep8 http://logs.openstack.org/00/663300/1/gate/openstack-tox-pep8/eca91b7/ : SUCCESS in 3m 12s\n- openstack-tox-py27 http://logs.openstack.org/00/663300/1/gate/openstack-tox-py27/7cc1622/ : SUCCESS in 3m 05s\n- openstack-tox-py36 http://logs.openstack.org/00/663300/1/gate/openstack-tox-py36/8b51c8e/ : SUCCESS in 2m 43s\n- openstack-tox-py37 http://logs.openstack.org/00/663300/1/gate/openstack-tox-py37/5db5f46/ : SUCCESS in 3m 31s\n- openstack-tox-docs http://logs.openstack.org/00/663300/1/gate/openstack-tox-docs/709011a/html/ : SUCCESS in 2m 43s\n- kolla-build-ubuntu-source http://logs.openstack.org/00/663300/1/gate/kolla-build-ubuntu-source/d7fac8b/ : SUCCESS in 31m 00s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2c4e4422e8ddd9d86c6c864f2e8891e83d65a3c8","revisions":{"2c4e4422e8ddd9d86c6c864f2e8891e83d65a3c8":{"kind":"REWORK","_number":1,"created":"2019-06-05 11:06:47.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/00/663300/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/00/663300/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/663300/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/663300/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/663300/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/00/663300/1"}}},"commit":{"parents":[{"commit":"00475feb7f62c192b6ff5f12bff1b77efd85da6f","subject":"Merge \"Ceph: Update to Nautilus on CentOS and OracleLinux\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/00475feb7f62c192b6ff5f12bff1b77efd85da6f"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-06-05 11:06:36.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-06-05 11:06:36.000000000","tz":120},"subject":"Use packagecloud/ubuntu rabbitmq repo for all archs","message":"Use packagecloud/ubuntu rabbitmq repo for all archs\n\nChange-Id: I880a33b56a2fe9b86de57326c80310a6a83c5e50\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2c4e4422e8ddd9d86c6c864f2e8891e83d65a3c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2c4e4422e8ddd9d86c6c864f2e8891e83d65a3c8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
