)]}'
{"id":"openstack%2Fkolla~663312","triplet_id":"openstack%2Fkolla~stable%2Fstein~I000398f587c5f4d6cc8995e34e162eebc77bc3e3","project":"openstack/kolla","branch":"stable/stein","topic":"bp/centos-ceph-nautilus-stable/stein","hashtags":[],"change_id":"I000398f587c5f4d6cc8995e34e162eebc77bc3e3","subject":"Ceph: Update to Nautilus on CentOS and OracleLinux","status":"MERGED","created":"2019-06-05 12:49:57.000000000","updated":"2019-06-06 07:47:10.000000000","submitted":"2019-06-06 07:47:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"663312-1559807230712-589de8e3","meta_rev_id":"3a132a3d53277cd915c682935f44e39a17f5d5af","_number":663312,"virtual_id_number":663312,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-06-06 07:47:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-06-06 07:18:08.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2019-06-06 06:40:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2019-06-06 06:36:47.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2019-06-06 06:40:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-06 06:40:56.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2019-06-06 07:18:08.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-06-06 07:47:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"36e21941a6ac866af9f6dc47ca85602608d9129d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-05 12:49:57.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"02942ddc3959f5ff2de91e7f3d60ffae47e7343a","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-05 12:50:03.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"cf72d73728af33c65a02b602540dea565383ef69","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-05 13:31:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/12/663312/1/openstack-check/tripleo-build-containers-fedora-28-master/b46a0ae/ : SUCCESS in 38m 23s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/12/663312/1/openstack-check/tripleo-build-containers-fedora-28-stein/f6a3d0e/ : SUCCESS in 37m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f8cb526c426c0408c3f645dea5c5663ba000598f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-05 14:48:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/12/663312/1/check/kolla-build-centos-binary/d1bf3c4/ : FAILURE in 41m 37s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/12/663312/1/check/kolla-build-centos-source/5b0b030/ : SUCCESS in 36m 26s\n- kolla-ansible-centos-source http://logs.openstack.org/12/663312/1/check/kolla-ansible-centos-source/7fd8bc1/ : SUCCESS in 1h 14m 10s\n- kolla-ansible-centos-binary http://logs.openstack.org/12/663312/1/check/kolla-ansible-centos-binary/02b453a/ : FAILURE in 27m 24s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/12/663312/1/check/kolla-ansible-centos-source-upgrade/8ab0788/ : SUCCESS in 1h 21m 55s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/12/663312/1/check/tripleo-build-containers-centos-7/0c95bcc/ : SUCCESS in 50m 06s\n- kolla-ansible-centos-source-ceph http://logs.openstack.org/12/663312/1/check/kolla-ansible-centos-source-ceph/3f4a7d3/ : SUCCESS in 1h 28m 33s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/12/663312/1/check/kolla-build-oraclelinux-binary/0f52480/ : FAILURE in 1h 00m 52s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/12/663312/1/check/kolla-build-oraclelinux-source/c1497c8/ : SUCCESS in 55m 00s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/12/663312/1/check/kolla-ansible-oraclelinux-source/5d5180c/ : SUCCESS in 1h 20m 32s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/12/663312/1/check/kolla-ansible-oraclelinux-binary/66d130f/ : FAILURE in 34m 09s (non-voting)\n- kolla-ansible-oraclelinux-source-ceph http://logs.openstack.org/12/663312/1/check/kolla-ansible-oraclelinux-source-ceph/e693596/ : SUCCESS in 1h 32m 25s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/12/663312/1/check/openstack-tox-lower-constraints/3046e01/ : SUCCESS in 3m 58s\n- openstack-tox-pep8 http://logs.openstack.org/12/663312/1/check/openstack-tox-pep8/09293af/ : SUCCESS in 2m 46s\n- openstack-tox-py27 http://logs.openstack.org/12/663312/1/check/openstack-tox-py27/c47d3a0/ : SUCCESS in 2m 59s\n- openstack-tox-py36 http://logs.openstack.org/12/663312/1/check/openstack-tox-py36/bb2c75c/ : SUCCESS in 3m 20s\n- openstack-tox-py37 http://logs.openstack.org/12/663312/1/check/openstack-tox-py37/2238589/ : SUCCESS in 3m 28s\n- openstack-tox-docs http://logs.openstack.org/12/663312/1/check/openstack-tox-docs/42652b2/html/ : SUCCESS in 3m 33s\n- kolla-build-ubuntu-binary http://logs.openstack.org/12/663312/1/check/kolla-build-ubuntu-binary/a76fc01/ : SUCCESS in 35m 19s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/12/663312/1/check/kolla-build-ubuntu-source/773c80a/ : SUCCESS in 31m 14s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/12/663312/1/check/kolla-ansible-ubuntu-source/c52c6ff/ : SUCCESS in 47m 41s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/12/663312/1/check/kolla-ansible-ubuntu-binary/2a11a6a/ : SUCCESS in 48m 14s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/12/663312/1/check/kolla-ansible-ubuntu-source-upgrade/7fdae3d/ : SUCCESS in 1h 09m 50s\n- kolla-ansible-ubuntu-source-ceph http://logs.openstack.org/12/663312/1/check/kolla-ansible-ubuntu-source-ceph/1288b6d/ : SUCCESS in 1h 14m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7603d8b87bfcd1e8adbe0e6fa0e25d38b4f7e804","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-06 06:11:12.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"012725868ad8716da5cbf99e42d06f795da43c7a","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-06 06:36:47.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"bf52131b8061f41ee988d30ce7f614e582719de1","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-06-06 06:40:56.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"78170a7640c871b89934148db41b6644a9c66ec9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 06:41:09.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"e7747c8bd4bfbb70bf33c05684a11e11faeb11ce","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-06 07:18:08.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/12/663312/3/openstack-check/tripleo-build-containers-fedora-28-master/0dde111/ : SUCCESS in 35m 56s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/12/663312/3/openstack-check/tripleo-build-containers-fedora-28-stein/4bfd071/ : SUCCESS in 36m 03s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"a22924680840bb43c069ad2e47a6d5e19c0676f1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 07:47:10.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source http://logs.openstack.org/12/663312/3/gate/kolla-build-centos-source/5f547e8/ : SUCCESS in 50m 14s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/12/663312/3/gate/tripleo-build-containers-centos-7/1ab1b7d/ : SUCCESS in 43m 21s\n- kolla-build-oraclelinux-source http://logs.openstack.org/12/663312/3/gate/kolla-build-oraclelinux-source/a7caf60/ : SUCCESS in 1h 00m 38s\n- openstack-tox-lower-constraints http://logs.openstack.org/12/663312/3/gate/openstack-tox-lower-constraints/3087347/ : SUCCESS in 3m 45s\n- openstack-tox-pep8 http://logs.openstack.org/12/663312/3/gate/openstack-tox-pep8/637b79b/ : SUCCESS in 3m 13s\n- openstack-tox-py27 http://logs.openstack.org/12/663312/3/gate/openstack-tox-py27/3f7647a/ : SUCCESS in 3m 19s\n- openstack-tox-py36 http://logs.openstack.org/12/663312/3/gate/openstack-tox-py36/a4ef4ba/ : SUCCESS in 3m 08s\n- openstack-tox-py37 http://logs.openstack.org/12/663312/3/gate/openstack-tox-py37/28c07e6/ : SUCCESS in 3m 57s\n- openstack-tox-docs http://logs.openstack.org/12/663312/3/gate/openstack-tox-docs/11a5007/html/ : SUCCESS in 2m 55s\n- kolla-build-ubuntu-source http://logs.openstack.org/12/663312/3/gate/kolla-build-ubuntu-source/7c2ba68/ : SUCCESS in 50m 37s","accounts_in_message":[],"_revision_number":3},{"id":"1868035e33cecc52f46bf09896bb158d0496c1f5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 07:47:10.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"78f862b0d7e390b16a57770a6af8519c27f05631","revisions":{"e7e537d638b214975ac431fe444383987544fb13":{"kind":"REWORK","_number":1,"created":"2019-06-05 12:49:57.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/12/663312/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/12/663312/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/663312/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/663312/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/663312/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/12/663312/1"}}},"commit":{"parents":[{"commit":"49f40264f6c4b3cebbd138d97732d0c6c578408d","subject":"Merge \"Fix multiple mon deployment failures\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/49f40264f6c4b3cebbd138d97732d0c6c578408d"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-05-09 10:47:33.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-06-05 12:49:57.000000000","tz":0},"subject":"Ceph: Update to Nautilus on CentOS and OracleLinux","message":"Ceph: Update to Nautilus on CentOS and OracleLinux\n\nThe CentOS Stein OpenStack distribution depends on the Ceph Nautilus\nrelease, while Kolla currently deploys Luminous on CentOS.\n\nThis change switches CentOS and OracleLinux builds to use Ceph Nautilus.\n\nSupport for auth UID has been removed in the Nautilus release [1][2], so\nwe have removed the --set-uid argument from calls to ceph-authtool.\n\nAlso ceph-osd bootstrap mode fails when no mon config is provided, so\nwe added --no-mon-config (mon config is injected later after bootstrap).\n\nDue to ceph-nfs build issues (ceph upstream nfs-ganesha rpms relied on older\nversion of userspace-rcu, than CentOS Storage SIG packages) we need to move\nto ceph upstream rpm repos.\n\n[1] https://github.com/ceph/ceph/commit/d6def8ba1126209f8dcb40e296977dc2b09a376e\n[2] http://docs.ceph.com/docs/master/releases/nautilus/\n\nCo-Authored-By: Michal Nasiadka \u003cmichal.nasiadka@nokia.com\u003e\n\nChange-Id: I000398f587c5f4d6cc8995e34e162eebc77bc3e3\nImplements: blueprint centos-ceph-nautilus\n(cherry picked from commit e0cae62c4aab001fa7b5cc324f178793f4e88f3c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e7e537d638b214975ac431fe444383987544fb13"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e7e537d638b214975ac431fe444383987544fb13"}]},"branch":"refs/heads/stable/stein"},"b67d54d4dadb86a4380c5aff3eb056cab4a11be6":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-06-06 06:11:12.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/12/663312/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/12/663312/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/663312/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/663312/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/663312/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/12/663312/2"}}},"commit":{"parents":[{"commit":"49f40264f6c4b3cebbd138d97732d0c6c578408d","subject":"Merge \"Fix multiple mon deployment failures\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/49f40264f6c4b3cebbd138d97732d0c6c578408d"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-05-09 10:47:33.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-06-06 06:11:01.000000000","tz":0},"subject":"Ceph: Update to Nautilus on CentOS and OracleLinux","message":"Ceph: Update to Nautilus on CentOS and OracleLinux\n\nThe CentOS Stein OpenStack distribution depends on the Ceph Nautilus\nrelease, while Kolla currently deploys Luminous on CentOS.\n\nThis change switches CentOS and OracleLinux builds to use Ceph Nautilus.\n\nSupport for auth UID has been removed in the Nautilus release [1][2], so\nwe have removed the --set-uid argument from calls to ceph-authtool.\n\nAlso ceph-osd bootstrap mode fails when no mon config is provided, so\nwe added --no-mon-config (mon config is injected later after bootstrap).\n\nDue to ceph-nfs build issues (ceph upstream nfs-ganesha rpms relied on older\nversion of userspace-rcu, than CentOS Storage SIG packages) we need to move\nto ceph upstream rpm repos.\n\n[1] https://github.com/ceph/ceph/commit/d6def8ba1126209f8dcb40e296977dc2b09a376e\n[2] http://docs.ceph.com/docs/master/releases/nautilus/\n\nCo-Authored-By: Michal Nasiadka \u003cmichal.nasiadka@nokia.com\u003e\n\nChange-Id: I000398f587c5f4d6cc8995e34e162eebc77bc3e3\nImplements: blueprint centos-ceph-nautilus\n(cherry picked from commit e0cae62c4aab001fa7b5cc324f178793f4e88f3c)\n\nDepends-On: Ifff5c60f5690098619bf5a43f25ae3895a96e0d2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b67d54d4dadb86a4380c5aff3eb056cab4a11be6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b67d54d4dadb86a4380c5aff3eb056cab4a11be6"}]},"branch":"refs/heads/stable/stein"},"78f862b0d7e390b16a57770a6af8519c27f05631":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2019-06-06 06:36:47.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/12/663312/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/12/663312/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/663312/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/663312/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/663312/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/12/663312/3"}}},"commit":{"parents":[{"commit":"49f40264f6c4b3cebbd138d97732d0c6c578408d","subject":"Merge \"Fix multiple mon deployment failures\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/49f40264f6c4b3cebbd138d97732d0c6c578408d"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-05-09 10:47:33.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-06-06 06:36:44.000000000","tz":0},"subject":"Ceph: Update to Nautilus on CentOS and OracleLinux","message":"Ceph: Update to Nautilus on CentOS and OracleLinux\n\nThe CentOS Stein OpenStack distribution depends on the Ceph Nautilus\nrelease, while Kolla currently deploys Luminous on CentOS.\n\nThis change switches CentOS and OracleLinux builds to use Ceph Nautilus.\n\nSupport for auth UID has been removed in the Nautilus release [1][2], so\nwe have removed the --set-uid argument from calls to ceph-authtool.\n\nAlso ceph-osd bootstrap mode fails when no mon config is provided, so\nwe added --no-mon-config (mon config is injected later after bootstrap).\n\nDue to ceph-nfs build issues (ceph upstream nfs-ganesha rpms relied on older\nversion of userspace-rcu, than CentOS Storage SIG packages) we need to move\nto ceph upstream rpm repos.\n\n[1] https://github.com/ceph/ceph/commit/d6def8ba1126209f8dcb40e296977dc2b09a376e\n[2] http://docs.ceph.com/docs/master/releases/nautilus/\n\nCo-Authored-By: Michal Nasiadka \u003cmichal.nasiadka@nokia.com\u003e\n\nChange-Id: I000398f587c5f4d6cc8995e34e162eebc77bc3e3\nImplements: blueprint centos-ceph-nautilus\n(cherry picked from commit e0cae62c4aab001fa7b5cc324f178793f4e88f3c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/78f862b0d7e390b16a57770a6af8519c27f05631"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/78f862b0d7e390b16a57770a6af8519c27f05631"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
