)]}'
{"id":"openstack%2Fkolla~663375","triplet_id":"openstack%2Fkolla~master~I0bff21384d88ea678608392de2db1ba418c96665","project":"openstack/kolla","branch":"master","topic":"bp/debian-ubuntu-python3","hashtags":[],"change_id":"I0bff21384d88ea678608392de2db1ba418c96665","subject":"Debian/Ubuntu: more python3 packages for binary","status":"MERGED","created":"2019-06-05 16:23:11.000000000","updated":"2019-06-07 17:59:49.000000000","submitted":"2019-06-07 08:29:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"663375-1559896149019-20b3d2df","meta_rev_id":"6889e969c959beddf8140c05937449dcf9158c65","_number":663375,"virtual_id_number":663375,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-06-07 08:29:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-06-06 11:44:47.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"recommended":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"all":[{"value":1,"date":"2019-06-07 04:19:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"value":2,"date":"2019-06-07 07:16:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},{"value":2,"date":"2019-06-06 12:53:21.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"value":1,"date":"2019-06-07 07:16:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2019-06-06 13:35:17.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-06 09:38:02.000000000","updated_by":{"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},"reviewer":{"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},"state":"REVIEWER"},{"updated":"2019-06-06 11:44:47.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-06-06 12:53:21.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-06-07 04:19:27.000000000","updated_by":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"reviewer":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"state":"REVIEWER"},{"updated":"2019-06-07 07:16:03.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2019-06-07 08:29:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7828aa873befa9cd70dea989e958f180679ef0ea","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-05 16:23:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"aa1b931cfdcb0e44af9eae741bb1c7891322a834","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-05 17:02:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/75/663375/1/openstack-check/tripleo-build-containers-fedora-28-master/440d54d/ : SUCCESS in 31m 28s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/75/663375/1/openstack-check/tripleo-build-containers-fedora-28-stein/e7db920/ : SUCCESS in 34m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4789e6510be503ba1c3c7b1fad714becb283bbab","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-05 18:32:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/75/663375/1/check/kolla-build-centos-binary/f4014f4/ : SUCCESS in 29m 22s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/75/663375/1/check/kolla-build-centos-source/2be03de/ : SUCCESS in 54m 26s\n- kolla-ansible-centos-source http://logs.openstack.org/75/663375/1/check/kolla-ansible-centos-source/38e8ca6/ : SUCCESS in 1h 24m 09s\n- kolla-ansible-centos-binary http://logs.openstack.org/75/663375/1/check/kolla-ansible-centos-binary/7b856fd/ : SUCCESS in 1h 08m 08s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/75/663375/1/check/kolla-ansible-centos-source-upgrade/83be0a1/ : SUCCESS in 1h 16m 54s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/75/663375/1/check/tripleo-build-containers-centos-7/372fa50/ : SUCCESS in 49m 35s\n- kolla-ansible-centos-source-ceph http://logs.openstack.org/75/663375/1/check/kolla-ansible-centos-source-ceph/0c009f9/ : SUCCESS in 1h 30m 07s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/75/663375/1/check/kolla-build-oraclelinux-binary/f2d3ef2/ : SUCCESS in 1h 04m 06s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/75/663375/1/check/kolla-build-oraclelinux-source/008513a/ : SUCCESS in 51m 40s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/75/663375/1/check/kolla-ansible-oraclelinux-source/e9c071d/ : SUCCESS in 1h 10m 39s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/75/663375/1/check/kolla-ansible-oraclelinux-binary/7228b22/ : SUCCESS in 1h 10m 32s (non-voting)\n- kolla-ansible-oraclelinux-source-ceph http://logs.openstack.org/75/663375/1/check/kolla-ansible-oraclelinux-source-ceph/9fa3f97/ : SUCCESS in 1h 29m 56s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/75/663375/1/check/openstack-tox-lower-constraints/5000a15/ : SUCCESS in 4m 09s\n- openstack-tox-pep8 http://logs.openstack.org/75/663375/1/check/openstack-tox-pep8/7e38e3f/ : SUCCESS in 2m 47s\n- openstack-tox-py27 http://logs.openstack.org/75/663375/1/check/openstack-tox-py27/6385f07/ : SUCCESS in 3m 06s\n- openstack-tox-py36 http://logs.openstack.org/75/663375/1/check/openstack-tox-py36/55d8c17/ : SUCCESS in 3m 07s\n- openstack-tox-py37 http://logs.openstack.org/75/663375/1/check/openstack-tox-py37/3fd1cf2/ : SUCCESS in 4m 12s\n- openstack-tox-docs http://logs.openstack.org/75/663375/1/check/openstack-tox-docs/00a23cd/html/ : SUCCESS in 3m 27s\n- kolla-build-ubuntu-binary http://logs.openstack.org/75/663375/1/check/kolla-build-ubuntu-binary/6d9e715/ : FAILURE in 30m 40s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/75/663375/1/check/kolla-build-ubuntu-source/a17edd6/ : SUCCESS in 34m 43s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/75/663375/1/check/kolla-ansible-ubuntu-source/fc77093/ : SUCCESS in 1h 08m 39s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/75/663375/1/check/kolla-ansible-ubuntu-binary/803fea9/ : SUCCESS in 45m 42s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/75/663375/1/check/kolla-ansible-ubuntu-source-upgrade/9b74272/ : SUCCESS in 1h 06m 57s\n- kolla-ansible-ubuntu-source-ceph http://logs.openstack.org/75/663375/1/check/kolla-ansible-ubuntu-source-ceph/cc89a87/ : SUCCESS in 1h 15m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2ebdf616477c6e6efd794eecb85dd2b50d7b0a18","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-05 19:15:52.000000000","message":"Patch Set 1:\n\n(2 comments)\n\nthanks for splitting it\n\nfound few more places","accounts_in_message":[],"_revision_number":1},{"id":"db78f15b80e7c57e7ceadeb7d726eb3680f96a2c","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-06 05:55:11.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6e3049e8d2492b29956c343c9ec3251715d8497d","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-06 06:37:07.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/75/663375/2/openstack-check/tripleo-build-containers-fedora-28-master/781c5ac/ : SUCCESS in 41m 12s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/75/663375/2/openstack-check/tripleo-build-containers-fedora-28-stein/3878506/ : SUCCESS in 39m 02s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0f881c05b7f92317d0712aa70603dc253a2a614d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 08:06:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/75/663375/2/check/kolla-build-centos-binary/b255b82/ : FAILURE in 3m 47s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/75/663375/2/check/kolla-build-centos-source/e760a4e/ : SUCCESS in 57m 10s\n- kolla-ansible-centos-source http://logs.openstack.org/75/663375/2/check/kolla-ansible-centos-source/74febcf/ : SUCCESS in 1h 18m 29s\n- kolla-ansible-centos-binary http://logs.openstack.org/75/663375/2/check/kolla-ansible-centos-binary/806a1d0/ : SUCCESS in 1h 05m 12s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/75/663375/2/check/kolla-ansible-centos-source-upgrade/2eee0a1/ : SUCCESS in 1h 21m 54s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/75/663375/2/check/tripleo-build-containers-centos-7/60e599f/ : SUCCESS in 41m 22s\n- kolla-ansible-centos-source-ceph http://logs.openstack.org/75/663375/2/check/kolla-ansible-centos-source-ceph/07d2a56/ : SUCCESS in 1h 41m 34s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/75/663375/2/check/kolla-build-oraclelinux-binary/b1096f9/ : SUCCESS in 33m 24s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/75/663375/2/check/kolla-build-oraclelinux-source/2aef5aa/ : SUCCESS in 50m 34s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/75/663375/2/check/kolla-ansible-oraclelinux-source/4d89f46/ : SUCCESS in 1h 12m 07s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/75/663375/2/check/kolla-ansible-oraclelinux-binary/3f87641/ : SUCCESS in 1h 10m 21s (non-voting)\n- kolla-ansible-oraclelinux-source-ceph http://logs.openstack.org/75/663375/2/check/kolla-ansible-oraclelinux-source-ceph/acfc18b/ : SUCCESS in 1h 33m 02s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/75/663375/2/check/openstack-tox-lower-constraints/8aa01f4/ : SUCCESS in 3m 43s\n- openstack-tox-pep8 http://logs.openstack.org/75/663375/2/check/openstack-tox-pep8/56b571a/ : SUCCESS in 2m 42s\n- openstack-tox-py27 http://logs.openstack.org/75/663375/2/check/openstack-tox-py27/c6d69df/ : SUCCESS in 3m 07s\n- openstack-tox-py36 http://logs.openstack.org/75/663375/2/check/openstack-tox-py36/ae4469e/ : SUCCESS in 3m 01s\n- openstack-tox-py37 http://logs.openstack.org/75/663375/2/check/openstack-tox-py37/91dcc76/ : SUCCESS in 2m 53s\n- openstack-tox-docs http://logs.openstack.org/75/663375/2/check/openstack-tox-docs/22497a0/html/ : SUCCESS in 2m 57s\n- kolla-build-ubuntu-binary http://logs.openstack.org/75/663375/2/check/kolla-build-ubuntu-binary/82dca22/ : SUCCESS in 42m 09s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/75/663375/2/check/kolla-build-ubuntu-source/d7c6037/ : SUCCESS in 44m 43s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/75/663375/2/check/kolla-ansible-ubuntu-source/c248891/ : SUCCESS in 53m 38s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/75/663375/2/check/kolla-ansible-ubuntu-binary/ba85c2d/ : SUCCESS in 44m 39s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/75/663375/2/check/kolla-ansible-ubuntu-source-upgrade/6d7cb18/ : SUCCESS in 1h 19m 49s\n- kolla-ansible-ubuntu-source-ceph http://logs.openstack.org/75/663375/2/check/kolla-ansible-ubuntu-source-ceph/6f16507/ : SUCCESS in 1h 07m 51s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"45a3da4f6fb2f17e7e6d1e744f2a5fe8c12bc076","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-06 08:11:15.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"18f975ba34ecc15b79280ff30a2d3d5fc4bf2e0e","author":{"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},"date":"2019-06-06 09:38:02.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"22cc165581ce2b1c9e749f37825b2619ff261ccb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 09:38:15.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"163b641085953b8eafe684ac2e2001a6ec2c3475","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-06 09:54:32.000000000","message":"Patch Set 2: Workflow-1\n\nHmm, still think we have a few to catch.","accounts_in_message":[],"_revision_number":2},{"id":"624c5f6d7a2fcd30d84cbbee51a0ca91db3d9e4e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-06 09:54:39.000000000","message":"Removed Workflow+1 by Surya Prakash (spsurya) \u003csingh.surya64mnnit@gmail.com\u003e\n","accounts_in_message":[],"_revision_number":2},{"id":"8fb2815ccdf84d2e12191665aa2165236eeeff9a","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-06 10:35:13.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"826419c5260a1f42d029aacc8d7095ccff458c03","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-06 10:46:15.000000000","message":"Patch Set 3:\n\nI think I found all relevant occurences of python-*. Tested via:\n\nkolla-build -t binary -b ubuntu --template-only\nfind /tmp/kolla-2019-06-06_11-24-46_OucDC5/docker -name Dockerfile.j2 -exec rm {} +\ngrep -rn python- /tmp/kolla-2019-06-06_11-24-46_OucDC5/docker\n\nThese remain, and I think they\u0027re all valid:\n\ndocker/helm-repository/Dockerfile:18:RUN apt-get update \u0026\u0026 apt-get -y install --no-install-recommends git jq python-pip \u0026\u0026 apt-get clean \u0026\u0026 rm -rf /var/lib/apt/lists/*\ndocker/rally/Dockerfile:21:RUN apt-get update \u0026\u0026 apt-get -y install --no-install-recommends python-os-testr rally \u0026\u0026 apt-get clean \u0026\u0026 rm -rf /var/lib/apt/lists/*\ndocker/neutron/neutron-base/Dockerfile:39:             # FIXED(jeffrey4l): UCA rocky release is not include python-vmware-nsx\ndocker/neutron/neutron-base/Dockerfile:40:             # 13.0.0 package now.  add \u0027python-vmware-nsx\u0027 into\ndocker/horizon/Dockerfile:19:\t#NOTE(hrw): check for python-murano-dashboard python-neutron-lbaas-dashboard\ndocker/swift/swift-proxy-server/Dockerfile:13:RUN apt-get update \u0026\u0026 apt-get -y install --no-install-recommends swift-proxy python-ceilometermiddleware \u0026\u0026 apt-get clean \u0026\u0026 rm -rf /var/lib/apt/lists/*\ndocker/ironic/ironic-conductor/Dockerfile:10:        # TODO(jeffrey4l): no python-dracclient, python-oneviewclient, UcsSdk,\ndocker/ironic/ironic-conductor/Dockerfile:11:        # ImcSdk package, python-sushy, ironic-staging-drivers\ndocker/kolla-toolbox/Dockerfile:22:RUN apt-get update \u0026\u0026 apt-get -y install --no-install-recommends build-essential ca-certificates crudini gdisk git jq libffi-dev libmariadb-dev-compat libssl-dev libxslt1-dev mariadb-client python-dev \u0026\u0026 apt-get clean \u0026\u0026 rm -rf /var/lib/apt/lists/*\ndocker/kolla-toolbox/Dockerfile:44:RUN pip --no-cache-dir install --upgrade ansible\u003d\u003d2.2.0.0 \"cmd2\u003c0.9.0\" influxdb mysqlclient os-client-config\u003d\u003d1.29.0 pbr\u003d\u003d4.0.0 pymongo python-openstackclient\u003d\u003d3.14.0 pytz pyudev shade\u003d\u003d1.27.1 \\\ndocker/openvswitch/openvswitch-base/Dockerfile:14:RUN apt-get update \u0026\u0026 apt-get -y install --no-install-recommends openvswitch-switch python-openvswitch python-netifaces tcpdump \u0026\u0026 apt-get clean \u0026\u0026 rm -rf /var/lib/apt/lists/*\ndocker/base/crane.repo:2:# NOTE python-crane package is not signed","accounts_in_message":[],"_revision_number":3},{"id":"5a050064f8f23cfdbb00fe5db9cf24dade8708f4","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-06 11:02:31.000000000","message":"Patch Set 3:\n\n\u003e These remain, and I think they\u0027re all valid:\n \u003e \n \u003e docker/helm-repository/Dockerfile:18:RUN apt-get update \u0026\u0026 apt-get\n \u003e -y install --no-install-recommends git jq python-pip \u0026\u0026 apt-get\n \u003e clean \u0026\u0026 rm -rf /var/lib/apt/lists/*\n\nThis one is weird. We install pip from package and then bootstrap it anyway. And then fetch binary...\n\n\n \u003e docker/rally/Dockerfile:21:RUN apt-get update \u0026\u0026 apt-get -y install\n \u003e --no-install-recommends python-os-testr rally \u0026\u0026 apt-get clean \u0026\u0026\n \u003e rm -rf /var/lib/apt/lists/*\n\nRally is py2 in UCA.\n\n \u003e docker/swift/swift-proxy-server/Dockerfile:13:RUN apt-get update \u0026\u0026\n \u003e apt-get -y install --no-install-recommends swift-proxy\n \u003e python-ceilometermiddleware \u0026\u0026 apt-get clean \u0026\u0026 rm -rf\n \u003e /var/lib/apt/lists/*\n\npy2 in UCA\n\n\n \u003e docker/kolla-toolbox/Dockerfile:22:RUN apt-get update \u0026\u0026 apt-get -y\n \u003e install --no-install-recommends build-essential ca-certificates\n \u003e crudini gdisk git jq libffi-dev libmariadb-dev-compat libssl-dev\n \u003e libxslt1-dev mariadb-client python-dev \u0026\u0026 apt-get clean \u0026\u0026 rm -rf\n \u003e /var/lib/apt/lists/*\n\nbase has python so python-dev is correct. Source patch changes it to python3.\n\n\n \u003e docker/openvswitch/openvswitch-base/Dockerfile:14:RUN apt-get\n \u003e update \u0026\u0026 apt-get -y install --no-install-recommends\n \u003e openvswitch-switch python-openvswitch python-netifaces tcpdump \u0026\u0026\n \u003e apt-get clean \u0026\u0026 rm -rf /var/lib/apt/lists/*\n\nopenvswitch-switch is py3 so here we need to change","accounts_in_message":[],"_revision_number":3},{"id":"0d61aa6281a86e43877f349936cc765e89e5c15f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-06 11:07:02.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"17353e0800eae1de165ff4ad4f2738a52f431153","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-06 11:07:15.000000000","message":"Patch Set 4:\n\n\u003e \u003e These remain, and I think they\u0027re all valid:\n \u003e \u003e\n \u003e \u003e docker/helm-repository/Dockerfile:18:RUN apt-get update \u0026\u0026\n \u003e apt-get\n \u003e \u003e -y install --no-install-recommends git jq python-pip \u0026\u0026 apt-get\n \u003e \u003e clean \u0026\u0026 rm -rf /var/lib/apt/lists/*\n \u003e \n \u003e This one is weird. We install pip from package and then bootstrap\n \u003e it anyway. And then fetch binary...\n \u003e \n \u003e \n \u003e \u003e docker/rally/Dockerfile:21:RUN apt-get update \u0026\u0026 apt-get -y\n \u003e install\n \u003e \u003e --no-install-recommends python-os-testr rally \u0026\u0026 apt-get clean \u0026\u0026\n \u003e \u003e rm -rf /var/lib/apt/lists/*\n \u003e \n \u003e Rally is py2 in UCA.\n \u003e \n \u003e \u003e docker/swift/swift-proxy-server/Dockerfile:13:RUN apt-get update\n \u003e \u0026\u0026\n \u003e \u003e apt-get -y install --no-install-recommends swift-proxy\n \u003e \u003e python-ceilometermiddleware \u0026\u0026 apt-get clean \u0026\u0026 rm -rf\n \u003e \u003e /var/lib/apt/lists/*\n \u003e \n \u003e py2 in UCA\n \u003e \n \u003e \n \u003e \u003e docker/kolla-toolbox/Dockerfile:22:RUN apt-get update \u0026\u0026 apt-get\n \u003e -y\n \u003e \u003e install --no-install-recommends build-essential ca-certificates\n \u003e \u003e crudini gdisk git jq libffi-dev libmariadb-dev-compat libssl-dev\n \u003e \u003e libxslt1-dev mariadb-client python-dev \u0026\u0026 apt-get clean \u0026\u0026 rm -rf\n \u003e \u003e /var/lib/apt/lists/*\n \u003e \n \u003e base has python so python-dev is correct. Source patch changes it\n \u003e to python3.\n \u003e \n \u003e \n \u003e \u003e docker/openvswitch/openvswitch-base/Dockerfile:14:RUN apt-get\n \u003e \u003e update \u0026\u0026 apt-get -y install --no-install-recommends\n \u003e \u003e openvswitch-switch python-openvswitch python-netifaces tcpdump \u0026\u0026\n \u003e \u003e apt-get clean \u0026\u0026 rm -rf /var/lib/apt/lists/*\n \u003e \n \u003e openvswitch-switch is py3 so here we need to change\n\nDone, good catch.","accounts_in_message":[],"_revision_number":4},{"id":"3751d8a5a90b667bce945ff8b278ce5faf7d0e75","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-06 11:44:47.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/75/663375/4/openstack-check/tripleo-build-containers-fedora-28-master/b1ce0cb/ : SUCCESS in 36m 23s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/75/663375/4/openstack-check/tripleo-build-containers-fedora-28-stein/89c9606/ : SUCCESS in 32m 52s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"a0285b49a7cd142423b43cb2bc893790defa4cfc","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-06 12:53:21.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"e3cc43e600fd8286a7fd838474f65fa8c556b400","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 12:55:47.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/75/663375/4/check/kolla-build-centos-binary/30bf9c3/ : SUCCESS in 45m 24s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/75/663375/4/check/kolla-build-centos-source/21ab81a/ : SUCCESS in 44m 55s\n- kolla-ansible-centos-source http://logs.openstack.org/75/663375/4/check/kolla-ansible-centos-source/0ba566a/ : SUCCESS in 1h 09m 41s\n- kolla-ansible-centos-binary http://logs.openstack.org/75/663375/4/check/kolla-ansible-centos-binary/8fe4829/ : SUCCESS in 1h 02m 26s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/75/663375/4/check/kolla-ansible-centos-source-upgrade/5c07853/ : SUCCESS in 1h 17m 22s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/75/663375/4/check/tripleo-build-containers-centos-7/cc9820a/ : SUCCESS in 39m 57s\n- kolla-ansible-centos-source-ceph http://logs.openstack.org/75/663375/4/check/kolla-ansible-centos-source-ceph/9fde4e9/ : SUCCESS in 1h 42m 47s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/75/663375/4/check/kolla-build-oraclelinux-binary/2ea9e6c/ : SUCCESS in 58m 01s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/75/663375/4/check/kolla-build-oraclelinux-source/50e8db0/ : SUCCESS in 51m 31s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/75/663375/4/check/kolla-ansible-oraclelinux-source/fc594c6/ : SUCCESS in 1h 09m 11s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/75/663375/4/check/kolla-ansible-oraclelinux-binary/6204d37/ : SUCCESS in 1h 10m 50s (non-voting)\n- kolla-ansible-oraclelinux-source-ceph http://logs.openstack.org/75/663375/4/check/kolla-ansible-oraclelinux-source-ceph/ba08d78/ : SUCCESS in 1h 32m 46s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/75/663375/4/check/openstack-tox-lower-constraints/f53953b/ : SUCCESS in 4m 00s\n- openstack-tox-pep8 http://logs.openstack.org/75/663375/4/check/openstack-tox-pep8/750812c/ : SUCCESS in 3m 01s\n- openstack-tox-py27 http://logs.openstack.org/75/663375/4/check/openstack-tox-py27/d740943/ : SUCCESS in 2m 46s\n- openstack-tox-py36 http://logs.openstack.org/75/663375/4/check/openstack-tox-py36/8d874e6/ : SUCCESS in 3m 09s\n- openstack-tox-py37 http://logs.openstack.org/75/663375/4/check/openstack-tox-py37/3df0d30/ : SUCCESS in 3m 06s\n- openstack-tox-docs http://logs.openstack.org/75/663375/4/check/openstack-tox-docs/9b65009/html/ : SUCCESS in 3m 17s\n- kolla-build-ubuntu-binary http://logs.openstack.org/75/663375/4/check/kolla-build-ubuntu-binary/9b49f95/ : SUCCESS in 32m 16s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/75/663375/4/check/kolla-build-ubuntu-source/9c222d7/ : SUCCESS in 32m 32s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/75/663375/4/check/kolla-ansible-ubuntu-source/028ccd8/ : SUCCESS in 51m 22s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/75/663375/4/check/kolla-ansible-ubuntu-binary/e4c57d5/ : SUCCESS in 44m 55s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/75/663375/4/check/kolla-ansible-ubuntu-source-upgrade/b76c452/ : SUCCESS in 1h 03m 55s\n- kolla-ansible-ubuntu-source-ceph http://logs.openstack.org/75/663375/4/check/kolla-ansible-ubuntu-source-ceph/86a03d9/ : SUCCESS in 1h 19m 25s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"661f457703b86b36ec2e39330c972870f7d712c7","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-06 13:35:17.000000000","message":"Patch Set 4: Review-Priority+1","accounts_in_message":[],"_revision_number":4},{"id":"2a6b0dc3178a3cc0846b2309f07e4feb1156516f","author":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"date":"2019-06-07 04:19:27.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"74267c80189b6e1d28e1b4c103ffa2ceb12ae881","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-06-07 07:16:03.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"de1e2bdcddc7b5073e4ea332b89c3b83e7cbed7b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 07:16:16.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"81590285e1c05096e1feae27a989a8f3de2fb843","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 08:29:08.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source http://logs.openstack.org/75/663375/4/gate/kolla-build-centos-source/953f743/ : SUCCESS in 1h 00m 52s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/75/663375/4/gate/tripleo-build-containers-centos-7/717c367/ : SUCCESS in 40m 27s\n- kolla-build-oraclelinux-source http://logs.openstack.org/75/663375/4/gate/kolla-build-oraclelinux-source/b93b77f/ : SUCCESS in 48m 22s\n- openstack-tox-lower-constraints http://logs.openstack.org/75/663375/4/gate/openstack-tox-lower-constraints/8d1e5f5/ : SUCCESS in 4m 06s\n- openstack-tox-pep8 http://logs.openstack.org/75/663375/4/gate/openstack-tox-pep8/ec113bb/ : SUCCESS in 3m 15s\n- openstack-tox-py27 http://logs.openstack.org/75/663375/4/gate/openstack-tox-py27/1bc95f9/ : SUCCESS in 3m 30s\n- openstack-tox-py36 http://logs.openstack.org/75/663375/4/gate/openstack-tox-py36/82e6a72/ : SUCCESS in 4m 14s\n- openstack-tox-py37 http://logs.openstack.org/75/663375/4/gate/openstack-tox-py37/021e867/ : SUCCESS in 3m 34s\n- openstack-tox-docs http://logs.openstack.org/75/663375/4/gate/openstack-tox-docs/c72c395/html/ : SUCCESS in 2m 46s\n- kolla-build-ubuntu-source http://logs.openstack.org/75/663375/4/gate/kolla-build-ubuntu-source/e8dd91e/ : SUCCESS in 33m 58s","accounts_in_message":[],"_revision_number":4},{"id":"d08a1c0f0c766c1531093343f841f9de36344ed5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 08:29:09.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"565c9c4b2b5438c95ae77ebfcca64c1c7b40ef43","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-07 17:59:49.000000000","message":"Patch Set 4: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit b71f2be433568e6c71ea1d54a92dfa6a449dead6","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"aa59f0db6f88ed0a03169b473fdf6d541d34df9f","revisions":{"f4a1388d223b1e905af84deacc36cbeee9d9d5dd":{"kind":"REWORK","_number":1,"created":"2019-06-05 16:23:11.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/75/663375/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/75/663375/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/75/663375/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/75/663375/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/75/663375/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/75/663375/1"}}},"commit":{"parents":[{"commit":"4e083196d14877de458ae16129331c37fe1f6f21","subject":"Merge \"Use the default policy.json for Qinling\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4e083196d14877de458ae16129331c37fe1f6f21"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-06-05 16:16:49.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-06-05 16:22:17.000000000","tz":60},"subject":"Debian/Ubuntu: more python3 packages for binary","message":"Debian/Ubuntu: more python3 packages for binary\n\nbackport: Stein\n\nDuring the switch to Stein UCA, we did not switch all packages to python\n3 for Debian/Ubuntu binary images. This change switches some more of\nthose packages.\n\nChange-Id: I0bff21384d88ea678608392de2db1ba418c96665\nCo-Authored-By: Marcin Juszkiewicz \u003cmarcin.juszkiewicz@linaro.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f4a1388d223b1e905af84deacc36cbeee9d9d5dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f4a1388d223b1e905af84deacc36cbeee9d9d5dd"}]},"branch":"refs/heads/master"},"e52938bdb2716599701fd98fc299ad6bf3dbc8c3":{"kind":"REWORK","_number":2,"created":"2019-06-06 05:55:11.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/75/663375/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/75/663375/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/75/663375/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/75/663375/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/75/663375/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/75/663375/2"}}},"commit":{"parents":[{"commit":"4e083196d14877de458ae16129331c37fe1f6f21","subject":"Merge \"Use the default policy.json for Qinling\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4e083196d14877de458ae16129331c37fe1f6f21"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-06-05 16:16:49.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-06-06 05:54:16.000000000","tz":120},"subject":"Debian/Ubuntu: more python3 packages for binary","message":"Debian/Ubuntu: more python3 packages for binary\n\nbackport: Stein\n\nDuring the switch to Stein UCA, we did not switch all packages to python\n3 for Debian/Ubuntu binary images. This change switches some more of\nthose packages.\n\nChange-Id: I0bff21384d88ea678608392de2db1ba418c96665\nCo-Authored-By: Marcin Juszkiewicz \u003cmarcin.juszkiewicz@linaro.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e52938bdb2716599701fd98fc299ad6bf3dbc8c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e52938bdb2716599701fd98fc299ad6bf3dbc8c3"}]},"branch":"refs/heads/master"},"3493796abb83aa79e4caf741f70bb093efaa952c":{"kind":"REWORK","_number":3,"created":"2019-06-06 10:35:13.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/75/663375/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/75/663375/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/75/663375/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/75/663375/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/75/663375/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/75/663375/3"}}},"commit":{"parents":[{"commit":"4e083196d14877de458ae16129331c37fe1f6f21","subject":"Merge \"Use the default policy.json for Qinling\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4e083196d14877de458ae16129331c37fe1f6f21"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-06-05 16:16:49.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-06-06 10:34:45.000000000","tz":60},"subject":"Debian/Ubuntu: more python3 packages for binary","message":"Debian/Ubuntu: more python3 packages for binary\n\nbackport: Stein\n\nDuring the switch to Stein UCA, we did not switch all packages to python\n3 for Debian/Ubuntu binary images. This change switches some more of\nthose packages.\n\nChange-Id: I0bff21384d88ea678608392de2db1ba418c96665\nCo-Authored-By: Marcin Juszkiewicz \u003cmarcin.juszkiewicz@linaro.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3493796abb83aa79e4caf741f70bb093efaa952c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3493796abb83aa79e4caf741f70bb093efaa952c"}]},"branch":"refs/heads/master"},"aa59f0db6f88ed0a03169b473fdf6d541d34df9f":{"kind":"REWORK","_number":4,"created":"2019-06-06 11:07:02.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/75/663375/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/75/663375/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/75/663375/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/75/663375/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/75/663375/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/75/663375/4"}}},"commit":{"parents":[{"commit":"4e083196d14877de458ae16129331c37fe1f6f21","subject":"Merge \"Use the default policy.json for Qinling\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4e083196d14877de458ae16129331c37fe1f6f21"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-06-05 16:16:49.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-06-06 11:06:58.000000000","tz":0},"subject":"Debian/Ubuntu: more python3 packages for binary","message":"Debian/Ubuntu: more python3 packages for binary\n\nbackport: Stein\n\nDuring the switch to Stein UCA, we did not switch all packages to python\n3 for Debian/Ubuntu binary images. This change switches some more of\nthose packages.\n\nChange-Id: I0bff21384d88ea678608392de2db1ba418c96665\nCo-Authored-By: Marcin Juszkiewicz \u003cmarcin.juszkiewicz@linaro.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/aa59f0db6f88ed0a03169b473fdf6d541d34df9f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/aa59f0db6f88ed0a03169b473fdf6d541d34df9f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
