)]}'
{"id":"openstack%2Fkolla~663409","triplet_id":"openstack%2Fkolla~stable%2Fstein~Ia6bb2c1de260e099cf44f47b190be032007ac8b1","project":"openstack/kolla","branch":"stable/stein","topic":"bp/python3-support-stable/stein","hashtags":[],"change_id":"Ia6bb2c1de260e099cf44f47b190be032007ac8b1","subject":"WIP: Fedora support for python3 package testing","status":"ABANDONED","created":"2019-06-05 19:39:16.000000000","updated":"2019-11-04 12:28:28.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"92cdcb44127a1fb158000660dbe4d0a921744f9f","_number":663409,"virtual_id_number":663409,"owner":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2019-06-18 22:32:26.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2019-06-18 19:20:26.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":-1,"date":"2019-06-18 18:36:02.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-05 19:39:16.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-06-08 18:55:03.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-06-18 19:20:26.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-06-18 22:32:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7162fe6c76d434239254e036d08ff55bafe0851e","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2019-06-05 19:39:16.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"470af35a1b5f502876f0ad9ed2433784b99e679a","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-05 20:16:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/09/663409/1/openstack-check/tripleo-build-containers-fedora-28-master/2a37421/ : SUCCESS in 36m 12s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/09/663409/1/openstack-check/tripleo-build-containers-fedora-28-stein/e9f6d53/ : SUCCESS in 35m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"03c794ffb9b0c10509357c135722a31c5c88acbd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-05 21:21:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/09/663409/1/check/kolla-build-centos-binary/c9a8c27/ : FAILURE in 45m 44s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/09/663409/1/check/kolla-build-centos-source/ec139e6/ : SUCCESS in 44m 26s\n- kolla-ansible-centos-source http://logs.openstack.org/09/663409/1/check/kolla-ansible-centos-source/87b9d6d/ : SUCCESS in 1h 17m 15s\n- kolla-ansible-centos-binary http://logs.openstack.org/09/663409/1/check/kolla-ansible-centos-binary/90ebcd7/ : FAILURE in 27m 03s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/09/663409/1/check/kolla-ansible-centos-source-upgrade/793f347/ : SUCCESS in 1h 31m 41s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/09/663409/1/check/tripleo-build-containers-centos-7/8631a61/ : SUCCESS in 42m 32s\n- kolla-ansible-bifrost-centos-source http://logs.openstack.org/09/663409/1/check/kolla-ansible-bifrost-centos-source/955df54/ : SUCCESS in 50m 18s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/09/663409/1/check/kolla-build-oraclelinux-binary/0161581/ : FAILURE in 49m 35s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/09/663409/1/check/kolla-build-oraclelinux-source/fbce53d/ : SUCCESS in 36m 39s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/09/663409/1/check/kolla-ansible-oraclelinux-source/e2b3cf5/ : SUCCESS in 1h 08m 23s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/09/663409/1/check/kolla-ansible-oraclelinux-binary/9376344/ : FAILURE in 39m 50s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/09/663409/1/check/openstack-tox-lower-constraints/c8a2fa5/ : SUCCESS in 3m 32s\n- openstack-tox-pep8 http://logs.openstack.org/09/663409/1/check/openstack-tox-pep8/b1d78e1/ : SUCCESS in 2m 51s\n- openstack-tox-py27 http://logs.openstack.org/09/663409/1/check/openstack-tox-py27/a06845d/ : SUCCESS in 4m 18s\n- openstack-tox-py36 http://logs.openstack.org/09/663409/1/check/openstack-tox-py36/23cc27b/ : SUCCESS in 3m 25s\n- openstack-tox-py37 http://logs.openstack.org/09/663409/1/check/openstack-tox-py37/51022b9/ : SUCCESS in 3m 30s\n- openstack-tox-docs http://logs.openstack.org/09/663409/1/check/openstack-tox-docs/362e509/html/ : SUCCESS in 2m 37s\n- kolla-build-ubuntu-binary http://logs.openstack.org/09/663409/1/check/kolla-build-ubuntu-binary/77182e6/ : SUCCESS in 1h 05m 11s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/09/663409/1/check/kolla-build-ubuntu-source/9012a61/ : SUCCESS in 1h 08m 33s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/09/663409/1/check/kolla-ansible-ubuntu-source/f3913b0/ : SUCCESS in 53m 15s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/09/663409/1/check/kolla-ansible-ubuntu-binary/33efb83/ : SUCCESS in 47m 13s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/09/663409/1/check/kolla-ansible-ubuntu-source-upgrade/4fdac18/ : SUCCESS in 58m 18s","accounts_in_message":[],"_revision_number":1},{"id":"417442cfd69f5bbe59a088f3709fc806a360c01d","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-08 18:55:03.000000000","message":"Patch Set 1: Code-Review-1 Workflow-1\n\nbackporting for not-merged-into-master patch?","accounts_in_message":[],"_revision_number":1},{"id":"f928ceff732d51ed64d79709a4507d0b8d37e81f","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2019-06-18 18:36:02.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"e3155d8fc4808c4780ce6e8e988174493e070bf3","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-18 19:20:26.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/09/663409/2/openstack-check/tripleo-build-containers-fedora-28-master/55c65e4/ : SUCCESS in 32m 11s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/09/663409/2/openstack-check/tripleo-build-containers-fedora-28-stein/77e308e/ : FAILURE in 5m 14s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0e9da28f66b69ffec09be4f174937e1d4679c89f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 22:32:26.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/09/663409/2/check/kolla-build-centos-binary/8c75658/ : SUCCESS in 29m 54s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/09/663409/2/check/kolla-build-centos-source/f40059a/ : SUCCESS in 50m 21s\n- kolla-ansible-centos-source http://logs.openstack.org/09/663409/2/check/kolla-ansible-centos-source/536322e/ : SUCCESS in 1h 14m 20s\n- kolla-ansible-centos-binary http://logs.openstack.org/09/663409/2/check/kolla-ansible-centos-binary/cccad6e/ : SUCCESS in 1h 02m 01s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/09/663409/2/check/kolla-ansible-centos-source-upgrade/41b886d/ : SUCCESS in 1h 14m 17s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/09/663409/2/check/tripleo-build-containers-centos-7/818d916/ : SUCCESS in 49m 00s\n- kolla-ansible-bifrost-centos-source http://logs.openstack.org/09/663409/2/check/kolla-ansible-bifrost-centos-source/ef92002/ : SUCCESS in 59m 04s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/09/663409/2/check/kolla-build-oraclelinux-binary/464c7ed/ : SUCCESS in 41m 58s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/09/663409/2/check/kolla-build-oraclelinux-source/913a9ef/ : SUCCESS in 40m 24s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/09/663409/2/check/kolla-ansible-oraclelinux-source/3b43253/ : SUCCESS in 1h 22m 34s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/09/663409/2/check/kolla-ansible-oraclelinux-binary/dea4bd9/ : SUCCESS in 1h 04m 09s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/09/663409/2/check/openstack-tox-lower-constraints/e082651/ : SUCCESS in 3m 59s\n- openstack-tox-pep8 http://logs.openstack.org/09/663409/2/check/openstack-tox-pep8/de9e9f2/ : SUCCESS in 4m 21s\n- openstack-tox-py27 http://logs.openstack.org/09/663409/2/check/openstack-tox-py27/d309152/ : SUCCESS in 3m 12s\n- openstack-tox-py36 http://logs.openstack.org/09/663409/2/check/openstack-tox-py36/0a1995c/ : SUCCESS in 3m 14s\n- openstack-tox-py37 http://logs.openstack.org/09/663409/2/check/openstack-tox-py37/b7d1e4a/ : SUCCESS in 3m 45s\n- openstack-tox-docs http://logs.openstack.org/09/663409/2/check/openstack-tox-docs/5736018/html/ : SUCCESS in 2m 44s\n- kolla-build-ubuntu-binary http://logs.openstack.org/09/663409/2/check/kolla-build-ubuntu-binary/a5ab6f6/ : SUCCESS in 25m 47s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/09/663409/2/check/kolla-build-ubuntu-source/9001d15/ : SUCCESS in 34m 39s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/09/663409/2/check/kolla-ansible-ubuntu-source/bbe7a11/ : SUCCESS in 52m 43s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/09/663409/2/check/kolla-ansible-ubuntu-binary/31679fd/ : SUCCESS in 54m 03s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/09/663409/2/check/kolla-ansible-ubuntu-source-upgrade/5e7ec3b/ : SUCCESS in 1h 04m 31s","accounts_in_message":[],"_revision_number":2},{"id":"d63370468978593a913f5ce4440c71065218b32e","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-11-04 12:28:28.000000000","message":"Abandoned\n\nStein closed for features","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"5f94734bc85b71a04b0e72a36eeda1bb68bee7bd","revisions":{"ac0c434e16b92b807d0eb7c44737939ef8188acc":{"kind":"REWORK","_number":1,"created":"2019-06-05 19:39:16.000000000","uploader":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"ref":"refs/changes/09/663409/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/09/663409/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/09/663409/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/09/663409/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/09/663409/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/09/663409/1"}}},"commit":{"parents":[{"commit":"49f40264f6c4b3cebbd138d97732d0c6c578408d","subject":"Merge \"Fix multiple mon deployment failures\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/49f40264f6c4b3cebbd138d97732d0c6c578408d"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-01-21 20:00:51.000000000","tz":-420},"committer":{"name":"wes hayutin","email":"whayutin@redhat.com","date":"2019-06-05 19:39:16.000000000","tz":0},"subject":"WIP: Fedora support for python3 package testing","message":"WIP: Fedora support for python3 package testing\n\nRDO is currently using a version of Fedora 28 to test the python3\npackaging for the next version of CentOS/RHEL.  This change updates adds\nbasic support to support Fedora28 as a build option.\n\nChange-Id: Ia6bb2c1de260e099cf44f47b190be032007ac8b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ac0c434e16b92b807d0eb7c44737939ef8188acc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ac0c434e16b92b807d0eb7c44737939ef8188acc"}]},"branch":"refs/heads/stable/stein"},"5f94734bc85b71a04b0e72a36eeda1bb68bee7bd":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-06-18 18:36:02.000000000","uploader":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"ref":"refs/changes/09/663409/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/09/663409/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/09/663409/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/09/663409/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/09/663409/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/09/663409/2"}}},"commit":{"parents":[{"commit":"e735b0985e3994d64f800a441e20bf3df092b74b","subject":"Merge \"Debian/Ubuntu: more python3 packages for binary\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e735b0985e3994d64f800a441e20bf3df092b74b"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-01-21 20:00:51.000000000","tz":-420},"committer":{"name":"wes hayutin","email":"whayutin@redhat.com","date":"2019-06-18 18:36:02.000000000","tz":0},"subject":"WIP: Fedora support for python3 package testing","message":"WIP: Fedora support for python3 package testing\n\nRDO is currently using a version of Fedora 28 to test the python3\npackaging for the next version of CentOS/RHEL.  This change updates adds\nbasic support to support Fedora28 as a build option.\n\nChange-Id: Ia6bb2c1de260e099cf44f47b190be032007ac8b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5f94734bc85b71a04b0e72a36eeda1bb68bee7bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5f94734bc85b71a04b0e72a36eeda1bb68bee7bd"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
