)]}'
{"id":"openstack%2Fkolla~663447","triplet_id":"openstack%2Fkolla~master~I8670db4dde4dac26df28502637eadc6b854261cf","project":"openstack/kolla","branch":"master","topic":"locals-base-option","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-01-02 10:44:00.000000000","reason":"Change was submitted"},"24072":{"account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"last_update":"2023-01-02 10:44:00.000000000","reason":"Change was submitted"},"29100":{"account":{"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},"last_update":"2023-01-02 10:44:00.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8670db4dde4dac26df28502637eadc6b854261cf","subject":"Add locals_base option","status":"MERGED","created":"2019-06-05 21:53:35.000000000","updated":"2023-01-02 10:45:11.000000000","submitted":"2023-01-02 10:44:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"663447-locals-base-option","meta_rev_id":"95aa85271e43b7a34dd6f82d8a74c191296de605","_number":663447,"virtual_id_number":663447,"owner":{"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-02 10:44:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2023-01-02 09:27:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2023-01-02 08:48:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2023-01-02 09:27:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-03 16:23:50.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-08-20 22:47:27.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-21 00:11:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-21 06:47:41.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2023-01-02 08:48:37.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2023-01-02 08:48:49.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2023-01-02 09:27:45.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"ac824a3c71c9d46780cac55fd4cdf5880ccddfe0","author":{"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},"date":"2019-06-05 21:53:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"05d1036a56e78f06b3d35ade83a0c2601437d7a5","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-05 22:32:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/47/663447/1/openstack-check/tripleo-build-containers-fedora-28-master/a13a00c/ : SUCCESS in 38m 07s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/47/663447/1/openstack-check/tripleo-build-containers-fedora-28-stein/258b2b8/ : SUCCESS in 35m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a6b24b2ecbc51f98b532b5bdcf980e7d1d6b7a09","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 00:27:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/47/663447/1/check/kolla-build-centos-binary/9b79cea/ : SUCCESS in 34m 58s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/47/663447/1/check/kolla-build-centos-source/b82a8ce/ : SUCCESS in 42m 44s\n- kolla-ansible-centos-source http://logs.openstack.org/47/663447/1/check/kolla-ansible-centos-source/5cfb0bf/ : SUCCESS in 1h 18m 11s\n- kolla-ansible-centos-binary http://logs.openstack.org/47/663447/1/check/kolla-ansible-centos-binary/49e1eb6/ : SUCCESS in 1h 09m 36s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/47/663447/1/check/kolla-ansible-centos-source-upgrade/baa6271/ : SUCCESS in 1h 28m 44s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/47/663447/1/check/tripleo-build-containers-centos-7/0596149/ : SUCCESS in 47m 00s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/47/663447/1/check/kolla-build-oraclelinux-binary/c8d44e8/ : SUCCESS in 31m 00s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/47/663447/1/check/kolla-build-oraclelinux-source/dfcd7ce/ : SUCCESS in 58m 23s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/47/663447/1/check/kolla-ansible-oraclelinux-source/b54053f/ : SUCCESS in 1h 10m 05s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/47/663447/1/check/kolla-ansible-oraclelinux-binary/674aca6/ : SUCCESS in 1h 05m 35s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/47/663447/1/check/openstack-tox-lower-constraints/1708391/ : SUCCESS in 4m 00s\n- openstack-tox-pep8 http://logs.openstack.org/47/663447/1/check/openstack-tox-pep8/9b68799/ : SUCCESS in 3m 17s\n- openstack-tox-py27 http://logs.openstack.org/47/663447/1/check/openstack-tox-py27/deab832/ : SUCCESS in 3m 20s\n- openstack-tox-py36 http://logs.openstack.org/47/663447/1/check/openstack-tox-py36/e44de41/ : SUCCESS in 3m 16s\n- openstack-tox-py37 http://logs.openstack.org/47/663447/1/check/openstack-tox-py37/c69d5b9/ : SUCCESS in 3m 25s\n- openstack-tox-docs http://logs.openstack.org/47/663447/1/check/openstack-tox-docs/4034978/html/ : SUCCESS in 2m 53s\n- kolla-build-ubuntu-binary http://logs.openstack.org/47/663447/1/check/kolla-build-ubuntu-binary/47fea06/ : SUCCESS in 1h 16m 29s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/47/663447/1/check/kolla-build-ubuntu-source/9f8ce9a/ : SUCCESS in 38m 49s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/47/663447/1/check/kolla-ansible-ubuntu-source/833d051/ : SUCCESS in 56m 49s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/47/663447/1/check/kolla-ansible-ubuntu-binary/3379e05/ : SUCCESS in 47m 13s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/47/663447/1/check/kolla-ansible-ubuntu-source-upgrade/9ba8f02/ : SUCCESS in 59m 55s","accounts_in_message":[],"_revision_number":1},{"id":"6eef86484f422104de5eddd33e14cc914eb6cf80","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-06 14:32:47.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nCould you add something in the docs to cover this?","accounts_in_message":[],"_revision_number":1},{"id":"fff99ec817737b26bdecdcf9079769d09a5741e3","author":{"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},"date":"2019-07-18 22:33:47.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"33d7956fc72de54fd90e91ff40b5d62aa5d53a73","author":{"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},"date":"2019-07-18 22:34:14.000000000","message":"Patch Set 2:\n\nI have added docs and also updated the commit message.","accounts_in_message":[],"_revision_number":2},{"id":"a0ac5408fec9901f4732c209f8d0530f4ef1db86","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-18 23:22:07.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/47/663447/2/openstack-check/tripleo-build-containers-fedora-28-master/3a41a2c/ : FAILURE in 5m 45s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/47/663447/2/openstack-check/tripleo-build-containers-fedora-28-stein/1c6bf26/ : SUCCESS in 46m 58s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f5b78048366ff6955a009af6f9e205e989f9d302","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-19 00:12:46.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/47/663447/2/check/kolla-build-centos-binary/b4ffc40/ : SUCCESS in 24m 56s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/47/663447/2/check/kolla-build-centos-source/c39ba91/ : SUCCESS in 45m 34s\n- kolla-ansible-centos-source http://logs.openstack.org/47/663447/2/check/kolla-ansible-centos-source/8165c3c/ : SUCCESS in 1h 14m 54s\n- kolla-ansible-centos-binary http://logs.openstack.org/47/663447/2/check/kolla-ansible-centos-binary/8badc98/ : SUCCESS in 1h 10m 57s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/47/663447/2/check/kolla-ansible-centos-source-upgrade/5d537b9/ : SUCCESS in 1h 29m 54s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/47/663447/2/check/tripleo-build-containers-centos-7/72ffd66/ : SUCCESS in 38m 25s\n- openstack-tox-lower-constraints http://logs.openstack.org/47/663447/2/check/openstack-tox-lower-constraints/63afaa5/ : SUCCESS in 5m 25s\n- openstack-tox-pep8 http://logs.openstack.org/47/663447/2/check/openstack-tox-pep8/7a087c9/ : SUCCESS in 6m 00s\n- openstack-tox-py27 http://logs.openstack.org/47/663447/2/check/openstack-tox-py27/9d75e5c/ : SUCCESS in 3m 30s\n- openstack-tox-py36 http://logs.openstack.org/47/663447/2/check/openstack-tox-py36/a5c8d03/ : SUCCESS in 3m 09s\n- openstack-tox-py37 http://logs.openstack.org/47/663447/2/check/openstack-tox-py37/3a00b6f/ : SUCCESS in 3m 43s\n- openstack-tox-docs http://logs.openstack.org/47/663447/2/check/openstack-tox-docs/c0e2779/ : POST_FAILURE in 2m 41s\n- kolla-build-ubuntu-binary http://logs.openstack.org/47/663447/2/check/kolla-build-ubuntu-binary/51dec33/ : SUCCESS in 28m 27s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/47/663447/2/check/kolla-build-ubuntu-source/38aad90/ : SUCCESS in 33m 06s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/47/663447/2/check/kolla-ansible-ubuntu-source/3b1e94d/ : SUCCESS in 48m 37s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/47/663447/2/check/kolla-ansible-ubuntu-binary/e17b582/ : SUCCESS in 45m 52s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/47/663447/2/check/kolla-ansible-ubuntu-source-upgrade/aa73d5f/ : SUCCESS in 57m 45s","accounts_in_message":[],"_revision_number":2},{"id":"6b3877718e54fea20f04115c417e2dfac45ce9c9","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-08-03 16:23:50.000000000","message":"Patch Set 2: Code-Review-1\n\nDocs build failed:\n\n2019-07-18 22:41:25.237614 | ubuntu-bionic | doc/source/admin/image-building.rst:213: D000 Content block expected for the \"code-block\" directive; none found.\n2019-07-18 22:41:25.237702 | ubuntu-bionic | doc/source/admin/image-building.rst:214: D000 Explicit markup ends without a blank line; unexpected unindent.","accounts_in_message":[],"_revision_number":2},{"id":"92c2a15bb93836124de3ee0f7e6b23088901a212","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-08-03 16:24:17.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"4a6c83465b35c61590b67b639a441ded52ab52ba","author":{"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},"date":"2019-08-20 22:08:28.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"c39cc289d10f05b0fc70d453dbc2223b84504532","author":{"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},"date":"2019-08-20 22:08:42.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"bf4434ea41df162c55199801d586e3f6e3f407d6","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-20 22:47:27.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/47/663447/3/openstack-check/tripleo-build-containers-rhel-8-master/ee894d4/ : SUCCESS in 37m 55s","accounts_in_message":[],"_revision_number":3},{"id":"adeab5e5f5c0c0493ed0882ad09d89f39429c82e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-21 00:11:54.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/6105a7bd13ce4ff98baf95b46efd085b : SUCCESS in 31m 53s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/3476427d8c3b4448b73c8740d5099ec1 : SUCCESS in 47m 13s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/a15a6671c6254df5a40cd764eaab5020 : SUCCESS in 1h 13m 52s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/42b56cbcbfed47619363e2d83e937df8 : SUCCESS in 1h 22m 47s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/2ee909d8c79647bb8ecf8cd98cf6d707 : SUCCESS in 1h 50m 25s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/29effa084e2842d7bb56ec7995e500e7 : SUCCESS in 43m 19s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/52d86d81cd964c8bb7f5e41fec15d790 : SUCCESS in 23m 10s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/26a96238ce9d4ba0a4f52c5d281ebde6 : SUCCESS in 6m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fbac728393d24883b8e2eca02fe51aec : SUCCESS in 4m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f315b6aac3a94ab09dd7459b6ae4efcd : SUCCESS in 5m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c066599fcc7b4adb93136a1dc625c293 : SUCCESS in 4m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9c88166b573749588519cc021cab723d : SUCCESS in 3m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cebfa31bd3464310a5fc779b9ff8c6fb : SUCCESS in 4m 02s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a85d26f4508049d9a7b6e3029010d36c : SUCCESS in 32m 27s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/545f2ec3375f4dd2af3c7feb1611a285 : SUCCESS in 34m 41s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/2f109d37b1ff4922a6031be7b9149573 : SUCCESS in 1h 01m 04s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9575cc1febe94249bc29cdf96e960003 : SUCCESS in 49m 18s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/304c33eafeb8499198546858f93c8b53 : SUCCESS in 1h 06m 10s","accounts_in_message":[],"_revision_number":3},{"id":"2986d4f5ec40c84875b07a14cf37b7c2b25ead8d","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-08-21 06:47:41.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"417063192b976223f0a90fdce38b88131b569d1f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-01-02 07:48:49.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"84d9c717053efa6ac989b37fadadcf8fd4bf739d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-01-02 07:49:33.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"1d05f98fbab6e5ef0df487777a048b3992c8413f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-01-02 07:51:07.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"f3c183a26bf3d1fc76ce73745e8cfccddfc1e910","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-01-02 07:52:09.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"1d33ec3797b6b5ca8db60a8f6b9748e40eb904ca","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-01-02 08:48:37.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"0a4fd644cf253ca31e0fd5f8a4bc3c1fae962f06","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-02 09:06:54.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/81c824ec9506420db451ef3644c7fd63\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/83bb0e553aea449f8b9586db990b527f : SUCCESS in 18m 48s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/6d3a0d93337a4f5981058133f90e605d : SUCCESS in 1h 04m 03s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/65408ea3b6e14d249227d466276f0261 : SUCCESS in 17m 24s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/9b5edb0704084efab54e660f89fb8391 : SUCCESS in 48m 15s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/b4026abdaee648d1aa16876e023af71d : SUCCESS in 1h 09m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c0f0777d6b004cefb3e5296b9bb4d202 : SUCCESS in 2m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/de19da8797624c638792382e6056fc52 : SUCCESS in 3m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d3c619db00f14647868bcdf0ddf4d094 : SUCCESS in 2m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/47bbc99b235a4c7aa8aade2930567f07 : SUCCESS in 2m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33fa3fef4e6f40f8acc5109de61eb637 : SUCCESS in 4m 17s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/a1fe80a1405747549f400a95e1625e1a : SUCCESS in 16m 02s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/e520f452b9114377830f068f391ec2b5 : SUCCESS in 49m 54s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/46a78d291832414aa1fd51198fe9fe02 : SUCCESS in 15m 30s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/53b630be43094bd69180d9dfda3c3eed : SUCCESS in 44m 11s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/70a7d4a85ac348389a5d3afd6ad48b4b : SUCCESS in 1h 02m 49s","accounts_in_message":[],"_revision_number":7},{"id":"392a3c85dd682c77d50c69b0330d06513ff12deb","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-01-02 09:27:45.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"ca0d52ab78adcd4427d350ad3b721364cf21910a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-02 09:28:07.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"29b41a21e0a7b5c2b25c169957f5e1a09971e68a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-02 10:36:41.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9d8026aa8b484fc689444847c24c6017\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/6958adacc3fd4ccf8a0c1b9ed310a748 : SUCCESS in 2h 43m 58s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/25df41c693b5406ab7eb97a3f8c4543b : SUCCESS in 2h 16m 43s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/62fd6f7037454b30a711c59373fe9c42 : SUCCESS in 2h 40m 50s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4fe2dcf9a1be49e58964e2d3d77f7909 : SUCCESS in 5m 37s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/43d1bc4690e748e882783b15af789104 : SUCCESS in 5m 43s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/f5b7dfaf66c549cbbf6fb0beb72e3660 : SUCCESS in 2h 09m 58s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/d789fe4694c040078d850d3886dcdfa7 : SUCCESS in 2h 16m 54s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"50e992ca1914537f4d5b8c0a8149d318fb7a9d99","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-02 10:44:00.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d246950cd66b430abc26bf5dba65988f\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/5aebd6935152435ab81804618df4c06f : SUCCESS in 17m 37s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/230d606c82034682bb420d656dfef3f2 : SUCCESS in 1h 06m 44s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/0409aea6bd7f47ffaac84971604d183d : SUCCESS in 1h 03m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/623060a3f8664dfcaf469d885501f6f2 : SUCCESS in 2m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ecba942eeea84b86969533b065bac9a0 : SUCCESS in 3m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c0d9fb99caba4d5f96449222cf1f617e : SUCCESS in 3m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c2e23d8f3fa24ceab1c985578c88a2ca : SUCCESS in 4m 27s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/a5e5c03cea344d598ec0127361344f13 : SUCCESS in 21m 00s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/a2743d0ffafa40728fb5c90b2637eb26 : SUCCESS in 1h 07m 32s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/f3bae0b29626420e924201cddf8b20bc : SUCCESS in 29m 07s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/ecfc203f37a34aa2b3315cd21e60b44e : SUCCESS in 46m 23s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/1cff25f777694468900ae054823a5002 : SUCCESS in 1h 09m 28s","accounts_in_message":[],"_revision_number":7},{"id":"e15c83bf752ed3c263deb9641e26f3790f876a9e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-02 10:44:00.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"95aa85271e43b7a34dd6f82d8a74c191296de605","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-02 10:45:11.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7477696e067d4003b312eb5d2937c03b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a4e4c9b3cf44c09ac1ee7f308a88597 : SUCCESS in 52s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"d079dd67e3b6f3cca560444809e0d6cedad9d9d3","revisions":{"8f9bb7ee38cae33a8ea88eb3deb16bca6b04600c":{"kind":"REWORK","_number":1,"created":"2019-06-05 21:53:35.000000000","uploader":{"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},"ref":"refs/changes/47/663447/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/47/663447/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/663447/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/663447/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/663447/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/47/663447/1"}}},"commit":{"parents":[{"commit":"f53fcc7e5b250118830615a90cc1675032ffd2aa","subject":"Ability to build container against docker swarm","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f53fcc7e5b250118830615a90cc1675032ffd2aa"}]}],"author":{"name":"Jason","email":"jasonanderson@uchicago.edu","date":"2019-01-10 23:48:15.000000000","tz":-360},"committer":{"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","date":"2019-06-05 21:52:14.000000000","tz":-300},"subject":"Add locals_base option","message":"Add locals_base option\n\nThis option allows you to define \u0027locals\u0027 style dependencies as relative\nto a $locals_base variable, similar to how $tarballs_base works. This\ncan help you avoid defining the same hard-coded path in kolla-build, and\ninstead pass it to kolla via a flag.\n\nSimilar to $tarballs_base, this works via oslo_config\u0027s existing\ninterpolation functionality[1].\n\nExample usage:\n  [neutron-base]\n  type \u003d local\n  location \u003d $locals_base/neutron/neutron-stable-rocky.tar.gz\n\n[1]:\nhttps://docs.openstack.org/oslo.config/latest/reference/configuration-files.html#option-value-interpolation\n\nChange-Id: I8670db4dde4dac26df28502637eadc6b854261cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8f9bb7ee38cae33a8ea88eb3deb16bca6b04600c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8f9bb7ee38cae33a8ea88eb3deb16bca6b04600c"}]},"branch":"refs/heads/master"},"4449303015e3a81851faaee20a8af871fee704d3":{"kind":"REWORK","_number":2,"created":"2019-07-18 22:33:47.000000000","uploader":{"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},"ref":"refs/changes/47/663447/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/47/663447/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/663447/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/663447/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/663447/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/47/663447/2"}}},"commit":{"parents":[{"commit":"0e766d57ad2cfe3c355b7b4ff9dad18ee0354129","subject":"Merge \"Delete default gnocchi \u0026 aodh wsgi conf in Ubuntu binary\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0e766d57ad2cfe3c355b7b4ff9dad18ee0354129"}]}],"author":{"name":"Jason","email":"jasonanderson@uchicago.edu","date":"2019-01-10 23:48:15.000000000","tz":-360},"committer":{"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","date":"2019-07-18 22:33:39.000000000","tz":-300},"subject":"Add locals_base option","message":"Add locals_base option\n\nThis option allows you to define \u0027locals\u0027 style dependencies as relative\nto a $locals_base variable, similar to how $tarballs_base works. This\ncan help you avoid defining the same hard-coded path in kolla-build, and\ninstead pass it to kolla via a flag.\n\nSimilar to $tarballs_base, this works via oslo_config\u0027s existing\ninterpolation functionality[1].\n\nExample usage:\n  [DEFAULT]\n  locals_base \u003d /opt/openstack\n\n  [neutron-base]\n  type \u003d local\n  location \u003d $locals_base/neutron/neutron-stable-rocky.tar.gz\n\n[1]:\nhttps://docs.openstack.org/oslo.config/latest/reference/configuration-files.html#option-value-interpolation\n\nChange-Id: I8670db4dde4dac26df28502637eadc6b854261cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4449303015e3a81851faaee20a8af871fee704d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4449303015e3a81851faaee20a8af871fee704d3"}]},"branch":"refs/heads/master"},"ebbf75e9b0c2846eee95cfb20736ee37c736e254":{"kind":"REWORK","_number":3,"created":"2019-08-20 22:08:28.000000000","uploader":{"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},"ref":"refs/changes/47/663447/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/47/663447/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/663447/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/663447/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/663447/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/47/663447/3"}}},"commit":{"parents":[{"commit":"0e766d57ad2cfe3c355b7b4ff9dad18ee0354129","subject":"Merge \"Delete default gnocchi \u0026 aodh wsgi conf in Ubuntu binary\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0e766d57ad2cfe3c355b7b4ff9dad18ee0354129"}]}],"author":{"name":"Jason","email":"jasonanderson@uchicago.edu","date":"2019-01-10 23:48:15.000000000","tz":-360},"committer":{"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","date":"2019-08-20 22:08:11.000000000","tz":-300},"subject":"Add locals_base option","message":"Add locals_base option\n\nThis option allows you to define \u0027locals\u0027 style dependencies as relative\nto a $locals_base variable, similar to how $tarballs_base works. This\ncan help you avoid defining the same hard-coded path in kolla-build, and\ninstead pass it to kolla via a flag.\n\nSimilar to $tarballs_base, this works via oslo_config\u0027s existing\ninterpolation functionality[1].\n\nExample usage:\n  [DEFAULT]\n  locals_base \u003d /opt/openstack\n\n  [neutron-base]\n  type \u003d local\n  location \u003d $locals_base/neutron/neutron-stable-rocky.tar.gz\n\n[1]:\nhttps://docs.openstack.org/oslo.config/latest/reference/configuration-files.html#option-value-interpolation\n\nChange-Id: I8670db4dde4dac26df28502637eadc6b854261cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ebbf75e9b0c2846eee95cfb20736ee37c736e254"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ebbf75e9b0c2846eee95cfb20736ee37c736e254"}]},"branch":"refs/heads/master"},"cf9020938a47d80c5f0a2432246cf2cdc75b96c8":{"kind":"REWORK","_number":4,"created":"2023-01-02 07:48:49.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/47/663447/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/47/663447/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/663447/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/663447/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/663447/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/47/663447/4"}}},"commit":{"parents":[{"commit":"8585495f451ca8d1a719e550d63949f4142cf4e6","subject":"Merge \"Revert \"[release] Change RDO\u0027s DELOREAN_DEPS repo to Zed\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8585495f451ca8d1a719e550d63949f4142cf4e6"}]}],"author":{"name":"Jason","email":"jasonanderson@uchicago.edu","date":"2019-01-10 23:48:15.000000000","tz":-360},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-02 07:48:46.000000000","tz":60},"subject":"Add locals_base option","message":"Add locals_base option\n\nThis option allows you to define \u0027locals\u0027 style dependencies as relative\nto a $locals_base variable, similar to how $tarballs_base works. This\ncan help you avoid defining the same hard-coded path in kolla-build, and\ninstead pass it to kolla via a flag.\n\nSimilar to $tarballs_base, this works via oslo_config\u0027s existing\ninterpolation functionality[1].\n\nExample usage:\n  [DEFAULT]\n  locals_base \u003d /opt/openstack\n\n  [neutron-base]\n  type \u003d local\n  location \u003d $locals_base/neutron/neutron-stable-rocky.tar.gz\n\n[1]:\nhttps://docs.openstack.org/oslo.config/latest/reference/configuration-files.html#option-value-interpolation\n\nChange-Id: I8670db4dde4dac26df28502637eadc6b854261cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cf9020938a47d80c5f0a2432246cf2cdc75b96c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cf9020938a47d80c5f0a2432246cf2cdc75b96c8"}]},"branch":"refs/heads/master"},"e3b8bc5e82f5fe21a9938319968f47fac7a5b2e6":{"kind":"REWORK","_number":5,"created":"2023-01-02 07:49:33.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/47/663447/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/47/663447/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/663447/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/663447/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/663447/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/47/663447/5"}}},"commit":{"parents":[{"commit":"8585495f451ca8d1a719e550d63949f4142cf4e6","subject":"Merge \"Revert \"[release] Change RDO\u0027s DELOREAN_DEPS repo to Zed\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8585495f451ca8d1a719e550d63949f4142cf4e6"}]}],"author":{"name":"Jason","email":"jasonanderson@uchicago.edu","date":"2019-01-10 23:48:15.000000000","tz":-360},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-02 07:49:27.000000000","tz":60},"subject":"Add locals_base option","message":"Add locals_base option\n\nThis option allows you to define \u0027locals\u0027 style dependencies as relative\nto a $locals_base variable, similar to how $tarballs_base works. This\ncan help you avoid defining the same hard-coded path in kolla-build, and\ninstead pass it to kolla via a flag.\n\nSimilar to $tarballs_base, this works via oslo_config\u0027s existing\ninterpolation functionality[1].\n\nExample usage:\n  [DEFAULT]\n  locals_base \u003d /opt/openstack\n\n  [neutron-base]\n  type \u003d local\n  location \u003d $locals_base/neutron/neutron-stable-rocky.tar.gz\n\n[1]:\nhttps://docs.openstack.org/oslo.config/latest/reference/configuration-files.html#option-value-interpolation\n\nChange-Id: I8670db4dde4dac26df28502637eadc6b854261cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e3b8bc5e82f5fe21a9938319968f47fac7a5b2e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e3b8bc5e82f5fe21a9938319968f47fac7a5b2e6"}]},"branch":"refs/heads/master"},"ff24f0bb71fc57bab673a423165ecd458b7da80f":{"kind":"REWORK","_number":6,"created":"2023-01-02 07:51:07.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/47/663447/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/47/663447/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/663447/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/663447/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/663447/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/47/663447/6"}}},"commit":{"parents":[{"commit":"8585495f451ca8d1a719e550d63949f4142cf4e6","subject":"Merge \"Revert \"[release] Change RDO\u0027s DELOREAN_DEPS repo to Zed\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8585495f451ca8d1a719e550d63949f4142cf4e6"}]}],"author":{"name":"Jason","email":"jasonanderson@uchicago.edu","date":"2019-01-10 23:48:15.000000000","tz":-360},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-02 07:51:01.000000000","tz":60},"subject":"Add locals_base option","message":"Add locals_base option\n\nThis option allows you to define \u0027locals\u0027 style dependencies as relative\nto a $locals_base variable, similar to how $tarballs_base works. This\ncan help you avoid defining the same hard-coded path in kolla-build, and\ninstead pass it to kolla via a flag.\n\nSimilar to $tarballs_base, this works via oslo_config\u0027s existing\ninterpolation functionality[1].\n\nExample usage:\n  [DEFAULT]\n  locals_base \u003d /opt/openstack\n\n  [neutron-base]\n  type \u003d local\n  location \u003d $locals_base/neutron/neutron-stable-rocky.tar.gz\n\n[1]:\nhttps://docs.openstack.org/oslo.config/latest/reference/configuration-files.html#option-value-interpolation\n\nChange-Id: I8670db4dde4dac26df28502637eadc6b854261cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ff24f0bb71fc57bab673a423165ecd458b7da80f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ff24f0bb71fc57bab673a423165ecd458b7da80f"}]},"branch":"refs/heads/master"},"d079dd67e3b6f3cca560444809e0d6cedad9d9d3":{"kind":"REWORK","_number":7,"created":"2023-01-02 07:52:09.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/47/663447/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/47/663447/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/663447/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/663447/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/663447/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/47/663447/7"}}},"commit":{"parents":[{"commit":"8585495f451ca8d1a719e550d63949f4142cf4e6","subject":"Merge \"Revert \"[release] Change RDO\u0027s DELOREAN_DEPS repo to Zed\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8585495f451ca8d1a719e550d63949f4142cf4e6"}]}],"author":{"name":"Jason","email":"jasonanderson@uchicago.edu","date":"2019-01-10 23:48:15.000000000","tz":-360},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-02 07:52:02.000000000","tz":60},"subject":"Add locals_base option","message":"Add locals_base option\n\nThis option allows you to define \u0027locals\u0027 style dependencies as relative\nto a $locals_base variable, similar to how $tarballs_base works. This\ncan help you avoid defining the same hard-coded path in kolla-build, and\ninstead pass it to kolla via a flag.\n\nSimilar to $tarballs_base, this works via oslo_config\u0027s existing\ninterpolation functionality[1].\n\nExample usage:\n  [DEFAULT]\n  locals_base \u003d /opt/openstack\n\n  [neutron-base]\n  type \u003d local\n  location \u003d $locals_base/neutron/neutron-stable-rocky.tar.gz\n\n[1]:\nhttps://docs.openstack.org/oslo.config/latest/reference/configuration-files.html#option-value-interpolation\n\nChange-Id: I8670db4dde4dac26df28502637eadc6b854261cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d079dd67e3b6f3cca560444809e0d6cedad9d9d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d079dd67e3b6f3cca560444809e0d6cedad9d9d3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
