)]}'
{"id":"openstack%2Fkolla~663892","triplet_id":"openstack%2Fkolla~master~I530fde5b57894afda2de98d7c217e735497e3feb","project":"openstack/kolla","branch":"master","hashtags":[],"change_id":"I530fde5b57894afda2de98d7c217e735497e3feb","subject":"Debian/Ubuntu: allow to use unsigned repositories","status":"MERGED","created":"2019-06-07 10:55:18.000000000","updated":"2019-06-13 18:12:07.000000000","submitted":"2019-06-10 12:22:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"663892-1560169338097-b35f9ced","meta_rev_id":"54a9dd5a5f5c6b8afad1f43e25c3dd109bde7ab2","_number":663892,"virtual_id_number":663892,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"date":"2019-06-07 15:56:25.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":2,"date":"2019-06-10 12:22:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30429,"name":"wangpeng","email":"wang_peng@awcloud.com"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},{"value":2,"date":"2019-06-10 08:54:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2019-06-07 16:59:02.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2019-06-07 15:59:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-06-10 08:58:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30429,"name":"wangpeng","email":"wang_peng@awcloud.com"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},{"value":1,"date":"2019-06-10 08:54:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30429,"name":"wangpeng","email":"wang_peng@awcloud.com"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30429,"name":"wangpeng","email":"wang_peng@awcloud.com"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"date":"2019-06-07 21:56:01.000000000","_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30429,"name":"wangpeng","email":"wang_peng@awcloud.com"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30429,"name":"wangpeng","email":"wang_peng@awcloud.com"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-07 15:46:07.000000000","updated_by":{"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},"reviewer":{"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},"state":"REVIEWER"},{"updated":"2019-06-07 15:59:49.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-06-07 16:59:02.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-06-07 21:56:01.000000000","updated_by":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"reviewer":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"state":"REVIEWER"},{"updated":"2019-06-10 08:54:41.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2019-06-10 08:58:14.000000000","updated_by":{"_account_id":30429,"name":"wangpeng","email":"wang_peng@awcloud.com"},"reviewer":{"_account_id":30429,"name":"wangpeng","email":"wang_peng@awcloud.com"},"state":"REVIEWER"},{"updated":"2019-06-10 12:22:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6d561426517cdedeb7a41a3dd53c4db060fa1c35","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-07 10:55:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"97f3adc956d798cc53e453156fb116bf3d9802d0","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-07 11:05:49.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d85c3168b380a40b0c3bbfc5e239acda0debd3bb","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-07 12:25:48.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c56634d2a41d5bf3a2abcda10f359363e19881cd","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-07 12:26:16.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nalso merged two RUN lines into one to avoid temporary container","accounts_in_message":[],"_revision_number":1},{"id":"27c59c75b9633e1d7a27c117a057682975b55a7b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 13:48:28.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/92/663892/2/check/kolla-build-centos-binary/8d7bba1/ : SUCCESS in 58m 43s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/92/663892/2/check/kolla-build-centos-source/425efcf/ : SUCCESS in 1h 05m 07s\n- kolla-ansible-centos-source http://logs.openstack.org/92/663892/2/check/kolla-ansible-centos-source/dcf8d28/ : SUCCESS in 1h 16m 54s\n- kolla-ansible-centos-binary http://logs.openstack.org/92/663892/2/check/kolla-ansible-centos-binary/d9b0dca/ : SUCCESS in 58m 48s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/92/663892/2/check/kolla-ansible-centos-source-upgrade/23c284a/ : SUCCESS in 1h 16m 38s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/92/663892/2/check/kolla-build-oraclelinux-binary/58420f3/ : SUCCESS in 51m 08s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/92/663892/2/check/kolla-build-oraclelinux-source/add164a/ : SUCCESS in 43m 01s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/92/663892/2/check/kolla-ansible-oraclelinux-source/5fee91f/ : SUCCESS in 1h 03m 09s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/92/663892/2/check/kolla-ansible-oraclelinux-binary/8ad9f2e/ : SUCCESS in 1h 02m 02s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/92/663892/2/check/openstack-tox-lower-constraints/9e3ef62/ : SUCCESS in 3m 45s\n- openstack-tox-pep8 http://logs.openstack.org/92/663892/2/check/openstack-tox-pep8/bcf754b/ : SUCCESS in 2m 57s\n- openstack-tox-py27 http://logs.openstack.org/92/663892/2/check/openstack-tox-py27/0d97448/ : SUCCESS in 3m 01s\n- openstack-tox-py36 http://logs.openstack.org/92/663892/2/check/openstack-tox-py36/45372ee/ : SUCCESS in 3m 02s\n- openstack-tox-py37 http://logs.openstack.org/92/663892/2/check/openstack-tox-py37/97fc661/ : SUCCESS in 3m 02s\n- openstack-tox-docs http://logs.openstack.org/92/663892/2/check/openstack-tox-docs/1178b4d/html/ : SUCCESS in 2m 19s\n- kolla-build-ubuntu-binary http://logs.openstack.org/92/663892/2/check/kolla-build-ubuntu-binary/3637351/ : SUCCESS in 30m 43s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/92/663892/2/check/kolla-build-ubuntu-source/207226d/ : SUCCESS in 31m 43s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/92/663892/2/check/kolla-ansible-ubuntu-source/8920310/ : SUCCESS in 44m 36s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/92/663892/2/check/kolla-ansible-ubuntu-binary/a3f18ce/ : SUCCESS in 52m 45s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/92/663892/2/check/kolla-ansible-ubuntu-source-upgrade/03d3da6/ : SUCCESS in 53m 44s","accounts_in_message":[],"_revision_number":2},{"id":"69d50309179dea9790db975b109d709502d4c94e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-07 15:01:46.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"ea4904bd216cce0d3908d42c999e994c0e11efde","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-07 15:36:53.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d34d374da0d2e95762cca6c2b4b82a0a7e105b15","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-07 15:56:06.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"ef45f89ce43e0ba450e2b2748d38c6004e3af583","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-07 15:56:25.000000000","message":"Patch Set 3:\n\n(1 comment)\n\napt configs renamed to 99*","accounts_in_message":[],"_revision_number":3},{"id":"fc299aae4b8cb9dce8a3e85569b21952456aa9f5","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-07 15:59:49.000000000","message":"Patch Set 3: Code-Review+1\n\nFinally worth of my +1 :-)","accounts_in_message":[],"_revision_number":3},{"id":"347c1d26f94f37f2a20a7138ceec4ed4dbe76f64","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-07 16:00:29.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c548d786dd4b75d6b33e77515e6bdcdc1d8baa2f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-07 16:59:02.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"0a6398264803777c6f0713ad2f1b08753dec84d2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 17:35:58.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/92/663892/3/check/kolla-build-centos-binary/98737cc/ : SUCCESS in 43m 11s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/92/663892/3/check/kolla-build-centos-source/5eb8254/ : SUCCESS in 45m 49s\n- kolla-ansible-centos-source http://logs.openstack.org/92/663892/3/check/kolla-ansible-centos-source/dad8d09/ : SUCCESS in 1h 05m 05s\n- kolla-ansible-centos-binary http://logs.openstack.org/92/663892/3/check/kolla-ansible-centos-binary/71d775c/ : SUCCESS in 1h 09m 56s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/92/663892/3/check/kolla-ansible-centos-source-upgrade/0e67563/ : SUCCESS in 1h 17m 13s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/92/663892/3/check/kolla-build-oraclelinux-binary/c9335ae/ : SUCCESS in 41m 39s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/92/663892/3/check/kolla-build-oraclelinux-source/5ba2d47/ : SUCCESS in 1h 04m 28s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/92/663892/3/check/kolla-ansible-oraclelinux-source/f0f7e68/ : SUCCESS in 1h 11m 54s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/92/663892/3/check/kolla-ansible-oraclelinux-binary/13ec721/ : SUCCESS in 1h 02m 48s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/92/663892/3/check/openstack-tox-lower-constraints/1178357/ : SUCCESS in 4m 24s\n- openstack-tox-pep8 http://logs.openstack.org/92/663892/3/check/openstack-tox-pep8/60e77ad/ : SUCCESS in 3m 16s\n- openstack-tox-py27 http://logs.openstack.org/92/663892/3/check/openstack-tox-py27/0cd75b2/ : SUCCESS in 3m 23s\n- openstack-tox-py36 http://logs.openstack.org/92/663892/3/check/openstack-tox-py36/b31c84c/ : SUCCESS in 2m 38s\n- openstack-tox-py37 http://logs.openstack.org/92/663892/3/check/openstack-tox-py37/d199e19/ : SUCCESS in 4m 01s\n- openstack-tox-docs http://logs.openstack.org/92/663892/3/check/openstack-tox-docs/89055de/html/ : SUCCESS in 3m 07s\n- kolla-build-ubuntu-binary http://logs.openstack.org/92/663892/3/check/kolla-build-ubuntu-binary/b609283/ : SUCCESS in 31m 08s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/92/663892/3/check/kolla-build-ubuntu-source/f6e8453/ : SUCCESS in 24m 08s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/92/663892/3/check/kolla-ansible-ubuntu-source/7e76bbd/ : SUCCESS in 50m 21s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/92/663892/3/check/kolla-ansible-ubuntu-binary/d15434c/ : SUCCESS in 1h 05m 15s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/92/663892/3/check/kolla-ansible-ubuntu-source-upgrade/d8b7bfa/ : SUCCESS in 56m 17s","accounts_in_message":[],"_revision_number":3},{"id":"534edc9db6a0641cf4f482066dcdc8949fd76cb1","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-06-10 08:54:41.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"ced30b3ec017c88a956e46b6a3917fa8b5209e9e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 08:54:53.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"9ed375526f1f15894cc6b7f5dbad0de658e9cce0","author":{"_account_id":30429,"name":"wangpeng","email":"wang_peng@awcloud.com"},"date":"2019-06-10 08:58:14.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"532bc3897cedf61d8d86b809080e86aed7f1608a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 09:47:23.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-source http://logs.openstack.org/92/663892/3/gate/kolla-build-centos-source/590519b/ : FAILURE in 16m 03s\n- kolla-build-oraclelinux-source http://logs.openstack.org/92/663892/3/gate/kolla-build-oraclelinux-source/745b82c/ : SUCCESS in 49m 45s\n- openstack-tox-lower-constraints http://logs.openstack.org/92/663892/3/gate/openstack-tox-lower-constraints/85dc8f7/ : SUCCESS in 5m 35s\n- openstack-tox-pep8 http://logs.openstack.org/92/663892/3/gate/openstack-tox-pep8/6a9bbda/ : SUCCESS in 4m 13s\n- openstack-tox-py27 http://logs.openstack.org/92/663892/3/gate/openstack-tox-py27/cc32dce/ : SUCCESS in 4m 06s\n- openstack-tox-py36 http://logs.openstack.org/92/663892/3/gate/openstack-tox-py36/93afcde/ : SUCCESS in 3m 27s\n- openstack-tox-py37 http://logs.openstack.org/92/663892/3/gate/openstack-tox-py37/9b6ee76/ : SUCCESS in 3m 29s\n- openstack-tox-docs http://logs.openstack.org/92/663892/3/gate/openstack-tox-docs/94c929d/html/ : SUCCESS in 3m 04s\n- kolla-build-ubuntu-source http://logs.openstack.org/92/663892/3/gate/kolla-build-ubuntu-source/9785e70/ : SUCCESS in 29m 30s","accounts_in_message":[],"_revision_number":3},{"id":"21e8135c2ffcabd1a0fa5e872c6e10b0998784f7","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-10 09:50:15.000000000","message":"Patch Set 3:\n\nrecheck\n\nSweet irony when Percona fix depends on a change that failed gate due to Percona. :-)","accounts_in_message":[],"_revision_number":3},{"id":"1402d7a4d168dcce833c2ba171f8989e59ab5604","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 11:25:21.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/92/663892/3/check/kolla-build-centos-binary/5c5bdee/ : SUCCESS in 40m 41s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/92/663892/3/check/kolla-build-centos-source/68dac17/ : SUCCESS in 1h 00m 14s\n- kolla-ansible-centos-source http://logs.openstack.org/92/663892/3/check/kolla-ansible-centos-source/5739a6d/ : SUCCESS in 1h 28m 21s\n- kolla-ansible-centos-binary http://logs.openstack.org/92/663892/3/check/kolla-ansible-centos-binary/c9a1a12/ : SUCCESS in 1h 03m 11s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/92/663892/3/check/kolla-ansible-centos-source-upgrade/4489043/ : SUCCESS in 1h 18m 04s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/92/663892/3/check/kolla-build-oraclelinux-binary/1c7dd78/ : SUCCESS in 48m 36s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/92/663892/3/check/kolla-build-oraclelinux-source/c4debe5/ : SUCCESS in 1h 04m 49s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/92/663892/3/check/kolla-ansible-oraclelinux-source/e57a1af/ : SUCCESS in 1h 10m 49s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/92/663892/3/check/kolla-ansible-oraclelinux-binary/96a25c4/ : SUCCESS in 1h 08m 02s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/92/663892/3/check/openstack-tox-lower-constraints/1b9db75/ : SUCCESS in 3m 34s\n- openstack-tox-pep8 http://logs.openstack.org/92/663892/3/check/openstack-tox-pep8/3f62257/ : SUCCESS in 3m 40s\n- openstack-tox-py27 http://logs.openstack.org/92/663892/3/check/openstack-tox-py27/6783519/ : SUCCESS in 3m 22s\n- openstack-tox-py36 http://logs.openstack.org/92/663892/3/check/openstack-tox-py36/5e30fce/ : SUCCESS in 3m 36s\n- openstack-tox-py37 http://logs.openstack.org/92/663892/3/check/openstack-tox-py37/3167185/ : SUCCESS in 3m 49s\n- openstack-tox-docs http://logs.openstack.org/92/663892/3/check/openstack-tox-docs/fb71ca2/html/ : SUCCESS in 2m 46s\n- kolla-build-ubuntu-binary http://logs.openstack.org/92/663892/3/check/kolla-build-ubuntu-binary/11a0e79/ : SUCCESS in 23m 53s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/92/663892/3/check/kolla-build-ubuntu-source/6b71a83/ : SUCCESS in 30m 10s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/92/663892/3/check/kolla-ansible-ubuntu-source/c36398f/ : SUCCESS in 45m 09s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/92/663892/3/check/kolla-ansible-ubuntu-binary/12375e5/ : SUCCESS in 45m 42s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/92/663892/3/check/kolla-ansible-ubuntu-source-upgrade/b81b5fb/ : SUCCESS in 54m 03s","accounts_in_message":[],"_revision_number":3},{"id":"30044e9eeda1429175e3c715733030c5b48a1fd7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 11:25:33.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"15a6196f51ca6e0f926bac70875d7b61c755db07","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-10 11:27:35.000000000","message":"Patch Set 3:\n\nfingers crossed","accounts_in_message":[],"_revision_number":3},{"id":"48295ff5f1a5e1a1c1fdba3b55b05de3ecffaf00","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 12:22:17.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source http://logs.openstack.org/92/663892/3/gate/kolla-build-centos-source/7c7a033/ : SUCCESS in 51m 01s\n- kolla-build-oraclelinux-source http://logs.openstack.org/92/663892/3/gate/kolla-build-oraclelinux-source/229fb8d/ : SUCCESS in 50m 45s\n- openstack-tox-lower-constraints http://logs.openstack.org/92/663892/3/gate/openstack-tox-lower-constraints/f807bcb/ : SUCCESS in 3m 59s\n- openstack-tox-pep8 http://logs.openstack.org/92/663892/3/gate/openstack-tox-pep8/7be6c48/ : SUCCESS in 3m 06s\n- openstack-tox-py27 http://logs.openstack.org/92/663892/3/gate/openstack-tox-py27/d6f38e0/ : SUCCESS in 3m 22s\n- openstack-tox-py36 http://logs.openstack.org/92/663892/3/gate/openstack-tox-py36/64806f7/ : SUCCESS in 3m 06s\n- openstack-tox-py37 http://logs.openstack.org/92/663892/3/gate/openstack-tox-py37/7316153/ : SUCCESS in 3m 56s\n- openstack-tox-docs http://logs.openstack.org/92/663892/3/gate/openstack-tox-docs/15e84a9/html/ : SUCCESS in 2m 57s\n- kolla-build-ubuntu-source http://logs.openstack.org/92/663892/3/gate/kolla-build-ubuntu-source/490a11b/ : SUCCESS in 30m 24s","accounts_in_message":[],"_revision_number":3},{"id":"23a2800a4c96bfcae05dfbc0dc57ab669a732c73","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 12:22:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"54a9dd5a5f5c6b8afad1f43e25c3dd109bde7ab2","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-13 18:12:07.000000000","message":"Patch Set 3: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit 46272752c03114a049c781ec432cfd645d81635a","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"5859180c2c78c62defca37446722b5f15fdbefb7","revisions":{"45335ce002e0f7b88ae9ebac16d3d4f82d6297be":{"kind":"REWORK","_number":1,"created":"2019-06-07 10:55:18.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/92/663892/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/92/663892/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/92/663892/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/92/663892/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/92/663892/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/92/663892/1"}}},"commit":{"parents":[{"commit":"9cfd3a88aa1c07054b66495dbb069a0000a20738","subject":"Merge \"Ignore EM releases in version-check.py\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9cfd3a88aa1c07054b66495dbb069a0000a20738"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-06-07 10:53:51.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-06-07 10:53:51.000000000","tz":120},"subject":"Debian/Ubuntu: allow to use unsigned repositories","message":"Debian/Ubuntu: allow to use unsigned repositories\n\nWhen we run Debian \u0027buster\u0027 or Ubuntu 18.04 in CI job they refuse to\nwork as infra mirror is not signed.\n\nThis patch instructs APT to use unsigned repos anyway.\n\nChange-Id: I530fde5b57894afda2de98d7c217e735497e3feb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/45335ce002e0f7b88ae9ebac16d3d4f82d6297be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/45335ce002e0f7b88ae9ebac16d3d4f82d6297be"}]},"branch":"refs/heads/master"},"3de629986a228e2da4bf03c8397958b2a5f0c960":{"kind":"REWORK","_number":2,"created":"2019-06-07 12:25:48.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/92/663892/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/92/663892/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/92/663892/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/92/663892/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/92/663892/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/92/663892/2"}}},"commit":{"parents":[{"commit":"9cfd3a88aa1c07054b66495dbb069a0000a20738","subject":"Merge \"Ignore EM releases in version-check.py\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9cfd3a88aa1c07054b66495dbb069a0000a20738"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-06-07 10:53:51.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-06-07 12:25:35.000000000","tz":120},"subject":"Debian/Ubuntu: allow to use unsigned repositories","message":"Debian/Ubuntu: allow to use unsigned repositories\n\nWhen we run Debian \u0027buster\u0027 or Ubuntu 18.04 in CI job they refuse to\nwork as infra mirror is not signed.\n\nThis patch instructs APT to use unsigned repos anyway.\n\nChange-Id: I530fde5b57894afda2de98d7c217e735497e3feb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3de629986a228e2da4bf03c8397958b2a5f0c960"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3de629986a228e2da4bf03c8397958b2a5f0c960"}]},"branch":"refs/heads/master"},"5859180c2c78c62defca37446722b5f15fdbefb7":{"kind":"REWORK","_number":3,"created":"2019-06-07 15:56:06.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/92/663892/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/92/663892/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/92/663892/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/92/663892/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/92/663892/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/92/663892/3"}}},"commit":{"parents":[{"commit":"9cfd3a88aa1c07054b66495dbb069a0000a20738","subject":"Merge \"Ignore EM releases in version-check.py\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9cfd3a88aa1c07054b66495dbb069a0000a20738"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-06-07 10:53:51.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-06-07 15:55:58.000000000","tz":0},"subject":"Debian/Ubuntu: allow to use unsigned repositories","message":"Debian/Ubuntu: allow to use unsigned repositories\n\nWhen we run Debian \u0027buster\u0027 or Ubuntu 18.04 in CI job they refuse to\nwork as infra mirror is not signed.\n\nThis patch instructs APT to use unsigned repos anyway.\n\nChange-Id: I530fde5b57894afda2de98d7c217e735497e3feb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5859180c2c78c62defca37446722b5f15fdbefb7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5859180c2c78c62defca37446722b5f15fdbefb7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
