)]}'
{"id":"openstack%2Fkolla~663893","triplet_id":"openstack%2Fkolla~master~Id27831f6b404762f9e44fd6bd1c4702bf06b05f8","project":"openstack/kolla","branch":"master","topic":"fix-percona","hashtags":[],"change_id":"Id27831f6b404762f9e44fd6bd1c4702bf06b05f8","subject":"Fix issues with Percona in Zuul and clean up a little bit","status":"MERGED","created":"2019-06-07 10:56:59.000000000","updated":"2019-06-13 18:15:28.000000000","submitted":"2019-06-10 15:12:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"663893-1560179521693-7fce86ad","meta_rev_id":"9d9319d8e36a5ffc6a709e9d302af828394acb1a","_number":663893,"virtual_id_number":663893,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"date":"2019-06-10 12:23:45.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":2,"date":"2019-06-10 15:12:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-06-10 13:09:24.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2019-06-10 08:14:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2019-06-08 18:58:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-06-10 08:10:32.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2019-06-10 08:14:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-07 21:27:43.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-06-08 18:58:22.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-06-10 08:10:32.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-06-10 08:14:38.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2019-06-10 13:09:24.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-06-10 15:12:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"81720afa22a32111a01a9a5f9896546487ac4fdc","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-07 10:56:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"184c1149376ff87e17196f1e5739d1cd9927964d","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-07 10:59:33.000000000","message":"Patch Set 1: Code-Review+1\n\nlet\u0027s see what zuul say","accounts_in_message":[],"_revision_number":1},{"id":"528ffc4b5ffef1ace63d9af399a00dfb231a57f3","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-07 11:27:03.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d483ffbe89e7d984135b648c9e09a2d3a03bd64f","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-07 12:04:13.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/93/663893/2/openstack-check/tripleo-build-containers-fedora-28-master/2a1f424/ : SUCCESS in 35m 18s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/93/663893/2/openstack-check/tripleo-build-containers-fedora-28-stein/d2cb10e/ : SUCCESS in 34m 53s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3eacce7f1966abcb8980bc0640dd034768837c94","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 13:01:49.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/93/663893/2/check/kolla-build-centos-binary/02c0ca5/ : SUCCESS in 25m 26s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/93/663893/2/check/kolla-build-centos-source/80271a0/ : SUCCESS in 58m 04s\n- kolla-ansible-centos-source http://logs.openstack.org/93/663893/2/check/kolla-ansible-centos-source/b530f05/ : SUCCESS in 1h 11m 27s\n- kolla-ansible-centos-binary http://logs.openstack.org/93/663893/2/check/kolla-ansible-centos-binary/5d84af1/ : SUCCESS in 1h 08m 46s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/93/663893/2/check/kolla-ansible-centos-source-upgrade/a0b1f5c/ : SUCCESS in 1h 21m 38s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/93/663893/2/check/tripleo-build-containers-centos-7/7682c35/ : SUCCESS in 48m 00s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/93/663893/2/check/kolla-build-oraclelinux-binary/638d849/ : SUCCESS in 50m 10s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/93/663893/2/check/kolla-build-oraclelinux-source/13b99b4/ : SUCCESS in 56m 09s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/93/663893/2/check/kolla-ansible-oraclelinux-source/70af49a/ : SUCCESS in 1h 03m 07s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/93/663893/2/check/kolla-ansible-oraclelinux-binary/8c0219a/ : SUCCESS in 1h 07m 14s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/93/663893/2/check/openstack-tox-lower-constraints/7ce9743/ : SUCCESS in 4m 08s\n- openstack-tox-pep8 http://logs.openstack.org/93/663893/2/check/openstack-tox-pep8/851b4a3/ : SUCCESS in 3m 21s\n- openstack-tox-py27 http://logs.openstack.org/93/663893/2/check/openstack-tox-py27/ada6f69/ : SUCCESS in 2m 56s\n- openstack-tox-py36 http://logs.openstack.org/93/663893/2/check/openstack-tox-py36/15c4a5f/ : SUCCESS in 3m 32s\n- openstack-tox-py37 http://logs.openstack.org/93/663893/2/check/openstack-tox-py37/18755b0/ : SUCCESS in 3m 17s\n- openstack-tox-docs http://logs.openstack.org/93/663893/2/check/openstack-tox-docs/232747d/html/ : SUCCESS in 2m 25s\n- kolla-build-ubuntu-binary http://logs.openstack.org/93/663893/2/check/kolla-build-ubuntu-binary/c5af5e0/ : SUCCESS in 25m 31s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/93/663893/2/check/kolla-build-ubuntu-source/ecec222/ : SUCCESS in 34m 49s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/93/663893/2/check/kolla-ansible-ubuntu-source/7d7964c/ : SUCCESS in 46m 37s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/93/663893/2/check/kolla-ansible-ubuntu-binary/a6e4d48/ : SUCCESS in 46m 17s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/93/663893/2/check/kolla-ansible-ubuntu-source-upgrade/fa9c7b4/ : SUCCESS in 1h 01m 33s","accounts_in_message":[],"_revision_number":2},{"id":"812c43da60d993fe45ba7bf3a195a4dc044abd53","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-07 13:11:34.000000000","message":"Patch Set 2:\n\nLooks good. Let\u0027s run it a few times.","accounts_in_message":[],"_revision_number":2},{"id":"91364d3d83316e354f4d1ee11f4f03fa9bf3644e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-07 13:11:40.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"b58529fe4b6fcba761acc2f489a67d62607006a3","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-07 13:52:54.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/93/663893/2/openstack-check/tripleo-build-containers-fedora-28-master/e03ee1f/ : SUCCESS in 38m 01s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/93/663893/2/openstack-check/tripleo-build-containers-fedora-28-stein/335310d/ : SUCCESS in 37m 43s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d6225644cf469aae7c7bb63594f4bbb1d8b254be","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 15:18:41.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/93/663893/2/check/kolla-build-centos-binary/9dd3964/ : SUCCESS in 49m 36s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/93/663893/2/check/kolla-build-centos-source/5335546/ : SUCCESS in 49m 11s\n- kolla-ansible-centos-source http://logs.openstack.org/93/663893/2/check/kolla-ansible-centos-source/36e2271/ : SUCCESS in 1h 14m 09s\n- kolla-ansible-centos-binary http://logs.openstack.org/93/663893/2/check/kolla-ansible-centos-binary/fd3af92/ : SUCCESS in 1h 07m 58s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/93/663893/2/check/kolla-ansible-centos-source-upgrade/d149999/ : SUCCESS in 1h 16m 20s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/93/663893/2/check/tripleo-build-containers-centos-7/557f37e/ : SUCCESS in 42m 26s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/93/663893/2/check/kolla-build-oraclelinux-binary/5fe1b83/ : SUCCESS in 50m 55s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/93/663893/2/check/kolla-build-oraclelinux-source/2270cc5/ : SUCCESS in 52m 58s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/93/663893/2/check/kolla-ansible-oraclelinux-source/7d53e70/ : SUCCESS in 1h 03m 29s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/93/663893/2/check/kolla-ansible-oraclelinux-binary/e8f1674/ : SUCCESS in 1h 07m 14s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/93/663893/2/check/openstack-tox-lower-constraints/5215471/ : SUCCESS in 4m 52s\n- openstack-tox-pep8 http://logs.openstack.org/93/663893/2/check/openstack-tox-pep8/a2e37e5/ : SUCCESS in 4m 23s\n- openstack-tox-py27 http://logs.openstack.org/93/663893/2/check/openstack-tox-py27/34cae1d/ : SUCCESS in 3m 00s\n- openstack-tox-py36 http://logs.openstack.org/93/663893/2/check/openstack-tox-py36/4d09e32/ : SUCCESS in 2m 55s\n- openstack-tox-py37 http://logs.openstack.org/93/663893/2/check/openstack-tox-py37/7a9c6dc/ : SUCCESS in 3m 12s\n- openstack-tox-docs http://logs.openstack.org/93/663893/2/check/openstack-tox-docs/57a3720/html/ : SUCCESS in 2m 57s\n- kolla-build-ubuntu-binary http://logs.openstack.org/93/663893/2/check/kolla-build-ubuntu-binary/35850bc/ : SUCCESS in 30m 08s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/93/663893/2/check/kolla-build-ubuntu-source/c93f906/ : SUCCESS in 34m 20s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/93/663893/2/check/kolla-ansible-ubuntu-source/cbf7e5e/ : SUCCESS in 43m 52s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/93/663893/2/check/kolla-ansible-ubuntu-binary/ae8c188/ : SUCCESS in 45m 31s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/93/663893/2/check/kolla-ansible-ubuntu-source-upgrade/8cd2783/ : SUCCESS in 1h 02m 53s","accounts_in_message":[],"_revision_number":2},{"id":"56d7a608bb23f2f9c688c55b6fae9e27626f5c0a","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-07 16:00:04.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"f94cda3522d4abc8b13959e747aaa2b4d344ae8d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-07 16:58:22.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"becbd4e0129a7771ac199cf01b9196042e519d86","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-07 17:40:19.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/93/663893/2/openstack-check/tripleo-build-containers-fedora-28-master/f6e4569/ : SUCCESS in 41m 11s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/93/663893/2/openstack-check/tripleo-build-containers-fedora-28-stein/a83fcae/ : SUCCESS in 35m 12s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"00389091eb2eafea624cd2149b0229465905976e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 18:54:37.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/93/663893/2/check/kolla-build-centos-binary/80fa855/ : SUCCESS in 43m 09s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/93/663893/2/check/kolla-build-centos-source/f4f8d30/ : SUCCESS in 49m 19s\n- kolla-ansible-centos-source http://logs.openstack.org/93/663893/2/check/kolla-ansible-centos-source/d38a1a1/ : SUCCESS in 1h 30m 08s\n- kolla-ansible-centos-binary http://logs.openstack.org/93/663893/2/check/kolla-ansible-centos-binary/d83b6c4/ : SUCCESS in 1h 04m 46s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/93/663893/2/check/kolla-ansible-centos-source-upgrade/c3bd2e2/ : SUCCESS in 1h 19m 55s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/93/663893/2/check/tripleo-build-containers-centos-7/d58b4b7/ : SUCCESS in 49m 39s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/93/663893/2/check/kolla-build-oraclelinux-binary/a30d48f/ : SUCCESS in 39m 59s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/93/663893/2/check/kolla-build-oraclelinux-source/f012efd/ : SUCCESS in 52m 36s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/93/663893/2/check/kolla-ansible-oraclelinux-source/8f03fb3/ : SUCCESS in 1h 34m 33s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/93/663893/2/check/kolla-ansible-oraclelinux-binary/d235b3a/ : SUCCESS in 1h 10m 23s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/93/663893/2/check/openstack-tox-lower-constraints/05fefc5/ : SUCCESS in 4m 02s\n- openstack-tox-pep8 http://logs.openstack.org/93/663893/2/check/openstack-tox-pep8/df648f7/ : SUCCESS in 3m 12s\n- openstack-tox-py27 http://logs.openstack.org/93/663893/2/check/openstack-tox-py27/2f4e72d/ : SUCCESS in 3m 19s\n- openstack-tox-py36 http://logs.openstack.org/93/663893/2/check/openstack-tox-py36/c0873a1/ : SUCCESS in 3m 00s\n- openstack-tox-py37 http://logs.openstack.org/93/663893/2/check/openstack-tox-py37/29b6e64/ : SUCCESS in 3m 22s\n- openstack-tox-docs http://logs.openstack.org/93/663893/2/check/openstack-tox-docs/8f9762d/html/ : SUCCESS in 2m 31s\n- kolla-build-ubuntu-binary http://logs.openstack.org/93/663893/2/check/kolla-build-ubuntu-binary/c723506/ : SUCCESS in 31m 09s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/93/663893/2/check/kolla-build-ubuntu-source/3a51287/ : SUCCESS in 34m 16s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/93/663893/2/check/kolla-ansible-ubuntu-source/09eb619/ : SUCCESS in 45m 41s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/93/663893/2/check/kolla-ansible-ubuntu-binary/2d098ed/ : SUCCESS in 52m 21s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/93/663893/2/check/kolla-ansible-ubuntu-source-upgrade/3a4fab7/ : SUCCESS in 53m 15s","accounts_in_message":[],"_revision_number":2},{"id":"51eeb67c850b6b333bc62f86c18788351ff8b346","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-07 18:58:05.000000000","message":"Patch Set 2:\n\nrecheck\n\nfinal one","accounts_in_message":[],"_revision_number":2},{"id":"c4fe3e16c64ee146b667813692bff3100a8e03cf","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-07 19:37:30.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/93/663893/2/openstack-check/tripleo-build-containers-fedora-28-master/a2877f6/ : SUCCESS in 37m 56s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/93/663893/2/openstack-check/tripleo-build-containers-fedora-28-stein/8b7eb4a/ : SUCCESS in 35m 16s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3f93ede4fffe75fcf9efdd5b4ac64dae07c247f8","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-06-07 21:27:43.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"7f61c70da1b444ffbc6a1069df36f744ef7817bf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 22:45:22.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/93/663893/2/check/kolla-build-centos-binary/b2b6e4c/ : SUCCESS in 39m 49s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/93/663893/2/check/kolla-build-centos-source/d3741ac/ : SUCCESS in 36m 37s\n- kolla-ansible-centos-source http://logs.openstack.org/93/663893/2/check/kolla-ansible-centos-source/a4af160/ : SUCCESS in 1h 09m 08s\n- kolla-ansible-centos-binary http://logs.openstack.org/93/663893/2/check/kolla-ansible-centos-binary/ee9c7e2/ : SUCCESS in 1h 07m 25s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/93/663893/2/check/kolla-ansible-centos-source-upgrade/e9fc14e/ : SUCCESS in 1h 21m 32s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/93/663893/2/check/tripleo-build-containers-centos-7/2e11eed/ : SUCCESS in 46m 58s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/93/663893/2/check/kolla-build-oraclelinux-binary/303c1ba/ : SUCCESS in 52m 14s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/93/663893/2/check/kolla-build-oraclelinux-source/f153b50/ : SUCCESS in 52m 40s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/93/663893/2/check/kolla-ansible-oraclelinux-source/a32c871/ : SUCCESS in 1h 19m 14s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/93/663893/2/check/kolla-ansible-oraclelinux-binary/116b5fd/ : SUCCESS in 1h 13m 02s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/93/663893/2/check/openstack-tox-lower-constraints/6f3e1be/ : SUCCESS in 6m 51s\n- openstack-tox-pep8 http://logs.openstack.org/93/663893/2/check/openstack-tox-pep8/a6c2ada/ : SUCCESS in 4m 03s\n- openstack-tox-py27 http://logs.openstack.org/93/663893/2/check/openstack-tox-py27/a1272ca/ : SUCCESS in 5m 24s\n- openstack-tox-py36 http://logs.openstack.org/93/663893/2/check/openstack-tox-py36/8ad0506/ : SUCCESS in 4m 06s\n- openstack-tox-py37 http://logs.openstack.org/93/663893/2/check/openstack-tox-py37/2b121ea/ : SUCCESS in 4m 35s\n- openstack-tox-docs http://logs.openstack.org/93/663893/2/check/openstack-tox-docs/b37c30c/html/ : SUCCESS in 3m 36s\n- kolla-build-ubuntu-binary http://logs.openstack.org/93/663893/2/check/kolla-build-ubuntu-binary/e3b6a97/ : SUCCESS in 26m 17s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/93/663893/2/check/kolla-build-ubuntu-source/ceff78c/ : SUCCESS in 29m 53s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/93/663893/2/check/kolla-ansible-ubuntu-source/c07cb5f/ : SUCCESS in 48m 09s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/93/663893/2/check/kolla-ansible-ubuntu-binary/67c0fd4/ : SUCCESS in 44m 00s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/93/663893/2/check/kolla-ansible-ubuntu-source-upgrade/2eeb08c/ : SUCCESS in 51m 55s","accounts_in_message":[],"_revision_number":2},{"id":"77a47706a5b0fe9d3d4651e9380792f8987c589d","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-08 07:35:23.000000000","message":"Patch Set 2:\n\nrecheck\n\none more because the extended approach failed and let\u0027s keep CI busy :-)","accounts_in_message":[],"_revision_number":2},{"id":"3b18e83df560c917d04ebafb5dc5e99be84b409a","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-08 07:36:05.000000000","message":"Patch Set 2:\n\nProbable candidate to backport to all supported releases.","accounts_in_message":[],"_revision_number":2},{"id":"79d282472c5310d1c5975f9a5cb486f4a031cbbf","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-08 08:14:14.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/93/663893/2/openstack-check/tripleo-build-containers-fedora-28-master/082de2c/ : SUCCESS in 37m 17s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/93/663893/2/openstack-check/tripleo-build-containers-fedora-28-stein/c543763/ : SUCCESS in 37m 33s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"74affc41518f39b6382c0bfc9c6eac22fe909450","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-08 09:18:05.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/93/663893/2/check/kolla-build-centos-binary/e4f2a37/ : SUCCESS in 37m 03s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/93/663893/2/check/kolla-build-centos-source/9d54a5d/ : SUCCESS in 40m 57s\n- kolla-ansible-centos-source http://logs.openstack.org/93/663893/2/check/kolla-ansible-centos-source/e7c662a/ : SUCCESS in 1h 05m 09s\n- kolla-ansible-centos-binary http://logs.openstack.org/93/663893/2/check/kolla-ansible-centos-binary/64afb4f/ : SUCCESS in 1h 01m 30s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/93/663893/2/check/kolla-ansible-centos-source-upgrade/c2e1b01/ : SUCCESS in 1h 14m 24s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/93/663893/2/check/tripleo-build-containers-centos-7/1193b3e/ : SUCCESS in 39m 39s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/93/663893/2/check/kolla-build-oraclelinux-binary/58c819a/ : SUCCESS in 44m 15s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/93/663893/2/check/kolla-build-oraclelinux-source/3a1f51e/ : SUCCESS in 48m 19s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/93/663893/2/check/kolla-ansible-oraclelinux-source/e9402fd/ : SUCCESS in 1h 11m 47s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/93/663893/2/check/kolla-ansible-oraclelinux-binary/93fbdb3/ : SUCCESS in 1h 00m 44s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/93/663893/2/check/openstack-tox-lower-constraints/c9f8504/ : SUCCESS in 4m 03s\n- openstack-tox-pep8 http://logs.openstack.org/93/663893/2/check/openstack-tox-pep8/bc8bf56/ : SUCCESS in 4m 30s\n- openstack-tox-py27 http://logs.openstack.org/93/663893/2/check/openstack-tox-py27/643e72b/ : SUCCESS in 3m 06s\n- openstack-tox-py36 http://logs.openstack.org/93/663893/2/check/openstack-tox-py36/edcf05a/ : SUCCESS in 4m 00s\n- openstack-tox-py37 http://logs.openstack.org/93/663893/2/check/openstack-tox-py37/ec0a945/ : SUCCESS in 3m 32s\n- openstack-tox-docs http://logs.openstack.org/93/663893/2/check/openstack-tox-docs/ba45f33/html/ : SUCCESS in 3m 04s\n- kolla-build-ubuntu-binary http://logs.openstack.org/93/663893/2/check/kolla-build-ubuntu-binary/c8a220f/ : SUCCESS in 23m 09s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/93/663893/2/check/kolla-build-ubuntu-source/918b770/ : SUCCESS in 30m 30s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/93/663893/2/check/kolla-ansible-ubuntu-source/dc4d3cc/ : SUCCESS in 44m 43s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/93/663893/2/check/kolla-ansible-ubuntu-binary/7f2d856/ : SUCCESS in 46m 30s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/93/663893/2/check/kolla-ansible-ubuntu-source-upgrade/5b7e854/ : FAILURE in 53m 38s","accounts_in_message":[],"_revision_number":2},{"id":"d95d0e0126996f90d776f3030e1ae7bed3bcbb21","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-08 09:22:55.000000000","message":"Patch Set 2:\n\nrecheck\n\nneither Ubuntu nor kolla-ansible touched by this change\n\nkolla-ansible-ubuntu-source-upgrade: tests/test-openstack.sh non-zero return code","accounts_in_message":[],"_revision_number":2},{"id":"ca052108c6d1afc4995b64a4fdc67d0cc8062b94","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-08 09:50:27.000000000","message":"Patch Set 2:\n\nrelevant log entries for failure look like random problem with connectivity: http://paste.openstack.org/show/752649/","accounts_in_message":[],"_revision_number":2},{"id":"8ccbf291468f8202c5fd4dbefd377bdbac3eb666","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-08 10:01:48.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/93/663893/2/openstack-check/tripleo-build-containers-fedora-28-master/35a67c7/ : SUCCESS in 35m 40s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/93/663893/2/openstack-check/tripleo-build-containers-fedora-28-stein/8668e39/ : SUCCESS in 37m 34s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9edadd37f09834b9ef2f487c51bfc023b247f70c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-08 10:48:20.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/93/663893/2/check/kolla-build-centos-binary/84db4d0/ : SUCCESS in 40m 04s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/93/663893/2/check/kolla-build-centos-source/78d87f0/ : SUCCESS in 50m 18s\n- kolla-ansible-centos-source http://logs.openstack.org/93/663893/2/check/kolla-ansible-centos-source/b67f24e/ : SUCCESS in 1h 19m 55s\n- kolla-ansible-centos-binary http://logs.openstack.org/93/663893/2/check/kolla-ansible-centos-binary/a2b77ac/ : SUCCESS in 1h 07m 46s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/93/663893/2/check/kolla-ansible-centos-source-upgrade/6d6276d/ : SUCCESS in 1h 16m 16s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/93/663893/2/check/tripleo-build-containers-centos-7/296a09f/ : SUCCESS in 41m 27s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/93/663893/2/check/kolla-build-oraclelinux-binary/dbad82b/ : SUCCESS in 46m 05s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/93/663893/2/check/kolla-build-oraclelinux-source/c356804/ : SUCCESS in 1h 05m 05s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/93/663893/2/check/kolla-ansible-oraclelinux-source/37717cb/ : SUCCESS in 1h 08m 49s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/93/663893/2/check/kolla-ansible-oraclelinux-binary/29de4bc/ : SUCCESS in 57m 09s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/93/663893/2/check/openstack-tox-lower-constraints/eb9902b/ : SUCCESS in 3m 43s\n- openstack-tox-pep8 http://logs.openstack.org/93/663893/2/check/openstack-tox-pep8/73d273d/ : SUCCESS in 2m 37s\n- openstack-tox-py27 http://logs.openstack.org/93/663893/2/check/openstack-tox-py27/4b9e1cf/ : SUCCESS in 2m 45s\n- openstack-tox-py36 http://logs.openstack.org/93/663893/2/check/openstack-tox-py36/4ce2866/ : SUCCESS in 2m 41s\n- openstack-tox-py37 http://logs.openstack.org/93/663893/2/check/openstack-tox-py37/f66d2f0/ : SUCCESS in 3m 04s\n- openstack-tox-docs http://logs.openstack.org/93/663893/2/check/openstack-tox-docs/5d541b7/html/ : SUCCESS in 2m 25s\n- kolla-build-ubuntu-binary http://logs.openstack.org/93/663893/2/check/kolla-build-ubuntu-binary/ceb02b6/ : SUCCESS in 25m 27s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/93/663893/2/check/kolla-build-ubuntu-source/6867d07/ : SUCCESS in 27m 06s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/93/663893/2/check/kolla-ansible-ubuntu-source/ce8a4c8/ : SUCCESS in 43m 36s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/93/663893/2/check/kolla-ansible-ubuntu-binary/682638f/ : SUCCESS in 44m 46s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/93/663893/2/check/kolla-ansible-ubuntu-source-upgrade/70c7219/ : SUCCESS in 47m 06s","accounts_in_message":[],"_revision_number":2},{"id":"b74c752faa9d3fafc672343b2e0f057f0a8fbf04","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-08 12:30:28.000000000","message":"Patch Set 2:\n\nOK, done with checks.","accounts_in_message":[],"_revision_number":2},{"id":"18e03a634005eda7e26472300afa51e8a790ce9a","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-08 16:55:06.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"0764584435b6a2ef1a89025e0d7b312325b26ce6","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-08 17:43:34.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/93/663893/3/openstack-check/tripleo-build-containers-fedora-28-master/6b1e451/ : SUCCESS in 45m 50s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/93/663893/3/openstack-check/tripleo-build-containers-fedora-28-stein/cc51994/ : SUCCESS in 45m 08s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"43238e8a49c6eec8de8ab154811bf5546fde8ec1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-08 18:44:12.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/93/663893/3/check/kolla-build-centos-binary/ca58382/ : SUCCESS in 47m 42s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/93/663893/3/check/kolla-build-centos-source/f1a3bb7/ : SUCCESS in 47m 40s\n- kolla-ansible-centos-source http://logs.openstack.org/93/663893/3/check/kolla-ansible-centos-source/31f44b2/ : SUCCESS in 1h 07m 19s\n- kolla-ansible-centos-binary http://logs.openstack.org/93/663893/3/check/kolla-ansible-centos-binary/f447c37/ : SUCCESS in 1h 03m 54s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/93/663893/3/check/kolla-ansible-centos-source-upgrade/23ed5b0/ : SUCCESS in 1h 15m 05s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/93/663893/3/check/tripleo-build-containers-centos-7/2023c52/ : SUCCESS in 41m 59s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/93/663893/3/check/kolla-build-oraclelinux-binary/0ccb815/ : SUCCESS in 47m 59s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/93/663893/3/check/kolla-build-oraclelinux-source/cb12148/ : SUCCESS in 50m 23s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/93/663893/3/check/kolla-ansible-oraclelinux-source/550bc50/ : SUCCESS in 1h 10m 46s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/93/663893/3/check/kolla-ansible-oraclelinux-binary/beded7a/ : SUCCESS in 1h 10m 54s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/93/663893/3/check/openstack-tox-lower-constraints/90e093c/ : SUCCESS in 3m 16s\n- openstack-tox-pep8 http://logs.openstack.org/93/663893/3/check/openstack-tox-pep8/bdf5a16/ : SUCCESS in 2m 47s\n- openstack-tox-py27 http://logs.openstack.org/93/663893/3/check/openstack-tox-py27/43076b8/ : SUCCESS in 2m 42s\n- openstack-tox-py36 http://logs.openstack.org/93/663893/3/check/openstack-tox-py36/3e575a7/ : SUCCESS in 3m 15s\n- openstack-tox-py37 http://logs.openstack.org/93/663893/3/check/openstack-tox-py37/45a6c09/ : SUCCESS in 3m 40s\n- openstack-tox-docs http://logs.openstack.org/93/663893/3/check/openstack-tox-docs/f7506fe/html/ : SUCCESS in 2m 31s\n- kolla-build-ubuntu-binary http://logs.openstack.org/93/663893/3/check/kolla-build-ubuntu-binary/49b8bb8/ : SUCCESS in 25m 57s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/93/663893/3/check/kolla-build-ubuntu-source/5cfca54/ : SUCCESS in 26m 01s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/93/663893/3/check/kolla-ansible-ubuntu-source/b61b541/ : SUCCESS in 45m 01s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/93/663893/3/check/kolla-ansible-ubuntu-binary/8ff6274/ : SUCCESS in 45m 36s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/93/663893/3/check/kolla-ansible-ubuntu-source-upgrade/1b97d28/ : SUCCESS in 57m 57s","accounts_in_message":[],"_revision_number":3},{"id":"a6e3bc57707e9404f90278eed50d55842e04d814","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-08 18:58:22.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)\n\nI am not a fan of repos enabled in just one image but it makes sense here.","accounts_in_message":[],"_revision_number":3},{"id":"dd49c24cf9402f56030937a54d895e51d06cdb70","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-08 19:08:07.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"284babd5edaae30ff9d23af38fe0902dc7287fc6","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-10 08:10:32.000000000","message":"Patch Set 3: Code-Review+2\n\nThanks for sorting this.","accounts_in_message":[],"_revision_number":3},{"id":"b0d239cf179785805bbb26fd2d8621f00a0ae98c","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-06-10 08:14:38.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nThanks!","accounts_in_message":[],"_revision_number":3},{"id":"2bcbd79e5e7f48fc46c9816a03d3dcaae6d47ef7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 08:54:55.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"a8cc7433e67df155626cea3bbbacef35a72b9afd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 09:47:23.000000000","message":"Patch Set 3: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":3},{"id":"0cb62a94a2067f876cd27142a36d83fa7e89e7e9","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-10 12:23:45.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"8a4d7b2092987dd949f007c5076ee336c8e6bc8b","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-10 13:09:24.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/93/663893/3/openstack-check/tripleo-build-containers-fedora-28-master/195e3ef/ : SUCCESS in 44m 00s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/93/663893/3/openstack-check/tripleo-build-containers-fedora-28-stein/c1b14fa/ : SUCCESS in 43m 06s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"898e1ea83a048f10e5c8fe0c5196ac30ac73ef13","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 14:09:07.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/93/663893/3/check/kolla-build-centos-binary/8e160a2/ : SUCCESS in 28m 44s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/93/663893/3/check/kolla-build-centos-source/7ebd0f6/ : SUCCESS in 52m 07s\n- kolla-ansible-centos-source http://logs.openstack.org/93/663893/3/check/kolla-ansible-centos-source/58502b2/ : SUCCESS in 1h 04m 55s\n- kolla-ansible-centos-binary http://logs.openstack.org/93/663893/3/check/kolla-ansible-centos-binary/9a1eb67/ : SUCCESS in 1h 09m 29s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/93/663893/3/check/kolla-ansible-centos-source-upgrade/1624faf/ : SUCCESS in 1h 19m 42s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/93/663893/3/check/tripleo-build-containers-centos-7/553172c/ : SUCCESS in 43m 25s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/93/663893/3/check/kolla-build-oraclelinux-binary/bca04b2/ : SUCCESS in 34m 00s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/93/663893/3/check/kolla-build-oraclelinux-source/4c07939/ : SUCCESS in 50m 44s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/93/663893/3/check/kolla-ansible-oraclelinux-source/dd871cb/ : SUCCESS in 1h 10m 30s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/93/663893/3/check/kolla-ansible-oraclelinux-binary/4b452de/ : SUCCESS in 1h 11m 47s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/93/663893/3/check/openstack-tox-lower-constraints/c8cd61f/ : SUCCESS in 4m 40s\n- openstack-tox-pep8 http://logs.openstack.org/93/663893/3/check/openstack-tox-pep8/ddd00cc/ : SUCCESS in 3m 44s\n- openstack-tox-py27 http://logs.openstack.org/93/663893/3/check/openstack-tox-py27/7b8bddf/ : SUCCESS in 3m 11s\n- openstack-tox-py36 http://logs.openstack.org/93/663893/3/check/openstack-tox-py36/53c5af7/ : SUCCESS in 4m 19s\n- openstack-tox-py37 http://logs.openstack.org/93/663893/3/check/openstack-tox-py37/dc87835/ : SUCCESS in 4m 55s\n- openstack-tox-docs http://logs.openstack.org/93/663893/3/check/openstack-tox-docs/f643bda/html/ : SUCCESS in 3m 59s\n- kolla-build-ubuntu-binary http://logs.openstack.org/93/663893/3/check/kolla-build-ubuntu-binary/0b71941/ : SUCCESS in 24m 43s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/93/663893/3/check/kolla-build-ubuntu-source/d361f76/ : SUCCESS in 28m 25s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/93/663893/3/check/kolla-ansible-ubuntu-source/2de43ce/ : SUCCESS in 42m 28s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/93/663893/3/check/kolla-ansible-ubuntu-binary/025c799/ : SUCCESS in 43m 14s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/93/663893/3/check/kolla-ansible-ubuntu-source-upgrade/f6c5743/ : SUCCESS in 50m 55s","accounts_in_message":[],"_revision_number":3},{"id":"1ddb530fa7b18a0cda75c3f7eb465332e0cc3989","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 14:09:19.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"f749e0eda611065df97b35e09bf646bc215e4812","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 15:12:01.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source http://logs.openstack.org/93/663893/3/gate/kolla-build-centos-source/1c436ba/ : SUCCESS in 57m 32s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/93/663893/3/gate/tripleo-build-containers-centos-7/be0bed8/ : SUCCESS in 48m 25s\n- kolla-build-oraclelinux-source http://logs.openstack.org/93/663893/3/gate/kolla-build-oraclelinux-source/675f67b/ : SUCCESS in 50m 21s\n- openstack-tox-lower-constraints http://logs.openstack.org/93/663893/3/gate/openstack-tox-lower-constraints/7907d9d/ : SUCCESS in 4m 24s\n- openstack-tox-pep8 http://logs.openstack.org/93/663893/3/gate/openstack-tox-pep8/9486604/ : SUCCESS in 4m 29s\n- openstack-tox-py27 http://logs.openstack.org/93/663893/3/gate/openstack-tox-py27/935c5e3/ : SUCCESS in 3m 27s\n- openstack-tox-py36 http://logs.openstack.org/93/663893/3/gate/openstack-tox-py36/6d8f428/ : SUCCESS in 4m 20s\n- openstack-tox-py37 http://logs.openstack.org/93/663893/3/gate/openstack-tox-py37/6f87f4f/ : SUCCESS in 4m 35s\n- openstack-tox-docs http://logs.openstack.org/93/663893/3/gate/openstack-tox-docs/d6434c6/html/ : SUCCESS in 2m 41s\n- kolla-build-ubuntu-source http://logs.openstack.org/93/663893/3/gate/kolla-build-ubuntu-source/5c4618b/ : SUCCESS in 28m 30s","accounts_in_message":[],"_revision_number":3},{"id":"914142273f6cec942d96f51f039da74b9b571c65","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 15:12:01.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"e30c040ec49fc8e0723d5d5e23d5b518957659fe","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-11 14:24:40.000000000","message":"Patch Set 3:\n\nShould this be backported?","accounts_in_message":[],"_revision_number":3},{"id":"be49f9eac5a8869ef3298023608d1e0929610d79","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-11 19:58:20.000000000","message":"Patch Set 3:\n\nAs I mentioned, yeah, to Stein, to ease our CI pain. Though there is this another CI breaker right now which has to be merged first for check to pass... I mean this: https://review.opendev.org/664676/","accounts_in_message":[],"_revision_number":3},{"id":"34565910cc60d4a3e0566fd378e39e5d2c59e7b5","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-13 18:15:28.000000000","message":"Patch Set 3:\n\nCherry-pick awaiting merge of https://review.opendev.org/665199","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"de58339a0e1704ae1ade4e032a73bb4708ecfaf1","revisions":{"a908a5313a0d5364c15cd79654d8c1fa92c768c5":{"kind":"REWORK","_number":1,"created":"2019-06-07 10:56:59.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/93/663893/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/93/663893/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/663893/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/663893/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/663893/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/93/663893/1"}}},"commit":{"parents":[{"commit":"9cfd3a88aa1c07054b66495dbb069a0000a20738","subject":"Merge \"Ignore EM releases in version-check.py\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9cfd3a88aa1c07054b66495dbb069a0000a20738"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-06-07 10:44:30.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-06-07 10:56:40.000000000","tz":120},"subject":"Fix issues with RHEL-family-based Percona in Zuul and clean up a little bit","message":"Fix issues with RHEL-family-based Percona in Zuul and clean up a little bit\n\nPercona site has connectivity issues from OVH hosts (of Zuul).\nThis patch applies to RHEL family and makes it so that:\n  - the internal repo is actually used\n  - RPM GPG keys are downloaded from the actually used repo\n\nChange-Id: Id27831f6b404762f9e44fd6bd1c4702bf06b05f8\nSigned-off-by: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a908a5313a0d5364c15cd79654d8c1fa92c768c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a908a5313a0d5364c15cd79654d8c1fa92c768c5"}]},"branch":"refs/heads/master"},"e235a7433598a10f3ced173beadd49d21d7dd0e3":{"kind":"REWORK","_number":2,"created":"2019-06-07 11:27:03.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/93/663893/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/93/663893/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/663893/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/663893/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/663893/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/93/663893/2"}}},"commit":{"parents":[{"commit":"9cfd3a88aa1c07054b66495dbb069a0000a20738","subject":"Merge \"Ignore EM releases in version-check.py\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9cfd3a88aa1c07054b66495dbb069a0000a20738"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-06-07 10:44:30.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-06-07 11:26:26.000000000","tz":120},"subject":"Fix issues with RHEL-family-based Percona in Zuul and clean up a little bit","message":"Fix issues with RHEL-family-based Percona in Zuul and clean up a little bit\n\nPercona site has connectivity issues from OVH hosts (of Zuul).\nThis patch applies to RHEL family and makes it so that:\n  - the internal repo is actually used\n  - RPM GPG keys are downloaded from the actually used repo\n\nChange-Id: Id27831f6b404762f9e44fd6bd1c4702bf06b05f8\nSigned-off-by: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e235a7433598a10f3ced173beadd49d21d7dd0e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e235a7433598a10f3ced173beadd49d21d7dd0e3"}]},"branch":"refs/heads/master"},"de58339a0e1704ae1ade4e032a73bb4708ecfaf1":{"kind":"REWORK","_number":3,"created":"2019-06-08 16:55:06.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/93/663893/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/93/663893/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/663893/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/663893/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/663893/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/93/663893/3"}}},"commit":{"parents":[{"commit":"5859180c2c78c62defca37446722b5f15fdbefb7","subject":"Debian/Ubuntu: allow to use unsigned repositories","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5859180c2c78c62defca37446722b5f15fdbefb7"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-06-07 10:44:30.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-06-08 16:54:19.000000000","tz":120},"subject":"Fix issues with Percona in Zuul and clean up a little bit","message":"Fix issues with Percona in Zuul and clean up a little bit\n\nPercona site has connectivity issues from OVH hosts (of Zuul).\nThis patch makes it so that:\n  - the internal repo is actually used\n  - RPM GPG keys are downloaded from the actually used repo\n  - Ubuntu mimics the RHEL behavior (i.e. the Percona repo is\n    enabled only in the xtrabackup image)\n    extra benefit: no need for APT preferences for Percona repo\n    to spare MariaDB deps\n\nChange-Id: Id27831f6b404762f9e44fd6bd1c4702bf06b05f8\nSigned-off-by: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/de58339a0e1704ae1ade4e032a73bb4708ecfaf1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/de58339a0e1704ae1ade4e032a73bb4708ecfaf1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
