)]}'
{"id":"openstack%2Fkolla~663942","triplet_id":"openstack%2Fkolla~master~I249b78cd263c97c4a9223523c77f8c325f3e649a","project":"openstack/kolla","branch":"master","topic":"remove-crane","hashtags":[],"change_id":"I249b78cd263c97c4a9223523c77f8c325f3e649a","subject":"Remove crane image","status":"MERGED","created":"2019-06-07 13:32:27.000000000","updated":"2019-06-18 16:03:49.000000000","submitted":"2019-06-18 16:03:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"663942-1560873829867-b95397b7","meta_rev_id":"e125e05b64997ca36da14d28421fa99a806c7125","_number":663942,"virtual_id_number":663942,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":2,"date":"2019-06-18 16:03:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-06-18 14:02:15.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"recommended":{"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},"all":[{"value":1,"date":"2019-06-10 15:57:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},{"value":2,"date":"2019-06-18 10:45:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2019-06-14 20:35:15.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2019-06-07 21:27:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2019-06-15 08:08:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},{"value":1,"date":"2019-06-18 10:45:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-07 21:27:54.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-06-10 15:57:40.000000000","updated_by":{"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},"reviewer":{"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},"state":"REVIEWER"},{"updated":"2019-06-14 20:35:15.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-06-15 08:08:08.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-06-18 10:45:20.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2019-06-18 14:02:15.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-06-18 16:03:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4e646df87da78f88aa3dc3d96b27f1fa13dbeed3","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-07 13:32:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"037db13da88dda161cbed90331a1c889cdb8801f","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-07 14:19:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/42/663942/1/openstack-check/tripleo-build-containers-fedora-28-master/851d3c7/ : SUCCESS in 39m 34s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/42/663942/1/openstack-check/tripleo-build-containers-fedora-28-stein/437a1d6/ : SUCCESS in 40m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6f4d46ab08313628fc867b496dfe7eced4e9b8f3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 15:18:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/42/663942/1/check/kolla-build-centos-binary/83ed556/ : SUCCESS in 47m 46s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/42/663942/1/check/kolla-build-centos-source/55afc25/ : SUCCESS in 47m 59s\n- kolla-ansible-centos-source http://logs.openstack.org/42/663942/1/check/kolla-ansible-centos-source/402f128/ : SUCCESS in 1h 16m 26s\n- kolla-ansible-centos-binary http://logs.openstack.org/42/663942/1/check/kolla-ansible-centos-binary/0744b92/ : SUCCESS in 1h 01m 57s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/42/663942/1/check/kolla-ansible-centos-source-upgrade/04f0e12/ : SUCCESS in 1h 38m 05s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/42/663942/1/check/tripleo-build-containers-centos-7/1f8703d/ : SUCCESS in 50m 01s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/42/663942/1/check/kolla-build-oraclelinux-binary/2be912f/ : SUCCESS in 49m 11s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/42/663942/1/check/kolla-build-oraclelinux-source/9648daf/ : SUCCESS in 50m 32s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/42/663942/1/check/kolla-ansible-oraclelinux-source/4218195/ : SUCCESS in 1h 08m 15s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/42/663942/1/check/kolla-ansible-oraclelinux-binary/f7668f2/ : SUCCESS in 1h 10m 50s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/42/663942/1/check/openstack-tox-lower-constraints/f2289a4/ : SUCCESS in 4m 23s\n- openstack-tox-pep8 http://logs.openstack.org/42/663942/1/check/openstack-tox-pep8/3f6a767/ : SUCCESS in 3m 25s\n- openstack-tox-py27 http://logs.openstack.org/42/663942/1/check/openstack-tox-py27/afd41a3/ : SUCCESS in 3m 56s\n- openstack-tox-py36 http://logs.openstack.org/42/663942/1/check/openstack-tox-py36/7bbb9ad/ : SUCCESS in 4m 10s\n- openstack-tox-py37 http://logs.openstack.org/42/663942/1/check/openstack-tox-py37/9e5bec7/ : SUCCESS in 4m 03s\n- openstack-tox-docs http://logs.openstack.org/42/663942/1/check/openstack-tox-docs/c0c498b/html/ : SUCCESS in 3m 11s\n- build-openstack-releasenotes http://logs.openstack.org/42/663942/1/check/build-openstack-releasenotes/a506dab/html/ : SUCCESS in 4m 04s\n- kolla-build-ubuntu-binary http://logs.openstack.org/42/663942/1/check/kolla-build-ubuntu-binary/84176ae/ : SUCCESS in 54m 48s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/42/663942/1/check/kolla-build-ubuntu-source/274ad66/ : SUCCESS in 28m 43s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/42/663942/1/check/kolla-ansible-ubuntu-source/93894e7/ : SUCCESS in 45m 14s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/42/663942/1/check/kolla-ansible-ubuntu-binary/cb2823d/ : SUCCESS in 45m 14s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/42/663942/1/check/kolla-ansible-ubuntu-source-upgrade/e2157b6/ : SUCCESS in 56m 07s","accounts_in_message":[],"_revision_number":1},{"id":"2b2d8a97047893667d463456e53e50f3b1beb45f","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-06-07 21:27:54.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"0af02e65f2f85a65941d712cd5218899d3549cdb","author":{"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},"date":"2019-06-10 15:57:40.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"359c892f00990299dd1b63d92c7bbc50e099a715","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-14 20:35:15.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d499061232cf116fcfe29b6769b49788faac7229","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-15 08:08:08.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"ed4be1b4fcb1e7d44d5944981ddf602b939036e4","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-06-18 10:45:20.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e049fcf32a21e264b2fb7baf60e58c06fa45d4b6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 10:46:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f2627ed281f97194a04fead825c7ac11fb447580","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 12:57:38.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-source http://logs.openstack.org/42/663942/1/gate/kolla-build-centos-source/9b34a20/ : SUCCESS in 50m 11s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/42/663942/1/gate/tripleo-build-containers-centos-7/d645a70/ : SUCCESS in 39m 31s\n- kolla-build-oraclelinux-source http://logs.openstack.org/42/663942/1/gate/kolla-build-oraclelinux-source/6f68272/ : SUCCESS in 55m 46s\n- openstack-tox-lower-constraints http://logs.openstack.org/42/663942/1/gate/openstack-tox-lower-constraints/d4cdac8/ : SUCCESS in 3m 27s\n- openstack-tox-pep8 http://logs.openstack.org/42/663942/1/gate/openstack-tox-pep8/c5d7f42/ : RETRY_LIMIT in 30m 56s\n- openstack-tox-py27 http://logs.openstack.org/42/663942/1/gate/openstack-tox-py27/245b76d/ : SUCCESS in 2m 44s\n- openstack-tox-py36 http://logs.openstack.org/42/663942/1/gate/openstack-tox-py36/68a2c0c/ : SUCCESS in 3m 27s\n- openstack-tox-py37 http://logs.openstack.org/42/663942/1/gate/openstack-tox-py37/fe73920/ : SUCCESS in 5m 19s\n- openstack-tox-docs http://logs.openstack.org/42/663942/1/gate/openstack-tox-docs/11c0812/html/ : SUCCESS in 2m 59s\n- build-openstack-releasenotes http://logs.openstack.org/42/663942/1/gate/build-openstack-releasenotes/b1975d1/html/ : SUCCESS in 3m 48s\n- kolla-build-ubuntu-source http://logs.openstack.org/42/663942/1/gate/kolla-build-ubuntu-source/f08bcb6/ : SUCCESS in 32m 11s","accounts_in_message":[],"_revision_number":1},{"id":"c0a86703417fc82159f33c66bb69332e331acaef","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-06-18 13:17:09.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6e9c0bc83eaf84926228801e4f60f7a9f1dace7a","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-18 14:02:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/42/663942/1/openstack-check/tripleo-build-containers-fedora-28-master/b519346/ : SUCCESS in 34m 04s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/42/663942/1/openstack-check/tripleo-build-containers-fedora-28-stein/b240b95/ : SUCCESS in 31m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ebb51f3a8a50ce93f849e36632ac7cad48d17024","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 14:48:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/42/663942/1/check/kolla-build-centos-binary/36c172e/ : RETRY_LIMIT in 36m 14s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/42/663942/1/check/kolla-build-centos-source/3f8659c/ : SUCCESS in 48m 09s\n- kolla-ansible-centos-source http://logs.openstack.org/42/663942/1/check/kolla-ansible-centos-source/c45eed3/ : SUCCESS in 1h 09m 38s\n- kolla-ansible-centos-binary http://logs.openstack.org/42/663942/1/check/kolla-ansible-centos-binary/83ecd19/ : SUCCESS in 1h 08m 44s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/42/663942/1/check/kolla-ansible-centos-source-upgrade/69b33da/ : SUCCESS in 1h 22m 50s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/42/663942/1/check/tripleo-build-containers-centos-7/430965e/ : SUCCESS in 41m 55s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/42/663942/1/check/kolla-build-oraclelinux-binary/4398dce/ : SUCCESS in 50m 50s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/42/663942/1/check/kolla-build-oraclelinux-source/9b1b19b/ : SUCCESS in 36m 52s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/42/663942/1/check/kolla-ansible-oraclelinux-source/5f5994a/ : RETRY_LIMIT in 36m 22s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/42/663942/1/check/kolla-ansible-oraclelinux-binary/34fce6f/ : SUCCESS in 1h 18m 17s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/42/663942/1/check/openstack-tox-lower-constraints/a5857c4/ : SUCCESS in 3m 37s\n- openstack-tox-pep8 http://logs.openstack.org/42/663942/1/check/openstack-tox-pep8/157577c/ : SUCCESS in 3m 16s\n- openstack-tox-py27 http://logs.openstack.org/42/663942/1/check/openstack-tox-py27/4fb6788/ : SUCCESS in 4m 43s\n- openstack-tox-py36 http://logs.openstack.org/42/663942/1/check/openstack-tox-py36/3371fcc/ : SUCCESS in 3m 25s\n- openstack-tox-py37 http://logs.openstack.org/42/663942/1/check/openstack-tox-py37/e1fe72d/ : SUCCESS in 3m 55s\n- openstack-tox-docs http://logs.openstack.org/42/663942/1/check/openstack-tox-docs/d77781d/html/ : SUCCESS in 3m 24s\n- build-openstack-releasenotes http://logs.openstack.org/42/663942/1/check/build-openstack-releasenotes/f98db48/html/ : SUCCESS in 4m 32s\n- kolla-build-ubuntu-binary http://logs.openstack.org/42/663942/1/check/kolla-build-ubuntu-binary/f4eb4d1/ : SUCCESS in 24m 17s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/42/663942/1/check/kolla-build-ubuntu-source/8110bee/ : SUCCESS in 30m 52s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/42/663942/1/check/kolla-ansible-ubuntu-source/8a80534/ : SUCCESS in 47m 16s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/42/663942/1/check/kolla-ansible-ubuntu-binary/24dabc4/ : SUCCESS in 45m 45s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/42/663942/1/check/kolla-ansible-ubuntu-source-upgrade/f141d74/ : SUCCESS in 52m 56s","accounts_in_message":[],"_revision_number":1},{"id":"f878be3bfefcd95e0e8f2b271934babb056c7633","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 14:48:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"dffe233c16ec5c8acc4db2005e45774ed5213e90","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 16:03:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"cf96253f0fc07bbd03df061edb9c83755b94f139","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 16:03:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source http://logs.openstack.org/42/663942/1/gate/kolla-build-centos-source/d7d456f/ : SUCCESS in 54m 40s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/42/663942/1/gate/tripleo-build-containers-centos-7/d471bb2/ : SUCCESS in 51m 29s\n- kolla-build-oraclelinux-source http://logs.openstack.org/42/663942/1/gate/kolla-build-oraclelinux-source/ecac40f/ : SUCCESS in 50m 57s\n- openstack-tox-lower-constraints http://logs.openstack.org/42/663942/1/gate/openstack-tox-lower-constraints/a2be045/ : SUCCESS in 3m 42s\n- openstack-tox-pep8 http://logs.openstack.org/42/663942/1/gate/openstack-tox-pep8/48b2383/ : SUCCESS in 3m 19s\n- openstack-tox-py27 http://logs.openstack.org/42/663942/1/gate/openstack-tox-py27/b2a43e3/ : SUCCESS in 3m 34s\n- openstack-tox-py36 http://logs.openstack.org/42/663942/1/gate/openstack-tox-py36/9e2631a/ : SUCCESS in 4m 25s\n- openstack-tox-py37 http://logs.openstack.org/42/663942/1/gate/openstack-tox-py37/7c4589c/ : SUCCESS in 3m 47s\n- openstack-tox-docs http://logs.openstack.org/42/663942/1/gate/openstack-tox-docs/89a5ac4/html/ : SUCCESS in 3m 08s\n- build-openstack-releasenotes http://logs.openstack.org/42/663942/1/gate/build-openstack-releasenotes/ff687f7/html/ : SUCCESS in 3m 50s\n- kolla-build-ubuntu-source http://logs.openstack.org/42/663942/1/gate/kolla-build-ubuntu-source/5368a22/ : SUCCESS in 37m 02s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c63f817730a66761b45607afa201ebae300452c2","revisions":{"c63f817730a66761b45607afa201ebae300452c2":{"kind":"REWORK","_number":1,"created":"2019-06-07 13:32:27.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/42/663942/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/42/663942/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/663942/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/663942/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/663942/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/42/663942/1"}}},"commit":{"parents":[{"commit":"03ca8c63256549ab2f204e89d225abb8b9ed0e6b","subject":"Merge \"Blacklist bandit 1.6.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/03ca8c63256549ab2f204e89d225abb8b9ed0e6b"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-06-07 13:28:56.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-06-07 13:32:08.000000000","tz":60},"subject":"Remove crane image","message":"Remove crane image\n\nThis image was deprecated in the Stein cycle.\n\nChange-Id: I249b78cd263c97c4a9223523c77f8c325f3e649a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c63f817730a66761b45607afa201ebae300452c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c63f817730a66761b45607afa201ebae300452c2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
