)]}'
{"id":"openstack%2Fkolla~664012","triplet_id":"openstack%2Fkolla~stable%2Fstein~I0bff21384d88ea678608392de2db1ba418c96665","project":"openstack/kolla","branch":"stable/stein","topic":"bp/debian-ubuntu-python3-stable/stein","hashtags":[],"change_id":"I0bff21384d88ea678608392de2db1ba418c96665","subject":"Debian/Ubuntu: more python3 packages for binary","status":"MERGED","created":"2019-06-07 17:59:49.000000000","updated":"2019-06-18 09:57:57.000000000","submitted":"2019-06-18 09:57:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"664012-1560851877940-d473cd44","meta_rev_id":"22efa8077459cc6d19d883b29051309853bc7c36","_number":664012,"virtual_id_number":664012,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-06-18 09:57:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-06-07 18:43:22.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2019-06-18 08:40:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2019-06-14 20:30:02.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":1,"date":"2019-06-16 21:58:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2019-06-18 08:40:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2019-06-08 08:13:00.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-07 18:43:22.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-06-14 20:30:02.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-06-16 21:58:27.000000000","updated_by":{"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"},"reviewer":{"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"},"state":"REVIEWER"},{"updated":"2019-06-18 08:40:29.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2019-06-18 09:57:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9b046fa4d826a882c981fc4a365ef14c9663edd5","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-07 17:59:49.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"f801566e4e87fe66e4ec2bf972470415975e7bae","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-07 18:43:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/12/664012/1/openstack-check/tripleo-build-containers-fedora-28-master/b3ebdd6/ : SUCCESS in 39m 38s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/12/664012/1/openstack-check/tripleo-build-containers-fedora-28-stein/3dc4193/ : SUCCESS in 42m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f1e8fc7273dff18dfe1b3312c5ecf693870fc58b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 19:49:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/12/664012/1/check/kolla-build-centos-binary/14367f5/ : SUCCESS in 42m 10s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/12/664012/1/check/kolla-build-centos-source/4b82b11/ : SUCCESS in 47m 49s\n- kolla-ansible-centos-source http://logs.openstack.org/12/664012/1/check/kolla-ansible-centos-source/bd63c42/ : SUCCESS in 1h 08m 21s\n- kolla-ansible-centos-binary http://logs.openstack.org/12/664012/1/check/kolla-ansible-centos-binary/6787f0c/ : SUCCESS in 1h 05m 25s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/12/664012/1/check/kolla-ansible-centos-source-upgrade/b132d77/ : SUCCESS in 1h 14m 18s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/12/664012/1/check/tripleo-build-containers-centos-7/2a1ed1c/ : SUCCESS in 42m 04s\n- kolla-ansible-centos-source-ceph finger://ze12.openstack.org/52a579c746494d9d862f8925d93315b3 : RETRY_LIMIT in 46m 34s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/12/664012/1/check/kolla-build-oraclelinux-binary/dea7456/ : SUCCESS in 45m 46s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/12/664012/1/check/kolla-build-oraclelinux-source/783cf1d/ : SUCCESS in 54m 22s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/12/664012/1/check/kolla-ansible-oraclelinux-source/40e10d1/ : SUCCESS in 1h 21m 28s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/12/664012/1/check/kolla-ansible-oraclelinux-binary/4e4110c/ : SUCCESS in 1h 07m 59s (non-voting)\n- kolla-ansible-oraclelinux-source-ceph http://logs.openstack.org/12/664012/1/check/kolla-ansible-oraclelinux-source-ceph/630e90f/ : SUCCESS in 1h 30m 16s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/12/664012/1/check/openstack-tox-lower-constraints/2b4e08b/ : SUCCESS in 3m 55s\n- openstack-tox-pep8 http://logs.openstack.org/12/664012/1/check/openstack-tox-pep8/638e553/ : SUCCESS in 3m 23s\n- openstack-tox-py27 http://logs.openstack.org/12/664012/1/check/openstack-tox-py27/eb16a1a/ : SUCCESS in 2m 58s\n- openstack-tox-py36 http://logs.openstack.org/12/664012/1/check/openstack-tox-py36/54b275e/ : SUCCESS in 3m 42s\n- openstack-tox-py37 http://logs.openstack.org/12/664012/1/check/openstack-tox-py37/96937f2/ : SUCCESS in 3m 51s\n- openstack-tox-docs http://logs.openstack.org/12/664012/1/check/openstack-tox-docs/ea42e8a/html/ : SUCCESS in 3m 02s\n- kolla-build-ubuntu-binary http://logs.openstack.org/12/664012/1/check/kolla-build-ubuntu-binary/e7e06bd/ : SUCCESS in 22m 32s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/12/664012/1/check/kolla-build-ubuntu-source/8a26474/ : SUCCESS in 36m 46s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/12/664012/1/check/kolla-ansible-ubuntu-source/98a3e91/ : SUCCESS in 53m 34s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/12/664012/1/check/kolla-ansible-ubuntu-binary/9882a90/ : SUCCESS in 42m 51s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/12/664012/1/check/kolla-ansible-ubuntu-source-upgrade/01ca7f1/ : SUCCESS in 1h 00m 41s\n- kolla-ansible-ubuntu-source-ceph http://logs.openstack.org/12/664012/1/check/kolla-ansible-ubuntu-source-ceph/5c753c0/ : SUCCESS in 1h 13m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a24f7cb4dbd84ff8310826afd612b90f97912c34","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-08 08:13:00.000000000","message":"Patch Set 1: Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"2ec74328564e1fc2392f39d65d72d2c7741f73f9","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-14 20:30:02.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a42a19ed0f259dcbf383099b9fdf083f8a6f1cd2","author":{"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"},"date":"2019-06-16 21:58:27.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"1deba4b17d4bf7f454d30917d30dfdfb254604c0","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-06-18 08:40:29.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"83ab0da047e4fb83ea3c883ab0995b8a981f3bbd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 08:40:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f04979e9f6950b99660859ce8f6c16476cfd899f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 09:57:57.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"00ef3a1a5ad3ddd10085ec4519d301a2fcc5f7c4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 09:57:57.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source http://logs.openstack.org/12/664012/1/gate/kolla-build-centos-source/3f3f7d1/ : SUCCESS in 1h 00m 51s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/12/664012/1/gate/tripleo-build-containers-centos-7/f1bd71e/ : SUCCESS in 43m 22s\n- kolla-build-oraclelinux-source http://logs.openstack.org/12/664012/1/gate/kolla-build-oraclelinux-source/bb2fa07/ : SUCCESS in 59m 15s\n- openstack-tox-lower-constraints http://logs.openstack.org/12/664012/1/gate/openstack-tox-lower-constraints/ddc8744/ : SUCCESS in 4m 48s\n- openstack-tox-pep8 http://logs.openstack.org/12/664012/1/gate/openstack-tox-pep8/f611b97/ : SUCCESS in 3m 13s\n- openstack-tox-py27 http://logs.openstack.org/12/664012/1/gate/openstack-tox-py27/e7c3881/ : SUCCESS in 3m 38s\n- openstack-tox-py36 http://logs.openstack.org/12/664012/1/gate/openstack-tox-py36/71682be/ : SUCCESS in 3m 46s\n- openstack-tox-py37 http://logs.openstack.org/12/664012/1/gate/openstack-tox-py37/bec3a80/ : SUCCESS in 3m 22s\n- openstack-tox-docs http://logs.openstack.org/12/664012/1/gate/openstack-tox-docs/53be890/html/ : SUCCESS in 2m 55s\n- kolla-build-ubuntu-source http://logs.openstack.org/12/664012/1/gate/kolla-build-ubuntu-source/951fe36/ : SUCCESS in 32m 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b71f2be433568e6c71ea1d54a92dfa6a449dead6","revisions":{"b71f2be433568e6c71ea1d54a92dfa6a449dead6":{"kind":"REWORK","_number":1,"created":"2019-06-07 17:59:49.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/12/664012/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/12/664012/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/664012/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/664012/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/664012/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/12/664012/1"}}},"commit":{"parents":[{"commit":"c5d2e9683d8c26e382356a1c1b6d431e78e0f398","subject":"Switch to RDO Stein release on CentOS \u0026 Oracle","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c5d2e9683d8c26e382356a1c1b6d431e78e0f398"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-06-05 16:16:49.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-06-07 17:59:49.000000000","tz":0},"subject":"Debian/Ubuntu: more python3 packages for binary","message":"Debian/Ubuntu: more python3 packages for binary\n\nbackport: Stein\n\nDuring the switch to Stein UCA, we did not switch all packages to python\n3 for Debian/Ubuntu binary images. This change switches some more of\nthose packages.\n\nChange-Id: I0bff21384d88ea678608392de2db1ba418c96665\nCo-Authored-By: Marcin Juszkiewicz \u003cmarcin.juszkiewicz@linaro.org\u003e\n(cherry picked from commit aa59f0db6f88ed0a03169b473fdf6d541d34df9f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b71f2be433568e6c71ea1d54a92dfa6a449dead6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b71f2be433568e6c71ea1d54a92dfa6a449dead6"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
