)]}'
{"id":"openstack%2Fkolla~664026","triplet_id":"openstack%2Fkolla~master~I1fe045e63757ebf8e83d86778e435fabcd892f11","project":"openstack/kolla","branch":"master","topic":"clean-repos","hashtags":[],"change_id":"I1fe045e63757ebf8e83d86778e435fabcd892f11","subject":"Clean Zuul repo mirrors for RHEL family","status":"ABANDONED","created":"2019-06-07 18:34:01.000000000","updated":"2019-06-22 07:54:38.000000000","total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f7067ec2c434eeec89c610c8ed1d9e7b8896dd69","_number":664026,"virtual_id_number":664026,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":-1,"date":"2019-06-08 19:48:33.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":-1,"date":"2019-06-08 17:36:45.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":-1,"date":"2019-06-08 20:12:04.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"date":"2019-06-08 17:36:45.000000000","permitted_voting_range":{"min":-1,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"date":"2019-06-08 17:36:45.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-08 07:32:38.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-06-08 08:23:13.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2019-06-08 08:28:14.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-06-08 19:48:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fa37ab8125003f4b43a9e295ceb529d23afbd975","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-07 18:34:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f3d211615174f44b92c0fe8fbf11a81bb3b3ad17","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 19:51:51.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary finger://ze04.openstack.org/fde32f308223498f95b62d1f88c652b8 : RETRY_LIMIT in 29m 20s (non-voting)\n- kolla-build-centos-source finger://ze07.openstack.org/3614fffa4cf34f2f8032d4fafd925fc9 : RETRY_LIMIT in 30m 06s\n- kolla-ansible-centos-source finger://ze04.openstack.org/4a4fe43f41864f319f13fe94b4f2ed31 : RETRY_LIMIT in 26m 29s\n- kolla-ansible-centos-binary finger://ze02.openstack.org/acce252cfb5e4bd6a7bfa3a6a86c26d1 : RETRY_LIMIT in 34m 09s (non-voting)\n- kolla-ansible-centos-source-upgrade finger://ze01.openstack.org/0c4d4fd62ffb4949a4e6000e5861e2ea : RETRY_LIMIT in 34m 05s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/26/664026/1/check/kolla-build-oraclelinux-binary/91f0d4c/ : RETRY_LIMIT in 36m 19s (non-voting)\n- kolla-build-oraclelinux-source finger://ze02.openstack.org/3b08c8a209a24cd0ad328d613118ae5a : RETRY_LIMIT in 31m 27s\n- kolla-ansible-oraclelinux-source finger://ze01.openstack.org/1ddad3dac31745599999c37badfaf749 : RETRY_LIMIT in 34m 09s (non-voting)\n- kolla-ansible-oraclelinux-binary finger://ze05.openstack.org/0d4ee7cf6eb545dd8e0a0ae3520a017c : RETRY_LIMIT in 33m 48s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/26/664026/1/check/openstack-tox-lower-constraints/868156c/ : SUCCESS in 4m 11s\n- openstack-tox-pep8 http://logs.openstack.org/26/664026/1/check/openstack-tox-pep8/3e4356f/ : SUCCESS in 3m 28s\n- openstack-tox-py27 http://logs.openstack.org/26/664026/1/check/openstack-tox-py27/13b0822/ : SUCCESS in 4m 33s\n- openstack-tox-py36 http://logs.openstack.org/26/664026/1/check/openstack-tox-py36/e9d620e/ : SUCCESS in 4m 06s\n- openstack-tox-py37 http://logs.openstack.org/26/664026/1/check/openstack-tox-py37/6d14422/ : SUCCESS in 3m 53s\n- openstack-tox-docs http://logs.openstack.org/26/664026/1/check/openstack-tox-docs/db68796/html/ : SUCCESS in 3m 50s\n- kolla-build-ubuntu-binary http://logs.openstack.org/26/664026/1/check/kolla-build-ubuntu-binary/354f06a/ : SUCCESS in 25m 52s (non-voting)\n- kolla-build-ubuntu-source finger://ze01.openstack.org/1139af84795044cd9857e6d32eb722b5 : RETRY_LIMIT in 29m 36s\n- kolla-ansible-ubuntu-source finger://ze07.openstack.org/538dc91fe89c401e829e5b861390afae : RETRY_LIMIT in 32m 58s\n- kolla-ansible-ubuntu-binary finger://ze11.openstack.org/31e9df14e4724402995439014a5ba37a : RETRY_LIMIT in 32m 58s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade finger://ze10.openstack.org/cbc4a9f3f17a46f5bc96fd45b854128c : RETRY_LIMIT in 32m 51s","accounts_in_message":[],"_revision_number":1},{"id":"e416cc2d6a8a0f211e0cd3bdb295fdf9fae9582f","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-08 07:31:52.000000000","message":"Patch Set 1: Workflow-1\n\nZuul mirrors for CentOS and EPEL lack the GPG keys (probably due to filtering for major version 7). The issue affects only CentOS\u0027 because EPEL\u0027s key is installed via the package epel-release. Other mirrors (port 8080) seem to be verbatim. I would prefer to complete mirrors rather than hack the sources further.","accounts_in_message":[],"_revision_number":1},{"id":"75b3733621ba62dc9a422ff7063cfce18c847544","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-08 08:28:14.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b5d4253efd261fbb89e10fa18aef5428d1e5e1d0","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-08 08:37:08.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"155ec5123e69fe37ce230dd9030bd24661ad2f33","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-08 12:39:10.000000000","message":"Patch Set 1:\n\nI also wonder why I got RETRY_LIMIT instead of normal FAILURE and all the other checks seem to have been cancelled as well.","accounts_in_message":[],"_revision_number":1},{"id":"05ba3ec40d9373349a4c508222779d46c4af5c6c","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-08 12:40:37.000000000","message":"Patch Set 1:\n\nrecheck\n\nTo see how you fail today.","accounts_in_message":[],"_revision_number":1},{"id":"f31e6ac029f3c75667ff9e6ed09a4194940141a0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-08 14:21:52.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/26/664026/1/check/kolla-build-centos-binary/c7de3f2/ : SUCCESS in 37m 10s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/26/664026/1/check/kolla-build-centos-source/b034ff9/ : SUCCESS in 37m 13s\n- kolla-ansible-centos-source http://logs.openstack.org/26/664026/1/check/kolla-ansible-centos-source/ae3f6a5/ : SUCCESS in 59m 00s\n- kolla-ansible-centos-binary http://logs.openstack.org/26/664026/1/check/kolla-ansible-centos-binary/bac63e2/ : SUCCESS in 1h 27m 56s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/26/664026/1/check/kolla-ansible-centos-source-upgrade/df8017f/ : SUCCESS in 1h 36m 31s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/26/664026/1/check/kolla-build-oraclelinux-binary/b688dcd/ : FAILURE in 35m 02s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/26/664026/1/check/kolla-build-oraclelinux-source/ec38e15/ : FAILURE in 54m 12s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/26/664026/1/check/kolla-ansible-oraclelinux-source/60e7af8/ : SUCCESS in 1h 29m 18s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/26/664026/1/check/kolla-ansible-oraclelinux-binary/be475b6/ : SUCCESS in 1h 26m 39s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/26/664026/1/check/openstack-tox-lower-constraints/6667f63/ : SUCCESS in 3m 56s\n- openstack-tox-pep8 http://logs.openstack.org/26/664026/1/check/openstack-tox-pep8/9d8c907/ : SUCCESS in 4m 07s\n- openstack-tox-py27 http://logs.openstack.org/26/664026/1/check/openstack-tox-py27/4484925/ : SUCCESS in 3m 17s\n- openstack-tox-py36 http://logs.openstack.org/26/664026/1/check/openstack-tox-py36/e4d929d/ : SUCCESS in 4m 14s\n- openstack-tox-py37 http://logs.openstack.org/26/664026/1/check/openstack-tox-py37/d8f020b/ : SUCCESS in 3m 22s\n- openstack-tox-docs http://logs.openstack.org/26/664026/1/check/openstack-tox-docs/7fe7c78/html/ : SUCCESS in 3m 52s\n- kolla-build-ubuntu-binary http://logs.openstack.org/26/664026/1/check/kolla-build-ubuntu-binary/26c3d4b/ : SUCCESS in 22m 46s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/26/664026/1/check/kolla-build-ubuntu-source/e453cb0/ : FAILURE in 1h 01m 39s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/26/664026/1/check/kolla-ansible-ubuntu-source/5025ec7/ : SUCCESS in 43m 39s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/26/664026/1/check/kolla-ansible-ubuntu-binary/7fdd305/ : SUCCESS in 1h 11m 33s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/26/664026/1/check/kolla-ansible-ubuntu-source-upgrade/13ec5b1/ : SUCCESS in 51m 14s","accounts_in_message":[],"_revision_number":1},{"id":"d493d472a12b4aa360d77ff14b2ed3f5f2fd3f89","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-08 15:50:04.000000000","message":"Patch Set 1:\n\nOK, only Oracle fails due to lacking keys.\n\nCentOS images built because CentOS obviously ships their GPG keys in base image. :-)\n\nkolla-build-ubuntu-source FAILURE is due to unreachable Percona mirror","accounts_in_message":[],"_revision_number":1},{"id":"bec0735f1efa3f1ffcfcd460487792d2e05b9011","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-08 17:36:45.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"5a8f0909b7ff33ec935ae2affd1ee505f452580f","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-08 17:37:28.000000000","message":"Topic changed from fix-percona to clean-repos","accounts_in_message":[],"_revision_number":2},{"id":"d18e6fbc73b7f18d506db58c744092f4169b4762","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-08 18:43:45.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"dea67bb4c699367da2af05aca071ce1831c92ebd","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-08 19:01:07.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3e83ec32812fb16e6797bf7e44d066ddd21bc1db","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-08 19:48:33.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/26/664026/2/check/kolla-build-centos-binary/3c28f26/ : SUCCESS in 22m 57s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/26/664026/2/check/kolla-build-centos-source/a5466da/ : SUCCESS in 40m 06s\n- kolla-ansible-centos-source http://logs.openstack.org/26/664026/2/check/kolla-ansible-centos-source/dd14b66/ : SUCCESS in 1h 23m 35s\n- kolla-ansible-centos-binary http://logs.openstack.org/26/664026/2/check/kolla-ansible-centos-binary/60c491a/ : SUCCESS in 1h 08m 34s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/26/664026/2/check/kolla-ansible-centos-source-upgrade/1330f47/ : SUCCESS in 1h 29m 11s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/26/664026/2/check/kolla-build-oraclelinux-binary/dc80115/ : FAILURE in 30m 06s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/26/664026/2/check/kolla-build-oraclelinux-source/666a213/ : FAILURE in 38m 34s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/26/664026/2/check/kolla-ansible-oraclelinux-source/93c0f3a/ : SUCCESS in 1h 10m 04s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/26/664026/2/check/kolla-ansible-oraclelinux-binary/eafec52/ : SUCCESS in 1h 07m 58s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/26/664026/2/check/openstack-tox-lower-constraints/6a3d246/ : SUCCESS in 3m 23s\n- openstack-tox-pep8 http://logs.openstack.org/26/664026/2/check/openstack-tox-pep8/6cf7a62/ : SUCCESS in 3m 24s\n- openstack-tox-py27 http://logs.openstack.org/26/664026/2/check/openstack-tox-py27/8277217/ : SUCCESS in 3m 55s\n- openstack-tox-py36 http://logs.openstack.org/26/664026/2/check/openstack-tox-py36/b3258ee/ : SUCCESS in 2m 42s\n- openstack-tox-py37 http://logs.openstack.org/26/664026/2/check/openstack-tox-py37/769ddf5/ : SUCCESS in 3m 28s\n- openstack-tox-docs http://logs.openstack.org/26/664026/2/check/openstack-tox-docs/43248f2/html/ : SUCCESS in 2m 58s\n- kolla-build-ubuntu-binary http://logs.openstack.org/26/664026/2/check/kolla-build-ubuntu-binary/cecbeaa/ : SUCCESS in 22m 52s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/26/664026/2/check/kolla-build-ubuntu-source/9ab2323/ : SUCCESS in 30m 59s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/26/664026/2/check/kolla-ansible-ubuntu-source/48873da/ : SUCCESS in 45m 57s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/26/664026/2/check/kolla-ansible-ubuntu-binary/bd33b0e/ : SUCCESS in 52m 48s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/26/664026/2/check/kolla-ansible-ubuntu-source-upgrade/e31549f/ : SUCCESS in 54m 10s","accounts_in_message":[],"_revision_number":2},{"id":"7cffc50b2cf6832d1a77c2ac6372fd747f0f4a0c","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-08 20:12:04.000000000","message":"Patch Set 2: Workflow-1\n\nOnly planned failures now. Awaiting decision on mirror incompleteness handling.","accounts_in_message":[],"_revision_number":2},{"id":"f7067ec2c434eeec89c610c8ed1d9e7b8896dd69","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-22 07:54:38.000000000","message":"Abandoned\n\nIn favor of: https://review.opendev.org/664151","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d37764b74b46e8ef804029c952aa7ba8dce328c8","revisions":{"606172635e8b4d146d700e01b852680ff6b9480b":{"kind":"REWORK","_number":1,"created":"2019-06-07 18:34:01.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/26/664026/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/26/664026/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/26/664026/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/26/664026/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/26/664026/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/26/664026/1"}}},"commit":{"parents":[{"commit":"e235a7433598a10f3ced173beadd49d21d7dd0e3","subject":"Fix issues with RHEL-family-based Percona in Zuul and clean up a little bit","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e235a7433598a10f3ced173beadd49d21d7dd0e3"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-06-07 18:21:15.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-06-07 18:21:15.000000000","tz":120},"subject":"Clean Zuul repo mirrors for RHEL family","message":"Clean Zuul repo mirrors for RHEL family\n\n- unify RDO mirror definition with others\n- drop definitions of unused mirrors: grafana, elastic\n- clean up the process of mirror substitution:\n  - actually OracleLinux and CentOS do not differ here at all\n    and in fact never should (both are RHEL-based after all)\n  - use the more common style of sed d command\n  - uncomment baseurl globally\n\nChange-Id: I1fe045e63757ebf8e83d86778e435fabcd892f11\nSigned-off-by: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/606172635e8b4d146d700e01b852680ff6b9480b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/606172635e8b4d146d700e01b852680ff6b9480b"}]},"branch":"refs/heads/master"},"d37764b74b46e8ef804029c952aa7ba8dce328c8":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-06-08 17:36:45.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/26/664026/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/26/664026/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/26/664026/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/26/664026/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/26/664026/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/26/664026/2"}}},"commit":{"parents":[{"commit":"de58339a0e1704ae1ade4e032a73bb4708ecfaf1","subject":"Fix issues with Percona in Zuul and clean up a little bit","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/de58339a0e1704ae1ade4e032a73bb4708ecfaf1"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-06-07 18:21:15.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-06-08 17:34:27.000000000","tz":120},"subject":"Clean Zuul repo mirrors for RHEL family","message":"Clean Zuul repo mirrors for RHEL family\n\n- unify RDO mirror definition with others\n- drop definitions of unused mirrors: grafana, elastic\n- clean up the process of mirror substitution:\n  - actually OracleLinux and CentOS do not differ here at all\n    and in fact never should (both are RHEL-based after all)\n  - use the more common style of sed d command\n  - uncomment baseurl globally\n\nChange-Id: I1fe045e63757ebf8e83d86778e435fabcd892f11\nSigned-off-by: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d37764b74b46e8ef804029c952aa7ba8dce328c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d37764b74b46e8ef804029c952aa7ba8dce328c8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
