)]}'
{"id":"openstack%2Fkolla~664252","triplet_id":"openstack%2Fkolla~stable%2Fstein~Ie268bb875b774a9bafd04f8c27c63b2b35dee5c8","project":"openstack/kolla","branch":"stable/stein","topic":"tripleo-ceph-nautilus","hashtags":[],"change_id":"Ie268bb875b774a9bafd04f8c27c63b2b35dee5c8","subject":"Add centos_yum_repos_to_disable overridable in base","status":"ABANDONED","created":"2019-06-10 12:18:20.000000000","updated":"2019-06-22 07:50:50.000000000","total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"5bf40cf98672116a43c5ca41ff59f272b9885d29","_number":664252,"virtual_id_number":664252,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":-1,"date":"2019-06-17 14:38:08.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},{"date":"2019-06-17 14:08:57.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":-1,"date":"2019-06-17 18:44:28.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-10 15:28:45.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2019-06-15 12:13:13.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-06-15 12:13:16.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2019-06-15 12:14:40.000000000","updated_by":{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},"reviewer":{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},"state":"REVIEWER"},{"updated":"2019-06-15 19:07:24.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-06-17 14:08:57.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-06-17 14:38:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"db595565ee1d3493421e065fc48d8a855b99fc6a","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-10 12:18:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"98ff59d92f407bcd3a1f43683348c6b48ee91204","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-10 12:21:24.000000000","message":"Patch Set 1: Workflow-1\n\nLet Zuul test it.","accounts_in_message":[],"_revision_number":1},{"id":"92767698e9226374ebc64360f77306a3e82dbe38","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-10 13:06:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/52/664252/1/openstack-check/tripleo-build-containers-fedora-28-master/dee87bb/ : SUCCESS in 42m 22s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/52/664252/1/openstack-check/tripleo-build-containers-fedora-28-stein/71d2c8e/ : SUCCESS in 44m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5bb050b91418cc4746428186757478a8158f5bd4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 13:55:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/52/664252/1/check/kolla-build-centos-binary/3e37683/ : FAILURE in 27m 37s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/52/664252/1/check/kolla-build-centos-source/b6c2bd0/ : FAILURE in 38m 36s\n- kolla-ansible-centos-source http://logs.openstack.org/52/664252/1/check/kolla-ansible-centos-source/fb073c0/ : SUCCESS in 1h 12m 43s\n- kolla-ansible-centos-binary http://logs.openstack.org/52/664252/1/check/kolla-ansible-centos-binary/c012b4e/ : SUCCESS in 1h 03m 27s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/52/664252/1/check/kolla-ansible-centos-source-upgrade/23b99ba/ : SUCCESS in 1h 14m 26s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/52/664252/1/check/tripleo-build-containers-centos-7/9e327e4/ : SUCCESS in 45m 45s\n- kolla-ansible-centos-source-ceph http://logs.openstack.org/52/664252/1/check/kolla-ansible-centos-source-ceph/9007ca9/ : FAILURE in 32m 21s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/52/664252/1/check/kolla-build-oraclelinux-binary/a167f6d/ : FAILURE in 41m 36s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/52/664252/1/check/kolla-build-oraclelinux-source/ba9d1e4/ : FAILURE in 47m 32s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/52/664252/1/check/kolla-ansible-oraclelinux-source/7bf1cc2/ : SUCCESS in 1h 07m 19s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/52/664252/1/check/kolla-ansible-oraclelinux-binary/0c20d69/ : SUCCESS in 1h 05m 28s (non-voting)\n- kolla-ansible-oraclelinux-source-ceph http://logs.openstack.org/52/664252/1/check/kolla-ansible-oraclelinux-source-ceph/79457a4/ : FAILURE in 43m 53s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/52/664252/1/check/openstack-tox-lower-constraints/7e04c1a/ : SUCCESS in 4m 46s\n- openstack-tox-pep8 http://logs.openstack.org/52/664252/1/check/openstack-tox-pep8/44130a7/ : SUCCESS in 3m 36s\n- openstack-tox-py27 http://logs.openstack.org/52/664252/1/check/openstack-tox-py27/acebc11/ : SUCCESS in 3m 49s\n- openstack-tox-py36 http://logs.openstack.org/52/664252/1/check/openstack-tox-py36/fe98882/ : SUCCESS in 4m 22s\n- openstack-tox-py37 http://logs.openstack.org/52/664252/1/check/openstack-tox-py37/45129a6/ : SUCCESS in 3m 58s\n- openstack-tox-docs http://logs.openstack.org/52/664252/1/check/openstack-tox-docs/3f4d514/html/ : SUCCESS in 2m 38s\n- kolla-build-ubuntu-binary http://logs.openstack.org/52/664252/1/check/kolla-build-ubuntu-binary/c8b6b9a/ : SUCCESS in 32m 40s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/52/664252/1/check/kolla-build-ubuntu-source/ddcbabc/ : SUCCESS in 34m 25s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/52/664252/1/check/kolla-ansible-ubuntu-source/df1700c/ : SUCCESS in 54m 01s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/52/664252/1/check/kolla-ansible-ubuntu-binary/7ee5136/ : SUCCESS in 53m 48s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/52/664252/1/check/kolla-ansible-ubuntu-source-upgrade/d3391d6/ : SUCCESS in 55m 09s\n- kolla-ansible-ubuntu-source-ceph http://logs.openstack.org/52/664252/1/check/kolla-ansible-ubuntu-source-ceph/2bdc06b/ : SUCCESS in 1h 16m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"204a9d98ff5cd4d27301f44954966881a2378552","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-10 14:02:26.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"12490ecf1175c59e2d35c11f3d3476ec1fe863be","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-10 14:02:39.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"801299ef17f4078f8294b0a327bc25f2128736e9","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-10 14:52:06.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/52/664252/2/openstack-check/tripleo-build-containers-fedora-28-master/a82b4c4/ : SUCCESS in 48m 54s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/52/664252/2/openstack-check/tripleo-build-containers-fedora-28-stein/c15bde0/ : SUCCESS in 45m 18s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"802a684ff4034908641d49e2b1edf47fe6893b0c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 15:59:36.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/52/664252/2/check/kolla-build-centos-binary/8c4153d/ : SUCCESS in 44m 54s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/52/664252/2/check/kolla-build-centos-source/c98bba5/ : SUCCESS in 54m 06s\n- kolla-ansible-centos-source http://logs.openstack.org/52/664252/2/check/kolla-ansible-centos-source/6899622/ : SUCCESS in 1h 11m 26s\n- kolla-ansible-centos-binary http://logs.openstack.org/52/664252/2/check/kolla-ansible-centos-binary/a430300/ : SUCCESS in 1h 07m 19s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/52/664252/2/check/kolla-ansible-centos-source-upgrade/e4bdf7f/ : SUCCESS in 1h 20m 53s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/52/664252/2/check/tripleo-build-containers-centos-7/d97b2c1/ : SUCCESS in 40m 13s\n- kolla-ansible-centos-source-ceph http://logs.openstack.org/52/664252/2/check/kolla-ansible-centos-source-ceph/3936ef4/ : SUCCESS in 1h 33m 33s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/52/664252/2/check/kolla-build-oraclelinux-binary/3d02555/ : SUCCESS in 43m 19s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/52/664252/2/check/kolla-build-oraclelinux-source/e1df356/ : SUCCESS in 51m 10s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/52/664252/2/check/kolla-ansible-oraclelinux-source/7b3231d/ : SUCCESS in 1h 04m 57s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/52/664252/2/check/kolla-ansible-oraclelinux-binary/7332968/ : SUCCESS in 1h 12m 41s (non-voting)\n- kolla-ansible-oraclelinux-source-ceph http://logs.openstack.org/52/664252/2/check/kolla-ansible-oraclelinux-source-ceph/91677e7/ : SUCCESS in 1h 42m 23s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/52/664252/2/check/openstack-tox-lower-constraints/969eff1/ : SUCCESS in 3m 58s\n- openstack-tox-pep8 http://logs.openstack.org/52/664252/2/check/openstack-tox-pep8/9017b24/ : SUCCESS in 3m 09s\n- openstack-tox-py27 http://logs.openstack.org/52/664252/2/check/openstack-tox-py27/1e31bbb/ : SUCCESS in 3m 46s\n- openstack-tox-py36 http://logs.openstack.org/52/664252/2/check/openstack-tox-py36/f4cf322/ : SUCCESS in 3m 43s\n- openstack-tox-py37 http://logs.openstack.org/52/664252/2/check/openstack-tox-py37/1641f5a/ : SUCCESS in 3m 36s\n- openstack-tox-docs http://logs.openstack.org/52/664252/2/check/openstack-tox-docs/e0e6ea5/html/ : SUCCESS in 3m 09s\n- kolla-build-ubuntu-binary http://logs.openstack.org/52/664252/2/check/kolla-build-ubuntu-binary/d56585e/ : SUCCESS in 50m 10s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/52/664252/2/check/kolla-build-ubuntu-source/2d0570f/ : SUCCESS in 27m 28s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/52/664252/2/check/kolla-ansible-ubuntu-source/efc8e95/ : SUCCESS in 57m 34s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/52/664252/2/check/kolla-ansible-ubuntu-binary/5cab560/ : SUCCESS in 52m 12s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/52/664252/2/check/kolla-ansible-ubuntu-source-upgrade/fe168a2/ : SUCCESS in 58m 43s\n- kolla-ansible-ubuntu-source-ceph http://logs.openstack.org/52/664252/2/check/kolla-ansible-ubuntu-source-ceph/4f6a138/ : SUCCESS in 1h 15m 13s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"338737c8df2258176553ba582d10fead54735935","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-15 08:00:08.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"3aa4ac342ebeffb2bd9f36afe8d817630497632c","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-15 08:00:54.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"893a5e6438d54aea9e739546cba295f1f9c22bf1","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-15 08:03:07.000000000","message":"Uploaded patch set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"d39cc70b786cfda7ccb62003ae0ef82fe7da43eb","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-15 08:37:37.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/52/664252/5/openstack-check/tripleo-build-containers-fedora-28-master/f0561c8/ : SUCCESS in 32m 09s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/52/664252/5/openstack-check/tripleo-build-containers-fedora-28-stein/96791ad/ : SUCCESS in 33m 12s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"93ca418fc0fb083f9a5f529651c31e8f3f8be28e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-15 09:42:32.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/52/664252/5/check/kolla-build-centos-binary/489934e/ : SUCCESS in 44m 25s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/52/664252/5/check/kolla-build-centos-source/73afe18/ : SUCCESS in 45m 28s\n- kolla-ansible-centos-source http://logs.openstack.org/52/664252/5/check/kolla-ansible-centos-source/a275821/ : SUCCESS in 1h 09m 49s\n- kolla-ansible-centos-binary http://logs.openstack.org/52/664252/5/check/kolla-ansible-centos-binary/b66bf2d/ : SUCCESS in 1h 00m 22s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/52/664252/5/check/kolla-ansible-centos-source-upgrade/be28ae2/ : SUCCESS in 1h 12m 01s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/52/664252/5/check/tripleo-build-containers-centos-7/ef9fbdd/ : SUCCESS in 56m 21s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/52/664252/5/check/kolla-build-oraclelinux-binary/a6a3113/ : SUCCESS in 46m 34s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/52/664252/5/check/kolla-build-oraclelinux-source/6a0ffa6/ : SUCCESS in 45m 15s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/52/664252/5/check/kolla-ansible-oraclelinux-source/8c4673c/ : SUCCESS in 1h 12m 15s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/52/664252/5/check/kolla-ansible-oraclelinux-binary/75bf67c/ : SUCCESS in 1h 08m 20s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/52/664252/5/check/openstack-tox-lower-constraints/e27eb41/ : FAILURE in 3m 24s\n- openstack-tox-pep8 http://logs.openstack.org/52/664252/5/check/openstack-tox-pep8/6632d50/ : SUCCESS in 3m 22s\n- openstack-tox-py27 http://logs.openstack.org/52/664252/5/check/openstack-tox-py27/95e82c3/ : SUCCESS in 4m 21s\n- openstack-tox-py36 http://logs.openstack.org/52/664252/5/check/openstack-tox-py36/f6c8bb7/ : SUCCESS in 4m 59s\n- openstack-tox-py37 http://logs.openstack.org/52/664252/5/check/openstack-tox-py37/4503972/ : SUCCESS in 3m 18s\n- openstack-tox-docs http://logs.openstack.org/52/664252/5/check/openstack-tox-docs/fe64f4e/html/ : SUCCESS in 3m 07s\n- kolla-build-ubuntu-binary http://logs.openstack.org/52/664252/5/check/kolla-build-ubuntu-binary/3783a7c/ : SUCCESS in 30m 25s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/52/664252/5/check/kolla-build-ubuntu-source/091956c/ : SUCCESS in 36m 01s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/52/664252/5/check/kolla-ansible-ubuntu-source/d1562c3/ : SUCCESS in 52m 54s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/52/664252/5/check/kolla-ansible-ubuntu-binary/434c3e0/ : SUCCESS in 48m 06s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/52/664252/5/check/kolla-ansible-ubuntu-source-upgrade/ea9b818/ : SUCCESS in 59m 05s","accounts_in_message":[],"_revision_number":5},{"id":"89a3c105e141c9543401e219027c9f4cdc3e121d","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-15 12:12:21.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"6cff4c41e07b2aef91c861ffc40f07c2527f4677","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-15 12:13:00.000000000","message":"Patch Set 6:\n\nRebased on lower-constraints fix to pass checks.","accounts_in_message":[],"_revision_number":6},{"id":"9a8592f8b896d14b8c578f82f09fe68456b0adfd","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-15 12:50:11.000000000","message":"Patch Set 6:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/52/664252/6/openstack-check/tripleo-build-containers-fedora-28-master/aab5f17/ : SUCCESS in 33m 33s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/52/664252/6/openstack-check/tripleo-build-containers-fedora-28-stein/3bfdd0c/ : SUCCESS in 36m 17s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"a6590ce742797cea5608a125f828fd3c65085550","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-15 13:46:35.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/52/664252/6/check/kolla-build-centos-binary/dc27b77/ : SUCCESS in 40m 56s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/52/664252/6/check/kolla-build-centos-source/6a9a87b/ : SUCCESS in 44m 24s\n- kolla-ansible-centos-source http://logs.openstack.org/52/664252/6/check/kolla-ansible-centos-source/b96eb4e/ : SUCCESS in 1h 07m 33s\n- kolla-ansible-centos-binary http://logs.openstack.org/52/664252/6/check/kolla-ansible-centos-binary/b89dbf9/ : SUCCESS in 1h 04m 51s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/52/664252/6/check/kolla-ansible-centos-source-upgrade/57a8f0c/ : SUCCESS in 1h 13m 07s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/52/664252/6/check/tripleo-build-containers-centos-7/a298908/ : SUCCESS in 43m 36s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/52/664252/6/check/kolla-build-oraclelinux-binary/3f137b5/ : SUCCESS in 54m 38s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/52/664252/6/check/kolla-build-oraclelinux-source/34b2090/ : SUCCESS in 47m 41s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/52/664252/6/check/kolla-ansible-oraclelinux-source/2849f52/ : SUCCESS in 1h 21m 33s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/52/664252/6/check/kolla-ansible-oraclelinux-binary/e9faaf2/ : SUCCESS in 59m 51s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/52/664252/6/check/openstack-tox-lower-constraints/591472d/ : SUCCESS in 3m 21s\n- openstack-tox-pep8 http://logs.openstack.org/52/664252/6/check/openstack-tox-pep8/00059ab/ : SUCCESS in 2m 34s\n- openstack-tox-py27 http://logs.openstack.org/52/664252/6/check/openstack-tox-py27/c730643/ : SUCCESS in 3m 11s\n- openstack-tox-py36 http://logs.openstack.org/52/664252/6/check/openstack-tox-py36/625a4c3/ : SUCCESS in 3m 01s\n- openstack-tox-py37 http://logs.openstack.org/52/664252/6/check/openstack-tox-py37/8bd5f31/ : SUCCESS in 3m 13s\n- openstack-tox-docs http://logs.openstack.org/52/664252/6/check/openstack-tox-docs/15f781e/html/ : SUCCESS in 2m 45s\n- kolla-build-ubuntu-binary http://logs.openstack.org/52/664252/6/check/kolla-build-ubuntu-binary/014a713/ : SUCCESS in 49m 25s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/52/664252/6/check/kolla-build-ubuntu-source/e54c831/ : SUCCESS in 30m 26s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/52/664252/6/check/kolla-ansible-ubuntu-source/8b5b363/ : SUCCESS in 48m 53s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/52/664252/6/check/kolla-ansible-ubuntu-binary/76697c5/ : SUCCESS in 46m 05s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/52/664252/6/check/kolla-ansible-ubuntu-source-upgrade/dec45ca/ : SUCCESS in 52m 59s","accounts_in_message":[],"_revision_number":6},{"id":"41392a51156076c3c9958e83f0bc145838fb5a96","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-15 19:06:53.000000000","message":"Patch Set 6: Code-Review+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":6},{"id":"ff4e2be68dc5da0c86c8944169bcce576f50d758","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-15 19:07:24.000000000","message":"Patch Set 6: Code-Review-1\n\nDo it in master and then backport.","accounts_in_message":[],"_revision_number":6},{"id":"f5637687e455bad7f9221d029f131f28c2065049","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-15 19:11:01.000000000","message":"Patch Set 6:\n\n(3 comments)","accounts_in_message":[],"_revision_number":6},{"id":"ec20bd8f777459e30173b47e3d588b7ea76e3f7d","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-15 19:11:53.000000000","message":"Patch Set 6:\n\nDo we want it in master? It will be empty.","accounts_in_message":[],"_revision_number":6},{"id":"bfebb07a6165a30ba0a19eefd01f35a4925aaaed","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-17 11:59:46.000000000","message":"Patch Set 6:\n\n\u003e Do we want it in master? It will be empty.\n\nIt is new function. Should land in master first. Then backported.\n\nWho knows, maybe someone will use it in master.","accounts_in_message":[],"_revision_number":6},{"id":"f8b00274d863262ec86a45fec0be2d740d27e5b0","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-17 12:02:20.000000000","message":"Patch Set 6:\n\nYeah, I thought the same later but forgot due to other things. Let me do some git magic.","accounts_in_message":[],"_revision_number":6},{"id":"661790f8240c259ed60b3f4f2f62cba9ee6661eb","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-17 12:58:45.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"aeacd2a381058dee926910c66ebcc93003fd7a5a","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-17 14:08:57.000000000","message":"Patch Set 7:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/52/664252/7/openstack-check/tripleo-build-containers-fedora-28-master/4b2536a/ : SUCCESS in 39m 01s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/52/664252/7/openstack-check/tripleo-build-containers-fedora-28-stein/d4882ce/ : SUCCESS in 38m 49s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"99ebac23d14b45c064764545ea352a79a35e0622","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-17 14:38:08.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/52/664252/7/check/kolla-build-centos-binary/474c18d/ : FAILURE in 3m 41s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/52/664252/7/check/kolla-build-centos-source/0ea5080/ : SUCCESS in 57m 29s\n- kolla-ansible-centos-source http://logs.openstack.org/52/664252/7/check/kolla-ansible-centos-source/f9fecc0/ : SUCCESS in 1h 00m 49s\n- kolla-ansible-centos-binary http://logs.openstack.org/52/664252/7/check/kolla-ansible-centos-binary/9931ccd/ : SUCCESS in 1h 07m 28s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/52/664252/7/check/kolla-ansible-centos-source-upgrade/b190587/ : SUCCESS in 1h 06m 16s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/52/664252/7/check/tripleo-build-containers-centos-7/f4491db/ : SUCCESS in 45m 25s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/52/664252/7/check/kolla-build-oraclelinux-binary/7a155fa/ : SUCCESS in 52m 07s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/52/664252/7/check/kolla-build-oraclelinux-source/da11506/ : SUCCESS in 48m 42s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/52/664252/7/check/kolla-ansible-oraclelinux-source/3c343d5/ : SUCCESS in 1h 16m 08s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/52/664252/7/check/kolla-ansible-oraclelinux-binary/9ced049/ : SUCCESS in 1h 04m 57s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/52/664252/7/check/openstack-tox-lower-constraints/93bac99/ : SUCCESS in 3m 34s\n- openstack-tox-pep8 http://logs.openstack.org/52/664252/7/check/openstack-tox-pep8/e4cdf24/ : SUCCESS in 3m 19s\n- openstack-tox-py27 http://logs.openstack.org/52/664252/7/check/openstack-tox-py27/af2fe7f/ : SUCCESS in 3m 54s\n- openstack-tox-py36 http://logs.openstack.org/52/664252/7/check/openstack-tox-py36/f41cc46/ : SUCCESS in 3m 28s\n- openstack-tox-py37 http://logs.openstack.org/52/664252/7/check/openstack-tox-py37/c509635/ : SUCCESS in 3m 40s\n- openstack-tox-docs http://logs.openstack.org/52/664252/7/check/openstack-tox-docs/5c7e4f5/html/ : SUCCESS in 2m 48s\n- kolla-build-ubuntu-binary http://logs.openstack.org/52/664252/7/check/kolla-build-ubuntu-binary/6a3815c/ : FAILURE in 5m 41s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/52/664252/7/check/kolla-build-ubuntu-source/76acc9e/ : FAILURE in 5m 31s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/52/664252/7/check/kolla-ansible-ubuntu-source/f5556ac/ : SUCCESS in 53m 35s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/52/664252/7/check/kolla-ansible-ubuntu-binary/1a44457/ : SUCCESS in 47m 56s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/52/664252/7/check/kolla-ansible-ubuntu-source-upgrade/d25e360/ : SUCCESS in 1h 01m 38s","accounts_in_message":[],"_revision_number":7},{"id":"8995f67456cba78543a3469c21216a59dba32bcf","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-17 18:44:28.000000000","message":"Patch Set 7: Workflow-1\n\nwaiting for master merge","accounts_in_message":[],"_revision_number":7},{"id":"53a77be445b9c7ceb4ba9d43cb8b05ddbcd46ee0","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-22 07:50:50.000000000","message":"Abandoned\n\nWe will be using CentOS Storage SIG Ceph Nautilus in Stein because ganesha is back. Master branch change: https://review.opendev.org/666015","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"190f2fec748b6f8e8ee29c43f50e8c57ce3f57ce","revisions":{"5c86f3511161ef84179b58a5a7a06fb7430966ff":{"kind":"REWORK","_number":1,"created":"2019-06-10 12:18:20.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/52/664252/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/52/664252/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/664252/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/664252/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/664252/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/52/664252/1"}}},"commit":{"parents":[{"commit":"13913023f02ca07a0c737c136c291a4544ef0d24","subject":"Merge \"Bump Monasca Fluentd plugin version\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/13913023f02ca07a0c737c136c291a4544ef0d24"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-06-10 12:16:14.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-06-10 12:16:14.000000000","tz":120},"subject":"WIP: Do not disable ceph-nautilus repo in base but ceph-base","message":"WIP: Do not disable ceph-nautilus repo in base but ceph-base\n\nFor TripleO compatibility.\nSee https://review.opendev.org/664244\n\nChange-Id: Ie268bb875b774a9bafd04f8c27c63b2b35dee5c8\nSigned-off-by: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5c86f3511161ef84179b58a5a7a06fb7430966ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5c86f3511161ef84179b58a5a7a06fb7430966ff"}]},"branch":"refs/heads/stable/stein"},"ca34a112185f5db0267220e2bb9eb49b929a5e47":{"kind":"REWORK","_number":2,"created":"2019-06-10 14:02:26.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/52/664252/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/52/664252/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/664252/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/664252/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/664252/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/52/664252/2"}}},"commit":{"parents":[{"commit":"13913023f02ca07a0c737c136c291a4544ef0d24","subject":"Merge \"Bump Monasca Fluentd plugin version\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/13913023f02ca07a0c737c136c291a4544ef0d24"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-06-10 12:16:14.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-06-10 14:01:54.000000000","tz":120},"subject":"WIP: Do not disable ceph-nautilus repo in base but ceph-base","message":"WIP: Do not disable ceph-nautilus repo in base but ceph-base\n\nFor TripleO compatibility.\nSee https://review.opendev.org/664244\n\nDNM: breaking aarch64 now\n\nChange-Id: Ie268bb875b774a9bafd04f8c27c63b2b35dee5c8\nSigned-off-by: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ca34a112185f5db0267220e2bb9eb49b929a5e47"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ca34a112185f5db0267220e2bb9eb49b929a5e47"}]},"branch":"refs/heads/stable/stein"},"6fd36dbe08e2808df3e75cea9be6b172e3c3ee3f":{"kind":"REWORK","_number":3,"created":"2019-06-15 08:00:08.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/52/664252/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/52/664252/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/664252/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/664252/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/664252/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/52/664252/3"}}},"commit":{"parents":[{"commit":"13913023f02ca07a0c737c136c291a4544ef0d24","subject":"Merge \"Bump Monasca Fluentd plugin version\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/13913023f02ca07a0c737c136c291a4544ef0d24"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-06-10 12:16:14.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-06-15 07:45:40.000000000","tz":120},"subject":"[stable/stein] Let others override repos disabled in base container","message":"[stable/stein] Let others override repos disabled in base container\n\nhttps://review.opendev.org/651601 disabled CentOS Ceph Nautilus\nrepo for Stein.\nIt turned out to be a problem for TripleO which does not use\nupstream Ceph and actually relies on CentOS Ceph Nautilus.\nWe did not leave the option to override our customization.\nThis patch implements it.\n\nSee https://review.opendev.org/664244 for the related TripleO\nchange.\n\nChange-Id: Ie268bb875b774a9bafd04f8c27c63b2b35dee5c8\nSigned-off-by: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6fd36dbe08e2808df3e75cea9be6b172e3c3ee3f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6fd36dbe08e2808df3e75cea9be6b172e3c3ee3f"}]},"branch":"refs/heads/stable/stein"},"e1e57294d711d14008312894fda7c53802544f42":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2019-06-15 08:00:54.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/52/664252/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/52/664252/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/664252/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/664252/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/664252/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/52/664252/4"}}},"commit":{"parents":[{"commit":"13913023f02ca07a0c737c136c291a4544ef0d24","subject":"Merge \"Bump Monasca Fluentd plugin version\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/13913023f02ca07a0c737c136c291a4544ef0d24"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-06-10 12:16:14.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-06-15 08:00:23.000000000","tz":120},"subject":"[stable/stein] Let others override repos disabled in base","message":"[stable/stein] Let others override repos disabled in base\n\nhttps://review.opendev.org/651601 disabled CentOS Ceph Nautilus\nrepo for Stein.\nIt turned out to be a problem for TripleO which does not use\nupstream Ceph and actually relies on CentOS Ceph Nautilus.\nWe did not leave the option to override our customization.\nThis patch implements it.\n\nSee https://review.opendev.org/664244 for the related TripleO\nchange.\n\nChange-Id: Ie268bb875b774a9bafd04f8c27c63b2b35dee5c8\nSigned-off-by: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e1e57294d711d14008312894fda7c53802544f42"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e1e57294d711d14008312894fda7c53802544f42"}]},"branch":"refs/heads/stable/stein"},"99cd2f8b4c1e43bf45ea2cf551d9566f6729a3cc":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2019-06-15 08:03:07.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/52/664252/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/52/664252/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/664252/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/664252/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/664252/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/52/664252/5"}}},"commit":{"parents":[{"commit":"13913023f02ca07a0c737c136c291a4544ef0d24","subject":"Merge \"Bump Monasca Fluentd plugin version\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/13913023f02ca07a0c737c136c291a4544ef0d24"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-06-10 12:16:14.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-06-15 08:02:07.000000000","tz":120},"subject":"[stable/stein] Let others override repos disabled in base","message":"[stable/stein] Let others override repos disabled in base\n\nhttps://review.opendev.org/651601 disabled CentOS Ceph\nNautilus repo for Stein.\nIt turned out to be a problem for TripleO which does not use\nupstream Ceph repo and actually relies on CentOS Ceph\nNautilus repo.\nWe did not leave the option to override our customization.\nThis patch implements it.\n\nSee https://review.opendev.org/664244 for the related TripleO\nchange.\n\nChange-Id: Ie268bb875b774a9bafd04f8c27c63b2b35dee5c8\nSigned-off-by: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/99cd2f8b4c1e43bf45ea2cf551d9566f6729a3cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/99cd2f8b4c1e43bf45ea2cf551d9566f6729a3cc"}]},"branch":"refs/heads/stable/stein"},"f5f41ce9426be854af1dceebefeddf64462c401e":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2019-06-15 12:12:21.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/52/664252/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/52/664252/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/664252/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/664252/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/664252/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/52/664252/6"}}},"commit":{"parents":[{"commit":"7a23046b280121cfe7c1b8f77eb97c6f099987a7","subject":"Fix lower-constraints test fail due to failing dep build","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7a23046b280121cfe7c1b8f77eb97c6f099987a7"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-06-10 12:16:14.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-06-15 12:12:21.000000000","tz":0},"subject":"[stable/stein] Let others override repos disabled in base","message":"[stable/stein] Let others override repos disabled in base\n\nhttps://review.opendev.org/651601 disabled CentOS Ceph\nNautilus repo for Stein.\nIt turned out to be a problem for TripleO which does not use\nupstream Ceph repo and actually relies on CentOS Ceph\nNautilus repo.\nWe did not leave the option to override our customization.\nThis patch implements it.\n\nSee https://review.opendev.org/664244 for the related TripleO\nchange.\n\nChange-Id: Ie268bb875b774a9bafd04f8c27c63b2b35dee5c8\nSigned-off-by: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f5f41ce9426be854af1dceebefeddf64462c401e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f5f41ce9426be854af1dceebefeddf64462c401e"}]},"branch":"refs/heads/stable/stein"},"190f2fec748b6f8e8ee29c43f50e8c57ce3f57ce":{"kind":"REWORK","_number":7,"created":"2019-06-17 12:58:45.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/52/664252/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/52/664252/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/664252/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/664252/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/664252/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/52/664252/7"}}},"commit":{"parents":[{"commit":"45ae94f9ad4412146a41451b2e17b0674b9e9bca","subject":"Fix issues with Percona in Zuul and clean up a little bit","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/45ae94f9ad4412146a41451b2e17b0674b9e9bca"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-06-10 12:16:14.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-06-17 12:53:23.000000000","tz":120},"subject":"Add centos_yum_repos_to_disable overridable in base","message":"Add centos_yum_repos_to_disable overridable in base\n\n[1] disabled CentOS Ceph Nautilus repo for Stein.\nIt turned out to be a problem for TripleO ([2])\nwhich does not use upstream Ceph repo and actually\nrelies on CentOS Ceph Nautilus repo.\nWe did not leave the option to override our customization.\nThis patch implements it in stable/stein.\n\n[1] https://review.opendev.org/651601\n[2] https://review.opendev.org/664244\n\nChange-Id: Ie268bb875b774a9bafd04f8c27c63b2b35dee5c8\nSigned-off-by: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n(cherry picked from commit 867556b13cf17f24477822368333bdcfb21c9dd0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/190f2fec748b6f8e8ee29c43f50e8c57ce3f57ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/190f2fec748b6f8e8ee29c43f50e8c57ce3f57ce"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
