)]}'
{"id":"openstack%2Fkolla~664676","triplet_id":"openstack%2Fkolla~master~I8ae19f212049cae5dfb4e8dd6ec6cd14a8c46e0e","project":"openstack/kolla","branch":"master","topic":"clean-reqs","hashtags":[],"change_id":"I8ae19f212049cae5dfb4e8dd6ec6cd14a8c46e0e","subject":"Fix lower-constraints test fail due to failing dep build","status":"MERGED","created":"2019-06-11 19:14:31.000000000","updated":"2019-06-13 17:36:11.000000000","submitted":"2019-06-13 17:31:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"664676-1560447111996-0785ed6c","meta_rev_id":"f749580104f0ddcff230d9a16ca19b658d561967","_number":664676,"virtual_id_number":664676,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":2,"date":"2019-06-13 17:31:51.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-06-11 20:54:27.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":2,"date":"2019-06-13 16:34:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2019-06-13 16:33:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2019-06-13 11:35:25.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2019-06-13 13:11:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":1,"date":"2019-06-13 16:34:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"date":"2019-06-13 11:35:40.000000000","_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},{"value":0,"date":"2019-06-13 11:35:29.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-11 20:54:27.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-06-13 11:35:25.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-06-13 11:35:29.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-06-13 11:35:40.000000000","updated_by":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"reviewer":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"state":"REVIEWER"},{"updated":"2019-06-13 13:11:04.000000000","updated_by":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"reviewer":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"state":"REVIEWER"},{"updated":"2019-06-13 16:33:50.000000000","updated_by":{"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},"reviewer":{"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},"state":"REVIEWER"},{"updated":"2019-06-13 16:34:04.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2019-06-13 17:31:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3ef1c7437c187cc01b3d5a79888ff617818ae195","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-11 19:14:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"74966127b2f37cf582f58e9efaeb86cd4df71db1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 19:41:38.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints http://logs.openstack.org/76/664676/1/check/openstack-tox-lower-constraints/d57a419/ : SUCCESS in 3m 46s\n- openstack-tox-pep8 http://logs.openstack.org/76/664676/1/check/openstack-tox-pep8/0587c38/ : SUCCESS in 4m 34s\n- openstack-tox-py27 http://logs.openstack.org/76/664676/1/check/openstack-tox-py27/348dcd7/ : SUCCESS in 3m 13s\n- openstack-tox-py36 http://logs.openstack.org/76/664676/1/check/openstack-tox-py36/cb2bbbc/ : SUCCESS in 3m 45s\n- openstack-tox-py37 http://logs.openstack.org/76/664676/1/check/openstack-tox-py37/5884c1e/ : SUCCESS in 3m 41s\n- requirements-check http://logs.openstack.org/76/664676/1/check/requirements-check/a193d45/ : FAILURE in 2m 17s\n- openstack-tox-docs http://logs.openstack.org/76/664676/1/check/openstack-tox-docs/4071607/html/ : SUCCESS in 3m 09s","accounts_in_message":[],"_revision_number":1},{"id":"bd7769b783e14819e4e8127ee8bf7e49d9660be7","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-11 19:52:40.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2810ffa956bde52dabe1ef6b5e4d44941c2ec4e0","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-11 20:54:27.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/76/664676/2/openstack-check/tripleo-build-containers-fedora-28-master/16274dd/ : SUCCESS in 57m 49s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/76/664676/2/openstack-check/tripleo-build-containers-fedora-28-stein/d028038/ : SUCCESS in 58m 40s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"85fa84a8b3589aa96364545cfde4b19f252b313d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 21:27:49.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/76/664676/2/check/kolla-build-centos-binary/751801f/ : SUCCESS in 41m 47s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/76/664676/2/check/kolla-build-centos-source/f690cdb/ : SUCCESS in 52m 50s\n- kolla-ansible-centos-source http://logs.openstack.org/76/664676/2/check/kolla-ansible-centos-source/9782f45/ : SUCCESS in 1h 15m 39s\n- kolla-ansible-centos-binary http://logs.openstack.org/76/664676/2/check/kolla-ansible-centos-binary/0fd3a0d/ : SUCCESS in 1h 08m 11s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/76/664676/2/check/kolla-ansible-centos-source-upgrade/6e7d5e0/ : SUCCESS in 1h 22m 45s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/76/664676/2/check/tripleo-build-containers-centos-7/1ec0ffa/ : SUCCESS in 47m 07s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/76/664676/2/check/kolla-build-oraclelinux-binary/23f9bca/ : SUCCESS in 54m 56s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/76/664676/2/check/kolla-build-oraclelinux-source/02eed00/ : SUCCESS in 1h 00m 07s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/76/664676/2/check/kolla-ansible-oraclelinux-source/0cbf07b/ : SUCCESS in 1h 13m 16s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/76/664676/2/check/kolla-ansible-oraclelinux-binary/f61dac2/ : SUCCESS in 1h 13m 23s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/76/664676/2/check/openstack-tox-lower-constraints/2c4d95b/ : SUCCESS in 3m 59s\n- openstack-tox-pep8 http://logs.openstack.org/76/664676/2/check/openstack-tox-pep8/bb9f722/ : SUCCESS in 3m 21s\n- openstack-tox-py27 http://logs.openstack.org/76/664676/2/check/openstack-tox-py27/1336408/ : SUCCESS in 3m 20s\n- openstack-tox-py36 http://logs.openstack.org/76/664676/2/check/openstack-tox-py36/f34e5bf/ : SUCCESS in 3m 07s\n- openstack-tox-py37 http://logs.openstack.org/76/664676/2/check/openstack-tox-py37/82b38e4/ : SUCCESS in 4m 16s\n- requirements-check http://logs.openstack.org/76/664676/2/check/requirements-check/2ae310a/ : SUCCESS in 2m 13s\n- openstack-tox-docs http://logs.openstack.org/76/664676/2/check/openstack-tox-docs/9f35c72/html/ : SUCCESS in 3m 06s\n- kolla-build-ubuntu-binary http://logs.openstack.org/76/664676/2/check/kolla-build-ubuntu-binary/0f4f9d9/ : SUCCESS in 24m 32s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/76/664676/2/check/kolla-build-ubuntu-source/12aeddc/ : SUCCESS in 31m 59s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/76/664676/2/check/kolla-ansible-ubuntu-source/17e124f/ : SUCCESS in 46m 34s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/76/664676/2/check/kolla-ansible-ubuntu-binary/991879f/ : SUCCESS in 46m 25s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/76/664676/2/check/kolla-ansible-ubuntu-source-upgrade/101a36f/ : SUCCESS in 56m 24s","accounts_in_message":[],"_revision_number":2},{"id":"29882ba8bef054c0dc48fd0c6a770c5b0b7b3df5","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-13 11:35:25.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"41f4d49e72aab69d1d02aaa9132539f500e9a32e","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2019-06-13 13:11:04.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"f8ca15385f93a204585978b2b0b55df51c2e9e18","author":{"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},"date":"2019-06-13 16:33:50.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"603fe82f7d79c71460aa461101726b99ba3511c5","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-06-13 16:34:04.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"3c75c1fadf6671bc3bea7cb3d2788c33442d6199","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-13 16:34:16.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"4807712c3aa235edb5975ab5b8f15781e57b9584","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-13 17:31:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"a4517c5077a75af204f046d8a0f0b0180133ef5a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-13 17:31:51.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source http://logs.openstack.org/76/664676/2/gate/kolla-build-centos-source/8afd3cf/ : SUCCESS in 48m 44s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/76/664676/2/gate/tripleo-build-containers-centos-7/e6982f0/ : SUCCESS in 50m 27s\n- kolla-build-oraclelinux-source http://logs.openstack.org/76/664676/2/gate/kolla-build-oraclelinux-source/3b0bfc7/ : SUCCESS in 45m 52s\n- openstack-tox-lower-constraints http://logs.openstack.org/76/664676/2/gate/openstack-tox-lower-constraints/10b5e43/ : SUCCESS in 4m 37s\n- openstack-tox-pep8 http://logs.openstack.org/76/664676/2/gate/openstack-tox-pep8/cc3e389/ : SUCCESS in 3m 59s\n- openstack-tox-py27 http://logs.openstack.org/76/664676/2/gate/openstack-tox-py27/e3a36e0/ : SUCCESS in 3m 45s\n- openstack-tox-py36 http://logs.openstack.org/76/664676/2/gate/openstack-tox-py36/210f586/ : SUCCESS in 3m 22s\n- openstack-tox-py37 http://logs.openstack.org/76/664676/2/gate/openstack-tox-py37/97a53cc/ : SUCCESS in 3m 44s\n- requirements-check http://logs.openstack.org/76/664676/2/gate/requirements-check/bc5f605/ : SUCCESS in 2m 52s\n- openstack-tox-docs http://logs.openstack.org/76/664676/2/gate/openstack-tox-docs/30873fd/html/ : SUCCESS in 3m 16s\n- kolla-build-ubuntu-source http://logs.openstack.org/76/664676/2/gate/kolla-build-ubuntu-source/9218268/ : SUCCESS in 32m 23s","accounts_in_message":[],"_revision_number":2},{"id":"a80c7942023438fadbd14c96b7b716629d20c6fb","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-13 17:36:11.000000000","message":"Patch Set 2: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit 728d34fe23c65ec35635dc6e5e2e659749490214","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"39a2cf066c4c79436629017155302d36243aa338","revisions":{"e312a9024f009fad52183e9e05e5c5d4294ae9e9":{"kind":"REWORK","_number":1,"created":"2019-06-11 19:14:31.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/76/664676/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/76/664676/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/76/664676/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/76/664676/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/76/664676/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/76/664676/1"}}},"commit":{"parents":[{"commit":"b5c8b595a7f9c4aebb35b0ceacc236ec5f4eb350","subject":"Merge \"Fix issues with Percona in Zuul and clean up a little bit\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b5c8b595a7f9c4aebb35b0ceacc236ec5f4eb350"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-06-11 19:04:38.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-06-11 19:12:54.000000000","tz":120},"subject":"Fix lower-constraints test fail due to failing dep build","message":"Fix lower-constraints test fail due to failing dep build\n\nUbuntu Bionic updated OpenSSL package to new release:\nfrom 1.1.0 to 1.1.1\nIt caused cryptography in version from lower-constraints\nto fail to build.\nAdditionally, constrained cryptography version did not\nmatch the requirement for constrained PyOpenSSL.\nThis commit bumps the version to the minimum required\nby constrained PyOpenSSL.\nThis version agrees also with upper-constraints for Queens.\n\nChange-Id: I8ae19f212049cae5dfb4e8dd6ec6cd14a8c46e0e\nSigned-off-by: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e312a9024f009fad52183e9e05e5c5d4294ae9e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e312a9024f009fad52183e9e05e5c5d4294ae9e9"}]},"branch":"refs/heads/master"},"39a2cf066c4c79436629017155302d36243aa338":{"kind":"REWORK","_number":2,"created":"2019-06-11 19:52:40.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/76/664676/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/76/664676/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/76/664676/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/76/664676/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/76/664676/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/76/664676/2"}}},"commit":{"parents":[{"commit":"b5c8b595a7f9c4aebb35b0ceacc236ec5f4eb350","subject":"Merge \"Fix issues with Percona in Zuul and clean up a little bit\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b5c8b595a7f9c4aebb35b0ceacc236ec5f4eb350"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-06-11 19:04:38.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-06-11 19:52:10.000000000","tz":120},"subject":"Fix lower-constraints test fail due to failing dep build","message":"Fix lower-constraints test fail due to failing dep build\n\nUbuntu Bionic updated OpenSSL package to new release:\nfrom 1.1.0 to 1.1.1\nIt caused cryptography in version from lower-constraints\nto fail to build in Zuul.\nAdditionally, constrained cryptography version did not\nmatch the requirement for constrained PyOpenSSL.\nThis commit bumps the version to the minimum required\nby constrained PyOpenSSL.\nIt works because manylinux1 (binary) wheel exists.\nThis version agrees also with upper-constraints for Queens.\n\nDependency on cryptography is removed as it is not used directly\nby kolla (and requirements checker requires lower-constraints\nmatching version - let\u0027s not create an illusion of usage).\n\nChange-Id: I8ae19f212049cae5dfb4e8dd6ec6cd14a8c46e0e\nSigned-off-by: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/39a2cf066c4c79436629017155302d36243aa338"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/39a2cf066c4c79436629017155302d36243aa338"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
