)]}'
{"id":"openstack%2Fkolla~664979","triplet_id":"openstack%2Fkolla~stable%2Frocky~I128559f9bcfe48b786a3f52fe4e04bd47cac7a7d","project":"openstack/kolla","branch":"stable/rocky","topic":"bug/1791896-stable/rocky","hashtags":[],"change_id":"I128559f9bcfe48b786a3f52fe4e04bd47cac7a7d","subject":"Handle role properly when it is changed in kolla_keystone_user module","status":"MERGED","created":"2019-06-12 17:13:03.000000000","updated":"2019-06-17 14:55:34.000000000","submitted":"2019-06-17 14:55:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"664979-1560783334652-44d5107d","meta_rev_id":"0a54555e7a118548cbdd2781c86ba8d0a808edae","_number":664979,"virtual_id_number":664979,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-06-17 14:55:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-06-12 18:03:50.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},{"value":2,"date":"2019-06-15 19:20:52.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2019-06-17 13:56:37.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2019-06-17 13:56:37.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"date":"2019-06-12 17:13:03.000000000","_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-12 17:13:03.000000000","updated_by":{"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},"reviewer":{"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},"state":"REVIEWER"},{"updated":"2019-06-12 18:03:50.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-06-15 19:20:52.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-06-17 13:56:37.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-06-17 14:55:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f1b3cb20c457f66c4a3f2bb0ae2250cf1a5defef","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-06-12 17:13:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a3da3d375f66d2969513f4541c5c4cf208ebe7bc","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-06-12 17:18:06.000000000","message":"Topic set to bug/1791896-stable/rocky","accounts_in_message":[],"_revision_number":1},{"id":"dff3f620efa685f73b2e897afd722546ce607056","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-12 18:03:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master finger://ze02.softwarefactory-project.io:7979/31bd0f9839f14656aec788e888e0a98f : RETRY_LIMIT in 18s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/79/664979/1/openstack-check/tripleo-build-containers-fedora-28-stein/aba6280/ : SUCCESS in 45m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6d0d94f013085ebfa2eaee1dc72b7cff5de052d5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-12 19:14:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/79/664979/1/check/kolla-build-centos-binary/e646cb0/ : SUCCESS in 39m 35s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/79/664979/1/check/kolla-build-centos-source/120520f/ : SUCCESS in 35m 27s\n- kolla-ansible-centos-source http://logs.openstack.org/79/664979/1/check/kolla-ansible-centos-source/e39fbc0/ : SUCCESS in 1h 24m 45s\n- kolla-ansible-centos-binary http://logs.openstack.org/79/664979/1/check/kolla-ansible-centos-binary/2a38d31/ : SUCCESS in 1h 10m 20s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/79/664979/1/check/kolla-build-oraclelinux-binary/e566cdc/ : SUCCESS in 55m 15s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/79/664979/1/check/kolla-build-oraclelinux-source/9d31d77/ : SUCCESS in 44m 22s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/79/664979/1/check/kolla-ansible-oraclelinux-source/21e3aeb/ : SUCCESS in 1h 08m 37s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/79/664979/1/check/kolla-ansible-oraclelinux-binary/d31fc28/ : SUCCESS in 1h 17m 07s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/79/664979/1/check/openstack-tox-lower-constraints/604f353/ : SUCCESS in 5m 16s\n- openstack-tox-pep8 http://logs.openstack.org/79/664979/1/check/openstack-tox-pep8/2ecfd08/ : SUCCESS in 4m 02s\n- openstack-tox-py27 http://logs.openstack.org/79/664979/1/check/openstack-tox-py27/8805a5f/ : SUCCESS in 3m 29s\n- openstack-tox-py35 http://logs.openstack.org/79/664979/1/check/openstack-tox-py35/20f8302/ : SUCCESS in 3m 40s\n- build-openstack-sphinx-docs http://logs.openstack.org/79/664979/1/check/build-openstack-sphinx-docs/66f3e32/html/ : SUCCESS in 3m 44s\n- kolla-build-ubuntu-binary http://logs.openstack.org/79/664979/1/check/kolla-build-ubuntu-binary/e37ba59/ : SUCCESS in 1h 26m 37s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/79/664979/1/check/kolla-build-ubuntu-source/7951ac2/ : SUCCESS in 38m 18s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/79/664979/1/check/kolla-ansible-ubuntu-source/4c0b0d4/ : SUCCESS in 1h 05m 05s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/79/664979/1/check/kolla-ansible-ubuntu-binary/6a7ec68/ : SUCCESS in 1h 56m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1fc8bdb361e52aa59615413089137ffebacd4987","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-15 19:20:52.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"29a29160929ce9aa6ae42b5d210ce61b1f5de514","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-17 13:56:37.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0ceb74fac44bfdf4a35d56d990e9e9c56792b15a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-17 13:57:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0400dd908c82fe03937875cc51356898894be377","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-17 14:55:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"0a54555e7a118548cbdd2781c86ba8d0a808edae","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-17 14:55:34.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source http://logs.openstack.org/79/664979/1/gate/kolla-build-centos-source/3b39ae3/ : SUCCESS in 38m 11s\n- kolla-build-oraclelinux-source http://logs.openstack.org/79/664979/1/gate/kolla-build-oraclelinux-source/495d3d8/ : SUCCESS in 43m 38s\n- openstack-tox-lower-constraints http://logs.openstack.org/79/664979/1/gate/openstack-tox-lower-constraints/9001f23/ : SUCCESS in 3m 29s\n- openstack-tox-pep8 http://logs.openstack.org/79/664979/1/gate/openstack-tox-pep8/2369e88/ : SUCCESS in 4m 23s\n- openstack-tox-py27 http://logs.openstack.org/79/664979/1/gate/openstack-tox-py27/6402d0e/ : SUCCESS in 3m 04s\n- openstack-tox-py35 http://logs.openstack.org/79/664979/1/gate/openstack-tox-py35/739a31d/ : SUCCESS in 3m 38s\n- build-openstack-sphinx-docs http://logs.openstack.org/79/664979/1/gate/build-openstack-sphinx-docs/75ad62a/html/ : SUCCESS in 4m 40s\n- kolla-build-ubuntu-source http://logs.openstack.org/79/664979/1/gate/kolla-build-ubuntu-source/c13d69e/ : SUCCESS in 32m 27s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"35c9f0ba43c3a897c43ba5a25c1f655649a1b821","revisions":{"35c9f0ba43c3a897c43ba5a25c1f655649a1b821":{"kind":"REWORK","_number":1,"created":"2019-06-12 17:13:03.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/79/664979/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/664979/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/664979/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/664979/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/664979/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/664979/1"}}},"commit":{"parents":[{"commit":"31f0ef285ac127a569c17d340dcf4cc1f6c285b8","subject":"Merge \"Bump Monasca Notification version in Rocky\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/31f0ef285ac127a569c17d340dcf4cc1f6c285b8"}]}],"author":{"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","date":"2018-09-11 07:20:31.000000000","tz":480},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-06-12 17:12:47.000000000","tz":60},"subject":"Handle role properly when it is changed in kolla_keystone_user module","message":"Handle role properly when it is changed in kolla_keystone_user module\n\nChange-Id: I128559f9bcfe48b786a3f52fe4e04bd47cac7a7d\nCloses-Bug: #1791896\n(cherry picked from commit e08da5f9837a91f2b6e0fc2f20877d5cec3aa14f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/35c9f0ba43c3a897c43ba5a25c1f655649a1b821"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/35c9f0ba43c3a897c43ba5a25c1f655649a1b821"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
