)]}'
{"id":"openstack%2Fkolla~664980","triplet_id":"openstack%2Fkolla~stable%2Fqueens~I128559f9bcfe48b786a3f52fe4e04bd47cac7a7d","project":"openstack/kolla","branch":"stable/queens","topic":"bug/1791896-stable/queens","hashtags":[],"change_id":"I128559f9bcfe48b786a3f52fe4e04bd47cac7a7d","subject":"Handle role properly when it is changed in kolla_keystone_user module","status":"MERGED","created":"2019-06-12 17:14:20.000000000","updated":"2019-06-17 14:55:36.000000000","submitted":"2019-06-17 14:55:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"664980-1560783336321-91b8215b","meta_rev_id":"1d05f305ebbe155951b8490d916dd35db37c0df0","_number":664980,"virtual_id_number":664980,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-06-17 14:55:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-06-12 18:01:03.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},{"value":2,"date":"2019-06-15 19:21:09.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2019-06-17 13:56:52.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2019-06-17 13:56:52.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"date":"2019-06-12 17:14:20.000000000","_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-12 17:14:20.000000000","updated_by":{"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},"reviewer":{"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},"state":"REVIEWER"},{"updated":"2019-06-12 18:01:03.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-06-15 19:21:09.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-06-17 13:56:52.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-06-17 14:55:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c7cc98429028d2997a3c7a7e8a66a083d4a8aa9f","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-06-12 17:14:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a4c209f95530174c0b3ef6571e29868fa347ca7b","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-06-12 17:17:59.000000000","message":"Topic set to bug/1791896-stable/queens","accounts_in_message":[],"_revision_number":1},{"id":"5a475e45903573e076d5398bc6219cab3aa999a6","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-12 18:01:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master finger://ze06.softwarefactory-project.io:7979/09dc5cb0bd6c4b3a87b287679da9350f : RETRY_LIMIT in 33s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/80/664980/1/openstack-check/tripleo-build-containers-fedora-28-stein/7683257/ : SUCCESS in 43m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"29919574983d2300ffcf434acedf5b7852e6cf30","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-12 18:58:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/80/664980/1/check/kolla-build-centos-binary/99d8af5/ : SUCCESS in 39m 01s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/80/664980/1/check/kolla-build-centos-source/6de2780/ : SUCCESS in 38m 24s\n- kolla-ansible-centos-source http://logs.openstack.org/80/664980/1/check/kolla-ansible-centos-source/a9b504d/ : SUCCESS in 1h 06m 26s (non-voting)\n- kolla-ansible-centos-binary http://logs.openstack.org/80/664980/1/check/kolla-ansible-centos-binary/cb49934/ : SUCCESS in 1h 02m 33s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/80/664980/1/check/kolla-build-oraclelinux-binary/39bbf2f/ : SUCCESS in 36m 21s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/80/664980/1/check/kolla-build-oraclelinux-source/88fa3a1/ : SUCCESS in 48m 43s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/80/664980/1/check/kolla-ansible-oraclelinux-source/54bb891/ : SUCCESS in 1h 06m 07s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/80/664980/1/check/kolla-ansible-oraclelinux-binary/aeff0aa/ : SUCCESS in 1h 01m 28s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/80/664980/1/check/openstack-tox-pep8/34eed99/ : SUCCESS in 4m 05s\n- openstack-tox-py27 http://logs.openstack.org/80/664980/1/check/openstack-tox-py27/75d03d2/ : SUCCESS in 4m 07s\n- openstack-tox-py35 http://logs.openstack.org/80/664980/1/check/openstack-tox-py35/088d0fa/ : SUCCESS in 4m 33s\n- build-openstack-sphinx-docs http://logs.openstack.org/80/664980/1/check/build-openstack-sphinx-docs/38f91e0/html/ : SUCCESS in 4m 45s\n- kolla-build-ubuntu-binary http://logs.openstack.org/80/664980/1/check/kolla-build-ubuntu-binary/efdb531/ : SUCCESS in 24m 53s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/80/664980/1/check/kolla-build-ubuntu-source/8c6cefd/ : SUCCESS in 27m 18s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/80/664980/1/check/kolla-ansible-ubuntu-source/40f015b/ : SUCCESS in 53m 56s (non-voting)\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/80/664980/1/check/kolla-ansible-ubuntu-binary/6680758/ : SUCCESS in 56m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ae64ff65f1838605509c2d5c7f6f6aae281ce81c","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-15 19:21:09.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a00ec58eca98bf60041bc9d9a568fe84c1c8f37d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-17 13:56:52.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5068859082f84d0e46d2e81edaf25c721c8b251a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-17 13:57:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f8929a79647dd7a0038e3b63a45f6650219df5f0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-17 14:55:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source http://logs.openstack.org/80/664980/1/gate/kolla-build-centos-source/15a5fbc/ : SUCCESS in 34m 04s\n- kolla-build-oraclelinux-source http://logs.openstack.org/80/664980/1/gate/kolla-build-oraclelinux-source/ca5db85/ : SUCCESS in 43m 30s\n- openstack-tox-pep8 http://logs.openstack.org/80/664980/1/gate/openstack-tox-pep8/28a9b6c/ : SUCCESS in 4m 01s\n- openstack-tox-py27 http://logs.openstack.org/80/664980/1/gate/openstack-tox-py27/b538944/ : SUCCESS in 3m 35s\n- openstack-tox-py35 http://logs.openstack.org/80/664980/1/gate/openstack-tox-py35/399ceb7/ : SUCCESS in 4m 17s\n- build-openstack-sphinx-docs http://logs.openstack.org/80/664980/1/gate/build-openstack-sphinx-docs/27e2037/html/ : SUCCESS in 4m 13s\n- kolla-build-ubuntu-source http://logs.openstack.org/80/664980/1/gate/kolla-build-ubuntu-source/5c6d924/ : SUCCESS in 28m 17s","accounts_in_message":[],"_revision_number":1},{"id":"1d05f305ebbe155951b8490d916dd35db37c0df0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-17 14:55:36.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fc49f6cfc50808593717bef91003c2f25f81dca9","revisions":{"fc49f6cfc50808593717bef91003c2f25f81dca9":{"kind":"REWORK","_number":1,"created":"2019-06-12 17:14:20.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/80/664980/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/80/664980/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/80/664980/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/80/664980/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/80/664980/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/80/664980/1"}}},"commit":{"parents":[{"commit":"6f7c18c15dc19a4736110a1b86ffd809b7ab255e","subject":"Merge \"Bump service versions for Queens\" into stable/queens","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6f7c18c15dc19a4736110a1b86ffd809b7ab255e"}]}],"author":{"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","date":"2018-09-11 07:20:31.000000000","tz":480},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-06-12 17:14:10.000000000","tz":60},"subject":"Handle role properly when it is changed in kolla_keystone_user module","message":"Handle role properly when it is changed in kolla_keystone_user module\n\nChange-Id: I128559f9bcfe48b786a3f52fe4e04bd47cac7a7d\nCloses-Bug: #1791896\n(cherry picked from commit e08da5f9837a91f2b6e0fc2f20877d5cec3aa14f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fc49f6cfc50808593717bef91003c2f25f81dca9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fc49f6cfc50808593717bef91003c2f25f81dca9"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
