)]}'
{"id":"openstack%2Fkolla~665116","triplet_id":"openstack%2Fkolla~master~I34c002a0ee605ca121d9d923b256e4fe1ea3d9aa","project":"openstack/kolla","branch":"master","topic":"more-flexible-image-skipping","hashtags":[],"change_id":"I34c002a0ee605ca121d9d923b256e4fe1ea3d9aa","subject":"Make SKIPPED_IMAGES more flexible","status":"MERGED","created":"2019-06-13 11:56:13.000000000","updated":"2019-06-25 08:29:16.000000000","submitted":"2019-06-25 08:29:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"665116-1561451356449-55b2706f","meta_rev_id":"1e90c81dd7c2c1d8d6add9d99bc20ee7ad890a07","_number":665116,"virtual_id_number":665116,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"date":"2019-06-24 13:10:03.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":2,"date":"2019-06-25 08:29:16.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-06-20 14:50:44.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2019-06-24 10:03:26.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2019-06-20 15:55:50.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2019-06-25 07:30:46.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-15 06:56:41.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2019-06-20 14:50:44.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-06-20 15:55:50.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-06-25 07:30:46.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-06-25 08:29:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"cd34864564bef2ebc356ee51685b0e44bbe2dc0a","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-13 11:56:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"11e5220d260423e69f5a38ab5d45955e4390d135","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-13 12:37:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/16/665116/1/openstack-check/tripleo-build-containers-fedora-28-master/65e07f0/ : SUCCESS in 34m 55s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/16/665116/1/openstack-check/tripleo-build-containers-fedora-28-stein/d1070b7/ : SUCCESS in 33m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"16aae89c9d5d0cda2fc6d1fe7ebd2ff798f54b62","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-13 13:30:14.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/16/665116/1/check/kolla-build-centos-binary/a059c86/ : SUCCESS in 47m 51s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/16/665116/1/check/kolla-build-centos-source/2f0ab41/ : SUCCESS in 56m 19s\n- kolla-ansible-centos-source http://logs.openstack.org/16/665116/1/check/kolla-ansible-centos-source/a4d72a1/ : SUCCESS in 1h 11m 16s\n- kolla-ansible-centos-binary http://logs.openstack.org/16/665116/1/check/kolla-ansible-centos-binary/d5f1c75/ : SUCCESS in 1h 02m 36s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/16/665116/1/check/kolla-ansible-centos-source-upgrade/81e28af/ : SUCCESS in 1h 22m 10s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/16/665116/1/check/tripleo-build-containers-centos-7/cd484d6/ : SUCCESS in 43m 46s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/16/665116/1/check/kolla-build-oraclelinux-binary/4dbe4fe/ : FAILURE in 37m 28s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/16/665116/1/check/kolla-build-oraclelinux-source/5d2ce8f/ : FAILURE in 50m 53s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/16/665116/1/check/kolla-ansible-oraclelinux-source/cc69f5f/ : SUCCESS in 1h 24m 02s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/16/665116/1/check/kolla-ansible-oraclelinux-binary/55d32f3/ : SUCCESS in 59m 38s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/16/665116/1/check/openstack-tox-lower-constraints/7f2b09e/ : SUCCESS in 3m 17s\n- openstack-tox-pep8 http://logs.openstack.org/16/665116/1/check/openstack-tox-pep8/520e3f1/ : SUCCESS in 3m 04s\n- openstack-tox-py27 http://logs.openstack.org/16/665116/1/check/openstack-tox-py27/fd9f80e/ : SUCCESS in 2m 53s\n- openstack-tox-py36 http://logs.openstack.org/16/665116/1/check/openstack-tox-py36/6f288eb/ : SUCCESS in 3m 19s\n- openstack-tox-py37 http://logs.openstack.org/16/665116/1/check/openstack-tox-py37/5f7bb4d/ : SUCCESS in 3m 28s\n- openstack-tox-docs http://logs.openstack.org/16/665116/1/check/openstack-tox-docs/0a306d9/html/ : SUCCESS in 3m 07s\n- build-openstack-releasenotes http://logs.openstack.org/16/665116/1/check/build-openstack-releasenotes/11d9374/html/ : SUCCESS in 4m 14s\n- kolla-build-ubuntu-binary http://logs.openstack.org/16/665116/1/check/kolla-build-ubuntu-binary/704d7cb/ : FAILURE in 24m 03s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/16/665116/1/check/kolla-build-ubuntu-source/b512389/ : FAILURE in 31m 03s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/16/665116/1/check/kolla-ansible-ubuntu-source/30647ce/ : SUCCESS in 45m 00s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/16/665116/1/check/kolla-ansible-ubuntu-binary/a72b58e/ : SUCCESS in 49m 39s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/16/665116/1/check/kolla-ansible-ubuntu-source-upgrade/06e3bbb/ : SUCCESS in 1h 07m 43s","accounts_in_message":[],"_revision_number":1},{"id":"d162794509b0cac1c95ae4c2b1fd688918aebb31","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-13 14:31:06.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"63de326fcb23a1a4677a19363de45e1db38c964c","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-13 15:34:42.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/16/665116/2/openstack-check/tripleo-build-containers-fedora-28-master/bb67d40/ : SUCCESS in 30m 27s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/16/665116/2/openstack-check/tripleo-build-containers-fedora-28-stein/bd86807/ : SUCCESS in 32m 42s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"99dd4c8fcf0ffcd1c5ea0d56be0f46f0db0a6498","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-13 17:48:36.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/16/665116/2/check/kolla-build-centos-binary/0f04b28/ : SUCCESS in 43m 48s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/16/665116/2/check/kolla-build-centos-source/a0cd093/ : SUCCESS in 48m 18s\n- kolla-ansible-centos-source http://logs.openstack.org/16/665116/2/check/kolla-ansible-centos-source/bf7f61c/ : SUCCESS in 1h 08m 23s\n- kolla-ansible-centos-binary http://logs.openstack.org/16/665116/2/check/kolla-ansible-centos-binary/bd8a199/ : SUCCESS in 1h 11m 29s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/16/665116/2/check/kolla-ansible-centos-source-upgrade/9109cc5/ : SUCCESS in 1h 18m 42s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/16/665116/2/check/tripleo-build-containers-centos-7/a22e8cc/ : SUCCESS in 49m 52s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/16/665116/2/check/kolla-build-oraclelinux-binary/e8d8654/ : SUCCESS in 47m 37s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/16/665116/2/check/kolla-build-oraclelinux-source/f6c2de7/ : SUCCESS in 49m 51s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/16/665116/2/check/kolla-ansible-oraclelinux-source/906105c/ : SUCCESS in 1h 05m 48s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/16/665116/2/check/kolla-ansible-oraclelinux-binary/c9b67da/ : SUCCESS in 1h 05m 55s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/16/665116/2/check/openstack-tox-lower-constraints/7f41485/ : SUCCESS in 3m 27s\n- openstack-tox-pep8 http://logs.openstack.org/16/665116/2/check/openstack-tox-pep8/c7e9582/ : SUCCESS in 3m 54s\n- openstack-tox-py27 http://logs.openstack.org/16/665116/2/check/openstack-tox-py27/d733cd2/ : SUCCESS in 4m 02s\n- openstack-tox-py36 http://logs.openstack.org/16/665116/2/check/openstack-tox-py36/55be706/ : SUCCESS in 3m 55s\n- openstack-tox-py37 http://logs.openstack.org/16/665116/2/check/openstack-tox-py37/7b46637/ : SUCCESS in 5m 18s\n- openstack-tox-docs http://logs.openstack.org/16/665116/2/check/openstack-tox-docs/19f328b/html/ : SUCCESS in 3m 29s\n- build-openstack-releasenotes http://logs.openstack.org/16/665116/2/check/build-openstack-releasenotes/e4702e6/html/ : SUCCESS in 4m 19s\n- kolla-build-ubuntu-binary http://logs.openstack.org/16/665116/2/check/kolla-build-ubuntu-binary/d2e203e/ : SUCCESS in 27m 54s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/16/665116/2/check/kolla-build-ubuntu-source/10d2a5c/ : SUCCESS in 27m 53s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/16/665116/2/check/kolla-ansible-ubuntu-source/12771b6/ : SUCCESS in 57m 47s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/16/665116/2/check/kolla-ansible-ubuntu-binary/f6ce0f1/ : SUCCESS in 1h 02m 08s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/16/665116/2/check/kolla-ansible-ubuntu-source-upgrade/17ef937/ : SUCCESS in 1h 05m 22s","accounts_in_message":[],"_revision_number":2},{"id":"ecfe6750a1aa2e47e7fb860782e02d09c0ba5cd1","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-15 18:55:23.000000000","message":"Patch Set 2: Code-Review+1\n\nhave to check on machine","accounts_in_message":[],"_revision_number":2},{"id":"d29c4409c959ecaffb5c29f686bda08b5edab890","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-19 13:11:18.000000000","message":"Patch Set 2: Code-Review-1\n\n(3 comments)\n\nMakes sense, I checked all the skips by expanding the new format into the old and diffing. The crane removal is probably what caused the merge conflict.","accounts_in_message":[],"_revision_number":2},{"id":"29e8a4170b7ecbe1d4a2625304e37c3ae379e73b","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-19 15:44:44.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nI suspect so too. I will do a rebase and hope to get merge soon so that you can\u0027t afford to remove more images. :P","accounts_in_message":[],"_revision_number":2},{"id":"3bb6e1bfd3155d420e4a4bb9b34416bee9a0143e","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-19 17:33:09.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"7cb0321e2940cfe3b0bef3c33ed4d594fb618243","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-19 17:52:49.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"86908b9f61ca697e19c1f9ca250c767d565c0004","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-19 17:54:51.000000000","message":"Patch Set 4:\n\nrebased, confirmed only crane removed, reordered distro sections","accounts_in_message":[],"_revision_number":4},{"id":"d6fc05240833f6085d439b38451689a8b751c614","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-19 18:31:50.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/16/665116/4/openstack-check/tripleo-build-containers-fedora-28-master/502178f/ : FAILURE in 5m 51s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/16/665116/4/openstack-check/tripleo-build-containers-fedora-28-stein/e419654/ : SUCCESS in 35m 53s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"fa6179ce0e3be6ae7d221f6d5bb4155c80dc08cb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-19 19:21:59.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/16/665116/4/check/kolla-build-centos-binary/5726ecf/ : SUCCESS in 26m 07s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/16/665116/4/check/kolla-build-centos-source/264c8d0/ : SUCCESS in 52m 51s\n- kolla-ansible-centos-source http://logs.openstack.org/16/665116/4/check/kolla-ansible-centos-source/ed63a20/ : SUCCESS in 1h 20m 13s\n- kolla-ansible-centos-binary http://logs.openstack.org/16/665116/4/check/kolla-ansible-centos-binary/727b0c7/ : SUCCESS in 1h 04m 41s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/16/665116/4/check/kolla-ansible-centos-source-upgrade/d90f05d/ : SUCCESS in 1h 19m 29s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/16/665116/4/check/tripleo-build-containers-centos-7/64a5a32/ : FAILURE in 17m 32s\n- openstack-tox-lower-constraints http://logs.openstack.org/16/665116/4/check/openstack-tox-lower-constraints/9b9cb84/ : SUCCESS in 3m 33s\n- openstack-tox-pep8 http://logs.openstack.org/16/665116/4/check/openstack-tox-pep8/0bae9d6/ : SUCCESS in 3m 29s\n- openstack-tox-py27 http://logs.openstack.org/16/665116/4/check/openstack-tox-py27/21b3f95/ : SUCCESS in 3m 24s\n- openstack-tox-py36 http://logs.openstack.org/16/665116/4/check/openstack-tox-py36/a307b44/ : SUCCESS in 4m 34s\n- openstack-tox-py37 http://logs.openstack.org/16/665116/4/check/openstack-tox-py37/9dc7b48/ : SUCCESS in 3m 32s\n- openstack-tox-docs http://logs.openstack.org/16/665116/4/check/openstack-tox-docs/b2ee51a/html/ : SUCCESS in 2m 33s\n- build-openstack-releasenotes http://logs.openstack.org/16/665116/4/check/build-openstack-releasenotes/56d7a63/html/ : SUCCESS in 4m 11s\n- kolla-build-ubuntu-binary http://logs.openstack.org/16/665116/4/check/kolla-build-ubuntu-binary/14533f2/ : FAILURE in 20m 05s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/16/665116/4/check/kolla-build-ubuntu-source/d0fa3e5/ : FAILURE in 27m 45s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/16/665116/4/check/kolla-ansible-ubuntu-source/ab79756/ : SUCCESS in 53m 25s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/16/665116/4/check/kolla-ansible-ubuntu-binary/aea4fe1/ : SUCCESS in 42m 39s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/16/665116/4/check/kolla-ansible-ubuntu-source-upgrade/4e8baaa/ : SUCCESS in 45m 43s","accounts_in_message":[],"_revision_number":4},{"id":"cceacf0fb5502e9da1d0298c3c123f63cabbd0ca","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-20 08:28:35.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"412f929f256a4ddbef92ad3c45c5cbb99a1e7fc1","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-20 09:12:08.000000000","message":"Patch Set 4:\n\n(1 comment)\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"add81254ec7debfcc3b1095b43d45c4d7aa79270","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-20 14:14:30.000000000","message":"Patch Set 4:\n\nrecheck\n\ndid not pick previous one, odd","accounts_in_message":[],"_revision_number":4},{"id":"89d9f86c12b84176561f3bbe994742f6f7c55dd7","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-20 14:50:44.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/16/665116/4/openstack-check/tripleo-build-containers-fedora-28-master/dc66a26/ : FAILURE in 5m 38s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/16/665116/4/openstack-check/tripleo-build-containers-fedora-28-stein/a5397c3/ : SUCCESS in 34m 28s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"db4704b30ecc604095f36a04925aa02a5e70c151","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-20 15:47:18.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/16/665116/4/check/kolla-build-centos-binary/fdbc655/ : SUCCESS in 1h 09m 19s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/16/665116/4/check/kolla-build-centos-source/0104cd6/ : SUCCESS in 55m 41s\n- kolla-ansible-centos-source http://logs.openstack.org/16/665116/4/check/kolla-ansible-centos-source/ad8bf6e/ : SUCCESS in 1h 00m 43s\n- kolla-ansible-centos-binary http://logs.openstack.org/16/665116/4/check/kolla-ansible-centos-binary/66846df/ : SUCCESS in 1h 09m 16s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/16/665116/4/check/kolla-ansible-centos-source-upgrade/7c4557d/ : SUCCESS in 1h 09m 58s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/16/665116/4/check/tripleo-build-containers-centos-7/f62da0f/ : SUCCESS in 47m 21s\n- openstack-tox-lower-constraints http://logs.openstack.org/16/665116/4/check/openstack-tox-lower-constraints/07eff1f/ : SUCCESS in 9m 59s\n- openstack-tox-pep8 http://logs.openstack.org/16/665116/4/check/openstack-tox-pep8/42ba6fd/ : SUCCESS in 3m 10s\n- openstack-tox-py27 http://logs.openstack.org/16/665116/4/check/openstack-tox-py27/12b4bb2/ : SUCCESS in 2m 47s\n- openstack-tox-py36 http://logs.openstack.org/16/665116/4/check/openstack-tox-py36/1f39a2b/ : SUCCESS in 3m 19s\n- openstack-tox-py37 http://logs.openstack.org/16/665116/4/check/openstack-tox-py37/833e417/ : SUCCESS in 3m 39s\n- openstack-tox-docs http://logs.openstack.org/16/665116/4/check/openstack-tox-docs/cb263cf/html/ : SUCCESS in 2m 39s\n- build-openstack-releasenotes http://logs.openstack.org/16/665116/4/check/build-openstack-releasenotes/40ea302/html/ : SUCCESS in 3m 51s\n- kolla-build-ubuntu-binary http://logs.openstack.org/16/665116/4/check/kolla-build-ubuntu-binary/d780d3c/ : SUCCESS in 53m 23s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/16/665116/4/check/kolla-build-ubuntu-source/7665bf6/ : SUCCESS in 51m 14s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/16/665116/4/check/kolla-ansible-ubuntu-source/f875daf/ : SUCCESS in 1h 21m 13s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/16/665116/4/check/kolla-ansible-ubuntu-binary/f5b1d1c/ : SUCCESS in 1h 09m 21s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/16/665116/4/check/kolla-ansible-ubuntu-source-upgrade/78e90df/ : SUCCESS in 1h 26m 38s","accounts_in_message":[],"_revision_number":4},{"id":"70a39deca078332cde09dcd47c661d1837ff07f4","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-20 15:55:50.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"5f4d0ba6c88105f800d740db8450fa43c1c26667","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-24 10:03:26.000000000","message":"Patch Set 4: Code-Review+2\n\nCan you add example of \u0027distro-type-arch\u0027?","accounts_in_message":[],"_revision_number":4},{"id":"820d230c158bbc6ddfc77512933593c98326c931","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-24 13:10:03.000000000","message":"Patch Set 4:\n\nYou can go with any of \u0027aarch64\u0027, \u0027debian+aarch64\u0027, \u0027binary+aarch64\u0027, \u0027debian+source+aarch64\u0027. Actually the order does not really even matter - you get the point.","accounts_in_message":[],"_revision_number":4},{"id":"e16c3de8c95a646291dbc4575937c277dd6935f9","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-25 07:30:46.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"763c5d169edb6cf2e7411139ec1c4e51330ec504","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-25 07:30:58.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"388b6b8982e0939627e9a546ab4e1d98f1dcc60a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-25 08:29:16.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"b89130e680a9c82ed681f43ef29305d1cd7b412b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-25 08:29:16.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source http://logs.openstack.org/16/665116/4/gate/kolla-build-centos-source/907239d/ : SUCCESS in 41m 42s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/16/665116/4/gate/tripleo-build-containers-centos-7/e1f1d31/ : SUCCESS in 52m 49s\n- openstack-tox-lower-constraints http://logs.openstack.org/16/665116/4/gate/openstack-tox-lower-constraints/b784a83/ : SUCCESS in 4m 08s\n- openstack-tox-pep8 http://logs.openstack.org/16/665116/4/gate/openstack-tox-pep8/e135a25/ : SUCCESS in 3m 14s\n- openstack-tox-py27 http://logs.openstack.org/16/665116/4/gate/openstack-tox-py27/5a19926/ : SUCCESS in 3m 57s\n- openstack-tox-py36 http://logs.openstack.org/16/665116/4/gate/openstack-tox-py36/c834ab7/ : SUCCESS in 3m 22s\n- openstack-tox-py37 http://logs.openstack.org/16/665116/4/gate/openstack-tox-py37/db2c534/ : SUCCESS in 3m 52s\n- openstack-tox-docs http://logs.openstack.org/16/665116/4/gate/openstack-tox-docs/7bf7770/html/ : SUCCESS in 2m 48s\n- build-openstack-releasenotes http://logs.openstack.org/16/665116/4/gate/build-openstack-releasenotes/038dced/html/ : SUCCESS in 3m 50s\n- kolla-build-ubuntu-source http://logs.openstack.org/16/665116/4/gate/kolla-build-ubuntu-source/3d5cc6f/ : SUCCESS in 28m 48s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"be54999eb14b7d3fb83fc4515b0e61645b217105","revisions":{"f4ab4ca32f2ea2ce2e8feda4f318bf256216bad6":{"kind":"REWORK","_number":1,"created":"2019-06-13 11:56:13.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/16/665116/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/665116/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/665116/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/665116/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/665116/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/665116/1"}}},"commit":{"parents":[{"commit":"39a2cf066c4c79436629017155302d36243aa338","subject":"Fix lower-constraints test fail due to failing dep build","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/39a2cf066c4c79436629017155302d36243aa338"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-06-13 11:38:34.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-06-13 11:38:34.000000000","tz":120},"subject":"Make SKIPPED_IMAGES more flexible","message":"Make SKIPPED_IMAGES more flexible\n\nWith introduction of new architectures, it seems reasonable to allow\nskipping of images based on their architecture.\nAdditionally, the existing framework had unneccessary duplication.\nThis commit makes the feature more flexible and cleaner.\nExisting skips have been programmatically rewritten.\n\nChange-Id: I34c002a0ee605ca121d9d923b256e4fe1ea3d9aa\nSigned-off-by: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f4ab4ca32f2ea2ce2e8feda4f318bf256216bad6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f4ab4ca32f2ea2ce2e8feda4f318bf256216bad6"}]},"branch":"refs/heads/master"},"41f28170a5601567bdd2943535efe9e77e8a0f14":{"kind":"REWORK","_number":2,"created":"2019-06-13 14:31:06.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/16/665116/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/665116/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/665116/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/665116/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/665116/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/665116/2"}}},"commit":{"parents":[{"commit":"39a2cf066c4c79436629017155302d36243aa338","subject":"Fix lower-constraints test fail due to failing dep build","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/39a2cf066c4c79436629017155302d36243aa338"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-06-13 11:38:34.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-06-13 14:30:50.000000000","tz":120},"subject":"Make SKIPPED_IMAGES more flexible","message":"Make SKIPPED_IMAGES more flexible\n\nWith introduction of new architectures, it seems reasonable to allow\nskipping of images based on their architecture.\nAdditionally, the existing framework had unneccessary duplication.\nThis commit makes the feature more flexible and cleaner.\nExisting skips have been programmatically rewritten.\n\nChange-Id: I34c002a0ee605ca121d9d923b256e4fe1ea3d9aa\nSigned-off-by: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/41f28170a5601567bdd2943535efe9e77e8a0f14"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/41f28170a5601567bdd2943535efe9e77e8a0f14"}]},"branch":"refs/heads/master"},"bf4ae0f8fed95a5c8bfd5b2107eaff994ecc8f6d":{"kind":"REWORK","_number":3,"created":"2019-06-19 17:33:09.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/16/665116/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/665116/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/665116/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/665116/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/665116/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/665116/3"}}},"commit":{"parents":[{"commit":"86907688802f2633d165271eff08cdcdc0284b33","subject":"Merge \"Disable and remove OracleLinux CI jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/86907688802f2633d165271eff08cdcdc0284b33"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-06-13 11:38:34.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-06-19 17:32:37.000000000","tz":120},"subject":"Make SKIPPED_IMAGES more flexible","message":"Make SKIPPED_IMAGES more flexible\n\nWith introduction of new architectures, it seems reasonable to allow\nskipping of images based on their architecture.\nAdditionally, the existing framework had unneccessary duplication.\nThis commit makes the feature more flexible and cleaner.\nExisting skips have been programmatically rewritten.\n\nChange-Id: I34c002a0ee605ca121d9d923b256e4fe1ea3d9aa\nSigned-off-by: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bf4ae0f8fed95a5c8bfd5b2107eaff994ecc8f6d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bf4ae0f8fed95a5c8bfd5b2107eaff994ecc8f6d"}]},"branch":"refs/heads/master"},"be54999eb14b7d3fb83fc4515b0e61645b217105":{"kind":"REWORK","_number":4,"created":"2019-06-19 17:52:49.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/16/665116/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/665116/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/665116/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/665116/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/665116/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/665116/4"}}},"commit":{"parents":[{"commit":"86907688802f2633d165271eff08cdcdc0284b33","subject":"Merge \"Disable and remove OracleLinux CI jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/86907688802f2633d165271eff08cdcdc0284b33"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-06-13 11:38:34.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-06-19 17:52:38.000000000","tz":120},"subject":"Make SKIPPED_IMAGES more flexible","message":"Make SKIPPED_IMAGES more flexible\n\nWith introduction of new architectures, it seems reasonable to allow\nskipping of images based on their architecture.\nAdditionally, the existing framework had unneccessary duplication.\nThis commit makes the feature more flexible and cleaner.\nExisting skips have been programmatically rewritten.\n\nChange-Id: I34c002a0ee605ca121d9d923b256e4fe1ea3d9aa\nSigned-off-by: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/be54999eb14b7d3fb83fc4515b0e61645b217105"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/be54999eb14b7d3fb83fc4515b0e61645b217105"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
