)]}'
{"id":"openstack%2Fkolla~665117","triplet_id":"openstack%2Fkolla~master~I5f2a3b9d49e0a1eb7c351d2e7be0989c5f209ecf","project":"openstack/kolla","branch":"master","topic":"drop-privileges-ovn","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2025-06-23 16:37:28.000000000","reason":"Change was marked work in progress"},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2025-06-23 16:37:28.000000000","reason":"Change was marked work in progress"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2025-06-23 16:37:28.000000000","reason":"Change was marked work in progress"},"11082":{"account":{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},"last_update":"2025-06-23 16:37:28.000000000","reason":"Change was marked work in progress"},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2025-06-23 16:37:28.000000000","reason":"Change was marked work in progress"},"27339":{"account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"last_update":"2025-06-23 16:37:28.000000000","reason":"Change was marked work in progress"}},"hashtags":[],"change_id":"I5f2a3b9d49e0a1eb7c351d2e7be0989c5f209ecf","subject":"Drop root privileges for ovn","status":"ABANDONED","created":"2019-06-13 12:11:22.000000000","updated":"2026-01-28 07:05:01.000000000","total_comment_count":18,"unresolved_comment_count":2,"work_in_progress":true,"has_review_started":true,"meta_rev_id":"7509366a3c1f7a542e7904efb61ee311eee3e207","_number":665117,"virtual_id_number":665117,"owner":{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-06-23 17:04:42.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}],"CC":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"reviewer_updates":[{"updated":"2019-06-13 12:50:08.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-06-13 13:59:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-05 09:42:12.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"CC"},{"updated":"2021-11-05 09:42:12.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"},{"updated":"2021-11-17 17:30:46.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-11-25 12:40:08.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2021-11-26 08:50:27.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2021-12-16 15:23:21.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"CC"},{"updated":"2023-03-25 20:26:43.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REMOVED"},{"updated":"2023-08-26 19:49:42.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2024-02-23 11:20:13.000000000","updated_by":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"reviewer":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"state":"REVIEWER"}],"messages":[{"id":"f6f3f477365ffbfcb21c9a22c0b46ec69195ce82","author":{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},"date":"2019-06-13 12:11:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1cdbe08f5c3aaeae3da81f2b5cc6a312329d7618","author":{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},"date":"2019-06-13 12:11:34.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"552c0bb22798dec3b5f7916ca01b2481a893db50","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-13 12:50:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/17/665117/1/openstack-check/tripleo-build-containers-fedora-28-master/2eeccb0/ : SUCCESS in 30m 46s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/17/665117/1/openstack-check/tripleo-build-containers-fedora-28-stein/53931c1/ : SUCCESS in 35m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cf8cc0b5da10a64eeef61d878ecca40e658d17e5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-13 13:59:52.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/17/665117/1/check/kolla-build-centos-binary/54d9612/ : SUCCESS in 40m 40s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/17/665117/1/check/kolla-build-centos-source/4e37d97/ : SUCCESS in 40m 48s\n- kolla-ansible-centos-source http://logs.openstack.org/17/665117/1/check/kolla-ansible-centos-source/c5072b5/ : SUCCESS in 1h 11m 54s\n- kolla-ansible-centos-binary http://logs.openstack.org/17/665117/1/check/kolla-ansible-centos-binary/40b3e2c/ : SUCCESS in 1h 01m 01s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/17/665117/1/check/kolla-ansible-centos-source-upgrade/454ec08/ : SUCCESS in 1h 20m 22s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/17/665117/1/check/tripleo-build-containers-centos-7/1e748c4/ : SUCCESS in 48m 25s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/17/665117/1/check/kolla-build-oraclelinux-binary/adc8b9c/ : SUCCESS in 46m 58s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/17/665117/1/check/kolla-build-oraclelinux-source/2b4613e/ : SUCCESS in 52m 32s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/17/665117/1/check/kolla-ansible-oraclelinux-source/974cf92/ : SUCCESS in 1h 11m 17s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/17/665117/1/check/kolla-ansible-oraclelinux-binary/9e1ed01/ : SUCCESS in 1h 04m 32s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/17/665117/1/check/openstack-tox-lower-constraints/187e666/ : FAILURE in 3m 27s\n- openstack-tox-pep8 http://logs.openstack.org/17/665117/1/check/openstack-tox-pep8/9008bd0/ : FAILURE in 5m 18s\n- openstack-tox-py27 http://logs.openstack.org/17/665117/1/check/openstack-tox-py27/b3fec4d/ : SUCCESS in 5m 07s\n- openstack-tox-py36 http://logs.openstack.org/17/665117/1/check/openstack-tox-py36/791f4c5/ : SUCCESS in 4m 23s\n- openstack-tox-py37 http://logs.openstack.org/17/665117/1/check/openstack-tox-py37/a11e5f0/ : SUCCESS in 3m 39s\n- openstack-tox-docs http://logs.openstack.org/17/665117/1/check/openstack-tox-docs/a69c3c2/html/ : SUCCESS in 3m 02s\n- kolla-build-ubuntu-binary http://logs.openstack.org/17/665117/1/check/kolla-build-ubuntu-binary/fa97453/ : FAILURE in 38m 41s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/17/665117/1/check/kolla-build-ubuntu-source/672e8c5/ : FAILURE in 33m 09s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/17/665117/1/check/kolla-ansible-ubuntu-source/5778803/ : SUCCESS in 53m 20s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/17/665117/1/check/kolla-ansible-ubuntu-binary/24f986c/ : SUCCESS in 51m 50s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/17/665117/1/check/kolla-ansible-ubuntu-source-upgrade/2d9485d/ : SUCCESS in 1h 12m 51s","accounts_in_message":[],"_revision_number":1},{"id":"8970dce434c1ff468aadafa12f629883e98cb9a6","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},"date":"2021-11-05 09:07:59.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1},{"id":"343079a47c9cd3759b6044b23c88d178ad6f12b5","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-11-05 09:42:12.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"09fdcb2dae6c2da8d75cca5a835c58d69308c963","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-11-05 09:58:15.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"33145c2f64ee1af28073ba2d6a2ec5cd5dcf9940","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-11-05 10:29:42.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3d24701a567bbc5580b921cf906a44d2e5850653","tag":"autogenerated:gerrit:restore","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-11-17 13:37:09.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"12735586d45037452426666341aa01bee46be7c6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-11-17 13:37:15.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"027ecdd96f84027fb0bec95a3ba8cf7d31e39df2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-11-17 13:38:27.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"ab774d7ce677fadf21ff10c450b05bd779dcf07d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-11-17 13:39:55.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"08a82b29ae800cd9360f6e2743c1f7ae9234a093","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-11-17 14:11:35.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"e8e736e3e2c5e2b115bebcf25c8a26aa03abef74","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-17 15:40:34.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/0a2ae12e726c4ebbbf0799fe321d989c : SUCCESS in 1h 27m 56s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/73aa2bd58b7a4b17a0506cfed8a797bc : SUCCESS in 41m 42s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/f3118470db75494e8b9ca12b36288437 : SUCCESS in 1h 11m 02s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a4db57b4d09243fa87f48db5512499f6 : SUCCESS in 4m 31s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c61090d3c7024866afea8ab4020c6388 : SUCCESS in 4m 27s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/c78af9fc5b15427bb3cc9d89951e993a : SUCCESS in 59m 44s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"fe43364fcb31d4b3374841d7447b6731b4b4f8f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-17 15:52:12.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/0bb22beed77240c5b11480140a7e98a2 : SUCCESS in 38m 06s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/94179a86fbe0441e9a0139774c136924 : SUCCESS in 35m 31s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/e40642229fc642248cf469d19f6218f6 : SUCCESS in 1h 22m 25s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/123d5c359e1d4d9295de41f63dd1cea4 : SUCCESS in 1h 18m 35s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/0e55a62ea36e46d7aa0b8bc166a31031 : SUCCESS in 1h 37m 08s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/09839527c86f4ad1904d2bd40e7f3207 : SUCCESS in 31m 37s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/12820d555a274e14a95256c96462a9e5 : SUCCESS in 22m 27s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/e8dbf58db704421b8751bc51cdfc8ae7 : SUCCESS in 59m 35s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/30ae89be938d45acaf533df5fd943104 : SUCCESS in 1h 10m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/85a4e53bfb9046d5a7afed696d0bd0ce : SUCCESS in 4m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/82adca928f724a75afa3bd19722518fe : SUCCESS in 4m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4f57cdfa2d224302a76ba3b9e03fd07d : SUCCESS in 3m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/46db352514af4f4ba8bca25e717f78ee : SUCCESS in 5m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e3ba466c9127437cb91e25a35e050375 : SUCCESS in 5m 59s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/5a06eb4725c847278495f9e03d8afe8d : SUCCESS in 21m 35s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/a1cd02b715124eeeacb508d475f450e4 : SUCCESS in 26m 26s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/567e62ce614d47a19ebbf625a8f1f354 : SUCCESS in 47m 12s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/184132f808a54b2589f220a5c5b2a4ea : SUCCESS in 53m 22s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/95b49b057db0438383aadf9f550c0e23 : SUCCESS in 1h 17m 04s","accounts_in_message":[],"_revision_number":5},{"id":"406886fe8eccdbc21641237fda1f75609b97fe10","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-11-25 09:10:24.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"fbad79c42420bd266901af614d3f130cdfd61355","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-11-25 10:25:45.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"45b608de68ec6c35b1c0909ffc729fd7f86bfed5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-25 10:53:23.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/ef31947a527749adb917a6728f0f2c11 : SUCCESS in 28m 42s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/c0ecf10a5f694c3dbcd0aec19be08ab3 : SUCCESS in 36m 08s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/8e5a7bd4e899453ea5e2746dc576c82a : SUCCESS in 1h 04m 21s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/373f539411bd455fb82a61be74baae7f : SUCCESS in 1h 17m 01s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/5c6b698f0b2f489f8a1697af03ee7872 : SUCCESS in 1h 35m 42s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/a57d1ad1febc4e25b59ad9f77e01659c : SUCCESS in 26m 52s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/749dafc66fc6408eb57bc2498c586c61 : SUCCESS in 17m 34s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/4171eb17c7f54322848b26ca6fd3dfbf : SUCCESS in 57m 54s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/0dd14ab8ecf94f0da2c1fee7ef10c4eb : SUCCESS in 1h 21m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f7037d96ff141aa872121b2937061e0 : SUCCESS in 5m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/faf1cab2444345a3a9d1b00a19e3ff63 : SUCCESS in 4m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e768a4fe20fa4a41a4272f2adb799020 : SUCCESS in 3m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/016c0a5644e64e999fd3b919baad105a : SUCCESS in 8m 08s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/2ce0b8144b444aba97e808b21a587d9e : SUCCESS in 17m 51s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/cdfb7eeb0e7545cb8c21dfcd094655c1 : SUCCESS in 39m 50s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/9c0494aec63744f48e75434116506b95 : SUCCESS in 54m 02s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d4b9f89168f344b5aa97c824fac04944 : SUCCESS in 53m 29s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/cac6103d19064b56a1680b452e7bb13a : SUCCESS in 1h 07m 11s","accounts_in_message":[],"_revision_number":5},{"id":"bba51c6bcec842deb5b586df1ac0fc904d692b41","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-25 10:55:46.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/e839d7ecce8444abbc773cec9c5897b4 : SUCCESS in 1h 41m 48s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/73c1be6009e64e6fa5e37ce4f1ac7705 : SUCCESS in 1h 34m 27s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/73088d9305954c2a8735ffc89299d177 : SUCCESS in 1h 25m 27s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fe8d33fe1ce04786b6636d999643761b : SUCCESS in 5m 42s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2a736d08e04846459553897df26dbb70 : SUCCESS in 5m 43s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/a883bf34435f418dadb43d798b595dc8 : SUCCESS in 1h 38m 24s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"9add74ba46599580ce50a199275ba5dd77e6307d","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-11-25 11:03:03.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"4b0f018eebc96ff844edf8ef02aa731cacc25377","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-11-25 12:40:08.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"c7d0e738fe40da1513361e8ccbad211ffbdeced6","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-25 13:39:47.000000000","message":"Patch Set 5:\n\nBuild failed (experimental pipeline).\n\n- kolla-build-no-infra-wheels-centos8s-source https://zuul.opendev.org/t/openstack/build/bb439debdb2e4c3db260a0b3d1ef5346 : TIMED_OUT in 3h 12m 40s\n- kolla-ansible-centos8s-source-ovn https://zuul.opendev.org/t/openstack/build/a9b6e56470584c0ca714d8f2d5fa1aa9 : FAILURE in 1h 10m 10s (non-voting)\n- kolla-build-no-infra-wheels-debian-source https://zuul.opendev.org/t/openstack/build/58363c47a7c24a719deb5e79de3f796b : SUCCESS in 39m 27s\n- kolla-build-no-infra-wheels-ubuntu-source https://zuul.opendev.org/t/openstack/build/39c3466333804eb091d56c53617b74e0 : SUCCESS in 38m 54s\n- kolla-ansible-ubuntu-source-ovn https://zuul.opendev.org/t/openstack/build/0fbf27cddc8d40099af3de1a36189787 : FAILURE in 54m 18s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"c6f06666185d90a865cfab2342d13f82988c2220","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-11-25 16:45:18.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"29023421561f26413af0150de8c3656c05a5087e","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-11-25 16:45:25.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"52f596cf4cd4287796dc274801ccce39ba63f04e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-25 18:23:30.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/cf7f43336f74422ebac2360ee07166d2 : SUCCESS in 28m 48s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/6e5e28a959c0492d8df6dca90bdec88e : SUCCESS in 43m 36s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/211e9c38b06a4d6d8a5605200542a17e : SUCCESS in 1h 13m 10s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/df2160b441044a73aa90db3e94260703 : SUCCESS in 1h 12m 56s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/923b5cda03234bae94be1673043ddd33 : SUCCESS in 1h 31m 33s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/150476d8e1154204afd4503ee667feae : SUCCESS in 28m 38s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/cd6bc35fb6814b1cac3c9b66daf38995 : SUCCESS in 17m 16s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/38ab5c211e4942158d4afc070ceb4a93 : SUCCESS in 1h 00m 29s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/ad1bdefc21dc491a8c206436bfea00ac : SUCCESS in 1h 09m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/090562c9d6fb4d4d81c9143e78135da8 : SUCCESS in 4m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c987c7034ae7451c8066c840548a37a3 : SUCCESS in 4m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cbfe263217694eefa3f11655a3dfe148 : SUCCESS in 4m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea755d4da9c149d99f8ffcc24fcdc2f2 : SUCCESS in 5m 55s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d3fa6294e2b04cf2aa9c63d0da829c89 : SUCCESS in 20m 18s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/f841bce7d49a4603a6b4f5a0ea551df4 : SUCCESS in 31m 48s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/09fd7a76bd0b46a29cc352870fb5b3b2 : SUCCESS in 56m 32s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/3035e9d5f312466c9509c88863c8d647 : SUCCESS in 56m 23s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/60a0e5251b2643b9b5dede7a358b3530 : SUCCESS in 1h 01m 36s","accounts_in_message":[],"_revision_number":6},{"id":"a92e65d298dc7e0bef87e733556e171136aecd80","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-11-25 18:27:08.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"4063054790ab12282dbcfe24e16c4bda9f061065","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-25 18:36:38.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/e02561feabee4e1fa6bbd3b0ea69bf9c : SUCCESS in 1h 47m 59s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/2c09b76cad3945ca9127bed3d5d15ebe : SUCCESS in 45m 05s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/1bfc73c4848a46478bff0d90d844ca89 : SUCCESS in 1h 22m 43s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1927d4e46ad24fc6b9b03d7e8a9ddd6f : SUCCESS in 3m 48s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/96660b9b82de4171b5e939cb9efb50ce : SUCCESS in 4m 26s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/a02f392d1a194ca1aa0ad088eed44678 : SUCCESS in 35m 09s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"5a35cd293eedf04336743a4b7486d423e5ff0fce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-11-25 18:44:36.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"30745b4fbccc6d821df49d1d92d8e1f44889351f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-11-25 18:49:00.000000000","message":"Patch Set 8: Published edit on patch set 7.","accounts_in_message":[],"_revision_number":8},{"id":"9e6ca9e786960e18ed4f32a680c33da968196e4c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-11-25 18:51:08.000000000","message":"Patch Set 9: Published edit on patch set 8.","accounts_in_message":[],"_revision_number":9},{"id":"8b2ff2eb14bb72e5eb6b72ae3642f1aebe7d01fb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-11-25 19:18:19.000000000","message":"Patch Set 10: Published edit on patch set 9.","accounts_in_message":[],"_revision_number":10},{"id":"c7961c317a5ef5b73f34fc1e494073fba6e940dc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-25 20:31:20.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/b391f30ba073456eaa2d0f81f5ea71ab : SUCCESS in 1h 08m 10s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/f845cc9468774899a92070637ecf9e66 : SUCCESS in 29m 45s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/ad6cbc3ad96546b897835421c44bc79c : SUCCESS in 1h 06m 55s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6d6b4d4060934416837f25ec8469779a : SUCCESS in 5m 08s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/db9d21d0426e444dbc781bf33e5f1db0 : SUCCESS in 4m 15s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/7ac5a2c6ecfc42348ff56a8802442557 : SUCCESS in 51m 36s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"97cbe600438025c09a7e50bc940c224c6bbac73c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-25 22:37:01.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/ed078e3bfdb24bb8808e72ea8fc69b03 : SUCCESS in 36m 14s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/355c1b00d70d47f4878d5fdcd4b21567 : TIMED_OUT in 3h 10m 28s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/77e1ef27c7494849aed51b0a560d3e64 : SUCCESS in 1h 09m 14s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/4888c53db40f4e939081385aa8a4f39d : SUCCESS in 1h 06m 50s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/f9e8caa817ee4f1193cea1bcb2b339da : SUCCESS in 1h 16m 48s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/d77bf636bc0d4a0e8dc26577c790cfe1 : SUCCESS in 25m 25s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/47bcbaed63a040e7ab7b1afdc1baca2d : SUCCESS in 15m 41s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/e24e2ad24da8474a9af4dbeb0f24340a : SUCCESS in 52m 48s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/069c19117c314b49bea14cdd20e08940 : SUCCESS in 1h 03m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/582ea49e5ac846a5a0cf3c719c5da4d1 : SUCCESS in 3m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0c04e7275726451683df191470973204 : SUCCESS in 3m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5f4d9328099746ee851c20f359f9c318 : SUCCESS in 3m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf58d06aa53246bdbf3675748705ccf0 : SUCCESS in 6m 14s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/6d5a3c49f0cb4f6aa2b4adbda9c91883 : SUCCESS in 28m 57s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/df55b172eeb74de89d24245973bd5e3a : SUCCESS in 35m 47s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e8f43abd29f74b598d5efbdca3fd2ba9 : SUCCESS in 53m 04s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d1a10e70ec1a4a34b669142486c169fc : SUCCESS in 51m 54s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/3dbda5db90a0438a810ff9b26edd0282 : SUCCESS in 1h 05m 24s","accounts_in_message":[],"_revision_number":10},{"id":"ed4bb43b70149741c92bc63d29bce8bd49c0bb35","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-11-26 08:21:56.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"d8129ab8b0abffbaade3909f2249fa5c6414318a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-11-26 08:34:16.000000000","message":"Patch Set 11: Published edit on patch set 10.","accounts_in_message":[],"_revision_number":11},{"id":"adf617c4cbaa1e31d616b79892073a0277bbb193","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2021-11-26 08:50:27.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"5852c342f3e74db6fe499a0923d328b8a48dc5b0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-11-26 08:54:44.000000000","message":"Patch Set 12: Published edit on patch set 11.","accounts_in_message":[],"_revision_number":12},{"id":"ec1fe29f60d11badc6445454eae8e734eb76a602","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-11-26 08:55:16.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"95d8daee487f5cb43e44b2f52702c110d7ab39e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-26 10:24:04.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/e7fbb9d745834a8a8664a31e1129ab87 : SUCCESS in 56m 00s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/2cfe3bb23a154de4adb218f927a6a708 : SUCCESS in 42m 33s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/bf70d121df3f4b8fa42eb01ee199682e : SUCCESS in 1h 10m 36s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/bc72f629f07046ac9ca9e3f869563682 : SUCCESS in 1h 11m 56s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/29da24d3adb9421895088ca7bbc95b1b : SUCCESS in 1h 27m 24s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/912bd51ed23d4e41addc5be19fc2c98d : SUCCESS in 32m 24s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/f2b945a650cc47e284f4e74ce7efe9d3 : SUCCESS in 22m 07s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/3eb96904627f4affb96ce760523584b0 : SUCCESS in 50m 53s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/99bad5e0c2e2462286a7a116ff054b7d : SUCCESS in 1h 22m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c42966d9eed47cca520aa5f3a1751f2 : SUCCESS in 4m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/436b7589e63548d5be483973e417b1d6 : SUCCESS in 3m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dea134a8f0504e3a833934df09cf0fa4 : SUCCESS in 3m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/013d068b2e4b4d9cbf25f82e47f120b4 : SUCCESS in 6m 14s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/cfeac75c476548769851608694c47d85 : SUCCESS in 20m 00s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d2f8f73bca00431fb687c25588e63ca6 : SUCCESS in 28m 49s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/d3d69f9c34c743ff8af45ce75f5f0c11 : SUCCESS in 57m 10s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/2040ccd11e384da89df30284d576c88c : SUCCESS in 51m 49s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ec2fc0433dad472c98d8b33e7ccb7085 : FAILURE in 1h 13m 14s","accounts_in_message":[],"_revision_number":12},{"id":"99edfa85330f3ce762f4b241e986917e308e11c4","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-11-26 10:28:34.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"3303f9adbbda805eaf31c4f49b1e0172dfb5a584","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-26 10:39:24.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/7a68f93ef71142f2811012eaa1c8fa45 : SUCCESS in 55m 05s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/725a98383f3147cfa08ca44da57d51b0 : SUCCESS in 44m 02s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/8d8de1a8f25b40478102d588be08d513 : SUCCESS in 1h 34m 29s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1b8c7c4504b34fceb17c1d20e3b5a22b : SUCCESS in 3m 53s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/06f53f103a8f4b42af4f92997c25200e : SUCCESS in 4m 36s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/9a66af4e62ca4000a8f8340435ebb590 : SUCCESS in 41m 47s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"0598583e4571dce78842ef6125a0a766c110fb36","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-11-26 12:08:42.000000000","message":"Patch Set 13: Published edit on patch set 12.","accounts_in_message":[],"_revision_number":13},{"id":"8b658a34032f2f920af35234812a7e4f6a88dfb4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-26 13:44:12.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/6c0fabc3082942509bb62b0f784c4b55 : SUCCESS in 46m 35s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/eee202ccde2d4ff2ad2debb559b7e7d8 : SUCCESS in 57m 58s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/b692130de44e44d58c470003052a1d1f : SUCCESS in 1h 24m 26s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/882fea3d115a408b95e085c1983e406f : SUCCESS in 1h 12m 12s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/4f77274156a4481f8149d976b4612a43 : SUCCESS in 1h 33m 24s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/38a38e6efe514b5cb2da3421a5860724 : SUCCESS in 36m 57s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/8610ed2dbec540a5bea5ceee36be3327 : SUCCESS in 16m 42s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/4b76c5df12a24741a8be346fafd0a1c9 : SUCCESS in 1h 14m 33s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/917cbdb01fae42e9b9645eb0cbb0e892 : SUCCESS in 57m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5024c89ab6a43b2a13d220d6ef66d24 : SUCCESS in 4m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3034a8ea7ab14dffb5dd0e8c12c92a52 : SUCCESS in 3m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/eb9e6d255a424b3294151ec97ef8313c : SUCCESS in 3m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/abba4bbfab014cee8394af6d175f69ee : SUCCESS in 5m 49s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/dac70c2e50434160828b327d11a4518f : SUCCESS in 19m 19s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/5d98ce562a054202b61dd17c92a9f30d : SUCCESS in 23m 29s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/6b9b92b38ccf431ea16816a8054b8bdb : SUCCESS in 52m 38s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e10c77d11f994a5cb8aa49a1d62161f4 : SUCCESS in 57m 51s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/35a31f23da454b6eb5164a68087fa0cb : SUCCESS in 1h 16m 31s","accounts_in_message":[],"_revision_number":13},{"id":"1a9dfa28ca8416910e20375e31e50b9226571835","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-26 14:32:19.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/0b831e568ee64260a01fd9132b062c01 : SUCCESS in 2h 20m 41s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/5f1685c6f31046b3b9c228b2bbef8c2a : SUCCESS in 2h 10m 00s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/76402d70508049f5baafadbc5095c8ae : SUCCESS in 1h 22m 25s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/62a3df57f6f741158bd6cd92af8dec32 : SUCCESS in 5m 35s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b32ee04bba5840409d93de4a1c9ddb91 : SUCCESS in 5m 47s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/a28d6453db5949558c0b643cae7cc445 : SUCCESS in 1h 40m 37s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"e72afbb50faed3d4814521a9c9f60f5d71f53d45","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2021-11-26 14:38:53.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"a446cce80746fb29b280155e730e12d6789ef344","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-26 17:50:52.000000000","message":"Patch Set 13:\n\nBuild failed (experimental pipeline).\n\n- kolla-build-no-infra-wheels-centos8s-source https://zuul.opendev.org/t/openstack/build/9c540ca3a96f4e549a5bafa32db06453 : TIMED_OUT in 3h 07m 00s\n- kolla-ansible-centos8s-source-ovn https://zuul.opendev.org/t/openstack/build/656884fd73494cd3becce712b2335a3a : FAILURE in 1h 03m 53s (non-voting)\n- kolla-build-no-infra-wheels-debian-source https://zuul.opendev.org/t/openstack/build/cb5be45b14874084bd2c0be156c49fb0 : SUCCESS in 35m 55s\n- kolla-build-no-infra-wheels-ubuntu-source https://zuul.opendev.org/t/openstack/build/fb361f9941e149e292adcf6d6c2dca39 : SUCCESS in 32m 03s\n- kolla-ansible-ubuntu-source-ovn https://zuul.opendev.org/t/openstack/build/b810f6b28da04aa1965592c76bf70480 : FAILURE in 31m 55s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"20efae53fdebc6fddc319c7cef287ab4d6544197","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-12-16 11:39:57.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"3b1217f00055de8d34c10308560907ce23546e39","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-16 13:15:44.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/78cc9b0e14ea47488f077f81a685253a : SUCCESS in 29m 06s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/d73519c67b1f4546b39a153d2cafe76d : SUCCESS in 42m 00s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/fc830bf2bc7d4713af8ce4df94f915b4 : SUCCESS in 1h 19m 30s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/3fb9ccda38d74df28b3d46152d377345 : SUCCESS in 1h 27m 12s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/a30e8c52a32447bba052e843125258c7 : SUCCESS in 1h 26m 45s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/947287e1dd0d45318ab23017f444a69a : SUCCESS in 27m 30s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/ba6e64791e4946ae9ed1e0c28bae664e : SUCCESS in 16m 33s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/a036a6898c3e4d76ba0c4292541abc7e : SUCCESS in 1h 09m 41s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/cf228be1c8f94c3887b588769247e2ce : SUCCESS in 1h 18m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/838b9626a11e441d91e7e1e894841b9d : SUCCESS in 4m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a461b7b371024b93837300924e984c21 : SUCCESS in 4m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6f8f8a7dee0b4783b40306cdee5f7ecf : SUCCESS in 3m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea6941e1fbd945399d6509cde5856f02 : SUCCESS in 6m 15s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c76a15da16284d3cae504097e0298a54 : SUCCESS in 18m 10s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/f3b3567abc174c39a13f1f88aefda323 : SUCCESS in 28m 36s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/3895368523d3458e98dd987d0e9d5cf6 : SUCCESS in 50m 53s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/2217252b1c2d4a6eb5d441bfb8ad36f3 : SUCCESS in 50m 20s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/972fb12e8374404cb5c769c789bcb6ec : SUCCESS in 1h 01m 18s","accounts_in_message":[],"_revision_number":14},{"id":"8a23d457d82a79e44e2a7e0da2c4e41008a23a37","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-16 13:41:27.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/ee330e89120f4b70bcc39dff97e0c068 : SUCCESS in 1h 16m 35s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/f33d22a66c984cd78e5b3d557b5b5b87 : SUCCESS in 1h 16m 42s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/6d5e123f73ad45dfb6447cbf8cd9dc36 : SUCCESS in 1h 52m 29s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/eaa3f1e71a4941339b53fa866c2e221a : SUCCESS in 5m 19s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d7e88cd398b64b148212e1be846adbe1 : SUCCESS in 5m 03s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/2398d421ddff48d6a5f5785f19d6c922 : SUCCESS in 1h 10m 07s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"41ea68cc1d4fa2095040b662e2de43bff062e9a5","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-12-16 13:49:09.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"3da26a4b4f7f76ce1ed2cffbd13fa962eec98b56","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-12-16 14:54:19.000000000","message":"Patch Set 15: Published edit on patch set 14.","accounts_in_message":[],"_revision_number":15},{"id":"dc633e10b651f7474e6865417c51ef8f92481b0b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-16 16:08:09.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/c986d4ff3c724ebbbf69078b4e8f5159 : SUCCESS in 25m 10s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/2e4f358524fb45a79e51419ef1fce483 : SUCCESS in 39m 27s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/8daf9617f67a41859ccb2941653533ce : SUCCESS in 1h 06m 48s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/664e8ac609314cb6b8be68841688ee4a : SUCCESS in 1h 00m 48s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/b7a48cce564847ec84f0b026753d811b : SUCCESS in 1h 10m 47s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/7707c233781e41f49f4816b2f81479c0 : SUCCESS in 33m 49s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/461aafe84b1c4886a75181fc088fb9b8 : SUCCESS in 16m 29s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/1c696faea041410ba5973d324a7471ed : SUCCESS in 1h 06m 00s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/8d5a52f68faa49abbea4006c6fbf722a : SUCCESS in 1h 05m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51bf82dd55804750a819ad768babc2e3 : SUCCESS in 4m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e18d7a7d95c741c3a4cdb4b0f9719fd9 : SUCCESS in 4m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d85945583eb4499fac588ce84b61671a : SUCCESS in 5m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a959224db3e64e2a99ccd889575c8ac9 : SUCCESS in 6m 08s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/113016c8547f4cb3988fa0be1b9ce04f : SUCCESS in 20m 48s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/bf5980bb70944245a24fe160544f682d : SUCCESS in 33m 04s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/7b57c8b83c424a068ea86143a99c439d : SUCCESS in 43m 40s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/50e2c7d659ba48cc95226c61be313555 : SUCCESS in 55m 04s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/1b76158d798d42d9aa383e2b5d458394 : SUCCESS in 53m 30s","accounts_in_message":[],"_revision_number":15},{"id":"7986e05b016582d3133e9fa029530f8e0fb51682","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-12-16 16:20:07.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"28022b8b0d4b2bd41e23d0e0d7b99c3419b794e9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-16 17:09:56.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/2e0f9383a4bd41529836d313d4814419 : SUCCESS in 47m 19s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/14c93fac3c7b4c3db4eeacce445a94b0 : SUCCESS in 39m 53s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/eada563ad56149caad35095ec1cac1fc : SUCCESS in 2h 13m 19s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/bf020113b90c4d4988f887412d6b8697 : SUCCESS in 4m 56s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0197c6c7e39245d49ed42e0f75a10c8c : SUCCESS in 4m 32s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/d3d5f49fd06d4027b5bf45a1136c92c2 : SUCCESS in 1h 20m 28s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"3794335e43a90ee9d6dacd6ef29b21254dd1fa72","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-16 17:37:43.000000000","message":"Patch Set 15:\n\nBuild succeeded (experimental pipeline).\n\n- kolla-build-no-infra-wheels-centos8s-source https://zuul.opendev.org/t/openstack/build/425f3944ab934e7fa4576b2556c76f70 : SUCCESS in 36m 09s\n- kolla-ansible-centos8s-source-ovn https://zuul.opendev.org/t/openstack/build/bb99b8014be348e8ac58287b194c0571 : FAILURE in 1h 12m 24s (non-voting)\n- kolla-build-no-infra-wheels-debian-source https://zuul.opendev.org/t/openstack/build/13b870c25e7046268f0b66e8ea0d5e3e : SUCCESS in 28m 53s\n- kolla-build-no-infra-wheels-ubuntu-source https://zuul.opendev.org/t/openstack/build/47ddd4e08c9c41e5b2afbee4ccacabd5 : SUCCESS in 31m 57s\n- kolla-ansible-ubuntu-source-ovn https://zuul.opendev.org/t/openstack/build/4f66a7a6edb64af3a8f7ad0a0dc5be84 : FAILURE in 53m 59s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"a1345c202c4c112408dc172effae3fa72e22800f","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2023-03-25 20:26:43.000000000","message":"Removed reviewer \u003cGERRIT_ACCOUNT_30491\u003e.","accounts_in_message":[{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}],"_revision_number":15},{"id":"ce80d39f74c4dd74d55eeb1549e2067c44f1856f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-01-05 13:44:25.000000000","message":"Uploaded patch set 16.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":16},{"id":"e0f0308357336a881361618083fc5e1ec6ef3946","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-05 15:17:14.000000000","message":"Patch Set 16:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1a629b712f5d41ed9de7f517fa477890\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/e323e9a6d5e242909fb8afb7a12e5353 : SUCCESS in 20m 53s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/58b18edf871146a8888abe59e1f1ed29 : SUCCESS in 16m 25s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/6853a641209e49f7b04397ec3695b389 : SUCCESS in 1h 04m 48s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5fa3a5b7f8064e838fa01c785f5c1e1e : SUCCESS in 4m 47s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ce444f7c37f9491aa57e4bf5a4d3a575 : SUCCESS in 2m 51s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/ec77ee2dbf094487b1f4c9fea517d3a3 : SUCCESS in 4m 55s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/ff416afe0e884a2cb8c770a1e49ae0c1 : SUCCESS in 20m 06s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/6fa5f9802aef41b488ee5dc41c94fcc4 : SUCCESS in 1h 22m 37s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"d57abe27ec7af45b4ed0cfbc4d9b7e717ef81396","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-05 15:30:26.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3b2f5e24c1e54fbda0caf7adca921bea\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/42630348b0da4034abb0243cd18dca39 : SUCCESS in 24m 38s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/02aff7d225dc40af9f844e7dd52b1d17 : SUCCESS in 52m 29s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/e52475e94a5e44b2b630d863d4ebbc87 : SUCCESS in 18m 40s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/edb1efdf95af41ddbf8aa0bf9540f19e : SUCCESS in 28m 54s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/c6654e03b3f0484bafb134292497d293 : SUCCESS in 1h 03m 49s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/62b9c3f5971f4bffb9d2b1cf962ea040 : SUCCESS in 1h 19m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/03945fd5c9f840e2a4846211490be9df : SUCCESS in 2m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f742f76beba44c0689b788f5aad56c49 : SUCCESS in 2m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/34d89a19354a4e4c9b8009ae8f5ccbbe : SUCCESS in 2m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/34ca3c4d42874603a1a22049bf282ee0 : SUCCESS in 2m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aace7638072c4528bee92d38a4c01431 : SUCCESS in 4m 13s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/9bbc031550544c829b4c7c37b3b9346f : SUCCESS in 20m 03s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/f51bdf9f67b645228f17ed657f1da09c : SUCCESS in 33m 40s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/22f3db8927424db994946781252a49ec : SUCCESS in 1h 10m 00s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/227bee32061e42b58d5ee17fb6b2bdf2 : SUCCESS in 1h 36m 09s\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/8ccc436b4ec74f7894dd8b5d47cef6bc : SUCCESS in 1h 19m 10s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/2f0254f6cbd849f198c9d10daf5607a5 : FAILURE in 52m 22s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/7e86eb201d0a4bceab2adfeaf72c7850 : SUCCESS in 20m 39s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/6c7f1cb44604413090da58ba08a7e17e : SUCCESS in 58m 12s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/62251ee5d5444e7997db221641560573 : SUCCESS in 56m 03s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/21e1c9d7bc0b4794a75aa141d99b2661 : SUCCESS in 1h 29m 56s\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/63e5f51049fe43ee823c741482d3fe67 : SUCCESS in 59m 55s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/84159a0a12414ca8b91c0723f35a7781 : FAILURE in 39m 53s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"d5dc5e6bfd405842c01138894859cc14099de844","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-02-23 11:20:13.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"336906df074a6751138ddac80d6573b09b224c3c","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-02-23 11:30:29.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"92a8b20548aeab8a4e781d456dd55a5f282f8291","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-23 12:51:16.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/bd04c887f5bb47e3a28707a92e61940b\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/8e3bf4938f0c4a3c85c5ec0a54051d4f : SUCCESS in 33m 05s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/bb660d37eb994e6884c2ec8dfd38dfe3 : SUCCESS in 1h 11m 07s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/78685e52334e4245bd9e09c39396321a : FAILURE in 16m 21s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/79bde39701be4f4c9f9499e2b0e39d79 : FAILURE in 30m 25s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/8247a1f00bb24975b762bb23c285bb63 : FAILURE in 13m 31s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/977e2cd6f67d45c9b053b9140859288f : FAILURE in 22m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/14e51bb9faec4b008ddd318184cb0348 : SUCCESS in 2m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2433dd3a38b34938a37418721b40d745 : SUCCESS in 2m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c92ca88f5a34428b9a400f957da1c830 : SUCCESS in 2m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c805b0fb93884c2d84c65248dea1b180 : SUCCESS in 2m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef9a1bea81da4e0aa15ec91cefb24de7 : SUCCESS in 4m 19s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/4c4c645ecfdc40389afa01c59d3a5db3 : SUCCESS in 24m 50s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/cb64dca38bec43c58bb12b1d91e2e63a : SUCCESS in 32m 03s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/6d102ef6f588412880afbc2bc2941e0a : SUCCESS in 1h 11m 02s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/b992aed02c4d4915a060bed8320bf6f0 : SUCCESS in 1h 26m 29s\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/9c4ab3c66db4473681250aa6eebe5167 : FAILURE in 1h 20m 27s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/320d19a7771a43c6bb5a67fb3996277c : FAILURE in 51m 32s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/b6adf74bd48147a4bc71f6d359d92643 : FAILURE in 14m 40s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/87730c2b2bb341c0a3ffec2df9a4c096 : FAILURE in 59m 06s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/917547f1702c4becb0be2cd5d0a22caa : FAILURE in 12m 47s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/7ae4e43a80324e2eb62667a8aa8d3a26 : FAILURE in 16m 04s\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/d2891d81d47849bfa60aa5596f03d9e9 : FAILURE in 15m 55s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/602ab212abf3456b96414aadc8bef5bd : FAILURE in 15m 48s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"21d17202bc44bcc199cac8c7e34b8e1ae7edc37b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-23 13:08:47.000000000","message":"Patch Set 16:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/063490d767ed4f95a1b106388bdfc23e\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/a59af78a4e684ffa8289b4bdf495bc3f : FAILURE in 1h 46m 23s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/196e580ef2f649249ab8154c4a9f1821 : FAILURE in 34m 16s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/150f6d3428a941ccb58a9ada745fc66f : FAILURE in 47m 27s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1ca8f2013d7a497d899cfd5b53e14608 : SUCCESS in 3m 38s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a9e3ecad271c44d09f34aa7baa5eabbc : SUCCESS in 5m 31s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/631f2fb3ea074ae09347b6423b256f1f : SUCCESS in 3m 34s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/def6352a3cd74150983e1343bdd7e8af : FAILURE in 1h 27m 51s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/486c19d6e9294b2caadc6716135d470a : FAILURE in 29m 59s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"40cc2ff558eb093361144ebb66ed2443e00d57d5","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-06-23 16:37:28.000000000","message":"Patch Set 17: Patch Set 16 was rebased\n\nThe following files contain Git conflicts:\n* kolla/common/users.py\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":17},{"id":"3028af53a594d1bb4e1c08efb2a4bcd16f67d32e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-23 17:04:42.000000000","message":"Patch Set 17: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/dc7664b9b7ab4e8e9c957d0c63f0b22d\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/0d2643e2170648a8869d2e3b495097ca : FAILURE in 2m 40s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/05024fe02ebe49e3a2d4f921728cb01a : FAILURE in 3m 58s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/26b7dec625974c69886150cb7f2d57c3 : FAILURE in 9m 12s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/6cc0a61d9698469ea0d970adafdbde48 : FAILURE in 10m 16s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/49ba2c098fa74a4a9db0f5841ae84d27 : FAILURE in 2m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3539b29dc95441eea5657dcf7b3ab751 : FAILURE in 2m 39s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/da285d2016f84825ada2fb7c4c6bcf10 : FAILURE in 3m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c30e16913c184c3d97cfdaeab6bc9150 : FAILURE in 7m 29s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0935156dc59643fb9f8a052939f8b0dc : FAILURE in 3m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7dcbaa4593bf444383822774a129afe5 : SUCCESS in 6m 59s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/7b60d6a0beee4ac68f425c84c2a4d4fc : FAILURE in 3m 03s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/02ee065cca5643e89edbe4056c3df066 : FAILURE in 1m 48s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/b4402accffde47db8d52577df480d92d : FAILURE in 3m 23s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/07f0b22d3aae47a48f2fe607a23a77b6 : FAILURE in 6m 57s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/197ee9a5e46740079fd83b4d9850c090 : FAILURE in 7m 13s\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/616b0a5b0e11476da021e3c850c3dda4 : FAILURE in 8m 14s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/a2e97384aae3447db677ef270bf8067b : FAILURE in 7m 29s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"c3c3f627c66832beba66e44bf8a8b6926e86cdbe","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-23 17:46:57.000000000","message":"Patch Set 17:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9cc38701cf6c46078dbb16a17fe7fcd9\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/6e15ce22a5e6407380aa0ecd7e7f2e95 : FAILURE in 3m 56s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/5899f3db8b0242e0be54262acb727509 : FAILURE in 8m 36s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/33ca2de357e34e9cbda60dba328534fb : FAILURE in 3m 02s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/64804b732852408e94925f8d67668b6d : FAILURE in 4m 51s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/dc959be57835493f9858d65ff80c8190 : FAILURE in 9m 31s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/591d903b1fff41ac82095a74c493b60e : FAILURE in 6m 12s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"7509366a3c1f7a542e7904efb61ee311eee3e207","tag":"autogenerated:gerrit:abandon","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-01-28 07:05:01.000000000","message":"Abandoned\n\nLet\u0027s create a new patch for this","accounts_in_message":[],"_revision_number":17}],"current_revision_number":17,"current_revision":"49395d80dde134abf82ca5b6dfeebe283b6f9b13","revisions":{"c0565b36992970b3e31fee3e2a4b994d4594f015":{"kind":"REWORK","_number":1,"created":"2019-06-13 12:11:22.000000000","uploader":{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},"ref":"refs/changes/17/665117/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/17/665117/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/665117/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/665117/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/665117/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/17/665117/1"}}},"commit":{"parents":[{"commit":"b5c8b595a7f9c4aebb35b0ceacc236ec5f4eb350","subject":"Merge \"Fix issues with Percona in Zuul and clean up a little bit\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b5c8b595a7f9c4aebb35b0ceacc236ec5f4eb350"}]}],"author":{"name":"Kamil Sambor","email":"ksambor@redhat.com","date":"2019-06-13 11:52:16.000000000","tz":120},"committer":{"name":"Kamil Sambor","email":"ksambor@redhat.com","date":"2019-06-13 12:11:19.000000000","tz":120},"subject":"Drop root privileges for ovn","message":"Drop root privileges for ovn\n\nDrop root privileges for ovn from security reasones.\n\nChange-Id: I5f2a3b9d49e0a1eb7c351d2e7be0989c5f209ecf\nRelated-Bug: #1781534\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c0565b36992970b3e31fee3e2a4b994d4594f015"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c0565b36992970b3e31fee3e2a4b994d4594f015"}]},"branch":"refs/heads/master"},"9fa40c41f0a3227560ebf2ab0b4347491a75ffef":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-11-17 13:37:15.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/17/665117/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/17/665117/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/665117/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/665117/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/665117/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/17/665117/2"}}},"commit":{"parents":[{"commit":"7b6f6a054c1a91451ebf655d613c291a818fa8de","subject":"Merge \"Debian: we want Openvswitch and ovn from backports\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7b6f6a054c1a91451ebf655d613c291a818fa8de"}]}],"author":{"name":"Kamil Sambor","email":"ksambor@redhat.com","date":"2019-06-13 11:52:16.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-11-17 13:37:15.000000000","tz":0},"subject":"Drop root privileges for ovn","message":"Drop root privileges for ovn\n\nDrop root privileges for ovn from security reasones.\n\nChange-Id: I5f2a3b9d49e0a1eb7c351d2e7be0989c5f209ecf\nRelated-Bug: #1781534\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9fa40c41f0a3227560ebf2ab0b4347491a75ffef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9fa40c41f0a3227560ebf2ab0b4347491a75ffef"}]},"branch":"refs/heads/master","description":"Rebase"},"e89fc2ec3833c9ed7b7b7953ca2ff8415e7e76f8":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2021-11-17 13:38:27.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/17/665117/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/17/665117/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/665117/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/665117/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/665117/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/17/665117/3"}}},"commit":{"parents":[{"commit":"7b6f6a054c1a91451ebf655d613c291a818fa8de","subject":"Merge \"Debian: we want Openvswitch and ovn from backports\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7b6f6a054c1a91451ebf655d613c291a818fa8de"}]}],"author":{"name":"Kamil Sambor","email":"ksambor@redhat.com","date":"2019-06-13 11:52:16.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-11-17 13:38:27.000000000","tz":0},"subject":"Drop root privileges for ovn","message":"Drop root privileges for ovn\n\nChange-Id: I5f2a3b9d49e0a1eb7c351d2e7be0989c5f209ecf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e89fc2ec3833c9ed7b7b7953ca2ff8415e7e76f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e89fc2ec3833c9ed7b7b7953ca2ff8415e7e76f8"}]},"branch":"refs/heads/master","description":"Edit commit message"},"fa56d0e95d8c0e32a64a1e8c3f205ca2146b32e0":{"kind":"REWORK","_number":4,"created":"2021-11-17 13:39:55.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/17/665117/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/17/665117/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/665117/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/665117/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/665117/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/17/665117/4"}}},"commit":{"parents":[{"commit":"7b6f6a054c1a91451ebf655d613c291a818fa8de","subject":"Merge \"Debian: we want Openvswitch and ovn from backports\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7b6f6a054c1a91451ebf655d613c291a818fa8de"}]}],"author":{"name":"Kamil Sambor","email":"ksambor@redhat.com","date":"2019-06-13 11:52:16.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-11-17 13:39:55.000000000","tz":0},"subject":"Drop root privileges for ovn","message":"Drop root privileges for ovn\n\nChange-Id: I5f2a3b9d49e0a1eb7c351d2e7be0989c5f209ecf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fa56d0e95d8c0e32a64a1e8c3f205ca2146b32e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fa56d0e95d8c0e32a64a1e8c3f205ca2146b32e0"}]},"branch":"refs/heads/master"},"771673dcaf5fbe3cab3f25c6eda733df928ed397":{"kind":"REWORK","_number":5,"created":"2021-11-17 14:11:35.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/17/665117/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/17/665117/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/665117/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/665117/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/665117/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/17/665117/5"}}},"commit":{"parents":[{"commit":"7b6f6a054c1a91451ebf655d613c291a818fa8de","subject":"Merge \"Debian: we want Openvswitch and ovn from backports\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7b6f6a054c1a91451ebf655d613c291a818fa8de"}]}],"author":{"name":"Kamil Sambor","email":"ksambor@redhat.com","date":"2019-06-13 11:52:16.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-11-17 14:11:28.000000000","tz":0},"subject":"Drop root privileges for ovn","message":"Drop root privileges for ovn\n\nChange-Id: I5f2a3b9d49e0a1eb7c351d2e7be0989c5f209ecf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/771673dcaf5fbe3cab3f25c6eda733df928ed397"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/771673dcaf5fbe3cab3f25c6eda733df928ed397"}]},"branch":"refs/heads/master"},"1eb411696df2482eec76c57fe3030776964d10f3":{"kind":"REWORK","_number":6,"created":"2021-11-25 16:45:18.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/17/665117/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/17/665117/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/665117/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/665117/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/665117/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/17/665117/6"}}},"commit":{"parents":[{"commit":"7b6f6a054c1a91451ebf655d613c291a818fa8de","subject":"Merge \"Debian: we want Openvswitch and ovn from backports\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7b6f6a054c1a91451ebf655d613c291a818fa8de"}]}],"author":{"name":"Kamil Sambor","email":"ksambor@redhat.com","date":"2019-06-13 11:52:16.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-11-25 16:45:16.000000000","tz":0},"subject":"Drop root privileges for ovn","message":"Drop root privileges for ovn\n\nChange-Id: I5f2a3b9d49e0a1eb7c351d2e7be0989c5f209ecf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1eb411696df2482eec76c57fe3030776964d10f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1eb411696df2482eec76c57fe3030776964d10f3"}]},"branch":"refs/heads/master"},"9eb3cf9a0772b126bcaaeb4db3f91e50cd1c6fe1":{"kind":"REWORK","_number":7,"created":"2021-11-25 18:44:36.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/17/665117/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/17/665117/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/665117/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/665117/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/665117/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/17/665117/7"}}},"commit":{"parents":[{"commit":"7b6f6a054c1a91451ebf655d613c291a818fa8de","subject":"Merge \"Debian: we want Openvswitch and ovn from backports\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7b6f6a054c1a91451ebf655d613c291a818fa8de"}]}],"author":{"name":"Kamil Sambor","email":"ksambor@redhat.com","date":"2019-06-13 11:52:16.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-11-25 18:44:17.000000000","tz":0},"subject":"Drop root privileges for ovn","message":"Drop root privileges for ovn\n\nChange-Id: I5f2a3b9d49e0a1eb7c351d2e7be0989c5f209ecf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9eb3cf9a0772b126bcaaeb4db3f91e50cd1c6fe1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9eb3cf9a0772b126bcaaeb4db3f91e50cd1c6fe1"}]},"branch":"refs/heads/master"},"06abde22bf6d38ac47f73093a881a8262ec83db6":{"kind":"REWORK","_number":8,"created":"2021-11-25 18:49:00.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/17/665117/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/17/665117/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/665117/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/665117/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/665117/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/17/665117/8"}}},"commit":{"parents":[{"commit":"7b6f6a054c1a91451ebf655d613c291a818fa8de","subject":"Merge \"Debian: we want Openvswitch and ovn from backports\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7b6f6a054c1a91451ebf655d613c291a818fa8de"}]}],"author":{"name":"Kamil Sambor","email":"ksambor@redhat.com","date":"2019-06-13 11:52:16.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-11-25 18:48:59.000000000","tz":0},"subject":"Drop root privileges for ovn","message":"Drop root privileges for ovn\n\nChange-Id: I5f2a3b9d49e0a1eb7c351d2e7be0989c5f209ecf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/06abde22bf6d38ac47f73093a881a8262ec83db6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/06abde22bf6d38ac47f73093a881a8262ec83db6"}]},"branch":"refs/heads/master"},"cfe540ef0379277a62044861ba11588b18a25c77":{"kind":"REWORK","_number":9,"created":"2021-11-25 18:51:08.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/17/665117/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/17/665117/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/665117/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/665117/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/665117/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/17/665117/9"}}},"commit":{"parents":[{"commit":"7b6f6a054c1a91451ebf655d613c291a818fa8de","subject":"Merge \"Debian: we want Openvswitch and ovn from backports\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7b6f6a054c1a91451ebf655d613c291a818fa8de"}]}],"author":{"name":"Kamil Sambor","email":"ksambor@redhat.com","date":"2019-06-13 11:52:16.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-11-25 18:51:08.000000000","tz":0},"subject":"Drop root privileges for ovn","message":"Drop root privileges for ovn\n\nChange-Id: I5f2a3b9d49e0a1eb7c351d2e7be0989c5f209ecf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cfe540ef0379277a62044861ba11588b18a25c77"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cfe540ef0379277a62044861ba11588b18a25c77"}]},"branch":"refs/heads/master"},"19f220ae8909f4b5ce80a7c8847f4bf267b152b5":{"kind":"REWORK","_number":10,"created":"2021-11-25 19:18:19.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/17/665117/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/17/665117/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/665117/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/665117/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/665117/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/17/665117/10"}}},"commit":{"parents":[{"commit":"7b6f6a054c1a91451ebf655d613c291a818fa8de","subject":"Merge \"Debian: we want Openvswitch and ovn from backports\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7b6f6a054c1a91451ebf655d613c291a818fa8de"}]}],"author":{"name":"Kamil Sambor","email":"ksambor@redhat.com","date":"2019-06-13 11:52:16.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-11-25 19:18:19.000000000","tz":0},"subject":"Drop root privileges for ovn","message":"Drop root privileges for ovn\n\nChange-Id: I5f2a3b9d49e0a1eb7c351d2e7be0989c5f209ecf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/19f220ae8909f4b5ce80a7c8847f4bf267b152b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/19f220ae8909f4b5ce80a7c8847f4bf267b152b5"}]},"branch":"refs/heads/master"},"b6fa1946d4e79042b7d5c86683f8523b18f20516":{"kind":"REWORK","_number":11,"created":"2021-11-26 08:34:16.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/17/665117/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/17/665117/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/665117/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/665117/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/665117/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/17/665117/11"}}},"commit":{"parents":[{"commit":"7b6f6a054c1a91451ebf655d613c291a818fa8de","subject":"Merge \"Debian: we want Openvswitch and ovn from backports\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7b6f6a054c1a91451ebf655d613c291a818fa8de"}]}],"author":{"name":"Kamil Sambor","email":"ksambor@redhat.com","date":"2019-06-13 11:52:16.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-11-26 08:34:16.000000000","tz":0},"subject":"Drop root privileges for ovn","message":"Drop root privileges for ovn\n\nChange-Id: I5f2a3b9d49e0a1eb7c351d2e7be0989c5f209ecf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b6fa1946d4e79042b7d5c86683f8523b18f20516"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b6fa1946d4e79042b7d5c86683f8523b18f20516"}]},"branch":"refs/heads/master"},"33bca772a74f2973cf83afb13ad8bb3d156a7e9a":{"kind":"REWORK","_number":12,"created":"2021-11-26 08:54:44.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/17/665117/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/17/665117/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/665117/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/665117/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/665117/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/17/665117/12"}}},"commit":{"parents":[{"commit":"7b6f6a054c1a91451ebf655d613c291a818fa8de","subject":"Merge \"Debian: we want Openvswitch and ovn from backports\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7b6f6a054c1a91451ebf655d613c291a818fa8de"}]}],"author":{"name":"Kamil Sambor","email":"ksambor@redhat.com","date":"2019-06-13 11:52:16.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-11-26 08:54:44.000000000","tz":0},"subject":"Drop root privileges for ovn","message":"Drop root privileges for ovn\n\nChange-Id: I5f2a3b9d49e0a1eb7c351d2e7be0989c5f209ecf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/33bca772a74f2973cf83afb13ad8bb3d156a7e9a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/33bca772a74f2973cf83afb13ad8bb3d156a7e9a"}]},"branch":"refs/heads/master"},"889a054385c24a1e2469b45f2a4b23a0797ba920":{"kind":"REWORK","_number":13,"created":"2021-11-26 12:08:42.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/17/665117/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/17/665117/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/665117/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/665117/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/665117/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/17/665117/13"}}},"commit":{"parents":[{"commit":"7b6f6a054c1a91451ebf655d613c291a818fa8de","subject":"Merge \"Debian: we want Openvswitch and ovn from backports\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7b6f6a054c1a91451ebf655d613c291a818fa8de"}]}],"author":{"name":"Kamil Sambor","email":"ksambor@redhat.com","date":"2019-06-13 11:52:16.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-11-26 12:08:40.000000000","tz":0},"subject":"Drop root privileges for ovn","message":"Drop root privileges for ovn\n\nChange-Id: I5f2a3b9d49e0a1eb7c351d2e7be0989c5f209ecf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/889a054385c24a1e2469b45f2a4b23a0797ba920"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/889a054385c24a1e2469b45f2a4b23a0797ba920"}]},"branch":"refs/heads/master"},"80de72fcf88a4f7bfa48f8e42c3ae55e4d201e19":{"kind":"REWORK","_number":14,"created":"2021-12-16 11:39:57.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/17/665117/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/17/665117/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/665117/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/665117/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/665117/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/17/665117/14"}}},"commit":{"parents":[{"commit":"7b6f6a054c1a91451ebf655d613c291a818fa8de","subject":"Merge \"Debian: we want Openvswitch and ovn from backports\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7b6f6a054c1a91451ebf655d613c291a818fa8de"}]}],"author":{"name":"Kamil Sambor","email":"ksambor@redhat.com","date":"2019-06-13 11:52:16.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-12-16 11:39:55.000000000","tz":0},"subject":"Drop root privileges for ovn","message":"Drop root privileges for ovn\n\nChange-Id: I5f2a3b9d49e0a1eb7c351d2e7be0989c5f209ecf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/80de72fcf88a4f7bfa48f8e42c3ae55e4d201e19"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/80de72fcf88a4f7bfa48f8e42c3ae55e4d201e19"}]},"branch":"refs/heads/master"},"56f707228e378c3506ff829322fcdab9c25d7970":{"kind":"REWORK","_number":15,"created":"2021-12-16 14:54:19.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/17/665117/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/17/665117/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/665117/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/665117/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/665117/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/17/665117/15"}}},"commit":{"parents":[{"commit":"7b6f6a054c1a91451ebf655d613c291a818fa8de","subject":"Merge \"Debian: we want Openvswitch and ovn from backports\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7b6f6a054c1a91451ebf655d613c291a818fa8de"}]}],"author":{"name":"Kamil Sambor","email":"ksambor@redhat.com","date":"2019-06-13 11:52:16.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-12-16 14:54:19.000000000","tz":0},"subject":"Drop root privileges for ovn","message":"Drop root privileges for ovn\n\nChange-Id: I5f2a3b9d49e0a1eb7c351d2e7be0989c5f209ecf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/56f707228e378c3506ff829322fcdab9c25d7970"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/56f707228e378c3506ff829322fcdab9c25d7970"}]},"branch":"refs/heads/master"},"5235b873b0f7b99941db95ae1f19b653faf9cb45":{"kind":"REWORK","_number":16,"created":"2024-01-05 13:44:25.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/17/665117/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/17/665117/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/665117/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/665117/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/665117/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/17/665117/16"}}},"commit":{"parents":[{"commit":"9ebd51f7991a76059e91be98ac579ff705954337","subject":"Merge \"Revert \"CentOS/Rocky: use CentOS Cloud SIG repo instead of Delorean (Bobcat)\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9ebd51f7991a76059e91be98ac579ff705954337"}]}],"author":{"name":"Kamil Sambor","email":"ksambor@redhat.com","date":"2019-06-13 11:52:16.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-01-05 13:44:14.000000000","tz":60},"subject":"Drop root privileges for ovn","message":"Drop root privileges for ovn\n\nChange-Id: I5f2a3b9d49e0a1eb7c351d2e7be0989c5f209ecf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5235b873b0f7b99941db95ae1f19b653faf9cb45"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5235b873b0f7b99941db95ae1f19b653faf9cb45"}]},"branch":"refs/heads/master"},"49395d80dde134abf82ca5b6dfeebe283b6f9b13":{"kind":"REWORK","_number":17,"created":"2025-06-23 16:37:28.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/17/665117/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/17/665117/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/665117/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/665117/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/665117/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/17/665117/17"}}},"commit":{"parents":[{"commit":"466fbc53e61fb73a37496b086f817ccaacfc51e8","subject":"Merge \"rabbitmq: Improve healthcheck script output\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/466fbc53e61fb73a37496b086f817ccaacfc51e8"}]}],"author":{"name":"Kamil Sambor","email":"ksambor@redhat.com","date":"2019-06-13 11:52:16.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-06-23 16:37:28.000000000","tz":0},"subject":"Drop root privileges for ovn","message":"Drop root privileges for ovn\n\nChange-Id: I5f2a3b9d49e0a1eb7c351d2e7be0989c5f209ecf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/49395d80dde134abf82ca5b6dfeebe283b6f9b13"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/49395d80dde134abf82ca5b6dfeebe283b6f9b13"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"UNSATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-1"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
