)]}'
{"id":"openstack%2Fkolla~665361","triplet_id":"openstack%2Fkolla~master~I7c979dacd89ebead404b751d596ba120b38f00f3","project":"openstack/kolla","branch":"master","topic":"no_versioned","hashtags":[],"change_id":"I7c979dacd89ebead404b751d596ba120b38f00f3","subject":"doc: Start using openstackdoctheme\u0027s extlink extension","status":"MERGED","created":"2019-06-14 11:11:09.000000000","updated":"2019-06-28 06:46:23.000000000","submitted":"2019-06-28 06:46:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"665361-1561704383961-3d1d4c99","meta_rev_id":"c90aee9888123c59994358ec4352cf9a57578ada","_number":665361,"virtual_id_number":665361,"owner":{"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-06-28 06:46:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"all":[{"value":2,"date":"2019-06-26 06:47:40.000000000","_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":2,"date":"2019-06-28 06:36:15.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":1,"date":"2019-06-28 06:36:15.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"date":"2019-06-25 13:03:03.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-25 13:03:03.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-06-26 06:47:40.000000000","updated_by":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"reviewer":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"state":"REVIEWER"},{"updated":"2019-06-28 06:36:15.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-06-28 06:46:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"88bc3e1f795f339d4ab0179e3b698b90c024b579","author":{"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},"date":"2019-06-14 11:11:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7fcf5eda0dbd274eebc7932f012a594e051876f8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-14 11:29:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/61/665361/1/check/openstack-tox-pep8/64cb08b/ : FAILURE in 3m 13s\n- requirements-check http://logs.openstack.org/61/665361/1/check/requirements-check/9081647/ : SUCCESS in 2m 10s\n- openstack-tox-docs http://logs.openstack.org/61/665361/1/check/openstack-tox-docs/6b2e961/ : POST_FAILURE in 4m 25s\n- build-openstack-releasenotes http://logs.openstack.org/61/665361/1/check/build-openstack-releasenotes/5035f04/html/ : SUCCESS in 4m 07s","accounts_in_message":[],"_revision_number":1},{"id":"6b9566d5b9cf7510285189e25feaff6f8971737e","author":{"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},"date":"2019-06-14 11:43:19.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9f7453134c4f5911719ea14d542551968ec63b22","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-14 11:57:18.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/61/665361/2/check/openstack-tox-pep8/2d8dd3a/ : SUCCESS in 3m 52s\n- requirements-check http://logs.openstack.org/61/665361/2/check/requirements-check/99317a7/ : SUCCESS in 2m 24s\n- openstack-tox-docs http://logs.openstack.org/61/665361/2/check/openstack-tox-docs/074e44d/ : POST_FAILURE in 2m 50s\n- build-openstack-releasenotes http://logs.openstack.org/61/665361/2/check/build-openstack-releasenotes/af05448/html/ : SUCCESS in 4m 00s","accounts_in_message":[],"_revision_number":2},{"id":"2ad56cf9de7008e5dbe604ceeeac5a27c95f8abc","author":{"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},"date":"2019-06-14 13:23:27.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"627028db7aec20a8e4171a05171f66f6500e9eb2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-14 13:46:47.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/61/665361/3/check/openstack-tox-pep8/fba9a4f/ : SUCCESS in 2m 46s\n- requirements-check http://logs.openstack.org/61/665361/3/check/requirements-check/15c37d6/ : SUCCESS in 1m 59s\n- openstack-tox-docs http://logs.openstack.org/61/665361/3/check/openstack-tox-docs/5163837/html/ : SUCCESS in 3m 11s\n- build-openstack-releasenotes http://logs.openstack.org/61/665361/3/check/build-openstack-releasenotes/dc7cf41/html/ : SUCCESS in 4m 44s","accounts_in_message":[],"_revision_number":3},{"id":"daf293a7233f0d52214790812713d750d8af69be","author":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"date":"2019-06-26 06:47:40.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"94923ba03d39a7b2dd25e918171a86d5d87c636f","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-28 06:36:15.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"dededd907717a90cc1fb19ec13036f81cd5f0484","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-28 06:36:28.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"e82aeb4a4c7b08b92aeb02748fb701a8cd21b78c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-28 06:46:23.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"570b2993325807f8ae76520316b57e72474ddc87","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-28 06:46:23.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/61/665361/3/gate/openstack-tox-pep8/c29afa7/ : SUCCESS in 3m 47s\n- requirements-check http://logs.openstack.org/61/665361/3/gate/requirements-check/b74b669/ : SUCCESS in 3m 19s\n- openstack-tox-docs http://logs.openstack.org/61/665361/3/gate/openstack-tox-docs/a8a7414/html/ : SUCCESS in 3m 06s\n- build-openstack-releasenotes http://logs.openstack.org/61/665361/3/gate/build-openstack-releasenotes/8a7d782/html/ : SUCCESS in 4m 42s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"ae1e2e680867559fb97da869086ccfcfc4cbd62d","revisions":{"b5821d85991834fa9163b9d926b160bebeed4809":{"kind":"REWORK","_number":1,"created":"2019-06-14 11:11:09.000000000","uploader":{"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},"ref":"refs/changes/61/665361/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/61/665361/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/61/665361/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/61/665361/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/61/665361/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/61/665361/1"}}},"commit":{"parents":[{"commit":"39a2cf066c4c79436629017155302d36243aa338","subject":"Fix lower-constraints test fail due to failing dep build","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/39a2cf066c4c79436629017155302d36243aa338"}]}],"author":{"name":"chenxing","email":"chason.chan@foxmail.com","date":"2019-06-14 11:10:25.000000000","tz":480},"committer":{"name":"chenxing","email":"chason.chan@foxmail.com","date":"2019-06-14 11:10:25.000000000","tz":480},"subject":"doc: Start using openstackdoctheme\u0027s extlink extension","message":"doc: Start using openstackdoctheme\u0027s extlink extension\n\nThis ensures we have version-specific references to other projects [1].\nNote that this doesn\u0027t mean the URLs are actually valid - we need to do\nmore work (linkcheck?) here, but it\u0027s an improvement nonetheless.\n\n[1] https://docs.openstack.org/openstackdocstheme/latest/#external-link-helper\n\nChange-Id: I7c979dacd89ebead404b751d596ba120b38f00f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b5821d85991834fa9163b9d926b160bebeed4809"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b5821d85991834fa9163b9d926b160bebeed4809"}]},"branch":"refs/heads/master"},"08f8f5b50e90acf43e5b3b204000f1d2e69674ee":{"kind":"REWORK","_number":2,"created":"2019-06-14 11:43:19.000000000","uploader":{"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},"ref":"refs/changes/61/665361/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/61/665361/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/61/665361/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/61/665361/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/61/665361/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/61/665361/2"}}},"commit":{"parents":[{"commit":"39a2cf066c4c79436629017155302d36243aa338","subject":"Fix lower-constraints test fail due to failing dep build","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/39a2cf066c4c79436629017155302d36243aa338"}]}],"author":{"name":"chenxing","email":"chason.chan@foxmail.com","date":"2019-06-14 11:10:25.000000000","tz":480},"committer":{"name":"chenxing","email":"chason.chan@foxmail.com","date":"2019-06-14 11:43:04.000000000","tz":480},"subject":"doc: Start using openstackdoctheme\u0027s extlink extension","message":"doc: Start using openstackdoctheme\u0027s extlink extension\n\nThis ensures we have version-specific references to other projects [1].\nNote that this doesn\u0027t mean the URLs are actually valid - we need to do\nmore work (linkcheck?) here, but it\u0027s an improvement nonetheless.\n\n[1] https://docs.openstack.org/openstackdocstheme/latest/#external-link-helper\n\nChange-Id: I7c979dacd89ebead404b751d596ba120b38f00f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/08f8f5b50e90acf43e5b3b204000f1d2e69674ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/08f8f5b50e90acf43e5b3b204000f1d2e69674ee"}]},"branch":"refs/heads/master"},"ae1e2e680867559fb97da869086ccfcfc4cbd62d":{"kind":"REWORK","_number":3,"created":"2019-06-14 13:23:27.000000000","uploader":{"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},"ref":"refs/changes/61/665361/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/61/665361/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/61/665361/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/61/665361/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/61/665361/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/61/665361/3"}}},"commit":{"parents":[{"commit":"39a2cf066c4c79436629017155302d36243aa338","subject":"Fix lower-constraints test fail due to failing dep build","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/39a2cf066c4c79436629017155302d36243aa338"}]}],"author":{"name":"chenxing","email":"chason.chan@foxmail.com","date":"2019-06-14 11:10:25.000000000","tz":480},"committer":{"name":"chenxing","email":"chason.chan@foxmail.com","date":"2019-06-14 13:23:10.000000000","tz":480},"subject":"doc: Start using openstackdoctheme\u0027s extlink extension","message":"doc: Start using openstackdoctheme\u0027s extlink extension\n\nThis ensures we have version-specific references to other projects [1].\nNote that this doesn\u0027t mean the URLs are actually valid - we need to do\nmore work (linkcheck?) here, but it\u0027s an improvement nonetheless.\n\n[1] https://docs.openstack.org/openstackdocstheme/latest/#external-link-helper\n\nChange-Id: I7c979dacd89ebead404b751d596ba120b38f00f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ae1e2e680867559fb97da869086ccfcfc4cbd62d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ae1e2e680867559fb97da869086ccfcfc4cbd62d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
