)]}'
{"id":"openstack%2Fkolla~665532","triplet_id":"openstack%2Fkolla~stable%2Fstein~Icb8d572b7f084804e0423a8a3171dd66646b3885","project":"openstack/kolla","branch":"stable/stein","topic":"neutron-migrate-fix-stable/stein","hashtags":[],"change_id":"Icb8d572b7f084804e0423a8a3171dd66646b3885","subject":"[neutron] Do not migrate all subprojects","status":"MERGED","created":"2019-06-16 08:19:36.000000000","updated":"2019-06-17 15:49:54.000000000","submitted":"2019-06-17 15:49:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"665532-1560786594795-b95397b7","meta_rev_id":"438afb603b6b5059b99faf229c7dad5f37ae61c1","_number":665532,"virtual_id_number":665532,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},{"value":2,"date":"2019-06-17 15:49:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-06-16 09:04:31.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2019-06-17 12:03:26.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2019-06-17 14:39:05.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2019-06-17 14:39:05.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"date":"2019-06-16 08:19:36.000000000","_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-16 08:19:36.000000000","updated_by":{"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},"reviewer":{"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},"state":"REVIEWER"},{"updated":"2019-06-16 09:04:31.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-06-17 12:03:26.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-06-17 14:39:05.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-06-17 15:49:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"83f4d709774c3c93576e1bbcfb266ec42bcd41ef","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-16 08:19:36.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"5c8913f1bca6c021b620eed16f91a5bf43dc250e","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-16 09:04:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/32/665532/1/openstack-check/tripleo-build-containers-fedora-28-master/d869879/ : SUCCESS in 33m 36s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/32/665532/1/openstack-check/tripleo-build-containers-fedora-28-stein/b817962/ : SUCCESS in 32m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"540033613371b486fd9b6bdf17594d881cf0fdd9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-16 09:59:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/32/665532/1/check/kolla-build-centos-binary/8ab92ac/ : FAILURE in 3m 50s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/32/665532/1/check/kolla-build-centos-source/ab5cfd0/ : SUCCESS in 57m 44s\n- kolla-ansible-centos-source http://logs.openstack.org/32/665532/1/check/kolla-ansible-centos-source/250242a/ : SUCCESS in 1h 03m 10s\n- kolla-ansible-centos-binary http://logs.openstack.org/32/665532/1/check/kolla-ansible-centos-binary/77d85c1/ : SUCCESS in 1h 05m 32s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/32/665532/1/check/kolla-ansible-centos-source-upgrade/23d70ed/ : SUCCESS in 1h 17m 01s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/32/665532/1/check/tripleo-build-containers-centos-7/820367d/ : SUCCESS in 54m 42s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/32/665532/1/check/kolla-build-oraclelinux-binary/d237aac/ : SUCCESS in 47m 08s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/32/665532/1/check/kolla-build-oraclelinux-source/78cdd40/ : SUCCESS in 44m 41s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/32/665532/1/check/kolla-ansible-oraclelinux-source/c1be882/ : SUCCESS in 1h 13m 17s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/32/665532/1/check/kolla-ansible-oraclelinux-binary/97bf1de/ : SUCCESS in 1h 03m 47s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/32/665532/1/check/openstack-tox-lower-constraints/242efeb/ : SUCCESS in 5m 07s\n- openstack-tox-pep8 http://logs.openstack.org/32/665532/1/check/openstack-tox-pep8/8a07450/ : SUCCESS in 3m 37s\n- openstack-tox-py27 http://logs.openstack.org/32/665532/1/check/openstack-tox-py27/a94aaf4/ : SUCCESS in 3m 56s\n- openstack-tox-py36 http://logs.openstack.org/32/665532/1/check/openstack-tox-py36/9b01c0f/ : SUCCESS in 3m 23s\n- openstack-tox-py37 http://logs.openstack.org/32/665532/1/check/openstack-tox-py37/0150d83/ : SUCCESS in 3m 23s\n- openstack-tox-docs http://logs.openstack.org/32/665532/1/check/openstack-tox-docs/4af21cc/html/ : SUCCESS in 2m 51s\n- kolla-build-ubuntu-binary http://logs.openstack.org/32/665532/1/check/kolla-build-ubuntu-binary/895c96a/ : SUCCESS in 29m 30s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/32/665532/1/check/kolla-build-ubuntu-source/51da760/ : SUCCESS in 31m 20s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/32/665532/1/check/kolla-ansible-ubuntu-source/c6ba94f/ : SUCCESS in 51m 39s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/32/665532/1/check/kolla-ansible-ubuntu-binary/49b62ae/ : SUCCESS in 45m 01s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/32/665532/1/check/kolla-ansible-ubuntu-source-upgrade/9af943d/ : SUCCESS in 57m 26s","accounts_in_message":[],"_revision_number":1},{"id":"a02b306e7f8653bbae36040dd0f1c8ab59f58e3d","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-17 12:03:26.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"aac9179432e95c7b4417854d5b1a74cd6d504c58","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-17 14:39:05.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"35f78d3338f4fcf5bf9ea5ccf652d02d5959ad20","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-17 14:39:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"733f27778302184530e219defffbd51d80233036","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-17 15:49:54.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"fd2cd08c368c3374666bf9615dee1cc1f17e3e2b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-17 15:49:54.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source http://logs.openstack.org/32/665532/1/gate/kolla-build-centos-source/645bcff/ : SUCCESS in 48m 46s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/32/665532/1/gate/tripleo-build-containers-centos-7/c89d6c7/ : SUCCESS in 45m 58s\n- kolla-build-oraclelinux-source http://logs.openstack.org/32/665532/1/gate/kolla-build-oraclelinux-source/fda28f7/ : SUCCESS in 1h 02m 50s\n- openstack-tox-lower-constraints http://logs.openstack.org/32/665532/1/gate/openstack-tox-lower-constraints/60eb244/ : SUCCESS in 3m 52s\n- openstack-tox-pep8 http://logs.openstack.org/32/665532/1/gate/openstack-tox-pep8/c81f16f/ : SUCCESS in 3m 30s\n- openstack-tox-py27 http://logs.openstack.org/32/665532/1/gate/openstack-tox-py27/82a4ff2/ : SUCCESS in 3m 23s\n- openstack-tox-py36 http://logs.openstack.org/32/665532/1/gate/openstack-tox-py36/08ca23a/ : SUCCESS in 3m 50s\n- openstack-tox-py37 http://logs.openstack.org/32/665532/1/gate/openstack-tox-py37/b808caf/ : SUCCESS in 3m 51s\n- openstack-tox-docs http://logs.openstack.org/32/665532/1/gate/openstack-tox-docs/b288fe6/html/ : SUCCESS in 3m 35s\n- kolla-build-ubuntu-source http://logs.openstack.org/32/665532/1/gate/kolla-build-ubuntu-source/c34c4fa/ : SUCCESS in 34m 33s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1fe505ffe4d98854452ce0e3c72bc5aeb1ae47d3","revisions":{"1fe505ffe4d98854452ce0e3c72bc5aeb1ae47d3":{"kind":"REWORK","_number":1,"created":"2019-06-16 08:19:36.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/32/665532/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/32/665532/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/32/665532/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/32/665532/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/32/665532/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/32/665532/1"}}},"commit":{"parents":[{"commit":"7a23046b280121cfe7c1b8f77eb97c6f099987a7","subject":"Fix lower-constraints test fail due to failing dep build","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7a23046b280121cfe7c1b8f77eb97c6f099987a7"}]}],"author":{"name":"Jason","email":"jasonanderson@uchicago.edu","date":"2019-04-30 16:44:53.000000000","tz":-300},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-06-16 08:19:36.000000000","tz":0},"subject":"[neutron] Do not migrate all subprojects","message":"[neutron] Do not migrate all subprojects\n\nInstead of migrating all Neutron subprojects at bootstrap time, target\nspecifically the neutron, neutron-fwaas, and neutron-vpnaas subprojects\nspeficially. If a subproject is not specified, all subprojects are\nmigrated, which is not the intended behavior.\n\nChange-Id: Icb8d572b7f084804e0423a8a3171dd66646b3885\n(cherry picked from commit f36d7aacf28bc75a4b0ece73b45aa09591cd262c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1fe505ffe4d98854452ce0e3c72bc5aeb1ae47d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1fe505ffe4d98854452ce0e3c72bc5aeb1ae47d3"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
