)]}'
{"id":"openstack%2Fkolla~665634","triplet_id":"openstack%2Fkolla~master~I0514c02aaee1a413a7979c3a426e0484df839861","project":"openstack/kolla","branch":"master","topic":"bug/1832860","hashtags":[],"change_id":"I0514c02aaee1a413a7979c3a426e0484df839861","subject":"Run nova api_db sync before db sync","status":"MERGED","created":"2019-06-17 10:04:07.000000000","updated":"2019-06-27 14:25:07.000000000","submitted":"2019-06-27 11:41:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"665634-1561635709551-a50dbb77","meta_rev_id":"42cb879a44b19f08907773e19be3ba9a452afc8e","_number":665634,"virtual_id_number":665634,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-06-27 11:41:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-06-17 10:46:02.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"all":[{"value":2,"date":"2019-06-27 10:54:34.000000000","_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":2,"date":"2019-06-18 10:49:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"all":[{"value":1,"date":"2019-06-27 10:54:34.000000000","_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2019-06-27 09:21:13.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-17 10:46:02.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-06-18 10:49:30.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2019-06-27 10:54:34.000000000","updated_by":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"reviewer":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"state":"REVIEWER"},{"updated":"2019-06-27 11:41:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"224d96cffbe88e1761586c6afe3dbe7b38a8360c","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-17 10:04:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c12ace0f7390daa0e9173903732c3a1d0cf061e0","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-17 10:07:23.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"c0b0fe9b6bf91ab915f7b21b16cd3febfa661f2d","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-17 10:46:02.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/34/665634/2/openstack-check/tripleo-build-containers-fedora-28-master/84a20ea/ : SUCCESS in 35m 36s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/34/665634/2/openstack-check/tripleo-build-containers-fedora-28-stein/6c837c4/ : SUCCESS in 34m 57s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f1d9138ae4304a71ce84fb27ffe13bd7bc47abc1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-17 11:59:42.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/34/665634/2/check/kolla-build-centos-binary/187b312/ : SUCCESS in 36m 16s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/34/665634/2/check/kolla-build-centos-source/84e69ec/ : SUCCESS in 40m 19s\n- kolla-ansible-centos-source http://logs.openstack.org/34/665634/2/check/kolla-ansible-centos-source/7738391/ : SUCCESS in 1h 15m 45s\n- kolla-ansible-centos-binary http://logs.openstack.org/34/665634/2/check/kolla-ansible-centos-binary/d940536/ : SUCCESS in 1h 06m 53s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/34/665634/2/check/kolla-ansible-centos-source-upgrade/bd2f348/ : SUCCESS in 1h 26m 47s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/34/665634/2/check/tripleo-build-containers-centos-7/7828cbb/ : SUCCESS in 42m 10s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/34/665634/2/check/kolla-build-oraclelinux-binary/7512604/ : SUCCESS in 46m 05s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/34/665634/2/check/kolla-build-oraclelinux-source/aa568f1/ : SUCCESS in 48m 07s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/34/665634/2/check/kolla-ansible-oraclelinux-source/b6010e3/ : SUCCESS in 1h 13m 45s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/34/665634/2/check/kolla-ansible-oraclelinux-binary/fb4b503/ : SUCCESS in 1h 02m 38s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/34/665634/2/check/openstack-tox-lower-constraints/4d81615/ : SUCCESS in 4m 08s\n- openstack-tox-pep8 http://logs.openstack.org/34/665634/2/check/openstack-tox-pep8/410ad5f/ : SUCCESS in 3m 38s\n- openstack-tox-py27 http://logs.openstack.org/34/665634/2/check/openstack-tox-py27/e065e45/ : SUCCESS in 3m 45s\n- openstack-tox-py36 http://logs.openstack.org/34/665634/2/check/openstack-tox-py36/a18b79d/ : SUCCESS in 3m 26s\n- openstack-tox-py37 http://logs.openstack.org/34/665634/2/check/openstack-tox-py37/7e41a87/ : SUCCESS in 3m 22s\n- openstack-tox-docs http://logs.openstack.org/34/665634/2/check/openstack-tox-docs/637c822/html/ : SUCCESS in 4m 16s\n- kolla-build-ubuntu-binary http://logs.openstack.org/34/665634/2/check/kolla-build-ubuntu-binary/94f6cf3/ : SUCCESS in 23m 12s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/34/665634/2/check/kolla-build-ubuntu-source/e5a30cf/ : SUCCESS in 31m 18s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/34/665634/2/check/kolla-ansible-ubuntu-source/2f36747/ : SUCCESS in 58m 46s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/34/665634/2/check/kolla-ansible-ubuntu-binary/8ebb961/ : SUCCESS in 45m 57s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/34/665634/2/check/kolla-ansible-ubuntu-source-upgrade/f249489/ : SUCCESS in 48m 23s","accounts_in_message":[],"_revision_number":2},{"id":"70d084784f6d68461b3db4d689c654d93b4337e1","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-06-18 10:49:30.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"5ac568152c0674678846473b5fb132c68ef21114","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-27 09:21:13.000000000","message":"Patch Set 2: Review-Priority+1","accounts_in_message":[],"_revision_number":2},{"id":"1f4d0dc5dce2355895cadbcbda7e9c413177bba3","author":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"date":"2019-06-27 10:54:34.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"2261c1403708d22cab9aeab53f63cc5af3edfacf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 10:54:48.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"66b8e3a54d06e6325501f3160da971bba6c11032","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 11:41:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"d207d12c7d2f2057e5a997d8770f78875ade70fd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 11:41:49.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source http://logs.openstack.org/34/665634/2/gate/kolla-build-centos-source/3a53dcd/ : SUCCESS in 45m 29s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/34/665634/2/gate/tripleo-build-containers-centos-7/c0f990d/ : SUCCESS in 40m 44s\n- openstack-tox-lower-constraints http://logs.openstack.org/34/665634/2/gate/openstack-tox-lower-constraints/95127d3/ : SUCCESS in 4m 11s\n- openstack-tox-pep8 http://logs.openstack.org/34/665634/2/gate/openstack-tox-pep8/cfe6dbc/ : SUCCESS in 3m 04s\n- openstack-tox-py27 http://logs.openstack.org/34/665634/2/gate/openstack-tox-py27/0cba6ee/ : SUCCESS in 3m 07s\n- openstack-tox-py36 http://logs.openstack.org/34/665634/2/gate/openstack-tox-py36/b4dd662/ : SUCCESS in 3m 30s\n- openstack-tox-py37 http://logs.openstack.org/34/665634/2/gate/openstack-tox-py37/1898aa2/ : SUCCESS in 3m 09s\n- openstack-tox-docs http://logs.openstack.org/34/665634/2/gate/openstack-tox-docs/d872e82/html/ : SUCCESS in 2m 34s\n- kolla-build-ubuntu-source http://logs.openstack.org/34/665634/2/gate/kolla-build-ubuntu-source/c10d39c/ : SUCCESS in 32m 11s","accounts_in_message":[],"_revision_number":2},{"id":"c0b17797c3eae05e7c6e443fb14f923a425dd4c6","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-27 14:24:57.000000000","message":"Patch Set 2: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit 09226a59866e5ace7d573bac2e0772109a2554a0","accounts_in_message":[],"_revision_number":2},{"id":"0b2192d6a8b0be591ba3ffedd5a46516c7e24755","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-27 14:25:07.000000000","message":"Patch Set 2: Cherry Picked\n\nThis patchset was cherry picked to branch stable/rocky as commit b68632ff7e1e782712d94af6005c6348f6b1c640","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"84a07102e2db92333ce416a520151f1a1856f162","revisions":{"d766291b8ff4a880123a9e61d154e3dd29ccc4f9":{"kind":"REWORK","_number":1,"created":"2019-06-17 10:04:07.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/34/665634/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/34/665634/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/34/665634/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/34/665634/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/34/665634/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/34/665634/1"}}},"commit":{"parents":[{"commit":"498c57fcac31da9268918ec6c188fe2aae0c4906","subject":"Merge \"Use packagecloud/ubuntu rabbitmq repo for all archs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/498c57fcac31da9268918ec6c188fe2aae0c4906"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-06-17 09:49:21.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-06-17 09:49:21.000000000","tz":60},"subject":"Run nova api_db sync before db sync","message":"Run nova api_db sync before db sync\n\nSince cells v2 was introduced, nova operators must run two commands to\nmigrate the database schemas of nova\u0027s databases - nova-manage api_db\nsync and nova-manage db sync. It is necessary to run them in this order,\nsince the db sync may depend on schema changes made to the api\ndatabase in the api_db sync. Executing the db sync first may fail,\nfor example with the following seen in a Queens to Rocky upgrade:\n\n    nova-manage db sync\n    ERROR: Could not access cell0.\n    Has the nova_api database been created?\n    Has the nova_cell0 database been created?\n    Has \"nova-manage api_db sync\" been run?\n    Has \"nova-manage cell_v2 map_cell0\" been run?\n    Is [api_database]/connection set in nova.conf?\n    Is the cell0 database connection URL correct?\n    Error: (pymysql.err.InternalError) (1054, u\"Unknown column\n            \u0027cell_mappings.disabled\u0027 in \u0027field list\u0027\") [SQL: u\u0027SELECT\n    cell_mappings.created_at AS cell_mappings_created_at,\n    cell_mappings.updated_at AS cell_mappings_updated_at,\n    cell_mappings.id AS cell_mappings_id, cell_mappings.uuid AS\n    cell_mappings_uuid, cell_mappings.name AS cell_mappings_name,\n    cell_mappings.transport_url AS cell_mappings_transport_url,\n    cell_mappings.database_connection AS\n    cell_mappings_database_connection, cell_mappings.disabled AS\n    cell_mappings_disabled \\nFROM cell_mappings \\nWHERE\n    cell_mappings.uuid \u003d %(uuid_1)s \\n LIMIT %(param_1)s\u0027] [parameters:\n    {u\u0027uuid_1\u0027: \u002700000000-0000-0000-0000-000000000000\u0027, u\u0027param_1\u0027: 1}]\n    (Background on this error at: http://sqlalche.me/e/2j85)\n\nDespite this error, the command actually exits zero, so kolla does not notice\nthe failure.\n\nThis change reorders these two commands to run the api_db sync before the db\nsync.\n\nChange-Id: I0514c02aaee1a413a7979c3a426e0484df839861\nCloses-Bug: #1832860\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d766291b8ff4a880123a9e61d154e3dd29ccc4f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d766291b8ff4a880123a9e61d154e3dd29ccc4f9"}]},"branch":"refs/heads/master"},"84a07102e2db92333ce416a520151f1a1856f162":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-06-17 10:07:23.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/34/665634/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/34/665634/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/34/665634/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/34/665634/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/34/665634/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/34/665634/2"}}},"commit":{"parents":[{"commit":"498c57fcac31da9268918ec6c188fe2aae0c4906","subject":"Merge \"Use packagecloud/ubuntu rabbitmq repo for all archs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/498c57fcac31da9268918ec6c188fe2aae0c4906"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-06-17 09:49:21.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-06-17 10:07:01.000000000","tz":60},"subject":"Run nova api_db sync before db sync","message":"Run nova api_db sync before db sync\n\nBackport: Stein, Rocky\n\nSince cells v2 was introduced, nova operators must run two commands to\nmigrate the database schemas of nova\u0027s databases - nova-manage api_db\nsync and nova-manage db sync. It is necessary to run them in this order,\nsince the db sync may depend on schema changes made to the api\ndatabase in the api_db sync. Executing the db sync first may fail,\nfor example with the following seen in a Queens to Rocky upgrade:\n\n    nova-manage db sync\n    ERROR: Could not access cell0.\n    Has the nova_api database been created?\n    Has the nova_cell0 database been created?\n    Has \"nova-manage api_db sync\" been run?\n    Has \"nova-manage cell_v2 map_cell0\" been run?\n    Is [api_database]/connection set in nova.conf?\n    Is the cell0 database connection URL correct?\n    Error: (pymysql.err.InternalError) (1054, u\"Unknown column\n            \u0027cell_mappings.disabled\u0027 in \u0027field list\u0027\") [SQL: u\u0027SELECT\n    cell_mappings.created_at AS cell_mappings_created_at,\n    cell_mappings.updated_at AS cell_mappings_updated_at,\n    cell_mappings.id AS cell_mappings_id, cell_mappings.uuid AS\n    cell_mappings_uuid, cell_mappings.name AS cell_mappings_name,\n    cell_mappings.transport_url AS cell_mappings_transport_url,\n    cell_mappings.database_connection AS\n    cell_mappings_database_connection, cell_mappings.disabled AS\n    cell_mappings_disabled \\nFROM cell_mappings \\nWHERE\n    cell_mappings.uuid \u003d %(uuid_1)s \\n LIMIT %(param_1)s\u0027] [parameters:\n    {u\u0027uuid_1\u0027: \u002700000000-0000-0000-0000-000000000000\u0027, u\u0027param_1\u0027: 1}]\n    (Background on this error at: http://sqlalche.me/e/2j85)\n\nDespite this error, the command actually exits zero, so kolla does not notice\nthe failure.\n\nThis change reorders these two commands to run the api_db sync before the db\nsync.\n\nChange-Id: I0514c02aaee1a413a7979c3a426e0484df839861\nCloses-Bug: #1832860\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/84a07102e2db92333ce416a520151f1a1856f162"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/84a07102e2db92333ce416a520151f1a1856f162"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
