)]}'
{"id":"openstack%2Fkolla~665662","triplet_id":"openstack%2Fkolla~master~I33fbc77562d3806695345ffd7504ef3385f7564a","project":"openstack/kolla","branch":"master","hashtags":[],"change_id":"I33fbc77562d3806695345ffd7504ef3385f7564a","subject":"monasca: add \u0027monasca-common\u0027 to base","status":"MERGED","created":"2019-06-17 13:12:26.000000000","updated":"2019-10-02 11:15:37.000000000","submitted":"2019-10-02 11:14:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"665662-1570014875530-2741a3e8","meta_rev_id":"c669c0e333677ea8714f76745a29b0b540040ff8","_number":665662,"virtual_id_number":665662,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-02 11:14:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-10-02 08:06:54.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2019-10-02 09:12:58.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-10-02 07:29:19.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2019-10-02 09:12:58.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-17 13:30:42.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2019-06-27 08:35:27.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2019-10-02 07:29:19.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-10-02 08:06:54.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-10-02 09:12:58.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-10-02 11:14:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5047192e043f80d6685e342398ea4793950a585d","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-17 13:12:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dcf8b6d2f219bda3b276bad33d1d491705558b23","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-17 13:36:38.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3a5b6c69d54853232825769b3d06da5cca815f16","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-06-17 13:40:54.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"38291ee80520132836fccda2939bf6608894557c","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-17 13:51:40.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f739d4ea515cf12d51e2b2d9e01097c0822b9e2b","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-06-17 13:55:34.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"08a54220c14bb2624cee9e2f37f4dfc64175b7cc","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-17 14:04:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/62/665662/1/openstack-check/tripleo-build-containers-fedora-28-master/d8b2625/ : SUCCESS in 35m 00s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/62/665662/1/openstack-check/tripleo-build-containers-fedora-28-stein/ff1abf1/ : SUCCESS in 37m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d026e60e9c47ba1ebf24aa12deb10dff4c9f1117","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-17 14:55:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/62/665662/1/check/kolla-build-centos-binary/3ab5724/ : SUCCESS in 43m 07s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/62/665662/1/check/kolla-build-centos-source/dcc0508/ : SUCCESS in 57m 25s\n- kolla-ansible-centos-source http://logs.openstack.org/62/665662/1/check/kolla-ansible-centos-source/bad60d0/ : SUCCESS in 1h 08m 18s\n- kolla-ansible-centos-binary http://logs.openstack.org/62/665662/1/check/kolla-ansible-centos-binary/0e47427/ : SUCCESS in 59m 34s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/62/665662/1/check/kolla-ansible-centos-source-upgrade/e386228/ : SUCCESS in 1h 15m 21s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/62/665662/1/check/tripleo-build-containers-centos-7/79dcf61/ : SUCCESS in 46m 49s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/62/665662/1/check/kolla-build-oraclelinux-binary/9f68451/ : SUCCESS in 31m 27s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/62/665662/1/check/kolla-build-oraclelinux-source/51ff684/ : SUCCESS in 48m 04s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/62/665662/1/check/kolla-ansible-oraclelinux-source/43422d0/ : FAILURE in 39m 25s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/62/665662/1/check/kolla-ansible-oraclelinux-binary/d54553b/ : SUCCESS in 1h 10m 37s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/62/665662/1/check/openstack-tox-lower-constraints/cd81f91/ : SUCCESS in 3m 41s\n- openstack-tox-pep8 http://logs.openstack.org/62/665662/1/check/openstack-tox-pep8/7ba1809/ : SUCCESS in 3m 24s\n- openstack-tox-py27 http://logs.openstack.org/62/665662/1/check/openstack-tox-py27/69761dc/ : SUCCESS in 3m 40s\n- openstack-tox-py36 http://logs.openstack.org/62/665662/1/check/openstack-tox-py36/acaec1f/ : SUCCESS in 3m 52s\n- openstack-tox-py37 http://logs.openstack.org/62/665662/1/check/openstack-tox-py37/c42c803/ : SUCCESS in 3m 59s\n- openstack-tox-docs http://logs.openstack.org/62/665662/1/check/openstack-tox-docs/72ac395/html/ : SUCCESS in 2m 50s\n- kolla-build-ubuntu-binary http://logs.openstack.org/62/665662/1/check/kolla-build-ubuntu-binary/118c332/ : SUCCESS in 26m 34s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/62/665662/1/check/kolla-build-ubuntu-source/7fb1662/ : SUCCESS in 26m 54s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/62/665662/1/check/kolla-ansible-ubuntu-source/62c5ae0/ : SUCCESS in 50m 27s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/62/665662/1/check/kolla-ansible-ubuntu-binary/b769837/ : SUCCESS in 45m 15s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/62/665662/1/check/kolla-ansible-ubuntu-source-upgrade/8f08d25/ : SUCCESS in 52m 57s","accounts_in_message":[],"_revision_number":1},{"id":"2ba4dbcb99ac9462897118a44800ba4d7a1f5d59","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-18 07:32:53.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5cbbd09386aae67c58b26be5e7257adf90f85bfe","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-18 08:16:03.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0ea25515e7acb3fb168c14714a73eb1bf2dc1513","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-06-18 08:16:25.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a9ed08b08fef3e794af8d8139baa5b001f7f67ef","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-26 21:57:46.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"0cfb27be2367ca0c764a83be8d64cfbd14b08c2c","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-26 21:57:50.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"23539463689eb5fcc102b177ec356d19487c695b","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-26 22:31:37.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/62/665662/2/openstack-check/tripleo-build-containers-fedora-28-master/3950706/ : FAILURE in 5m 01s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/62/665662/2/openstack-check/tripleo-build-containers-fedora-28-stein/825309b/ : SUCCESS in 33m 00s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c505253848fbd0a358afc737b6e721c26106fa03","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 00:34:49.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/62/665662/2/check/kolla-build-centos-binary/07fe7b1/ : SUCCESS in 41m 28s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/62/665662/2/check/kolla-build-centos-source/8b88bd1/ : SUCCESS in 2h 24m 49s\n- kolla-ansible-centos-source http://logs.openstack.org/62/665662/2/check/kolla-ansible-centos-source/9f67888/ : TIMED_OUT in 2h 01m 13s\n- kolla-ansible-centos-binary http://logs.openstack.org/62/665662/2/check/kolla-ansible-centos-binary/650db0d/ : SUCCESS in 1h 03m 55s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/62/665662/2/check/kolla-ansible-centos-source-upgrade/2ef418c/ : TIMED_OUT in 2h 01m 03s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/62/665662/2/check/tripleo-build-containers-centos-7/109e326/ : SUCCESS in 46m 24s\n- openstack-tox-lower-constraints http://logs.openstack.org/62/665662/2/check/openstack-tox-lower-constraints/14f35f6/ : SUCCESS in 3m 44s\n- openstack-tox-pep8 http://logs.openstack.org/62/665662/2/check/openstack-tox-pep8/f49e336/ : SUCCESS in 3m 27s\n- openstack-tox-py27 http://logs.openstack.org/62/665662/2/check/openstack-tox-py27/42c18eb/ : SUCCESS in 4m 04s\n- openstack-tox-py36 http://logs.openstack.org/62/665662/2/check/openstack-tox-py36/c1d5b19/ : SUCCESS in 3m 11s\n- openstack-tox-py37 http://logs.openstack.org/62/665662/2/check/openstack-tox-py37/e9387d1/ : SUCCESS in 3m 26s\n- openstack-tox-docs http://logs.openstack.org/62/665662/2/check/openstack-tox-docs/14ca7a4/html/ : SUCCESS in 3m 33s\n- kolla-build-ubuntu-binary http://logs.openstack.org/62/665662/2/check/kolla-build-ubuntu-binary/18449a3/ : SUCCESS in 29m 23s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/62/665662/2/check/kolla-build-ubuntu-source/4e4da56/ : SUCCESS in 28m 49s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/62/665662/2/check/kolla-ansible-ubuntu-source/f656620/ : SUCCESS in 58m 50s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/62/665662/2/check/kolla-ansible-ubuntu-binary/6f8ad05/ : SUCCESS in 46m 23s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/62/665662/2/check/kolla-ansible-ubuntu-source-upgrade/63f01ce/ : SUCCESS in 1h 07m 08s","accounts_in_message":[],"_revision_number":2},{"id":"874d4c370bba2e40a468a8ea6b7b611dfca0246b","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-27 08:11:48.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"f8494facf101bfd4e6dab9a03f929a22e5c669da","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-27 08:29:39.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"45354b659d2c7ef7aed187818b8563bf36e18b8a","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-06-27 08:35:27.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d05411a11d0908d9c41464962b2e788f4493f15d","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-27 08:48:46.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/62/665662/3/openstack-check/tripleo-build-containers-fedora-28-master/1cfff2a/ : FAILURE in 5m 42s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/62/665662/3/openstack-check/tripleo-build-containers-fedora-28-stein/15c4ba3/ : SUCCESS in 33m 18s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"9df9a1c0ff4ab583f850072e8e059b3bf2e13f6e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 10:34:46.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/62/665662/3/check/kolla-build-centos-binary/b2ae552/ : SUCCESS in 30m 34s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/62/665662/3/check/kolla-build-centos-source/ac20586/ : SUCCESS in 54m 44s\n- kolla-ansible-centos-source http://logs.openstack.org/62/665662/3/check/kolla-ansible-centos-source/4806fae/ : SUCCESS in 1h 03m 58s\n- kolla-ansible-centos-binary http://logs.openstack.org/62/665662/3/check/kolla-ansible-centos-binary/9f8bdcd/ : SUCCESS in 1h 03m 25s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/62/665662/3/check/kolla-ansible-centos-source-upgrade/17fce78/ : SUCCESS in 1h 20m 08s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/62/665662/3/check/tripleo-build-containers-centos-7/d0bf160/ : SUCCESS in 47m 09s\n- openstack-tox-lower-constraints http://logs.openstack.org/62/665662/3/check/openstack-tox-lower-constraints/6c7d3ed/ : SUCCESS in 4m 45s\n- openstack-tox-pep8 http://logs.openstack.org/62/665662/3/check/openstack-tox-pep8/a09afff/ : SUCCESS in 3m 15s\n- openstack-tox-py27 http://logs.openstack.org/62/665662/3/check/openstack-tox-py27/d41e465/ : SUCCESS in 3m 24s\n- openstack-tox-py36 http://logs.openstack.org/62/665662/3/check/openstack-tox-py36/34b937f/ : SUCCESS in 3m 08s\n- openstack-tox-py37 http://logs.openstack.org/62/665662/3/check/openstack-tox-py37/6d895b6/ : SUCCESS in 3m 35s\n- openstack-tox-docs http://logs.openstack.org/62/665662/3/check/openstack-tox-docs/293a560/html/ : SUCCESS in 2m 26s\n- kolla-build-ubuntu-binary http://logs.openstack.org/62/665662/3/check/kolla-build-ubuntu-binary/32505a2/ : SUCCESS in 23m 04s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/62/665662/3/check/kolla-build-ubuntu-source/70de031/ : SUCCESS in 28m 10s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/62/665662/3/check/kolla-ansible-ubuntu-source/9825912/ : SUCCESS in 45m 06s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/62/665662/3/check/kolla-ansible-ubuntu-binary/30a8ab5/ : SUCCESS in 53m 32s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/62/665662/3/check/kolla-ansible-ubuntu-source-upgrade/eb0ad51/ : SUCCESS in 1h 03m 06s","accounts_in_message":[],"_revision_number":3},{"id":"09a6a6c8e354ea9096f779d404e819d2e7fa7c43","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-27 11:02:25.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"251eccad50c5c3b769fff491d6a93f427a0ceeb0","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-10-02 07:27:48.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"440b03598f595f748a3f413b1464af2a72c0329f","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-10-02 07:28:11.000000000","message":"Patch Set 4:\n\nadded comma as it needed a retest after 3 months anyway :-)","accounts_in_message":[],"_revision_number":4},{"id":"368fcb941755e38d67e5a4d55c410ca6ba258040","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-10-02 07:29:19.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"ee4a4d0ea1709fdc5b703174dacb2840f8797b19","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-10-02 08:06:54.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/62/665662/4/openstack-check/tripleo-build-containers-rhel-8-master/209fa6e/ : SUCCESS in 38m 17s","accounts_in_message":[],"_revision_number":4},{"id":"08c8b55e3f8dfa755c3c81dec0f48d5cba9d23da","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-02 09:12:58.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\nI guess it can\u0027t hurt","accounts_in_message":[],"_revision_number":4},{"id":"1a0a3605243fe90144db90bc4d6cf6f178d8dd37","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-02 09:26:38.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/56a4fb6b83684634a3eb414eb0f6f714 : SUCCESS in 53m 37s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/83d8cb4af4624a028825d2b611c12451 : SUCCESS in 52m 40s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/ea321e606b2140949553c4b9b99ce1c7 : SUCCESS in 1h 50m 27s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/0b174a22533e4719a947ceb4c62bc81f : SUCCESS in 1h 38m 01s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/8fe08cead85e41659e5ce4398568b842 : SUCCESS in 1h 57m 12s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/25380bb8f30047e2850dba28c92163f5 : SUCCESS in 1h 07m 23s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/4327a6f5e31e4c2489c015271d86a0bf : SUCCESS in 37m 34s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/8264394b36a84928af76c96f818ac284 : SUCCESS in 38m 13s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/886aa7e124194de0b105c408a296300e : SUCCESS in 3m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa01d85681b5447ea477249de02c895c : SUCCESS in 4m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/69b3b88312d94e23b3845ae28b8b3f31 : SUCCESS in 3m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/da3f90547414497e81dbb60a75a33e66 : SUCCESS in 4m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d14d59c88e454feb8ce2334fc7880ffc : SUCCESS in 5m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50779945a9794c36ac1068e4c753f44a : SUCCESS in 7m 09s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a2dd579ad66e4403ba414c08b1d49bf7 : SUCCESS in 43m 37s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/cc4c96ad493141219836b698d4a4b69b : SUCCESS in 38m 33s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/13ef38e656324ea2845602506bed5d94 : SUCCESS in 1h 08m 34s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/91ea96e85dff4c91a764a0f66ef610ce : SUCCESS in 54m 41s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/60950778dad042c5a5e84744cd052443 : SUCCESS in 1h 16m 49s","accounts_in_message":[],"_revision_number":4},{"id":"44d525b1257bc1962432bb529a239366e314d46d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-02 09:26:52.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"d023473dfb114bc314693704c3921fab712505db","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-02 11:14:35.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"f24f7f60cf6b8d1c2b10a3a4048b00b6db2b8653","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-02 11:14:35.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/6cba95b1cb5f41b4b20fb54f31d3c0eb : SUCCESS in 58m 31s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/cc2585dfe8cd47518ce3681c278124be : SUCCESS in 1h 24m 49s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/4bb4e1c913234b06aa2c182adc088738 : SUCCESS in 1h 29m 25s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/27c0e2e9734e4a3abbef6c66154dc4ff : SUCCESS in 56m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8212fcb9b4984d6d97b3a1aca5ccb0ad : SUCCESS in 3m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac5ff69cce4245c3970e5c98b0a5ac00 : SUCCESS in 3m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0a6ad69d3e3740058790e745d5fd0e58 : SUCCESS in 3m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4c1ff358375f4d4faf34ae091564edde : SUCCESS in 3m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3b8f0ded2024465eb66043804694f28f : SUCCESS in 4m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9831c77861174db2adfb634b5f68a745 : SUCCESS in 4m 17s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/022f9f7a82f14d1b9fa01122cb767483 : SUCCESS in 24m 41s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/cd130a32e65f4760ae3e212d2a70ff4d : SUCCESS in 59m 36s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/f20a319697d44c0d95a59e84ac7dee05 : SUCCESS in 1h 15m 23s","accounts_in_message":[],"_revision_number":4},{"id":"c669c0e333677ea8714f76745a29b0b540040ff8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-02 11:15:37.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32836eb8abe548ac8e249c7204bf27f0 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"004f6f4ad249eb9f042b5cde475801bf22c51905","revisions":{"c6405eaac2f5188719d4ee346e2737cf4d3c0593":{"kind":"REWORK","_number":1,"created":"2019-06-17 13:12:26.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/62/665662/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/62/665662/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/62/665662/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/62/665662/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/62/665662/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/62/665662/1"}}},"commit":{"parents":[{"commit":"45ef7176aead8d8b59135f8c168094600ea05f31","subject":"Merge \"[neutron] Do not migrate all subprojects\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/45ef7176aead8d8b59135f8c168094600ea05f31"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-06-17 13:08:31.000000000","tz":0},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-06-17 13:12:19.000000000","tz":0},"subject":"monasca: add \u0027confluent-kafka\u0027 to base","message":"monasca: add \u0027confluent-kafka\u0027 to base\n\n\u0027confluent-kafka\u0027 is used by several monasca images:\n\n- agent\n- api\n- log-api\n- notification\n- persister\n\nSo build it once.\n\nChange-Id: I33fbc77562d3806695345ffd7504ef3385f7564a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c6405eaac2f5188719d4ee346e2737cf4d3c0593"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c6405eaac2f5188719d4ee346e2737cf4d3c0593"}]},"branch":"refs/heads/master"},"5d526d05b1fda1d2ba00295211025f395cfaa02a":{"kind":"REWORK","_number":2,"created":"2019-06-26 21:57:46.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/62/665662/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/62/665662/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/62/665662/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/62/665662/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/62/665662/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/62/665662/2"}}},"commit":{"parents":[{"commit":"45ef7176aead8d8b59135f8c168094600ea05f31","subject":"Merge \"[neutron] Do not migrate all subprojects\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/45ef7176aead8d8b59135f8c168094600ea05f31"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-06-17 13:08:31.000000000","tz":0},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-06-26 21:56:28.000000000","tz":0},"subject":"monasca: add \u0027confluent-kafka\u0027 to base","message":"monasca: add \u0027confluent-kafka\u0027 to base\n\n\u0027confluent-kafka\u0027 is used by several monasca images:\n\n- agent\n- api\n- log-api\n- notification\n- persister\n\nSo build it once.\n\nChange-Id: I33fbc77562d3806695345ffd7504ef3385f7564a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5d526d05b1fda1d2ba00295211025f395cfaa02a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5d526d05b1fda1d2ba00295211025f395cfaa02a"}]},"branch":"refs/heads/master"},"ea1f352f4234beaf429a623e6cf6affb778f5578":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2019-06-27 08:11:48.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/62/665662/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/62/665662/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/62/665662/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/62/665662/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/62/665662/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/62/665662/3"}}},"commit":{"parents":[{"commit":"45ef7176aead8d8b59135f8c168094600ea05f31","subject":"Merge \"[neutron] Do not migrate all subprojects\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/45ef7176aead8d8b59135f8c168094600ea05f31"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-06-17 13:08:31.000000000","tz":0},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-06-27 08:11:36.000000000","tz":0},"subject":"monasca: add \u0027monasca-common\u0027 to base","message":"monasca: add \u0027monasca-common\u0027 to base\n\n\u0027monasca-common\u0027 is used by several monasca images:\n\n- agent\n- api\n- log-api\n- notification\n- persister\n\nSo build it once.\n\nChange-Id: I33fbc77562d3806695345ffd7504ef3385f7564a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ea1f352f4234beaf429a623e6cf6affb778f5578"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ea1f352f4234beaf429a623e6cf6affb778f5578"}]},"branch":"refs/heads/master"},"004f6f4ad249eb9f042b5cde475801bf22c51905":{"kind":"REWORK","_number":4,"created":"2019-10-02 07:27:48.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/62/665662/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/62/665662/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/62/665662/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/62/665662/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/62/665662/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/62/665662/4"}}},"commit":{"parents":[{"commit":"45ef7176aead8d8b59135f8c168094600ea05f31","subject":"Merge \"[neutron] Do not migrate all subprojects\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/45ef7176aead8d8b59135f8c168094600ea05f31"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-06-17 13:08:31.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-10-02 07:27:45.000000000","tz":0},"subject":"monasca: add \u0027monasca-common\u0027 to base","message":"monasca: add \u0027monasca-common\u0027 to base\n\n\u0027monasca-common\u0027 is used by several monasca images:\n\n- agent\n- api\n- log-api\n- notification\n- persister\n\nSo build it once.\n\nChange-Id: I33fbc77562d3806695345ffd7504ef3385f7564a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/004f6f4ad249eb9f042b5cde475801bf22c51905"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/004f6f4ad249eb9f042b5cde475801bf22c51905"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
