)]}'
{"id":"openstack%2Fkolla~665724","triplet_id":"openstack%2Fkolla~master~I34b9d56559eada96abec13814a2aa91b57055710","project":"openstack/kolla","branch":"master","hashtags":[],"change_id":"I34b9d56559eada96abec13814a2aa91b57055710","subject":"base: do not use non-free repositories for Ubuntu","status":"MERGED","created":"2019-06-17 16:36:36.000000000","updated":"2019-06-18 17:52:19.000000000","submitted":"2019-06-18 09:50:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"665724-1560851441428-1e9fde82","meta_rev_id":"d73b1ec0eccda1c7e69524e6cde1af1c658efe77","_number":665724,"virtual_id_number":665724,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":2,"date":"2019-06-18 09:50:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-06-18 08:42:36.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2019-06-18 08:39:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2019-06-17 18:17:24.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2019-06-17 16:37:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2019-06-18 08:39:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2019-06-17 16:38:19.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-17 16:37:21.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-06-17 18:17:24.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-06-18 08:39:27.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2019-06-18 08:42:36.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-06-18 09:50:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bee286827653fa9b85e8f8d37ed34268fccdfe1b","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-17 16:36:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cd867f6cf34ec680517db30d0f5a5d4e6007a469","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-17 16:37:21.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"e57f2b0a461661eb3ed51506df243b6303a266d7","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-17 16:38:19.000000000","message":"Patch Set 1: Review-Priority+2\n\nWe moved to use infra mirrors for ubuntu CI jobs. And non-free is not there \u003d\u003d jobs start to fail.","accounts_in_message":[],"_revision_number":1},{"id":"e31b08a80e12e5abd49b0b7b5b0bd92eb74a44bf","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-17 16:40:08.000000000","message":"Patch Set 1:\n\nI have not seen a job that failed due to that. Only warnings, hmm.","accounts_in_message":[],"_revision_number":1},{"id":"9084d2793f845c7cccbe8a77346f4f70dc3759c7","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-17 17:17:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/24/665724/1/openstack-check/tripleo-build-containers-fedora-28-master/3b45747/ : SUCCESS in 40m 16s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/24/665724/1/openstack-check/tripleo-build-containers-fedora-28-stein/6e7fe08/ : SUCCESS in 37m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ed202dacddf45141e6301d4cac3d8260f8a35fa5","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-17 17:55:40.000000000","message":"Patch Set 1:\n\n\u003e I have not seen a job that failed due to that. Only warnings, hmm.\n\nhttps://review.opendev.org/#/c/665621/ fails due to that:\n\nINFO:kolla.common.utils.base:E: Failed to fetch http://mirror.iad.rax.opendev.org/ubuntu/dists/bionic-updates/multiverse/binary-amd64/Packages  404  Not Found [IP: 104.130.4.160 80]","accounts_in_message":[],"_revision_number":1},{"id":"2d5f53c6b4339d4ff13b8c2480004ff2717be00f","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-17 18:00:04.000000000","message":"Patch Set 1:\n\nOdd, after so many passes of our changes... Something must have changed in the source image again. I issued a recheck to satisfy my curiosity. If it fails, then let\u0027s put an alarm: every Monday Ubuntu base image goes awry, expect problems.","accounts_in_message":[],"_revision_number":1},{"id":"a84789591c9a36ebd8a590761dcff77432eeba17","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-17 18:10:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/24/665724/1/check/kolla-build-centos-binary/e72b012/ : SUCCESS in 39m 13s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/24/665724/1/check/kolla-build-centos-source/aa7bf35/ : SUCCESS in 47m 31s\n- kolla-ansible-centos-source http://logs.openstack.org/24/665724/1/check/kolla-ansible-centos-source/5a73db2/ : SUCCESS in 1h 11m 58s\n- kolla-ansible-centos-binary http://logs.openstack.org/24/665724/1/check/kolla-ansible-centos-binary/c8e8603/ : SUCCESS in 1h 03m 58s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/24/665724/1/check/kolla-ansible-centos-source-upgrade/00b1f9c/ : SUCCESS in 1h 26m 32s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/24/665724/1/check/tripleo-build-containers-centos-7/4573187/ : SUCCESS in 44m 25s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/24/665724/1/check/kolla-build-oraclelinux-binary/86d74cc/ : SUCCESS in 45m 40s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/24/665724/1/check/kolla-build-oraclelinux-source/7d09b63/ : SUCCESS in 43m 31s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/24/665724/1/check/kolla-ansible-oraclelinux-source/12141f5/ : SUCCESS in 1h 10m 17s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/24/665724/1/check/kolla-ansible-oraclelinux-binary/0d5bf81/ : SUCCESS in 1h 11m 32s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/24/665724/1/check/openstack-tox-lower-constraints/ab37f88/ : SUCCESS in 4m 06s\n- openstack-tox-pep8 http://logs.openstack.org/24/665724/1/check/openstack-tox-pep8/7cff886/ : SUCCESS in 4m 06s\n- openstack-tox-py27 http://logs.openstack.org/24/665724/1/check/openstack-tox-py27/2c136db/ : SUCCESS in 4m 13s\n- openstack-tox-py36 http://logs.openstack.org/24/665724/1/check/openstack-tox-py36/dd4303a/ : SUCCESS in 3m 26s\n- openstack-tox-py37 http://logs.openstack.org/24/665724/1/check/openstack-tox-py37/4549028/ : SUCCESS in 4m 18s\n- openstack-tox-docs http://logs.openstack.org/24/665724/1/check/openstack-tox-docs/4355a13/html/ : SUCCESS in 3m 08s\n- kolla-build-ubuntu-binary http://logs.openstack.org/24/665724/1/check/kolla-build-ubuntu-binary/d075a47/ : SUCCESS in 26m 40s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/24/665724/1/check/kolla-build-ubuntu-source/92479b0/ : SUCCESS in 30m 42s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/24/665724/1/check/kolla-ansible-ubuntu-source/406ecbc/ : SUCCESS in 57m 26s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/24/665724/1/check/kolla-ansible-ubuntu-binary/68635a6/ : SUCCESS in 51m 38s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/24/665724/1/check/kolla-ansible-ubuntu-source-upgrade/b995090/ : SUCCESS in 1h 02m 58s","accounts_in_message":[],"_revision_number":1},{"id":"be8e8ea10fe685b957c5d8b31440ae91c6a7c3df","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-17 18:17:24.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0c40eb283142493ef88dc2396dc30177847b877f","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-18 06:17:09.000000000","message":"Patch Set 1:\n\n\u003e \u003e I have not seen a job that failed due to that. Only warnings,\n \u003e hmm.\n \u003e \n \u003e https://review.opendev.org/#/c/665621/ fails due to that:\n \u003e \n \u003e INFO:kolla.common.utils.base:E: Failed to fetch http://mirror.iad.rax.opendev.org/ubuntu/dists/bionic-updates/multiverse/binary-amd64/Packages\n \u003e  404  Not Found [IP: 104.130.4.160 80]\n\nSeems like a bug in apt which stopped being precise (it\u0027s bionic now!, yeah, I know, bad pun :P ). The difference between passing and failing was in this line:\nThe repository \u0027http://mirror.iad.rax.opendev.org/ubuntu bionic-updates Release\u0027 does not have a Release file.\nSo it would fail regardless of this fix (which I like anyway).\nInfra team said they have 3 different AFS versions - mirrors are going to be fun.","accounts_in_message":[],"_revision_number":1},{"id":"1e33edc32e7518bd882edf089e949b856973b11f","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-18 06:17:33.000000000","message":"Patch Set 1:\n\nrecheck\n\nto see mirrors today","accounts_in_message":[],"_revision_number":1},{"id":"da4cde4522ced7b11c6ee1dd6f58d688e2e107bd","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-18 07:06:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/24/665724/1/openstack-check/tripleo-build-containers-fedora-28-master/e8e6fe5/ : SUCCESS in 35m 25s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/24/665724/1/openstack-check/tripleo-build-containers-fedora-28-stein/a0c0e57/ : SUCCESS in 37m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"db4696d7b12b90b1ff7ec668eef52907d0518eca","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 07:56:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/24/665724/1/check/kolla-build-centos-binary/ad48d03/ : SUCCESS in 25m 48s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/24/665724/1/check/kolla-build-centos-source/02487ac/ : SUCCESS in 40m 58s\n- kolla-ansible-centos-source http://logs.openstack.org/24/665724/1/check/kolla-ansible-centos-source/dfac45f/ : SUCCESS in 1h 12m 58s\n- kolla-ansible-centos-binary http://logs.openstack.org/24/665724/1/check/kolla-ansible-centos-binary/9e66bc9/ : SUCCESS in 1h 28m 36s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/24/665724/1/check/kolla-ansible-centos-source-upgrade/bf6991e/ : SUCCESS in 1h 18m 31s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/24/665724/1/check/tripleo-build-containers-centos-7/c5328c5/ : SUCCESS in 46m 28s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/24/665724/1/check/kolla-build-oraclelinux-binary/ab69b74/ : SUCCESS in 48m 56s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/24/665724/1/check/kolla-build-oraclelinux-source/b5c5527/ : SUCCESS in 57m 49s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/24/665724/1/check/kolla-ansible-oraclelinux-source/92c1ba0/ : SUCCESS in 1h 26m 52s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/24/665724/1/check/kolla-ansible-oraclelinux-binary/9ce72ca/ : SUCCESS in 1h 16m 53s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/24/665724/1/check/openstack-tox-lower-constraints/b4b412e/ : SUCCESS in 4m 11s\n- openstack-tox-pep8 http://logs.openstack.org/24/665724/1/check/openstack-tox-pep8/9f76d67/ : SUCCESS in 3m 51s\n- openstack-tox-py27 http://logs.openstack.org/24/665724/1/check/openstack-tox-py27/410d604/ : SUCCESS in 3m 31s\n- openstack-tox-py36 http://logs.openstack.org/24/665724/1/check/openstack-tox-py36/3bf36a3/ : SUCCESS in 3m 29s\n- openstack-tox-py37 http://logs.openstack.org/24/665724/1/check/openstack-tox-py37/fbe6ce3/ : SUCCESS in 3m 45s\n- openstack-tox-docs http://logs.openstack.org/24/665724/1/check/openstack-tox-docs/5c60c5b/html/ : SUCCESS in 2m 49s\n- kolla-build-ubuntu-binary http://logs.openstack.org/24/665724/1/check/kolla-build-ubuntu-binary/454fedc/ : SUCCESS in 37m 58s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/24/665724/1/check/kolla-build-ubuntu-source/f5c84a7/ : SUCCESS in 29m 47s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/24/665724/1/check/kolla-ansible-ubuntu-source/d88010b/ : SUCCESS in 50m 38s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/24/665724/1/check/kolla-ansible-ubuntu-binary/95eff17/ : SUCCESS in 43m 03s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/24/665724/1/check/kolla-ansible-ubuntu-source-upgrade/247e363/ : SUCCESS in 56m 51s","accounts_in_message":[],"_revision_number":1},{"id":"4c16163d56293719dc6083124ee272a9334fa3e7","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-18 08:03:06.000000000","message":"Patch Set 1:\n\nrecheck\n\nalways different mirrors :-)","accounts_in_message":[],"_revision_number":1},{"id":"1909fe429bee87c46c0fd8a155c288a01ebc96be","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-06-18 08:39:27.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"caea7aa2faf33baecf6b56b1d09c3134f4fcf0fc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 08:39:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8441560822c511aaf2a55aab4cc372cfa31bc8a8","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-18 08:42:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/24/665724/1/openstack-check/tripleo-build-containers-fedora-28-master/a96920b/ : SUCCESS in 38m 25s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/24/665724/1/openstack-check/tripleo-build-containers-fedora-28-stein/9736335/ : SUCCESS in 35m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"850a07a13072d743f748018e0c9b0e5c0cca3fc9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 09:48:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/24/665724/1/check/kolla-build-centos-binary/bd43827/ : SUCCESS in 40m 30s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/24/665724/1/check/kolla-build-centos-source/e91e913/ : SUCCESS in 39m 18s\n- kolla-ansible-centos-source http://logs.openstack.org/24/665724/1/check/kolla-ansible-centos-source/ea3e8d3/ : SUCCESS in 1h 20m 52s\n- kolla-ansible-centos-binary http://logs.openstack.org/24/665724/1/check/kolla-ansible-centos-binary/4146d0b/ : SUCCESS in 1h 04m 48s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/24/665724/1/check/kolla-ansible-centos-source-upgrade/864d113/ : SUCCESS in 1h 19m 36s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/24/665724/1/check/tripleo-build-containers-centos-7/cf7b2d9/ : SUCCESS in 47m 31s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/24/665724/1/check/kolla-build-oraclelinux-binary/fb32ddc/ : SUCCESS in 44m 51s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/24/665724/1/check/kolla-build-oraclelinux-source/007c115/ : SUCCESS in 59m 02s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/24/665724/1/check/kolla-ansible-oraclelinux-source/653d619/ : SUCCESS in 1h 22m 30s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/24/665724/1/check/kolla-ansible-oraclelinux-binary/81d9863/ : SUCCESS in 1h 02m 14s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/24/665724/1/check/openstack-tox-lower-constraints/d8ff50f/ : SUCCESS in 3m 54s\n- openstack-tox-pep8 http://logs.openstack.org/24/665724/1/check/openstack-tox-pep8/e77de01/ : SUCCESS in 3m 41s\n- openstack-tox-py27 http://logs.openstack.org/24/665724/1/check/openstack-tox-py27/6b92924/ : SUCCESS in 3m 55s\n- openstack-tox-py36 http://logs.openstack.org/24/665724/1/check/openstack-tox-py36/9275a71/ : SUCCESS in 3m 05s\n- openstack-tox-py37 http://logs.openstack.org/24/665724/1/check/openstack-tox-py37/e8766ba/ : SUCCESS in 4m 01s\n- openstack-tox-docs http://logs.openstack.org/24/665724/1/check/openstack-tox-docs/76843b1/html/ : SUCCESS in 2m 40s\n- kolla-build-ubuntu-binary http://logs.openstack.org/24/665724/1/check/kolla-build-ubuntu-binary/e36eac2/ : SUCCESS in 24m 53s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/24/665724/1/check/kolla-build-ubuntu-source/0aae0d3/ : SUCCESS in 28m 50s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/24/665724/1/check/kolla-ansible-ubuntu-source/15c1567/ : SUCCESS in 49m 31s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/24/665724/1/check/kolla-ansible-ubuntu-binary/09cfac8/ : SUCCESS in 42m 48s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/24/665724/1/check/kolla-ansible-ubuntu-source-upgrade/7286471/ : SUCCESS in 52m 35s","accounts_in_message":[],"_revision_number":1},{"id":"1a253aabcbd5a841258166e8350b706e3b1cf27b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 09:50:41.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source http://logs.openstack.org/24/665724/1/gate/kolla-build-centos-source/94b8b89/ : SUCCESS in 56m 59s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/24/665724/1/gate/tripleo-build-containers-centos-7/98f4199/ : SUCCESS in 41m 25s\n- kolla-build-oraclelinux-source http://logs.openstack.org/24/665724/1/gate/kolla-build-oraclelinux-source/5c2228e/ : SUCCESS in 47m 57s\n- openstack-tox-lower-constraints http://logs.openstack.org/24/665724/1/gate/openstack-tox-lower-constraints/60580ae/ : SUCCESS in 3m 37s\n- openstack-tox-pep8 http://logs.openstack.org/24/665724/1/gate/openstack-tox-pep8/746c1b2/ : SUCCESS in 4m 32s\n- openstack-tox-py27 http://logs.openstack.org/24/665724/1/gate/openstack-tox-py27/a46e895/ : SUCCESS in 3m 41s\n- openstack-tox-py36 http://logs.openstack.org/24/665724/1/gate/openstack-tox-py36/e3d093e/ : SUCCESS in 3m 26s\n- openstack-tox-py37 http://logs.openstack.org/24/665724/1/gate/openstack-tox-py37/7aa9130/ : SUCCESS in 3m 54s\n- openstack-tox-docs http://logs.openstack.org/24/665724/1/gate/openstack-tox-docs/6a44640/html/ : SUCCESS in 2m 54s\n- kolla-build-ubuntu-source http://logs.openstack.org/24/665724/1/gate/kolla-build-ubuntu-source/ccb2f13/ : SUCCESS in 30m 23s","accounts_in_message":[],"_revision_number":1},{"id":"e15db717a4eef14a7940a37deb9363973db29781","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 09:50:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"d73b1ec0eccda1c7e69524e6cde1af1c658efe77","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-18 17:52:19.000000000","message":"Patch Set 1:\n\nShould this be backported?","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"96460c4bfe258c2a39f271a9b5591e76b8b3e783","revisions":{"96460c4bfe258c2a39f271a9b5591e76b8b3e783":{"kind":"REWORK","_number":1,"created":"2019-06-17 16:36:36.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/24/665724/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/24/665724/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/24/665724/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/24/665724/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/24/665724/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/24/665724/1"}}},"commit":{"parents":[{"commit":"b07241ff5537e6f3807943c1bb9bc26a9badc7aa","subject":"Merge \"grafana/aarch64: use package from upstream repositories\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b07241ff5537e6f3807943c1bb9bc26a9badc7aa"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin@juszkiewicz.com.pl","date":"2019-06-17 16:19:57.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin@juszkiewicz.com.pl","date":"2019-06-17 16:35:33.000000000","tz":120},"subject":"base: do not use non-free repositories for Ubuntu","message":"base: do not use non-free repositories for Ubuntu\n\nRemove restricted and multiverse from ubuntu.\n\nrestricted is supported software that is non-free.\nmultiverse is unsupported software that is non-free.\n\nWe should not use any of them. And they are not mirrored by OpenStack\ninfra.\n\nChange-Id: I34b9d56559eada96abec13814a2aa91b57055710\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/96460c4bfe258c2a39f271a9b5591e76b8b3e783"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/96460c4bfe258c2a39f271a9b5591e76b8b3e783"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
