)]}'
{"id":"openstack%2Fkolla~665846","triplet_id":"openstack%2Fkolla~master~I14d9135937109a803c3d53ccf369725d61b73dca","project":"openstack/kolla","branch":"master","topic":"add_kapacitor","hashtags":[],"change_id":"I14d9135937109a803c3d53ccf369725d61b73dca","subject":"add kapacitor","status":"ABANDONED","created":"2019-06-18 02:29:05.000000000","updated":"2021-03-25 18:48:01.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"226c275c117166774753fb4061a6aeb251335614","_number":665846,"virtual_id_number":665846,"owner":{"_account_id":24061,"name":"pangliye","email":"pangliye@inspur.com","username":"pangliye"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2019-06-19 09:21:31.000000000","_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":-1,"date":"2019-06-18 11:27:22.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2019-07-01 11:48:44.000000000","_account_id":24061,"name":"pangliye","email":"pangliye@inspur.com","username":"pangliye"},{"date":"2019-06-18 10:18:01.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":24061,"name":"pangliye","email":"pangliye@inspur.com","username":"pangliye"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":24061,"name":"pangliye","email":"pangliye@inspur.com","username":"pangliye"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24061,"name":"pangliye","email":"pangliye@inspur.com","username":"pangliye"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24061,"name":"pangliye","email":"pangliye@inspur.com","username":"pangliye"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24061,"name":"pangliye","email":"pangliye@inspur.com","username":"pangliye"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-18 06:21:38.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-06-18 10:18:01.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-06-18 11:27:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-06-19 09:21:31.000000000","updated_by":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"reviewer":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"state":"REVIEWER"}],"messages":[{"id":"e9a3f939d350fb5cda72bf0a683f55a7e71ccbb6","author":{"_account_id":24061,"name":"pangliye","email":"pangliye@inspur.com","username":"pangliye"},"date":"2019-06-18 02:29:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6ed0623e0649a715ab0456bfd4ba9ff9c92605e4","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-18 03:14:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/46/665846/1/openstack-check/tripleo-build-containers-fedora-28-master/81d012a/ : SUCCESS in 34m 04s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/46/665846/1/openstack-check/tripleo-build-containers-fedora-28-stein/e84295d/ : SUCCESS in 31m 44s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2328372de33f2a5e742526709b1b06035550f6a0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 04:00:17.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/46/665846/1/check/test-release-openstack/b36f782/ : SUCCESS in 3m 10s\n- kolla-build-centos-binary http://logs.openstack.org/46/665846/1/check/kolla-build-centos-binary/5173678/ : SUCCESS in 28m 44s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/46/665846/1/check/kolla-build-centos-source/f11d6a8/ : SUCCESS in 1h 08m 04s\n- kolla-ansible-centos-source http://logs.openstack.org/46/665846/1/check/kolla-ansible-centos-source/59e0801/ : SUCCESS in 1h 03m 04s\n- kolla-ansible-centos-binary http://logs.openstack.org/46/665846/1/check/kolla-ansible-centos-binary/f26ec2f/ : SUCCESS in 58m 28s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/46/665846/1/check/kolla-ansible-centos-source-upgrade/e322fd7/ : SUCCESS in 1h 15m 08s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/46/665846/1/check/tripleo-build-containers-centos-7/d7c19de/ : SUCCESS in 42m 03s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/46/665846/1/check/kolla-build-oraclelinux-binary/40d0158/ : SUCCESS in 42m 26s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/46/665846/1/check/kolla-build-oraclelinux-source/aaccb34/ : SUCCESS in 50m 00s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/46/665846/1/check/kolla-ansible-oraclelinux-source/deda6d1/ : SUCCESS in 1h 11m 44s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/46/665846/1/check/kolla-ansible-oraclelinux-binary/a058eab/ : SUCCESS in 1h 07m 09s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/46/665846/1/check/openstack-tox-lower-constraints/b3b792d/ : SUCCESS in 3m 19s\n- openstack-tox-pep8 http://logs.openstack.org/46/665846/1/check/openstack-tox-pep8/81047da/ : SUCCESS in 4m 13s\n- openstack-tox-py27 http://logs.openstack.org/46/665846/1/check/openstack-tox-py27/80d9159/ : SUCCESS in 3m 53s\n- openstack-tox-py36 http://logs.openstack.org/46/665846/1/check/openstack-tox-py36/717ada0/ : SUCCESS in 3m 20s\n- openstack-tox-py37 http://logs.openstack.org/46/665846/1/check/openstack-tox-py37/f8d4e14/ : SUCCESS in 3m 42s\n- openstack-tox-docs http://logs.openstack.org/46/665846/1/check/openstack-tox-docs/c5cb565/html/ : SUCCESS in 2m 24s\n- kolla-build-ubuntu-binary http://logs.openstack.org/46/665846/1/check/kolla-build-ubuntu-binary/57cb24f/ : SUCCESS in 23m 20s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/46/665846/1/check/kolla-build-ubuntu-source/4599cf0/ : SUCCESS in 29m 01s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/46/665846/1/check/kolla-ansible-ubuntu-source/10b3aee/ : SUCCESS in 44m 42s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/46/665846/1/check/kolla-ansible-ubuntu-binary/feba810/ : SUCCESS in 45m 46s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/46/665846/1/check/kolla-ansible-ubuntu-source-upgrade/37a1396/ : FAILURE in 29m 53s","accounts_in_message":[],"_revision_number":1},{"id":"d2706ef68ddccbd2870956e01b4b1c9dee9316ff","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-18 06:20:15.000000000","message":"Patch Set 1:\n\nrecheck\n\n\u003e INFO:kolla.common.utils.fluentd:\u001b[91mERROR:  While executing gem ... (Gem::RemoteFetcher::FetchError)\n\u003e INFO:kolla.common.utils.fluentd:    too many connection resets (https://api.rubygems.org/gems/msgpack-1.2.10.gem)","accounts_in_message":[],"_revision_number":1},{"id":"a92669d9034d5d169bd7a2e5292d2604b74856a3","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-18 06:21:38.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"0031b299a058d329d14380dfced297efff27d60b","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-18 06:56:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/46/665846/1/openstack-check/tripleo-build-containers-fedora-28-master/e340ddd/ : SUCCESS in 32m 48s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/46/665846/1/openstack-check/tripleo-build-containers-fedora-28-stein/db54bc4/ : SUCCESS in 35m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9b064e327de7f49bea610867732aad7e3c0a5fbe","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 08:21:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/46/665846/1/check/test-release-openstack/d80bbc5/ : SUCCESS in 2m 05s\n- kolla-build-centos-binary http://logs.openstack.org/46/665846/1/check/kolla-build-centos-binary/29c6578/ : SUCCESS in 50m 25s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/46/665846/1/check/kolla-build-centos-source/112a46f/ : SUCCESS in 55m 53s\n- kolla-ansible-centos-source http://logs.openstack.org/46/665846/1/check/kolla-ansible-centos-source/62e447e/ : SUCCESS in 1h 15m 42s\n- kolla-ansible-centos-binary http://logs.openstack.org/46/665846/1/check/kolla-ansible-centos-binary/ce8295f/ : SUCCESS in 1h 15m 53s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/46/665846/1/check/kolla-ansible-centos-source-upgrade/b4d4609/ : SUCCESS in 1h 15m 50s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/46/665846/1/check/tripleo-build-containers-centos-7/d8e2adb/ : SUCCESS in 46m 54s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/46/665846/1/check/kolla-build-oraclelinux-binary/38b1022/ : SUCCESS in 47m 20s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/46/665846/1/check/kolla-build-oraclelinux-source/d0be9e5/ : SUCCESS in 55m 03s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/46/665846/1/check/kolla-ansible-oraclelinux-source/4c1b461/ : SUCCESS in 1h 02m 11s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/46/665846/1/check/kolla-ansible-oraclelinux-binary/0fd923b/ : SUCCESS in 1h 05m 49s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/46/665846/1/check/openstack-tox-lower-constraints/c76fee6/ : SUCCESS in 3m 25s\n- openstack-tox-pep8 http://logs.openstack.org/46/665846/1/check/openstack-tox-pep8/06d6d99/ : SUCCESS in 4m 48s\n- openstack-tox-py27 http://logs.openstack.org/46/665846/1/check/openstack-tox-py27/bfc9a37/ : SUCCESS in 4m 20s\n- openstack-tox-py36 http://logs.openstack.org/46/665846/1/check/openstack-tox-py36/cd8da46/ : SUCCESS in 3m 53s\n- openstack-tox-py37 http://logs.openstack.org/46/665846/1/check/openstack-tox-py37/83a0f0f/ : SUCCESS in 3m 33s\n- openstack-tox-docs http://logs.openstack.org/46/665846/1/check/openstack-tox-docs/25f5b78/html/ : SUCCESS in 3m 25s\n- kolla-build-ubuntu-binary http://logs.openstack.org/46/665846/1/check/kolla-build-ubuntu-binary/99545ba/ : SUCCESS in 24m 24s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/46/665846/1/check/kolla-build-ubuntu-source/f218eaf/ : SUCCESS in 29m 19s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/46/665846/1/check/kolla-ansible-ubuntu-source/98f839f/ : SUCCESS in 49m 33s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/46/665846/1/check/kolla-ansible-ubuntu-binary/a3b2959/ : SUCCESS in 44m 26s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/46/665846/1/check/kolla-ansible-ubuntu-source-upgrade/a7e2498/ : SUCCESS in 50m 42s","accounts_in_message":[],"_revision_number":1},{"id":"e5e81a3cb4851dfbba81379eadd8106334692326","author":{"_account_id":24061,"name":"pangliye","email":"pangliye@inspur.com","username":"pangliye"},"date":"2019-06-18 09:37:38.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"1996f9c09f7bb32d5886fd317588277149b67399","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-18 10:18:01.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/46/665846/2/openstack-check/tripleo-build-containers-fedora-28-master/2e39499/ : SUCCESS in 37m 05s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/46/665846/2/openstack-check/tripleo-build-containers-fedora-28-stein/c5ca63b/ : SUCCESS in 39m 09s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"13add12f4f84d2d755d36dc4e3ec4efdcac9df8b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 11:27:22.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/46/665846/2/check/test-release-openstack/9bdea96/ : SUCCESS in 2m 22s\n- kolla-build-centos-binary http://logs.openstack.org/46/665846/2/check/kolla-build-centos-binary/71a27bd/ : SUCCESS in 28m 46s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/46/665846/2/check/kolla-build-centos-source/a3c254b/ : SUCCESS in 40m 30s\n- kolla-ansible-centos-source http://logs.openstack.org/46/665846/2/check/kolla-ansible-centos-source/c933fb0/ : FAILURE in 31m 45s\n- kolla-ansible-centos-binary http://logs.openstack.org/46/665846/2/check/kolla-ansible-centos-binary/cf25ac8/ : SUCCESS in 1h 12m 50s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/46/665846/2/check/kolla-ansible-centos-source-upgrade/e54eb33/ : SUCCESS in 1h 28m 28s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/46/665846/2/check/tripleo-build-containers-centos-7/6c2fae8/ : SUCCESS in 46m 58s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/46/665846/2/check/kolla-build-oraclelinux-binary/5a00ce0/ : FAILURE in 1h 10m 32s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/46/665846/2/check/kolla-build-oraclelinux-source/def20ec/ : FAILURE in 55m 14s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/46/665846/2/check/kolla-ansible-oraclelinux-source/eae65c3/ : SUCCESS in 1h 08m 29s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/46/665846/2/check/kolla-ansible-oraclelinux-binary/7d557a9/ : SUCCESS in 1h 05m 35s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/46/665846/2/check/openstack-tox-lower-constraints/589af9a/ : SUCCESS in 3m 58s\n- openstack-tox-pep8 http://logs.openstack.org/46/665846/2/check/openstack-tox-pep8/2db6f4c/ : SUCCESS in 3m 17s\n- openstack-tox-py27 http://logs.openstack.org/46/665846/2/check/openstack-tox-py27/62ef8c1/ : SUCCESS in 3m 34s\n- openstack-tox-py36 http://logs.openstack.org/46/665846/2/check/openstack-tox-py36/6b5be1e/ : SUCCESS in 3m 01s\n- openstack-tox-py37 http://logs.openstack.org/46/665846/2/check/openstack-tox-py37/119f381/ : SUCCESS in 3m 09s\n- openstack-tox-docs http://logs.openstack.org/46/665846/2/check/openstack-tox-docs/01f615d/html/ : SUCCESS in 4m 21s\n- kolla-build-ubuntu-binary http://logs.openstack.org/46/665846/2/check/kolla-build-ubuntu-binary/a5f834d/ : FAILURE in 27m 01s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/46/665846/2/check/kolla-build-ubuntu-source/2b5deda/ : SUCCESS in 31m 36s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/46/665846/2/check/kolla-ansible-ubuntu-source/eb10c00/ : SUCCESS in 51m 26s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/46/665846/2/check/kolla-ansible-ubuntu-binary/51dfb40/ : FAILURE in 29m 11s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/46/665846/2/check/kolla-ansible-ubuntu-source-upgrade/7852b6a/ : SUCCESS in 53m 31s","accounts_in_message":[],"_revision_number":2},{"id":"aef4323ae27b2df9568b2b7df02e49c3715290b6","author":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"date":"2019-06-19 09:21:31.000000000","message":"Patch Set 2:\n\nis there any blueprint or spec for this new image/service?","accounts_in_message":[],"_revision_number":2},{"id":"13b07fb762ca53bac851ca00ee703033d47c5da0","author":{"_account_id":24061,"name":"pangliye","email":"pangliye@inspur.com","username":"pangliye"},"date":"2019-07-01 11:48:44.000000000","message":"Patch Set 2:\n\nWe use  Kapacitor in our clusters","accounts_in_message":[],"_revision_number":2},{"id":"226c275c117166774753fb4061a6aeb251335614","tag":"autogenerated:gerrit:abandon","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-03-25 18:48:01.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"569f0d8fd73b0dc8669b286502a02a56dbc3f5cf","revisions":{"5968d429d40dbbd32b9e549930603fce4fb7fcbc":{"kind":"REWORK","_number":1,"created":"2019-06-18 02:29:05.000000000","uploader":{"_account_id":24061,"name":"pangliye","email":"pangliye@inspur.com","username":"pangliye"},"ref":"refs/changes/46/665846/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/665846/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/665846/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/665846/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/665846/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/665846/1"}}},"commit":{"parents":[{"commit":"b07241ff5537e6f3807943c1bb9bc26a9badc7aa","subject":"Merge \"grafana/aarch64: use package from upstream repositories\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b07241ff5537e6f3807943c1bb9bc26a9badc7aa"}]}],"author":{"name":"pangliye","email":"pangliye@inspur.com","date":"2019-06-18 02:27:02.000000000","tz":480},"committer":{"name":"pangliye","email":"pangliye@inspur.com","date":"2019-06-18 02:28:45.000000000","tz":480},"subject":"add kapacitor","message":"add kapacitor\n\nadd kapacitor.\nhttps://review.opendev.org/#/c/665841/\n\nChange-Id: I14d9135937109a803c3d53ccf369725d61b73dca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5968d429d40dbbd32b9e549930603fce4fb7fcbc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5968d429d40dbbd32b9e549930603fce4fb7fcbc"}]},"branch":"refs/heads/master"},"569f0d8fd73b0dc8669b286502a02a56dbc3f5cf":{"kind":"REWORK","_number":2,"created":"2019-06-18 09:37:38.000000000","uploader":{"_account_id":24061,"name":"pangliye","email":"pangliye@inspur.com","username":"pangliye"},"ref":"refs/changes/46/665846/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/665846/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/665846/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/665846/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/665846/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/665846/2"}}},"commit":{"parents":[{"commit":"b07241ff5537e6f3807943c1bb9bc26a9badc7aa","subject":"Merge \"grafana/aarch64: use package from upstream repositories\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b07241ff5537e6f3807943c1bb9bc26a9badc7aa"}]}],"author":{"name":"pangliye","email":"pangliye@inspur.com","date":"2019-06-18 02:27:02.000000000","tz":480},"committer":{"name":"pangliye","email":"pangliye@inspur.com","date":"2019-06-18 09:36:37.000000000","tz":0},"subject":"add kapacitor","message":"add kapacitor\n\nadd kapacitor.\nhttps://review.opendev.org/#/c/665841/\n\nChange-Id: I14d9135937109a803c3d53ccf369725d61b73dca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/569f0d8fd73b0dc8669b286502a02a56dbc3f5cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/569f0d8fd73b0dc8669b286502a02a56dbc3f5cf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
