)]}'
{"id":"openstack%2Fkolla~665908","triplet_id":"openstack%2Fkolla~master~If75db0f9fc1bfe82a4ace2cd5bfe603cfe5a5396","project":"openstack/kolla","branch":"master","hashtags":[],"change_id":"If75db0f9fc1bfe82a4ace2cd5bfe603cfe5a5396","subject":"ubuntu: use UCA/train","status":"MERGED","created":"2019-06-18 08:31:38.000000000","updated":"2019-07-03 07:37:49.000000000","submitted":"2019-07-03 07:37:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"665908-1562139469025-d5a15b68","meta_rev_id":"226071cc3f2b35b7b88d24bc7b6268c996189f83","_number":665908,"virtual_id_number":665908,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":2,"date":"2019-07-03 07:37:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-06-27 10:20:08.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2019-07-03 06:02:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2019-06-27 14:13:00.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2019-06-27 13:51:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2019-07-03 06:02:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2019-06-27 13:51:01.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-27 10:20:08.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-06-27 13:51:41.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-06-27 14:13:00.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-07-03 06:02:48.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2019-07-03 07:37:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8288e2dd8c6d2d7b3dacf5f28155f4d331f6293e","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-18 08:31:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"aa24af502b6a5dd9e272324e4a186f530eff34e8","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-18 08:45:18.000000000","message":"Patch Set 1:\n\nhttps://review.opendev.org/#/c/665896/","accounts_in_message":[],"_revision_number":1},{"id":"a07339dcc0e6b56cb6d194928974fae68c7f902e","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-18 09:05:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/08/665908/1/openstack-check/tripleo-build-containers-fedora-28-master/81efd2c/ : SUCCESS in 32m 50s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/08/665908/1/openstack-check/tripleo-build-containers-fedora-28-stein/a219537/ : SUCCESS in 31m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d12b13bd792525a2100cd4130b1d31ec0bbcc533","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 10:45:59.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/08/665908/1/check/kolla-build-centos-binary/ac499da/ : SUCCESS in 46m 28s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/08/665908/1/check/kolla-build-centos-source/ac06e26/ : SUCCESS in 51m 21s\n- kolla-ansible-centos-source http://logs.openstack.org/08/665908/1/check/kolla-ansible-centos-source/1562b41/ : SUCCESS in 1h 10m 12s\n- kolla-ansible-centos-binary http://logs.openstack.org/08/665908/1/check/kolla-ansible-centos-binary/d603b85/ : SUCCESS in 1h 03m 38s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/08/665908/1/check/kolla-ansible-centos-source-upgrade/5475de5/ : SUCCESS in 1h 25m 55s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/08/665908/1/check/tripleo-build-containers-centos-7/a3e6216/ : SUCCESS in 55m 16s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/08/665908/1/check/kolla-build-oraclelinux-binary/da73b26/ : SUCCESS in 58m 42s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/08/665908/1/check/kolla-build-oraclelinux-source/d19a091/ : SUCCESS in 1h 01m 55s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/08/665908/1/check/kolla-ansible-oraclelinux-source/fdd85d8/ : SUCCESS in 1h 23m 03s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/08/665908/1/check/kolla-ansible-oraclelinux-binary/4bd78b0/ : SUCCESS in 1h 09m 38s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/08/665908/1/check/openstack-tox-lower-constraints/f0c4384/ : SUCCESS in 3m 22s\n- openstack-tox-pep8 http://logs.openstack.org/08/665908/1/check/openstack-tox-pep8/c7ae4e7/ : SUCCESS in 3m 01s\n- openstack-tox-py27 http://logs.openstack.org/08/665908/1/check/openstack-tox-py27/2759fd8/ : SUCCESS in 3m 07s\n- openstack-tox-py36 http://logs.openstack.org/08/665908/1/check/openstack-tox-py36/9ab098a/ : SUCCESS in 3m 41s\n- openstack-tox-py37 http://logs.openstack.org/08/665908/1/check/openstack-tox-py37/ad22105/ : SUCCESS in 3m 19s\n- openstack-tox-docs http://logs.openstack.org/08/665908/1/check/openstack-tox-docs/0b7e4aa/html/ : SUCCESS in 2m 43s\n- kolla-build-ubuntu-binary http://logs.openstack.org/08/665908/1/check/kolla-build-ubuntu-binary/d35f63d/ : FAILURE in 5m 58s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/08/665908/1/check/kolla-build-ubuntu-source/0e1b309/ : FAILURE in 5m 52s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/08/665908/1/check/kolla-ansible-ubuntu-source/4045829/ : SUCCESS in 44m 20s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/08/665908/1/check/kolla-ansible-ubuntu-binary/4bd1c83/ : SUCCESS in 43m 42s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/08/665908/1/check/kolla-ansible-ubuntu-source-upgrade/9b14b51/ : SUCCESS in 52m 30s","accounts_in_message":[],"_revision_number":1},{"id":"9b83bdb484fbccf9ca6f8e821695953f4a81cd02","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-18 12:04:27.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"9a299fe399657940788508ae71f79a25caba890f","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-18 12:39:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master finger://ze04.softwarefactory-project.io:7979/189ed1d9ddc14a28808700a5c0b1af01 : RETRY_LIMIT in 6s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/08/665908/1/openstack-check/tripleo-build-containers-fedora-28-stein/790b6f2/ : SUCCESS in 33m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"07e9e8dc4a0f4ec4cd1abf21aa24462182f9364d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 14:09:21.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/08/665908/1/check/kolla-build-centos-binary/03c2eb0/ : SUCCESS in 40m 59s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/08/665908/1/check/kolla-build-centos-source/1b966d8/ : RETRY_LIMIT in 50m 57s\n- kolla-ansible-centos-source http://logs.openstack.org/08/665908/1/check/kolla-ansible-centos-source/8a4dc66/ : RETRY_LIMIT in 50m 59s\n- kolla-ansible-centos-binary http://logs.openstack.org/08/665908/1/check/kolla-ansible-centos-binary/34e02b7/ : RETRY_LIMIT in 51m 08s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/08/665908/1/check/kolla-ansible-centos-source-upgrade/b3844e6/ : FAILURE in 58m 52s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/08/665908/1/check/tripleo-build-containers-centos-7/6636c1b/ : SUCCESS in 37m 04s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/08/665908/1/check/kolla-build-oraclelinux-binary/563f5a6/ : RETRY_LIMIT in 4m 59s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/08/665908/1/check/kolla-build-oraclelinux-source/9da6b60/ : SUCCESS in 51m 49s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/08/665908/1/check/kolla-ansible-oraclelinux-source/5db998e/ : SUCCESS in 1h 11m 46s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/08/665908/1/check/kolla-ansible-oraclelinux-binary/b334635/ : SUCCESS in 1h 06m 50s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/08/665908/1/check/openstack-tox-lower-constraints/a569920/ : SUCCESS in 4m 05s\n- openstack-tox-pep8 http://logs.openstack.org/08/665908/1/check/openstack-tox-pep8/01813e7/ : SUCCESS in 3m 35s\n- openstack-tox-py27 http://logs.openstack.org/08/665908/1/check/openstack-tox-py27/b25104d/ : SUCCESS in 3m 13s\n- openstack-tox-py36 http://logs.openstack.org/08/665908/1/check/openstack-tox-py36/9daff6a/ : SUCCESS in 3m 12s\n- openstack-tox-py37 http://logs.openstack.org/08/665908/1/check/openstack-tox-py37/3b078c3/ : SUCCESS in 3m 27s\n- openstack-tox-docs http://logs.openstack.org/08/665908/1/check/openstack-tox-docs/512b587/html/ : SUCCESS in 2m 30s\n- kolla-build-ubuntu-binary http://logs.openstack.org/08/665908/1/check/kolla-build-ubuntu-binary/4c99bba/ : FAILURE in 6m 13s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/08/665908/1/check/kolla-build-ubuntu-source/00d32bd/ : FAILURE in 5m 37s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/08/665908/1/check/kolla-ansible-ubuntu-source/171197a/ : SUCCESS in 46m 22s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/08/665908/1/check/kolla-ansible-ubuntu-binary/8c13743/ : SUCCESS in 46m 52s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/08/665908/1/check/kolla-ansible-ubuntu-source-upgrade/3554dcf/ : SUCCESS in 52m 24s","accounts_in_message":[],"_revision_number":1},{"id":"8e95c02abdac82183fa52d6b32aa456f808462d6","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-20 10:30:47.000000000","message":"Patch Set 1:\n\ndepends on infra doing train UCA mirror","accounts_in_message":[],"_revision_number":1},{"id":"1ad1de9dc1944e1ea4858d2e4d3f11b93c4b6600","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-20 10:32:48.000000000","message":"Patch Set 1:\n\nhttps://review.opendev.org/666533 sent","accounts_in_message":[],"_revision_number":1},{"id":"3c5609f27a51060f24befdae78a9a538515d8592","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-25 07:30:02.000000000","message":"Patch Set 1:\n\nrecheck\n\nmirrors got train packages","accounts_in_message":[],"_revision_number":1},{"id":"3dba66db4af0c65b74beaa166daad0a47eee61d7","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-25 08:10:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/08/665908/1/openstack-check/tripleo-build-containers-fedora-28-master/964203e/ : FAILURE in 5m 19s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/08/665908/1/openstack-check/tripleo-build-containers-fedora-28-stein/b36bfcc/ : SUCCESS in 37m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"59c23eafac1c501d5a8de5c2f23070de3201d175","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-25 09:08:32.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary finger://ze01.openstack.org/be9d40c3dda143ee98926b4d5e390de8 : RETRY_LIMIT in 34m 01s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/08/665908/1/check/kolla-build-centos-source/d2ad846/ : SUCCESS in 38m 29s\n- kolla-ansible-centos-source finger://ze02.openstack.org/29978e4aa2724d40842903a01df1a42f : RETRY_LIMIT in 1h 32m 12s\n- kolla-ansible-centos-binary http://logs.openstack.org/08/665908/1/check/kolla-ansible-centos-binary/1797cdc/ : SUCCESS in 1h 09m 38s (non-voting)\n- kolla-ansible-centos-source-upgrade finger://ze06.openstack.org/6984fcc5fbab4890b64cc3c72ff1a677 : RETRY_LIMIT in 1h 32m 24s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/08/665908/1/check/tripleo-build-containers-centos-7/fb88002/ : SUCCESS in 52m 07s\n- openstack-tox-lower-constraints http://logs.openstack.org/08/665908/1/check/openstack-tox-lower-constraints/388c0c8/ : SUCCESS in 3m 28s\n- openstack-tox-pep8 http://logs.openstack.org/08/665908/1/check/openstack-tox-pep8/2eadf8d/ : SUCCESS in 2m 44s\n- openstack-tox-py27 http://logs.openstack.org/08/665908/1/check/openstack-tox-py27/3c53068/ : SUCCESS in 3m 16s\n- openstack-tox-py36 http://logs.openstack.org/08/665908/1/check/openstack-tox-py36/eb60ee3/ : SUCCESS in 3m 31s\n- openstack-tox-py37 http://logs.openstack.org/08/665908/1/check/openstack-tox-py37/e78490b/ : SUCCESS in 3m 37s\n- openstack-tox-docs http://logs.openstack.org/08/665908/1/check/openstack-tox-docs/a88ad2a/html/ : SUCCESS in 2m 58s\n- kolla-build-ubuntu-binary http://logs.openstack.org/08/665908/1/check/kolla-build-ubuntu-binary/1c7cd9d/ : SUCCESS in 24m 49s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/08/665908/1/check/kolla-build-ubuntu-source/193ef83/ : SUCCESS in 31m 01s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/08/665908/1/check/kolla-ansible-ubuntu-source/b93293c/ : SUCCESS in 46m 51s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/08/665908/1/check/kolla-ansible-ubuntu-binary/da177fd/ : SUCCESS in 53m 46s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/08/665908/1/check/kolla-ansible-ubuntu-source-upgrade/083e058/ : SUCCESS in 1h 05m 20s","accounts_in_message":[],"_revision_number":1},{"id":"09b3c34aa20cd49b3db0f1e7128688367f591a43","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-25 09:11:13.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e3d1942be40c1816322f41d06cf7af2fe0ffb373","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-25 09:50:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/08/665908/1/openstack-check/tripleo-build-containers-fedora-28-master/9fe20a2/ : FAILURE in 5m 28s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/08/665908/1/openstack-check/tripleo-build-containers-fedora-28-stein/9378630/ : SUCCESS in 38m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9120511349691ac645f6fb9cd3e095f9b4195920","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-25 11:57:01.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/08/665908/1/check/kolla-build-centos-binary/5ef337f/ : SUCCESS in 44m 43s (non-voting)\n- kolla-build-centos-source finger://ze06.openstack.org/a6c2a087ffea4f35a1e4d48d56195679 : RETRY_LIMIT in 49m 53s\n- kolla-ansible-centos-source finger://ze10.openstack.org/ac8394e9a8ce4310882fca03fd92dd8a : RETRY_LIMIT in 50m 51s\n- kolla-ansible-centos-binary finger://ze06.openstack.org/d98671852ca940808f6f7b9d46f15c1f : RETRY_LIMIT in 50m 57s (non-voting)\n- kolla-ansible-centos-source-upgrade finger://ze03.openstack.org/ab1004ee36e14972a7c08c3f6a3f972e : RETRY_LIMIT in 55m 37s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/08/665908/1/check/tripleo-build-containers-centos-7/25b89a3/ : SUCCESS in 41m 40s\n- openstack-tox-lower-constraints http://logs.openstack.org/08/665908/1/check/openstack-tox-lower-constraints/8ea4b10/ : SUCCESS in 4m 18s\n- openstack-tox-pep8 http://logs.openstack.org/08/665908/1/check/openstack-tox-pep8/98a9bae/ : SUCCESS in 3m 06s\n- openstack-tox-py27 http://logs.openstack.org/08/665908/1/check/openstack-tox-py27/30a13c1/ : SUCCESS in 4m 04s\n- openstack-tox-py36 http://logs.openstack.org/08/665908/1/check/openstack-tox-py36/c18b656/ : SUCCESS in 3m 32s\n- openstack-tox-py37 http://logs.openstack.org/08/665908/1/check/openstack-tox-py37/67137c4/ : SUCCESS in 3m 05s\n- openstack-tox-docs http://logs.openstack.org/08/665908/1/check/openstack-tox-docs/fb61669/html/ : SUCCESS in 2m 52s\n- kolla-build-ubuntu-binary http://logs.openstack.org/08/665908/1/check/kolla-build-ubuntu-binary/886f200/ : FAILURE in 17m 01s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/08/665908/1/check/kolla-build-ubuntu-source/32a4d73/ : FAILURE in 16m 05s\n- kolla-ansible-ubuntu-source finger://ze04.openstack.org/64be849e22804659b33071aa3667f9bd : RETRY_LIMIT in 49m 26s\n- kolla-ansible-ubuntu-binary finger://ze09.openstack.org/484e5e172c9c473ea15381eb20dbcb25 : RETRY_LIMIT in 50m 22s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade finger://ze08.openstack.org/7f74ee0127344e95a6618209004f19e5 : RETRY_LIMIT in 51m 43s","accounts_in_message":[],"_revision_number":1},{"id":"e0a8a05ec24c8831f367282b92fc130c6eea0892","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-25 21:11:33.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"afcc3a32ea39c32db48ebdb011d2675b99167fd1","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-25 22:27:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/08/665908/1/openstack-check/tripleo-build-containers-fedora-28-master/f6aee25/ : FAILURE in 5m 22s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/08/665908/1/openstack-check/tripleo-build-containers-fedora-28-stein/49d511c/ : SUCCESS in 1h 14m 44s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9c00c3c7507f484c1e0326a88d077bf455001c69","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-26 00:12:11.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/08/665908/1/check/kolla-build-centos-binary/a69080a/ : SUCCESS in 32m 51s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/08/665908/1/check/kolla-build-centos-source/7f3f3e4/ : TIMED_OUT in 2h 53m 26s\n- kolla-ansible-centos-source http://logs.openstack.org/08/665908/1/check/kolla-ansible-centos-source/5c020f2/ : TIMED_OUT in 2h 01m 03s\n- kolla-ansible-centos-binary http://logs.openstack.org/08/665908/1/check/kolla-ansible-centos-binary/59b1146/ : SUCCESS in 1h 03m 10s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/08/665908/1/check/kolla-ansible-centos-source-upgrade/22acf76/ : TIMED_OUT in 2h 01m 09s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/08/665908/1/check/tripleo-build-containers-centos-7/7d94ed0/ : SUCCESS in 47m 49s\n- openstack-tox-lower-constraints http://logs.openstack.org/08/665908/1/check/openstack-tox-lower-constraints/36ced7f/ : SUCCESS in 3m 21s\n- openstack-tox-pep8 http://logs.openstack.org/08/665908/1/check/openstack-tox-pep8/8ab9d45/ : SUCCESS in 4m 38s\n- openstack-tox-py27 http://logs.openstack.org/08/665908/1/check/openstack-tox-py27/c0ab39c/ : SUCCESS in 3m 11s\n- openstack-tox-py36 http://logs.openstack.org/08/665908/1/check/openstack-tox-py36/39aa3d9/ : SUCCESS in 3m 31s\n- openstack-tox-py37 http://logs.openstack.org/08/665908/1/check/openstack-tox-py37/392d42d/ : SUCCESS in 3m 51s\n- openstack-tox-docs http://logs.openstack.org/08/665908/1/check/openstack-tox-docs/d3ed06d/html/ : SUCCESS in 3m 41s\n- kolla-build-ubuntu-binary http://logs.openstack.org/08/665908/1/check/kolla-build-ubuntu-binary/d963fc7/ : SUCCESS in 24m 53s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/08/665908/1/check/kolla-build-ubuntu-source/fa0c2c4/ : SUCCESS in 42m 44s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/08/665908/1/check/kolla-ansible-ubuntu-source/fac9b1b/ : SUCCESS in 47m 43s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/08/665908/1/check/kolla-ansible-ubuntu-binary/affe629/ : SUCCESS in 44m 08s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/08/665908/1/check/kolla-ansible-ubuntu-source-upgrade/8585f6d/ : SUCCESS in 58m 58s","accounts_in_message":[],"_revision_number":1},{"id":"7d2a539b4884985fefca70ba45eb77f1dc164ce5","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-27 07:59:58.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1e12963ed436c7bf4b03d941d706f23f693ae756","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-27 08:41:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/08/665908/1/openstack-check/tripleo-build-containers-fedora-28-master/570d8ef/ : FAILURE in 6m 00s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/08/665908/1/openstack-check/tripleo-build-containers-fedora-28-stein/97f656c/ : SUCCESS in 37m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"eaac76b811bac295ad3c503082ba02e73654198b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 09:42:35.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/08/665908/1/check/kolla-build-centos-binary/54313a7/ : RETRY_LIMIT in 6m 40s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/08/665908/1/check/kolla-build-centos-source/ae34225/ : RETRY_LIMIT in 6m 31s\n- kolla-ansible-centos-source http://logs.openstack.org/08/665908/1/check/kolla-ansible-centos-source/b281c36/ : SUCCESS in 1h 10m 40s\n- kolla-ansible-centos-binary http://logs.openstack.org/08/665908/1/check/kolla-ansible-centos-binary/22561cc/ : SUCCESS in 1h 01m 57s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/08/665908/1/check/kolla-ansible-centos-source-upgrade/9d4f231/ : RETRY_LIMIT in 6m 36s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/08/665908/1/check/tripleo-build-containers-centos-7/ba9131e/ : SUCCESS in 42m 07s\n- openstack-tox-lower-constraints http://logs.openstack.org/08/665908/1/check/openstack-tox-lower-constraints/0cfb260/ : SUCCESS in 3m 45s\n- openstack-tox-pep8 http://logs.openstack.org/08/665908/1/check/openstack-tox-pep8/fa9dab4/ : SUCCESS in 2m 50s\n- openstack-tox-py27 http://logs.openstack.org/08/665908/1/check/openstack-tox-py27/53504ce/ : SUCCESS in 3m 46s\n- openstack-tox-py36 http://logs.openstack.org/08/665908/1/check/openstack-tox-py36/37b681d/ : SUCCESS in 3m 35s\n- openstack-tox-py37 http://logs.openstack.org/08/665908/1/check/openstack-tox-py37/3ab715f/ : SUCCESS in 4m 32s\n- openstack-tox-docs http://logs.openstack.org/08/665908/1/check/openstack-tox-docs/2bffb1c/html/ : SUCCESS in 3m 02s\n- kolla-build-ubuntu-binary http://logs.openstack.org/08/665908/1/check/kolla-build-ubuntu-binary/d054ec0/ : RETRY_LIMIT in 14m 38s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/08/665908/1/check/kolla-build-ubuntu-source/1f66822/ : SUCCESS in 34m 36s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/08/665908/1/check/kolla-ansible-ubuntu-source/bb144e6/ : SUCCESS in 48m 20s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/08/665908/1/check/kolla-ansible-ubuntu-binary/1ba5905/ : FAILURE in 15m 39s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/08/665908/1/check/kolla-ansible-ubuntu-source-upgrade/6290a37/ : SUCCESS in 1h 02m 34s","accounts_in_message":[],"_revision_number":1},{"id":"fcb9d41c0f7e16bd06fd760871d6d0b62dc61ad6","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-27 09:45:01.000000000","message":"Patch Set 1:\n\nrecheck\n\nmirror, you know which one","accounts_in_message":[],"_revision_number":1},{"id":"5cdef4340b623d7df19d37e859706fbf3fa0b1ab","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-27 10:20:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/08/665908/1/openstack-check/tripleo-build-containers-fedora-28-master/60212f9/ : FAILURE in 5m 58s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/08/665908/1/openstack-check/tripleo-build-containers-fedora-28-stein/c962d35/ : SUCCESS in 34m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1acef045b9ca74b252e59620d6b68b3a605b7d8b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 12:35:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/08/665908/1/check/kolla-build-centos-binary/3ae850b/ : SUCCESS in 47m 57s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/08/665908/1/check/kolla-build-centos-source/4e2da22/ : SUCCESS in 52m 52s\n- kolla-ansible-centos-source http://logs.openstack.org/08/665908/1/check/kolla-ansible-centos-source/a2081dc/ : SUCCESS in 1h 01m 53s\n- kolla-ansible-centos-binary http://logs.openstack.org/08/665908/1/check/kolla-ansible-centos-binary/adbf8fe/ : SUCCESS in 1h 00m 49s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/08/665908/1/check/kolla-ansible-centos-source-upgrade/b5de367/ : SUCCESS in 1h 13m 10s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/08/665908/1/check/tripleo-build-containers-centos-7/55ae169/ : SUCCESS in 54m 05s\n- openstack-tox-lower-constraints http://logs.openstack.org/08/665908/1/check/openstack-tox-lower-constraints/84f4d0a/ : SUCCESS in 3m 31s\n- openstack-tox-pep8 http://logs.openstack.org/08/665908/1/check/openstack-tox-pep8/de247d0/ : SUCCESS in 3m 00s\n- openstack-tox-py27 http://logs.openstack.org/08/665908/1/check/openstack-tox-py27/eb34fe0/ : SUCCESS in 3m 38s\n- openstack-tox-py36 http://logs.openstack.org/08/665908/1/check/openstack-tox-py36/8caeca0/ : SUCCESS in 3m 20s\n- openstack-tox-py37 http://logs.openstack.org/08/665908/1/check/openstack-tox-py37/14bbf82/ : SUCCESS in 3m 42s\n- openstack-tox-docs http://logs.openstack.org/08/665908/1/check/openstack-tox-docs/d05f36f/html/ : SUCCESS in 3m 06s\n- kolla-build-ubuntu-binary http://logs.openstack.org/08/665908/1/check/kolla-build-ubuntu-binary/dca0a07/ : SUCCESS in 24m 09s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/08/665908/1/check/kolla-build-ubuntu-source/bdf8c78/ : SUCCESS in 32m 45s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/08/665908/1/check/kolla-ansible-ubuntu-source/4834b90/ : SUCCESS in 46m 46s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/08/665908/1/check/kolla-ansible-ubuntu-binary/bd4372f/ : SUCCESS in 42m 58s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/08/665908/1/check/kolla-ansible-ubuntu-source-upgrade/269ea04/ : SUCCESS in 57m 24s","accounts_in_message":[],"_revision_number":1},{"id":"802bcafb83ba233b0db2c1c6068aa4ec0f4a0a05","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-27 13:51:01.000000000","message":"Patch Set 1: Review-Priority+1\n\nWe need to move to Train packages sooner rather than later.","accounts_in_message":[],"_revision_number":1},{"id":"261489b3e3a03a1f097f5ea6b89e6d249e249e47","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-27 13:51:41.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"18c99609aea8b8d4adc2040abb8b80f28c54a442","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-27 14:13:00.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f9f2d719b9b519a61b943002f1f73734817e3436","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-07-03 06:02:48.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c0c65f2e16f0599ca254eeca760379a4c1f79cb2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-03 06:03:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4e6ceb7d6dfeb53d7743d96239abaf1fe04edc32","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-03 07:37:48.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source http://logs.openstack.org/08/665908/1/gate/kolla-build-centos-source/ab5623e/ : SUCCESS in 57m 37s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/08/665908/1/gate/tripleo-build-containers-centos-7/ac62dbc/ : SUCCESS in 43m 35s\n- openstack-tox-lower-constraints http://logs.openstack.org/08/665908/1/gate/openstack-tox-lower-constraints/8f77e60/ : SUCCESS in 5m 43s\n- openstack-tox-pep8 http://logs.openstack.org/08/665908/1/gate/openstack-tox-pep8/e64b6e6/ : SUCCESS in 4m 57s\n- openstack-tox-py27 http://logs.openstack.org/08/665908/1/gate/openstack-tox-py27/630e7cb/ : SUCCESS in 3m 52s\n- openstack-tox-py36 http://logs.openstack.org/08/665908/1/gate/openstack-tox-py36/074de34/ : SUCCESS in 3m 09s\n- openstack-tox-py37 http://logs.openstack.org/08/665908/1/gate/openstack-tox-py37/961e17f/ : SUCCESS in 5m 36s\n- openstack-tox-docs http://logs.openstack.org/08/665908/1/gate/openstack-tox-docs/603a5b2/html/ : SUCCESS in 3m 58s\n- kolla-build-ubuntu-source http://logs.openstack.org/08/665908/1/gate/kolla-build-ubuntu-source/f564a8b/ : SUCCESS in 26m 38s","accounts_in_message":[],"_revision_number":1},{"id":"226071cc3f2b35b7b88d24bc7b6268c996189f83","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-03 07:37:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7dab1fedd044a95fa1203b9ca809e13330f5936b","revisions":{"7dab1fedd044a95fa1203b9ca809e13330f5936b":{"kind":"REWORK","_number":1,"created":"2019-06-18 08:31:38.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/08/665908/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/08/665908/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/08/665908/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/08/665908/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/08/665908/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/08/665908/1"}}},"commit":{"parents":[{"commit":"b07241ff5537e6f3807943c1bb9bc26a9badc7aa","subject":"Merge \"grafana/aarch64: use package from upstream repositories\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b07241ff5537e6f3807943c1bb9bc26a9badc7aa"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-06-18 08:30:40.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-06-18 08:31:18.000000000","tz":120},"subject":"ubuntu: use UCA/train","message":"ubuntu: use UCA/train\n\nChange-Id: If75db0f9fc1bfe82a4ace2cd5bfe603cfe5a5396\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7dab1fedd044a95fa1203b9ca809e13330f5936b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7dab1fedd044a95fa1203b9ca809e13330f5936b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
