)]}'
{"id":"openstack%2Fkolla~665935","triplet_id":"openstack%2Fkolla~stable%2Focata~I52b1bf9b7b92b0874d7a59f0d6fa786368e82d2f","project":"openstack/kolla","branch":"stable/ocata","hashtags":[],"change_id":"I52b1bf9b7b92b0874d7a59f0d6fa786368e82d2f","subject":"Mark kolla-build-*-source jobs non-voting","status":"MERGED","created":"2019-06-18 10:10:56.000000000","updated":"2019-06-20 11:07:41.000000000","submitted":"2019-06-20 11:07:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"665935-1561028861445-420ce86b","meta_rev_id":"943c733231cdf50246336a20f8cbda1f6480cffd","_number":665935,"virtual_id_number":665935,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2019-06-20 11:07:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"all":[{"value":2,"date":"2019-06-20 09:36:51.000000000","_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":1,"date":"2019-06-19 15:42:19.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":2,"date":"2019-06-19 20:40:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2019-06-20 10:29:25.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2019-06-20 10:29:25.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-19 15:42:19.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2019-06-19 20:40:06.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2019-06-20 09:36:51.000000000","updated_by":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"reviewer":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"state":"REVIEWER"},{"updated":"2019-06-20 10:29:25.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-06-20 11:07:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5a73e5c252495e8a4288025dfc0f4caa2ab8316c","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-18 10:10:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d3593d19cb1c29fcbe1cbe0c706623fa08ccb308","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 12:17:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/35/665935/1/check/openstack-tox-pep8/8b8e0c9/ : SUCCESS in 3m 27s\n- openstack-tox-py27 http://logs.openstack.org/35/665935/1/check/openstack-tox-py27/d7568b8/ : SUCCESS in 3m 41s\n- openstack-tox-py35 http://logs.openstack.org/35/665935/1/check/openstack-tox-py35/701dff5/ : SUCCESS in 3m 56s\n- build-openstack-sphinx-docs http://logs.openstack.org/35/665935/1/check/build-openstack-sphinx-docs/83ec3d1/html/ : SUCCESS in 3m 43s\n- kolla-build-centos-binary http://logs.openstack.org/35/665935/1/check/kolla-build-centos-binary/4333894/ : FAILURE in 8m 39s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/35/665935/1/check/kolla-build-centos-source/a5aeb3e/ : FAILURE in 10m 17s (non-voting)\n- kolla-build-ubuntu-binary http://logs.openstack.org/35/665935/1/check/kolla-build-ubuntu-binary/cc48206/ : RETRY_LIMIT in 1h 39m 38s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/35/665935/1/check/kolla-build-ubuntu-source/7a3b481/ : FAILURE in 27m 35s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/35/665935/1/check/kolla-build-oraclelinux-binary/8f3a0dd/ : FAILURE in 8m 46s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/35/665935/1/check/kolla-build-oraclelinux-source/5e1d9fe/ : FAILURE in 8m 47s (non-voting)\n- kolla-ansible-centos-source http://logs.openstack.org/35/665935/1/check/kolla-ansible-centos-source/b2c6c3f/ : FAILURE in 9m 50s (non-voting)\n- kolla-ansible-centos-binary http://logs.openstack.org/35/665935/1/check/kolla-ansible-centos-binary/731b893/ : FAILURE in 9m 24s (non-voting)\n- kolla-ansible-ubuntu-source http://logs.openstack.org/35/665935/1/check/kolla-ansible-ubuntu-source/a7e8d43/ : FAILURE in 16m 26s (non-voting)\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/35/665935/1/check/kolla-ansible-ubuntu-binary/592b665/ : FAILURE in 25m 50s (non-voting)\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/35/665935/1/check/kolla-ansible-oraclelinux-source/30fbd41/ : FAILURE in 9m 22s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/35/665935/1/check/kolla-ansible-oraclelinux-binary/e999024/ : FAILURE in 9m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7d1cee325a32e9a84f845ff7bad8da7f60392f4a","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-06-19 15:42:19.000000000","message":"Patch Set 1: Code-Review+1\n\nSeems sensible.","accounts_in_message":[],"_revision_number":1},{"id":"312daa8467029e5d387965c2059392265de0f1d4","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-06-19 20:40:06.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1daecd1a9e56ac24038d2191fbd09067142b9ba0","author":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"date":"2019-06-20 09:36:51.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d738110d82f92de6ee0d0452da32128cb90bfef2","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-20 10:29:25.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2383b1d538ab13d34823f09d8d5e95a0aa51793e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-20 10:29:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"39cfb0d25dc36996cbd3c03c85373355709f8539","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-20 11:07:41.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/35/665935/1/gate/openstack-tox-pep8/cbc3223/ : SUCCESS in 13m 25s\n- openstack-tox-py27 http://logs.openstack.org/35/665935/1/gate/openstack-tox-py27/72a6a1c/ : SUCCESS in 4m 53s\n- openstack-tox-py35 http://logs.openstack.org/35/665935/1/gate/openstack-tox-py35/9cfadd9/ : SUCCESS in 3m 14s\n- build-openstack-sphinx-docs http://logs.openstack.org/35/665935/1/gate/build-openstack-sphinx-docs/61d125b/html/ : SUCCESS in 3m 31s\n- kolla-build-centos-source http://logs.openstack.org/35/665935/1/gate/kolla-build-centos-source/ef4d1d9/ : FAILURE in 9m 29s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/35/665935/1/gate/kolla-build-ubuntu-source/ac402b5/ : FAILURE in 32m 25s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/35/665935/1/gate/kolla-build-oraclelinux-source/53657da/ : FAILURE in 8m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"943c733231cdf50246336a20f8cbda1f6480cffd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-20 11:07:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bb261c429c5f091d8f5d2abbfbeba965adfab5ca","revisions":{"bb261c429c5f091d8f5d2abbfbeba965adfab5ca":{"kind":"REWORK","_number":1,"created":"2019-06-18 10:10:56.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/35/665935/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/35/665935/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/35/665935/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/35/665935/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/35/665935/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/35/665935/1"}}},"commit":{"parents":[{"commit":"2c69b170b99c6b93883a1f519b0e1aae0e515503","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2c69b170b99c6b93883a1f519b0e1aae0e515503"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-06-18 10:09:28.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-06-18 10:10:37.000000000","tz":60},"subject":"Mark kolla-build-*-source jobs non-voting","message":"Mark kolla-build-*-source jobs non-voting\n\nOcata only\n\nAll build jobs are failing on Ocata, and we cannot disable the image\npublishing jobs without removing them or marking as non-voting.\n\nChange-Id: I52b1bf9b7b92b0874d7a59f0d6fa786368e82d2f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bb261c429c5f091d8f5d2abbfbeba965adfab5ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bb261c429c5f091d8f5d2abbfbeba965adfab5ca"}]},"branch":"refs/heads/stable/ocata"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
