)]}'
{"id":"openstack%2Fkolla~666000","triplet_id":"openstack%2Fkolla~master~I0b25bf33a1ab77a0e84c1eac1caa3184601849d9","project":"openstack/kolla","branch":"master","hashtags":[],"change_id":"I0b25bf33a1ab77a0e84c1eac1caa3184601849d9","subject":"cleanup image squashing","status":"ABANDONED","created":"2019-06-18 14:39:35.000000000","updated":"2019-08-08 08:27:46.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"bfc2ffedf351ce8aa615a635424b20ec4bcd8743","_number":666000,"virtual_id_number":666000,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":-1,"date":"2019-06-18 16:13:46.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2019-06-18 15:19:09.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":-1,"date":"2019-06-18 14:40:13.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-18 15:19:09.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-06-18 16:13:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a8a7e247f1eada6e8d595b1df5ac082754f7b2a0","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-18 14:39:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"837118c826dc82574ef73ff8b78119921d8ba3ca","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-18 14:40:13.000000000","message":"Patch Set 1: Workflow-1\n\nhave to find time and test some improvement ideas","accounts_in_message":[],"_revision_number":1},{"id":"53565b0157ca4f0986d29edf238c0a071677da3f","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-18 15:19:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/00/666000/1/openstack-check/tripleo-build-containers-fedora-28-master/2f6cd42/ : SUCCESS in 32m 32s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/00/666000/1/openstack-check/tripleo-build-containers-fedora-28-stein/f0a1f49/ : SUCCESS in 34m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"153811ba96ba1a721822521c9a15c88d0a09af6e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 16:13:46.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/00/666000/1/check/kolla-build-centos-binary/f18f6c6/ : SUCCESS in 46m 15s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/00/666000/1/check/kolla-build-centos-source/fe12c0c/ : RETRY_LIMIT in 2m 00s\n- kolla-ansible-centos-source http://logs.openstack.org/00/666000/1/check/kolla-ansible-centos-source/65b0672/ : SUCCESS in 1h 25m 49s\n- kolla-ansible-centos-binary http://logs.openstack.org/00/666000/1/check/kolla-ansible-centos-binary/ba81671/ : SUCCESS in 1h 09m 30s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/00/666000/1/check/kolla-ansible-centos-source-upgrade/a82b968/ : SUCCESS in 1h 26m 51s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/00/666000/1/check/tripleo-build-containers-centos-7/330966c/ : SUCCESS in 40m 52s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/00/666000/1/check/kolla-build-oraclelinux-binary/ec7c6ce/ : SUCCESS in 54m 13s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/00/666000/1/check/kolla-build-oraclelinux-source/98a7de4/ : SUCCESS in 44m 00s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/00/666000/1/check/kolla-ansible-oraclelinux-source/1cc6646/ : SUCCESS in 1h 12m 47s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/00/666000/1/check/kolla-ansible-oraclelinux-binary/129ba83/ : SUCCESS in 1h 08m 24s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/00/666000/1/check/openstack-tox-lower-constraints/7d3238c/ : FAILURE in 3m 44s\n- openstack-tox-pep8 http://logs.openstack.org/00/666000/1/check/openstack-tox-pep8/9e16b17/ : FAILURE in 3m 05s\n- openstack-tox-py27 http://logs.openstack.org/00/666000/1/check/openstack-tox-py27/6a0581c/ : FAILURE in 2m 55s\n- openstack-tox-py36 http://logs.openstack.org/00/666000/1/check/openstack-tox-py36/31c5d96/ : FAILURE in 3m 10s\n- openstack-tox-py37 http://logs.openstack.org/00/666000/1/check/openstack-tox-py37/572ec8c/ : FAILURE in 3m 05s\n- openstack-tox-docs http://logs.openstack.org/00/666000/1/check/openstack-tox-docs/9f00ab8/html/ : SUCCESS in 3m 14s\n- kolla-build-ubuntu-binary http://logs.openstack.org/00/666000/1/check/kolla-build-ubuntu-binary/72f9640/ : SUCCESS in 22m 07s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/00/666000/1/check/kolla-build-ubuntu-source/498966d/ : SUCCESS in 29m 36s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/00/666000/1/check/kolla-ansible-ubuntu-source/601188b/ : SUCCESS in 45m 54s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/00/666000/1/check/kolla-ansible-ubuntu-binary/9bbb419/ : SUCCESS in 39m 46s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/00/666000/1/check/kolla-ansible-ubuntu-source-upgrade/0adeab2/ : SUCCESS in 53m 19s","accounts_in_message":[],"_revision_number":1},{"id":"bfc2ffedf351ce8aa615a635424b20ec4bcd8743","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-08-08 08:27:46.000000000","message":"Abandoned\n\ndo not have time for it","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8b1bda59eba4e3e8362b0cb9497af41ba92f5894","revisions":{"8b1bda59eba4e3e8362b0cb9497af41ba92f5894":{"kind":"REWORK","_number":1,"created":"2019-06-18 14:39:35.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/00/666000/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/00/666000/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/666000/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/666000/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/666000/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/00/666000/1"}}},"commit":{"parents":[{"commit":"3b61de2d3193611506ebee37afa3f622ba1b8ecc","subject":"Merge \"Add sha256 in ceph fetch keys\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3b61de2d3193611506ebee37afa3f622ba1b8ecc"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-06-18 14:14:41.000000000","tz":0},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-06-18 14:39:10.000000000","tz":0},"subject":"cleanup image squashing","message":"cleanup image squashing\n\nDocker has image squashing as experimental feature. So let\u0027s use it\ninstead of \u0027docker-squash\u0027 command.\n\nRequires \u0027Experimental\u0027: true in daemon.json configuration of Docker.\n\nFull of issues anyway - building Ceph images means base is OK,\nceph-base is OK but rest of ceph-* images fail as parent layer\ndissappeared.\n\nChange-Id: I0b25bf33a1ab77a0e84c1eac1caa3184601849d9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8b1bda59eba4e3e8362b0cb9497af41ba92f5894"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8b1bda59eba4e3e8362b0cb9497af41ba92f5894"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
