)]}'
{"id":"openstack%2Fkolla~666068","triplet_id":"openstack%2Fkolla~master~I410cb943b766284a6b7bf11b0dd53bd2cdc75749","project":"openstack/kolla","branch":"master","hashtags":[],"change_id":"I410cb943b766284a6b7bf11b0dd53bd2cdc75749","subject":"Enable CI jobs for ceph upgrade, NFV, cinder LVM, ironic","status":"MERGED","created":"2019-06-18 17:29:09.000000000","updated":"2019-07-03 07:38:00.000000000","submitted":"2019-07-03 07:38:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"666068-1562139480682-3639b714","meta_rev_id":"4d3cf77d72e6d0da97abff79d1b12baeb1ffaf18","_number":666068,"virtual_id_number":666068,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"date":"2019-06-27 16:34:50.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":2,"date":"2019-07-03 07:38:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},{"value":2,"date":"2019-07-03 06:10:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2019-06-28 06:45:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":2,"date":"2019-06-28 06:38:01.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},{"value":1,"date":"2019-07-03 06:10:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-18 17:36:37.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-06-20 13:12:33.000000000","updated_by":{"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},"reviewer":{"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},"state":"REVIEWER"},{"updated":"2019-06-27 16:34:50.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-06-28 06:38:01.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-06-28 06:45:32.000000000","updated_by":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"reviewer":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"state":"REVIEWER"},{"updated":"2019-07-03 06:10:50.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2019-07-03 07:38:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"748cd1e18a5578b829b0cf343bb18dfa576c696f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-18 17:29:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"63ec4506435529a2f92dfa6eb36bd30dc6522658","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-18 17:31:31.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"384d31d65a99837254a8ecb79e02666a4e46a4c3","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-18 17:36:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/68/666068/1/openstack-check/tripleo-build-containers-fedora-28-master/c6a5d54/ : FAILURE in 5m 23s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/68/666068/1/openstack-check/tripleo-build-containers-fedora-28-stein/2ca5314/ : FAILURE in 5m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"11c73d952353186c0a08ee2f587595742139d0f7","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-18 17:36:48.000000000","message":"Patch Set 1:\n\n(2 comments)\n\nWe default to ceph for glance if enabled.","accounts_in_message":[],"_revision_number":1},{"id":"c523a15153c300b84eadc27a0ba0502496c48f42","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 19:31:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/68/666068/1/check/kolla-build-centos-binary/be32fad/ : SUCCESS in 49m 07s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/68/666068/1/check/kolla-build-centos-source/83a2fcf/ : SUCCESS in 56m 59s\n- kolla-ansible-centos-source http://logs.openstack.org/68/666068/1/check/kolla-ansible-centos-source/9cde2c2/ : SUCCESS in 1h 09m 02s\n- kolla-ansible-centos-binary http://logs.openstack.org/68/666068/1/check/kolla-ansible-centos-binary/04a8bd1/ : SUCCESS in 1h 10m 02s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/68/666068/1/check/kolla-ansible-centos-source-upgrade/11bbadf/ : SUCCESS in 1h 27m 39s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/68/666068/1/check/tripleo-build-containers-centos-7/e06ca71/ : SUCCESS in 47m 02s\n- kolla-ansible-centos-source-ceph http://logs.openstack.org/68/666068/1/check/kolla-ansible-centos-source-ceph/0f3f35a/ : SUCCESS in 1h 31m 26s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph http://logs.openstack.org/68/666068/1/check/kolla-ansible-centos-source-upgrade-ceph/9c7ac02/ : FAILURE in 1h 24m 49s (non-voting)\n- kolla-ansible-centos-source-cinder-lvm http://logs.openstack.org/68/666068/1/check/kolla-ansible-centos-source-cinder-lvm/94b78b1/ : SUCCESS in 1h 18m 43s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv http://logs.openstack.org/68/666068/1/check/kolla-ansible-centos-source-scenario-nfv/8c7d7db/ : SUCCESS in 1h 38m 23s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/68/666068/1/check/kolla-build-oraclelinux-binary/34bce92/ : SUCCESS in 1h 00m 43s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/68/666068/1/check/kolla-build-oraclelinux-source/2fff7e8/ : SUCCESS in 53m 47s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/68/666068/1/check/kolla-ansible-oraclelinux-source/5cce702/ : SUCCESS in 1h 49m 21s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/68/666068/1/check/kolla-ansible-oraclelinux-binary/4b152fa/ : SUCCESS in 1h 07m 42s (non-voting)\n- kolla-ansible-oraclelinux-source-ceph http://logs.openstack.org/68/666068/1/check/kolla-ansible-oraclelinux-source-ceph/3eb53ad/ : SUCCESS in 1h 38m 51s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/68/666068/1/check/openstack-tox-lower-constraints/2bac60d/ : SUCCESS in 4m 30s\n- openstack-tox-pep8 http://logs.openstack.org/68/666068/1/check/openstack-tox-pep8/92e966c/ : SUCCESS in 2m 56s\n- openstack-tox-py27 http://logs.openstack.org/68/666068/1/check/openstack-tox-py27/50889d1/ : SUCCESS in 3m 03s\n- openstack-tox-py36 http://logs.openstack.org/68/666068/1/check/openstack-tox-py36/e2ed044/ : SUCCESS in 3m 20s\n- openstack-tox-py37 http://logs.openstack.org/68/666068/1/check/openstack-tox-py37/3becfb9/ : SUCCESS in 3m 52s\n- openstack-tox-docs http://logs.openstack.org/68/666068/1/check/openstack-tox-docs/f92470d/html/ : SUCCESS in 3m 13s\n- kolla-build-ubuntu-binary http://logs.openstack.org/68/666068/1/check/kolla-build-ubuntu-binary/102a7c7/ : SUCCESS in 23m 09s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/68/666068/1/check/kolla-build-ubuntu-source/24cac3e/ : SUCCESS in 25m 58s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/68/666068/1/check/kolla-ansible-ubuntu-source/2f2b557/ : SUCCESS in 48m 24s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/68/666068/1/check/kolla-ansible-ubuntu-binary/77845e7/ : SUCCESS in 50m 08s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/68/666068/1/check/kolla-ansible-ubuntu-source-upgrade/106e70b/ : SUCCESS in 57m 24s\n- kolla-ansible-ubuntu-source-ceph http://logs.openstack.org/68/666068/1/check/kolla-ansible-ubuntu-source-ceph/f75fedc/ : SUCCESS in 1h 14m 39s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph http://logs.openstack.org/68/666068/1/check/kolla-ansible-ubuntu-source-upgrade-ceph/256da2b/ : SUCCESS in 1h 23m 25s (non-voting)\n- kolla-ansible-ubuntu-source-cinder-lvm http://logs.openstack.org/68/666068/1/check/kolla-ansible-ubuntu-source-cinder-lvm/c74911c/ : SUCCESS in 49m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"42bb7d5e96b092f48e67f0e7df42521e7c9f2c89","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-19 16:05:11.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"844e1095f7425e471e479c408eaa2fc10c43624b","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-19 16:09:11.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"fe6561b21d3dda940f0905762bae018252b33369","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-19 16:10:57.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"f5bcc83f6146c599febaaa5f7f23f7088331133b","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-19 16:12:45.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"8f104e2da15ffa38330ab647959f026138ed2d8e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-19 17:49:07.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/68/666068/3/check/kolla-build-centos-binary/41a3530/ : SUCCESS in 47m 06s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/68/666068/3/check/kolla-build-centos-source/0199ba1/ : SUCCESS in 47m 14s\n- kolla-ansible-centos-source http://logs.openstack.org/68/666068/3/check/kolla-ansible-centos-source/736cc8c/ : SUCCESS in 1h 10m 38s\n- kolla-ansible-centos-binary http://logs.openstack.org/68/666068/3/check/kolla-ansible-centos-binary/b8552ec/ : SUCCESS in 1h 07m 09s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/68/666068/3/check/kolla-ansible-centos-source-upgrade/6efe33b/ : SUCCESS in 1h 19m 02s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/68/666068/3/check/tripleo-build-containers-centos-7/7bab8bd/ : FAILURE in 17m 12s\n- openstack-tox-lower-constraints http://logs.openstack.org/68/666068/3/check/openstack-tox-lower-constraints/819e3bf/ : SUCCESS in 3m 33s\n- openstack-tox-pep8 http://logs.openstack.org/68/666068/3/check/openstack-tox-pep8/724fd09/ : SUCCESS in 3m 32s\n- openstack-tox-py27 http://logs.openstack.org/68/666068/3/check/openstack-tox-py27/7250b7f/ : SUCCESS in 4m 05s\n- openstack-tox-py36 http://logs.openstack.org/68/666068/3/check/openstack-tox-py36/9596ab3/ : SUCCESS in 3m 22s\n- openstack-tox-py37 http://logs.openstack.org/68/666068/3/check/openstack-tox-py37/e946ab6/ : SUCCESS in 3m 36s\n- openstack-tox-docs http://logs.openstack.org/68/666068/3/check/openstack-tox-docs/88fc09e/html/ : SUCCESS in 3m 35s\n- kolla-build-ubuntu-binary http://logs.openstack.org/68/666068/3/check/kolla-build-ubuntu-binary/a1b12d8/ : FAILURE in 24m 11s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/68/666068/3/check/kolla-build-ubuntu-source/3ec4a74/ : SUCCESS in 33m 53s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/68/666068/3/check/kolla-ansible-ubuntu-source/d54243e/ : SUCCESS in 44m 13s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/68/666068/3/check/kolla-ansible-ubuntu-binary/f7e919b/ : SUCCESS in 48m 36s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/68/666068/3/check/kolla-ansible-ubuntu-source-upgrade/2e11c4b/ : SUCCESS in 55m 17s","accounts_in_message":[],"_revision_number":3},{"id":"ca0741321d24e32d33af77977b491b262946e3b7","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-19 18:04:37.000000000","message":"Patch Set 3:\n\nrecheck\n\ntripleo-build-* failed due to:\nCannot find a valid baseurl for repo: centos-nfs-ganesha28/7/x86_64\n\nkolla-build-ubuntu-binary e.g. due to:\ndesignate-common : Depends: sqlite3 but it is not going to be installed\n(looks like mirror issue?)","accounts_in_message":[],"_revision_number":3},{"id":"71be20bed15d801a89cc330e12002a514d0a685f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-19 19:46:35.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/68/666068/3/check/kolla-build-centos-binary/e7037ae/ : SUCCESS in 40m 58s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/68/666068/3/check/kolla-build-centos-source/adc82e3/ : SUCCESS in 51m 58s\n- kolla-ansible-centos-source http://logs.openstack.org/68/666068/3/check/kolla-ansible-centos-source/5ad74d1/ : SUCCESS in 1h 11m 15s\n- kolla-ansible-centos-binary http://logs.openstack.org/68/666068/3/check/kolla-ansible-centos-binary/47fdade/ : SUCCESS in 1h 07m 34s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/68/666068/3/check/kolla-ansible-centos-source-upgrade/4c480ef/ : SUCCESS in 1h 17m 21s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/68/666068/3/check/tripleo-build-containers-centos-7/872b2a0/ : FAILURE in 18m 02s\n- openstack-tox-lower-constraints http://logs.openstack.org/68/666068/3/check/openstack-tox-lower-constraints/e54d5b5/ : SUCCESS in 3m 26s\n- openstack-tox-pep8 http://logs.openstack.org/68/666068/3/check/openstack-tox-pep8/f8313a4/ : SUCCESS in 3m 14s\n- openstack-tox-py27 http://logs.openstack.org/68/666068/3/check/openstack-tox-py27/4522566/ : SUCCESS in 3m 48s\n- openstack-tox-py36 http://logs.openstack.org/68/666068/3/check/openstack-tox-py36/da7cfb7/ : SUCCESS in 3m 33s\n- openstack-tox-py37 http://logs.openstack.org/68/666068/3/check/openstack-tox-py37/77b85b9/ : SUCCESS in 4m 03s\n- openstack-tox-docs http://logs.openstack.org/68/666068/3/check/openstack-tox-docs/d2fb47a/html/ : SUCCESS in 3m 19s\n- kolla-build-ubuntu-binary http://logs.openstack.org/68/666068/3/check/kolla-build-ubuntu-binary/4c1a291/ : FAILURE in 26m 03s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/68/666068/3/check/kolla-build-ubuntu-source/8e83898/ : FAILURE in 33m 52s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/68/666068/3/check/kolla-ansible-ubuntu-source/05b44f8/ : SUCCESS in 49m 31s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/68/666068/3/check/kolla-ansible-ubuntu-binary/9d8d83f/ : SUCCESS in 46m 42s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/68/666068/3/check/kolla-ansible-ubuntu-source-upgrade/5eeb46d/ : SUCCESS in 54m 51s","accounts_in_message":[],"_revision_number":3},{"id":"d4f214b45a8323b5d327de45072f1b767d0e5b58","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-20 11:29:59.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"19ac64e31748c99410b16a49b42cb9fe9415e69a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-20 13:02:06.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/68/666068/3/check/kolla-build-centos-binary/19b9f2f/ : SUCCESS in 26m 00s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/68/666068/3/check/kolla-build-centos-source/9f72162/ : SUCCESS in 42m 39s\n- kolla-ansible-centos-source http://logs.openstack.org/68/666068/3/check/kolla-ansible-centos-source/3462c09/ : SUCCESS in 1h 11m 36s\n- kolla-ansible-centos-binary http://logs.openstack.org/68/666068/3/check/kolla-ansible-centos-binary/8f01369/ : SUCCESS in 1h 02m 20s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/68/666068/3/check/kolla-ansible-centos-source-upgrade/045c2dc/ : SUCCESS in 1h 16m 42s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/68/666068/3/check/tripleo-build-containers-centos-7/52014b5/ : SUCCESS in 45m 47s\n- openstack-tox-lower-constraints http://logs.openstack.org/68/666068/3/check/openstack-tox-lower-constraints/d9ab329/ : SUCCESS in 3m 38s\n- openstack-tox-pep8 http://logs.openstack.org/68/666068/3/check/openstack-tox-pep8/fc594f3/ : SUCCESS in 3m 35s\n- openstack-tox-py27 http://logs.openstack.org/68/666068/3/check/openstack-tox-py27/2fd8cce/ : SUCCESS in 3m 44s\n- openstack-tox-py36 http://logs.openstack.org/68/666068/3/check/openstack-tox-py36/3d37020/ : SUCCESS in 3m 17s\n- openstack-tox-py37 http://logs.openstack.org/68/666068/3/check/openstack-tox-py37/20a6ba3/ : SUCCESS in 3m 57s\n- openstack-tox-docs http://logs.openstack.org/68/666068/3/check/openstack-tox-docs/c908e9a/html/ : SUCCESS in 3m 18s\n- kolla-build-ubuntu-binary http://logs.openstack.org/68/666068/3/check/kolla-build-ubuntu-binary/3abfae3/ : SUCCESS in 28m 29s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/68/666068/3/check/kolla-build-ubuntu-source/3f87b46/ : SUCCESS in 41m 26s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/68/666068/3/check/kolla-ansible-ubuntu-source/2019c3a/ : SUCCESS in 48m 05s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/68/666068/3/check/kolla-ansible-ubuntu-binary/b2aec40/ : SUCCESS in 51m 47s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/68/666068/3/check/kolla-ansible-ubuntu-source-upgrade/28b05da/ : SUCCESS in 1h 09m 43s","accounts_in_message":[],"_revision_number":3},{"id":"5342774b2c942be4f5520b79113cc6c578cd30cc","author":{"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},"date":"2019-06-20 13:12:33.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"5a6602d955ce25e2aaee53506eaa83ec16bd6f48","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-27 09:19:50.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"a76952d9b2a4b419bfaa236dcb34b9b432a2db5a","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-27 09:50:03.000000000","message":"Patch Set 4:\n\n(1 comment)\n\nBase and openstack-base yeah, nice catch.","accounts_in_message":[],"_revision_number":4},{"id":"398306a2caab4e2b2084902e84b6913bd47f5772","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-27 10:10:56.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"661727b6e82acaadad64bbceb18713498199d26b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-27 10:11:16.000000000","message":"Patch Set 6: Commit message was updated.","accounts_in_message":[],"_revision_number":6},{"id":"0ffe27079fc44301a3736ecb3bac41a90131031c","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-27 10:23:44.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"826bdddb344bf0f32aa73b7c10876b3030dbefe2","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-27 10:26:23.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"ad0705327a3cafcefcb1e5552262d78b7a9f1ab3","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-27 10:27:54.000000000","message":"Patch Set 6: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"f00cf01db5aa7960693bd9d18e3d4641d9e17beb","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-27 10:37:33.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"1e733720cb47ccc2c2cb95133af90f3a47d4f8f1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 13:45:10.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/68/666068/7/check/kolla-build-centos-binary/dd5f47c/ : SUCCESS in 45m 42s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/68/666068/7/check/kolla-build-centos-source/3478446/ : SUCCESS in 48m 36s\n- kolla-ansible-centos-source http://logs.openstack.org/68/666068/7/check/kolla-ansible-centos-source/e48adf0/ : SUCCESS in 1h 07m 50s\n- kolla-ansible-centos-binary http://logs.openstack.org/68/666068/7/check/kolla-ansible-centos-binary/8afd52a/ : SUCCESS in 54m 04s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/68/666068/7/check/kolla-ansible-centos-source-upgrade/572396d/ : SUCCESS in 1h 05m 37s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/68/666068/7/check/tripleo-build-containers-centos-7/28b3683/ : SUCCESS in 48m 36s\n- openstack-tox-lower-constraints http://logs.openstack.org/68/666068/7/check/openstack-tox-lower-constraints/cb5456a/ : SUCCESS in 3m 23s\n- openstack-tox-pep8 http://logs.openstack.org/68/666068/7/check/openstack-tox-pep8/8284224/ : SUCCESS in 4m 28s\n- openstack-tox-py27 http://logs.openstack.org/68/666068/7/check/openstack-tox-py27/2510397/ : SUCCESS in 3m 47s\n- openstack-tox-py36 http://logs.openstack.org/68/666068/7/check/openstack-tox-py36/2a9d761/ : SUCCESS in 3m 59s\n- openstack-tox-py37 http://logs.openstack.org/68/666068/7/check/openstack-tox-py37/b50965e/ : SUCCESS in 3m 37s\n- openstack-tox-docs http://logs.openstack.org/68/666068/7/check/openstack-tox-docs/c0cd15e/html/ : SUCCESS in 2m 49s\n- kolla-build-ubuntu-binary http://logs.openstack.org/68/666068/7/check/kolla-build-ubuntu-binary/25af815/ : SUCCESS in 23m 10s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/68/666068/7/check/kolla-build-ubuntu-source/856d99f/ : FAILURE in 32m 43s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/68/666068/7/check/kolla-ansible-ubuntu-source/e598afd/ : SUCCESS in 44m 38s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/68/666068/7/check/kolla-ansible-ubuntu-binary/44923b2/ : SUCCESS in 43m 36s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/68/666068/7/check/kolla-ansible-ubuntu-source-upgrade/b1197ec/ : SUCCESS in 51m 31s","accounts_in_message":[],"_revision_number":7},{"id":"f82189e54a99688403ba352a1e87930a02e16ba8","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-27 13:50:43.000000000","message":"Patch Set 7:\n\nrecheck\n\nmirror","accounts_in_message":[],"_revision_number":7},{"id":"e7a4b2d92f91d8c07ec1efa416d6dcd5afabb723","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 16:27:17.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/68/666068/7/check/kolla-build-centos-binary/2309113/ : SUCCESS in 27m 02s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/68/666068/7/check/kolla-build-centos-source/a9dafba/ : SUCCESS in 51m 49s\n- kolla-ansible-centos-source http://logs.openstack.org/68/666068/7/check/kolla-ansible-centos-source/d056395/ : SUCCESS in 1h 10m 02s\n- kolla-ansible-centos-binary http://logs.openstack.org/68/666068/7/check/kolla-ansible-centos-binary/0ae33fa/ : SUCCESS in 1h 03m 07s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/68/666068/7/check/kolla-ansible-centos-source-upgrade/7449429/ : SUCCESS in 1h 18m 52s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/68/666068/7/check/tripleo-build-containers-centos-7/6821060/ : SUCCESS in 54m 06s\n- openstack-tox-lower-constraints http://logs.openstack.org/68/666068/7/check/openstack-tox-lower-constraints/26cdfa2/ : SUCCESS in 4m 13s\n- openstack-tox-pep8 http://logs.openstack.org/68/666068/7/check/openstack-tox-pep8/c2aaca0/ : SUCCESS in 3m 51s\n- openstack-tox-py27 http://logs.openstack.org/68/666068/7/check/openstack-tox-py27/2d9d4fc/ : SUCCESS in 2m 57s\n- openstack-tox-py36 http://logs.openstack.org/68/666068/7/check/openstack-tox-py36/710c739/ : SUCCESS in 2m 55s\n- openstack-tox-py37 http://logs.openstack.org/68/666068/7/check/openstack-tox-py37/4f93d9d/ : SUCCESS in 3m 09s\n- openstack-tox-docs http://logs.openstack.org/68/666068/7/check/openstack-tox-docs/80f8e7f/html/ : SUCCESS in 2m 33s\n- kolla-build-ubuntu-binary http://logs.openstack.org/68/666068/7/check/kolla-build-ubuntu-binary/82bdb3b/ : SUCCESS in 22m 00s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/68/666068/7/check/kolla-build-ubuntu-source/7d573f8/ : FAILURE in 17m 24s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/68/666068/7/check/kolla-ansible-ubuntu-source/eb33310/ : SUCCESS in 49m 10s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/68/666068/7/check/kolla-ansible-ubuntu-binary/b6e786b/ : SUCCESS in 45m 22s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/68/666068/7/check/kolla-ansible-ubuntu-source-upgrade/09bd0eb/ : SUCCESS in 53m 14s","accounts_in_message":[],"_revision_number":7},{"id":"f8dab43cc92a4954ada3dccef0e5c13134efe0ee","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-27 16:34:50.000000000","message":"Patch Set 7:\n\nrecheck\n\nmirror again","accounts_in_message":[],"_revision_number":7},{"id":"076735d54575f89664775ddeaeb9e244f60ef54d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 18:19:39.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/68/666068/7/check/kolla-build-centos-binary/b40bb58/ : SUCCESS in 38m 25s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/68/666068/7/check/kolla-build-centos-source/3bbd4cd/ : SUCCESS in 37m 09s\n- kolla-ansible-centos-source http://logs.openstack.org/68/666068/7/check/kolla-ansible-centos-source/f687a64/ : SUCCESS in 1h 09m 37s\n- kolla-ansible-centos-binary http://logs.openstack.org/68/666068/7/check/kolla-ansible-centos-binary/6241c69/ : SUCCESS in 1h 06m 57s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/68/666068/7/check/kolla-ansible-centos-source-upgrade/b967c19/ : SUCCESS in 1h 16m 14s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/68/666068/7/check/tripleo-build-containers-centos-7/f597e04/ : SUCCESS in 44m 07s\n- openstack-tox-lower-constraints http://logs.openstack.org/68/666068/7/check/openstack-tox-lower-constraints/edd24cc/ : SUCCESS in 4m 38s\n- openstack-tox-pep8 http://logs.openstack.org/68/666068/7/check/openstack-tox-pep8/a417d7d/ : SUCCESS in 3m 29s\n- openstack-tox-py27 http://logs.openstack.org/68/666068/7/check/openstack-tox-py27/8cc66bf/ : SUCCESS in 2m 53s\n- openstack-tox-py36 http://logs.openstack.org/68/666068/7/check/openstack-tox-py36/a166484/ : SUCCESS in 3m 41s\n- openstack-tox-py37 http://logs.openstack.org/68/666068/7/check/openstack-tox-py37/96ba9c1/ : SUCCESS in 3m 35s\n- openstack-tox-docs http://logs.openstack.org/68/666068/7/check/openstack-tox-docs/5b2c38b/html/ : SUCCESS in 3m 41s\n- kolla-build-ubuntu-binary http://logs.openstack.org/68/666068/7/check/kolla-build-ubuntu-binary/6e4ed7a/ : SUCCESS in 23m 11s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/68/666068/7/check/kolla-build-ubuntu-source/01216b5/ : SUCCESS in 27m 55s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/68/666068/7/check/kolla-ansible-ubuntu-source/689eb6c/ : SUCCESS in 54m 22s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/68/666068/7/check/kolla-ansible-ubuntu-binary/262797d/ : SUCCESS in 46m 16s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/68/666068/7/check/kolla-ansible-ubuntu-source-upgrade/15fb55d/ : SUCCESS in 54m 44s","accounts_in_message":[],"_revision_number":7},{"id":"d63e512f52a86e8142884d00bd687635319f5b43","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-28 06:38:01.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"d217ec50fb038d8e2dd85a526cc53896e9936959","author":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"date":"2019-06-28 06:45:32.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"3b0655f27b4c0a0c62d0aefc6cf9c93e7a328f43","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-07-03 06:10:50.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"b028abffb3ccb4ea2312fc712dc756ceea63b262","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-03 06:11:11.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"a6bb0579905aabc46f8eddc6d33193dd9db31116","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-03 07:38:00.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7},{"id":"4d3cf77d72e6d0da97abff79d1b12baeb1ffaf18","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-03 07:38:00.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source http://logs.openstack.org/68/666068/7/gate/kolla-build-centos-source/f420c34/ : SUCCESS in 1h 00m 15s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/68/666068/7/gate/tripleo-build-containers-centos-7/2f9bc7c/ : SUCCESS in 53m 02s\n- openstack-tox-lower-constraints http://logs.openstack.org/68/666068/7/gate/openstack-tox-lower-constraints/544b0c6/ : SUCCESS in 4m 33s\n- openstack-tox-pep8 http://logs.openstack.org/68/666068/7/gate/openstack-tox-pep8/8ee650e/ : SUCCESS in 4m 09s\n- openstack-tox-py27 http://logs.openstack.org/68/666068/7/gate/openstack-tox-py27/4f2ccf8/ : SUCCESS in 3m 33s\n- openstack-tox-py36 http://logs.openstack.org/68/666068/7/gate/openstack-tox-py36/7c0aff1/ : SUCCESS in 3m 15s\n- openstack-tox-py37 http://logs.openstack.org/68/666068/7/gate/openstack-tox-py37/0d360f8/ : SUCCESS in 3m 31s\n- openstack-tox-docs http://logs.openstack.org/68/666068/7/gate/openstack-tox-docs/ef1c946/html/ : SUCCESS in 2m 56s\n- kolla-build-ubuntu-source http://logs.openstack.org/68/666068/7/gate/kolla-build-ubuntu-source/bc285c1/ : SUCCESS in 31m 59s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"0a4649aadd2166ca0819a5b427fd2e2b533bdedc","revisions":{"fce1a8632fc3a0def48af444ed491d4864c86739":{"kind":"REWORK","_number":1,"created":"2019-06-18 17:29:09.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/68/666068/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/68/666068/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/68/666068/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/68/666068/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/68/666068/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/68/666068/1"}}},"commit":{"parents":[{"commit":"45ef7176aead8d8b59135f8c168094600ea05f31","subject":"Merge \"[neutron] Do not migrate all subprojects\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/45ef7176aead8d8b59135f8c168094600ea05f31"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-06-18 17:28:27.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-06-18 17:28:27.000000000","tz":60},"subject":"Enable CI jobs for ceph upgrade, NFV, cinder LVM","message":"Enable CI jobs for ceph upgrade, NFV, cinder LVM\n\nChange-Id: I410cb943b766284a6b7bf11b0dd53bd2cdc75749\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fce1a8632fc3a0def48af444ed491d4864c86739"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fce1a8632fc3a0def48af444ed491d4864c86739"}]},"branch":"refs/heads/master"},"8fbbc0a2e641c2c92d1ef434cb4d91f10cf215f3":{"kind":"REWORK","_number":2,"created":"2019-06-19 16:05:11.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/68/666068/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/68/666068/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/68/666068/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/68/666068/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/68/666068/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/68/666068/2"}}},"commit":{"parents":[{"commit":"45ef7176aead8d8b59135f8c168094600ea05f31","subject":"Merge \"[neutron] Do not migrate all subprojects\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/45ef7176aead8d8b59135f8c168094600ea05f31"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-06-18 17:28:27.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-06-19 16:04:46.000000000","tz":60},"subject":"Enable CI jobs for ceph upgrade, NFV, cinder LVM","message":"Enable CI jobs for ceph upgrade, NFV, cinder LVM\n\nChange-Id: I410cb943b766284a6b7bf11b0dd53bd2cdc75749\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8fbbc0a2e641c2c92d1ef434cb4d91f10cf215f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8fbbc0a2e641c2c92d1ef434cb4d91f10cf215f3"}]},"branch":"refs/heads/master"},"e5659b2a1ac4c51cad45c15871132fdfcf0bf15e":{"kind":"REWORK","_number":3,"created":"2019-06-19 16:10:57.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/68/666068/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/68/666068/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/68/666068/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/68/666068/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/68/666068/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/68/666068/3"}}},"commit":{"parents":[{"commit":"45ef7176aead8d8b59135f8c168094600ea05f31","subject":"Merge \"[neutron] Do not migrate all subprojects\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/45ef7176aead8d8b59135f8c168094600ea05f31"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-06-18 17:28:27.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-06-19 16:10:54.000000000","tz":0},"subject":"Enable CI jobs for ceph upgrade, NFV, cinder LVM","message":"Enable CI jobs for ceph upgrade, NFV, cinder LVM\n\nChange-Id: I410cb943b766284a6b7bf11b0dd53bd2cdc75749\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e5659b2a1ac4c51cad45c15871132fdfcf0bf15e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e5659b2a1ac4c51cad45c15871132fdfcf0bf15e"}]},"branch":"refs/heads/master"},"75a7352cb06c7b5be7bbff382ec0ecf4e44fd247":{"kind":"REWORK","_number":4,"created":"2019-06-27 09:19:50.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/68/666068/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/68/666068/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/68/666068/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/68/666068/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/68/666068/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/68/666068/4"}}},"commit":{"parents":[{"commit":"45ef7176aead8d8b59135f8c168094600ea05f31","subject":"Merge \"[neutron] Do not migrate all subprojects\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/45ef7176aead8d8b59135f8c168094600ea05f31"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-06-18 17:28:27.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-06-27 09:19:21.000000000","tz":0},"subject":"Enable CI jobs for ceph upgrade, NFV, cinder LVM","message":"Enable CI jobs for ceph upgrade, NFV, cinder LVM\n\nChange-Id: I410cb943b766284a6b7bf11b0dd53bd2cdc75749\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/75a7352cb06c7b5be7bbff382ec0ecf4e44fd247"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/75a7352cb06c7b5be7bbff382ec0ecf4e44fd247"}]},"branch":"refs/heads/master"},"f9961a1840c1c89eeacfb04fe2b2fd9de70550d3":{"kind":"REWORK","_number":5,"created":"2019-06-27 10:10:56.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/68/666068/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/68/666068/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/68/666068/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/68/666068/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/68/666068/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/68/666068/5"}}},"commit":{"parents":[{"commit":"45ef7176aead8d8b59135f8c168094600ea05f31","subject":"Merge \"[neutron] Do not migrate all subprojects\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/45ef7176aead8d8b59135f8c168094600ea05f31"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-06-18 17:28:27.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-06-27 10:10:53.000000000","tz":0},"subject":"Enable CI jobs for ceph upgrade, NFV, cinder LVM","message":"Enable CI jobs for ceph upgrade, NFV, cinder LVM\n\nChange-Id: I410cb943b766284a6b7bf11b0dd53bd2cdc75749\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f9961a1840c1c89eeacfb04fe2b2fd9de70550d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f9961a1840c1c89eeacfb04fe2b2fd9de70550d3"}]},"branch":"refs/heads/master"},"84449495cb75c947b5cbf52a7ad544529ad457d4":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2019-06-27 10:11:16.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/68/666068/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/68/666068/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/68/666068/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/68/666068/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/68/666068/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/68/666068/6"}}},"commit":{"parents":[{"commit":"45ef7176aead8d8b59135f8c168094600ea05f31","subject":"Merge \"[neutron] Do not migrate all subprojects\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/45ef7176aead8d8b59135f8c168094600ea05f31"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-06-18 17:28:27.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-06-27 10:11:12.000000000","tz":0},"subject":"Enable CI jobs for ceph upgrade, NFV, cinder LVM, ironic","message":"Enable CI jobs for ceph upgrade, NFV, cinder LVM, ironic\n\nChange-Id: I410cb943b766284a6b7bf11b0dd53bd2cdc75749\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/84449495cb75c947b5cbf52a7ad544529ad457d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/84449495cb75c947b5cbf52a7ad544529ad457d4"}]},"branch":"refs/heads/master"},"0a4649aadd2166ca0819a5b427fd2e2b533bdedc":{"kind":"REWORK","_number":7,"created":"2019-06-27 10:37:33.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/68/666068/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/68/666068/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/68/666068/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/68/666068/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/68/666068/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/68/666068/7"}}},"commit":{"parents":[{"commit":"45ef7176aead8d8b59135f8c168094600ea05f31","subject":"Merge \"[neutron] Do not migrate all subprojects\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/45ef7176aead8d8b59135f8c168094600ea05f31"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-06-18 17:28:27.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-06-27 10:37:19.000000000","tz":0},"subject":"Enable CI jobs for ceph upgrade, NFV, cinder LVM, ironic","message":"Enable CI jobs for ceph upgrade, NFV, cinder LVM, ironic\n\nChange-Id: I410cb943b766284a6b7bf11b0dd53bd2cdc75749\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0a4649aadd2166ca0819a5b427fd2e2b533bdedc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0a4649aadd2166ca0819a5b427fd2e2b533bdedc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
