)]}'
{"id":"openstack%2Fkolla~666077","triplet_id":"openstack%2Fkolla~stable%2Fstein~I19447727f19dbd9c0a3e17d218b48ddc4c253587","project":"openstack/kolla","branch":"stable/stein","topic":"bug/1832026-stable/stein","hashtags":[],"change_id":"I19447727f19dbd9c0a3e17d218b48ddc4c253587","subject":"Fix ownership of ironic inspector DHCP hostsdir","status":"MERGED","created":"2019-06-18 17:43:10.000000000","updated":"2019-06-27 14:36:01.000000000","submitted":"2019-06-27 14:36:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"666077-1561646161370-90acf6b2","meta_rev_id":"18b73d208971d8e30435a26cf5aba8e7cc7aaa7c","_number":666077,"virtual_id_number":666077,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-06-27 14:36:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"},{"value":0,"date":"2019-06-27 11:11:22.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"recommended":{"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},"all":[{"value":1,"date":"2019-06-20 13:24:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},{"value":2,"date":"2019-06-19 20:40:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2019-06-24 10:11:33.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-06-19 22:45:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2019-06-24 10:11:33.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2019-06-20 14:09:47.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-19 20:40:19.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2019-06-19 22:45:27.000000000","updated_by":{"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"},"reviewer":{"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"},"state":"REVIEWER"},{"updated":"2019-06-20 13:24:34.000000000","updated_by":{"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},"reviewer":{"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},"state":"REVIEWER"},{"updated":"2019-06-24 10:11:33.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-06-27 11:11:22.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-06-27 14:36:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f751f332ba39546d12a6e885de148a8bb1eb0adb","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-18 17:43:10.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"86c39d1c0a39d9c0d1742753bdb03d2a6d5f5ef3","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-18 17:49:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/77/666077/1/openstack-check/tripleo-build-containers-fedora-28-master/8c400f0/ : FAILURE in 5m 05s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/77/666077/1/openstack-check/tripleo-build-containers-fedora-28-stein/5dbb238/ : FAILURE in 4m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"95720befa5559e9c59603cbb56a82769244eb906","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 20:19:14.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/77/666077/1/check/kolla-build-centos-binary/fb5146c/ : SUCCESS in 42m 11s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/77/666077/1/check/kolla-build-centos-source/8562ac3/ : SUCCESS in 43m 43s\n- kolla-ansible-centos-source http://logs.openstack.org/77/666077/1/check/kolla-ansible-centos-source/efd7150/ : SUCCESS in 1h 19m 47s\n- kolla-ansible-centos-binary http://logs.openstack.org/77/666077/1/check/kolla-ansible-centos-binary/a9c1713/ : RETRY_LIMIT in 2m 10s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/77/666077/1/check/kolla-ansible-centos-source-upgrade/627cb64/ : RETRY_LIMIT in 1m 59s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/77/666077/1/check/tripleo-build-containers-centos-7/8b809ca/ : SUCCESS in 37m 43s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/77/666077/1/check/kolla-build-oraclelinux-binary/9662d80/ : SUCCESS in 52m 39s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/77/666077/1/check/kolla-build-oraclelinux-source/63a1d1a/ : SUCCESS in 44m 38s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/77/666077/1/check/kolla-ansible-oraclelinux-source/854dcad/ : SUCCESS in 1h 13m 15s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/77/666077/1/check/kolla-ansible-oraclelinux-binary/0892dfc/ : SUCCESS in 1h 05m 16s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/77/666077/1/check/openstack-tox-lower-constraints/55d0d5a/ : SUCCESS in 4m 05s\n- openstack-tox-pep8 http://logs.openstack.org/77/666077/1/check/openstack-tox-pep8/282d2f6/ : SUCCESS in 3m 44s\n- openstack-tox-py27 http://logs.openstack.org/77/666077/1/check/openstack-tox-py27/f74cd98/ : SUCCESS in 2m 50s\n- openstack-tox-py36 http://logs.openstack.org/77/666077/1/check/openstack-tox-py36/0d6bae7/ : SUCCESS in 2m 59s\n- openstack-tox-py37 http://logs.openstack.org/77/666077/1/check/openstack-tox-py37/e07d7df/ : SUCCESS in 3m 22s\n- openstack-tox-docs http://logs.openstack.org/77/666077/1/check/openstack-tox-docs/9ee2d84/html/ : SUCCESS in 4m 05s\n- kolla-build-ubuntu-binary http://logs.openstack.org/77/666077/1/check/kolla-build-ubuntu-binary/cf39725/ : SUCCESS in 22m 51s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/77/666077/1/check/kolla-build-ubuntu-source/fcad523/ : SUCCESS in 27m 02s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/77/666077/1/check/kolla-ansible-ubuntu-source/ed52243/ : SUCCESS in 1h 28m 15s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/77/666077/1/check/kolla-ansible-ubuntu-binary/eb8f422/ : SUCCESS in 44m 42s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/77/666077/1/check/kolla-ansible-ubuntu-source-upgrade/9be2c83/ : SUCCESS in 1h 07m 02s","accounts_in_message":[],"_revision_number":1},{"id":"57d8c5c0bdc8b1f14d02babb04d114797c03fb0f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-19 11:49:28.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"74f9f9e39dca3f87261efbf73257e71fdaccfd11","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-19 12:24:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/77/666077/1/openstack-check/tripleo-build-containers-fedora-28-master/337d243/ : FAILURE in 5m 45s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/77/666077/1/openstack-check/tripleo-build-containers-fedora-28-stein/a70c9f1/ : SUCCESS in 33m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4c3aab10914bcc5b02ac0f366be5501cdd626d7d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-19 13:45:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/77/666077/1/check/kolla-build-centos-binary/440a86c/ : SUCCESS in 48m 00s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/77/666077/1/check/kolla-build-centos-source/3246404/ : SUCCESS in 40m 53s\n- kolla-ansible-centos-source http://logs.openstack.org/77/666077/1/check/kolla-ansible-centos-source/730af7c/ : SUCCESS in 1h 12m 15s\n- kolla-ansible-centos-binary http://logs.openstack.org/77/666077/1/check/kolla-ansible-centos-binary/0bb6b0e/ : SUCCESS in 1h 14m 31s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/77/666077/1/check/kolla-ansible-centos-source-upgrade/e25b81c/ : SUCCESS in 1h 12m 40s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/77/666077/1/check/tripleo-build-containers-centos-7/2775316/ : SUCCESS in 50m 24s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/77/666077/1/check/kolla-build-oraclelinux-binary/6799acc/ : SUCCESS in 53m 27s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/77/666077/1/check/kolla-build-oraclelinux-source/894cc81/ : SUCCESS in 42m 52s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/77/666077/1/check/kolla-ansible-oraclelinux-source/d07b54b/ : SUCCESS in 1h 16m 17s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/77/666077/1/check/kolla-ansible-oraclelinux-binary/a8c8d0a/ : SUCCESS in 1h 09m 23s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/77/666077/1/check/openstack-tox-lower-constraints/e18e0ec/ : SUCCESS in 3m 34s\n- openstack-tox-pep8 http://logs.openstack.org/77/666077/1/check/openstack-tox-pep8/65ec214/ : SUCCESS in 3m 12s\n- openstack-tox-py27 http://logs.openstack.org/77/666077/1/check/openstack-tox-py27/0a65c8e/ : SUCCESS in 4m 40s\n- openstack-tox-py36 http://logs.openstack.org/77/666077/1/check/openstack-tox-py36/ef010c7/ : SUCCESS in 2m 50s\n- openstack-tox-py37 http://logs.openstack.org/77/666077/1/check/openstack-tox-py37/e14b94f/ : SUCCESS in 4m 02s\n- openstack-tox-docs http://logs.openstack.org/77/666077/1/check/openstack-tox-docs/2371c1f/html/ : SUCCESS in 3m 13s\n- kolla-build-ubuntu-binary http://logs.openstack.org/77/666077/1/check/kolla-build-ubuntu-binary/6d3679c/ : SUCCESS in 24m 09s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/77/666077/1/check/kolla-build-ubuntu-source/ee15037/ : SUCCESS in 36m 06s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/77/666077/1/check/kolla-ansible-ubuntu-source/7add1cb/ : SUCCESS in 59m 41s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/77/666077/1/check/kolla-ansible-ubuntu-binary/a5fc576/ : SUCCESS in 46m 41s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/77/666077/1/check/kolla-ansible-ubuntu-source-upgrade/d200c59/ : SUCCESS in 1h 00m 33s","accounts_in_message":[],"_revision_number":1},{"id":"20b8eab5a4468d89892ee0e073263c1a592b3c37","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-06-19 20:40:19.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"384ad472d50aa27930355a5185408b6c36d8ed16","author":{"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"},"date":"2019-06-19 22:45:27.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"e356104d3250df4c5d82b054682d32a0a4d3d231","author":{"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},"date":"2019-06-20 13:24:34.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"9277550e6196d6470c0f9327f857e18f7bb79c5d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-20 14:09:47.000000000","message":"Patch Set 1: Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"1464f1f560254117b4389f813a231e3e246aabc3","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-24 10:11:33.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"62d113f2989381f318fe97754d84f7d15e56b0c8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-24 10:11:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4ec37f3af88ac3ecd9656219ea9631323378ae70","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-24 15:26:14.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-source http://logs.openstack.org/77/666077/1/gate/kolla-build-centos-source/ad6ccc8/ : FAILURE in 55m 11s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/77/666077/1/gate/tripleo-build-containers-centos-7/27058d4/ : SUCCESS in 45m 45s\n- kolla-build-oraclelinux-source http://logs.openstack.org/77/666077/1/gate/kolla-build-oraclelinux-source/2b00702/ : SUCCESS in 52m 28s\n- openstack-tox-lower-constraints http://logs.openstack.org/77/666077/1/gate/openstack-tox-lower-constraints/2f51528/ : SUCCESS in 3m 39s\n- openstack-tox-pep8 http://logs.openstack.org/77/666077/1/gate/openstack-tox-pep8/902f86f/ : SUCCESS in 3m 37s\n- openstack-tox-py27 http://logs.openstack.org/77/666077/1/gate/openstack-tox-py27/39ae5e1/ : SUCCESS in 3m 25s\n- openstack-tox-py36 http://logs.openstack.org/77/666077/1/gate/openstack-tox-py36/3dd3f06/ : SUCCESS in 3m 47s\n- openstack-tox-py37 http://logs.openstack.org/77/666077/1/gate/openstack-tox-py37/760930c/ : SUCCESS in 4m 04s\n- openstack-tox-docs http://logs.openstack.org/77/666077/1/gate/openstack-tox-docs/71ddce0/html/ : SUCCESS in 3m 01s\n- kolla-build-ubuntu-source http://logs.openstack.org/77/666077/1/gate/kolla-build-ubuntu-source/80c7f05/ : SUCCESS in 33m 06s","accounts_in_message":[],"_revision_number":1},{"id":"bad72cefa8b173aac51e2d7d1ba053eb6f7f2799","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-06-27 06:44:25.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2135a47ced7cca45089e3a42df8040aaa3360c27","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-27 07:22:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/77/666077/1/openstack-check/tripleo-build-containers-fedora-28-master/5df9daf/ : FAILURE in 7m 09s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/77/666077/1/openstack-check/tripleo-build-containers-fedora-28-stein/e7bb414/ : SUCCESS in 36m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"98adfe09c0e890052357632db137ab88980f5b12","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 08:58:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/77/666077/1/check/kolla-build-centos-binary/a3f328d/ : SUCCESS in 43m 59s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/77/666077/1/check/kolla-build-centos-source/eda163f/ : SUCCESS in 52m 21s\n- kolla-ansible-centos-source http://logs.openstack.org/77/666077/1/check/kolla-ansible-centos-source/31071e6/ : SUCCESS in 1h 18m 24s\n- kolla-ansible-centos-binary http://logs.openstack.org/77/666077/1/check/kolla-ansible-centos-binary/a37b2ce/ : SUCCESS in 1h 02m 15s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/77/666077/1/check/kolla-ansible-centos-source-upgrade/b142eb1/ : SUCCESS in 1h 18m 50s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/77/666077/1/check/tripleo-build-containers-centos-7/475fdc6/ : SUCCESS in 49m 48s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/77/666077/1/check/kolla-build-oraclelinux-binary/da5e1ac/ : SUCCESS in 32m 28s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/77/666077/1/check/kolla-build-oraclelinux-source/f75c6ac/ : SUCCESS in 48m 00s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/77/666077/1/check/kolla-ansible-oraclelinux-source/1080500/ : SUCCESS in 1h 06m 42s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/77/666077/1/check/kolla-ansible-oraclelinux-binary/8c86de4/ : SUCCESS in 1h 06m 25s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/77/666077/1/check/openstack-tox-lower-constraints/d7cdf95/ : SUCCESS in 3m 27s\n- openstack-tox-pep8 http://logs.openstack.org/77/666077/1/check/openstack-tox-pep8/6a5de08/ : SUCCESS in 3m 03s\n- openstack-tox-py27 http://logs.openstack.org/77/666077/1/check/openstack-tox-py27/49128bf/ : SUCCESS in 3m 06s\n- openstack-tox-py36 http://logs.openstack.org/77/666077/1/check/openstack-tox-py36/565dc9b/ : SUCCESS in 3m 18s\n- openstack-tox-py37 http://logs.openstack.org/77/666077/1/check/openstack-tox-py37/414a2d9/ : SUCCESS in 3m 31s\n- openstack-tox-docs http://logs.openstack.org/77/666077/1/check/openstack-tox-docs/e62a0ea/html/ : SUCCESS in 2m 39s\n- kolla-build-ubuntu-binary http://logs.openstack.org/77/666077/1/check/kolla-build-ubuntu-binary/85f376a/ : SUCCESS in 25m 54s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/77/666077/1/check/kolla-build-ubuntu-source/14aa80f/ : SUCCESS in 41m 36s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/77/666077/1/check/kolla-ansible-ubuntu-source/9905a2b/ : SUCCESS in 52m 38s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/77/666077/1/check/kolla-ansible-ubuntu-binary/d99d56b/ : SUCCESS in 46m 44s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/77/666077/1/check/kolla-ansible-ubuntu-source-upgrade/2e278eb/ : SUCCESS in 59m 46s","accounts_in_message":[],"_revision_number":1},{"id":"360c39956906d833ea9e3d024a3db32740965c4d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 08:58:20.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a945b72524d7a760615638079d9e1bb429e847f4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 10:28:50.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-source http://logs.openstack.org/77/666077/1/gate/kolla-build-centos-source/48f3c11/ : RETRY_LIMIT in 1m 49s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/77/666077/1/gate/tripleo-build-containers-centos-7/0b4ee1d/ : SUCCESS in 45m 12s\n- kolla-build-oraclelinux-source http://logs.openstack.org/77/666077/1/gate/kolla-build-oraclelinux-source/a8878bf/ : SUCCESS in 46m 20s\n- openstack-tox-lower-constraints http://logs.openstack.org/77/666077/1/gate/openstack-tox-lower-constraints/93cc183/ : SUCCESS in 3m 49s\n- openstack-tox-pep8 http://logs.openstack.org/77/666077/1/gate/openstack-tox-pep8/0f4caef/ : SUCCESS in 3m 21s\n- openstack-tox-py27 http://logs.openstack.org/77/666077/1/gate/openstack-tox-py27/eb79459/ : SUCCESS in 3m 13s\n- openstack-tox-py36 http://logs.openstack.org/77/666077/1/gate/openstack-tox-py36/ca433ca/ : SUCCESS in 3m 31s\n- openstack-tox-py37 http://logs.openstack.org/77/666077/1/gate/openstack-tox-py37/5e78a66/ : SUCCESS in 3m 42s\n- openstack-tox-docs http://logs.openstack.org/77/666077/1/gate/openstack-tox-docs/9ab0481/html/ : SUCCESS in 2m 58s\n- kolla-build-ubuntu-source http://logs.openstack.org/77/666077/1/gate/kolla-build-ubuntu-source/1f93ad5/ : SUCCESS in 28m 36s","accounts_in_message":[],"_revision_number":1},{"id":"f4c93984d272b9b83343f69c87a9992f4699393e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-27 10:33:43.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ff989eaed649f52e8b983b3a112be7d4772134a3","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-27 11:11:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/77/666077/1/openstack-check/tripleo-build-containers-fedora-28-master/6885869/ : FAILURE in 6m 27s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/77/666077/1/openstack-check/tripleo-build-containers-fedora-28-stein/881502e/ : SUCCESS in 33m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ddf6e61ee82affb6ab4308caf5b9be9b6f4f4719","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 13:25:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/77/666077/1/check/kolla-build-centos-binary/eeeac84/ : SUCCESS in 37m 51s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/77/666077/1/check/kolla-build-centos-source/ce6d547/ : SUCCESS in 47m 00s\n- kolla-ansible-centos-source http://logs.openstack.org/77/666077/1/check/kolla-ansible-centos-source/4c8732b/ : SUCCESS in 1h 11m 14s\n- kolla-ansible-centos-binary http://logs.openstack.org/77/666077/1/check/kolla-ansible-centos-binary/523a11b/ : SUCCESS in 1h 04m 12s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/77/666077/1/check/kolla-ansible-centos-source-upgrade/81a5851/ : SUCCESS in 1h 13m 33s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/77/666077/1/check/tripleo-build-containers-centos-7/6116996/ : SUCCESS in 47m 30s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/77/666077/1/check/kolla-build-oraclelinux-binary/d33e021/ : SUCCESS in 48m 27s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/77/666077/1/check/kolla-build-oraclelinux-source/6d7f5ca/ : SUCCESS in 52m 50s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/77/666077/1/check/kolla-ansible-oraclelinux-source/8f083ef/ : SUCCESS in 1h 07m 51s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/77/666077/1/check/kolla-ansible-oraclelinux-binary/b0a89b2/ : SUCCESS in 1h 00m 51s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/77/666077/1/check/openstack-tox-lower-constraints/d9e3ab6/ : SUCCESS in 3m 36s\n- openstack-tox-pep8 http://logs.openstack.org/77/666077/1/check/openstack-tox-pep8/1a101f3/ : SUCCESS in 3m 19s\n- openstack-tox-py27 http://logs.openstack.org/77/666077/1/check/openstack-tox-py27/c4eca4c/ : SUCCESS in 3m 11s\n- openstack-tox-py36 http://logs.openstack.org/77/666077/1/check/openstack-tox-py36/9ea334b/ : SUCCESS in 3m 02s\n- openstack-tox-py37 http://logs.openstack.org/77/666077/1/check/openstack-tox-py37/9802018/ : SUCCESS in 3m 48s\n- openstack-tox-docs http://logs.openstack.org/77/666077/1/check/openstack-tox-docs/d139592/html/ : SUCCESS in 2m 50s\n- kolla-build-ubuntu-binary http://logs.openstack.org/77/666077/1/check/kolla-build-ubuntu-binary/bf3b393/ : SUCCESS in 23m 25s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/77/666077/1/check/kolla-build-ubuntu-source/6dcd30d/ : SUCCESS in 43m 39s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/77/666077/1/check/kolla-ansible-ubuntu-source/7cf6939/ : SUCCESS in 51m 39s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/77/666077/1/check/kolla-ansible-ubuntu-binary/a837f04/ : SUCCESS in 49m 38s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/77/666077/1/check/kolla-ansible-ubuntu-source-upgrade/bdbe07c/ : SUCCESS in 58m 13s","accounts_in_message":[],"_revision_number":1},{"id":"097a8b6314cdbef93b9842bd902e634d8c268770","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 13:26:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"40cbf88ba46238b219d76af5d7d49573597fd3fa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 14:36:01.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source http://logs.openstack.org/77/666077/1/gate/kolla-build-centos-source/4cd93d9/ : SUCCESS in 48m 45s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/77/666077/1/gate/tripleo-build-containers-centos-7/6fd2deb/ : SUCCESS in 43m 18s\n- kolla-build-oraclelinux-source http://logs.openstack.org/77/666077/1/gate/kolla-build-oraclelinux-source/f7010e8/ : SUCCESS in 46m 50s\n- openstack-tox-lower-constraints http://logs.openstack.org/77/666077/1/gate/openstack-tox-lower-constraints/aad4a74/ : SUCCESS in 3m 28s\n- openstack-tox-pep8 http://logs.openstack.org/77/666077/1/gate/openstack-tox-pep8/5488e55/ : SUCCESS in 3m 16s\n- openstack-tox-py27 http://logs.openstack.org/77/666077/1/gate/openstack-tox-py27/2b1ee1c/ : SUCCESS in 3m 08s\n- openstack-tox-py36 http://logs.openstack.org/77/666077/1/gate/openstack-tox-py36/4eaa289/ : SUCCESS in 3m 08s\n- openstack-tox-py37 http://logs.openstack.org/77/666077/1/gate/openstack-tox-py37/afd22c3/ : SUCCESS in 3m 46s\n- openstack-tox-docs http://logs.openstack.org/77/666077/1/gate/openstack-tox-docs/397dacd/html/ : SUCCESS in 3m 03s\n- kolla-build-ubuntu-source http://logs.openstack.org/77/666077/1/gate/kolla-build-ubuntu-source/ac1fa18/ : SUCCESS in 36m 18s","accounts_in_message":[],"_revision_number":1},{"id":"741a9888dfae6ab4cfa7015c3800b7a6ac86c4ab","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 14:36:01.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6acca35f5b3a7517881b5d494500e3b67152285d","revisions":{"6acca35f5b3a7517881b5d494500e3b67152285d":{"kind":"REWORK","_number":1,"created":"2019-06-18 17:43:10.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/77/666077/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/77/666077/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/666077/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/666077/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/666077/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/77/666077/1"}}},"commit":{"parents":[{"commit":"e735b0985e3994d64f800a441e20bf3df092b74b","subject":"Merge \"Debian/Ubuntu: more python3 packages for binary\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e735b0985e3994d64f800a441e20bf3df092b74b"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-06-07 17:49:46.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-06-18 17:43:10.000000000","tz":0},"subject":"Fix ownership of ironic inspector DHCP hostsdir","message":"Fix ownership of ironic inspector DHCP hostsdir\n\nIf you run with enable_ironic set to true and\nironic_inspector_pxe_filter set to dnsmasq (the default in stein),\nironic inspector can fail with the following in the logs:\n\nPermission denied: u\u0027/var/lib/ironic-inspector/dhcp-hostsdir/\u003cMAC address\u003e\u0027\n\nA node must be registered with a port for this to happen.\n\nWeirdly this happens on centos/source, ubuntu/source, and ubuntu/binary,\nbut not centos/binary.\n\nThis change changes the ownership of\n/var/lib/ironic-inspector/dhcp-hostsdir to ironic-inspector user to make\nit writeable.\n\nChange-Id: I19447727f19dbd9c0a3e17d218b48ddc4c253587\nCloses-Bug: #1832026\n(cherry picked from commit 41bcd7b4961fd52066b79804fdce9f81cc129735)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6acca35f5b3a7517881b5d494500e3b67152285d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6acca35f5b3a7517881b5d494500e3b67152285d"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
