)]}'
{"id":"openstack%2Fkolla~666083","triplet_id":"openstack%2Fkolla~stable%2Frocky~I32ba1abd204c9dba4bbe3132b5650b0e391616af","project":"openstack/kolla","branch":"stable/rocky","topic":"bug/1826327-stable/rocky","hashtags":[],"change_id":"I32ba1abd204c9dba4bbe3132b5650b0e391616af","subject":"Add sha256 in ceph fetch keys","status":"MERGED","created":"2019-06-18 17:53:20.000000000","updated":"2019-06-24 15:57:32.000000000","submitted":"2019-06-24 15:57:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"666083-1561391852861-f9ed76da","meta_rev_id":"c6748a71771e7423a941ce865eb5be872f114736","_number":666083,"virtual_id_number":666083,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-06-24 15:57:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-06-19 14:01:06.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2019-06-24 10:12:14.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2019-06-19 16:12:17.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2019-06-24 10:12:14.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"date":"2019-06-18 17:53:20.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-18 17:53:20.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2019-06-19 14:01:06.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-06-24 10:12:14.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-06-24 15:57:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b15e6fd28b32f32b419254ceb08fb8fe3fa3c478","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-18 17:53:20.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"8e7cc441a10da394b644d82b57beb6db1f31d192","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-18 18:05:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/83/666083/1/openstack-check/tripleo-build-containers-fedora-28-master/0ba1e86/ : RETRY_LIMIT in 2m 38s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/83/666083/1/openstack-check/tripleo-build-containers-fedora-28-stein/96fb801/ : FAILURE in 5m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8831dc7e91942c355270f85b5399b4e20c3f1f58","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 21:52:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/83/666083/1/check/kolla-build-centos-binary/9a7b00d/ : SUCCESS in 33m 51s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/83/666083/1/check/kolla-build-centos-source/8d6c8f3/ : SUCCESS in 55m 23s\n- kolla-ansible-centos-source http://logs.openstack.org/83/666083/1/check/kolla-ansible-centos-source/d2840cf/ : SUCCESS in 1h 18m 01s\n- kolla-ansible-centos-binary http://logs.openstack.org/83/666083/1/check/kolla-ansible-centos-binary/489ac7d/ : SUCCESS in 1h 26m 54s (non-voting)\n- kolla-ansible-centos-source-ceph http://logs.openstack.org/83/666083/1/check/kolla-ansible-centos-source-ceph/7b52410/ : POST_FAILURE in 2m 34s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/83/666083/1/check/kolla-build-oraclelinux-binary/aee801a/ : SUCCESS in 31m 34s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/83/666083/1/check/kolla-build-oraclelinux-source/ce7d94e/ : SUCCESS in 39m 35s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/83/666083/1/check/kolla-ansible-oraclelinux-source/689b51a/ : SUCCESS in 1h 37m 02s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/83/666083/1/check/kolla-ansible-oraclelinux-binary/6ead1fa/ : SUCCESS in 1h 23m 42s (non-voting)\n- kolla-ansible-oraclelinux-source-ceph http://logs.openstack.org/83/666083/1/check/kolla-ansible-oraclelinux-source-ceph/db88c47/ : POST_FAILURE in 2m 21s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/83/666083/1/check/openstack-tox-lower-constraints/5ff9a24/ : SUCCESS in 4m 02s\n- openstack-tox-pep8 http://logs.openstack.org/83/666083/1/check/openstack-tox-pep8/252749d/ : SUCCESS in 4m 20s\n- openstack-tox-py27 http://logs.openstack.org/83/666083/1/check/openstack-tox-py27/05a4256/ : SUCCESS in 4m 03s\n- openstack-tox-py35 http://logs.openstack.org/83/666083/1/check/openstack-tox-py35/95ea25c/ : SUCCESS in 4m 46s\n- build-openstack-sphinx-docs http://logs.openstack.org/83/666083/1/check/build-openstack-sphinx-docs/f1e09e1/html/ : SUCCESS in 3m 24s\n- kolla-build-ubuntu-binary http://logs.openstack.org/83/666083/1/check/kolla-build-ubuntu-binary/e2d4008/ : SUCCESS in 1h 26m 21s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/83/666083/1/check/kolla-build-ubuntu-source/d993afc/ : SUCCESS in 35m 06s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/83/666083/1/check/kolla-ansible-ubuntu-source/c5cd4da/ : SUCCESS in 1h 07m 54s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/83/666083/1/check/kolla-ansible-ubuntu-binary/cdd2954/ : TIMED_OUT in 2h 01m 10s (non-voting)\n- kolla-ansible-ubuntu-source-ceph http://logs.openstack.org/83/666083/1/check/kolla-ansible-ubuntu-source-ceph/ff0265d/ : POST_FAILURE in 2m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"da615f73a3a478141cb176233c5d286ef05287ca","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-19 11:49:57.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d3d3de3b078ab4d52e244459c732ad3a009f808a","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-19 12:30:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/83/666083/1/openstack-check/tripleo-build-containers-fedora-28-master/444b15b/ : RETRY_LIMIT in 2m 47s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/83/666083/1/openstack-check/tripleo-build-containers-fedora-28-stein/e555d67/ : SUCCESS in 32m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2d01beb672c48a1f5f1174c86bc36f8a1b6b94df","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-19 13:23:30.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/83/666083/1/check/kolla-build-centos-binary/04294a4/ : RETRY_LIMIT in 51m 05s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/83/666083/1/check/kolla-build-centos-source/7b2ebeb/ : SUCCESS in 51m 39s\n- kolla-ansible-centos-source http://logs.openstack.org/83/666083/1/check/kolla-ansible-centos-source/fde61e5/ : SUCCESS in 1h 16m 17s\n- kolla-ansible-centos-binary http://logs.openstack.org/83/666083/1/check/kolla-ansible-centos-binary/413bdd7/ : SUCCESS in 1h 01m 48s (non-voting)\n- kolla-ansible-centos-source-ceph http://logs.openstack.org/83/666083/1/check/kolla-ansible-centos-source-ceph/fcf1789/ : POST_FAILURE in 2m 24s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/83/666083/1/check/kolla-build-oraclelinux-binary/2826151/ : SUCCESS in 50m 49s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/83/666083/1/check/kolla-build-oraclelinux-source/c3f3772/ : RETRY_LIMIT in 50m 57s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/83/666083/1/check/kolla-ansible-oraclelinux-source/1d737e8/ : RETRY_LIMIT in 50m 59s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/83/666083/1/check/kolla-ansible-oraclelinux-binary/b127cf7/ : RETRY_LIMIT in 50m 59s (non-voting)\n- kolla-ansible-oraclelinux-source-ceph http://logs.openstack.org/83/666083/1/check/kolla-ansible-oraclelinux-source-ceph/529f01e/ : RETRY_LIMIT in 50m 53s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/83/666083/1/check/openstack-tox-lower-constraints/5f1c900/ : SUCCESS in 3m 59s\n- openstack-tox-pep8 http://logs.openstack.org/83/666083/1/check/openstack-tox-pep8/1535e34/ : SUCCESS in 3m 55s\n- openstack-tox-py27 http://logs.openstack.org/83/666083/1/check/openstack-tox-py27/ed1c311/ : SUCCESS in 2m 53s\n- openstack-tox-py35 http://logs.openstack.org/83/666083/1/check/openstack-tox-py35/e630af5/ : SUCCESS in 3m 33s\n- build-openstack-sphinx-docs http://logs.openstack.org/83/666083/1/check/build-openstack-sphinx-docs/0e27d65/html/ : SUCCESS in 3m 44s\n- kolla-build-ubuntu-binary http://logs.openstack.org/83/666083/1/check/kolla-build-ubuntu-binary/227d24e/ : SUCCESS in 57m 31s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/83/666083/1/check/kolla-build-ubuntu-source/8300fa4/ : RETRY_LIMIT in 53m 17s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/83/666083/1/check/kolla-ansible-ubuntu-source/85a1dff/ : SUCCESS in 53m 05s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/83/666083/1/check/kolla-ansible-ubuntu-binary/66fbdf0/ : RETRY_LIMIT in 42m 21s (non-voting)\n- kolla-ansible-ubuntu-source-ceph http://logs.openstack.org/83/666083/1/check/kolla-ansible-ubuntu-source-ceph/1488c11/ : POST_FAILURE in 2m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c10b5923c59437ca1eea759ac6795068c4828775","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-19 13:25:56.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"666fa78c56b63494416372eb74d1bfaba113ebb0","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-19 14:01:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/83/666083/1/openstack-check/tripleo-build-containers-fedora-28-master/d2fd53e/ : RETRY_LIMIT in 2m 53s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/83/666083/1/openstack-check/tripleo-build-containers-fedora-28-stein/1c2d597/ : SUCCESS in 31m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5534f495500b6942c7a70d86f1edf363b01ddc0e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-19 15:18:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/83/666083/1/check/kolla-build-centos-binary/7679a73/ : SUCCESS in 25m 07s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/83/666083/1/check/kolla-build-centos-source/3158129/ : SUCCESS in 40m 03s\n- kolla-ansible-centos-source http://logs.openstack.org/83/666083/1/check/kolla-ansible-centos-source/1187602/ : SUCCESS in 1h 32m 29s\n- kolla-ansible-centos-binary http://logs.openstack.org/83/666083/1/check/kolla-ansible-centos-binary/1d4a24d/ : SUCCESS in 1h 28m 09s (non-voting)\n- kolla-ansible-centos-source-ceph http://logs.openstack.org/83/666083/1/check/kolla-ansible-centos-source-ceph/86da5ba/ : POST_FAILURE in 3m 04s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/83/666083/1/check/kolla-build-oraclelinux-binary/5d9c939/ : SUCCESS in 29m 16s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/83/666083/1/check/kolla-build-oraclelinux-source/2f58707/ : SUCCESS in 1h 01m 40s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/83/666083/1/check/kolla-ansible-oraclelinux-source/80f0a88/ : SUCCESS in 1h 11m 36s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/83/666083/1/check/kolla-ansible-oraclelinux-binary/eb40972/ : SUCCESS in 1h 04m 15s (non-voting)\n- kolla-ansible-oraclelinux-source-ceph http://logs.openstack.org/83/666083/1/check/kolla-ansible-oraclelinux-source-ceph/81d735f/ : POST_FAILURE in 2m 57s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/83/666083/1/check/openstack-tox-lower-constraints/4ffe92d/ : SUCCESS in 3m 18s\n- openstack-tox-pep8 http://logs.openstack.org/83/666083/1/check/openstack-tox-pep8/d5c9e24/ : SUCCESS in 3m 59s\n- openstack-tox-py27 http://logs.openstack.org/83/666083/1/check/openstack-tox-py27/a8bc85a/ : SUCCESS in 4m 26s\n- openstack-tox-py35 http://logs.openstack.org/83/666083/1/check/openstack-tox-py35/152bbec/ : SUCCESS in 4m 04s\n- build-openstack-sphinx-docs http://logs.openstack.org/83/666083/1/check/build-openstack-sphinx-docs/6712a03/html/ : SUCCESS in 4m 07s\n- kolla-build-ubuntu-binary http://logs.openstack.org/83/666083/1/check/kolla-build-ubuntu-binary/9116ed3/ : SUCCESS in 1h 25m 49s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/83/666083/1/check/kolla-build-ubuntu-source/6bb2fdf/ : SUCCESS in 45m 55s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/83/666083/1/check/kolla-ansible-ubuntu-source/c9af4f9/ : SUCCESS in 1h 15m 26s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/83/666083/1/check/kolla-ansible-ubuntu-binary/e573ea9/ : SUCCESS in 1h 38m 44s (non-voting)\n- kolla-ansible-ubuntu-source-ceph http://logs.openstack.org/83/666083/1/check/kolla-ansible-ubuntu-source-ceph/291aebc/ : POST_FAILURE in 2m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"758e3b5ce3bc595d59472127ec24880efe52eefe","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-19 16:12:17.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"05be3a24e5d0231a7a7c21e31d7c5b4e9da06c4a","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-24 10:12:14.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2e571bdbb65022f018e88b434d89a7fca9d203a7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-24 10:12:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"be464cd3f3b3a657759fbcf90e7cded7373575e9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-24 15:57:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source http://logs.openstack.org/83/666083/1/gate/kolla-build-centos-source/6f3ed13/ : SUCCESS in 45m 29s\n- kolla-build-oraclelinux-source http://logs.openstack.org/83/666083/1/gate/kolla-build-oraclelinux-source/b1594ff/ : SUCCESS in 55m 49s\n- openstack-tox-lower-constraints http://logs.openstack.org/83/666083/1/gate/openstack-tox-lower-constraints/75f8f1a/ : SUCCESS in 3m 21s\n- openstack-tox-pep8 http://logs.openstack.org/83/666083/1/gate/openstack-tox-pep8/d8eaf7a/ : SUCCESS in 3m 50s\n- openstack-tox-py27 http://logs.openstack.org/83/666083/1/gate/openstack-tox-py27/6af312d/ : SUCCESS in 3m 15s\n- openstack-tox-py35 http://logs.openstack.org/83/666083/1/gate/openstack-tox-py35/023f23c/ : SUCCESS in 3m 30s\n- build-openstack-sphinx-docs http://logs.openstack.org/83/666083/1/gate/build-openstack-sphinx-docs/5ef2007/html/ : SUCCESS in 3m 34s\n- kolla-build-ubuntu-source http://logs.openstack.org/83/666083/1/gate/kolla-build-ubuntu-source/17c09c1/ : SUCCESS in 30m 23s","accounts_in_message":[],"_revision_number":1},{"id":"44393f7a49d9672452f7332c66ce5d887e0eb745","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-24 15:57:32.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8001de09236504b6bfd7aa01ec5657ef0fb2136e","revisions":{"8001de09236504b6bfd7aa01ec5657ef0fb2136e":{"kind":"REWORK","_number":1,"created":"2019-06-18 17:53:20.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/83/666083/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/83/666083/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/83/666083/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/83/666083/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/83/666083/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/83/666083/1"}}},"commit":{"parents":[{"commit":"6072e842924069dffc580afb77f49420a07ba086","subject":"Merge \"Fix lower-constraints test fail due to failing dep build\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6072e842924069dffc580afb77f49420a07ba086"}]}],"author":{"name":"Michal Nasiadka","email":"michal.nasiadka@nokia.com","date":"2019-04-25 07:06:28.000000000","tz":120},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-06-18 17:53:20.000000000","tz":0},"subject":"Add sha256 in ceph fetch keys","message":"Add sha256 in ceph fetch keys\n\nfetch_ceph_keys.py is using insecure sha1 function, this patch adds\nsha256 beside sha1 - in order to remove sha1 safely in U release.\n\nChange-Id: I32ba1abd204c9dba4bbe3132b5650b0e391616af\nCloses-Bug: #1826327\n(cherry picked from commit 0a2dc8aa1d267da3c1c2925b3f292bd96aa50edb)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8001de09236504b6bfd7aa01ec5657ef0fb2136e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8001de09236504b6bfd7aa01ec5657ef0fb2136e"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
