)]}'
{"id":"openstack%2Fkolla~666084","triplet_id":"openstack%2Fkolla~stable%2Fqueens~I32ba1abd204c9dba4bbe3132b5650b0e391616af","project":"openstack/kolla","branch":"stable/queens","topic":"bug/1826327-stable/queens","hashtags":[],"change_id":"I32ba1abd204c9dba4bbe3132b5650b0e391616af","subject":"Add sha256 in ceph fetch keys","status":"MERGED","created":"2019-06-18 17:53:32.000000000","updated":"2019-06-19 21:51:13.000000000","submitted":"2019-06-19 21:51:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"666084-1560981073716-1a2baadc","meta_rev_id":"9f4d6aebf7321620933b96f9a1586efa1a931245","_number":666084,"virtual_id_number":666084,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-06-19 21:51:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-06-18 18:05:09.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2019-06-19 20:40:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2019-06-19 11:49:39.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2019-06-19 20:40:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-18 18:05:09.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-06-19 20:40:44.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2019-06-19 21:51:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d0f31e99465f7ddfaf634cd153dc7bbb68d94097","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-18 17:53:32.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"40327433d7e640de4fc6b4592114b0dd90e451ef","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-18 18:05:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/84/666084/1/openstack-check/tripleo-build-containers-fedora-28-master/cdac0c2/ : RETRY_LIMIT in 2m 32s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/84/666084/1/openstack-check/tripleo-build-containers-fedora-28-stein/2658a14/ : FAILURE in 5m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d363ca4a631f9fb27790328636f50a3e0bd8d076","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 21:10:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/84/666084/1/check/kolla-build-centos-binary/5b07c04/ : SUCCESS in 36m 54s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/84/666084/1/check/kolla-build-centos-source/21d680c/ : SUCCESS in 28m 57s\n- kolla-ansible-centos-source http://logs.openstack.org/84/666084/1/check/kolla-ansible-centos-source/4fa011e/ : SUCCESS in 59m 05s (non-voting)\n- kolla-ansible-centos-binary http://logs.openstack.org/84/666084/1/check/kolla-ansible-centos-binary/f22f116/ : SUCCESS in 1h 00m 55s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/84/666084/1/check/kolla-build-oraclelinux-binary/11d9fc9/ : SUCCESS in 25m 15s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/84/666084/1/check/kolla-build-oraclelinux-source/3931e34/ : SUCCESS in 48m 42s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/84/666084/1/check/kolla-ansible-oraclelinux-source/0aa7f6b/ : SUCCESS in 1h 15m 49s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/84/666084/1/check/kolla-ansible-oraclelinux-binary/4e72948/ : FAILURE in 49m 21s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/84/666084/1/check/openstack-tox-pep8/0f80291/ : SUCCESS in 3m 38s\n- openstack-tox-py27 http://logs.openstack.org/84/666084/1/check/openstack-tox-py27/bd9bba3/ : SUCCESS in 3m 42s\n- openstack-tox-py35 http://logs.openstack.org/84/666084/1/check/openstack-tox-py35/da023f5/ : SUCCESS in 3m 26s\n- build-openstack-sphinx-docs http://logs.openstack.org/84/666084/1/check/build-openstack-sphinx-docs/79a532b/html/ : SUCCESS in 4m 01s\n- kolla-build-ubuntu-binary http://logs.openstack.org/84/666084/1/check/kolla-build-ubuntu-binary/7b900b4/ : SUCCESS in 31m 27s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/84/666084/1/check/kolla-build-ubuntu-source/7fa7b78/ : SUCCESS in 27m 33s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/84/666084/1/check/kolla-ansible-ubuntu-source/537c451/ : SUCCESS in 53m 17s (non-voting)\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/84/666084/1/check/kolla-ansible-ubuntu-binary/387e600/ : SUCCESS in 1h 00m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"576655063d92db275de53bee8631a879ac748e57","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-19 11:49:39.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1889d478caef70f46ff412c266dd3c48d1a729e7","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-06-19 20:40:44.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"88209116f4906fd932692c73568d231ef32b4022","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-19 20:41:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8bf3d56434d6d9ab6704592df7a69644595292b1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-19 21:51:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"1ef24786462d5d2e064327ebc70f665aae19322f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-19 21:51:13.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source http://logs.openstack.org/84/666084/1/gate/kolla-build-centos-source/ba56d43/ : SUCCESS in 42m 50s\n- kolla-build-oraclelinux-source http://logs.openstack.org/84/666084/1/gate/kolla-build-oraclelinux-source/d226964/ : SUCCESS in 41m 25s\n- openstack-tox-pep8 http://logs.openstack.org/84/666084/1/gate/openstack-tox-pep8/cd6f52c/ : SUCCESS in 4m 22s\n- openstack-tox-py27 http://logs.openstack.org/84/666084/1/gate/openstack-tox-py27/9e53b72/ : SUCCESS in 4m 00s\n- openstack-tox-py35 http://logs.openstack.org/84/666084/1/gate/openstack-tox-py35/62bb7e9/ : SUCCESS in 3m 16s\n- build-openstack-sphinx-docs http://logs.openstack.org/84/666084/1/gate/build-openstack-sphinx-docs/e7b253c/html/ : SUCCESS in 3m 22s\n- kolla-build-ubuntu-source http://logs.openstack.org/84/666084/1/gate/kolla-build-ubuntu-source/8f7f0b7/ : SUCCESS in 32m 21s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7951713790464eedaec1e59851a42febc4acc346","revisions":{"7951713790464eedaec1e59851a42febc4acc346":{"kind":"REWORK","_number":1,"created":"2019-06-18 17:53:32.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/84/666084/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/84/666084/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/84/666084/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/84/666084/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/84/666084/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/84/666084/1"}}},"commit":{"parents":[{"commit":"fc49f6cfc50808593717bef91003c2f25f81dca9","subject":"Handle role properly when it is changed in kolla_keystone_user module","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fc49f6cfc50808593717bef91003c2f25f81dca9"}]}],"author":{"name":"Michal Nasiadka","email":"michal.nasiadka@nokia.com","date":"2019-04-25 07:06:28.000000000","tz":120},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-06-18 17:53:32.000000000","tz":0},"subject":"Add sha256 in ceph fetch keys","message":"Add sha256 in ceph fetch keys\n\nfetch_ceph_keys.py is using insecure sha1 function, this patch adds\nsha256 beside sha1 - in order to remove sha1 safely in U release.\n\nChange-Id: I32ba1abd204c9dba4bbe3132b5650b0e391616af\nCloses-Bug: #1826327\n(cherry picked from commit 0a2dc8aa1d267da3c1c2925b3f292bd96aa50edb)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7951713790464eedaec1e59851a42febc4acc346"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7951713790464eedaec1e59851a42febc4acc346"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
