)]}'
{"id":"openstack%2Fkolla~666543","triplet_id":"openstack%2Fkolla~master~I6af152b179a1e2254e092fcb16fb8e421044d38e","project":"openstack/kolla","branch":"master","hashtags":[],"change_id":"I6af152b179a1e2254e092fcb16fb8e421044d38e","subject":"DNM: Test CentOS Ceph","status":"ABANDONED","created":"2019-06-20 11:32:35.000000000","updated":"2019-06-27 09:13:47.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ada6bc8c82e97c029a52f0e00c39c3f6a662bed8","_number":666543,"virtual_id_number":666543,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2019-06-20 12:41:50.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":1,"date":"2019-06-20 13:26:54.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2019-06-20 12:05:29.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-20 12:05:29.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-06-20 12:41:50.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-06-20 13:26:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b40ef1b7257e61ab5dd9be19f2b22f94d989b36f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-20 11:32:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a6f6a69dcf82cfa187d0f4af7912eea98ecc5381","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-20 12:05:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/43/666543/1/openstack-check/tripleo-build-containers-fedora-28-master/179caef/ : FAILURE in 5m 44s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/43/666543/1/openstack-check/tripleo-build-containers-fedora-28-stein/8a2ab40/ : SUCCESS in 31m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"120b5d3965c4a51cde22b58e59b34f90405e3b72","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-20 12:41:50.000000000","message":"Patch Set 1:\n\nstein: https://review.opendev.org/666030","accounts_in_message":[],"_revision_number":1},{"id":"ba5c03dbc51ab06d9e65224cc2c053ebd7a6d1ff","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-20 13:26:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/43/666543/1/check/kolla-build-centos-binary/8bb9115/ : SUCCESS in 41m 41s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/43/666543/1/check/kolla-build-centos-source/f0c38f2/ : SUCCESS in 47m 34s\n- kolla-ansible-centos-source http://logs.openstack.org/43/666543/1/check/kolla-ansible-centos-source/cb91ac8/ : SUCCESS in 1h 12m 14s\n- kolla-ansible-centos-binary http://logs.openstack.org/43/666543/1/check/kolla-ansible-centos-binary/7b1de78/ : SUCCESS in 1h 07m 42s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/43/666543/1/check/kolla-ansible-centos-source-upgrade/ed68ea9/ : SUCCESS in 1h 17m 03s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/43/666543/1/check/tripleo-build-containers-centos-7/d721501/ : SUCCESS in 48m 44s\n- kolla-ansible-centos-source-ceph http://logs.openstack.org/43/666543/1/check/kolla-ansible-centos-source-ceph/99eee1d/ : SUCCESS in 1h 42m 54s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph http://logs.openstack.org/43/666543/1/check/kolla-ansible-centos-source-upgrade-ceph/a751ae2/ : SUCCESS in 1h 42m 23s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/43/666543/1/check/openstack-tox-lower-constraints/0d1a059/ : SUCCESS in 10m 30s\n- openstack-tox-pep8 http://logs.openstack.org/43/666543/1/check/openstack-tox-pep8/4784ebc/ : SUCCESS in 3m 11s\n- openstack-tox-py27 http://logs.openstack.org/43/666543/1/check/openstack-tox-py27/391c553/ : SUCCESS in 3m 21s\n- openstack-tox-py36 http://logs.openstack.org/43/666543/1/check/openstack-tox-py36/ad93adf/ : SUCCESS in 8m 27s\n- openstack-tox-py37 http://logs.openstack.org/43/666543/1/check/openstack-tox-py37/31c5cff/ : SUCCESS in 8m 53s\n- openstack-tox-docs http://logs.openstack.org/43/666543/1/check/openstack-tox-docs/9b857b8/html/ : SUCCESS in 2m 58s\n- kolla-build-ubuntu-binary http://logs.openstack.org/43/666543/1/check/kolla-build-ubuntu-binary/9930426/ : SUCCESS in 27m 26s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/43/666543/1/check/kolla-build-ubuntu-source/afe3ee4/ : SUCCESS in 38m 04s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/43/666543/1/check/kolla-ansible-ubuntu-source/6fd3150/ : SUCCESS in 46m 47s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/43/666543/1/check/kolla-ansible-ubuntu-binary/4e1f332/ : SUCCESS in 42m 57s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/43/666543/1/check/kolla-ansible-ubuntu-source-upgrade/bdb40fc/ : SUCCESS in 1h 05m 02s\n- kolla-ansible-ubuntu-source-ceph http://logs.openstack.org/43/666543/1/check/kolla-ansible-ubuntu-source-ceph/2204523/ : SUCCESS in 1h 20m 58s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph http://logs.openstack.org/43/666543/1/check/kolla-ansible-ubuntu-source-upgrade-ceph/dcab891/ : SUCCESS in 1h 19m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ada6bc8c82e97c029a52f0e00c39c3f6a662bed8","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-27 09:13:47.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7b3822e217206b6227f9a7f96c207c71eb41c706","revisions":{"7b3822e217206b6227f9a7f96c207c71eb41c706":{"kind":"REWORK","_number":1,"created":"2019-06-20 11:32:35.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/43/666543/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/43/666543/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/43/666543/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/43/666543/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/43/666543/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/43/666543/1"}}},"commit":{"parents":[{"commit":"86907688802f2633d165271eff08cdcdc0284b33","subject":"Merge \"Disable and remove OracleLinux CI jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/86907688802f2633d165271eff08cdcdc0284b33"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-06-20 11:31:41.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-06-20 11:31:41.000000000","tz":60},"subject":"DNM: Test CentOS Ceph","message":"DNM: Test CentOS Ceph\n\nChange-Id: I6af152b179a1e2254e092fcb16fb8e421044d38e\nDepends-On: https://review.opendev.org/666068\nDepends-On: https://review.opendev.org/666015/\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7b3822e217206b6227f9a7f96c207c71eb41c706"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7b3822e217206b6227f9a7f96c207c71eb41c706"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
