)]}'
{"id":"openstack%2Fkolla~666647","triplet_id":"openstack%2Fkolla~master~I9aaee4c2932f1e4ea9fe780a64e96a28fa6bccfb","project":"openstack/kolla","branch":"master","topic":"2005920/34181","hashtags":[],"change_id":"I9aaee4c2932f1e4ea9fe780a64e96a28fa6bccfb","subject":"Implement Apache WSGI for Qinling","status":"MERGED","created":"2019-06-20 17:06:07.000000000","updated":"2019-06-24 17:08:40.000000000","submitted":"2019-06-24 17:08:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"666647-1561396120509-6b32c878","meta_rev_id":"97da4afc5dcef6ea0d27e97a380b4f90c62385ac","_number":666647,"virtual_id_number":666647,"owner":{"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-06-24 17:08:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-06-20 18:15:35.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"all":[{"value":2,"date":"2019-06-24 10:08:01.000000000","_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":2,"date":"2019-06-21 09:33:40.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":1,"date":"2019-06-24 13:38:55.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-20 18:15:35.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-06-24 10:08:01.000000000","updated_by":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"reviewer":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"state":"REVIEWER"},{"updated":"2019-06-24 13:38:55.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-06-24 17:08:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"06acae57c19792f59a36cf35f7b34c51d4b6b759","author":{"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},"date":"2019-06-20 17:06:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5e7e14064887e562c158b79b9712962a5cd3ca24","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-20 18:15:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/47/666647/1/openstack-check/tripleo-build-containers-fedora-28-master/6f35a73/ : FAILURE in 6m 02s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/47/666647/1/openstack-check/tripleo-build-containers-fedora-28-stein/7e3907b/ : SUCCESS in 34m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c0630ae09734663da80700c04120d904faa1cbc0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-20 18:32:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/47/666647/1/check/kolla-build-centos-binary/d2f888a/ : SUCCESS in 40m 34s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/47/666647/1/check/kolla-build-centos-source/ed57deb/ : SUCCESS in 53m 31s\n- kolla-ansible-centos-source http://logs.openstack.org/47/666647/1/check/kolla-ansible-centos-source/56bb1cc/ : SUCCESS in 1h 19m 56s\n- kolla-ansible-centos-binary http://logs.openstack.org/47/666647/1/check/kolla-ansible-centos-binary/e7aabf3/ : SUCCESS in 1h 04m 19s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/47/666647/1/check/kolla-ansible-centos-source-upgrade/66c5c60/ : SUCCESS in 1h 16m 27s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/47/666647/1/check/tripleo-build-containers-centos-7/1784c89/ : SUCCESS in 46m 04s\n- openstack-tox-lower-constraints http://logs.openstack.org/47/666647/1/check/openstack-tox-lower-constraints/e030361/ : SUCCESS in 10m 47s\n- openstack-tox-pep8 http://logs.openstack.org/47/666647/1/check/openstack-tox-pep8/55f46ec/ : SUCCESS in 3m 44s\n- openstack-tox-py27 http://logs.openstack.org/47/666647/1/check/openstack-tox-py27/f16cb37/ : SUCCESS in 2m 56s\n- openstack-tox-py36 http://logs.openstack.org/47/666647/1/check/openstack-tox-py36/159b31b/ : SUCCESS in 3m 21s\n- openstack-tox-py37 http://logs.openstack.org/47/666647/1/check/openstack-tox-py37/0faf982/ : SUCCESS in 5m 03s\n- openstack-tox-docs http://logs.openstack.org/47/666647/1/check/openstack-tox-docs/0e736a4/html/ : SUCCESS in 2m 34s\n- kolla-build-ubuntu-binary http://logs.openstack.org/47/666647/1/check/kolla-build-ubuntu-binary/138a92b/ : SUCCESS in 24m 00s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/47/666647/1/check/kolla-build-ubuntu-source/5babd14/ : SUCCESS in 29m 28s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/47/666647/1/check/kolla-ansible-ubuntu-source/635e48a/ : SUCCESS in 48m 39s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/47/666647/1/check/kolla-ansible-ubuntu-binary/bdde216/ : SUCCESS in 57m 27s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/47/666647/1/check/kolla-ansible-ubuntu-source-upgrade/87bc4f6/ : SUCCESS in 1h 02m 27s","accounts_in_message":[],"_revision_number":1},{"id":"9d51701644a8562b1150d548b01d6d0b0cb286ed","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-21 09:33:40.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d581f7a78b8abec93162e6ebe3e751ac240afe38","author":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"date":"2019-06-24 10:08:01.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1078c88bf0fe1b09ce8ebe52f71babfe51f82a0c","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-24 13:38:55.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7f3dc6bd4bc4f64142a796636f8fc1db5b62dfb5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-24 13:39:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e5f869c2e6f143d839110d39d314710c67187b51","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-24 17:08:40.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source http://logs.openstack.org/47/666647/1/gate/kolla-build-centos-source/a5eb870/ : SUCCESS in 2h 06m 55s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/47/666647/1/gate/tripleo-build-containers-centos-7/05f4d4d/ : SUCCESS in 54m 00s\n- openstack-tox-lower-constraints http://logs.openstack.org/47/666647/1/gate/openstack-tox-lower-constraints/04ef530/ : SUCCESS in 3m 41s\n- openstack-tox-pep8 http://logs.openstack.org/47/666647/1/gate/openstack-tox-pep8/f1993e0/ : SUCCESS in 3m 10s\n- openstack-tox-py27 http://logs.openstack.org/47/666647/1/gate/openstack-tox-py27/81bcd4c/ : SUCCESS in 3m 10s\n- openstack-tox-py36 http://logs.openstack.org/47/666647/1/gate/openstack-tox-py36/8acc59d/ : SUCCESS in 3m 10s\n- openstack-tox-py37 http://logs.openstack.org/47/666647/1/gate/openstack-tox-py37/d32b580/ : SUCCESS in 3m 43s\n- openstack-tox-docs http://logs.openstack.org/47/666647/1/gate/openstack-tox-docs/d3c7444/html/ : SUCCESS in 3m 04s\n- kolla-build-ubuntu-source http://logs.openstack.org/47/666647/1/gate/kolla-build-ubuntu-source/0e682e6/ : SUCCESS in 23m 35s","accounts_in_message":[],"_revision_number":1},{"id":"4fa91fd50162ad5af999a03c394cde18ae14ebb2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-24 17:08:40.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f5a09567368e47aa6df861b0870a91028bc34f05","revisions":{"f5a09567368e47aa6df861b0870a91028bc34f05":{"kind":"REWORK","_number":1,"created":"2019-06-20 17:06:07.000000000","uploader":{"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},"ref":"refs/changes/47/666647/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/47/666647/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/666647/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/666647/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/666647/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/47/666647/1"}}},"commit":{"parents":[{"commit":"86907688802f2633d165271eff08cdcdc0284b33","subject":"Merge \"Disable and remove OracleLinux CI jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/86907688802f2633d165271eff08cdcdc0284b33"}]}],"author":{"name":"Gaëtan Trellu","email":"gaetan.trellu@incloudus.com","date":"2019-06-20 17:03:31.000000000","tz":-240},"committer":{"name":"Gaëtan Trellu","email":"gaetan.trellu@incloudus.com","date":"2019-06-20 17:06:06.000000000","tz":-240},"subject":"Implement Apache WSGI for Qinling","message":"Implement Apache WSGI for Qinling\n\nSince this review[1], Qinling supports WSGI execution.\n\nFrom a production perspective, Qinling should be deployed\nusing Apache and mod_wsgi.\n\n[1] https://review.opendev.org/#/c/661851/\n\nChange-Id: I9aaee4c2932f1e4ea9fe780a64e96a28fa6bccfb\nStory: 2005920\nTask: 34181\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f5a09567368e47aa6df861b0870a91028bc34f05"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f5a09567368e47aa6df861b0870a91028bc34f05"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
