)]}'
{"id":"openstack%2Fkolla~667153","triplet_id":"openstack%2Fkolla~master~Ib82810a8d33dd91d28537058e873724a22067103","project":"openstack/kolla","branch":"master","topic":"fix-openstack-exporter","hashtags":[],"change_id":"Ib82810a8d33dd91d28537058e873724a22067103","subject":"[prometheus/openstack-exporter] Changed repository location and version.","status":"MERGED","created":"2019-06-24 16:29:51.000000000","updated":"2019-06-28 17:27:47.000000000","submitted":"2019-06-28 17:27:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"667153-1561742867902-1ef4935a","meta_rev_id":"289bcd615840cdbb8e2b5c119c2f787d027bc3cd","_number":667153,"virtual_id_number":667153,"owner":{"_account_id":2276,"name":"Jorge Niedbalski","email":"jorge.niedbalski@canonical.com","username":"niedbalski"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":2276,"name":"Jorge Niedbalski","email":"jorge.niedbalski@canonical.com","username":"niedbalski"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-06-28 17:27:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-06-28 15:14:01.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":2276,"name":"Jorge Niedbalski","email":"jorge.niedbalski@canonical.com","username":"niedbalski"},{"value":2,"date":"2019-06-25 13:57:02.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2019-06-25 17:37:46.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":2276,"name":"Jorge Niedbalski","email":"jorge.niedbalski@canonical.com","username":"niedbalski"},{"value":1,"date":"2019-06-28 14:36:31.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":2276,"name":"Jorge Niedbalski","email":"jorge.niedbalski@canonical.com","username":"niedbalski"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"date":"2019-06-25 13:50:55.000000000","_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":2276,"name":"Jorge Niedbalski","email":"jorge.niedbalski@canonical.com","username":"niedbalski"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2276,"name":"Jorge Niedbalski","email":"jorge.niedbalski@canonical.com","username":"niedbalski"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-24 16:30:11.000000000","updated_by":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"reviewer":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"state":"REVIEWER"},{"updated":"2019-06-25 17:37:46.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-06-28 14:36:31.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-06-28 15:14:01.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-06-28 17:27:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0f8be057c751b3366cbafbd979e091b74c2ac9dd","author":{"_account_id":2276,"name":"Jorge Niedbalski","email":"jorge.niedbalski@canonical.com","username":"niedbalski"},"date":"2019-06-24 16:29:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ad45f77699a8a9fc7dbee4d9b8f55134bc908d7f","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-24 17:13:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/53/667153/1/openstack-check/tripleo-build-containers-fedora-28-master/87dd88e/ : FAILURE in 6m 09s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/53/667153/1/openstack-check/tripleo-build-containers-fedora-28-stein/649a585/ : SUCCESS in 41m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"72c07d9e46ea3fa4f108a3c524c930c441913a25","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-24 18:47:32.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/53/667153/1/check/kolla-build-centos-binary/6b29396/ : SUCCESS in 39m 29s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/53/667153/1/check/kolla-build-centos-source/f7e65b9/ : SUCCESS in 1h 32m 26s\n- kolla-ansible-centos-source http://logs.openstack.org/53/667153/1/check/kolla-ansible-centos-source/582bc91/ : SUCCESS in 1h 56m 34s\n- kolla-ansible-centos-binary http://logs.openstack.org/53/667153/1/check/kolla-ansible-centos-binary/f3d2a47/ : SUCCESS in 1h 03m 17s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/53/667153/1/check/kolla-ansible-centos-source-upgrade/0946479/ : TIMED_OUT in 2h 01m 09s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/53/667153/1/check/tripleo-build-containers-centos-7/1ef6890/ : SUCCESS in 50m 16s\n- openstack-tox-lower-constraints http://logs.openstack.org/53/667153/1/check/openstack-tox-lower-constraints/e029cd6/ : SUCCESS in 3m 51s\n- openstack-tox-pep8 http://logs.openstack.org/53/667153/1/check/openstack-tox-pep8/f8142e1/ : SUCCESS in 3m 13s\n- openstack-tox-py27 http://logs.openstack.org/53/667153/1/check/openstack-tox-py27/2319d98/ : SUCCESS in 3m 48s\n- openstack-tox-py36 http://logs.openstack.org/53/667153/1/check/openstack-tox-py36/bf186bc/ : SUCCESS in 3m 36s\n- openstack-tox-py37 http://logs.openstack.org/53/667153/1/check/openstack-tox-py37/9a35f10/ : SUCCESS in 3m 30s\n- openstack-tox-docs http://logs.openstack.org/53/667153/1/check/openstack-tox-docs/2bc0afe/html/ : SUCCESS in 2m 51s\n- kolla-build-ubuntu-binary http://logs.openstack.org/53/667153/1/check/kolla-build-ubuntu-binary/cdd7dab/ : SUCCESS in 25m 17s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/53/667153/1/check/kolla-build-ubuntu-source/0099b42/ : SUCCESS in 28m 50s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/53/667153/1/check/kolla-ansible-ubuntu-source/ced442a/ : SUCCESS in 46m 44s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/53/667153/1/check/kolla-ansible-ubuntu-binary/c809402/ : SUCCESS in 52m 54s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/53/667153/1/check/kolla-ansible-ubuntu-source-upgrade/7b1a4e6/ : SUCCESS in 55m 38s","accounts_in_message":[],"_revision_number":1},{"id":"20d970a79b189ef1cc770a121d1066db29c1ac87","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-25 07:47:14.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6f2b008d81bcba53043c8060eabda00efffb31a1","author":{"_account_id":2276,"name":"Jorge Niedbalski","email":"jorge.niedbalski@canonical.com","username":"niedbalski"},"date":"2019-06-25 13:50:48.000000000","message":"Patch Set 1:\n\n\u003e (1 comment)\n\nYes, no compat changes here, just increased the version to the latest bug fix release.","accounts_in_message":[],"_revision_number":1},{"id":"a3b0dd9d69c4f658f5205f5a7e527c876534c2e8","author":{"_account_id":2276,"name":"Jorge Niedbalski","email":"jorge.niedbalski@canonical.com","username":"niedbalski"},"date":"2019-06-25 13:50:55.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"19989351b345de3e22be723f8056bcaf8a75c082","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-25 13:57:02.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"b042555d2d23718685eea334f69627d926e3fdc1","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-25 14:27:52.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/53/667153/2/openstack-check/tripleo-build-containers-fedora-28-master/48986cb/ : FAILURE in 6m 30s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/53/667153/2/openstack-check/tripleo-build-containers-fedora-28-stein/317fabc/ : SUCCESS in 35m 31s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e7927f2add3a5bf7b1348227f75f7737aabb67a9","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-25 17:37:46.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"0d7cbee462c4eb060d7c0efd8432d290a0a9788f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-25 19:26:02.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/53/667153/2/check/kolla-build-centos-binary/5af3912/ : SUCCESS in 40m 08s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/53/667153/2/check/kolla-build-centos-source/851a182/ : SUCCESS in 2h 21m 00s\n- kolla-ansible-centos-source http://logs.openstack.org/53/667153/2/check/kolla-ansible-centos-source/6ee8e1f/ : TIMED_OUT in 2h 01m 07s\n- kolla-ansible-centos-binary http://logs.openstack.org/53/667153/2/check/kolla-ansible-centos-binary/1101f59/ : SUCCESS in 1h 15m 25s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/53/667153/2/check/kolla-ansible-centos-source-upgrade/f79da3a/ : TIMED_OUT in 2h 00m 56s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/53/667153/2/check/tripleo-build-containers-centos-7/8440ba3/ : SUCCESS in 37m 03s\n- openstack-tox-lower-constraints http://logs.openstack.org/53/667153/2/check/openstack-tox-lower-constraints/a7987f5/ : SUCCESS in 4m 08s\n- openstack-tox-pep8 http://logs.openstack.org/53/667153/2/check/openstack-tox-pep8/b0b7f6b/ : SUCCESS in 3m 28s\n- openstack-tox-py27 http://logs.openstack.org/53/667153/2/check/openstack-tox-py27/44bc6ef/ : SUCCESS in 4m 19s\n- openstack-tox-py36 http://logs.openstack.org/53/667153/2/check/openstack-tox-py36/15804bc/ : SUCCESS in 3m 41s\n- openstack-tox-py37 http://logs.openstack.org/53/667153/2/check/openstack-tox-py37/a8ead90/ : SUCCESS in 4m 19s\n- openstack-tox-docs http://logs.openstack.org/53/667153/2/check/openstack-tox-docs/7e6902b/html/ : SUCCESS in 2m 45s\n- kolla-build-ubuntu-binary http://logs.openstack.org/53/667153/2/check/kolla-build-ubuntu-binary/cb852e9/ : SUCCESS in 29m 00s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/53/667153/2/check/kolla-build-ubuntu-source/46f26d2/ : SUCCESS in 29m 07s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/53/667153/2/check/kolla-ansible-ubuntu-source/82aaff2/ : SUCCESS in 46m 52s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/53/667153/2/check/kolla-ansible-ubuntu-binary/157b01e/ : SUCCESS in 56m 00s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/53/667153/2/check/kolla-ansible-ubuntu-source-upgrade/6604ba6/ : SUCCESS in 52m 41s","accounts_in_message":[],"_revision_number":2},{"id":"87f151fb4f5d24210ddd91692dc3bae203f52b22","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-28 14:36:31.000000000","message":"Patch Set 2: Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"c31b8ecc924e26b523017db2471fcb841dd72d29","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-28 15:14:01.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/53/667153/2/openstack-check/tripleo-build-containers-fedora-28-master/e0d1a4c/ : FAILURE in 9m 01s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/53/667153/2/openstack-check/tripleo-build-containers-fedora-28-stein/4bb4ec0/ : SUCCESS in 35m 05s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e1225f93bb8e299017a659eb1ddf4d3b702e000f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-28 16:22:36.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/53/667153/2/check/kolla-build-centos-binary/8d8b571/ : SUCCESS in 41m 56s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/53/667153/2/check/kolla-build-centos-source/44b6d32/ : SUCCESS in 47m 55s\n- kolla-ansible-centos-source http://logs.openstack.org/53/667153/2/check/kolla-ansible-centos-source/94a3921/ : SUCCESS in 1h 18m 00s\n- kolla-ansible-centos-binary http://logs.openstack.org/53/667153/2/check/kolla-ansible-centos-binary/1c13024/ : SUCCESS in 1h 03m 59s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/53/667153/2/check/kolla-ansible-centos-source-upgrade/99faa48/ : SUCCESS in 1h 22m 46s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/53/667153/2/check/tripleo-build-containers-centos-7/0b17708/ : SUCCESS in 48m 08s\n- openstack-tox-lower-constraints http://logs.openstack.org/53/667153/2/check/openstack-tox-lower-constraints/7d516dc/ : SUCCESS in 3m 40s\n- openstack-tox-pep8 http://logs.openstack.org/53/667153/2/check/openstack-tox-pep8/419514d/ : SUCCESS in 3m 11s\n- openstack-tox-py27 http://logs.openstack.org/53/667153/2/check/openstack-tox-py27/add71db/ : SUCCESS in 3m 07s\n- openstack-tox-py36 http://logs.openstack.org/53/667153/2/check/openstack-tox-py36/7695204/ : SUCCESS in 3m 27s\n- openstack-tox-py37 http://logs.openstack.org/53/667153/2/check/openstack-tox-py37/bc03c04/ : SUCCESS in 4m 56s\n- openstack-tox-docs http://logs.openstack.org/53/667153/2/check/openstack-tox-docs/2f782a2/html/ : SUCCESS in 2m 49s\n- kolla-build-ubuntu-binary http://logs.openstack.org/53/667153/2/check/kolla-build-ubuntu-binary/075e703/ : SUCCESS in 24m 41s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/53/667153/2/check/kolla-build-ubuntu-source/3c60625/ : SUCCESS in 27m 26s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/53/667153/2/check/kolla-ansible-ubuntu-source/f7db45f/ : SUCCESS in 45m 45s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/53/667153/2/check/kolla-ansible-ubuntu-binary/38e6e4a/ : SUCCESS in 51m 22s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/53/667153/2/check/kolla-ansible-ubuntu-source-upgrade/39d7ce8/ : SUCCESS in 1h 19m 08s","accounts_in_message":[],"_revision_number":2},{"id":"ae5f2763ca7d5cba76572d65b1eb8b470d2b5ea1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-28 16:22:49.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a4dd0efd447f9dcc98f3f3259cbba819e079a343","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-28 17:27:47.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source http://logs.openstack.org/53/667153/2/gate/kolla-build-centos-source/b9565ff/ : SUCCESS in 57m 18s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/53/667153/2/gate/tripleo-build-containers-centos-7/aa2a31a/ : SUCCESS in 46m 34s\n- openstack-tox-lower-constraints http://logs.openstack.org/53/667153/2/gate/openstack-tox-lower-constraints/2fa4bf4/ : SUCCESS in 3m 56s\n- openstack-tox-pep8 http://logs.openstack.org/53/667153/2/gate/openstack-tox-pep8/c6055c9/ : SUCCESS in 3m 27s\n- openstack-tox-py27 http://logs.openstack.org/53/667153/2/gate/openstack-tox-py27/13a665c/ : SUCCESS in 3m 30s\n- openstack-tox-py36 http://logs.openstack.org/53/667153/2/gate/openstack-tox-py36/3d46646/ : SUCCESS in 3m 06s\n- openstack-tox-py37 http://logs.openstack.org/53/667153/2/gate/openstack-tox-py37/a6a1e57/ : SUCCESS in 3m 36s\n- openstack-tox-docs http://logs.openstack.org/53/667153/2/gate/openstack-tox-docs/2e6704a/html/ : SUCCESS in 2m 59s\n- kolla-build-ubuntu-source http://logs.openstack.org/53/667153/2/gate/kolla-build-ubuntu-source/0067cf7/ : SUCCESS in 23m 08s","accounts_in_message":[],"_revision_number":2},{"id":"11c700fa36a903096f3c48eb6392bee00fe18c00","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-28 17:27:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b692e8215b02648dbfd33af3e586a81da7695ea2","revisions":{"a3125ef8aed5f1bf300edc1a2be675b622b12dcf":{"kind":"REWORK","_number":1,"created":"2019-06-24 16:29:51.000000000","uploader":{"_account_id":2276,"name":"Jorge Niedbalski","email":"jorge.niedbalski@canonical.com","username":"niedbalski"},"ref":"refs/changes/53/667153/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/53/667153/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/53/667153/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/53/667153/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/53/667153/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/53/667153/1"}}},"commit":{"parents":[{"commit":"a7f9817a3e9a104660f54e57c85a0dae77ec2572","subject":"Merge \"Add a variable to configure repos disabled in base CentOS image\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a7f9817a3e9a104660f54e57c85a0dae77ec2572"}]}],"author":{"name":"Jorge Niedbalski","email":"jnr@metaklass.org","date":"2019-06-24 16:21:28.000000000","tz":-240},"committer":{"name":"Jorge Niedbalski","email":"niedbalski@gmail.com","date":"2019-06-24 16:29:34.000000000","tz":-240},"subject":"[prometheus/openstack-exporter] Changed repository location and version.","message":"[prometheus/openstack-exporter] Changed repository location and version.\n\nVersion 0.0.9 has been released, also the default organization\nhas been changed to its own org[0].\n\n[0] https://github.com/openstack-exporter/openstack-exporter\n\nChange-Id: Ib82810a8d33dd91d28537058e873724a22067103\nSigned-off-by: Jorge Niedbalski \u003cniedbalski@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a3125ef8aed5f1bf300edc1a2be675b622b12dcf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a3125ef8aed5f1bf300edc1a2be675b622b12dcf"}]},"branch":"refs/heads/master"},"b692e8215b02648dbfd33af3e586a81da7695ea2":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-06-25 13:50:55.000000000","uploader":{"_account_id":2276,"name":"Jorge Niedbalski","email":"jorge.niedbalski@canonical.com","username":"niedbalski"},"ref":"refs/changes/53/667153/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/53/667153/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/53/667153/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/53/667153/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/53/667153/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/53/667153/2"}}},"commit":{"parents":[{"commit":"2777a0c0584e4c707c56ca6aaa8cf2f41a473a83","subject":"Merge \"Make SKIPPED_IMAGES more flexible\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2777a0c0584e4c707c56ca6aaa8cf2f41a473a83"}]}],"author":{"name":"Jorge Niedbalski","email":"jnr@metaklass.org","date":"2019-06-24 16:21:28.000000000","tz":-240},"committer":{"name":"Jorge Niedbalski","email":"jorge.niedbalski@canonical.com","date":"2019-06-25 13:50:55.000000000","tz":0},"subject":"[prometheus/openstack-exporter] Changed repository location and version.","message":"[prometheus/openstack-exporter] Changed repository location and version.\n\nVersion 0.0.9 has been released, also the default organization\nhas been changed to its own org[0].\n\n[0] https://github.com/openstack-exporter/openstack-exporter\n\nChange-Id: Ib82810a8d33dd91d28537058e873724a22067103\nSigned-off-by: Jorge Niedbalski \u003cniedbalski@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b692e8215b02648dbfd33af3e586a81da7695ea2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b692e8215b02648dbfd33af3e586a81da7695ea2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
