)]}'
{"id":"openstack%2Fkolla~667204","triplet_id":"openstack%2Fkolla~master~I737f80566dd263ed291e215968ec72a95ad41a5f","project":"openstack/kolla","branch":"master","topic":"speed-up-source-builds","hashtags":[],"change_id":"I737f80566dd263ed291e215968ec72a95ad41a5f","subject":"Speed up builds of horizon slowed down by recent upstream change","status":"MERGED","created":"2019-06-24 19:39:13.000000000","updated":"2019-06-26 23:21:29.000000000","submitted":"2019-06-26 23:21:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"667204-1561591289128-0124e28e","meta_rev_id":"f0e2da3dd3b26d2411762a4267eeb5c2d5c4bb66","_number":667204,"virtual_id_number":667204,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":2,"date":"2019-06-26 23:21:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-06-26 21:36:18.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":2,"date":"2019-06-26 19:48:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2019-06-26 19:43:56.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":1,"date":"2019-06-26 19:48:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"date":"2019-06-26 19:49:23.000000000","_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"date":"2019-06-26 19:49:15.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-26 19:43:56.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-06-26 19:48:42.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2019-06-26 19:49:15.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-06-26 19:49:23.000000000","updated_by":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"reviewer":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"state":"REVIEWER"},{"updated":"2019-06-26 21:36:18.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-06-26 23:21:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"041ff4711f5e51637429385bfa287142931b191a","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-24 19:39:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c9ec5d3f25e0e62737e3b0499b42608d26400876","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-24 19:41:26.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"d7c19f051596d4fadeac265103b9cb1cad07821c","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-24 19:46:05.000000000","message":"Patch Set 2:\n\nI could not find why it does not affect binary builds.\nThey use manage.py as source builds used to, yet do not run extra turns.","accounts_in_message":[],"_revision_number":2},{"id":"2d9321465aaa09f628bfe6fabd575cc46cb410e6","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-24 20:26:42.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/04/667204/2/openstack-check/tripleo-build-containers-fedora-28-master/86e0564/ : FAILURE in 5m 50s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/04/667204/2/openstack-check/tripleo-build-containers-fedora-28-stein/b76f15b/ : SUCCESS in 40m 50s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"22572238c839182c4fa2e0dec944f7fb419f5ad1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-24 21:13:51.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/04/667204/2/check/kolla-build-centos-binary/d0b80ba/ : SUCCESS in 38m 41s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/04/667204/2/check/kolla-build-centos-source/f6e57c5/ : SUCCESS in 51m 50s\n- kolla-ansible-centos-source http://logs.openstack.org/04/667204/2/check/kolla-ansible-centos-source/7d73385/ : SUCCESS in 1h 11m 26s\n- kolla-ansible-centos-binary http://logs.openstack.org/04/667204/2/check/kolla-ansible-centos-binary/b3e41e5/ : SUCCESS in 1h 10m 18s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/04/667204/2/check/kolla-ansible-centos-source-upgrade/2101562/ : SUCCESS in 1h 26m 39s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/04/667204/2/check/tripleo-build-containers-centos-7/a7e443f/ : SUCCESS in 44m 33s\n- openstack-tox-lower-constraints http://logs.openstack.org/04/667204/2/check/openstack-tox-lower-constraints/00a4a30/ : SUCCESS in 3m 54s\n- openstack-tox-pep8 http://logs.openstack.org/04/667204/2/check/openstack-tox-pep8/fbd31bc/ : SUCCESS in 3m 11s\n- openstack-tox-py27 http://logs.openstack.org/04/667204/2/check/openstack-tox-py27/a91d47b/ : SUCCESS in 3m 31s\n- openstack-tox-py36 http://logs.openstack.org/04/667204/2/check/openstack-tox-py36/68a030a/ : SUCCESS in 3m 28s\n- openstack-tox-py37 http://logs.openstack.org/04/667204/2/check/openstack-tox-py37/09fd0f6/ : SUCCESS in 3m 49s\n- openstack-tox-docs http://logs.openstack.org/04/667204/2/check/openstack-tox-docs/9e13020/html/ : SUCCESS in 2m 50s\n- kolla-build-ubuntu-binary http://logs.openstack.org/04/667204/2/check/kolla-build-ubuntu-binary/5ce584a/ : SUCCESS in 27m 27s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/04/667204/2/check/kolla-build-ubuntu-source/2f881de/ : SUCCESS in 30m 34s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/04/667204/2/check/kolla-ansible-ubuntu-source/4243669/ : SUCCESS in 45m 44s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/04/667204/2/check/kolla-ansible-ubuntu-binary/c2f6cf5/ : SUCCESS in 46m 35s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/04/667204/2/check/kolla-ansible-ubuntu-source-upgrade/7e4f033/ : SUCCESS in 50m 01s","accounts_in_message":[],"_revision_number":2},{"id":"495bd44bc20d033f8bdf3fcea53d908a4958e6fd","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-25 06:45:01.000000000","message":"Patch Set 2: Workflow-1\n\nWe have been hit by https://opendev.org/openstack/horizon/commit/4e911e2889ebe7f0a577a0323649dceb9cef363c\nBinary builds probably have not incorporated it yet.","accounts_in_message":[],"_revision_number":2},{"id":"c45a25003efa302cbe353a7b1b5b4a86ff806e23","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-25 09:19:41.000000000","message":"Patch Set 2:\n\nPre the affecting horizon commit: http://logs.openstack.org/78/659678/3/check/kolla-build-centos-source/512a3ed/logs/build/horizon.txt.gz\nPost the affecting horizon commit: http://logs.openstack.org/15/666015/3/gate/kolla-build-centos-source/423deec/logs/build/horizon.txt.gz","accounts_in_message":[],"_revision_number":2},{"id":"4690a5dd51276a8704d748c9d5be5fe16c1c4901","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-25 09:56:14.000000000","message":"Patch Set 2:\n\nInformed horizon on IRC and https://bugs.launchpad.net/horizon/+bug/1834167","accounts_in_message":[],"_revision_number":2},{"id":"dc529a2280c16ee8b5f7bf0d89f120ef8a20e29e","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-25 16:07:22.000000000","message":"Abandoned\n\nobsoleted by: https://review.opendev.org/667386","accounts_in_message":[],"_revision_number":2},{"id":"97ac6ecd7c96974c92c7652b5e5142fb66cc348e","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-26 17:26:43.000000000","message":"Restored\n\nThis might be the best choice for now.","accounts_in_message":[],"_revision_number":2},{"id":"79b4c2386ded124aee5552fe39f1a30298136d52","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-26 17:38:51.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"ff2f43d18d2522be2c5c4d8d37d0b10141a9fa24","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-26 17:45:24.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"625daeab3159c94282da6e94a4fa135039956a5a","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-26 18:21:26.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/04/667204/4/openstack-check/tripleo-build-containers-fedora-28-master/9ccd4ed/ : FAILURE in 5m 00s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/04/667204/4/openstack-check/tripleo-build-containers-fedora-28-stein/16d7250/ : SUCCESS in 34m 41s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"6e5b17b81531fe46c169654d9a91aee61c09a29e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-26 19:20:26.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/04/667204/4/check/kolla-build-centos-binary/f4ccf09/ : SUCCESS in 46m 02s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/04/667204/4/check/kolla-build-centos-source/714c9e2/ : SUCCESS in 50m 35s\n- kolla-ansible-centos-source http://logs.openstack.org/04/667204/4/check/kolla-ansible-centos-source/6d0e6f1/ : SUCCESS in 1h 12m 35s\n- kolla-ansible-centos-binary http://logs.openstack.org/04/667204/4/check/kolla-ansible-centos-binary/0c87e17/ : SUCCESS in 1h 09m 55s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/04/667204/4/check/kolla-ansible-centos-source-upgrade/618b235/ : SUCCESS in 1h 21m 47s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/04/667204/4/check/tripleo-build-containers-centos-7/35d0bd1/ : SUCCESS in 49m 07s\n- openstack-tox-lower-constraints http://logs.openstack.org/04/667204/4/check/openstack-tox-lower-constraints/888ad03/ : SUCCESS in 3m 58s\n- openstack-tox-pep8 http://logs.openstack.org/04/667204/4/check/openstack-tox-pep8/00a6eaf/ : SUCCESS in 3m 17s\n- openstack-tox-py27 http://logs.openstack.org/04/667204/4/check/openstack-tox-py27/ae209a7/ : SUCCESS in 2m 59s\n- openstack-tox-py36 http://logs.openstack.org/04/667204/4/check/openstack-tox-py36/643bb5c/ : SUCCESS in 3m 07s\n- openstack-tox-py37 http://logs.openstack.org/04/667204/4/check/openstack-tox-py37/2cc26d6/ : SUCCESS in 3m 18s\n- openstack-tox-docs http://logs.openstack.org/04/667204/4/check/openstack-tox-docs/3bf0e5a/html/ : SUCCESS in 2m 35s\n- kolla-build-ubuntu-binary http://logs.openstack.org/04/667204/4/check/kolla-build-ubuntu-binary/36eb8ff/ : SUCCESS in 32m 14s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/04/667204/4/check/kolla-build-ubuntu-source/fb90f85/ : SUCCESS in 30m 38s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/04/667204/4/check/kolla-ansible-ubuntu-source/3e91182/ : SUCCESS in 50m 36s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/04/667204/4/check/kolla-ansible-ubuntu-binary/687b01e/ : SUCCESS in 45m 02s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/04/667204/4/check/kolla-ansible-ubuntu-source-upgrade/f9f7ad4/ : SUCCESS in 59m 16s","accounts_in_message":[],"_revision_number":4},{"id":"0563aff6d384c50474237b1210505983af5272b8","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-26 19:43:56.000000000","message":"Patch Set 4: Code-Review+2\n\nLGTM whatever works as we need to finally get it working","accounts_in_message":[],"_revision_number":4},{"id":"7c27f5db97308b9b3b8f385c444000fa7468228c","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-06-26 19:48:42.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"7b050ea20721bd03cf5f3920c66ab02e1788d031","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-26 19:48:56.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"dd751d840b7530c947ae0ed64a14fec23fef5074","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-26 20:47:03.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-source http://logs.openstack.org/04/667204/4/gate/kolla-build-centos-source/cb84814/ : FAILURE in 45m 55s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/04/667204/4/gate/tripleo-build-containers-centos-7/fb9b606/ : SUCCESS in 51m 59s\n- openstack-tox-lower-constraints http://logs.openstack.org/04/667204/4/gate/openstack-tox-lower-constraints/7ccb639/ : SUCCESS in 3m 11s\n- openstack-tox-pep8 http://logs.openstack.org/04/667204/4/gate/openstack-tox-pep8/fda193e/ : SUCCESS in 3m 13s\n- openstack-tox-py27 http://logs.openstack.org/04/667204/4/gate/openstack-tox-py27/f460c9d/ : SUCCESS in 3m 23s\n- openstack-tox-py36 http://logs.openstack.org/04/667204/4/gate/openstack-tox-py36/da2a188/ : SUCCESS in 3m 11s\n- openstack-tox-py37 http://logs.openstack.org/04/667204/4/gate/openstack-tox-py37/5adbfeb/ : SUCCESS in 3m 30s\n- openstack-tox-docs http://logs.openstack.org/04/667204/4/gate/openstack-tox-docs/fe4f13e/html/ : SUCCESS in 3m 02s\n- kolla-build-ubuntu-source http://logs.openstack.org/04/667204/4/gate/kolla-build-ubuntu-source/b5ad4ff/ : FAILURE in 22m 39s","accounts_in_message":[],"_revision_number":4},{"id":"f6cf486c55995a4d8a95bff28c36f804f988a840","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-26 21:01:29.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"8eec7293719f17bb62368c5fb1b2bf3f9f936c73","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-26 21:36:18.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/04/667204/4/openstack-check/tripleo-build-containers-fedora-28-master/eb88dce/ : FAILURE in 6m 02s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/04/667204/4/openstack-check/tripleo-build-containers-fedora-28-stein/fd696fe/ : SUCCESS in 34m 00s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"2c07cf43bea2a6087f868daf53c9a138fff331f2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-26 22:27:10.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/04/667204/4/check/kolla-build-centos-binary/cbe2042/ : SUCCESS in 44m 30s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/04/667204/4/check/kolla-build-centos-source/b3f282e/ : SUCCESS in 51m 23s\n- kolla-ansible-centos-source http://logs.openstack.org/04/667204/4/check/kolla-ansible-centos-source/04244b1/ : SUCCESS in 1h 17m 59s\n- kolla-ansible-centos-binary http://logs.openstack.org/04/667204/4/check/kolla-ansible-centos-binary/09c7cf2/ : SUCCESS in 1h 04m 08s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/04/667204/4/check/kolla-ansible-centos-source-upgrade/bc309be/ : SUCCESS in 1h 19m 19s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/04/667204/4/check/tripleo-build-containers-centos-7/d5e3396/ : SUCCESS in 50m 55s\n- openstack-tox-lower-constraints http://logs.openstack.org/04/667204/4/check/openstack-tox-lower-constraints/aba152b/ : SUCCESS in 4m 17s\n- openstack-tox-pep8 http://logs.openstack.org/04/667204/4/check/openstack-tox-pep8/e0e1389/ : SUCCESS in 3m 27s\n- openstack-tox-py27 http://logs.openstack.org/04/667204/4/check/openstack-tox-py27/89348a3/ : SUCCESS in 3m 20s\n- openstack-tox-py36 http://logs.openstack.org/04/667204/4/check/openstack-tox-py36/4f04606/ : SUCCESS in 3m 07s\n- openstack-tox-py37 http://logs.openstack.org/04/667204/4/check/openstack-tox-py37/18799cf/ : SUCCESS in 3m 48s\n- openstack-tox-docs http://logs.openstack.org/04/667204/4/check/openstack-tox-docs/cdc7521/html/ : SUCCESS in 2m 49s\n- kolla-build-ubuntu-binary http://logs.openstack.org/04/667204/4/check/kolla-build-ubuntu-binary/9836261/ : SUCCESS in 23m 41s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/04/667204/4/check/kolla-build-ubuntu-source/57a7971/ : SUCCESS in 25m 45s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/04/667204/4/check/kolla-ansible-ubuntu-source/463cb14/ : SUCCESS in 46m 02s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/04/667204/4/check/kolla-ansible-ubuntu-binary/9d4f89e/ : SUCCESS in 43m 28s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/04/667204/4/check/kolla-ansible-ubuntu-source-upgrade/151d00c/ : SUCCESS in 1h 01m 22s","accounts_in_message":[],"_revision_number":4},{"id":"cd8c898f2a3d29a1c944c2cd6cfb34062f63e6ed","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-26 22:27:24.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"1fa110c38bdb4bf2e0994bfafd9fe1a9ede39132","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-26 23:21:28.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source http://logs.openstack.org/04/667204/4/gate/kolla-build-centos-source/4e53c8f/ : SUCCESS in 33m 08s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/04/667204/4/gate/tripleo-build-containers-centos-7/c2575ad/ : SUCCESS in 45m 56s\n- openstack-tox-lower-constraints http://logs.openstack.org/04/667204/4/gate/openstack-tox-lower-constraints/0a424db/ : SUCCESS in 3m 28s\n- openstack-tox-pep8 http://logs.openstack.org/04/667204/4/gate/openstack-tox-pep8/048933c/ : SUCCESS in 3m 21s\n- openstack-tox-py27 http://logs.openstack.org/04/667204/4/gate/openstack-tox-py27/b771e41/ : SUCCESS in 3m 05s\n- openstack-tox-py36 http://logs.openstack.org/04/667204/4/gate/openstack-tox-py36/3ce4f82/ : SUCCESS in 3m 19s\n- openstack-tox-py37 http://logs.openstack.org/04/667204/4/gate/openstack-tox-py37/2717e62/ : SUCCESS in 4m 12s\n- openstack-tox-docs http://logs.openstack.org/04/667204/4/gate/openstack-tox-docs/69402d0/html/ : SUCCESS in 2m 44s\n- kolla-build-ubuntu-source http://logs.openstack.org/04/667204/4/gate/kolla-build-ubuntu-source/8aaf412/ : SUCCESS in 38m 04s","accounts_in_message":[],"_revision_number":4},{"id":"036e06ca6515915c94d74135eb678045f080a160","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-26 23:21:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"b9591bdec9fe996b5072c2dea03313e5414be67d","revisions":{"d1011120e03bfd5130c0e8918cf88d9b6165d718":{"kind":"REWORK","_number":1,"created":"2019-06-24 19:39:13.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/04/667204/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/04/667204/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/04/667204/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/04/667204/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/04/667204/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/04/667204/1"}}},"commit":{"parents":[{"commit":"a7f9817a3e9a104660f54e57c85a0dae77ec2572","subject":"Merge \"Add a variable to configure repos disabled in base CentOS image\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a7f9817a3e9a104660f54e57c85a0dae77ec2572"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-06-24 19:29:57.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-06-24 19:36:29.000000000","tz":120},"subject":"Speed up source builds for horizon","message":"Speed up source builds for horizon\n\nSource builds of horizon needlessly ran many compilemessages\niterations for core horizon parts because of using horizon\nmanage.py which referenced openstack_dashboard.settings\nThis patch switches to django-admin for source builds.\n\nDue to manage.py not being run, the \"static\" directory\nmust be created here by us to prevent later breakage.\n\nChange-Id: I737f80566dd263ed291e215968ec72a95ad41a5f\nSigned-off-by: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d1011120e03bfd5130c0e8918cf88d9b6165d718"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d1011120e03bfd5130c0e8918cf88d9b6165d718"}]},"branch":"refs/heads/master"},"24c720fe7fffe5636e5f5cc06d9f59dd3eec2d8e":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-06-24 19:41:26.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/04/667204/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/04/667204/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/04/667204/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/04/667204/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/04/667204/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/04/667204/2"}}},"commit":{"parents":[{"commit":"a7f9817a3e9a104660f54e57c85a0dae77ec2572","subject":"Merge \"Add a variable to configure repos disabled in base CentOS image\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a7f9817a3e9a104660f54e57c85a0dae77ec2572"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-06-24 19:29:57.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-06-24 19:41:17.000000000","tz":0},"subject":"Speed up source builds for horizon","message":"Speed up source builds for horizon\n\nSource builds of horizon needlessly ran many compilemessages\niterations for core horizon parts because of using horizon\nmanage.py which referenced openstack_dashboard.settings\nIt caused CentOS builds to take a long time due to their\nslow gettext implementation which in turn caused failed jobs\nin CI.\n\nThis patch switches to django-admin for source builds.\n\nDue to manage.py not being run, the \"static\" directory\nmust be created here by us to prevent later breakage.\n\nChange-Id: I737f80566dd263ed291e215968ec72a95ad41a5f\nSigned-off-by: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/24c720fe7fffe5636e5f5cc06d9f59dd3eec2d8e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/24c720fe7fffe5636e5f5cc06d9f59dd3eec2d8e"}]},"branch":"refs/heads/master"},"44ddf2ff0e3db99659563bfbc3b5571149476ff5":{"kind":"REWORK","_number":3,"created":"2019-06-26 17:38:51.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/04/667204/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/04/667204/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/04/667204/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/04/667204/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/04/667204/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/04/667204/3"}}},"commit":{"parents":[{"commit":"a7f9817a3e9a104660f54e57c85a0dae77ec2572","subject":"Merge \"Add a variable to configure repos disabled in base CentOS image\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a7f9817a3e9a104660f54e57c85a0dae77ec2572"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-06-24 19:29:57.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-06-26 17:38:22.000000000","tz":0},"subject":"Speed up builds for horizon slowed down by recent upstream change","message":"Speed up builds for horizon slowed down by recent upstream change\n\nSource builds of horizon needlessly ran many compilemessages iterations\nfor core horizon parts because of using horizon\u0027s manage.py in a loop.\nThe behavior was caused by recent upstream change ([1]).\n\nIt caused CentOS source builds to take too long time and time out in CI.\n\nThis patch switches to using django-admin.\n\n[1] https://opendev.org/openstack/horizon/commit/4e911e2889ebe7f0a577a0323649dceb9cef363c\n\nChange-Id: I737f80566dd263ed291e215968ec72a95ad41a5f\nSigned-off-by: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/44ddf2ff0e3db99659563bfbc3b5571149476ff5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/44ddf2ff0e3db99659563bfbc3b5571149476ff5"}]},"branch":"refs/heads/master"},"b9591bdec9fe996b5072c2dea03313e5414be67d":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2019-06-26 17:45:24.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/04/667204/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/04/667204/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/04/667204/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/04/667204/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/04/667204/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/04/667204/4"}}},"commit":{"parents":[{"commit":"a7f9817a3e9a104660f54e57c85a0dae77ec2572","subject":"Merge \"Add a variable to configure repos disabled in base CentOS image\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a7f9817a3e9a104660f54e57c85a0dae77ec2572"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-06-24 19:29:57.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-06-26 17:45:22.000000000","tz":0},"subject":"Speed up builds of horizon slowed down by recent upstream change","message":"Speed up builds of horizon slowed down by recent upstream change\n\nBuilds of horizon needlessly ran many compilemessages iterations for\ncore horizon parts because of using horizon\u0027s manage.py in a loop.\nThe behavior was caused by recent upstream change ([1]).\n\nIt caused CentOS source builds to take too long time and time out in CI.\n\nThis patch switches to using django-admin.\n\n[1] https://opendev.org/openstack/horizon/commit/4e911e2889ebe7f0a577a0323649dceb9cef363c\n\nChange-Id: I737f80566dd263ed291e215968ec72a95ad41a5f\nSigned-off-by: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b9591bdec9fe996b5072c2dea03313e5414be67d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b9591bdec9fe996b5072c2dea03313e5414be67d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
