)]}'
{"id":"openstack%2Fkolla~667349","triplet_id":"openstack%2Fkolla~master~I5090eeb20c701a19f0b3f78cfbb44dc87df45948","project":"openstack/kolla","branch":"master","hashtags":[],"change_id":"I5090eeb20c701a19f0b3f78cfbb44dc87df45948","subject":"setup-gate: do not error if directory exists (CI can have it)","status":"ABANDONED","created":"2019-06-25 12:31:10.000000000","updated":"2019-06-28 06:30:29.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"32b6a1f157d913beade04c5957c0f0c16af82a22","_number":667349,"virtual_id_number":667349,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2019-06-25 15:33:55.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":-1,"date":"2019-06-25 14:03:09.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"disliked":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2019-06-25 12:41:30.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":-1,"date":"2019-06-25 15:30:51.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-25 12:41:30.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-06-25 14:03:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-06-25 15:30:51.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"}],"messages":[{"id":"8ab4818bbcc047203076148a3f2a624b6a6b80cf","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-25 12:31:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3761fb23852416e2f14833759543165fb0bab651","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-25 12:41:30.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"96312d7bb9e9a6a2d5c5eff66b1da91d94e5da90","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-25 14:03:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary finger://ze12.openstack.org/73082d90a1b3464fa107ccaf618be858 : RETRY_LIMIT in 40m 14s (non-voting)\n- kolla-build-centos-source finger://ze05.openstack.org/f4ace60c46a8457697bbf88ba7c6d3a5 : RETRY_LIMIT in 45m 35s\n- kolla-ansible-centos-source finger://ze02.openstack.org/1bc7e9c5bbeb4553aff45cb4fbc2732c : RETRY_LIMIT in 46m 04s\n- kolla-ansible-centos-binary finger://ze03.openstack.org/fbdc1b7f85984516b1b49458d5af20d1 : RETRY_LIMIT in 45m 54s (non-voting)\n- kolla-ansible-centos-source-upgrade finger://ze11.openstack.org/ab41e2d3e3234e4ab7c893abd2ec80fd : RETRY_LIMIT in 41m 31s\n- openstack-tox-lower-constraints http://logs.openstack.org/49/667349/1/check/openstack-tox-lower-constraints/a120a92/ : SUCCESS in 5m 58s\n- openstack-tox-pep8 http://logs.openstack.org/49/667349/1/check/openstack-tox-pep8/db52927/ : SUCCESS in 5m 59s\n- openstack-tox-py27 http://logs.openstack.org/49/667349/1/check/openstack-tox-py27/4904aae/ : SUCCESS in 5m 43s\n- openstack-tox-py36 http://logs.openstack.org/49/667349/1/check/openstack-tox-py36/329cce3/ : SUCCESS in 4m 53s\n- openstack-tox-py37 http://logs.openstack.org/49/667349/1/check/openstack-tox-py37/7ab4feb/ : SUCCESS in 5m 55s\n- openstack-tox-docs http://logs.openstack.org/49/667349/1/check/openstack-tox-docs/50bba8e/html/ : SUCCESS in 4m 53s\n- kolla-build-ubuntu-binary http://logs.openstack.org/49/667349/1/check/kolla-build-ubuntu-binary/9a13555/ : SUCCESS in 41m 16s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/49/667349/1/check/kolla-build-ubuntu-source/cb8d324/ : POST_FAILURE in 45m 26s\n- kolla-ansible-ubuntu-source finger://ze10.openstack.org/0cb37b3112814d9082d5db8a130b5e9e : RETRY_LIMIT in 42m 02s\n- kolla-ansible-ubuntu-binary finger://ze10.openstack.org/790d8861a88a4fd4955aedf1dfe801ce : RETRY_LIMIT in 45m 08s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade finger://ze09.openstack.org/865cf67aab6d4ec6acbe53100a111f5e : RETRY_LIMIT in 42m 53s","accounts_in_message":[],"_revision_number":1},{"id":"fb6b80fd0c3126d7a3a80d4eb4299a81a2bda600","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-25 14:27:41.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a7d88c7d290f0b36223f3b0cd7752c696fd78737","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-25 15:30:51.000000000","message":"Patch Set 1: Code-Review-1\n\nIt\u0027s not even used in CI. See https://review.opendev.org/664580\n\nWhere did it have it? :O","accounts_in_message":[],"_revision_number":1},{"id":"0491dccd7786fe98034cd2dde69be03eb64ea4f6","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-25 15:33:55.000000000","message":"Patch Set 1:\n\n\u003e It\u0027s not even used in CI. See https://review.opendev.org/664580\n \u003e \n \u003e Where did it have it? :O\n\nhttp://logs.openstack.org/08/665908/1/check/kolla-build-ubuntu-binary/886f200/ara-report/","accounts_in_message":[],"_revision_number":1},{"id":"2abcc7303078fc58c2a9899f9dca4acc549bae1e","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-25 15:39:52.000000000","message":"Patch Set 1:\n\n\u003e \u003e It\u0027s not even used in CI. See https://review.opendev.org/664580\n \u003e \u003e\n \u003e \u003e Where did it have it? :O\n \u003e \n \u003e http://logs.openstack.org/08/665908/1/check/kolla-build-ubuntu-binary/886f200/ara-report/\n\nOdd, compare ara with txt log and it looks as if Zuul ran two in parallel.","accounts_in_message":[],"_revision_number":1},{"id":"0b365c982f2871a443b4770aa0836c0cf74c0c14","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-25 17:14:58.000000000","message":"Patch Set 1:\n\n\u003e \u003e \u003e It\u0027s not even used in CI. See https://review.opendev.org/664580\n \u003e \u003e \u003e\n \u003e \u003e \u003e Where did it have it? :O\n \u003e \u003e\n \u003e \u003e http://logs.openstack.org/08/665908/1/check/kolla-build-ubuntu-binary/886f200/ara-report/\n \u003e \n \u003e Odd, compare ara with txt log and it looks as if Zuul ran two in\n \u003e parallel.\n\nI think zuul can support retries for jobs but it should not happen for us as we set \u0027attempts\u003d1\u0027 in the job config.","accounts_in_message":[],"_revision_number":1},{"id":"601ded824a74fa9cb9f8e38039cad4bade35cf17","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-25 17:23:26.000000000","message":"Patch Set 1:\n\n\u003e \u003e \u003e \u003e It\u0027s not even used in CI. See https://review.opendev.org/664580\n \u003e \u003e \u003e \u003e\n \u003e \u003e \u003e \u003e Where did it have it? :O\n \u003e \u003e \u003e\n \u003e \u003e \u003e http://logs.openstack.org/08/665908/1/check/kolla-build-ubuntu-binary/886f200/ara-report/\n \u003e \u003e\n \u003e \u003e Odd, compare ara with txt log and it looks as if Zuul ran two in\n \u003e \u003e parallel.\n \u003e \n \u003e I think zuul can support retries for jobs but it should not happen\n \u003e for us as we set \u0027attempts\u003d1\u0027 in the job config.\n\nInfra guys confirmed something wrong happened. Even with retries we should have separate logs.","accounts_in_message":[],"_revision_number":1},{"id":"fab55884b4ba88b44ecdb77f86388abaeffb49e8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-25 21:26:08.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/49/667349/1/check/kolla-build-centos-binary/aecaf88/ : SUCCESS in 38m 55s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/49/667349/1/check/kolla-build-centos-source/b404fc2/ : TIMED_OUT in 2h 53m 17s\n- kolla-ansible-centos-source http://logs.openstack.org/49/667349/1/check/kolla-ansible-centos-source/f461195/ : TIMED_OUT in 2h 00m 58s\n- kolla-ansible-centos-binary http://logs.openstack.org/49/667349/1/check/kolla-ansible-centos-binary/fa231a0/ : SUCCESS in 1h 14m 06s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/49/667349/1/check/kolla-ansible-centos-source-upgrade/16731e8/ : TIMED_OUT in 2h 01m 05s\n- openstack-tox-lower-constraints http://logs.openstack.org/49/667349/1/check/openstack-tox-lower-constraints/12b16e7/ : SUCCESS in 3m 53s\n- openstack-tox-pep8 http://logs.openstack.org/49/667349/1/check/openstack-tox-pep8/e2cfcab/ : SUCCESS in 3m 42s\n- openstack-tox-py27 http://logs.openstack.org/49/667349/1/check/openstack-tox-py27/b1215f3/ : SUCCESS in 3m 46s\n- openstack-tox-py36 http://logs.openstack.org/49/667349/1/check/openstack-tox-py36/40f0c4d/ : SUCCESS in 4m 49s\n- openstack-tox-py37 http://logs.openstack.org/49/667349/1/check/openstack-tox-py37/ba70e9c/ : SUCCESS in 3m 08s\n- openstack-tox-docs http://logs.openstack.org/49/667349/1/check/openstack-tox-docs/dde778c/html/ : SUCCESS in 4m 05s\n- kolla-build-ubuntu-binary http://logs.openstack.org/49/667349/1/check/kolla-build-ubuntu-binary/c1e9655/ : SUCCESS in 25m 44s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/49/667349/1/check/kolla-build-ubuntu-source/a0aba0f/ : SUCCESS in 25m 46s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/49/667349/1/check/kolla-ansible-ubuntu-source/b50df71/ : SUCCESS in 41m 57s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/49/667349/1/check/kolla-ansible-ubuntu-binary/eaf2a66/ : SUCCESS in 41m 34s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/49/667349/1/check/kolla-ansible-ubuntu-source-upgrade/642893d/ : SUCCESS in 58m 05s","accounts_in_message":[],"_revision_number":1},{"id":"32b6a1f157d913beade04c5957c0f0c16af82a22","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-28 06:30:29.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cf728a100a9d4137eab5c64f4134c04723627eac","revisions":{"cf728a100a9d4137eab5c64f4134c04723627eac":{"kind":"REWORK","_number":1,"created":"2019-06-25 12:31:10.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/49/667349/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/49/667349/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/49/667349/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/49/667349/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/49/667349/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/49/667349/1"}}},"commit":{"parents":[{"commit":"2777a0c0584e4c707c56ca6aaa8cf2f41a473a83","subject":"Merge \"Make SKIPPED_IMAGES more flexible\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2777a0c0584e4c707c56ca6aaa8cf2f41a473a83"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin@juszkiewicz.com.pl","date":"2019-06-25 12:30:33.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin@juszkiewicz.com.pl","date":"2019-06-25 12:30:33.000000000","tz":120},"subject":"setup-gate: do not error if directory exists (CI can have it)","message":"setup-gate: do not error if directory exists (CI can have it)\n\nChange-Id: I5090eeb20c701a19f0b3f78cfbb44dc87df45948\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cf728a100a9d4137eab5c64f4134c04723627eac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cf728a100a9d4137eab5c64f4134c04723627eac"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
