)]}'
{"id":"openstack%2Fkolla~667420","triplet_id":"openstack%2Fkolla~master~I527aba3635ef82e2d2a88ee7f5250bf03e420800","project":"openstack/kolla","branch":"master","topic":"ci-remove-ol-overrides","hashtags":[],"change_id":"I527aba3635ef82e2d2a88ee7f5250bf03e420800","subject":"Remove Oracle Linux stuff in CI code","status":"MERGED","created":"2019-06-25 17:42:25.000000000","updated":"2019-06-28 06:32:44.000000000","submitted":"2019-06-27 14:05:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"667420-1561644358592-663f6094","meta_rev_id":"c33b6fc18e94991e306cf03d56b9b7b471165296","_number":667420,"virtual_id_number":667420,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"date":"2019-06-27 09:52:58.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":2,"date":"2019-06-27 14:05:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"all":[{"value":2,"date":"2019-06-26 06:43:55.000000000","_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":2,"date":"2019-06-26 08:00:55.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":1,"date":"2019-06-26 08:00:55.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-26 06:43:55.000000000","updated_by":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"reviewer":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"state":"REVIEWER"},{"updated":"2019-06-26 08:00:55.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-06-27 14:05:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"774a9921bc0839dd6cbdb7b25864b428390c9efb","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-25 17:42:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"678dd6f4181b1d1b025a55ca6fa2297d061018c8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-25 20:51:01.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/20/667420/1/check/kolla-build-centos-binary/a7a97ad/ : SUCCESS in 34m 08s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/20/667420/1/check/kolla-build-centos-source/853cf67/ : SUCCESS in 47m 18s\n- kolla-ansible-centos-source http://logs.openstack.org/20/667420/1/check/kolla-ansible-centos-source/9c7b8c6/ : TIMED_OUT in 2h 01m 26s\n- kolla-ansible-centos-binary http://logs.openstack.org/20/667420/1/check/kolla-ansible-centos-binary/3f29e7a/ : SUCCESS in 1h 14m 18s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/20/667420/1/check/kolla-ansible-centos-source-upgrade/b410e3d/ : TIMED_OUT in 2h 01m 44s\n- openstack-tox-lower-constraints http://logs.openstack.org/20/667420/1/check/openstack-tox-lower-constraints/12e7a17/ : SUCCESS in 6m 16s\n- openstack-tox-pep8 http://logs.openstack.org/20/667420/1/check/openstack-tox-pep8/86c8bd2/ : SUCCESS in 3m 23s\n- openstack-tox-py27 http://logs.openstack.org/20/667420/1/check/openstack-tox-py27/6aad351/ : SUCCESS in 3m 30s\n- openstack-tox-py36 http://logs.openstack.org/20/667420/1/check/openstack-tox-py36/10176fa/ : SUCCESS in 3m 26s\n- openstack-tox-py37 http://logs.openstack.org/20/667420/1/check/openstack-tox-py37/5cd9250/ : SUCCESS in 3m 55s\n- openstack-tox-docs http://logs.openstack.org/20/667420/1/check/openstack-tox-docs/b38cf6b/html/ : SUCCESS in 2m 51s\n- kolla-build-ubuntu-binary http://logs.openstack.org/20/667420/1/check/kolla-build-ubuntu-binary/22c9ef2/ : SUCCESS in 21m 05s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/20/667420/1/check/kolla-build-ubuntu-source/226426d/ : SUCCESS in 29m 54s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/20/667420/1/check/kolla-ansible-ubuntu-source/b5ceac4/ : SUCCESS in 49m 19s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/20/667420/1/check/kolla-ansible-ubuntu-binary/4845f33/ : SUCCESS in 47m 50s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/20/667420/1/check/kolla-ansible-ubuntu-source-upgrade/db7faad/ : SUCCESS in 56m 25s","accounts_in_message":[],"_revision_number":1},{"id":"e76cd7e86f86cc4a94aaf3e3dc0085f6a8f9fd4f","author":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"date":"2019-06-26 06:43:55.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3b5baeef4633dbd5f69cd7a47b79b8cae7cd5037","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-26 08:00:55.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"be8c0ed782d231e27f1e2bdead0e2f76c70a467f","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-26 09:50:41.000000000","message":"Patch Set 1:\n\nrecheck\n\ngood luck","accounts_in_message":[],"_revision_number":1},{"id":"d035ccaf7347f3c7fc0e1d45eeced89492dddd75","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-26 11:26:04.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/20/667420/1/check/kolla-build-centos-binary/1ce3fd5/ : TIMED_OUT in 2h 51m 01s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/20/667420/1/check/kolla-build-centos-source/50671a4/ : SUCCESS in 2h 16m 19s\n- kolla-ansible-centos-source http://logs.openstack.org/20/667420/1/check/kolla-ansible-centos-source/f79b3b0/ : TIMED_OUT in 2h 01m 34s\n- kolla-ansible-centos-binary http://logs.openstack.org/20/667420/1/check/kolla-ansible-centos-binary/b144f5f/ : SUCCESS in 1h 11m 50s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/20/667420/1/check/kolla-ansible-centos-source-upgrade/6b5a263/ : TIMED_OUT in 2h 01m 03s\n- openstack-tox-lower-constraints http://logs.openstack.org/20/667420/1/check/openstack-tox-lower-constraints/bb6eda2/ : SUCCESS in 4m 53s\n- openstack-tox-pep8 http://logs.openstack.org/20/667420/1/check/openstack-tox-pep8/997a30e/ : SUCCESS in 3m 01s\n- openstack-tox-py27 http://logs.openstack.org/20/667420/1/check/openstack-tox-py27/00a4d9f/ : SUCCESS in 2m 58s\n- openstack-tox-py36 http://logs.openstack.org/20/667420/1/check/openstack-tox-py36/7412c15/ : SUCCESS in 4m 14s\n- openstack-tox-py37 http://logs.openstack.org/20/667420/1/check/openstack-tox-py37/183c5dc/ : SUCCESS in 3m 38s\n- openstack-tox-docs http://logs.openstack.org/20/667420/1/check/openstack-tox-docs/d28a39a/html/ : SUCCESS in 2m 53s\n- kolla-build-ubuntu-binary http://logs.openstack.org/20/667420/1/check/kolla-build-ubuntu-binary/139eff8/ : SUCCESS in 23m 55s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/20/667420/1/check/kolla-build-ubuntu-source/f13a314/ : SUCCESS in 30m 30s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/20/667420/1/check/kolla-ansible-ubuntu-source/f795b7b/ : SUCCESS in 45m 34s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/20/667420/1/check/kolla-ansible-ubuntu-binary/ec5561a/ : SUCCESS in 42m 05s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/20/667420/1/check/kolla-ansible-ubuntu-source-upgrade/ad1007d/ : SUCCESS in 1h 00m 01s","accounts_in_message":[],"_revision_number":1},{"id":"b650776176037bafda45be8b4017496d7069cf14","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-26 11:27:47.000000000","message":"Patch Set 1:\n\nrecheck\n\nrax mirror","accounts_in_message":[],"_revision_number":1},{"id":"3e8efec561bf7da58fdc02cb5eede384e7153d06","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-26 13:37:56.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/20/667420/1/check/kolla-build-centos-binary/70344a7/ : RETRY_LIMIT in 2m 48s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/20/667420/1/check/kolla-build-centos-source/f3d66ca/ : SUCCESS in 1h 34m 27s\n- kolla-ansible-centos-source http://logs.openstack.org/20/667420/1/check/kolla-ansible-centos-source/46e5449/ : TIMED_OUT in 2h 01m 04s\n- kolla-ansible-centos-binary http://logs.openstack.org/20/667420/1/check/kolla-ansible-centos-binary/9e1221f/ : RETRY_LIMIT in 3m 38s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/20/667420/1/check/kolla-ansible-centos-source-upgrade/e005a4b/ : TIMED_OUT in 2h 01m 12s\n- openstack-tox-lower-constraints http://logs.openstack.org/20/667420/1/check/openstack-tox-lower-constraints/a10193b/ : SUCCESS in 3m 39s\n- openstack-tox-pep8 http://logs.openstack.org/20/667420/1/check/openstack-tox-pep8/6b71f13/ : SUCCESS in 3m 40s\n- openstack-tox-py27 http://logs.openstack.org/20/667420/1/check/openstack-tox-py27/0538d77/ : SUCCESS in 4m 29s\n- openstack-tox-py36 http://logs.openstack.org/20/667420/1/check/openstack-tox-py36/1aafb50/ : SUCCESS in 3m 26s\n- openstack-tox-py37 http://logs.openstack.org/20/667420/1/check/openstack-tox-py37/34cfca4/ : SUCCESS in 3m 43s\n- openstack-tox-docs http://logs.openstack.org/20/667420/1/check/openstack-tox-docs/9715764/html/ : SUCCESS in 3m 40s\n- kolla-build-ubuntu-binary http://logs.openstack.org/20/667420/1/check/kolla-build-ubuntu-binary/e365e52/ : SUCCESS in 23m 03s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/20/667420/1/check/kolla-build-ubuntu-source/1237b09/ : SUCCESS in 32m 49s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/20/667420/1/check/kolla-ansible-ubuntu-source/542b425/ : SUCCESS in 48m 43s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/20/667420/1/check/kolla-ansible-ubuntu-binary/f794159/ : SUCCESS in 45m 18s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/20/667420/1/check/kolla-ansible-ubuntu-source-upgrade/862a474/ : SUCCESS in 52m 38s","accounts_in_message":[],"_revision_number":1},{"id":"33aa5e04f416ad468e5c7b80aea2dbd5a74640d6","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-27 06:27:06.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"524e1f961f8ee776984084c67b62df1b6ff07f0a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 08:47:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/20/667420/1/check/kolla-build-centos-binary/90951df/ : SUCCESS in 33m 03s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/20/667420/1/check/kolla-build-centos-source/c51c0fb/ : SUCCESS in 49m 18s\n- kolla-ansible-centos-source http://logs.openstack.org/20/667420/1/check/kolla-ansible-centos-source/1b59455/ : SUCCESS in 1h 15m 49s\n- kolla-ansible-centos-binary http://logs.openstack.org/20/667420/1/check/kolla-ansible-centos-binary/dfe9e55/ : SUCCESS in 59m 00s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/20/667420/1/check/kolla-ansible-centos-source-upgrade/9aadfcf/ : SUCCESS in 1h 15m 21s\n- openstack-tox-lower-constraints http://logs.openstack.org/20/667420/1/check/openstack-tox-lower-constraints/07cbafa/ : SUCCESS in 3m 11s\n- openstack-tox-pep8 http://logs.openstack.org/20/667420/1/check/openstack-tox-pep8/a6b9535/ : SUCCESS in 4m 11s\n- openstack-tox-py27 http://logs.openstack.org/20/667420/1/check/openstack-tox-py27/b4cd4ae/ : SUCCESS in 4m 08s\n- openstack-tox-py36 http://logs.openstack.org/20/667420/1/check/openstack-tox-py36/a2962be/ : SUCCESS in 3m 45s\n- openstack-tox-py37 http://logs.openstack.org/20/667420/1/check/openstack-tox-py37/f6ff7e8/ : SUCCESS in 3m 51s\n- openstack-tox-docs http://logs.openstack.org/20/667420/1/check/openstack-tox-docs/06ccb21/html/ : SUCCESS in 2m 52s\n- kolla-build-ubuntu-binary http://logs.openstack.org/20/667420/1/check/kolla-build-ubuntu-binary/a953ea0/ : SUCCESS in 52m 49s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/20/667420/1/check/kolla-build-ubuntu-source/0922c65/ : SUCCESS in 31m 46s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/20/667420/1/check/kolla-ansible-ubuntu-source/20845a3/ : SUCCESS in 51m 34s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/20/667420/1/check/kolla-ansible-ubuntu-binary/4797150/ : SUCCESS in 46m 36s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/20/667420/1/check/kolla-ansible-ubuntu-source-upgrade/73f568e/ : SUCCESS in 1h 05m 06s","accounts_in_message":[],"_revision_number":1},{"id":"c9a31d9b3ff889de9cbc4e189ed237e5637244ab","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 08:47:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4c11bde970430e769d7fd2f35cccea45ad9123b6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 09:51:49.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-source http://logs.openstack.org/20/667420/1/gate/kolla-build-centos-source/fc06c39/ : RETRY_LIMIT in 6m 35s\n- openstack-tox-lower-constraints http://logs.openstack.org/20/667420/1/gate/openstack-tox-lower-constraints/77de5b1/ : SUCCESS in 3m 14s\n- openstack-tox-pep8 http://logs.openstack.org/20/667420/1/gate/openstack-tox-pep8/ca799cb/ : SUCCESS in 3m 09s\n- openstack-tox-py27 http://logs.openstack.org/20/667420/1/gate/openstack-tox-py27/0b29d71/ : SUCCESS in 2m 52s\n- openstack-tox-py36 http://logs.openstack.org/20/667420/1/gate/openstack-tox-py36/a57fe64/ : SUCCESS in 3m 24s\n- openstack-tox-py37 http://logs.openstack.org/20/667420/1/gate/openstack-tox-py37/15b4895/ : SUCCESS in 3m 45s\n- openstack-tox-docs http://logs.openstack.org/20/667420/1/gate/openstack-tox-docs/3ae5b6a/html/ : SUCCESS in 3m 12s\n- kolla-build-ubuntu-source http://logs.openstack.org/20/667420/1/gate/kolla-build-ubuntu-source/eda2521/ : SUCCESS in 24m 21s","accounts_in_message":[],"_revision_number":1},{"id":"2f84fe3c2573e1659c0440677ed59f600ea78e9d","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-27 09:52:58.000000000","message":"Patch Set 1:\n\nrecheck\n\n...","accounts_in_message":[],"_revision_number":1},{"id":"437d25c4c2855b3f5c69f0e86a36845a5d97b5a5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 13:15:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/20/667420/1/check/kolla-build-centos-binary/47b801a/ : SUCCESS in 46m 40s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/20/667420/1/check/kolla-build-centos-source/1b539cd/ : SUCCESS in 54m 43s\n- kolla-ansible-centos-source http://logs.openstack.org/20/667420/1/check/kolla-ansible-centos-source/08f0bed/ : SUCCESS in 1h 12m 39s\n- kolla-ansible-centos-binary http://logs.openstack.org/20/667420/1/check/kolla-ansible-centos-binary/796da6b/ : SUCCESS in 1h 10m 07s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/20/667420/1/check/kolla-ansible-centos-source-upgrade/1fc0a72/ : SUCCESS in 1h 24m 00s\n- openstack-tox-lower-constraints http://logs.openstack.org/20/667420/1/check/openstack-tox-lower-constraints/2c54afe/ : SUCCESS in 3m 29s\n- openstack-tox-pep8 http://logs.openstack.org/20/667420/1/check/openstack-tox-pep8/2b53bf0/ : SUCCESS in 3m 40s\n- openstack-tox-py27 http://logs.openstack.org/20/667420/1/check/openstack-tox-py27/07114ba/ : SUCCESS in 3m 11s\n- openstack-tox-py36 http://logs.openstack.org/20/667420/1/check/openstack-tox-py36/9702699/ : SUCCESS in 3m 11s\n- openstack-tox-py37 http://logs.openstack.org/20/667420/1/check/openstack-tox-py37/a2769da/ : SUCCESS in 3m 17s\n- openstack-tox-docs http://logs.openstack.org/20/667420/1/check/openstack-tox-docs/f623272/html/ : SUCCESS in 2m 52s\n- kolla-build-ubuntu-binary http://logs.openstack.org/20/667420/1/check/kolla-build-ubuntu-binary/a8a4555/ : SUCCESS in 26m 33s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/20/667420/1/check/kolla-build-ubuntu-source/d1027a4/ : SUCCESS in 29m 35s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/20/667420/1/check/kolla-ansible-ubuntu-source/8dbe5c2/ : SUCCESS in 44m 52s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/20/667420/1/check/kolla-ansible-ubuntu-binary/b24328f/ : SUCCESS in 38m 15s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/20/667420/1/check/kolla-ansible-ubuntu-source-upgrade/2f7b5f2/ : SUCCESS in 52m 32s","accounts_in_message":[],"_revision_number":1},{"id":"efb70cce1995242203f89dba619ec450b7b8eda2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 13:16:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ff19b7662c0a2ff65863bd2e524908753957c387","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 14:05:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"563cc4c6f0c9368e9c136e984e9df03e5594be27","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 14:05:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source http://logs.openstack.org/20/667420/1/gate/kolla-build-centos-source/96232e1/ : SUCCESS in 44m 43s\n- openstack-tox-lower-constraints http://logs.openstack.org/20/667420/1/gate/openstack-tox-lower-constraints/c25f553/ : SUCCESS in 3m 52s\n- openstack-tox-pep8 http://logs.openstack.org/20/667420/1/gate/openstack-tox-pep8/0dada16/ : SUCCESS in 3m 04s\n- openstack-tox-py27 http://logs.openstack.org/20/667420/1/gate/openstack-tox-py27/cfdc91b/ : SUCCESS in 2m 56s\n- openstack-tox-py36 http://logs.openstack.org/20/667420/1/gate/openstack-tox-py36/68af813/ : SUCCESS in 3m 55s\n- openstack-tox-py37 http://logs.openstack.org/20/667420/1/gate/openstack-tox-py37/a53a16f/ : SUCCESS in 4m 09s\n- openstack-tox-docs http://logs.openstack.org/20/667420/1/gate/openstack-tox-docs/224b8da/html/ : SUCCESS in 2m 53s\n- kolla-build-ubuntu-source http://logs.openstack.org/20/667420/1/gate/kolla-build-ubuntu-source/aec1ec3/ : SUCCESS in 28m 39s","accounts_in_message":[],"_revision_number":1},{"id":"d756374a946cccad3fae9a71fea00ba55829bc81","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-28 06:32:44.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit 848a089485036b100fbfce817fc5e9ca4f4b0c29","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ccabadc342477b3c16b11ae52e0194c7916966c2","revisions":{"ccabadc342477b3c16b11ae52e0194c7916966c2":{"kind":"REWORK","_number":1,"created":"2019-06-25 17:42:25.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/20/667420/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/20/667420/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/20/667420/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/20/667420/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/20/667420/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/20/667420/1"}}},"commit":{"parents":[{"commit":"2777a0c0584e4c707c56ca6aaa8cf2f41a473a83","subject":"Merge \"Make SKIPPED_IMAGES more flexible\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2777a0c0584e4c707c56ca6aaa8cf2f41a473a83"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-06-25 17:29:46.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-06-25 17:29:46.000000000","tz":120},"subject":"Remove Oracle Linux stuff in CI code","message":"Remove Oracle Linux stuff in CI code\n\nOracle Linux is no longer tested via CI after [1].\nThis patch removes OL support from Zuul files.\n\n[1] https://review.opendev.org/664217\n\nChange-Id: I527aba3635ef82e2d2a88ee7f5250bf03e420800\nSigned-off-by: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ccabadc342477b3c16b11ae52e0194c7916966c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ccabadc342477b3c16b11ae52e0194c7916966c2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
