)]}'
{"id":"openstack%2Fkolla~667519","triplet_id":"openstack%2Fkolla~master~I4cd87ffad39dddc4d34c17f5dd7f536440090944","project":"openstack/kolla","branch":"master","topic":"dnm-ci-test","hashtags":[],"change_id":"I4cd87ffad39dddc4d34c17f5dd7f536440090944","subject":"DNM: kolla CI test (host: CentOS 7 -\u003e Ubuntu Bionic)","status":"ABANDONED","created":"2019-06-26 06:22:33.000000000","updated":"2019-06-28 17:27:37.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c82a259bd37ec467f5911457b152f8c9800ba23b","_number":667519,"virtual_id_number":667519,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2019-06-26 09:00:03.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-26 09:00:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d999d948abe48e58568b0bd63e5b3bf689f9a5ac","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-26 06:22:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4ce4b9934b84ec192d98bf5db2d251055408e9f1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-26 06:22:51.000000000","message":"Patch Set 1: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/kolla on branch stable/queens.  The error was:\n\n  Nodeset kolla-centos does not match existing definition in branch master\n\nThe error appears in the following nodeset stanza:\n\n  nodeset:\n      name: kolla-centos\n      nodes:\n        - name: primary\n          label: centos-7\n\n  in \"openstack/kolla/.zuul.d/base.yaml@stable/queens\", line 28, column 3","accounts_in_message":[],"_revision_number":1},{"id":"63d04bc7b9f90d1c8a53d7952db5d45642594ecb","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-26 06:30:08.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"cd80d7f29aa1611eca1cbb996ced67bb9bf84957","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-26 09:00:03.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/19/667519/2/check/kolla-build-centos-binary/a3c0ca2/ : SUCCESS in 31m 13s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/19/667519/2/check/kolla-build-centos-source/dd7b3e9/ : SUCCESS in 1h 39m 02s\n- kolla-ansible-centos-source http://logs.openstack.org/19/667519/2/check/kolla-ansible-centos-source/d558171/ : SUCCESS in 1h 40m 22s\n- kolla-ansible-centos-binary http://logs.openstack.org/19/667519/2/check/kolla-ansible-centos-binary/edb9631/ : SUCCESS in 49m 49s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/19/667519/2/check/kolla-ansible-centos-source-upgrade/69dc1e7/ : SUCCESS in 1h 29m 52s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/19/667519/2/check/tripleo-build-containers-centos-7/244c151/ : SUCCESS in 45m 59s\n- openstack-tox-lower-constraints http://logs.openstack.org/19/667519/2/check/openstack-tox-lower-constraints/1bca347/ : SUCCESS in 3m 50s\n- openstack-tox-pep8 http://logs.openstack.org/19/667519/2/check/openstack-tox-pep8/cd48822/ : SUCCESS in 3m 24s\n- openstack-tox-py27 http://logs.openstack.org/19/667519/2/check/openstack-tox-py27/affc29a/ : SUCCESS in 3m 20s\n- openstack-tox-py36 http://logs.openstack.org/19/667519/2/check/openstack-tox-py36/72856e9/ : SUCCESS in 3m 06s\n- openstack-tox-py37 http://logs.openstack.org/19/667519/2/check/openstack-tox-py37/caa49c2/ : SUCCESS in 3m 11s\n- openstack-tox-docs http://logs.openstack.org/19/667519/2/check/openstack-tox-docs/2e17465/html/ : SUCCESS in 3m 02s\n- kolla-build-ubuntu-binary http://logs.openstack.org/19/667519/2/check/kolla-build-ubuntu-binary/7681f53/ : SUCCESS in 23m 28s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/19/667519/2/check/kolla-build-ubuntu-source/e5ce4fd/ : SUCCESS in 47m 23s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/19/667519/2/check/kolla-ansible-ubuntu-source/1ef5026/ : SUCCESS in 1h 11m 54s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/19/667519/2/check/kolla-ansible-ubuntu-binary/4940ca2/ : SUCCESS in 1h 10m 39s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/19/667519/2/check/kolla-ansible-ubuntu-source-upgrade/0b8cdd4/ : SUCCESS in 1h 21m 17s","accounts_in_message":[],"_revision_number":2},{"id":"6e94e3046d931b5eb7c308b0c54da08b2e83778a","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-28 17:27:37.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"5a1993d57027750174edcd20ec08777c2e8b9dca","revisions":{"8880e5d461028e0fcb90a476a29908c44d9a10b8":{"kind":"REWORK","_number":1,"created":"2019-06-26 06:22:33.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/19/667519/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/19/667519/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/19/667519/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/19/667519/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/19/667519/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/19/667519/1"}}},"commit":{"parents":[{"commit":"2777a0c0584e4c707c56ca6aaa8cf2f41a473a83","subject":"Merge \"Make SKIPPED_IMAGES more flexible\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2777a0c0584e4c707c56ca6aaa8cf2f41a473a83"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-06-26 06:21:18.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-06-26 06:21:18.000000000","tz":120},"subject":"DNM: kolla CI test (host: CentOS 7 -\u003e Ubuntu Bionic)","message":"DNM: kolla CI test (host: CentOS 7 -\u003e Ubuntu Bionic)\n\nDepends-On: https://review.opendev.org/667518\nChange-Id: I4cd87ffad39dddc4d34c17f5dd7f536440090944\nSigned-off-by: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8880e5d461028e0fcb90a476a29908c44d9a10b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8880e5d461028e0fcb90a476a29908c44d9a10b8"}]},"branch":"refs/heads/master"},"5a1993d57027750174edcd20ec08777c2e8b9dca":{"kind":"REWORK","_number":2,"created":"2019-06-26 06:30:08.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/19/667519/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/19/667519/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/19/667519/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/19/667519/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/19/667519/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/19/667519/2"}}},"commit":{"parents":[{"commit":"2777a0c0584e4c707c56ca6aaa8cf2f41a473a83","subject":"Merge \"Make SKIPPED_IMAGES more flexible\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2777a0c0584e4c707c56ca6aaa8cf2f41a473a83"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-06-26 06:21:18.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-06-26 06:29:06.000000000","tz":120},"subject":"DNM: kolla CI test (host: CentOS 7 -\u003e Ubuntu Bionic)","message":"DNM: kolla CI test (host: CentOS 7 -\u003e Ubuntu Bionic)\n\nDepends-On: https://review.opendev.org/667518\nChange-Id: I4cd87ffad39dddc4d34c17f5dd7f536440090944\nSigned-off-by: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5a1993d57027750174edcd20ec08777c2e8b9dca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5a1993d57027750174edcd20ec08777c2e8b9dca"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
