)]}'
{"id":"openstack%2Fkolla~667587","triplet_id":"openstack%2Fkolla~master~I78e9574ea01d84051dbc649fea1b5b9556778a15","project":"openstack/kolla","branch":"master","hashtags":[],"change_id":"I78e9574ea01d84051dbc649fea1b5b9556778a15","subject":"base: do not put grafana repo twice","status":"MERGED","created":"2019-06-26 12:05:02.000000000","updated":"2019-07-03 07:38:04.000000000","submitted":"2019-07-03 07:38:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"667587-1562139484467-cff29fd9","meta_rev_id":"501964640ee7903ce28e2c96518c8fe96fb0ad6e","_number":667587,"virtual_id_number":667587,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"date":"2019-06-27 13:50:31.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":2,"date":"2019-07-03 07:38:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-06-27 14:31:01.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2019-07-03 06:12:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2019-06-26 15:24:44.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2019-06-26 12:08:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2019-07-03 06:12:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-26 12:08:18.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-06-26 15:24:44.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-06-27 14:31:01.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-03 06:12:04.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2019-07-03 07:38:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ade5e89688b5863a019178e1c4ba569f7b31dbe5","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-26 12:05:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9f8c34a747a5889f0ee6fc3c44bba63e3cbeb7c7","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-26 12:08:18.000000000","message":"Patch Set 1: Code-Review+1\n\nThey say: the more the merrier, but we know better.","accounts_in_message":[],"_revision_number":1},{"id":"d6effe6e8d8caa08f6458d3261f0c9da4ce73ebb","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-26 12:39:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/87/667587/1/openstack-check/tripleo-build-containers-fedora-28-master/ba76a32/ : FAILURE in 6m 16s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/87/667587/1/openstack-check/tripleo-build-containers-fedora-28-stein/24ce1b7/ : SUCCESS in 32m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ed8c45a6a087da1c080205d1aaa4e87b6732ddef","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-26 14:23:04.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/87/667587/1/check/kolla-build-centos-binary/f0bd7ed/ : SUCCESS in 42m 40s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/87/667587/1/check/kolla-build-centos-source/51a6a65/ : SUCCESS in 2h 12m 17s\n- kolla-ansible-centos-source http://logs.openstack.org/87/667587/1/check/kolla-ansible-centos-source/f4a24e3/ : TIMED_OUT in 2h 00m 55s\n- kolla-ansible-centos-binary http://logs.openstack.org/87/667587/1/check/kolla-ansible-centos-binary/f5e3f5b/ : SUCCESS in 1h 02m 05s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/87/667587/1/check/kolla-ansible-centos-source-upgrade/fdde89c/ : TIMED_OUT in 2h 01m 08s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/87/667587/1/check/tripleo-build-containers-centos-7/55b0b66/ : SUCCESS in 47m 53s\n- openstack-tox-lower-constraints http://logs.openstack.org/87/667587/1/check/openstack-tox-lower-constraints/50426a2/ : SUCCESS in 3m 39s\n- openstack-tox-pep8 http://logs.openstack.org/87/667587/1/check/openstack-tox-pep8/9ef5036/ : SUCCESS in 4m 12s\n- openstack-tox-py27 http://logs.openstack.org/87/667587/1/check/openstack-tox-py27/8973dc9/ : SUCCESS in 2m 56s\n- openstack-tox-py36 http://logs.openstack.org/87/667587/1/check/openstack-tox-py36/3e73ef2/ : SUCCESS in 3m 03s\n- openstack-tox-py37 http://logs.openstack.org/87/667587/1/check/openstack-tox-py37/b81e063/ : SUCCESS in 3m 23s\n- openstack-tox-docs http://logs.openstack.org/87/667587/1/check/openstack-tox-docs/e5c5bb8/html/ : SUCCESS in 2m 47s\n- kolla-build-ubuntu-binary http://logs.openstack.org/87/667587/1/check/kolla-build-ubuntu-binary/6af676a/ : SUCCESS in 23m 48s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/87/667587/1/check/kolla-build-ubuntu-source/e76757b/ : SUCCESS in 29m 21s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/87/667587/1/check/kolla-ansible-ubuntu-source/e0992f0/ : SUCCESS in 53m 33s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/87/667587/1/check/kolla-ansible-ubuntu-binary/f790488/ : SUCCESS in 45m 07s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/87/667587/1/check/kolla-ansible-ubuntu-source-upgrade/5108038/ : SUCCESS in 52m 02s","accounts_in_message":[],"_revision_number":1},{"id":"6ad68fa41e9d00509c0e5c33cb44b4712d3c94df","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-26 15:24:44.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ad6b6593cf41e9456de7039cacc30254965f7d62","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-26 15:24:51.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f00da44488609e95de8666f49d2fc0cc17068ef4","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-26 16:06:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/87/667587/1/openstack-check/tripleo-build-containers-fedora-28-master/05a4b80/ : FAILURE in 5m 26s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/87/667587/1/openstack-check/tripleo-build-containers-fedora-28-stein/3cd3bc9/ : SUCCESS in 32m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4e6afca53564013c4ac691c0d5fb3849581d28ea","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-26 17:49:14.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/87/667587/1/check/kolla-build-centos-binary/7ecdb5e/ : SUCCESS in 25m 02s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/87/667587/1/check/kolla-build-centos-source/a8d6ef2/ : SUCCESS in 37m 56s\n- kolla-ansible-centos-source http://logs.openstack.org/87/667587/1/check/kolla-ansible-centos-source/0532f56/ : TIMED_OUT in 2h 01m 04s\n- kolla-ansible-centos-binary http://logs.openstack.org/87/667587/1/check/kolla-ansible-centos-binary/a3ecf59/ : SUCCESS in 1h 13m 49s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/87/667587/1/check/kolla-ansible-centos-source-upgrade/6411bd2/ : TIMED_OUT in 2h 01m 11s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/87/667587/1/check/tripleo-build-containers-centos-7/f61485d/ : SUCCESS in 50m 19s\n- openstack-tox-lower-constraints http://logs.openstack.org/87/667587/1/check/openstack-tox-lower-constraints/c26f57a/ : SUCCESS in 3m 14s\n- openstack-tox-pep8 http://logs.openstack.org/87/667587/1/check/openstack-tox-pep8/45bb688/ : SUCCESS in 3m 13s\n- openstack-tox-py27 http://logs.openstack.org/87/667587/1/check/openstack-tox-py27/7ae8505/ : SUCCESS in 3m 37s\n- openstack-tox-py36 http://logs.openstack.org/87/667587/1/check/openstack-tox-py36/819e7f5/ : SUCCESS in 3m 50s\n- openstack-tox-py37 http://logs.openstack.org/87/667587/1/check/openstack-tox-py37/e87a23e/ : SUCCESS in 3m 00s\n- openstack-tox-docs http://logs.openstack.org/87/667587/1/check/openstack-tox-docs/9da1a29/html/ : SUCCESS in 2m 40s\n- kolla-build-ubuntu-binary http://logs.openstack.org/87/667587/1/check/kolla-build-ubuntu-binary/3822fc3/ : SUCCESS in 25m 24s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/87/667587/1/check/kolla-build-ubuntu-source/a379e54/ : SUCCESS in 28m 22s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/87/667587/1/check/kolla-ansible-ubuntu-source/59117e7/ : SUCCESS in 46m 42s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/87/667587/1/check/kolla-ansible-ubuntu-binary/4cc848f/ : SUCCESS in 44m 14s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/87/667587/1/check/kolla-ansible-ubuntu-source-upgrade/2494ebe/ : SUCCESS in 54m 56s","accounts_in_message":[],"_revision_number":1},{"id":"05480dd4e14bfb4d69cf8368276c0a703afb17f8","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-26 17:55:53.000000000","message":"Patch Set 1:\n\nLet\u0027s not recheck until we fix the CI issue.","accounts_in_message":[],"_revision_number":1},{"id":"97eb006d4610c614fd2cabfc1ca167c48973357f","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-27 07:59:26.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"967cf419d8942f67f4179b40138f029eaea674e7","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-27 08:37:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/87/667587/1/openstack-check/tripleo-build-containers-fedora-28-master/afa7235/ : FAILURE in 6m 34s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/87/667587/1/openstack-check/tripleo-build-containers-fedora-28-stein/ee66dfe/ : SUCCESS in 33m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5fbf674f7ef90c1229ad1dabbd071edbd2cb8579","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 09:46:06.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/87/667587/1/check/kolla-build-centos-binary/caa29ab/ : SUCCESS in 34m 05s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/87/667587/1/check/kolla-build-centos-source/728e6fa/ : SUCCESS in 33m 20s\n- kolla-ansible-centos-source http://logs.openstack.org/87/667587/1/check/kolla-ansible-centos-source/5c1928e/ : SUCCESS in 1h 03m 48s\n- kolla-ansible-centos-binary http://logs.openstack.org/87/667587/1/check/kolla-ansible-centos-binary/094682e/ : SUCCESS in 1h 05m 31s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/87/667587/1/check/kolla-ansible-centos-source-upgrade/bcf43c1/ : SUCCESS in 1h 17m 43s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/87/667587/1/check/tripleo-build-containers-centos-7/781caf1/ : SUCCESS in 48m 38s\n- openstack-tox-lower-constraints http://logs.openstack.org/87/667587/1/check/openstack-tox-lower-constraints/b9b6dbe/ : RETRY_LIMIT in 2m 47s\n- openstack-tox-pep8 http://logs.openstack.org/87/667587/1/check/openstack-tox-pep8/7d9fbc2/ : SUCCESS in 3m 08s\n- openstack-tox-py27 http://logs.openstack.org/87/667587/1/check/openstack-tox-py27/85030ae/ : SUCCESS in 3m 09s\n- openstack-tox-py36 http://logs.openstack.org/87/667587/1/check/openstack-tox-py36/8a8d159/ : SUCCESS in 2m 43s\n- openstack-tox-py37 http://logs.openstack.org/87/667587/1/check/openstack-tox-py37/32cc4a9/ : SUCCESS in 4m 23s\n- openstack-tox-docs http://logs.openstack.org/87/667587/1/check/openstack-tox-docs/2b00ad6/html/ : SUCCESS in 2m 43s\n- kolla-build-ubuntu-binary http://logs.openstack.org/87/667587/1/check/kolla-build-ubuntu-binary/31546b2/ : SUCCESS in 28m 53s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/87/667587/1/check/kolla-build-ubuntu-source/6c6389a/ : SUCCESS in 30m 23s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/87/667587/1/check/kolla-ansible-ubuntu-source/2542be5/ : SUCCESS in 42m 32s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/87/667587/1/check/kolla-ansible-ubuntu-binary/c4efc23/ : SUCCESS in 43m 54s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/87/667587/1/check/kolla-ansible-ubuntu-source-upgrade/bc2fb1e/ : SUCCESS in 52m 22s","accounts_in_message":[],"_revision_number":1},{"id":"b68f5914c9ae4731070d320a5bb3c4f535adca30","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-27 13:50:31.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"8fd8b74ef452b17c40e6c5f07c7aaab6d661b479","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-27 14:31:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/87/667587/1/openstack-check/tripleo-build-containers-fedora-28-master/0a23399/ : FAILURE in 6m 18s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/87/667587/1/openstack-check/tripleo-build-containers-fedora-28-stein/9621a32/ : SUCCESS in 38m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"276e71444c8a5bd64e8134335f6e2e8a5709c8ce","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 16:01:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/87/667587/1/check/kolla-build-centos-binary/129e27e/ : SUCCESS in 43m 45s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/87/667587/1/check/kolla-build-centos-source/8e0015a/ : SUCCESS in 52m 55s\n- kolla-ansible-centos-source http://logs.openstack.org/87/667587/1/check/kolla-ansible-centos-source/64cdbf0/ : SUCCESS in 1h 06m 01s\n- kolla-ansible-centos-binary http://logs.openstack.org/87/667587/1/check/kolla-ansible-centos-binary/ba8dde4/ : SUCCESS in 1h 25m 35s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/87/667587/1/check/kolla-ansible-centos-source-upgrade/f152a76/ : SUCCESS in 1h 15m 01s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/87/667587/1/check/tripleo-build-containers-centos-7/588defb/ : SUCCESS in 52m 16s\n- openstack-tox-lower-constraints http://logs.openstack.org/87/667587/1/check/openstack-tox-lower-constraints/ee2dd26/ : SUCCESS in 3m 39s\n- openstack-tox-pep8 http://logs.openstack.org/87/667587/1/check/openstack-tox-pep8/2889f8f/ : SUCCESS in 3m 24s\n- openstack-tox-py27 http://logs.openstack.org/87/667587/1/check/openstack-tox-py27/cad7e77/ : SUCCESS in 3m 37s\n- openstack-tox-py36 http://logs.openstack.org/87/667587/1/check/openstack-tox-py36/eb2849c/ : SUCCESS in 3m 17s\n- openstack-tox-py37 http://logs.openstack.org/87/667587/1/check/openstack-tox-py37/6b33af5/ : SUCCESS in 3m 49s\n- openstack-tox-docs http://logs.openstack.org/87/667587/1/check/openstack-tox-docs/2b3bcec/html/ : SUCCESS in 3m 06s\n- kolla-build-ubuntu-binary http://logs.openstack.org/87/667587/1/check/kolla-build-ubuntu-binary/79ca6c9/ : SUCCESS in 25m 48s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/87/667587/1/check/kolla-build-ubuntu-source/1285653/ : SUCCESS in 40m 59s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/87/667587/1/check/kolla-ansible-ubuntu-source/40a7212/ : SUCCESS in 50m 29s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/87/667587/1/check/kolla-ansible-ubuntu-binary/9dcbffc/ : SUCCESS in 44m 24s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/87/667587/1/check/kolla-ansible-ubuntu-source-upgrade/3155987/ : SUCCESS in 56m 57s","accounts_in_message":[],"_revision_number":1},{"id":"ad1aaa7ae3b013d193a0d1806631e6c335bf7e4f","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-07-03 06:12:04.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4e20f70acda28a1840fd51eae62bda1ece6784dd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-03 06:12:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"66d423f8230dff56588a0e0c9c3714a654390c8c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-03 07:38:04.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source http://logs.openstack.org/87/667587/1/gate/kolla-build-centos-source/5e29279/ : SUCCESS in 37m 14s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/87/667587/1/gate/tripleo-build-containers-centos-7/cad8dcf/ : SUCCESS in 44m 11s\n- openstack-tox-lower-constraints http://logs.openstack.org/87/667587/1/gate/openstack-tox-lower-constraints/a7b9d7b/ : SUCCESS in 3m 36s\n- openstack-tox-pep8 http://logs.openstack.org/87/667587/1/gate/openstack-tox-pep8/19cdaaf/ : SUCCESS in 3m 08s\n- openstack-tox-py27 http://logs.openstack.org/87/667587/1/gate/openstack-tox-py27/8b01c41/ : SUCCESS in 3m 04s\n- openstack-tox-py36 http://logs.openstack.org/87/667587/1/gate/openstack-tox-py36/78b97fa/ : SUCCESS in 3m 24s\n- openstack-tox-py37 http://logs.openstack.org/87/667587/1/gate/openstack-tox-py37/56193bd/ : SUCCESS in 7m 44s\n- openstack-tox-docs http://logs.openstack.org/87/667587/1/gate/openstack-tox-docs/597a0ab/html/ : SUCCESS in 2m 44s\n- kolla-build-ubuntu-source http://logs.openstack.org/87/667587/1/gate/kolla-build-ubuntu-source/7be18b5/ : SUCCESS in 36m 24s","accounts_in_message":[],"_revision_number":1},{"id":"501964640ee7903ce28e2c96518c8fe96fb0ad6e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-03 07:38:04.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"91b7bcd297515354af2aa83d714638787ad68db5","revisions":{"91b7bcd297515354af2aa83d714638787ad68db5":{"kind":"REWORK","_number":1,"created":"2019-06-26 12:05:02.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/87/667587/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/87/667587/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/87/667587/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/87/667587/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/87/667587/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/87/667587/1"}}},"commit":{"parents":[{"commit":"414d61836fb909b16a76f9dbcfcefd5d5523e248","subject":"Merge \"Fix deployment of ceph-nfs container failing due to wrong USER\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/414d61836fb909b16a76f9dbcfcefd5d5523e248"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin@juszkiewicz.com.pl","date":"2019-06-26 11:53:52.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin@juszkiewicz.com.pl","date":"2019-06-26 12:04:14.000000000","tz":120},"subject":"base: do not put grafana repo twice","message":"base: do not put grafana repo twice\n\nBy mistake I added it second time in\n1e7a78932dc0461c60d9fd6e613b252d8b22ff01 commit.\n\nChange-Id: I78e9574ea01d84051dbc649fea1b5b9556778a15\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/91b7bcd297515354af2aa83d714638787ad68db5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/91b7bcd297515354af2aa83d714638787ad68db5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
