)]}'
{"id":"openstack%2Fkolla~667588","triplet_id":"openstack%2Fkolla~master~Id13545224fa0fbb729d46386a10505e0989fc3c6","project":"openstack/kolla","branch":"master","topic":"667588","hashtags":[],"change_id":"Id13545224fa0fbb729d46386a10505e0989fc3c6","subject":"build: add timestamps to start/end of build","status":"MERGED","created":"2019-06-26 12:05:39.000000000","updated":"2019-07-03 07:38:02.000000000","submitted":"2019-07-03 07:38:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"667588-1562139482596-2d6c1d39","meta_rev_id":"de46cbe0736a3adff8c376ba78cc075f87dff00a","_number":667588,"virtual_id_number":667588,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"date":"2019-06-27 07:46:33.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":2,"date":"2019-07-03 07:38:02.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-06-27 19:35:44.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2019-07-03 06:11:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2019-06-26 16:22:03.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2019-06-26 16:15:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2019-07-03 06:11:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-26 16:15:19.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-06-26 16:22:03.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-06-27 19:35:44.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-03 06:11:29.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2019-07-03 07:38:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f70fc1c9ec0ef3884d5d472c6736f45ac007e420","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-26 12:05:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"09d9de712060b12c944b6a3f94bb74512b95c921","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-26 12:11:19.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a8b7ca91f0eef96da93c56c83a869c2f7c3859b4","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-26 12:40:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/88/667588/1/openstack-check/tripleo-build-containers-fedora-28-master/9dc89d6/ : FAILURE in 6m 24s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/88/667588/1/openstack-check/tripleo-build-containers-fedora-28-stein/0eb43b2/ : SUCCESS in 32m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"edc48b6901e67bef4109d7c3722139911af85ee8","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-26 13:32:17.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"601d85cfc531f89ed352a428f63afc82e26198df","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-26 13:32:25.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"47e1bc19533714ff299155399c18745db396663b","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-26 14:07:01.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/88/667588/2/openstack-check/tripleo-build-containers-fedora-28-master/1499467/ : FAILURE in 5m 46s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/88/667588/2/openstack-check/tripleo-build-containers-fedora-28-stein/f4248c6/ : SUCCESS in 32m 37s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3558d4924c8582100d3395409fa2fdee181e7bdb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-26 16:07:36.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/88/667588/2/check/kolla-build-centos-binary/8ceef4e/ : SUCCESS in 27m 30s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/88/667588/2/check/kolla-build-centos-source/3913845/ : SUCCESS in 2h 12m 54s\n- kolla-ansible-centos-source http://logs.openstack.org/88/667588/2/check/kolla-ansible-centos-source/62b978e/ : TIMED_OUT in 2h 01m 11s\n- kolla-ansible-centos-binary http://logs.openstack.org/88/667588/2/check/kolla-ansible-centos-binary/e5ff3f6/ : SUCCESS in 1h 03m 18s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/88/667588/2/check/kolla-ansible-centos-source-upgrade/70a5e94/ : TIMED_OUT in 2h 01m 10s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/88/667588/2/check/tripleo-build-containers-centos-7/ffbaea5/ : SUCCESS in 45m 52s\n- openstack-tox-lower-constraints http://logs.openstack.org/88/667588/2/check/openstack-tox-lower-constraints/1e1a531/ : SUCCESS in 3m 13s\n- openstack-tox-pep8 http://logs.openstack.org/88/667588/2/check/openstack-tox-pep8/ec7daf0/ : SUCCESS in 2m 50s\n- openstack-tox-py27 http://logs.openstack.org/88/667588/2/check/openstack-tox-py27/66108f4/ : SUCCESS in 2m 43s\n- openstack-tox-py36 http://logs.openstack.org/88/667588/2/check/openstack-tox-py36/f31c4c5/ : SUCCESS in 3m 35s\n- openstack-tox-py37 http://logs.openstack.org/88/667588/2/check/openstack-tox-py37/204a1d6/ : SUCCESS in 3m 36s\n- openstack-tox-docs http://logs.openstack.org/88/667588/2/check/openstack-tox-docs/6404fcb/html/ : SUCCESS in 3m 00s\n- kolla-build-ubuntu-binary http://logs.openstack.org/88/667588/2/check/kolla-build-ubuntu-binary/dc89e37/ : SUCCESS in 24m 26s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/88/667588/2/check/kolla-build-ubuntu-source/3ab4996/ : SUCCESS in 29m 15s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/88/667588/2/check/kolla-ansible-ubuntu-source/6aa2653/ : SUCCESS in 47m 19s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/88/667588/2/check/kolla-ansible-ubuntu-binary/29d3ed8/ : SUCCESS in 45m 52s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/88/667588/2/check/kolla-ansible-ubuntu-source-upgrade/2d2a19d/ : SUCCESS in 52m 36s","accounts_in_message":[],"_revision_number":2},{"id":"b55c5adfbe44d2ef9d4e424772d4ff1798b2bea1","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-26 16:15:19.000000000","message":"Patch Set 2: Code-Review+1\n\nIt\u0027s fine even for failures because CI gives us modtime and the log will give us the start time. This especially applies to cases we want to watch with it - i.e. timeouts. ;-)\n\nLet\u0027s not check it again before we fix CI issues.","accounts_in_message":[],"_revision_number":2},{"id":"5671b622ff95d28605c44df0a71790ecc0c35d17","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-26 16:22:03.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"56f01e4ed85317b837dd2ece2a1dc7df6b5fa0ea","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-27 07:46:33.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"292489d652ac94a5deabdf509cc0ec24f5cb5c18","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-27 08:29:25.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/88/667588/2/openstack-check/tripleo-build-containers-fedora-28-master/51c5779/ : FAILURE in 5m 12s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/88/667588/2/openstack-check/tripleo-build-containers-fedora-28-stein/c23a1bb/ : SUCCESS in 39m 35s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"930439cd16457f7ed105f87614498c6a8577de3b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 09:28:49.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/88/667588/2/check/kolla-build-centos-binary/e879de1/ : SUCCESS in 29m 58s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/88/667588/2/check/kolla-build-centos-source/b2acfc3/ : SUCCESS in 52m 27s\n- kolla-ansible-centos-source http://logs.openstack.org/88/667588/2/check/kolla-ansible-centos-source/268b760/ : FAILURE in 17m 25s\n- kolla-ansible-centos-binary http://logs.openstack.org/88/667588/2/check/kolla-ansible-centos-binary/fed2f8a/ : SUCCESS in 1h 17m 23s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/88/667588/2/check/kolla-ansible-centos-source-upgrade/f111f60/ : SUCCESS in 1h 16m 18s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/88/667588/2/check/tripleo-build-containers-centos-7/5cda74d/ : SUCCESS in 53m 57s\n- openstack-tox-lower-constraints http://logs.openstack.org/88/667588/2/check/openstack-tox-lower-constraints/d2591fe/ : SUCCESS in 3m 22s\n- openstack-tox-pep8 http://logs.openstack.org/88/667588/2/check/openstack-tox-pep8/b78a019/ : SUCCESS in 4m 00s\n- openstack-tox-py27 http://logs.openstack.org/88/667588/2/check/openstack-tox-py27/1bc5b19/ : SUCCESS in 3m 12s\n- openstack-tox-py36 http://logs.openstack.org/88/667588/2/check/openstack-tox-py36/8bfd7e2/ : SUCCESS in 4m 05s\n- openstack-tox-py37 http://logs.openstack.org/88/667588/2/check/openstack-tox-py37/a8429d7/ : SUCCESS in 5m 13s\n- openstack-tox-docs http://logs.openstack.org/88/667588/2/check/openstack-tox-docs/28511c2/html/ : SUCCESS in 3m 03s\n- kolla-build-ubuntu-binary http://logs.openstack.org/88/667588/2/check/kolla-build-ubuntu-binary/aeefc8a/ : FAILURE in 47m 08s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/88/667588/2/check/kolla-build-ubuntu-source/fd50194/ : SUCCESS in 27m 26s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/88/667588/2/check/kolla-ansible-ubuntu-source/74f6eac/ : SUCCESS in 45m 05s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/88/667588/2/check/kolla-ansible-ubuntu-binary/9a2c745/ : SUCCESS in 44m 09s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/88/667588/2/check/kolla-ansible-ubuntu-source-upgrade/16462be/ : SUCCESS in 55m 33s","accounts_in_message":[],"_revision_number":2},{"id":"fd55a570911a253785c754ab4379ed9b56bb9b0d","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-27 16:47:30.000000000","message":"Patch Set 2:\n\nrecheck\n\nThis will also help infra pinpoint our issues with mirror to their logs as we can just show them our logs then.","accounts_in_message":[],"_revision_number":2},{"id":"fde82691e8aa4e0685388aeb13f909823bf509f0","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-27 17:27:57.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/88/667588/2/openstack-check/tripleo-build-containers-fedora-28-master/24084d5/ : FAILURE in 5m 46s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/88/667588/2/openstack-check/tripleo-build-containers-fedora-28-stein/a924282/ : SUCCESS in 37m 57s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"723365436689c39ed9f02a1ade1ed7c834e08a14","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 18:23:53.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/88/667588/2/check/kolla-build-centos-binary/2f1881f/ : FAILURE in 47m 44s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/88/667588/2/check/kolla-build-centos-source/9a38d01/ : SUCCESS in 48m 20s\n- kolla-ansible-centos-source http://logs.openstack.org/88/667588/2/check/kolla-ansible-centos-source/8a01da3/ : FAILURE in 35m 26s\n- kolla-ansible-centos-binary http://logs.openstack.org/88/667588/2/check/kolla-ansible-centos-binary/ad588c5/ : SUCCESS in 59m 24s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/88/667588/2/check/kolla-ansible-centos-source-upgrade/d53f79f/ : SUCCESS in 1h 18m 40s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/88/667588/2/check/tripleo-build-containers-centos-7/72753c6/ : SUCCESS in 44m 22s\n- openstack-tox-lower-constraints http://logs.openstack.org/88/667588/2/check/openstack-tox-lower-constraints/143c5e2/ : SUCCESS in 3m 43s\n- openstack-tox-pep8 http://logs.openstack.org/88/667588/2/check/openstack-tox-pep8/0a91170/ : SUCCESS in 3m 14s\n- openstack-tox-py27 http://logs.openstack.org/88/667588/2/check/openstack-tox-py27/a9b6a75/ : SUCCESS in 3m 13s\n- openstack-tox-py36 http://logs.openstack.org/88/667588/2/check/openstack-tox-py36/a1454b2/ : SUCCESS in 4m 50s\n- openstack-tox-py37 http://logs.openstack.org/88/667588/2/check/openstack-tox-py37/a16ddad/ : SUCCESS in 5m 08s\n- openstack-tox-docs http://logs.openstack.org/88/667588/2/check/openstack-tox-docs/a338e59/html/ : SUCCESS in 3m 01s\n- kolla-build-ubuntu-binary http://logs.openstack.org/88/667588/2/check/kolla-build-ubuntu-binary/db58e58/ : SUCCESS in 22m 25s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/88/667588/2/check/kolla-build-ubuntu-source/9ca81b5/ : SUCCESS in 26m 12s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/88/667588/2/check/kolla-ansible-ubuntu-source/126eda0/ : SUCCESS in 45m 38s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/88/667588/2/check/kolla-ansible-ubuntu-binary/6320861/ : SUCCESS in 46m 30s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/88/667588/2/check/kolla-ansible-ubuntu-source-upgrade/634d76b/ : SUCCESS in 53m 48s","accounts_in_message":[],"_revision_number":2},{"id":"3c2b9d03b06b4a55ccc658f11c7cb44376324ad4","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-27 18:43:34.000000000","message":"Patch Set 2:\n\nrecheck\n\nkolla-ansible-centos-source -\u003e external EPEL mirror 404 errors\n\nkolla-build-centos-binary -\u003e https://packagecloud.io/rabbitmq/rabbitmq-server/el/7/x86_64/repodata/repomd.xml: [Errno 14] HTTPS Error 302 - Found\nand\nhttps://repos.influxdata.com/rhel/7/x86_64/stable/repodata/repomd.xml: [Errno 12] Timeout on https://repos.influxdata.com/rhel/7/x86_64/stable/repodata/repomd.xml: (28, \u0027Operation timed out after 30000 milliseconds with 0 out of 0 bytes received\u0027)\nand\nhttps://repos.influxdata.com/rhel/7/x86_64/stable/repodata/repomd.xml: [Errno 14] curl#35 - \"TCP connection reset by peer\"\nand\nhttps://repos.influxdata.com/rhel/7/x86_64/stable/repodata/8129f03bbfd898c763e649a3819602dd0020c04f-primary.sqlite.bz2: [Errno 14] curl#7 - \"Failed to connect to 2600:9000:20d1:e800:7:7790:e740:93a1: Network is unreachable\"","accounts_in_message":[],"_revision_number":2},{"id":"35b5021bf252750c930bcf6152cd7089ee38c004","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-27 19:35:44.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/88/667588/2/openstack-check/tripleo-build-containers-fedora-28-master/555af96/ : FAILURE in 5m 59s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/88/667588/2/openstack-check/tripleo-build-containers-fedora-28-stein/6ab1715/ : SUCCESS in 36m 56s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"347195ef7d3524cda3e2ece0ad0989bba2a27a08","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 20:22:31.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/88/667588/2/check/kolla-build-centos-binary/4a76dc3/ : SUCCESS in 42m 43s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/88/667588/2/check/kolla-build-centos-source/0f21ad6/ : SUCCESS in 55m 54s\n- kolla-ansible-centos-source http://logs.openstack.org/88/667588/2/check/kolla-ansible-centos-source/689abaa/ : SUCCESS in 1h 09m 09s\n- kolla-ansible-centos-binary http://logs.openstack.org/88/667588/2/check/kolla-ansible-centos-binary/80281f9/ : SUCCESS in 1h 02m 47s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/88/667588/2/check/kolla-ansible-centos-source-upgrade/01e7a3b/ : SUCCESS in 1h 18m 39s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/88/667588/2/check/tripleo-build-containers-centos-7/1f5f51f/ : SUCCESS in 46m 07s\n- openstack-tox-lower-constraints http://logs.openstack.org/88/667588/2/check/openstack-tox-lower-constraints/13c9c72/ : SUCCESS in 3m 41s\n- openstack-tox-pep8 http://logs.openstack.org/88/667588/2/check/openstack-tox-pep8/397bf49/ : SUCCESS in 3m 26s\n- openstack-tox-py27 http://logs.openstack.org/88/667588/2/check/openstack-tox-py27/5fd25e6/ : SUCCESS in 3m 29s\n- openstack-tox-py36 http://logs.openstack.org/88/667588/2/check/openstack-tox-py36/9fc50f1/ : SUCCESS in 3m 13s\n- openstack-tox-py37 http://logs.openstack.org/88/667588/2/check/openstack-tox-py37/6eb470c/ : SUCCESS in 3m 36s\n- openstack-tox-docs http://logs.openstack.org/88/667588/2/check/openstack-tox-docs/ac0cf0c/html/ : SUCCESS in 3m 30s\n- kolla-build-ubuntu-binary http://logs.openstack.org/88/667588/2/check/kolla-build-ubuntu-binary/1b23eb4/ : SUCCESS in 25m 38s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/88/667588/2/check/kolla-build-ubuntu-source/587fde0/ : SUCCESS in 30m 45s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/88/667588/2/check/kolla-ansible-ubuntu-source/19ab6d2/ : SUCCESS in 44m 53s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/88/667588/2/check/kolla-ansible-ubuntu-binary/45081b9/ : SUCCESS in 43m 45s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/88/667588/2/check/kolla-ansible-ubuntu-source-upgrade/88a9a90/ : SUCCESS in 54m 38s","accounts_in_message":[],"_revision_number":2},{"id":"cf5a767cb9f2753ce7d2a89ac7ee854a00e93bea","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-07-03 06:11:29.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"1c359f95581daca4aeb4fe97ececff5c8e7bd2d6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-03 06:11:47.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ce5aef22914bcb6c92316a5d1b88335408adbf54","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-03 07:38:02.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"c3d793096c7d570a5d2db4cb33ebdce80e59a7be","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-03 07:38:02.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source http://logs.openstack.org/88/667588/2/gate/kolla-build-centos-source/f956031/ : SUCCESS in 46m 52s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/88/667588/2/gate/tripleo-build-containers-centos-7/4c43d31/ : SUCCESS in 52m 45s\n- openstack-tox-lower-constraints http://logs.openstack.org/88/667588/2/gate/openstack-tox-lower-constraints/efa5a11/ : SUCCESS in 4m 09s\n- openstack-tox-pep8 http://logs.openstack.org/88/667588/2/gate/openstack-tox-pep8/f6164d7/ : SUCCESS in 3m 37s\n- openstack-tox-py27 http://logs.openstack.org/88/667588/2/gate/openstack-tox-py27/bfd0971/ : SUCCESS in 3m 10s\n- openstack-tox-py36 http://logs.openstack.org/88/667588/2/gate/openstack-tox-py36/ec0acb5/ : SUCCESS in 4m 22s\n- openstack-tox-py37 http://logs.openstack.org/88/667588/2/gate/openstack-tox-py37/58e7267/ : SUCCESS in 3m 53s\n- openstack-tox-docs http://logs.openstack.org/88/667588/2/gate/openstack-tox-docs/f18ad8d/html/ : SUCCESS in 3m 04s\n- kolla-build-ubuntu-source http://logs.openstack.org/88/667588/2/gate/kolla-build-ubuntu-source/af8abb1/ : SUCCESS in 27m 04s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6e505affde45168ca3246f9377fa880dbe624c76","revisions":{"ad4ad4fb87924f9acd985c72890436497614c19a":{"kind":"REWORK","_number":1,"created":"2019-06-26 12:05:39.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/88/667588/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/88/667588/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/667588/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/667588/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/667588/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/88/667588/1"}}},"commit":{"parents":[{"commit":"414d61836fb909b16a76f9dbcfcefd5d5523e248","subject":"Merge \"Fix deployment of ceph-nfs container failing due to wrong USER\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/414d61836fb909b16a76f9dbcfcefd5d5523e248"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin@juszkiewicz.com.pl","date":"2019-06-26 11:55:32.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin@juszkiewicz.com.pl","date":"2019-06-26 12:05:22.000000000","tz":120},"subject":"build: add timestamps to start/end of build","message":"build: add timestamps to start/end of build\n\nDuring recent CI issues I got an idea that it would be nice to have\ninformation how much time building specific image took. On local systems\nwe can do log and then check \u0027modified_time - creation_time\u0027 but not on\nCI jobs.\n\nFresh image:\n\nINFO:kolla.common.utils.base:Building started at 2019-06-26 13:53:08.393544\nINFO:kolla.common.utils.base:Ended at 2019-06-26 13:56:40.354761 (took 0:03:31.961225)\n\nRebuild:\n\nINFO:kolla.common.utils.base:Building started at 2019-06-26 14:00:34.913194\nINFO:kolla.common.utils.base:Built at 2019-06-26 14:00:38.988874 (took 0:00:04.075686)\n\nFailed images do not get second line due to logger.exception() being\nused.\n\nChange-Id: Id13545224fa0fbb729d46386a10505e0989fc3c6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ad4ad4fb87924f9acd985c72890436497614c19a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ad4ad4fb87924f9acd985c72890436497614c19a"}]},"branch":"refs/heads/master"},"6e505affde45168ca3246f9377fa880dbe624c76":{"kind":"REWORK","_number":2,"created":"2019-06-26 13:32:25.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/88/667588/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/88/667588/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/667588/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/667588/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/667588/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/88/667588/2"}}},"commit":{"parents":[{"commit":"414d61836fb909b16a76f9dbcfcefd5d5523e248","subject":"Merge \"Fix deployment of ceph-nfs container failing due to wrong USER\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/414d61836fb909b16a76f9dbcfcefd5d5523e248"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin@juszkiewicz.com.pl","date":"2019-06-26 11:55:32.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin@juszkiewicz.com.pl","date":"2019-06-26 13:31:58.000000000","tz":120},"subject":"build: add timestamps to start/end of build","message":"build: add timestamps to start/end of build\n\nDuring recent CI issues I got an idea that it would be nice to have\ninformation how much time building specific image took. On local systems\nwe can do log and then check \u0027modified_time - creation_time\u0027 but not on\nCI jobs.\n\nFresh image:\n\nINFO:kolla.common.utils.base:Building started at 2019-06-26 13:53:08.393544\nINFO:kolla.common.utils.base:Ended at 2019-06-26 13:56:40.354761 (took 0:03:31.961225)\n\nRebuild:\n\nINFO:kolla.common.utils.base:Building started at 2019-06-26 14:00:34.913194\nINFO:kolla.common.utils.base:Built at 2019-06-26 14:00:38.988874 (took 0:00:04.075686)\n\nFailed images do not get second line due to logger.exception() being\nused.\n\nChange-Id: Id13545224fa0fbb729d46386a10505e0989fc3c6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6e505affde45168ca3246f9377fa880dbe624c76"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6e505affde45168ca3246f9377fa880dbe624c76"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
