)]}'
{"id":"openstack%2Fkolla~667797","triplet_id":"openstack%2Fkolla~master~Ie259a38aad84bd8f5392f19791edfbd1f8ec4b1a","project":"openstack/kolla","branch":"master","topic":"667797","hashtags":[],"change_id":"Ie259a38aad84bd8f5392f19791edfbd1f8ec4b1a","subject":"Add more entries into SKIPPED_IMAGES","status":"MERGED","created":"2019-06-27 07:15:19.000000000","updated":"2019-07-03 07:37:58.000000000","submitted":"2019-07-03 07:37:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"667797-1562139478591-7257798a","meta_rev_id":"dd1fd79132edd47accc5271327606e829c99a534","_number":667797,"virtual_id_number":667797,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"date":"2019-06-29 16:13:15.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":2,"date":"2019-07-03 07:37:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-06-28 15:35:49.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2019-07-03 06:09:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2019-07-01 11:01:32.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2019-06-30 06:58:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2019-07-03 06:09:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-28 15:35:49.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-06-30 06:58:58.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-07-01 11:01:32.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-07-03 06:09:39.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2019-07-03 07:37:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2c892b9c2f85741c3e85fa2aa76b252b9cf75962","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-27 07:15:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6758cd27cd2ad94e9dbeb3ddd8ba7489b13f4766","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-27 08:00:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/97/667797/1/openstack-check/tripleo-build-containers-fedora-28-master/2eef85a/ : FAILURE in 6m 03s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/97/667797/1/openstack-check/tripleo-build-containers-fedora-28-stein/533c0ef/ : SUCCESS in 39m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f6af4ace04793053d7e39fbd13f920a301cb3aec","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-27 08:24:46.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"5144c109020a238f483e7ad68f88b90143dced3c","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-27 08:25:59.000000000","message":"Patch Set 2: Code-Review+1\n\nReplaced \u0027cyborg-agent\u0027 with \u0027cyborg-base\u0027 as it is done in the other cases. Reordered to be consistent with the current ordering.","accounts_in_message":[],"_revision_number":2},{"id":"aee371a4aba7dfcd632002bbda7cad6992ae9084","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-27 09:01:06.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/97/667797/2/openstack-check/tripleo-build-containers-fedora-28-master/122455b/ : FAILURE in 6m 15s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/97/667797/2/openstack-check/tripleo-build-containers-fedora-28-stein/71d9b0d/ : SUCCESS in 32m 42s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d7b43d5f270d8efdc1ad7f89e57540549b2112f5","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-27 09:06:00.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"d9df0b0bbc0db003cca540f30d83b28870daa586","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 10:34:06.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/97/667797/2/check/kolla-build-centos-binary/0bd564b/ : SUCCESS in 42m 10s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/97/667797/2/check/kolla-build-centos-source/ab4a7da/ : RETRY_LIMIT in 5m 45s\n- kolla-ansible-centos-source http://logs.openstack.org/97/667797/2/check/kolla-ansible-centos-source/147833b/ : RETRY_LIMIT in 6m 24s\n- kolla-ansible-centos-binary http://logs.openstack.org/97/667797/2/check/kolla-ansible-centos-binary/6f9bfde/ : RETRY_LIMIT in 6m 03s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/97/667797/2/check/kolla-ansible-centos-source-upgrade/498723a/ : RETRY_LIMIT in 6m 34s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/97/667797/2/check/tripleo-build-containers-centos-7/fcc0b23/ : SUCCESS in 45m 59s\n- openstack-tox-lower-constraints http://logs.openstack.org/97/667797/2/check/openstack-tox-lower-constraints/e03b6fd/ : SUCCESS in 3m 29s\n- openstack-tox-pep8 http://logs.openstack.org/97/667797/2/check/openstack-tox-pep8/7970657/ : SUCCESS in 3m 11s\n- openstack-tox-py27 http://logs.openstack.org/97/667797/2/check/openstack-tox-py27/9d10b47/ : SUCCESS in 3m 15s\n- openstack-tox-py36 http://logs.openstack.org/97/667797/2/check/openstack-tox-py36/d5515e9/ : SUCCESS in 3m 53s\n- openstack-tox-py37 http://logs.openstack.org/97/667797/2/check/openstack-tox-py37/1183fa9/ : SUCCESS in 4m 47s\n- openstack-tox-docs http://logs.openstack.org/97/667797/2/check/openstack-tox-docs/89bf891/html/ : SUCCESS in 3m 02s\n- kolla-build-ubuntu-binary http://logs.openstack.org/97/667797/2/check/kolla-build-ubuntu-binary/b5bb7c7/ : SUCCESS in 21m 46s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/97/667797/2/check/kolla-build-ubuntu-source/af87fa7/ : SUCCESS in 29m 30s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/97/667797/2/check/kolla-ansible-ubuntu-source/b591ddd/ : SUCCESS in 53m 13s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/97/667797/2/check/kolla-ansible-ubuntu-binary/2880dd6/ : SUCCESS in 52m 34s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/97/667797/2/check/kolla-ansible-ubuntu-source-upgrade/4f74a9e/ : SUCCESS in 56m 20s","accounts_in_message":[],"_revision_number":2},{"id":"3bad2bb74c49c1b0e27503f6f2c9fab869393d61","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-27 13:49:59.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"382ef20be17d1498975b1e51382ffbe20f444fc6","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-27 14:28:39.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/97/667797/2/openstack-check/tripleo-build-containers-fedora-28-master/ae2ed96/ : FAILURE in 5m 22s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/97/667797/2/openstack-check/tripleo-build-containers-fedora-28-stein/c828829/ : SUCCESS in 33m 56s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d5b874fb757c73c6c88d8a7b4b219f566868e7de","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 15:28:19.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/97/667797/2/check/kolla-build-centos-binary/ba8b7ba/ : SUCCESS in 47m 46s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/97/667797/2/check/kolla-build-centos-source/095e566/ : SUCCESS in 47m 00s\n- kolla-ansible-centos-source http://logs.openstack.org/97/667797/2/check/kolla-ansible-centos-source/ffb3d46/ : SUCCESS in 1h 06m 25s\n- kolla-ansible-centos-binary http://logs.openstack.org/97/667797/2/check/kolla-ansible-centos-binary/5ccacb3/ : SUCCESS in 1h 11m 58s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/97/667797/2/check/kolla-ansible-centos-source-upgrade/1494085/ : SUCCESS in 1h 27m 16s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/97/667797/2/check/tripleo-build-containers-centos-7/bc09bcb/ : SUCCESS in 48m 07s\n- openstack-tox-lower-constraints http://logs.openstack.org/97/667797/2/check/openstack-tox-lower-constraints/993bf23/ : SUCCESS in 3m 12s\n- openstack-tox-pep8 http://logs.openstack.org/97/667797/2/check/openstack-tox-pep8/31680f1/ : SUCCESS in 3m 21s\n- openstack-tox-py27 http://logs.openstack.org/97/667797/2/check/openstack-tox-py27/bc9ab7a/ : SUCCESS in 3m 14s\n- openstack-tox-py36 http://logs.openstack.org/97/667797/2/check/openstack-tox-py36/82fa3f2/ : SUCCESS in 3m 33s\n- openstack-tox-py37 http://logs.openstack.org/97/667797/2/check/openstack-tox-py37/c43890f/ : SUCCESS in 3m 51s\n- openstack-tox-docs http://logs.openstack.org/97/667797/2/check/openstack-tox-docs/b4e604b/html/ : SUCCESS in 3m 10s\n- kolla-build-ubuntu-binary http://logs.openstack.org/97/667797/2/check/kolla-build-ubuntu-binary/8700f20/ : SUCCESS in 36m 11s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/97/667797/2/check/kolla-build-ubuntu-source/5a47edc/ : SUCCESS in 30m 35s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/97/667797/2/check/kolla-ansible-ubuntu-source/a203bd1/ : SUCCESS in 49m 54s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/97/667797/2/check/kolla-ansible-ubuntu-binary/44d9634/ : SUCCESS in 45m 36s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/97/667797/2/check/kolla-ansible-ubuntu-source-upgrade/ab7b835/ : SUCCESS in 53m 22s","accounts_in_message":[],"_revision_number":2},{"id":"52e03107e90f6e96149c32a1f36a288bb5129e85","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-28 14:52:58.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"1cae7bcfbc9b2c6dfbaa4450383174a5ef299d6a","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-28 14:55:14.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"68ccd6698809c1b3e5b067a8d008bf84899b8779","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-28 15:35:49.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/97/667797/3/openstack-check/tripleo-build-containers-fedora-28-master/87faca9/ : FAILURE in 7m 26s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/97/667797/3/openstack-check/tripleo-build-containers-fedora-28-stein/2d41dd2/ : SUCCESS in 33m 48s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e79bc345c0b96459074174e08a6beb9c45fc135d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-28 16:26:07.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/97/667797/3/check/kolla-build-centos-binary/2dd7085/ : SUCCESS in 37m 18s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/97/667797/3/check/kolla-build-centos-source/e69340a/ : SUCCESS in 48m 06s\n- kolla-ansible-centos-source http://logs.openstack.org/97/667797/3/check/kolla-ansible-centos-source/f47558a/ : SUCCESS in 1h 15m 59s\n- kolla-ansible-centos-binary http://logs.openstack.org/97/667797/3/check/kolla-ansible-centos-binary/a5c29dc/ : SUCCESS in 1h 00m 44s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/97/667797/3/check/kolla-ansible-centos-source-upgrade/b9ad156/ : SUCCESS in 1h 18m 16s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/97/667797/3/check/tripleo-build-containers-centos-7/c259d8f/ : SUCCESS in 41m 21s\n- openstack-tox-lower-constraints http://logs.openstack.org/97/667797/3/check/openstack-tox-lower-constraints/4e47e66/ : SUCCESS in 3m 43s\n- openstack-tox-pep8 http://logs.openstack.org/97/667797/3/check/openstack-tox-pep8/2502dce/ : SUCCESS in 3m 04s\n- openstack-tox-py27 http://logs.openstack.org/97/667797/3/check/openstack-tox-py27/b072ab0/ : SUCCESS in 3m 15s\n- openstack-tox-py36 http://logs.openstack.org/97/667797/3/check/openstack-tox-py36/bc04e4d/ : SUCCESS in 3m 20s\n- openstack-tox-py37 http://logs.openstack.org/97/667797/3/check/openstack-tox-py37/881c1b8/ : SUCCESS in 3m 49s\n- openstack-tox-docs http://logs.openstack.org/97/667797/3/check/openstack-tox-docs/eb44b07/html/ : SUCCESS in 4m 16s\n- kolla-build-ubuntu-binary http://logs.openstack.org/97/667797/3/check/kolla-build-ubuntu-binary/270f4e1/ : SUCCESS in 23m 12s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/97/667797/3/check/kolla-build-ubuntu-source/4a02f19/ : SUCCESS in 30m 41s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/97/667797/3/check/kolla-ansible-ubuntu-source/4ed2abb/ : SUCCESS in 46m 42s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/97/667797/3/check/kolla-ansible-ubuntu-binary/6a192b4/ : SUCCESS in 47m 46s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/97/667797/3/check/kolla-ansible-ubuntu-source-upgrade/f064d79/ : SUCCESS in 53m 20s","accounts_in_message":[],"_revision_number":3},{"id":"f8094e25ade4c3d4c6c8b56616d217f7eeb16191","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-29 16:13:15.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ce6bbecedc2c5ecdd1c7aed06107c808c8dedc2c","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-30 06:58:58.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"13f8bdc0fcb3ba14454c4f1103b3cef69029cfa8","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-07-01 11:01:32.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"7107d9dc90296cde9f51e45217fd7d0599084975","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-07-03 06:09:39.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nI like the comments approach :)","accounts_in_message":[],"_revision_number":3},{"id":"5411164b2862c0413da44458983d3be435ca2e77","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-03 06:09:58.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"a0cdce60b8b2be0423995daa42a4ef67ec40bf4e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-03 07:37:58.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source http://logs.openstack.org/97/667797/3/gate/kolla-build-centos-source/d21b1f8/ : SUCCESS in 55m 02s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/97/667797/3/gate/tripleo-build-containers-centos-7/9c72c51/ : SUCCESS in 43m 32s\n- openstack-tox-lower-constraints http://logs.openstack.org/97/667797/3/gate/openstack-tox-lower-constraints/52e8e63/ : SUCCESS in 3m 05s\n- openstack-tox-pep8 http://logs.openstack.org/97/667797/3/gate/openstack-tox-pep8/d3a9033/ : SUCCESS in 2m 55s\n- openstack-tox-py27 http://logs.openstack.org/97/667797/3/gate/openstack-tox-py27/6ce686d/ : SUCCESS in 3m 05s\n- openstack-tox-py36 http://logs.openstack.org/97/667797/3/gate/openstack-tox-py36/47b206d/ : SUCCESS in 2m 54s\n- openstack-tox-py37 http://logs.openstack.org/97/667797/3/gate/openstack-tox-py37/8bd72b5/ : SUCCESS in 3m 18s\n- openstack-tox-docs http://logs.openstack.org/97/667797/3/gate/openstack-tox-docs/6c6540f/html/ : SUCCESS in 3m 55s\n- kolla-build-ubuntu-source http://logs.openstack.org/97/667797/3/gate/kolla-build-ubuntu-source/09842cb/ : SUCCESS in 34m 34s","accounts_in_message":[],"_revision_number":3},{"id":"146497c1dce6b4baa0cf58cf2526fe21b0ff1812","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-03 07:37:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"fadabf0e46d89c85e656ff3b86c7cd01f0356219","revisions":{"27405d4af42d4afcc710f1280d95d63b63a2078d":{"kind":"REWORK","_number":1,"created":"2019-06-27 07:15:19.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/97/667797/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/97/667797/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/97/667797/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/97/667797/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/97/667797/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/97/667797/1"}}},"commit":{"parents":[{"commit":"c9547ea34f49fd5d5b3d703abcb83d764c908e0e","subject":"Merge \"Speed up builds of horizon slowed down by recent upstream change\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c9547ea34f49fd5d5b3d703abcb83d764c908e0e"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-06-27 07:12:19.000000000","tz":0},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-06-27 07:12:19.000000000","tz":0},"subject":"Add more entries into SKIPPED_IMAGES","message":"Add more entries into SKIPPED_IMAGES\n\nDid build of all images for AArch64, looked at failures and here is a\nlist of ones which never build due to lack of packages.\n\nThere are some images which could get to the list but there are already\npatches for them in Gerrit.\n\nAlso added three more entries for Debian due to missing packages.\n\nChange-Id: Ie259a38aad84bd8f5392f19791edfbd1f8ec4b1a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/27405d4af42d4afcc710f1280d95d63b63a2078d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/27405d4af42d4afcc710f1280d95d63b63a2078d"}]},"branch":"refs/heads/master"},"71079b562c293f02a2ae626d22c98ad8d6f84f34":{"kind":"REWORK","_number":2,"created":"2019-06-27 08:24:46.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/97/667797/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/97/667797/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/97/667797/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/97/667797/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/97/667797/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/97/667797/2"}}},"commit":{"parents":[{"commit":"c9547ea34f49fd5d5b3d703abcb83d764c908e0e","subject":"Merge \"Speed up builds of horizon slowed down by recent upstream change\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c9547ea34f49fd5d5b3d703abcb83d764c908e0e"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-06-27 07:12:19.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-06-27 08:23:45.000000000","tz":0},"subject":"Add more entries into SKIPPED_IMAGES","message":"Add more entries into SKIPPED_IMAGES\n\nDid build of all images for AArch64, looked at failures and here is a\nlist of ones which never build due to lack of packages.\n\nThere are some images which could get to the list but there are already\npatches for them in Gerrit.\n\nAlso added three more entries for Debian due to missing packages.\n\nChange-Id: Ie259a38aad84bd8f5392f19791edfbd1f8ec4b1a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/71079b562c293f02a2ae626d22c98ad8d6f84f34"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/71079b562c293f02a2ae626d22c98ad8d6f84f34"}]},"branch":"refs/heads/master"},"fadabf0e46d89c85e656ff3b86c7cd01f0356219":{"kind":"REWORK","_number":3,"created":"2019-06-28 14:52:58.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/97/667797/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/97/667797/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/97/667797/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/97/667797/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/97/667797/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/97/667797/3"}}},"commit":{"parents":[{"commit":"c9547ea34f49fd5d5b3d703abcb83d764c908e0e","subject":"Merge \"Speed up builds of horizon slowed down by recent upstream change\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c9547ea34f49fd5d5b3d703abcb83d764c908e0e"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-06-27 07:12:19.000000000","tz":0},"committer":{"name":"Marcin Juszkiewicz","email":"marcin@juszkiewicz.com.pl","date":"2019-06-28 14:52:33.000000000","tz":120},"subject":"Add more entries into SKIPPED_IMAGES","message":"Add more entries into SKIPPED_IMAGES\n\nDid build of all images for AArch64, looked at failures and here is a\nlist of ones which never build due to lack of packages.\n\nThere are some images which could get to the list but there are already\npatches for them in Gerrit.\n\nAlso added three more entries for Debian due to missing packages.\n\nChange-Id: Ie259a38aad84bd8f5392f19791edfbd1f8ec4b1a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fadabf0e46d89c85e656ff3b86c7cd01f0356219"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fadabf0e46d89c85e656ff3b86c7cd01f0356219"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
