)]}'
{"id":"openstack%2Fkolla~667824","triplet_id":"openstack%2Fkolla~stable%2Fstein~I1d69ec0ca97dbbacccfe737fca03c9739244b900","project":"openstack/kolla","branch":"stable/stein","topic":"fix-ceph-nfs-deploy-stable/stein","hashtags":[],"change_id":"I1d69ec0ca97dbbacccfe737fca03c9739244b900","subject":"Fix deployment of ceph-nfs container failing due to wrong USER","status":"MERGED","created":"2019-06-27 09:10:34.000000000","updated":"2019-06-27 22:25:07.000000000","submitted":"2019-06-27 22:25:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"667824-1561674307667-c8b2cb04","meta_rev_id":"8933e0e5e6b99f2af81b00387c21001437da988f","_number":667824,"virtual_id_number":667824,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":2,"date":"2019-06-27 22:25:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-06-27 09:48:48.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2019-06-27 21:22:47.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2019-06-27 14:24:00.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2019-06-27 09:10:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":1,"date":"2019-06-27 21:22:47.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-27 09:10:47.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-06-27 09:48:48.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-06-27 21:22:47.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-06-27 22:25:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"dccaac8dda1900725ff49126af22be876cccdaea","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-27 09:10:34.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"c94ea215cc266727b1ec61d82f0d31cde4879bad","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-27 09:10:47.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"0d10e725f5be1d3a2f793d055a924cf05fa5136d","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-27 09:48:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/24/667824/1/openstack-check/tripleo-build-containers-fedora-28-master/cce3ea3/ : FAILURE in 5m 31s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/24/667824/1/openstack-check/tripleo-build-containers-fedora-28-stein/d74db87/ : SUCCESS in 35m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"11554e8955cc65828c190476b3a49214c4c405bd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 12:31:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/24/667824/1/check/kolla-build-centos-binary/fcd0b5a/ : SUCCESS in 40m 03s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/24/667824/1/check/kolla-build-centos-source/59ffdf3/ : SUCCESS in 1h 01m 52s\n- kolla-ansible-centos-source http://logs.openstack.org/24/667824/1/check/kolla-ansible-centos-source/e237731/ : SUCCESS in 1h 19m 06s\n- kolla-ansible-centos-binary http://logs.openstack.org/24/667824/1/check/kolla-ansible-centos-binary/8a8d555/ : SUCCESS in 1h 07m 50s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/24/667824/1/check/kolla-ansible-centos-source-upgrade/40085bf/ : SUCCESS in 1h 13m 36s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/24/667824/1/check/tripleo-build-containers-centos-7/e8d8c6c/ : SUCCESS in 44m 14s\n- kolla-ansible-centos-source-ceph http://logs.openstack.org/24/667824/1/check/kolla-ansible-centos-source-ceph/04d0107/ : SUCCESS in 1h 33m 58s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/24/667824/1/check/kolla-build-oraclelinux-binary/47e5354/ : SUCCESS in 50m 17s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/24/667824/1/check/kolla-build-oraclelinux-source/272c52d/ : SUCCESS in 34m 47s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/24/667824/1/check/kolla-ansible-oraclelinux-source/abe06fe/ : SUCCESS in 1h 09m 57s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/24/667824/1/check/kolla-ansible-oraclelinux-binary/fb589c8/ : SUCCESS in 1h 08m 37s (non-voting)\n- kolla-ansible-oraclelinux-source-ceph http://logs.openstack.org/24/667824/1/check/kolla-ansible-oraclelinux-source-ceph/38a0105/ : TIMED_OUT in 2h 31m 53s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/24/667824/1/check/openstack-tox-lower-constraints/87c777e/ : SUCCESS in 3m 43s\n- openstack-tox-pep8 http://logs.openstack.org/24/667824/1/check/openstack-tox-pep8/14f8ff5/ : SUCCESS in 3m 27s\n- openstack-tox-py27 http://logs.openstack.org/24/667824/1/check/openstack-tox-py27/46c6be2/ : SUCCESS in 3m 11s\n- openstack-tox-py36 http://logs.openstack.org/24/667824/1/check/openstack-tox-py36/f315a73/ : SUCCESS in 3m 20s\n- openstack-tox-py37 http://logs.openstack.org/24/667824/1/check/openstack-tox-py37/256e6af/ : SUCCESS in 3m 41s\n- openstack-tox-docs http://logs.openstack.org/24/667824/1/check/openstack-tox-docs/95312f1/html/ : SUCCESS in 2m 43s\n- kolla-build-ubuntu-binary http://logs.openstack.org/24/667824/1/check/kolla-build-ubuntu-binary/f1802de/ : SUCCESS in 23m 49s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/24/667824/1/check/kolla-build-ubuntu-source/33b848d/ : SUCCESS in 33m 07s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/24/667824/1/check/kolla-ansible-ubuntu-source/ba40148/ : SUCCESS in 59m 55s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/24/667824/1/check/kolla-ansible-ubuntu-binary/11bfaa0/ : SUCCESS in 42m 50s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/24/667824/1/check/kolla-ansible-ubuntu-source-upgrade/00ad277/ : SUCCESS in 57m 17s\n- kolla-ansible-ubuntu-source-ceph http://logs.openstack.org/24/667824/1/check/kolla-ansible-ubuntu-source-ceph/f81ed82/ : SUCCESS in 1h 14m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9077c08523773da44b18c7b2e370a466ad78fd58","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-27 14:24:00.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"6a56210f9c87643157c18e1f4f0e9bf88ddf6201","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-27 21:22:47.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4a1f67e0d3dc06f0c87fe179f257a9697b15b30f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 21:23:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0c9a14d58a2bf911f5e6c70d0d32bee5c61d0240","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 22:25:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"b1ec9cac8309fffdeff0d21b51fe607fcc993f53","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 22:25:07.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source http://logs.openstack.org/24/667824/1/gate/kolla-build-centos-source/572b106/ : SUCCESS in 44m 21s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/24/667824/1/gate/tripleo-build-containers-centos-7/9100c5f/ : SUCCESS in 46m 01s\n- kolla-build-oraclelinux-source http://logs.openstack.org/24/667824/1/gate/kolla-build-oraclelinux-source/a37d24e/ : SUCCESS in 55m 49s\n- openstack-tox-lower-constraints http://logs.openstack.org/24/667824/1/gate/openstack-tox-lower-constraints/4c1de6b/ : SUCCESS in 3m 15s\n- openstack-tox-pep8 http://logs.openstack.org/24/667824/1/gate/openstack-tox-pep8/3328d46/ : SUCCESS in 3m 26s\n- openstack-tox-py27 http://logs.openstack.org/24/667824/1/gate/openstack-tox-py27/a0ef4df/ : SUCCESS in 2m 50s\n- openstack-tox-py36 http://logs.openstack.org/24/667824/1/gate/openstack-tox-py36/e36423c/ : SUCCESS in 3m 05s\n- openstack-tox-py37 http://logs.openstack.org/24/667824/1/gate/openstack-tox-py37/b954bf6/ : SUCCESS in 3m 46s\n- openstack-tox-docs http://logs.openstack.org/24/667824/1/gate/openstack-tox-docs/725018e/html/ : SUCCESS in 2m 52s\n- kolla-build-ubuntu-source http://logs.openstack.org/24/667824/1/gate/kolla-build-ubuntu-source/50603fc/ : SUCCESS in 28m 24s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f13852289da0ba996cf03b5773facacb9ab2085b","revisions":{"f13852289da0ba996cf03b5773facacb9ab2085b":{"kind":"REWORK","_number":1,"created":"2019-06-27 09:10:34.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/24/667824/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/24/667824/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/24/667824/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/24/667824/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/24/667824/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/24/667824/1"}}},"commit":{"parents":[{"commit":"4edaf8a7031308372b3c9c6171bd2a25af96446c","subject":"Move back to CentOS StorageSIG repos for Ceph and Ganesha","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4edaf8a7031308372b3c9c6171bd2a25af96446c"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-06-19 07:45:24.000000000","tz":120},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-06-27 09:10:34.000000000","tz":0},"subject":"Fix deployment of ceph-nfs container failing due to wrong USER","message":"Fix deployment of ceph-nfs container failing due to wrong USER\n\nThe startup script requires to be run as root, not ceph.\nAll other ceph images follow this rule.\n\nAlso removed redundant blank line.\n\nChange-Id: I1d69ec0ca97dbbacccfe737fca03c9739244b900\nSigned-off-by: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n(cherry picked from commit dc879dff3bf5cea9d64469395fe140b2ac7e4a3d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f13852289da0ba996cf03b5773facacb9ab2085b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f13852289da0ba996cf03b5773facacb9ab2085b"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
