)]}'
{"id":"openstack%2Fkolla~667825","triplet_id":"openstack%2Fkolla~stable%2Frocky~I1d69ec0ca97dbbacccfe737fca03c9739244b900","project":"openstack/kolla","branch":"stable/rocky","topic":"fix-ceph-nfs-deploy-stable/rocky","hashtags":[],"change_id":"I1d69ec0ca97dbbacccfe737fca03c9739244b900","subject":"Fix deployment of ceph-nfs container failing due to wrong USER","status":"MERGED","created":"2019-06-27 09:10:54.000000000","updated":"2019-07-03 07:37:54.000000000","submitted":"2019-07-03 07:37:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"667825-1562139474075-2d6c1d39","meta_rev_id":"1e87d96d6effbe3c5eef02506af8ba6087813584","_number":667825,"virtual_id_number":667825,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"date":"2019-07-01 10:41:37.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":2,"date":"2019-07-03 07:37:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-07-01 11:23:11.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"all":[{"value":2,"date":"2019-06-28 07:25:54.000000000","_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":2,"date":"2019-07-03 06:04:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2019-06-28 06:34:16.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2019-06-27 09:11:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":1,"date":"2019-07-03 06:04:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-27 09:11:03.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-06-28 06:34:16.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-06-28 07:25:54.000000000","updated_by":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"reviewer":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"state":"REVIEWER"},{"updated":"2019-07-01 11:23:11.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-03 06:04:39.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2019-07-03 07:37:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7a4fc1a94bbde7d400b5190382b57ccf3adb1eb8","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-27 09:10:54.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"ee77a06f286f3ec5a9cdd26e5ccd8fbca72071a6","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-27 09:11:03.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"d6244c097358fecc6427abb9fbf3bc0eec583eca","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-27 09:49:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/25/667825/1/openstack-check/tripleo-build-containers-fedora-28-master/9ef87ed/ : RETRY_LIMIT in 2m 42s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/25/667825/1/openstack-check/tripleo-build-containers-fedora-28-stein/07569de/ : SUCCESS in 34m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ade8ab118e6db0a0cca7f833818091fe54cc9dbf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 11:27:04.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/25/667825/1/check/kolla-build-centos-binary/da514e8/ : SUCCESS in 26m 18s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/25/667825/1/check/kolla-build-centos-source/6fdcddc/ : SUCCESS in 37m 16s\n- kolla-ansible-centos-source http://logs.openstack.org/25/667825/1/check/kolla-ansible-centos-source/d513bc6/ : POST_FAILURE in 5m 09s\n- kolla-ansible-centos-binary http://logs.openstack.org/25/667825/1/check/kolla-ansible-centos-binary/4e5857c/ : POST_FAILURE in 2m 31s (non-voting)\n- kolla-ansible-centos-source-ceph http://logs.openstack.org/25/667825/1/check/kolla-ansible-centos-source-ceph/9f2c657/ : POST_FAILURE in 2m 59s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/25/667825/1/check/kolla-build-oraclelinux-binary/56611cf/ : SUCCESS in 37m 27s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/25/667825/1/check/kolla-build-oraclelinux-source/e1044c7/ : SUCCESS in 42m 36s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/25/667825/1/check/kolla-ansible-oraclelinux-source/9b9526b/ : POST_FAILURE in 3m 29s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/25/667825/1/check/kolla-ansible-oraclelinux-binary/0de7fce/ : POST_FAILURE in 2m 47s (non-voting)\n- kolla-ansible-oraclelinux-source-ceph http://logs.openstack.org/25/667825/1/check/kolla-ansible-oraclelinux-source-ceph/583eccc/ : POST_FAILURE in 2m 45s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/25/667825/1/check/openstack-tox-lower-constraints/a62f7c8/ : SUCCESS in 3m 26s\n- openstack-tox-pep8 http://logs.openstack.org/25/667825/1/check/openstack-tox-pep8/9323dfe/ : SUCCESS in 4m 35s\n- openstack-tox-py27 http://logs.openstack.org/25/667825/1/check/openstack-tox-py27/593c833/ : SUCCESS in 3m 05s\n- openstack-tox-py35 http://logs.openstack.org/25/667825/1/check/openstack-tox-py35/74d9aee/ : SUCCESS in 2m 59s\n- build-openstack-sphinx-docs http://logs.openstack.org/25/667825/1/check/build-openstack-sphinx-docs/4dd98d9/html/ : SUCCESS in 3m 27s\n- kolla-build-ubuntu-binary http://logs.openstack.org/25/667825/1/check/kolla-build-ubuntu-binary/a580a9f/ : SUCCESS in 1h 26m 50s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/25/667825/1/check/kolla-build-ubuntu-source/ab577be/ : SUCCESS in 32m 27s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/25/667825/1/check/kolla-ansible-ubuntu-source/d074ee6/ : POST_FAILURE in 2m 59s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/25/667825/1/check/kolla-ansible-ubuntu-binary/fe6a57b/ : POST_FAILURE in 2m 35s (non-voting)\n- kolla-ansible-ubuntu-source-ceph http://logs.openstack.org/25/667825/1/check/kolla-ansible-ubuntu-source-ceph/3b36da2/ : POST_FAILURE in 2m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0422ba7a1a7e98ed68f4485e1cd75c7e8620e6c6","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-28 06:34:16.000000000","message":"Patch Set 1: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e3790630b6ba524ee32cc7bc45693ce055cf32df","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-28 07:14:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/25/667825/1/openstack-check/tripleo-build-containers-fedora-28-master/72d8e3a/ : RETRY_LIMIT in 2m 54s (non-voting)\n- tripleo-build-containers-fedora-28-stein finger://ze06.softwarefactory-project.io:7979/f7ae378ae1af4f1ba9351f2f1651ea49 : RETRY_LIMIT in 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a619a370d4bed14e5fce8cec8dfaa7b68e81bd98","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2019-06-28 07:25:54.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3f6f252024fba2743bb6b342ca0a9ef58d2354be","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-28 08:49:25.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/25/667825/1/check/kolla-build-centos-binary/70ce5d0/ : SUCCESS in 22m 26s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/25/667825/1/check/kolla-build-centos-source/e2a258e/ : SUCCESS in 47m 36s\n- kolla-ansible-centos-source http://logs.openstack.org/25/667825/1/check/kolla-ansible-centos-source/cf739aa/ : POST_FAILURE in 3m 27s\n- kolla-ansible-centos-binary http://logs.openstack.org/25/667825/1/check/kolla-ansible-centos-binary/e133657/ : POST_FAILURE in 3m 06s (non-voting)\n- kolla-ansible-centos-source-ceph http://logs.openstack.org/25/667825/1/check/kolla-ansible-centos-source-ceph/1393598/ : POST_FAILURE in 2m 22s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/25/667825/1/check/kolla-build-oraclelinux-binary/6a425a0/ : SUCCESS in 48m 16s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/25/667825/1/check/kolla-build-oraclelinux-source/376ce5c/ : SUCCESS in 36m 27s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/25/667825/1/check/kolla-ansible-oraclelinux-source/45db6fb/ : POST_FAILURE in 4m 07s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/25/667825/1/check/kolla-ansible-oraclelinux-binary/85ed001/ : POST_FAILURE in 3m 04s (non-voting)\n- kolla-ansible-oraclelinux-source-ceph http://logs.openstack.org/25/667825/1/check/kolla-ansible-oraclelinux-source-ceph/6581b40/ : POST_FAILURE in 2m 49s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/25/667825/1/check/openstack-tox-lower-constraints/12e53cc/ : SUCCESS in 4m 10s\n- openstack-tox-pep8 http://logs.openstack.org/25/667825/1/check/openstack-tox-pep8/f86771d/ : SUCCESS in 3m 44s\n- openstack-tox-py27 http://logs.openstack.org/25/667825/1/check/openstack-tox-py27/0d8a2eb/ : SUCCESS in 2m 45s\n- openstack-tox-py35 http://logs.openstack.org/25/667825/1/check/openstack-tox-py35/691bc7a/ : SUCCESS in 3m 16s\n- build-openstack-sphinx-docs http://logs.openstack.org/25/667825/1/check/build-openstack-sphinx-docs/d9a51c2/html/ : SUCCESS in 4m 16s\n- kolla-build-ubuntu-binary http://logs.openstack.org/25/667825/1/check/kolla-build-ubuntu-binary/cb80a44/ : SUCCESS in 1h 22m 24s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/25/667825/1/check/kolla-build-ubuntu-source/5f0f6b8/ : SUCCESS in 31m 33s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/25/667825/1/check/kolla-ansible-ubuntu-source/01af8b4/ : POST_FAILURE in 4m 27s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/25/667825/1/check/kolla-ansible-ubuntu-binary/54ae17b/ : POST_FAILURE in 2m 52s (non-voting)\n- kolla-ansible-ubuntu-source-ceph http://logs.openstack.org/25/667825/1/check/kolla-ansible-ubuntu-source-ceph/ba202df/ : POST_FAILURE in 2m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8996d772f6634e1ff706e14b1b07697f6683d12d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-28 08:52:23.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"84ea591181f5214f3c4525a65b396af44f8066f2","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-28 09:15:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/25/667825/1/openstack-check/tripleo-build-containers-fedora-28-master/6ad7acd/ : RETRY_LIMIT in 9m 42s (non-voting)\n- tripleo-build-containers-fedora-28-stein finger://ze03.softwarefactory-project.io:7979/bea192d4cfcf43de9ed86e7c97ea38e4 : RETRY_LIMIT in 8m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f5c9a0a01e53bee67e8f41a0db2919ed0ae5595f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-28 10:57:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/25/667825/1/check/kolla-build-centos-binary/b5eb720/ : SUCCESS in 34m 18s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/25/667825/1/check/kolla-build-centos-source/8a4aac3/ : SUCCESS in 43m 46s\n- kolla-ansible-centos-source http://logs.openstack.org/25/667825/1/check/kolla-ansible-centos-source/d5bddda/ : SUCCESS in 1h 30m 46s\n- kolla-ansible-centos-binary http://logs.openstack.org/25/667825/1/check/kolla-ansible-centos-binary/8466afe/ : SUCCESS in 1h 04m 26s (non-voting)\n- kolla-ansible-centos-source-ceph http://logs.openstack.org/25/667825/1/check/kolla-ansible-centos-source-ceph/0a53649/ : POST_FAILURE in 2m 49s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/25/667825/1/check/kolla-build-oraclelinux-binary/59ef301/ : SUCCESS in 46m 19s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/25/667825/1/check/kolla-build-oraclelinux-source/bc2ceca/ : SUCCESS in 55m 01s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/25/667825/1/check/kolla-ansible-oraclelinux-source/a682dd7/ : SUCCESS in 1h 12m 20s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/25/667825/1/check/kolla-ansible-oraclelinux-binary/968a2a3/ : SUCCESS in 1h 05m 30s (non-voting)\n- kolla-ansible-oraclelinux-source-ceph http://logs.openstack.org/25/667825/1/check/kolla-ansible-oraclelinux-source-ceph/0cd775c/ : POST_FAILURE in 2m 44s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/25/667825/1/check/openstack-tox-lower-constraints/4b64aa1/ : SUCCESS in 5m 37s\n- openstack-tox-pep8 http://logs.openstack.org/25/667825/1/check/openstack-tox-pep8/3b6c3b3/ : SUCCESS in 3m 43s\n- openstack-tox-py27 http://logs.openstack.org/25/667825/1/check/openstack-tox-py27/9d98e43/ : SUCCESS in 3m 42s\n- openstack-tox-py35 http://logs.openstack.org/25/667825/1/check/openstack-tox-py35/d502cb7/ : SUCCESS in 3m 33s\n- build-openstack-sphinx-docs http://logs.openstack.org/25/667825/1/check/build-openstack-sphinx-docs/e89b3ab/html/ : SUCCESS in 3m 25s\n- kolla-build-ubuntu-binary http://logs.openstack.org/25/667825/1/check/kolla-build-ubuntu-binary/299ff53/ : SUCCESS in 1h 23m 36s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/25/667825/1/check/kolla-build-ubuntu-source/6bdeaa6/ : SUCCESS in 30m 36s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/25/667825/1/check/kolla-ansible-ubuntu-source/9287595/ : SUCCESS in 49m 26s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/25/667825/1/check/kolla-ansible-ubuntu-binary/f27bd4e/ : SUCCESS in 1h 55m 55s (non-voting)\n- kolla-ansible-ubuntu-source-ceph http://logs.openstack.org/25/667825/1/check/kolla-ansible-ubuntu-source-ceph/0250a31/ : POST_FAILURE in 2m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d39909468c53fa8c79131f22466e2e4710ff8f0c","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-07-01 10:41:37.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d74162283c53479ac325a093a0d4b12a7404c672","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-01 11:23:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/25/667825/1/openstack-check/tripleo-build-containers-fedora-28-master/628f563/ : RETRY_LIMIT in 3m 12s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/25/667825/1/openstack-check/tripleo-build-containers-fedora-28-stein/27feb04/ : SUCCESS in 36m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a7a4b7757e08232ffc24497254b820d50191157d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-01 12:24:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/25/667825/1/check/kolla-build-centos-binary/0896f73/ : SUCCESS in 41m 13s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/25/667825/1/check/kolla-build-centos-source/0c69ff2/ : SUCCESS in 44m 17s\n- kolla-ansible-centos-source http://logs.openstack.org/25/667825/1/check/kolla-ansible-centos-source/9d282ff/ : SUCCESS in 1h 10m 23s\n- kolla-ansible-centos-binary http://logs.openstack.org/25/667825/1/check/kolla-ansible-centos-binary/96e137d/ : SUCCESS in 1h 10m 25s (non-voting)\n- kolla-ansible-centos-source-ceph http://logs.openstack.org/25/667825/1/check/kolla-ansible-centos-source-ceph/c991a4d/ : POST_FAILURE in 4m 40s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/25/667825/1/check/kolla-build-oraclelinux-binary/dc0a78c/ : SUCCESS in 37m 55s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/25/667825/1/check/kolla-build-oraclelinux-source/cb7e054/ : SUCCESS in 42m 12s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/25/667825/1/check/kolla-ansible-oraclelinux-source/183bc81/ : SUCCESS in 1h 06m 42s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/25/667825/1/check/kolla-ansible-oraclelinux-binary/2a9d675/ : SUCCESS in 1h 04m 31s (non-voting)\n- kolla-ansible-oraclelinux-source-ceph http://logs.openstack.org/25/667825/1/check/kolla-ansible-oraclelinux-source-ceph/9e703dc/ : POST_FAILURE in 2m 17s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/25/667825/1/check/openstack-tox-lower-constraints/4a70865/ : SUCCESS in 3m 55s\n- openstack-tox-pep8 http://logs.openstack.org/25/667825/1/check/openstack-tox-pep8/f854c4a/ : SUCCESS in 10m 26s\n- openstack-tox-py27 http://logs.openstack.org/25/667825/1/check/openstack-tox-py27/45472da/ : SUCCESS in 3m 10s\n- openstack-tox-py35 http://logs.openstack.org/25/667825/1/check/openstack-tox-py35/60ffd1a/ : SUCCESS in 3m 19s\n- build-openstack-sphinx-docs http://logs.openstack.org/25/667825/1/check/build-openstack-sphinx-docs/1ce38c6/html/ : SUCCESS in 3m 39s\n- kolla-build-ubuntu-binary http://logs.openstack.org/25/667825/1/check/kolla-build-ubuntu-binary/090b095/ : SUCCESS in 1h 29m 03s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/25/667825/1/check/kolla-build-ubuntu-source/3454fc4/ : SUCCESS in 33m 36s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/25/667825/1/check/kolla-ansible-ubuntu-source/80a894f/ : SUCCESS in 1h 01m 13s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/25/667825/1/check/kolla-ansible-ubuntu-binary/23ef546/ : SUCCESS in 1h 29m 22s (non-voting)\n- kolla-ansible-ubuntu-source-ceph http://logs.openstack.org/25/667825/1/check/kolla-ansible-ubuntu-source-ceph/4326c5a/ : POST_FAILURE in 2m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fe2f283718a14e97dde01006ce083e1da8a0ed8e","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-07-03 06:04:39.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"755ab218f4f249f130dfbc1561410e6ff10ec3dc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-03 06:04:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d00fd043993aa1a97863f09892d2fc184f4c40e5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-03 07:37:53.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source http://logs.openstack.org/25/667825/1/gate/kolla-build-centos-source/f39d517/ : SUCCESS in 40m 06s\n- kolla-build-oraclelinux-source http://logs.openstack.org/25/667825/1/gate/kolla-build-oraclelinux-source/860ec88/ : SUCCESS in 36m 16s\n- openstack-tox-lower-constraints http://logs.openstack.org/25/667825/1/gate/openstack-tox-lower-constraints/c1f0ae0/ : SUCCESS in 5m 30s\n- openstack-tox-pep8 http://logs.openstack.org/25/667825/1/gate/openstack-tox-pep8/7c5cd91/ : SUCCESS in 5m 04s\n- openstack-tox-py27 http://logs.openstack.org/25/667825/1/gate/openstack-tox-py27/e61c2d1/ : SUCCESS in 4m 11s\n- openstack-tox-py35 http://logs.openstack.org/25/667825/1/gate/openstack-tox-py35/a4f7e83/ : SUCCESS in 3m 24s\n- build-openstack-sphinx-docs http://logs.openstack.org/25/667825/1/gate/build-openstack-sphinx-docs/a984b04/html/ : SUCCESS in 3m 51s\n- kolla-build-ubuntu-source http://logs.openstack.org/25/667825/1/gate/kolla-build-ubuntu-source/8155565/ : SUCCESS in 30m 20s","accounts_in_message":[],"_revision_number":1},{"id":"a799f8c25871e609b890c4f94a42fb5206fd44fe","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-03 07:37:54.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"eaccc3b27d4e7fda400fa29a01bad83714496048","revisions":{"eaccc3b27d4e7fda400fa29a01bad83714496048":{"kind":"REWORK","_number":1,"created":"2019-06-27 09:10:54.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/25/667825/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/25/667825/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/25/667825/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/25/667825/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/25/667825/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/25/667825/1"}}},"commit":{"parents":[{"commit":"cdb88009abd650de588da3ddfde353a2493a07bb","subject":"Merge \"Add sha256 in ceph fetch keys\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cdb88009abd650de588da3ddfde353a2493a07bb"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-06-19 07:45:24.000000000","tz":120},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-06-27 09:10:54.000000000","tz":0},"subject":"Fix deployment of ceph-nfs container failing due to wrong USER","message":"Fix deployment of ceph-nfs container failing due to wrong USER\n\nThe startup script requires to be run as root, not ceph.\nAll other ceph images follow this rule.\n\nAlso removed redundant blank line.\n\nChange-Id: I1d69ec0ca97dbbacccfe737fca03c9739244b900\nSigned-off-by: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n(cherry picked from commit dc879dff3bf5cea9d64469395fe140b2ac7e4a3d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/eaccc3b27d4e7fda400fa29a01bad83714496048"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/eaccc3b27d4e7fda400fa29a01bad83714496048"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
