)]}'
{"id":"openstack%2Fkolla~667938","triplet_id":"openstack%2Fkolla~stable%2Fstein~I0514c02aaee1a413a7979c3a426e0484df839861","project":"openstack/kolla","branch":"stable/stein","topic":"bug/1832860-stable/stein","hashtags":[],"change_id":"I0514c02aaee1a413a7979c3a426e0484df839861","subject":"Run nova api_db sync before db sync","status":"MERGED","created":"2019-06-27 14:24:57.000000000","updated":"2019-06-28 08:11:31.000000000","submitted":"2019-06-28 08:11:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"667938-1561709491110-c2951fcc","meta_rev_id":"98a2e293e17a62ecf51603e2ed6df21aef1acf45","_number":667938,"virtual_id_number":667938,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2019-06-28 08:11:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-06-28 07:05:47.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"all":[{"value":2,"date":"2019-06-28 07:20:35.000000000","_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":1,"date":"2019-06-28 07:19:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":2,"date":"2019-06-28 06:33:52.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"all":[{"value":1,"date":"2019-06-28 07:20:35.000000000","_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-28 06:33:52.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-06-28 07:05:47.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-06-28 07:19:35.000000000","updated_by":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"reviewer":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"state":"REVIEWER"},{"updated":"2019-06-28 07:20:35.000000000","updated_by":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"reviewer":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"state":"REVIEWER"},{"updated":"2019-06-28 08:11:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a487b2238e52e5bd21112c5454d0f2c9bda27ebe","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-27 14:24:57.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"b0a05d1d0eec4e546a6f37fee65acaf0c10d6f63","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-27 14:59:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/38/667938/1/openstack-check/tripleo-build-containers-fedora-28-master/c50f6fc/ : FAILURE in 5m 37s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/38/667938/1/openstack-check/tripleo-build-containers-fedora-28-stein/c4ac201/ : SUCCESS in 31m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"56cce703038bac3c83b1bf2951fd936e8186228e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 16:54:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/38/667938/1/check/kolla-build-centos-binary/fd96bed/ : SUCCESS in 47m 41s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/38/667938/1/check/kolla-build-centos-source/bf0d600/ : SUCCESS in 54m 29s\n- kolla-ansible-centos-source http://logs.openstack.org/38/667938/1/check/kolla-ansible-centos-source/e74196d/ : SUCCESS in 1h 16m 41s\n- kolla-ansible-centos-binary http://logs.openstack.org/38/667938/1/check/kolla-ansible-centos-binary/42ac253/ : RETRY_LIMIT in 3m 55s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/38/667938/1/check/kolla-ansible-centos-source-upgrade/aa3a7ae/ : SUCCESS in 1h 39m 22s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/38/667938/1/check/tripleo-build-containers-centos-7/20be3f5/ : SUCCESS in 45m 07s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/38/667938/1/check/kolla-build-oraclelinux-binary/6702f26/ : RETRY_LIMIT in 2m 04s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/38/667938/1/check/kolla-build-oraclelinux-source/4930a92/ : SUCCESS in 32m 51s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/38/667938/1/check/kolla-ansible-oraclelinux-source/b98751f/ : FAILURE in 33m 58s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/38/667938/1/check/kolla-ansible-oraclelinux-binary/567d8f8/ : SUCCESS in 1h 12m 43s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/38/667938/1/check/openstack-tox-lower-constraints/cc51e80/ : SUCCESS in 4m 53s\n- openstack-tox-pep8 http://logs.openstack.org/38/667938/1/check/openstack-tox-pep8/2371040/ : SUCCESS in 3m 55s\n- openstack-tox-py27 http://logs.openstack.org/38/667938/1/check/openstack-tox-py27/7a9d90e/ : SUCCESS in 4m 07s\n- openstack-tox-py36 http://logs.openstack.org/38/667938/1/check/openstack-tox-py36/0360cef/ : SUCCESS in 3m 36s\n- openstack-tox-py37 http://logs.openstack.org/38/667938/1/check/openstack-tox-py37/21abbf7/ : SUCCESS in 3m 40s\n- openstack-tox-docs http://logs.openstack.org/38/667938/1/check/openstack-tox-docs/09989dd/html/ : SUCCESS in 2m 52s\n- kolla-build-ubuntu-binary http://logs.openstack.org/38/667938/1/check/kolla-build-ubuntu-binary/5f9093c/ : SUCCESS in 22m 16s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/38/667938/1/check/kolla-build-ubuntu-source/654731a/ : SUCCESS in 28m 53s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/38/667938/1/check/kolla-ansible-ubuntu-source/98f0cbb/ : SUCCESS in 53m 18s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/38/667938/1/check/kolla-ansible-ubuntu-binary/fb48dc1/ : SUCCESS in 54m 21s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/38/667938/1/check/kolla-ansible-ubuntu-source-upgrade/f7a6243/ : SUCCESS in 1h 00m 12s","accounts_in_message":[],"_revision_number":1},{"id":"c22d8b50ea47afc2a2b67433f66bb28c0858a077","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-28 06:33:52.000000000","message":"Patch Set 1: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"409bf29e95a31bf5fdcc6ceb3b6abed77fef16dd","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-28 07:05:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master finger://ze05.softwarefactory-project.io:7979/61e653b7b8d5478ba6653bfa9615293f : RETRY_LIMIT in 52s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/38/667938/1/openstack-check/tripleo-build-containers-fedora-28-stein/bb2b9cd/ : FAILURE in 18m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2ee133b8971648f986c5209427f6b2539fd1e72a","author":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"date":"2019-06-28 07:19:35.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"0613065ca08dd131e1dbe9f7331d239d1c62d1b9","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2019-06-28 07:20:35.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"78e369c997bb86401322d1ebda808ed56c7934ff","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-28 07:20:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9245b295cc27470b7b21c86c2a06eec3ee1a4cfc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-28 08:09:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/38/667938/1/check/kolla-build-centos-binary/6c5eb1f/ : SUCCESS in 43m 39s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/38/667938/1/check/kolla-build-centos-source/d9b9717/ : SUCCESS in 45m 44s\n- kolla-ansible-centos-source http://logs.openstack.org/38/667938/1/check/kolla-ansible-centos-source/4ab421e/ : SUCCESS in 1h 11m 30s\n- kolla-ansible-centos-binary http://logs.openstack.org/38/667938/1/check/kolla-ansible-centos-binary/a4f340b/ : SUCCESS in 1h 08m 14s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/38/667938/1/check/kolla-ansible-centos-source-upgrade/788aad4/ : SUCCESS in 1h 12m 19s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/38/667938/1/check/tripleo-build-containers-centos-7/f23f9ac/ : SUCCESS in 37m 17s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/38/667938/1/check/kolla-build-oraclelinux-binary/eef3c4f/ : SUCCESS in 28m 42s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/38/667938/1/check/kolla-build-oraclelinux-source/69ce79e/ : SUCCESS in 39m 59s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/38/667938/1/check/kolla-ansible-oraclelinux-source/fc8754f/ : SUCCESS in 1h 08m 15s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/38/667938/1/check/kolla-ansible-oraclelinux-binary/6ea923e/ : SUCCESS in 1h 03m 59s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/38/667938/1/check/openstack-tox-lower-constraints/dd31178/ : SUCCESS in 3m 29s\n- openstack-tox-pep8 http://logs.openstack.org/38/667938/1/check/openstack-tox-pep8/5ace2a7/ : SUCCESS in 3m 24s\n- openstack-tox-py27 http://logs.openstack.org/38/667938/1/check/openstack-tox-py27/bfb4c09/ : SUCCESS in 3m 23s\n- openstack-tox-py36 http://logs.openstack.org/38/667938/1/check/openstack-tox-py36/e19c577/ : SUCCESS in 3m 02s\n- openstack-tox-py37 http://logs.openstack.org/38/667938/1/check/openstack-tox-py37/39dcd3a/ : SUCCESS in 3m 24s\n- openstack-tox-docs http://logs.openstack.org/38/667938/1/check/openstack-tox-docs/55d3f50/html/ : SUCCESS in 3m 15s\n- kolla-build-ubuntu-binary http://logs.openstack.org/38/667938/1/check/kolla-build-ubuntu-binary/7e2ee90/ : SUCCESS in 23m 17s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/38/667938/1/check/kolla-build-ubuntu-source/e890fa1/ : SUCCESS in 28m 15s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/38/667938/1/check/kolla-ansible-ubuntu-source/3a0d28f/ : SUCCESS in 51m 28s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/38/667938/1/check/kolla-ansible-ubuntu-binary/8002341/ : SUCCESS in 46m 28s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/38/667938/1/check/kolla-ansible-ubuntu-source-upgrade/5d3c085/ : SUCCESS in 58m 06s","accounts_in_message":[],"_revision_number":1},{"id":"8db8e2153233d195607a350a1fd8531bd4ad7350","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-28 08:11:30.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source http://logs.openstack.org/38/667938/1/gate/kolla-build-centos-source/84b9921/ : SUCCESS in 45m 12s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/38/667938/1/gate/tripleo-build-containers-centos-7/70e73dc/ : SUCCESS in 39m 57s\n- kolla-build-oraclelinux-source http://logs.openstack.org/38/667938/1/gate/kolla-build-oraclelinux-source/5195b82/ : SUCCESS in 43m 24s\n- openstack-tox-lower-constraints http://logs.openstack.org/38/667938/1/gate/openstack-tox-lower-constraints/24a6be4/ : SUCCESS in 3m 08s\n- openstack-tox-pep8 http://logs.openstack.org/38/667938/1/gate/openstack-tox-pep8/82bd6eb/ : SUCCESS in 2m 52s\n- openstack-tox-py27 http://logs.openstack.org/38/667938/1/gate/openstack-tox-py27/7edae33/ : SUCCESS in 2m 46s\n- openstack-tox-py36 http://logs.openstack.org/38/667938/1/gate/openstack-tox-py36/d93df92/ : SUCCESS in 3m 36s\n- openstack-tox-py37 http://logs.openstack.org/38/667938/1/gate/openstack-tox-py37/d6d28c3/ : SUCCESS in 3m 25s\n- openstack-tox-docs http://logs.openstack.org/38/667938/1/gate/openstack-tox-docs/98f6c37/html/ : SUCCESS in 2m 50s\n- kolla-build-ubuntu-source http://logs.openstack.org/38/667938/1/gate/kolla-build-ubuntu-source/eea0b5d/ : SUCCESS in 33m 33s","accounts_in_message":[],"_revision_number":1},{"id":"8d985cf98ec434d2d9216b00fcf20000d142c1c6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-28 08:11:31.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"09226a59866e5ace7d573bac2e0772109a2554a0","revisions":{"09226a59866e5ace7d573bac2e0772109a2554a0":{"kind":"REWORK","_number":1,"created":"2019-06-27 14:24:57.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/38/667938/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/667938/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/667938/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/667938/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/667938/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/667938/1"}}},"commit":{"parents":[{"commit":"efbd5ffb46afe37f505523c9569d898db0c028d8","subject":"Merge \"Add sha256 in ceph fetch keys\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/efbd5ffb46afe37f505523c9569d898db0c028d8"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-06-17 09:49:21.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-06-27 14:24:57.000000000","tz":0},"subject":"Run nova api_db sync before db sync","message":"Run nova api_db sync before db sync\n\nBackport: Stein, Rocky\n\nSince cells v2 was introduced, nova operators must run two commands to\nmigrate the database schemas of nova\u0027s databases - nova-manage api_db\nsync and nova-manage db sync. It is necessary to run them in this order,\nsince the db sync may depend on schema changes made to the api\ndatabase in the api_db sync. Executing the db sync first may fail,\nfor example with the following seen in a Queens to Rocky upgrade:\n\n    nova-manage db sync\n    ERROR: Could not access cell0.\n    Has the nova_api database been created?\n    Has the nova_cell0 database been created?\n    Has \"nova-manage api_db sync\" been run?\n    Has \"nova-manage cell_v2 map_cell0\" been run?\n    Is [api_database]/connection set in nova.conf?\n    Is the cell0 database connection URL correct?\n    Error: (pymysql.err.InternalError) (1054, u\"Unknown column\n            \u0027cell_mappings.disabled\u0027 in \u0027field list\u0027\") [SQL: u\u0027SELECT\n    cell_mappings.created_at AS cell_mappings_created_at,\n    cell_mappings.updated_at AS cell_mappings_updated_at,\n    cell_mappings.id AS cell_mappings_id, cell_mappings.uuid AS\n    cell_mappings_uuid, cell_mappings.name AS cell_mappings_name,\n    cell_mappings.transport_url AS cell_mappings_transport_url,\n    cell_mappings.database_connection AS\n    cell_mappings_database_connection, cell_mappings.disabled AS\n    cell_mappings_disabled \\nFROM cell_mappings \\nWHERE\n    cell_mappings.uuid \u003d %(uuid_1)s \\n LIMIT %(param_1)s\u0027] [parameters:\n    {u\u0027uuid_1\u0027: \u002700000000-0000-0000-0000-000000000000\u0027, u\u0027param_1\u0027: 1}]\n    (Background on this error at: http://sqlalche.me/e/2j85)\n\nDespite this error, the command actually exits zero, so kolla does not notice\nthe failure.\n\nThis change reorders these two commands to run the api_db sync before the db\nsync.\n\nChange-Id: I0514c02aaee1a413a7979c3a426e0484df839861\nCloses-Bug: #1832860\n(cherry picked from commit 84a07102e2db92333ce416a520151f1a1856f162)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/09226a59866e5ace7d573bac2e0772109a2554a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/09226a59866e5ace7d573bac2e0772109a2554a0"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
