)]}'
{"id":"openstack%2Fkolla~667962","triplet_id":"openstack%2Fkolla~stable%2Frocky~I046e72ccb03da4fbb67ef4be3ac365425b2c91c7","project":"openstack/kolla","branch":"stable/rocky","topic":"periodic-stable-rocky","hashtags":[],"change_id":"I046e72ccb03da4fbb67ef4be3ac365425b2c91c7","subject":"CI: Move to periodic-stable","status":"MERGED","created":"2019-06-27 15:30:03.000000000","updated":"2019-07-08 09:47:55.000000000","submitted":"2019-07-08 09:47:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"667962-1562579275019-72c518ec","meta_rev_id":"067beed075fd4640f8a5cd5c5b454c74b18302cb","_number":667962,"virtual_id_number":667962,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"date":"2019-07-04 09:14:09.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":2,"date":"2019-07-08 09:47:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-07-04 16:15:46.000000000","_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"all":[{"value":2,"date":"2019-07-08 08:32:51.000000000","_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":2,"date":"2019-07-03 06:06:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2019-07-08 08:52:05.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-03 06:06:00.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2019-07-04 16:15:46.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2019-07-08 08:32:51.000000000","updated_by":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"reviewer":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"state":"REVIEWER"},{"updated":"2019-07-08 08:52:05.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-07-08 09:47:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"dbcb306859f1faac49ad8debc5bc7c76afb6c4cf","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-27 15:30:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e38472fb4c2d896d68996f9af119b6b6db3a7b3d","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-27 15:31:43.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"8ee7a566e526909583c2c1e20e6ef4f0fdf0ad85","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 17:12:45.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/62/667962/2/check/kolla-build-centos-binary/7fa7b04/ : SUCCESS in 36m 31s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/62/667962/2/check/kolla-build-centos-source/948592f/ : SUCCESS in 46m 04s\n- kolla-ansible-centos-source http://logs.openstack.org/62/667962/2/check/kolla-ansible-centos-source/8acf492/ : POST_FAILURE in 3m 38s\n- kolla-ansible-centos-binary http://logs.openstack.org/62/667962/2/check/kolla-ansible-centos-binary/987fa88/ : POST_FAILURE in 2m 42s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/62/667962/2/check/kolla-build-oraclelinux-binary/9a7a458/ : SUCCESS in 26m 52s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/62/667962/2/check/kolla-build-oraclelinux-source/99aa154/ : SUCCESS in 50m 55s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/62/667962/2/check/kolla-ansible-oraclelinux-source/fa7569d/ : POST_FAILURE in 3m 22s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/62/667962/2/check/kolla-ansible-oraclelinux-binary/d68bc85/ : POST_FAILURE in 4m 49s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/62/667962/2/check/openstack-tox-lower-constraints/02b266d/ : SUCCESS in 5m 07s\n- openstack-tox-pep8 http://logs.openstack.org/62/667962/2/check/openstack-tox-pep8/3c6e6a7/ : SUCCESS in 4m 20s\n- openstack-tox-py27 http://logs.openstack.org/62/667962/2/check/openstack-tox-py27/eeb2629/ : SUCCESS in 4m 34s\n- openstack-tox-py35 http://logs.openstack.org/62/667962/2/check/openstack-tox-py35/92e6f1b/ : SUCCESS in 4m 50s\n- build-openstack-sphinx-docs http://logs.openstack.org/62/667962/2/check/build-openstack-sphinx-docs/6b2401b/html/ : SUCCESS in 4m 04s\n- kolla-build-ubuntu-binary http://logs.openstack.org/62/667962/2/check/kolla-build-ubuntu-binary/9e3181f/ : SUCCESS in 1h 18m 44s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/62/667962/2/check/kolla-build-ubuntu-source/eec3668/ : SUCCESS in 35m 13s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/62/667962/2/check/kolla-ansible-ubuntu-source/3b64c53/ : POST_FAILURE in 2m 59s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/62/667962/2/check/kolla-ansible-ubuntu-binary/ead62f7/ : POST_FAILURE in 2m 44s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d8aead03f10b18c90bc0ab29dabae71981ea9ef4","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-28 17:41:36.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"35dab457493ed568fbfbacaa3230dcc639859ad3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-28 19:14:32.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/62/667962/2/check/kolla-build-centos-binary/cf75b98/ : SUCCESS in 34m 33s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/62/667962/2/check/kolla-build-centos-source/4dc76d3/ : SUCCESS in 45m 17s\n- kolla-ansible-centos-source http://logs.openstack.org/62/667962/2/check/kolla-ansible-centos-source/1f6fadc/ : SUCCESS in 1h 10m 24s\n- kolla-ansible-centos-binary http://logs.openstack.org/62/667962/2/check/kolla-ansible-centos-binary/dc85193/ : SUCCESS in 1h 10m 10s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/62/667962/2/check/kolla-build-oraclelinux-binary/d93c727/ : SUCCESS in 39m 38s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/62/667962/2/check/kolla-build-oraclelinux-source/16d2570/ : SUCCESS in 57m 24s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/62/667962/2/check/kolla-ansible-oraclelinux-source/5435530/ : SUCCESS in 1h 13m 52s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/62/667962/2/check/kolla-ansible-oraclelinux-binary/b7c7d72/ : SUCCESS in 1h 09m 26s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/62/667962/2/check/openstack-tox-lower-constraints/51a392d/ : SUCCESS in 3m 41s\n- openstack-tox-pep8 http://logs.openstack.org/62/667962/2/check/openstack-tox-pep8/75e8441/ : SUCCESS in 3m 53s\n- openstack-tox-py27 http://logs.openstack.org/62/667962/2/check/openstack-tox-py27/82d7ddc/ : SUCCESS in 3m 35s\n- openstack-tox-py35 http://logs.openstack.org/62/667962/2/check/openstack-tox-py35/236051d/ : SUCCESS in 3m 34s\n- build-openstack-sphinx-docs http://logs.openstack.org/62/667962/2/check/build-openstack-sphinx-docs/c236707/html/ : SUCCESS in 3m 27s\n- kolla-build-ubuntu-binary http://logs.openstack.org/62/667962/2/check/kolla-build-ubuntu-binary/d9bec02/ : SUCCESS in 1h 27m 25s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/62/667962/2/check/kolla-build-ubuntu-source/4c9b038/ : SUCCESS in 31m 09s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/62/667962/2/check/kolla-ansible-ubuntu-source/00487d5/ : SUCCESS in 1h 01m 13s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/62/667962/2/check/kolla-ansible-ubuntu-binary/4dcab20/ : SUCCESS in 1h 26m 28s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f2306b98322dd70107350a4e055f56cea851287f","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-07-03 06:06:00.000000000","message":"Patch Set 2: Code-Review+2\n\nGood catch","accounts_in_message":[],"_revision_number":2},{"id":"93f417b215e985f2f0bec56204c4db8f821fbf79","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-07-04 08:21:55.000000000","message":"Patch Set 2:\n\nI thought the idea was that you should not need to update your zuul config when you branch. What does this get us?","accounts_in_message":[],"_revision_number":2},{"id":"4ce87d6e70513058592c5f90a3093b032455a372","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-07-04 08:25:26.000000000","message":"Patch Set 2:\n\nperiodic-stable runs an hour later than periodic and uses mail notifications. Other than that, there are no real differences I am aware of.","accounts_in_message":[],"_revision_number":2},{"id":"8abe1d56cf753f6dcdba0ef3936a82626afed636","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-07-04 08:47:06.000000000","message":"Patch Set 2:\n\n\u003e periodic-stable runs an hour later than periodic and uses mail\n \u003e notifications. Other than that, there are no real differences I am\n \u003e aware of.\n\nMail notifications would be nice. It would be nice if we had a way to avoid needing to change this in new stable branches in future. If we changed to periodic-stable in master, would zuul know to skip jobs on master in that pipeline?","accounts_in_message":[],"_revision_number":2},{"id":"3389f8726e01baa4c58f2d8751653c3f5da52823","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-07-04 09:14:09.000000000","message":"Patch Set 2:\n\nDue to July 4 in the USA, we need to postpone this discussion till Zuul masters return from their holidays.","accounts_in_message":[],"_revision_number":2},{"id":"de74e4465ff2bbd9d824b327079f11cf543e70a3","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2019-07-04 16:15:46.000000000","message":"Patch Set 2:\n\nI feel like this is more a question for https://docs.openstack.org/project-team-guide/stable-branches.html whether they want jobs for deployment projects reporting via the http://lists.openstack.org/cgi-bin/mailman/listinfo/openstack-stable-maint ML.\n\nAs for explicit branch matchers on periodic jobs in master branches, that should work (negative lookahead eliminating master for the stable-periodic pipeline, and a master branch only match for the periodic pipeline). Subsequently branching that configuration will continue to have the desired effect.","accounts_in_message":[],"_revision_number":2},{"id":"d881586f146ee2473445842f662dcfc1c084317b","author":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"date":"2019-07-08 08:32:51.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"91ce1e3d50e9396fd7d42e2280e657857ba11c8e","author":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"date":"2019-07-08 08:33:03.000000000","message":"Patch Set 2: -Workflow","accounts_in_message":[],"_revision_number":2},{"id":"2fb1dbf24222cd7f66d864dfbc78a644bdfef77b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-08 08:33:13.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"cdecd675a6d32fe289c0c6c20c7cb0e83fe40ddc","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-07-08 08:52:05.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"73cd0768ce61fae62539bb7307d064503d927da8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-08 09:47:54.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source http://logs.openstack.org/62/667962/2/gate/kolla-build-centos-source/41406cc/ : SUCCESS in 57m 23s\n- kolla-build-oraclelinux-source http://logs.openstack.org/62/667962/2/gate/kolla-build-oraclelinux-source/a5ffe79/ : SUCCESS in 50m 00s\n- openstack-tox-lower-constraints http://logs.openstack.org/62/667962/2/gate/openstack-tox-lower-constraints/940b5c1/ : SUCCESS in 4m 40s\n- openstack-tox-pep8 http://logs.openstack.org/62/667962/2/gate/openstack-tox-pep8/3aec1c2/ : SUCCESS in 4m 49s\n- openstack-tox-py27 http://logs.openstack.org/62/667962/2/gate/openstack-tox-py27/7da3708/ : SUCCESS in 3m 33s\n- openstack-tox-py35 http://logs.openstack.org/62/667962/2/gate/openstack-tox-py35/3887dc6/ : SUCCESS in 3m 48s\n- build-openstack-sphinx-docs http://logs.openstack.org/62/667962/2/gate/build-openstack-sphinx-docs/d4ba6cc/html/ : SUCCESS in 5m 09s\n- kolla-build-ubuntu-source http://logs.openstack.org/62/667962/2/gate/kolla-build-ubuntu-source/8291d55/ : SUCCESS in 41m 03s","accounts_in_message":[],"_revision_number":2},{"id":"1f895c73c2c6f7b418cf3458ae4f4d3726be457d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-08 09:47:55.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8c5e20e0cfddb5e60d952bb32aa67c88225f9baf","revisions":{"696fa18dab88cb0ab44616be325a00a0fe243057":{"kind":"REWORK","_number":1,"created":"2019-06-27 15:30:03.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/62/667962/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/62/667962/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/62/667962/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/62/667962/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/62/667962/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/62/667962/1"}}},"commit":{"parents":[{"commit":"cdb88009abd650de588da3ddfde353a2493a07bb","subject":"Merge \"Add sha256 in ceph fetch keys\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cdb88009abd650de588da3ddfde353a2493a07bb"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-06-27 15:29:36.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-06-27 15:29:36.000000000","tz":120},"subject":"CI: Move to periodic-stable","message":"CI: Move to periodic-stable\n\nChange-Id: I046e72ccb03da4fbb67ef4be3ac365425b2c91c7\nSigned-off-by: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/696fa18dab88cb0ab44616be325a00a0fe243057"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/696fa18dab88cb0ab44616be325a00a0fe243057"}]},"branch":"refs/heads/stable/rocky"},"8c5e20e0cfddb5e60d952bb32aa67c88225f9baf":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-06-27 15:31:43.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/62/667962/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/62/667962/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/62/667962/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/62/667962/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/62/667962/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/62/667962/2"}}},"commit":{"parents":[{"commit":"cdb88009abd650de588da3ddfde353a2493a07bb","subject":"Merge \"Add sha256 in ceph fetch keys\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cdb88009abd650de588da3ddfde353a2493a07bb"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-06-27 15:29:36.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-06-27 15:31:37.000000000","tz":0},"subject":"CI: Move to periodic-stable","message":"CI: Move to periodic-stable\n\nFor stable branches we should be using periodic-stable pipeline instead\nof periodic.\n\nChange-Id: I046e72ccb03da4fbb67ef4be3ac365425b2c91c7\nSigned-off-by: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8c5e20e0cfddb5e60d952bb32aa67c88225f9baf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8c5e20e0cfddb5e60d952bb32aa67c88225f9baf"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
