)]}'
{"id":"openstack%2Fkolla~668094","triplet_id":"openstack%2Fkolla~stable%2Fstein~I527aba3635ef82e2d2a88ee7f5250bf03e420800","project":"openstack/kolla","branch":"stable/stein","topic":"ci-remove-ol-overrides-stable/stein","hashtags":[],"change_id":"I527aba3635ef82e2d2a88ee7f5250bf03e420800","subject":"Remove Oracle Linux stuff in CI code","status":"ABANDONED","created":"2019-06-28 06:32:44.000000000","updated":"2019-07-23 08:42:53.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d2cb74a34fd5ff5e5b252f3794e4ba331f9f5348","_number":668094,"virtual_id_number":668094,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":1,"date":"2019-06-28 08:03:10.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":-1,"date":"2019-06-28 07:02:29.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-28 07:02:29.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-06-28 08:03:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"86feac748915be3c136bf432b58593ae0c7b864c","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-06-28 06:32:44.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"ce6f821fd1345991794989b49342ab886de2934c","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-28 07:02:29.000000000","message":"Patch Set 1: Code-Review-1\n\nBut we test OL for stein still...","accounts_in_message":[],"_revision_number":1},{"id":"06c861fd67bdfcc5a0e6c3383d28d9b37576bc1b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-28 08:03:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/94/668094/1/check/kolla-build-centos-binary/a69efb0/ : SUCCESS in 32m 06s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/94/668094/1/check/kolla-build-centos-source/528abc9/ : SUCCESS in 36m 30s\n- kolla-ansible-centos-source http://logs.openstack.org/94/668094/1/check/kolla-ansible-centos-source/412cffc/ : SUCCESS in 1h 06m 52s\n- kolla-ansible-centos-binary http://logs.openstack.org/94/668094/1/check/kolla-ansible-centos-binary/a2ee1f3/ : SUCCESS in 1h 02m 35s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/94/668094/1/check/kolla-ansible-centos-source-upgrade/5f1c1c1/ : SUCCESS in 1h 15m 52s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/94/668094/1/check/kolla-build-oraclelinux-binary/faf7637/ : SUCCESS in 49m 44s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/94/668094/1/check/kolla-build-oraclelinux-source/494f0eb/ : SUCCESS in 46m 22s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/94/668094/1/check/kolla-ansible-oraclelinux-source/a506a91/ : SUCCESS in 1h 12m 23s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/94/668094/1/check/kolla-ansible-oraclelinux-binary/2386197/ : SUCCESS in 1h 04m 35s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/94/668094/1/check/openstack-tox-lower-constraints/d51fd64/ : SUCCESS in 3m 47s\n- openstack-tox-pep8 http://logs.openstack.org/94/668094/1/check/openstack-tox-pep8/f2a1569/ : SUCCESS in 3m 25s\n- openstack-tox-py27 http://logs.openstack.org/94/668094/1/check/openstack-tox-py27/d2ddeb7/ : SUCCESS in 3m 07s\n- openstack-tox-py36 http://logs.openstack.org/94/668094/1/check/openstack-tox-py36/7674c3c/ : SUCCESS in 4m 27s\n- openstack-tox-py37 http://logs.openstack.org/94/668094/1/check/openstack-tox-py37/4426895/ : SUCCESS in 3m 41s\n- openstack-tox-docs http://logs.openstack.org/94/668094/1/check/openstack-tox-docs/15a0dd9/html/ : SUCCESS in 3m 03s\n- kolla-build-ubuntu-binary http://logs.openstack.org/94/668094/1/check/kolla-build-ubuntu-binary/8c499a3/ : SUCCESS in 23m 01s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/94/668094/1/check/kolla-build-ubuntu-source/798f5dd/ : SUCCESS in 28m 41s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/94/668094/1/check/kolla-ansible-ubuntu-source/80effb1/ : SUCCESS in 54m 06s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/94/668094/1/check/kolla-ansible-ubuntu-binary/b7a366c/ : SUCCESS in 43m 14s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/94/668094/1/check/kolla-ansible-ubuntu-source-upgrade/85eb908/ : SUCCESS in 55m 57s","accounts_in_message":[],"_revision_number":1},{"id":"344586300211e99955ad71f28b67973d25a42b48","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-07-23 08:42:53.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"848a089485036b100fbfce817fc5e9ca4f4b0c29","revisions":{"848a089485036b100fbfce817fc5e9ca4f4b0c29":{"kind":"REWORK","_number":1,"created":"2019-06-28 06:32:44.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/94/668094/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/94/668094/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/94/668094/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/94/668094/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/94/668094/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/94/668094/1"}}},"commit":{"parents":[{"commit":"4923e714b0ec7666405b91f2b3665ef36e72ae89","subject":"Merge \"Fix deployment of ceph-nfs container failing due to wrong USER\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4923e714b0ec7666405b91f2b3665ef36e72ae89"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-06-25 17:29:46.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-06-28 06:32:44.000000000","tz":0},"subject":"Remove Oracle Linux stuff in CI code","message":"Remove Oracle Linux stuff in CI code\n\nOracle Linux is no longer tested via CI after [1].\nThis patch removes OL support from Zuul files.\n\n[1] https://review.opendev.org/664217\n\nChange-Id: I527aba3635ef82e2d2a88ee7f5250bf03e420800\nSigned-off-by: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n(cherry picked from commit ccabadc342477b3c16b11ae52e0194c7916966c2)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/848a089485036b100fbfce817fc5e9ca4f4b0c29"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/848a089485036b100fbfce817fc5e9ca4f4b0c29"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
