)]}'
{"id":"openstack%2Fkolla~668222","triplet_id":"openstack%2Fkolla~master~I98405e3fb32cbec2d9863817a913ed71c5597e15","project":"openstack/kolla","branch":"master","hashtags":[],"change_id":"I98405e3fb32cbec2d9863817a913ed71c5597e15","subject":"Fixes ceph-osd deploys with loop devices.","status":"MERGED","created":"2019-06-28 18:04:48.000000000","updated":"2019-07-03 07:37:55.000000000","submitted":"2019-07-03 07:37:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"668222-1562139475238-5fda7c2d","meta_rev_id":"085d32db92ecb6f5cbb596d0cdafb2ed14ad1e8e","_number":668222,"virtual_id_number":668222,"owner":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":2,"date":"2019-07-03 07:37:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-07-01 20:24:03.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2019-07-03 06:07:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2019-07-03 03:35:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":2,"date":"2019-07-01 08:17:56.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2019-06-28 18:08:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2019-07-03 06:07:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-28 18:08:00.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-07-01 08:17:56.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-07-01 20:24:03.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-03 03:35:09.000000000","updated_by":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"reviewer":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"state":"REVIEWER"},{"updated":"2019-07-03 06:07:22.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2019-07-03 07:37:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1aaa48075f0d89018e9a62f50d70cae5f8f0a8d4","author":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"date":"2019-06-28 18:04:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0278776002fd9407214adab3c6867a18d2f5735e","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-28 18:08:00.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"c7e852212a7ad275eb6c87e59e83109f27e76f8c","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-28 18:45:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/22/668222/1/openstack-check/tripleo-build-containers-fedora-28-master/f53d3c6/ : FAILURE in 6m 02s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/22/668222/1/openstack-check/tripleo-build-containers-fedora-28-stein/acd25b8/ : SUCCESS in 39m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"685c48261348b932bd780d29bf660dcce86534f8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-28 19:40:15.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/22/668222/1/check/kolla-build-centos-binary/aaaf927/ : SUCCESS in 41m 19s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/22/668222/1/check/kolla-build-centos-source/48948b2/ : SUCCESS in 37m 35s\n- kolla-ansible-centos-source http://logs.openstack.org/22/668222/1/check/kolla-ansible-centos-source/ee72129/ : SUCCESS in 1h 29m 15s\n- kolla-ansible-centos-binary http://logs.openstack.org/22/668222/1/check/kolla-ansible-centos-binary/2b1b99b/ : SUCCESS in 1h 00m 05s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/22/668222/1/check/kolla-ansible-centos-source-upgrade/026189c/ : FAILURE in 1h 10m 05s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/22/668222/1/check/tripleo-build-containers-centos-7/a7d2518/ : SUCCESS in 38m 26s\n- kolla-ansible-centos-source-ceph http://logs.openstack.org/22/668222/1/check/kolla-ansible-centos-source-ceph/2bce75f/ : SUCCESS in 1h 29m 25s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/22/668222/1/check/openstack-tox-lower-constraints/5aa62e7/ : SUCCESS in 3m 48s\n- openstack-tox-pep8 http://logs.openstack.org/22/668222/1/check/openstack-tox-pep8/ecf2b35/ : SUCCESS in 3m 20s\n- openstack-tox-py27 http://logs.openstack.org/22/668222/1/check/openstack-tox-py27/3d67fb0/ : SUCCESS in 2m 59s\n- openstack-tox-py36 http://logs.openstack.org/22/668222/1/check/openstack-tox-py36/e4f88b1/ : SUCCESS in 3m 13s\n- openstack-tox-py37 http://logs.openstack.org/22/668222/1/check/openstack-tox-py37/03b5a5a/ : SUCCESS in 3m 41s\n- openstack-tox-docs http://logs.openstack.org/22/668222/1/check/openstack-tox-docs/351c8d2/html/ : SUCCESS in 2m 57s\n- kolla-build-ubuntu-binary http://logs.openstack.org/22/668222/1/check/kolla-build-ubuntu-binary/d815bcb/ : SUCCESS in 23m 29s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/22/668222/1/check/kolla-build-ubuntu-source/4d80251/ : SUCCESS in 27m 08s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/22/668222/1/check/kolla-ansible-ubuntu-source/4570ed5/ : SUCCESS in 51m 25s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/22/668222/1/check/kolla-ansible-ubuntu-binary/8756d4d/ : SUCCESS in 40m 32s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/22/668222/1/check/kolla-ansible-ubuntu-source-upgrade/4468c35/ : FAILURE in 35m 01s\n- kolla-ansible-ubuntu-source-ceph http://logs.openstack.org/22/668222/1/check/kolla-ansible-ubuntu-source-ceph/be4186d/ : SUCCESS in 1h 06m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c6d6432beab25472ac189bbc9fa200e6c7fa64d5","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-29 06:48:56.000000000","message":"Patch Set 1:\n\nirrelevant - CI broken on common : Restart fluentd container","accounts_in_message":[],"_revision_number":1},{"id":"2a221d16f26c385b73c53448bd8f1be68c08c3af","author":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"date":"2019-06-29 07:38:47.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/rocky as commit e3c1e3fe5178810ca9bd273896823e24788838de","accounts_in_message":[],"_revision_number":1},{"id":"7aa322970cc55e162e44450779f9251a633f4aee","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-29 13:58:41.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit 5e8c54615fe6c3a380f945eb1879b607d2bc96a9","accounts_in_message":[],"_revision_number":1},{"id":"82463bac7bf96ec143a386f8747e357c3a6ae080","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-07-01 08:17:56.000000000","message":"Patch Set 1: Code-Review+2\n\nThanks","accounts_in_message":[],"_revision_number":1},{"id":"2c708f807ce9d150d6e5004e62c932a6eb8a5047","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-07-01 19:33:46.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f05d551cd7576d75918b0bf963c29cb6507cbee0","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-01 20:24:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/22/668222/1/openstack-check/tripleo-build-containers-fedora-28-master/c994a8d/ : FAILURE in 6m 25s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/22/668222/1/openstack-check/tripleo-build-containers-fedora-28-stein/4454ab2/ : SUCCESS in 46m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"23c09bf23c258e1e149c59c5080def7f114bdfe1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-01 21:28:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/22/668222/1/check/kolla-build-centos-binary/7b98fe8/ : SUCCESS in 44m 35s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/22/668222/1/check/kolla-build-centos-source/6a71b57/ : SUCCESS in 58m 51s\n- kolla-ansible-centos-source http://logs.openstack.org/22/668222/1/check/kolla-ansible-centos-source/39394dc/ : SUCCESS in 1h 23m 35s\n- kolla-ansible-centos-binary http://logs.openstack.org/22/668222/1/check/kolla-ansible-centos-binary/f01330f/ : SUCCESS in 1h 09m 15s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/22/668222/1/check/kolla-ansible-centos-source-upgrade/e0cf28d/ : SUCCESS in 1h 15m 19s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/22/668222/1/check/tripleo-build-containers-centos-7/a28552f/ : SUCCESS in 50m 14s\n- kolla-ansible-centos-source-ceph http://logs.openstack.org/22/668222/1/check/kolla-ansible-centos-source-ceph/6134a55/ : SUCCESS in 1h 32m 44s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/22/668222/1/check/openstack-tox-lower-constraints/2f47482/ : SUCCESS in 4m 52s\n- openstack-tox-pep8 http://logs.openstack.org/22/668222/1/check/openstack-tox-pep8/c199d29/ : SUCCESS in 4m 34s\n- openstack-tox-py27 http://logs.openstack.org/22/668222/1/check/openstack-tox-py27/6a97db6/ : SUCCESS in 3m 58s\n- openstack-tox-py36 http://logs.openstack.org/22/668222/1/check/openstack-tox-py36/29e0330/ : SUCCESS in 3m 03s\n- openstack-tox-py37 http://logs.openstack.org/22/668222/1/check/openstack-tox-py37/89b4c4e/ : SUCCESS in 3m 06s\n- openstack-tox-docs http://logs.openstack.org/22/668222/1/check/openstack-tox-docs/11b297c/html/ : SUCCESS in 3m 34s\n- kolla-build-ubuntu-binary http://logs.openstack.org/22/668222/1/check/kolla-build-ubuntu-binary/60a1746/ : SUCCESS in 34m 34s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/22/668222/1/check/kolla-build-ubuntu-source/9299897/ : SUCCESS in 24m 49s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/22/668222/1/check/kolla-ansible-ubuntu-source/75f5c97/ : SUCCESS in 44m 58s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/22/668222/1/check/kolla-ansible-ubuntu-binary/ad39b84/ : SUCCESS in 47m 08s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/22/668222/1/check/kolla-ansible-ubuntu-source-upgrade/1896980/ : SUCCESS in 52m 39s\n- kolla-ansible-ubuntu-source-ceph http://logs.openstack.org/22/668222/1/check/kolla-ansible-ubuntu-source-ceph/9efccbd/ : SUCCESS in 1h 10m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"550c01b78d78b7f19fce3deaad50485d8272fbdf","author":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"date":"2019-07-03 03:35:09.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"701641a769faba1abc3644da42e4939d8eb37362","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-07-03 06:07:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"572d11787782f7235b33067c3c72074298f3521f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-03 06:07:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0c6323d89a6a0777d9b69693371cc9efb42a41d2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-03 07:37:54.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source http://logs.openstack.org/22/668222/1/gate/kolla-build-centos-source/c73810e/ : SUCCESS in 1h 00m 47s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/22/668222/1/gate/tripleo-build-containers-centos-7/c3bb1da/ : SUCCESS in 44m 52s\n- openstack-tox-lower-constraints http://logs.openstack.org/22/668222/1/gate/openstack-tox-lower-constraints/966208c/ : SUCCESS in 4m 36s\n- openstack-tox-pep8 http://logs.openstack.org/22/668222/1/gate/openstack-tox-pep8/58ebb42/ : SUCCESS in 4m 05s\n- openstack-tox-py27 http://logs.openstack.org/22/668222/1/gate/openstack-tox-py27/aa2d313/ : SUCCESS in 4m 06s\n- openstack-tox-py36 http://logs.openstack.org/22/668222/1/gate/openstack-tox-py36/5fb8375/ : SUCCESS in 3m 41s\n- openstack-tox-py37 http://logs.openstack.org/22/668222/1/gate/openstack-tox-py37/e27c174/ : SUCCESS in 3m 40s\n- openstack-tox-docs http://logs.openstack.org/22/668222/1/gate/openstack-tox-docs/c77db93/html/ : SUCCESS in 3m 01s\n- kolla-build-ubuntu-source http://logs.openstack.org/22/668222/1/gate/kolla-build-ubuntu-source/c48591d/ : SUCCESS in 34m 33s","accounts_in_message":[],"_revision_number":1},{"id":"085d32db92ecb6f5cbb596d0cdafb2ed14ad1e8e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-03 07:37:55.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f5ac80e332c2a057ee68d95983492bc7e1156b9a","revisions":{"f5ac80e332c2a057ee68d95983492bc7e1156b9a":{"kind":"REWORK","_number":1,"created":"2019-06-28 18:04:48.000000000","uploader":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"ref":"refs/changes/22/668222/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/22/668222/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/668222/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/668222/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/668222/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/22/668222/1"}}},"commit":{"parents":[{"commit":"5ecb9107ea286bd0c1a45d4d95d064c764d33296","subject":"Merge \"[prometheus/openstack-exporter] Changed repository location and version.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5ecb9107ea286bd0c1a45d4d95d064c764d33296"}]}],"author":{"name":"Dincer Celik","email":"hello@dincercelik.com","date":"2019-06-28 18:00:18.000000000","tz":180},"committer":{"name":"Dincer Celik","email":"hello@dincercelik.com","date":"2019-06-28 18:04:45.000000000","tz":180},"subject":"Fixes ceph-osd deploys with loop devices.","message":"Fixes ceph-osd deploys with loop devices.\n\nChange-Id: I98405e3fb32cbec2d9863817a913ed71c5597e15\nCloses-Bug: #1834678\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f5ac80e332c2a057ee68d95983492bc7e1156b9a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f5ac80e332c2a057ee68d95983492bc7e1156b9a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
